)]}'
{"id":"openstack%2Fswift~928968","triplet_id":"openstack%2Fswift~feature%2Fmpu~I0c0b519fdadf212a65bf4c073f403a46e3520013","project":"openstack/swift","branch":"feature/mpu","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-01 03:37:48.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I0c0b519fdadf212a65bf4c073f403a46e3520013","subject":"Shorten MPU upload-ids","status":"MERGED","created":"2024-09-11 20:30:20.000000000","updated":"2024-10-01 05:01:08.000000000","submitted":"2024-10-01 04:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"928968","meta_rev_id":"2938e5547085101fc2e94edc72c2bb8d829c856f","_number":928968,"virtual_id_number":928968,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-01 04:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"all":[{"value":1,"date":"2024-09-26 23:25:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-01 03:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-01 03:37:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-11 21:32:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 21:56:46.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-09-26 23:25:50.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"}],"messages":[{"id":"18ad6bfc4dac8c38d79116e21a0f03bd1721a27e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-11 20:30:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da34b58e35a25ae4d39c3a0d87153fb863b42936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 21:32:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20de7e128fde49afb309ab37eff3dba5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11cad36e9f044eb490f509e5f34d77c6 : SUCCESS in 9m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e695e3548fd4e46aa61c08c0ee17bf3 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b29268c9c729420c92cf944cc26b1a6f : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/491e0475d5eb49989e31e216a92cbe2c : SUCCESS in 13m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bca4d5a96c8442a881a53fa199a0027d : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/726e3771b90f4769b4ffd14e4c411d47 : SUCCESS in 22m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7f3a787f06c74f9c9991f9e1d53bb165 : SUCCESS in 15m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7ffb3f73fe564ba0be2ed0e9b5f9bd0e : SUCCESS in 10m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f4ebe8ed0f4c49c994f17d22f9c65f3d : SUCCESS in 22m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/882e9309990c40a9ad05d69d8c144ab6 : SUCCESS in 13m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af5bf25af2ff413ea130b0d222fda020 : SUCCESS in 19m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/426631b3438a49fc858cff795cf549db : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e307985f8e1646418572b3829c4d0f05 : SUCCESS in 24m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f2f382f1d26a4feb8bb0ed7db932b332 : SUCCESS in 10m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/326c84ddcbce43d797d3c4388bcb1e50 : SUCCESS in 17m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0ee8ab58da7a43e18939eb8093a2ef01 : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d33285034a94617b4c22d2f4dfb2b2a : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4dc160d3f38e4fabb4a5dcc1b0576b3c : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6606c5c341644ff09b1943dfcc59d600 : SUCCESS in 7m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c1d8dd9bf694705b6074c78f16e1120 : SUCCESS in 56m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2d728e80e294d43a4f2c3ea8022e127 : SUCCESS in 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90c08c65fbfe42bca29278a1ca423a0a : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14a8b8d9d7ba4c6282b9dbf7b9951117 : SUCCESS in 18m 12s","accounts_in_message":[],"_revision_number":1},{"id":"28549947944a7e52e782f595a7cd154781934a04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:15:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fda0a3ca2bd94f19bbdac650a09c91d9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2942f32b31048ce95ec01383437c2bd : TIMED_OUT in 1h 31m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6cc301d27aae43158f1cae35f55a290d : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a410c4d2d4fe4c41bf85f251d94c1887 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/075e47820d1843fda434190f8ed0973a : TIMED_OUT in 1h 31m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2ce31ed06bb7d4cc5ed48df2c6ef93c020f81deb","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-26 22:52:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0dd285af6ba8ccbe4894dcbd7c148e3bf24de3c5","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-26 23:25:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60b3e53aa6f798f811d4df75fbaf5ae40c36c335","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-01 03:37:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c260d27c709cc1d67d350631619bc5728df5e2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 03:38:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47fc8cd88859f6cb68921083b4f22872a9c3d1e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 04:59:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e37a7037ca2d43688c65ce7ba6fd1501\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bb72bfe6db4b2e9014108147f9a38e : SUCCESS in 7m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a2cfdfa88a594836b684c81cf92d9688 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0be973bf61d24d1d9fe056414b200239 : SUCCESS in 15m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ec440f9c66d94968a3fe79f68442d51e : SUCCESS in 15m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4f050a116b0646869a73f63ee1d427a0 : SUCCESS in 19m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/63f3e79c59c04183a3f8fb117a544013 : SUCCESS in 14m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a8f9b9258c924df5993484eb4dc6a794 : SUCCESS in 18m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb116483b1094a56bc2e3651bd9ba9ee : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dd174913b2474adbb03cbceff6ffc758 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/52b4bc96f64343df84cf308828206e98 : SUCCESS in 17m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e4330dedd5814274a6d25ab2a5e90445 : SUCCESS in 12m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/054c2602e6e64579b5fb978d00b0cb50 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ab7123def1744751b3d984991a512444 : SUCCESS in 19m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f9d8113313a4c198b5bdb9bf1f2a68c : SUCCESS in 6m 23s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1f7c300bdf894b38973501ac7e3d4ea0 : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/14b4b47db57641e0990493cb7b03ff05 : SUCCESS in 50m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd3984ebe5614f8dbcea05b4ef6c8632 : SUCCESS in 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c95285394e7e48a19b79cfada8780580 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1926a3274081415eb710474c018677da : SUCCESS in 18m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5c9eb450b5af5bae1bd3e4e4e20443963aa85b80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 04:59:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2938e5547085101fc2e94edc72c2bb8d829c856f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 05:01:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0faf5030bd264087af4f9257c68355b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a04dc5df5f482d925ab79b9c52e57d : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/abc9ee011cc24037a56b192157252d04 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71e3be7c5bb1c8ea675af9ba7e0ebe9e7076c2b9","revisions":{"71e3be7c5bb1c8ea675af9ba7e0ebe9e7076c2b9":{"kind":"REWORK","_number":1,"created":"2024-09-11 20:30:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/928968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/928968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/928968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/928968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/928968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/928968/1"}}},"commit":{"parents":[{"commit":"31b3823148bbda17d97a34b1fd7dd5cb389a91bb","subject":"Merge branch \u0027master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31b3823148bbda17d97a34b1fd7dd5cb389a91bb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-11 20:29:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-11 20:29:17.000000000","tz":-420},"subject":"Shorten MPU upload-ids","message":"Shorten MPU upload-ids\n\nThere\u0027s no reason we need carry on the b64-encoded-hex craziness we\ninherited from swift3.\n\nChange-Id: I0c0b519fdadf212a65bf4c073f403a46e3520013\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71e3be7c5bb1c8ea675af9ba7e0ebe9e7076c2b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71e3be7c5bb1c8ea675af9ba7e0ebe9e7076c2b9"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
