)]}'
{"id":"openstack%2Fswift~928973","triplet_id":"openstack%2Fswift~master~I56165dc6d877bd9e56d7160dd28c6c60f339d730","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-12 01:18:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I56165dc6d877bd9e56d7160dd28c6c60f339d730","subject":"CI: Bump up ARM timeouts","status":"MERGED","created":"2024-09-11 22:29:30.000000000","updated":"2024-09-12 04:00:04.000000000","submitted":"2024-09-12 03:25:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928973","meta_rev_id":"df51988be5f150b1722d962a0b6b804d2b4e7002","_number":928973,"virtual_id_number":928973,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-12 03:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-12 00:13:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-12 00:13:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-11 23:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e3e341913b33cba84557470515bb23e2dfb872a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-11 22:29:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f123a12f4e29c083315fa73df02e4d67277c0bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 23:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9036d7957a4849f09ccdc03ff10b3a77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2348a19f7c5846d9a2814d90545e7a4a : SUCCESS in 7m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/56dbafb4b1394eceb767d1d480b4da4c : SUCCESS in 52m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8c52a7e3db024289a808306b7a82ebbe : SUCCESS in 34m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ddf3b2c5174482ba3f5c6a55fad44a0 : SUCCESS in 44m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5cf86cff89424f8eab07efcae3fe403c : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d2203719a424183b47694b342031627 : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/397d4b24c8464b51a19fa7def1c6249b : SUCCESS in 21m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d5dfffea4d64de0acc8d48f3bfacd71 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f613cdc2f3954708bbcb16f8b5555c07 : SUCCESS in 20m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fdb0d0bfac604863ac1b3260144bb7c1 : SUCCESS in 17m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7d48d2e474524ea4915066c8dd1364cf : SUCCESS in 20m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f170040d34c7493e97feb7fe044ae261 : SUCCESS in 21m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e996698e4750407496c57a87087f94f9 : SUCCESS in 22m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c185808ea24e434eb94d6fab864b7439 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5099effac6754b4691a0676ba64a3f76 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/71919c3a108a4f1f9e8a9d5528fceb29 : SUCCESS in 19m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2e12b0d648ca4dfeac6576d13f5fde50 : SUCCESS in 10m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/35883aa33752485b8c2a8788393322d1 : SUCCESS in 20m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d732991c03784a0dbe525c297f8914db : SUCCESS in 16m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e9fa59d1144749b59159768c54a05f49 : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b470160294164e2f81499d887d374cf7 : SUCCESS in 9m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/042c0316c2224fe1a63ed41b6579ef13 : SUCCESS in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/42f22f441a9d48e8ad1d0f6d8fe06fdc : SUCCESS in 36m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46ddabbe0bb843ec8aca28b87379ccf4 : SUCCESS in 51m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a65e51f953654bfaa209e8208b6ca902 : SUCCESS in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4c65bd4a63048d78b4457635aa44822 : SUCCESS in 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae419bb204184768932fe8f928873b6b : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f67f813dcf24ecfa2d9c04d8b4eee85 : SUCCESS in 17m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bed18eb04d94ecea3772dd63f9f7f49 : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c683356af5e18410d74146037fa0b06ec07eed4a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-12 00:13:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db5ce9ca5a458a6c75a444fbefb3caf4d3a62b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 00:14:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34779fe269d3194981542aa86bd660aa4f2d5cfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 00:58:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c750be59b8124648abf4ee84fb67b75d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff8142ab5a8a4905a0f467d7e793c577 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/a4e44af0209f431bb1fe077d2d5f9eb7 : SUCCESS in 41m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5616f47b01654c458abeba87fa229ac7 : SUCCESS in 37m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44a91d8b8a554141af2d00369d17e304 : SUCCESS in 43m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/56b687bd2b144c33881ef77e9a276c4a : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0faf6ad19704c96975ac8bb579df987 : SUCCESS in 16m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03678946d7844b1bb385135ddf7d3a05 : SUCCESS in 9m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3139cbce39d746c18c9017dc78256b82 : SUCCESS in 20m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9bd64169020d43268a1c198b30bd64d8 : SUCCESS in 16m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db5cc7ab7ad0405ba26ea91792b9e2ad : SUCCESS in 17m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/baa130e1dd124742aee9444529f16b14 : SUCCESS in 11m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a82abcf283354ce086cd8d24b7aad789 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8bbdb62c574d41d0bcc2f22794effebd : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a4ae844a13fe4eb89347ea2d9cac3cf9 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c41e943edca04f41acc7e086bb2a54f2 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/71be0e89696f4ebe97d5073e47348772 : SUCCESS in 10m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d4c3c7dd54e412c937b0d169599aeea : SUCCESS in 5m 44s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3316f173427f4e48a715788ad7903c34 : SUCCESS in 6m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6ba0522e0704350ad6a986bad5fe757 : FAILURE in 43m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee952aa99f424506ad1334d3723a4ae0 : SUCCESS in 30m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4033f65293c45babf6cd2f521f4dde9 : SUCCESS in 32m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5019a16a4e594bb3b52ef4e32ce276f6 : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5069a31df40d4ab89abbeb1209a934ff : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65b85ec64d6d4e1796d17d113f2e48eb : SUCCESS in 15m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e67128c212f4248bf7af75c2a07d5c4 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4bf0e7bfc59c964bf75bb65454919bea9ee59b39","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-12 01:18:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79be9c77b81aa900350b6e0bbfa437e15ef1d1ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 02:04:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2245e9e2ef4b43a2a4eb9a826c1024ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fea0c3c52bf493e90a63a4868129325 : SUCCESS in 7m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/58eb2da5351349be9124d1c356bc0a97 : SUCCESS in 44m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab4e34ca582a450c9efeb9f46a63bd7f : SUCCESS in 36m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c27b3013c1c43069d1ea4e6801eb3de : SUCCESS in 44m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/778d00896a2741f78f8e19fe0d8f935f : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3bb4665ef21948189fe05f1897e0d438 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86f2f28a0bc34bc7a3ccf78aa1dc3e0a : SUCCESS in 17m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42ffbdf33bbc47b9a005ec0e1a5ca787 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b9487e62bb04f26a926a162894b18d0 : SUCCESS in 16m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7f484008b8014759b33eb335eb91d3b2 : SUCCESS in 16m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3243c0f0b816400cace7052641011bd8 : SUCCESS in 17m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bd4fba61b1c64bb3b6fa3055a3eee34e : SUCCESS in 26m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0e841ab680b543f189ac9fc139cf323a : SUCCESS in 13m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3fefe9e5b464498ca3fbd719e3d8e6dd : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f7f46fd281344979fe2a27bee9694f8 : SUCCESS in 12m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/34adff197c6940e69d4559e8a276ad29 : SUCCESS in 20m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ef68928706924b10b75648fb0265326d : SUCCESS in 10m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ce268e21dc96453387679aae7c40b7c4 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e6fbc0f05374a7390ddb29f4cd6d123 : SUCCESS in 11m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/420a62250b8a4f0eb79909e1c4828b8a : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/91f7c04fbf364ddc9d3041d569239a95 : SUCCESS in 9m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b525d7773f18400f9235afc8a4854472 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4784cd63848f4c4c8aa67f26ba32ed80 : SUCCESS in 35m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/598fc687e64b481094f725b928b1c2cf : SUCCESS in 35m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e56a56e5270241399692dad51468efda : SUCCESS in 39m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62d7a097cc0a4c3db4ff563596e1ecdd : SUCCESS in 17m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c496441cb4ed40e696ca396e6dca07c1 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/171c048de6194c0abbdb3ac1dcbae272 : SUCCESS in 17m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4f3d6a8d357446ea1ddc27144fb6f96 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":1},{"id":"cedb1af8a252b388f9049b0a7db710a3b5225c28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 02:12:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e43d63d929bcc2e8e9bbbf3f8572b4ae558c865","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 03:25:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf64df780cf4efca096b80ae9d2ca58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2996fe81c803473d8a9357116704c573 : SUCCESS in 6m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/3fabd305665a4c1fae726863e053b4fe : SUCCESS in 56m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7ed3f4ae5d04a8ca9c4a0b9cbc56145 : SUCCESS in 51m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/976b316f5f944b6284c6841efb554bc7 : SUCCESS in 57m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/421817b766994a7ea2fd0f1335795787 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96c33059e1f5497da802ef2e5f2e87b2 : SUCCESS in 19m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2adb76a8157b44b194a9f8fc678623a5 : SUCCESS in 20m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b57b78a407e3459b82c11a43db1591fa : SUCCESS in 22m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e0ab1d4ba8bb4856b89784a268b530df : SUCCESS in 28m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a56cd93c3e0e4054b375fdc73c7829d4 : SUCCESS in 26m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/138602251b764f42aa5e79e436860bb8 : SUCCESS in 26m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b56ad46a05be4c6499a5ef7db6c263c6 : SUCCESS in 20m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/164e3aa4c6d64519b976fec8c92bd95f : SUCCESS in 18m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/58c266fda82f4972bc9215ee5a1cef2b : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/835e4ac9d41747dc868da651977b2f57 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8b44ad994eeb4cb1972f520d7de62228 : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74817c6199074350b7c07e532fcd9f7d : SUCCESS in 8m 47s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4fea5859d6804fa194862ed055ae9705 : SUCCESS in 6m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac9eb33176d04e45b21e9f229cc6b7de : SUCCESS in 48m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa56d845b666465b957d4b711581cd15 : SUCCESS in 41m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/608cf1572bb44bcc8ee96f3cbaccb13f : SUCCESS in 48m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08bb218def1f45b3bf72bffc92d950db : SUCCESS in 20m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ed9733c3214150bae0c1103494fdec : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2643b32ce1c34ab18467a7966a2b18eb : SUCCESS in 16m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/482cbeee0bd64aca8da6690c689f7bc7 : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ce6b28dc4c21016fa7f24f23d64999278e27b8d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 03:25:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7b81416f833ba4810cf5782bedafd646efd47b97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 03:27:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34cb996fe2674b48bfd091c2920018dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a5567df2e07479b923e77a7b0a5b2dc : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e53c86b77135445a928e7fffa3186fc2 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df51988be5f150b1722d962a0b6b804d2b4e7002","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 04:00:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dfbf0a6a2db44cf8b361d047345b70d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76891156889a43afbe6876b750cfdd95 : SUCCESS in 1h 48m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f779ea6c9a94fefb287182779884897 : TIMED_OUT in 2h 01m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e4d671b6b13488d91607267cc69a245 : TIMED_OUT in 2h 01m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6daf6fc8e7654e85a9d14002625799c1 : TIMED_OUT in 2h 01m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14fb7df5789071ebb5eb2776fd14e15a12f6e495","revisions":{"14fb7df5789071ebb5eb2776fd14e15a12f6e495":{"kind":"REWORK","_number":1,"created":"2024-09-11 22:29:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/928973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/928973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/928973/1"}}},"commit":{"parents":[{"commit":"0e41e179003fac2a0ca4bc4c913e9a6d574ddf68","subject":"AUTHORS/CHANGELOG for 2.34.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e41e179003fac2a0ca4bc4c913e9a6d574ddf68"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-11 22:16:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-11 22:16:36.000000000","tz":-420},"subject":"CI: Bump up ARM timeouts","message":"CI: Bump up ARM timeouts\n\nWe used to have a fairly bimodal distribution of successful ARM test\ntimes: roughly 2/3 of the time, tests would take 800-1800s (~15-30min),\nand 1/3 of the time they would take more like 2000-5400s (~30-90min).\n\nStarting around August, the bimodal distribution is gone, with only the\nlonger expected test times remaining. Timeouts have increased, leaving\nthe ARM jobs frequently failing spuriously. Increase the timeouts to\nget better signal from the tests.\n\nChange-Id: I56165dc6d877bd9e56d7160dd28c6c60f339d730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14fb7df5789071ebb5eb2776fd14e15a12f6e495"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14fb7df5789071ebb5eb2776fd14e15a12f6e495"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
