)]}'
{"id":"openstack%2Fswift~929438","triplet_id":"openstack%2Fswift~master~I97a958700b38a8e4ff7871b0ba3ec56e168f6626","project":"openstack/swift","branch":"master","topic":"p-slo-manifest-etag","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-20 16:12:46.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I97a958700b38a8e4ff7871b0ba3ec56e168f6626","subject":"slo: return manifest etag with manifest PUT response","status":"NEW","created":"2024-09-16 10:55:36.000000000","updated":"2024-09-20 17:24:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a34f9ddc7842679abc556912852b11923318cf8e","_number":929438,"virtual_id_number":929438,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-16 12:13:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2024-09-20 17:24:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2024-09-20 17:24:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-16 12:13:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85f88d65bcd601e83496f014d5434d6f188d3dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-16 10:55:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1a48308fc2270831784f51710d38f90e8435c8f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-16 10:57:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3110dda0a27fea2550e9cb3df0b2194f345c1db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 12:13:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fea95cc02db8414584c29db38986c801\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f694986d50b5491b80b04b833a554125 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/8a281cbd667943d094469c2c2a7406c3 : SUCCESS in 58m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/862ed53bcf1148f0853b677a673da751 : SUCCESS in 45m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cd74c0d5b674fdeb79c0411fb26142d : SUCCESS in 19m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ae31670c5824d229605b7fcf56683a7 : SUCCESS in 25m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5027994854f64ab39a646e51e628d51d : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe3ceca6b54d4c9f893ec039fc7a29e8 : SUCCESS in 21m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e87924d5b072450fa8f3a75203ded449 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17abe8b9ce274f2b94c0dc7c0ef45ebd : SUCCESS in 20m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c71d0ba9982948ce9380f0490dd3740e : SUCCESS in 11m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ff9ca0a629404200aab6e426bbc3abe5 : SUCCESS in 20m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/81c9b1cc4a714006903b12986a5e52d1 : SUCCESS in 22m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/acb46ac5915e4a0098d39c3f53c7a26e : SUCCESS in 28m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/452e88d0b3c84b1a8e2dcc0a310fe639 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f469447a24db4ab9918adddcd153210c : SUCCESS in 20m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/50959b5b4b994a55abe2ebd377b6a6d6 : SUCCESS in 21m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6366d6a429094d5da1517f0d8eb266f9 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4143f45c643c4035bea55012f75d790e : SUCCESS in 18m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8453b98bb08e40f2a92c1201d56075ad : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/717c69bf2d6f44d3bb6df3e4caec293a : SUCCESS in 8m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0b0c1f1d671d4d49aab98e427dd6950c : SUCCESS in 9m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/789113a0b8bd4831820a5edeef448e53 : SUCCESS in 7m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bf1df866fd6459b88b57cad479802b5 : SUCCESS in 48m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5837cb7c0f3a4bf7bd93ca840224893c : RETRY_LIMIT in 11m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/866a08eac1a24cf5a4117220835632cb : RETRY_LIMIT in 12m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6caab6ae9464baeaffa9838503fff1d : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb3c5cd508344f9acf863dbe993800d : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b47a6e8e180645df82f86dce14b0939b : SUCCESS in 16m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40ac8bc84f364bd09bcdcf112026ec88 : SUCCESS in 1h 08m 44s","accounts_in_message":[],"_revision_number":1},{"id":"69aece2ee2fb003d4576deacc632d33c2090ed54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 13:34:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2900fb2cbddd4732b5daa00058167a13\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4450fc77aacf47439a16120de458e9ef : TIMED_OUT in 2h 00m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60fd8993831c401eb62c02fd5b4730c8 : TIMED_OUT in 2h 00m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f528cb33d104700a7482254975778ba : FAILURE in 2h 00m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/de41833f0bfd4335926146155a22fa4c : SUCCESS in 1h 39m 15s","accounts_in_message":[],"_revision_number":1},{"id":"66399b6e1ad4c8a585c8db260bbd7f479dd17f75","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-20 16:12:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a34f9ddc7842679abc556912852b11923318cf8e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-20 17:24:16.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7dd6f42a1c8598f1efd22e63c54498f288c2055d","revisions":{"7dd6f42a1c8598f1efd22e63c54498f288c2055d":{"kind":"REWORK","_number":1,"created":"2024-09-16 10:55:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/929438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/929438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/929438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/929438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/929438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/929438/1"}}},"commit":{"parents":[{"commit":"e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-24 10:31:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-16 10:53:36.000000000","tz":60},"subject":"slo: return manifest etag with manifest PUT response","message":"slo: return manifest etag with manifest PUT response\n\nWhen possible, include the manifest etag as an X-Manifest-Etag header\nwith the response to a multipart manifest PUT.  When the request has\nheartbeat\u003don, include the manifest etag in the body of the response.\n\nIf you want to create a \"static\" symlink to an SLO manifest by sending\nan \"X-Symlink-Target-Etag: \u003cetag\u003e\" header then the value of \u003cetag\u003e\nmust be the manifest etag rather than the large object etag. However,\nthe manifest etag was not previously included in the response to an\nSLO manifest PUT request.\n\nIn non-heartbeat mode the etags are known when the manifest PUT\nresponse starts, so the manifest etag can be included in response\nheaders. In heartbeat mode neither etags are not known when the\nresponse starts, so they cannot be included in response\nheaders. However, the large object etag is already included in a json\ndictionary in the response body. This patch adds the manifest etag to\nthe response body.\n\nChange-Id: I97a958700b38a8e4ff7871b0ba3ec56e168f6626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dd6f42a1c8598f1efd22e63c54498f288c2055d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dd6f42a1c8598f1efd22e63c54498f288c2055d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
