)]}'
{"id":"openstack%2Fswift~929654","triplet_id":"openstack%2Fswift~feature%2Fmpu~I1a59b5da9f86d3b816422fbcbf9cee86c32155f8","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-audit-sessions","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-15 15:19:51.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I1a59b5da9f86d3b816422fbcbf9cee86c32155f8","subject":"completeUpload: Parse user manifest before modifying session","status":"MERGED","created":"2024-09-17 14:45:04.000000000","updated":"2024-10-15 17:18:28.000000000","submitted":"2024-10-15 17:15:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929654-p-mpu-audit-sessions","meta_rev_id":"ed0395eefaa1aca58101be2faeb111df3436e035","_number":929654,"virtual_id_number":929654,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-15 17:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-10-15 15:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-10-15 15:19:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 15:43:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-01 21:24:55.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"51865d8cdd5d53db9d25d6a292b037c77ead9017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-17 14:45:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed0a219eac908cbe0578a432853badf4c3334769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:43:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60e91c6353fe4f699268bbb2e4865790\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c80c11792e84ca8942a8dc9d58b9f0f : SUCCESS in 9m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fd047e335ed4dfe873545fe66aa74f7 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd34c82f0db04186a626d49080ead907 : SUCCESS in 13m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b877901067b4e3c8790692bf57d8c9c : SUCCESS in 21m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/669292f5892f45dfb338f0854c853b55 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac102d25bf3c4c7a891ac6e15b96683f : SUCCESS in 19m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cc8c289546448a7bf069655917558fc : SUCCESS in 24m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9665029169cc4c6d9f3c7563ddbe2580 : SUCCESS in 22m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b4e8d39fc766459a82530566ec9cb761 : SUCCESS in 31m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/09dcbe997df34af785c9137d537a2dcd : SUCCESS in 34m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21589c1be3bd4ff3a859116a3d41fa08 : FAILURE in 22m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4dbc0c6d66da42caa13252c52a41282a : FAILURE in 23m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/71093dccc23e4233b441057c9f68883a : FAILURE in 19m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/79c2be342208489e958769c436dbc1d4 : SUCCESS in 24m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/513469bc1ecc4a6c862adb48c4ba33c5 : SUCCESS in 20m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dfaac56636894d06855321110d1ad82c : SUCCESS in 21m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c4c23c8270e44cd9af7a4214627cce5b : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4f9047e3acd74bb1844f6bd66952bca5 : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1af3fcbd954b4700a9fa591ad3958501 : SUCCESS in 9m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8486dcdec7df4b4692cefc43d805f29f : SUCCESS in 54m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e11b4dcd3534e33912c25f35827ded9 : SUCCESS in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46fc3d2bb52144bf9c9ee01d681754b1 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17fdf729fa3848c0b65867cc513e1132 : SUCCESS in 23m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2de2fd78a40c13de008507c7cb640d23836c1fe0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 19:11:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4eae8f8924644e5a5b821653ba95d30\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af2b1f89e9904a5f98b40b6eae4f312c : SUCCESS in 1h 02m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f27b5aa351e4f9db387bcb33d411a91 : SUCCESS in 1h 10m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ea656f1d63c4496b7901a14d949045f : SUCCESS in 57m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c564fb0381944a3ae581b920bafc0e2 : SUCCESS in 1h 13m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bd493456eb3855b1500e3d71ec51d6e9e0e26426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-20 15:52:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e41778a8a399a4089f564a0be5a5496558e71b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 16:39:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc42dbc1839d447f91e103425b0f86e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abf123d1497b402aa45164025f76b1dd : SUCCESS in 7m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f66375ca9dd94e1fa85a81784525b089 : SUCCESS in 23m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bfb19a9f9cc4d8792ed4d58ddf740ce : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e57e55057ce4e6b8c8a7525f2cd8aa7 : SUCCESS in 24m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67a800c34fca47eb9de9f78e968c4909 : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/243d9b982d2d4b5697c69428e107b863 : SUCCESS in 19m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35fb1d0dca91436783a47abd3634b40f : SUCCESS in 11m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2ac05c25decd4b6e9bba6530b375e170 : SUCCESS in 23m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5c4fe22f313d4bd2bf33ddc5fee09ec5 : SUCCESS in 22m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/973185446473491db316927c57b7e476 : SUCCESS in 29m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4ebeb16e87ad4f248857973c5f92c546 : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/774dacf4b5f24a67bdc4418c76c42890 : SUCCESS in 21m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8f07162b6e8743bd82479cd2cedd422a : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2f4e3807d00849859f20604400d93b89 : SUCCESS in 20m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fef91e4a719047119007c565b0b7a675 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d26bc8262b074347ba4167654209bef0 : SUCCESS in 18m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/80b8a850de7946529018ee31830ed737 : SUCCESS in 5m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d29235f1a279412b9bc5657601bf3a55 : SUCCESS in 13m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f10a24ab78f7490895b4adf5a37e6871 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc7ac6138cf04244afb45e3d6c39df73 : SUCCESS in 44m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/153d1f192e2844b8a839a40b94155ec4 : SUCCESS in 21m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/738dd10278ff4f26a1acf9758a21e290 : FAILURE in 3m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03ed617a13344889a0bcbb0c80ec7a12 : SUCCESS in 24m 59s\n\nWarning:\n  Comments left for invalid file test/functional/test_mpu.py","accounts_in_message":[],"_revision_number":2},{"id":"a13c54d5d4f6495522cad04cd2923a5fb04db32a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 19:59:03.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63642bb604924a6984f3cabbc5fd9acf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4b796bc0fd84294ae25684b18ed4b8d : TIMED_OUT in 1h 30m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7905271e646f40d282de403eba0593e9 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c29c84d236d4f6c84c73560e0de0f00 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fef4710137fa448bb32a986905c82066 : SUCCESS in 1h 30m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5aef794d67367eb8a71a1285865bfb1f8ce7d8aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-23 18:26:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b254d297ac367c17a32f36da2d824ab48df1730b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 19:20:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5bc38e577f743809196c66ed32d8a35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/112ac43d851a4b6abfa763c140ab4bd9 : SUCCESS in 9m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3ea8947f7254613af32c1f79f8431d4 : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58c0f0198818441dba6da9aeb8525ca6 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff6b18ebe32b476c9b74f1aa24bb928a : SUCCESS in 20m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41195eb31e97442eada6949cc5160c9a : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/859ea2c2bc2a453d85bd59886ab0cedc : SUCCESS in 18m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/888d3469e2d645f18c9f60ff48b88b1f : SUCCESS in 20m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a2e2e5ab0b20456e98f7da3774aec6eb : SUCCESS in 24m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b83ba225394c4fefb8e6e24ba2ec12a7 : SUCCESS in 25m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f6c73dd6ad0242c79e61d0b897add998 : SUCCESS in 31m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/38c4098e7bb843e58318019a8c41b561 : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf869e3fae5944fd8204bf29e8a08114 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a51ba2806f464381895b67a84021bea4 : SUCCESS in 20m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/413a82e2ffb1423f83785b6e006123ae : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/224f5c93b6e14495b85d038b36c9a301 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6b87e2c6810a4b7ab14a17bf8ae00050 : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30626bdb219b4c629b687f4f940b9c6c : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/757527bef26b4818999ebb0af39cf4fb : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/614926fd6ebd4045bbc0ba7fe4d32144 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a1b713536d546f1a3417386e2a4d917 : SUCCESS in 42m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bcc7b6d57a64839a1be00cb3eab45bc : SUCCESS in 16m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8433bfffc6514468852b84dd95912d1c : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2291ee64fd164ac1ac6cc30bac5fc443 : SUCCESS in 18m 15s","accounts_in_message":[],"_revision_number":3},{"id":"0de23bf6891316ff47e56417ba2614f2823e9224","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 02:16:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f53a852d32e40dab3a43e23a90d9c83\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8309c9c9d5e4382863f21c8f0cf76a3 : SUCCESS in 52m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/141143e8b3c047b3aec5d45aa47431fb : SUCCESS in 52m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c320a6dbd93c48fb8d8e181ab2b04abe : SUCCESS in 1h 12m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd0889ab663e4b94b270dcf7d7c62076 : SUCCESS in 1h 21m 24s","accounts_in_message":[],"_revision_number":3},{"id":"4fa544afc5ab8e74d48f2ede09385a44cbeec108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-27 15:58:04.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0353d2943e38057d651efd85aeb0b2852302b6de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 16:53:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be81ca5e4e1848508a0561f3e8186c9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cdf9c495a424b7ea84ce48602c34bcd : SUCCESS in 16m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03d97d25af8e460aa67a21fedb8cb31e : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b10dd6c61ee47799ed3d8ad079ccb86 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa3155a0a5954d67904d0cb28883b41e : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/168c40a305bd4beea26f56fde685f946 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b745b10fa364c20a39af0614acee7ab : SUCCESS in 22m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4479a6d9812c4f19bf3561123171741f : SUCCESS in 9m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0eab33ce4c034ea18933d105f57aa7b9 : SUCCESS in 24m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c25dcbd8bcdb441b8905a301c2adaf4d : SUCCESS in 22m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b31d31d8e8dd46faaa06bc1b2eed1e67 : SUCCESS in 31m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a5867e81f5c4aa8aa168d7d1e351824 : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/85eb274b48f64917aeb7f9f10dd80e1a : SUCCESS in 20m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d33fafd8bc61459ebed495dd961b8ab9 : SUCCESS in 21m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fdb3025f81c94dca87377d7603030c50 : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/33c0067261e341f397c4c1a588a2e0bf : SUCCESS in 17m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ceb29295800b4498aa8e2e71346ca7ca : SUCCESS in 19m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f8facceac75454fada788aebf66bb39 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7223d460c0344df29e825ad201289d40 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/34fe380a98a2471cb5a467082c7cd4df : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0392e90f3c3c40419930453f85638da0 : SUCCESS in 49m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bcedb82169348fd9ac36890caf50321 : SUCCESS in 21m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c5e6bf7076e4dddb04fee2752010d4f : SUCCESS in 8m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c19f0c07f65941118b4f75325f20fd13 : SUCCESS in 20m 28s","accounts_in_message":[],"_revision_number":4},{"id":"70bd22501d2a0e786fb50099d1fba035d805a2e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 19:47:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09fd89509b5c45a3970b0e8ba55188ea\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70102ebe91e64095a35a6402da302f69 : SUCCESS in 1h 44m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa47fca3538d40d9922fed39f13e27c3 : SUCCESS in 1h 48m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a911282072a431cb1461fd61da15a38 : SUCCESS in 1h 33m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c71f48f5bad74b258806816305fdf0e2 : SUCCESS in 1h 21m 52s","accounts_in_message":[],"_revision_number":4},{"id":"bb1c33ce6992bf10f7f5697ed737cba18089fb47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 14:59:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6bff6611ec1c4072ce8cc12db7af6c91e6c55932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 16:21:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/519e21ee8dfe4ccc9a555ea0c388834f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b09b5f51e0f4ad8b54f11b101d11ce2 : SUCCESS in 8m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/110737026dea4b2f817ca9d9d8948505 : SUCCESS in 14m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ff254d5f3194a2fa98c6b9e82b0aee9 : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f34e8ce1362401aa2de1840d89f7169 : SUCCESS in 23m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1de3d4764f4c4862af441550e87ce450 : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac37ad8ca74048f9bc5bfa9913fda284 : SUCCESS in 14m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f8d05b8310b8494a9d2d3b4bf9e5617d : SUCCESS in 24m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/22afbad667c4429b99ce740ab5026ee3 : SUCCESS in 17m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/05d9c15020fe4d63a801f466fcd08b63 : SUCCESS in 15m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0d3a1bc2fa5a49bfb0254a4953176c6a : SUCCESS in 25m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/90097c98b6294b8a89517be936ff20e1 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/91ba51e43f3c4f5484dbf9969841bb9e : SUCCESS in 20m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/51e9c351f1ed4623bc482b9d0f6f4184 : SUCCESS in 17m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ae1c9dd3347f470393f109e05c6bed12 : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1beff7a1e79e4d06bf613cb2673b8530 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/39df25312a0f477e90d263c1e099dbe5 : SUCCESS in 9m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c7d6a946040f45749721e5a69af172da : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/129a86f1649d438d9091140e9e43946e : SUCCESS in 13m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/061a7806bcfc4c86ab7cb8a0f1f8a46d : SUCCESS in 6m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f615b3f28e5d405f9890858e9299ea2e : SUCCESS in 55m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dd056a1a09542fcb530d486e6eda72f : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e42451c81ed4c33849cde78f8dbe809 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2291fd06027a4c3f85b06a1a6a6ba5f0 : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":5},{"id":"cfdd18d51728ce8ba6a31f9a91ed7449a789d714","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:12:07.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf1da2379b64dfd84533b256e49c31d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/270b837a77584a9f89853704bfde9f97 : SUCCESS in 1h 22m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eac17d3f1bd840c0b4f5d2d6e6320968 : SUCCESS in 1h 04m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/344e4780e7734c0fa73d6e6a8c1f3393 : TIMED_OUT in 2h 00m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f3f5b71ead24a51b01a4efa8827ebbb : SUCCESS in 1h 46m 36s","accounts_in_message":[],"_revision_number":5},{"id":"4bbdb4005327acbe29f91896535664a99065b71b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-04 14:15:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e9fc7acefab83e5408bbe23a11885653261638e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 15:22:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2d05eed8d5c49148301da868f9097f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ec10a4e6564149ad94b6c667b5dfa3 : SUCCESS in 8m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd7f5a5f6a40408abafce7425d64a0d3 : SUCCESS in 10m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ed3858851f64bb09aaac2fd1744e6c5 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/493ae58f7cbc477da34ad3e301800f16 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/926d638cc67f47738b5ec40b1b489282 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c3d832cb11af42c3b97c651e4a54a029 : SUCCESS in 21m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/334102ca71a14132a06d496971fd261b : SUCCESS in 17m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/96d49305c7fd4be59d92b40428b44936 : SUCCESS in 20m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/205a386480e3400eb8e20fde0678eef2 : SUCCESS in 23m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0789a724e47641b3ae3c483db25d5bd8 : SUCCESS in 30m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14781ac7a79e444dbd5c8d839c3aaa87 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7a180a73285c44e599f4fe32170758cb : SUCCESS in 11m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d22700c746714a3e9dad70ae4baa795f : SUCCESS in 11m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/17f408ba184a498eb237af6167195953 : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b59b80285214140ac3f8aee50ed34b8 : SUCCESS in 9m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/10a7ea8350d6473a944f098e4fd93de1 : SUCCESS in 18m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/775d678729f646a9b2399d4346badd69 : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a58ae6a7bc8940f5a00d86d7bfa49ddc : SUCCESS in 13m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/55ad88428d76408da10ebbf3a7d78c66 : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1217578e7f644b8b205ac0142ec258b : SUCCESS in 56m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82a501810dad4829968f8d4c47a33fb3 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b438d9aa97a940d4becfc8d7ca7b1f5c : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d02a222355184de1aea0f718547817eb : SUCCESS in 17m 01s","accounts_in_message":[],"_revision_number":6},{"id":"2f0c1cc06ddae61ae162a13157da88e0201a960b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 16:59:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6566dcd345a5432e86b703edeabc9f23\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ede34224f1a94efa8c422dcf24de8a60 : SUCCESS in 1h 09m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/727593d242a84975a16b9081aaa53716 : SUCCESS in 1h 18m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6134345eeea248e8b9b230e7c41a2a6c : SUCCESS in 1h 25m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b04834638e4b492f825a44f3c9eab30e : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":6},{"id":"08953da6a0c9e9b8cc352eba69dd26c636d3c3f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 13:09:46.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"fc4fef830997002d64a74c8b8e2635c01309f6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 14:11:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bcd8bfe6c044fbc8e338af5079fbb93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9f67b233349418abe6c8575f9767f07 : SUCCESS in 8m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e23a7c5866364d6699acb4f3600bcdbe : SUCCESS in 21m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4091580de20f4864839b4f4b254e6b53 : SUCCESS in 21m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/649174e6266a4a55bd79303a7da7304f : SUCCESS in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a266b2f06634e7c8f53f2e96e545fca : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06aa9c2c6a3042289f70a5aeb7ee6446 : SUCCESS in 20m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5742b3b835a47c99c71b83f3b561dd0 : SUCCESS in 20m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2ec65acac48b4aacbc1abc73960c92bc : SUCCESS in 20m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5e131fd505904ec78fb5a76e3e83f4a8 : SUCCESS in 24m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b06ed2f638f54f9d9bf6b7de11536086 : SUCCESS in 27m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a7787cb65e242168c287a60e9a64fb9 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22b817d1ced44acebbb0e04d0a7a018c : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60a76d897260463e961720a363a931a4 : SUCCESS in 21m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ab3a37ae37fe4c3f98f9633108794fe6 : SUCCESS in 25m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cf424d334c7c47ecacdb77ad2abdfd6f : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/361de930d8d9451399ac2251d56ac945 : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf2b2c64d278409dadee830628a5e452 : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d5bef6c7b5dd403aacb0fceac333a696 : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee36dab0bb864cca9aeed0ef8972c0de : SUCCESS in 9m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/96a274f7dfab4d958b489ad88b719bbc : SUCCESS in 46m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b21467db118949b79f61e37a14177b4a : SUCCESS in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27e554bad4047d9b01d753a27962fd1 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b0fd849d8a54f7e89ae0665831204d8 : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":7},{"id":"7a2e7588ac12a2dab1a58841d31b2325ff221b26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 14:55:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28aac3c693a540d1aa9eed8af7029e06\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10ecf16ddf374886ab8d7af46445b74a : SUCCESS in 1h 22m 05s","accounts_in_message":[],"_revision_number":7},{"id":"d6c5d2b11970b142a4b7e64cf74d44b9c5605f5f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 15:19:51.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b3859ef2955ceb1ea4bf88a702f73c4dd259a766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:20:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a228a4d84382e687b42ee6719d8543598cfc1249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:15:59.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ecf7f6dadf642429ad42f50a5def36b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14261410fcc94204a4d2a44ec89e04a3 : SUCCESS in 8m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7168217ec3aa4d55b60bc789e3d71150 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6f6bf1e746fc425f912aba9f0151517b : SUCCESS in 21m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6a0c14377bd0429aa0620ee39ed5d0e5 : SUCCESS in 20m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/38e4d94b26404c0f92e234099d305b7d : SUCCESS in 18m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d7daab18f80247948a763c1e8dfc3ad6 : SUCCESS in 19m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/55575fe474ca4da791a00efa2ddad7a4 : SUCCESS in 25m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6de9db3e09374d97887d1b290eab8982 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/74e2f65be1254eb1abf4af14fbab6ecc : SUCCESS in 19m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c2955d096ef04f3b9f1bb3febe964dd6 : SUCCESS in 18m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/00a62f0ed73e48f08ee1dc2f1e208805 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c18adf1a5d7f4939ac3e4bc18fca28e3 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/64d686c9e24e46a98ff17163f3c35bea : SUCCESS in 10m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/935d84561e454770a499bc54dfb92311 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4932470d4db4cb8a2ce5b5170957ed6 : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3756fafe90514471b7b787c23ce1a001 : SUCCESS in 52m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de6cbb487de5483098de9f40384569bc : SUCCESS in 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aca384f7919485eae8058bbff809d0f : SUCCESS in 3m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26e6c4098e4d44a2aa3b132c0dcc6939 : SUCCESS in 18m 28s","accounts_in_message":[],"_revision_number":7},{"id":"f6668410bcc7b65f84db8f0dd627a3a3de563191","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:15:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"ed0395eefaa1aca58101be2faeb111df3436e035","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:18:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c269568d31742d49f424dce98b27491\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe4879abe1a44a685e20719bbcc8cb5 : SUCCESS in 57s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0a5b9e237add420794f56ffc35fd911d : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"44174116807b80e50c48e96534aef55ea5777a6e","revisions":{"f3c04b1dcc4beb1d754f80614bbd5261237ff2b1":{"kind":"REWORK","_number":1,"created":"2024-09-17 14:45:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/54/929654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/929654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/929654/1"}}},"commit":{"parents":[{"commit":"b052f780e60d913b396903b4cc20b6881894aab3","subject":"modify MPU entries in container listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b052f780e60d913b396903b4cc20b6881894aab3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-17 14:42:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-17 14:42:21.000000000","tz":60},"subject":"completeUpload: Parse user manifest before modifying session","message":"completeUpload: Parse user manifest before modifying session\n\nValidate user manifest is valid before setting session state to\n\u0027completing\u0027.\n\nChange-Id: I1a59b5da9f86d3b816422fbcbf9cee86c32155f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3c04b1dcc4beb1d754f80614bbd5261237ff2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3c04b1dcc4beb1d754f80614bbd5261237ff2b1"}]},"branch":"refs/heads/feature/mpu"},"5f8ca461468a6dfd4539ec7fc61455ff9482b67a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-20 15:52:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/54/929654/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/929654/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/929654/2"}}},"commit":{"parents":[{"commit":"685301e676c02bd983cc6d393eeb91a2fc4efe62","subject":"modify MPU entries in container listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/685301e676c02bd983cc6d393eeb91a2fc4efe62"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-17 14:42:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-20 15:51:24.000000000","tz":60},"subject":"completeUpload: Parse user manifest before modifying session","message":"completeUpload: Parse user manifest before modifying session\n\nValidate user manifest is valid before setting session state to\n\u0027completing\u0027.\n\nChange-Id: I1a59b5da9f86d3b816422fbcbf9cee86c32155f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f8ca461468a6dfd4539ec7fc61455ff9482b67a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f8ca461468a6dfd4539ec7fc61455ff9482b67a"}]},"branch":"refs/heads/feature/mpu"},"1a3fe24b3124119b1928bb6d55eeb06fe971736f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-09-23 18:26:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/54/929654/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/929654/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/929654/3"}}},"commit":{"parents":[{"commit":"30af7847cf89d73570fc47dfba0b11b331a6b0af","subject":"modify MPU entries in container listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30af7847cf89d73570fc47dfba0b11b331a6b0af"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-17 14:42:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-23 18:11:24.000000000","tz":60},"subject":"completeUpload: Parse user manifest before modifying session","message":"completeUpload: Parse user manifest before modifying session\n\nValidate user manifest is valid before setting session state to\n\u0027completing\u0027.\n\nChange-Id: I1a59b5da9f86d3b816422fbcbf9cee86c32155f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a3fe24b3124119b1928bb6d55eeb06fe971736f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a3fe24b3124119b1928bb6d55eeb06fe971736f"}]},"branch":"refs/heads/feature/mpu"},"07115fea3bf9e79a0cc9a5edab0718a0d2261d92":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-09-27 15:58:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/54/929654/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/929654/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/929654/4"}}},"commit":{"parents":[{"commit":"7b55f1bdfef4b0ccb6ab6a1dbbe599797b2c8a86","subject":"modify MPU entries in container listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b55f1bdfef4b0ccb6ab6a1dbbe599797b2c8a86"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-17 14:42:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-27 15:57:08.000000000","tz":60},"subject":"completeUpload: Parse user manifest before modifying session","message":"completeUpload: Parse user manifest before modifying session\n\nValidate user manifest is valid before setting session state to\n\u0027completing\u0027.\n\nChange-Id: I1a59b5da9f86d3b816422fbcbf9cee86c32155f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07115fea3bf9e79a0cc9a5edab0718a0d2261d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07115fea3bf9e79a0cc9a5edab0718a0d2261d92"}]},"branch":"refs/heads/feature/mpu"},"6571480321864ba6689c9dd44688767ee96d8f6f":{"kind":"REWORK","_number":5,"created":"2024-10-02 14:59:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/54/929654/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/929654/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/929654/5"}}},"commit":{"parents":[{"commit":"8858d57b4da0d0d8a7f3ba22c765fee0572c61b6","subject":"modify MPU entries in container listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8858d57b4da0d0d8a7f3ba22c765fee0572c61b6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-17 14:42:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 14:59:09.000000000","tz":60},"subject":"completeUpload: Parse user manifest before modifying session","message":"completeUpload: Parse user manifest before modifying session\n\nValidate user manifest is valid before setting session state to\n\u0027completing\u0027.\n\nChange-Id: I1a59b5da9f86d3b816422fbcbf9cee86c32155f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6571480321864ba6689c9dd44688767ee96d8f6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6571480321864ba6689c9dd44688767ee96d8f6f"}]},"branch":"refs/heads/feature/mpu"},"7ce60dffd4d19b59bbffea390db1ee7dd6e97035":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-04 14:15:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/54/929654/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/929654/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/929654/6"}}},"commit":{"parents":[{"commit":"6c2a8c3bde05d17dcf0b7333054c546646d50670","subject":"modify MPU entries in container listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c2a8c3bde05d17dcf0b7333054c546646d50670"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-17 14:42:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 11:48:02.000000000","tz":60},"subject":"completeUpload: Parse user manifest before modifying session","message":"completeUpload: Parse user manifest before modifying session\n\nValidate user manifest is valid before setting session state to\n\u0027completing\u0027.\n\nChange-Id: I1a59b5da9f86d3b816422fbcbf9cee86c32155f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ce60dffd4d19b59bbffea390db1ee7dd6e97035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ce60dffd4d19b59bbffea390db1ee7dd6e97035"}]},"branch":"refs/heads/feature/mpu"},"44174116807b80e50c48e96534aef55ea5777a6e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-10-15 13:09:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/54/929654/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/929654/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/929654/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/929654/7"}}},"commit":{"parents":[{"commit":"8639253088472422f294afc7876e19f76b9e6335","subject":"modify MPU entries in container listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8639253088472422f294afc7876e19f76b9e6335"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-17 14:42:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-15 13:05:20.000000000","tz":60},"subject":"completeUpload: Parse user manifest before modifying session","message":"completeUpload: Parse user manifest before modifying session\n\nValidate user manifest is valid before setting session state to\n\u0027completing\u0027.\n\nChange-Id: I1a59b5da9f86d3b816422fbcbf9cee86c32155f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44174116807b80e50c48e96534aef55ea5777a6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44174116807b80e50c48e96534aef55ea5777a6e"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
