)]}'
{"id":"openstack%2Fswift~929688","triplet_id":"openstack%2Fswift~master~Id5b0bf6efcb2c16e10334d2edb268194fa1ec008","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-18 15:23:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id5b0bf6efcb2c16e10334d2edb268194fa1ec008","subject":"CI: Include --domain in more openstack commands","status":"MERGED","created":"2024-09-17 17:24:29.000000000","updated":"2024-09-18 23:23:34.000000000","submitted":"2024-09-18 17:59:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929688","meta_rev_id":"e0531edccbc498507c99313ab29d02cf3229548a","_number":929688,"virtual_id_number":929688,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-18 17:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-17 20:12:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-17 20:12:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 20:10:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8af49a242d81b5d781a72685cbdb69d8a5c3afcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-17 17:24:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a07f354019d18ccc69ec750127608cb02267cc2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-17 17:30:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cba7ca30d001eb354cff6a8782495ee5094da585","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-17 18:13:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7890c709b1125e3b6e8e61ec3221c3e685eca103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-17 19:05:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"59e5a4229cc711ca5d604b3b40b35625e394c4bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 20:10:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d44f49c77bff4dd98cd5f37ac1fe360a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e1ef94a75324f97a276f728f855b3c9 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/cf729f3b894e4d96b92dca62a2cbd076 : SUCCESS in 1h 03m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/41ce5b857ac74afa8ba537764869390d : SUCCESS in 54m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2e255e663df492a87a451fac0460e6d : SUCCESS in 22m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87d817bf114b46b5bce23fa97e41b832 : SUCCESS in 22m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bcac888be0a54e85a3d12fc64f6abd7c : SUCCESS in 22m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e401673a004b4da389c20520e70f397c : SUCCESS in 14m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/361f85e256cd4c9c9fc9c6a75eac5aa1 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e307401d727f4befa3a1e283566097ea : SUCCESS in 11m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6a81a2d8cc7c4232a0dc75a15b152c2a : SUCCESS in 10m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2a80195b6dd84fb5866c705f3e3880df : SUCCESS in 21m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2725db9254544589a6a18a248be8511a : SUCCESS in 29m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b9e49b075a664ac199937ef87d2d903a : SUCCESS in 29m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b7da29be58204b8d9fa67d5aaccb9143 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd84f255e3f9482f8266dff50ee71158 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c013b715fab24f0a9db6038f585d417f : SUCCESS in 20m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e22beb84231a4e7ba7e6e7c0c26dc489 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7dde7bf84a54456282467437f61babb8 : SUCCESS in 18m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aee8287da9d74332b8ed5a08f7e806aa : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2640c25196643289be8869d17dbc2fc : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47f8e97a69714aa08a7b1210f1f1a80d : SUCCESS in 12m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/33c8be4e488d495b97bd3b719cc210d8 : SUCCESS in 6m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/abd15ed4c1204fcab7b23405530851a2 : SUCCESS in 50m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59ef790be3104b1a9b8e6df0dd949e15 : SUCCESS in 50m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/905ada04fef84946acece0407e3cbe3b : SUCCESS in 37m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0692aad9116e4df394dc082db9d0e1ea : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7875eb1097ff4544b6f2c22d95551518 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b3f31f7bada42e5afa6bfbba521c2d8 : SUCCESS in 17m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a151a4a5e7aa4a70bcfe4c5e4f8da189 : SUCCESS in 41m 44s","accounts_in_message":[],"_revision_number":4},{"id":"68b2b6cf75f46e1224dc7406cd24cdd5e2ce84c0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-17 20:12:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c13932c71fd57620b4b691ccbf20a6851052972c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 20:13:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e4734ac7e150c386189ec17111d262c9bd0792c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 21:19:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e152482ad4647e58e84393e54727db7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3b35e3ca8a946ff927b0b506ba3e288 : SUCCESS in 1h 19m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b55894d81c93400ea2b4a1920799b29f : SUCCESS in 1h 09m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c90df331af96410da4e8b2ecb3b323e2 : SUCCESS in 1h 11m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d34d1439ed6141cb82f052998ab96e31 : SUCCESS in 1h 20m 44s","accounts_in_message":[],"_revision_number":4},{"id":"12f8b5f3f92c3974b390983c7aed900afc65a8c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 22:15:54.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75f044b59d2e400787ad862102e6e56a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45afeec9e40c4d9bb250012fb80e249a : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/801297a41177445fbd1030629022dbb6 : SUCCESS in 54m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89b34dd23cf043388ead7e915944d1db : SUCCESS in 44m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0fc1f2fd82f470590fd295ff50673c2 : SUCCESS in 22m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2292c1d491e441289ccf2ad57770f943 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/72fdc232c2e3406b86f794fbaca90823 : SUCCESS in 22m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a2666e9b19d146429b71391cdcb13a8d : SUCCESS in 11m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/259e133f57ed4ba79ab66d5cb9ca59fa : SUCCESS in 19m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bd50a78eda194c5893109f1cbb31799c : SUCCESS in 22m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/57e85c198e7d44d8a0c968adc3845704 : SUCCESS in 34m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aeb60a1692cd400f9ec195806eb86d41 : SUCCESS in 12m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/64882ce09eac49b9a9482e44a1888016 : SUCCESS in 19m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c2c76f1b6def4cb6ba8284f3704366f6 : SUCCESS in 19m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5eba82f45d9044f198c7f140ebce9075 : SUCCESS in 10m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/36791c2fad884c3793029121b7941298 : SUCCESS in 21m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3252eaeb4b54962acdc461d8c7fd49e : SUCCESS in 17m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8510c5fb66a4430a9594c4e2ffcaf1ae : SUCCESS in 6m 40s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/99981e90be294dd08744e5d16ba34cdc : SUCCESS in 8m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ed0560c57754d519653a6a04293a146 : TIMED_OUT in 2h 01m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f349768805a44afbb20f0eb3e5fa7db4 : SUCCESS in 32m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/016ae98992f94f48a69ef75e5244d1e8 : SUCCESS in 37m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe0b51f6b4de4ecea6f59c98dc8847c3 : SUCCESS in 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ab3d2ca0194be9ae728ac97359665b : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e968108cdd144b65b0d42133d9420a52 : SUCCESS in 22m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/311382535a344792ab62b68d880de536 : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":4},{"id":"1595a2306a6e2f708ce9a05b99f06aa341f1dc70","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-18 01:15:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"30563ef3dd0cc7e5eda152c542cb691615d24faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 02:22:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbe14cbf07a7449dae6d5f78362823c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eaab66d20334aa4864adec893d6adbf : SUCCESS in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/3d29b699473d44ed97ac68b0a1d2d83f : SUCCESS in 1h 02m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd575c5a2aa744919c9fa931b4c57ae1 : SUCCESS in 46m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16b6bab0c62148b29cb82a9f04e85446 : SUCCESS in 27m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75fd760852c24d21af560d7c42370b1e : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31545cbb98794dd0a96b755a0808660b : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6ac70c61d9a4c63849aefddb127b1a6 : SUCCESS in 19m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f89e053ed70a4611a032181c938392ad : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2655305db3804312af5f0ed5c7bd1ec5 : SUCCESS in 18m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b33a280034c1420596e4986f394811cb : SUCCESS in 18m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c7893e9c68ef40b0af6b2276d66b47e2 : SUCCESS in 22m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/754388c5ed214fdbab610a4ce447b992 : SUCCESS in 14m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d64d7ea9024a4df8852f8185cac91b8a : SUCCESS in 18m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/69ac148987e043a380849590f9daa28d : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7cf69b787aa40a4a7a191f650589456 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3bea9b48125b4aba8dfda43e8e9b3208 : SUCCESS in 19m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7102b3f757944b4b87d2c8ee6ae6cee8 : SUCCESS in 18m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/06c104c1665c40a3a0869537c5fd92ff : SUCCESS in 17m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c315b83ee7144ce093ebe45d2dc74aa8 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/48b0a361fb6d496a925d7c59138860fc : SUCCESS in 5m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/35ba53043577410a80f53fb52927399c : SUCCESS in 13m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7ab8453529e24da7b0b3d1458b2c465b : SUCCESS in 5m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc1c8c3d15c0471c95fd7e5b5a74e1c7 : SUCCESS in 58m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b134cd170d784753a2bfaccd5c9ec58a : SUCCESS in 33m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68e40a9fca1e491ba05f9e4f4e969a65 : SUCCESS in 49m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f27c39aa52c4b8c9934ac9da4a57f46 : SUCCESS in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2710e92c800846208200b26b35522c96 : SUCCESS in 3m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3fd93d4076624c82a45c7610410da581 : SUCCESS in 17m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b83341c14490484d8f674bf044bf3f55 : SUCCESS in 38m 08s","accounts_in_message":[],"_revision_number":4},{"id":"128831f7bc0da1ace84136b299a0e579687ed478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 02:23:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"24a59ff1c4c716ccf02bcdba0d4a2a8be1d40b1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 02:44:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b7b5b837834230ae5d8dd22cbc5282\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/639cf305a52942e99801249d65c2bf2d : SUCCESS in 1h 18m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e1572ece8b34d6e9e4d66dd6005e86c : SUCCESS in 1h 20m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1905cabbd2d4a0ab59a3f8e947fab16 : SUCCESS in 1h 27m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c466cf8c641486aa4538791e1f6b138 : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":4},{"id":"d445079df769d98264bc4e3f23f1266de984edad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 03:36:40.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9284fdd3a1604ea39b0d00b6de03acc6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebbe7e655bac43aa80d5a528afec1c4b : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/0bb214604d7c4bb9b2fa1435c7696f38 : SUCCESS in 56m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9017076d9f2d422fbe93c71b5e6b973d : SUCCESS in 52m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45d6164561294dd48a4961867ed6813f : SUCCESS in 32m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9bbfffcad8ef4bf2b9e0a323ec0bdac6 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c570ebb6c6749e280e18f284abd9821 : SUCCESS in 18m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b287fbe562f4421d95c99993741a8184 : SUCCESS in 20m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ee3ce541150e4c8f95d32a2768716474 : SUCCESS in 22m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3ae431613d054acc828e312a45b06d07 : SUCCESS in 17m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/de1f5c8e735b4b7d98b26bc4bf6713ff : SUCCESS in 34m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b62de517c3624ad08d6dec06430b6a40 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6098fab88e8043059dff3554f8262e53 : SUCCESS in 20m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2534a5ab33d84ae782125b6e20e41143 : SUCCESS in 18m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/45ecb1d43d9740be9c342441b0143803 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b57914a7a9140c297722554bda3ecd3 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/754e31873eec4057a4201ca41a96053b : SUCCESS in 17m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c570095acafd49c780b2a96bc2b8c69b : SUCCESS in 5m 57s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f4d2787b70e4403497dca9d95be03a6c : SUCCESS in 9m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9ae50418b1d42a68a30d0352e71bc77 : SUCCESS in 46m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f65f8b864644ba386d30b20ab0d7d3a : POST_FAILURE in 54m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f135ff222fb342c3bcb4c9e1c69a0e61 : SUCCESS in 47m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/001f2ff9a7dd487ebed93547c307a712 : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1564499f90e84025b68d1d279f7e0f96 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e287adfbe51242e9a68461198f549f84 : SUCCESS in 20m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e574a59a4c934df08b1df3fe270b2c71 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":4},{"id":"4bdb29408627fa6239624671d8a5c6f13a9fe04d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-18 15:23:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aeba49fc23352ab9ceb30a67d9f362bd9634d882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 16:52:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9ded52ab6514373984a9cc797ad20a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95549e19d9e846d483a3df262ebac8c6 : SUCCESS in 7m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/8847d1ae23ce4af1acecd823b6dd4f5c : SUCCESS in 1h 01m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c0850da76ca4c45966151c9ee216d17 : SUCCESS in 1h 06m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97577b8d8fcc4bed82d82065a312ecfa : SUCCESS in 43m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e71e674708e747779fbc4213ac80a698 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d42b210ed4944d8a9261ab59bb1777b : SUCCESS in 13m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43ff24e66efb483c86df06b6a336b575 : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2fa25f2dfb88415cb4fcc2190f4a643b : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/827c1339489b451e8ecae5060e9ec725 : SUCCESS in 20m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c3dff25da9142efb72981331a1035de : SUCCESS in 21m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4122ebbc058645018ef610cee4d07ce1 : SUCCESS in 11m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/090ecea05a27487f80a765c6f520c446 : SUCCESS in 17m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6187b299171846c0ae05421c4c6cb05d : SUCCESS in 26m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25a2d7ab1a774128ba039c1dbcf52b94 : SUCCESS in 27m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/caa2c99248e64c568a6564cd9728c4e8 : SUCCESS in 21m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d4d51056c514b1799fd53cc2f4919d1 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b8935faedaf444518f23125eddc0976f : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0c8b0c28853b47a3bb26f10997b0a31a : SUCCESS in 18m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b73bffaf189648e18544cc50b3d9328a : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/595a6a0a4a3a41ffa6b0df5a8700690c : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af9a4debe54f44c5980fb3315641b5ba : SUCCESS in 12m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ae2cbe00217342c9a467c45350e71b47 : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0beefef12834bcdad1998908b585d39 : SUCCESS in 47m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6b4376308de849de876e5349076d487d : SUCCESS in 50m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8cb4a4ad89b64bc38a294cac7f9b47d7 : SUCCESS in 45m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77be4518148f4cbf8fac372cda6796ee : SUCCESS in 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82a23ef23f104356aff607daf2e532f4 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/387b016b1c2742899e0998c795974785 : SUCCESS in 18m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a169c42ace3e430cb9d8153a4908e5fc : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":4},{"id":"26a9ea3bf324190057654ad2bef84bcff6283c74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 16:53:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c2d2f9ca6225d769e1e96fd729b94cf5fefbaea9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 17:59:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c7b3649a40a4d34bd2be6f8201e14dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1f405a7de9458ebda93a9ef2d6e3cd : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/69884b45a1254613bf4570c3250478ad : SUCCESS in 40m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a2dec5135f77478d8e423d38b4d83676 : SUCCESS in 52m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5f85e0006c4432d98e6ebd8a9d9d0aa : SUCCESS in 23m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e63694fb02e849a1a499df8cc401f009 : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2160d51ebf8646249a5ef17900b28307 : SUCCESS in 21m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/383f179d11b74a409e2f4791de81d138 : SUCCESS in 10m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/69b2ccf7a6e842389d3045a7fd694db2 : SUCCESS in 21m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/faba1f6a606649558a633df5b8ad84e3 : SUCCESS in 10m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/34acf15d419247d5a9a1d4ee601cd9de : SUCCESS in 30m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d1cab6b4f7e4786b9ddc2b8bdfdc574 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54f41ad594814313b4f726a374011ae5 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e973d5d85ef04718a305addf3630856f : SUCCESS in 13m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0f6d713401f94e36bb979dab594463ff : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0e4a089096124775902eb454b23a8d58 : SUCCESS in 10m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4a218b7456a948a5b914e6da94c9c4f7 : SUCCESS in 20m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f5329a464b049dda0c0a19fb337f97e : SUCCESS in 5m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb2acba244784ea0b520f78abb6135f8 : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/26c95591f08a46e791e5cb17d9bbd892 : SUCCESS in 51m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ddae702dd84c441383136261f33a267c : SUCCESS in 52m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/16b2592abc8243cb8c68bd782a7aea0a : SUCCESS in 1h 00m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aa49a3a575a41b78f21bf7f2ef3ebb2 : SUCCESS in 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3047f5630d294fbf8f1507cef39e441e : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90d7649354fc4d5d9aed4bd2b3b7b501 : SUCCESS in 19m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a281f99a27e4b6e82e4196626bb7acb : SUCCESS in 52m 11s","accounts_in_message":[],"_revision_number":4},{"id":"305585b7f7453c97942bc72a6166494a355ca394","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 17:59:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4c01a532c639136e3f130ebb75da18dacf038786","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 18:00:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b1ad396d1a245739667901abf22a719\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed412a5e97bd4d8a951d08694854624c : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7f41d47dbf5441c48bf2e27d19f35f30 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e0531edccbc498507c99313ab29d02cf3229548a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 23:23:34.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11efbe63317046e8bced88a8c1717203\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2803202e00d742809084337d2479adca : SUCCESS in 1h 27m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7873c182af744b2ca7db5caec6875a8f : SUCCESS in 1h 32m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/32aa83ab85a043608cf3d8b3113f14a1 : FAILURE in 1h 33m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa0dad4ff97d4603af0d7dc89839f9ab : SUCCESS in 1h 42m 26s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"95b7e16c7f9db6efbf3abe9305289a10774ee4d6","revisions":{"85546ebc726a653a3c0b4883e2f22f4456b83e70":{"kind":"REWORK","_number":1,"created":"2024-09-17 17:24:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/929688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/929688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/929688/1"}}},"commit":{"parents":[{"commit":"e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-17 17:23:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-17 17:23:32.000000000","tz":-420},"subject":"CI: Include --domain in more openstack commands","message":"CI: Include --domain in more openstack commands\n\nChange-Id: Id5b0bf6efcb2c16e10334d2edb268194fa1ec008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85546ebc726a653a3c0b4883e2f22f4456b83e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85546ebc726a653a3c0b4883e2f22f4456b83e70"}]},"branch":"refs/heads/master"},"6a9bd1ba2e123ee7fea2ee76e91fbec2ef2be4fc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-17 17:30:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/929688/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/929688/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/929688/2"}}},"commit":{"parents":[{"commit":"e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-17 17:23:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-17 17:28:07.000000000","tz":-420},"subject":"CI: Include --domain in more openstack commands","message":"CI: Include --domain in more openstack commands\n\nRelated-Bug: #2080600\nChange-Id: Id5b0bf6efcb2c16e10334d2edb268194fa1ec008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a9bd1ba2e123ee7fea2ee76e91fbec2ef2be4fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a9bd1ba2e123ee7fea2ee76e91fbec2ef2be4fc"}]},"branch":"refs/heads/master"},"7ba884b0015012693a8f202030aa3e491be67bec":{"kind":"REWORK","_number":3,"created":"2024-09-17 18:13:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/929688/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/929688/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/929688/3"}}},"commit":{"parents":[{"commit":"e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-17 17:23:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-17 18:12:55.000000000","tz":-420},"subject":"CI: Include --domain in more openstack commands","message":"CI: Include --domain in more openstack commands\n\nRelated-Bug: #2080600\nChange-Id: Id5b0bf6efcb2c16e10334d2edb268194fa1ec008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ba884b0015012693a8f202030aa3e491be67bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ba884b0015012693a8f202030aa3e491be67bec"}]},"branch":"refs/heads/master"},"95b7e16c7f9db6efbf3abe9305289a10774ee4d6":{"kind":"REWORK","_number":4,"created":"2024-09-17 19:05:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/929688/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/929688/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/929688/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/929688/4"}}},"commit":{"parents":[{"commit":"e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-17 17:23:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-17 19:05:17.000000000","tz":-420},"subject":"CI: Include --domain in more openstack commands","message":"CI: Include --domain in more openstack commands\n\nRelated-Bug: #2080600\nChange-Id: Id5b0bf6efcb2c16e10334d2edb268194fa1ec008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95b7e16c7f9db6efbf3abe9305289a10774ee4d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95b7e16c7f9db6efbf3abe9305289a10774ee4d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
