)]}'
{"id":"openstack%2Fswift~930095","triplet_id":"openstack%2Fswift~master~I24ec702cd2ed074ca9df084cefc896418cece394","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-10-03 11:22:21.000000000","reason":"Change was submitted"},"37271":{"account":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"last_update":"2024-10-03 11:22:21.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-02 16:29:05.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-10-02 16:40:29.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I24ec702cd2ed074ca9df084cefc896418cece394","subject":"Make object-expirer respect internal_client_conf_path","status":"MERGED","created":"2024-09-20 22:36:10.000000000","updated":"2024-10-03 11:51:21.000000000","submitted":"2024-10-03 11:22:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":57,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930095","meta_rev_id":"dfb69d15cce94782531b3d17c25a31a0da1fc5dd","_number":930095,"virtual_id_number":930095,"owner":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 11:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2024-10-02 16:29:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2024-10-02 16:40:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-10-03 08:58:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2024-10-02 16:40:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-20 23:51:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-24 03:29:34.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-09-24 10:28:04.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-09-27 16:07:00.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-09-30 20:39:32.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"344ef2d7cd00a3febb95b9bed232d0c92d939d6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-20 22:36:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04c4a76300f497af57ea00dda12d30f17042a639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 23:51:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45028de6dcb84552abee0ca3f5044cb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f362ad87a3d744ef9d0cc9a708946a86 : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/0af0a4a4913948cfa6a0bce9746b2f6a : SUCCESS in 1h 09m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09be01b5b9894662b4bef1df4cf5ba4e : SUCCESS in 1h 06m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb49f6da5f134ec99e547226de6b9fcf : SUCCESS in 29m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac297587bbf0415c8bfe51be663eec59 : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a0206102bb14ae1b42c107356e8755d : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bc664710fa74bcb992dca9dcc3806b9 : SUCCESS in 19m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ebc02bb669604a5382d553ab16ae897f : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c9393bf40d2b4190a8346c7bb547fe3a : FAILURE in 14m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/47fb5411683c4f15a30fc83d2306168c : FAILURE in 10m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/26799c1a628b449aacc8545d0e5da12f : FAILURE in 15m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/54f14f3fe28d4998ab92975ecab7f179 : FAILURE in 10m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b0124a74a6e84b1eb28985463f6ded03 : FAILURE in 18m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/562dc5aedc834d4d81d92d63e62a4427 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1e574e8c11954557ab9d17fc496ee4de : SUCCESS in 10m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e915649e16bc4551abb9c75e64bcb57e : SUCCESS in 13m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/31d4232241d24d198819f264dee71651 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/501d73f2261542e79517115015ca8219 : SUCCESS in 14m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b11c064c477b495f85b335e79f537bc7 : SUCCESS in 8m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c077066e928d4da59eceeb605690a248 : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8fbe92576edf4831b5c599a17198409d : SUCCESS in 9m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ddc36ab263494a3babe7de5adb98f7b7 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/76a2ce64ee134f95a4ca7bcf28ea0d4b : SUCCESS in 36m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c507788425a4acfb2e5c9fba9d9a16e : SUCCESS in 44m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7489ffad19a4981b99c806f9d1a03f8 : SUCCESS in 53m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d33c765a2f14710abe7f37fba571421 : FAILURE in 20m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6b190703e9f4b6fa71348f088aabe7f : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd64e05e457c4dd7937898021dda6492 : SUCCESS in 18m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8f93d57be004063bdf66a8ad6127ba0 : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":1},{"id":"59ead504eb9c4f1e683daa55031949a4c999a65d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 00:38:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5181916955c14aad990411523d99ce50\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94de54ce704c4aca83523a63c4402d74 : FAILURE in 1h 09m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b762c5703b88419c86f3192d9eeaa386 : FAILURE in 1h 27m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2993b0dc926f46f08bde6a7fe0b602ca : TIMED_OUT in 2h 00m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/abb4e10eba594ef1a546a7586e068faa : SUCCESS in 1h 29m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8491965597e422e69cd7e27aff5dea0801e00eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-23 16:18:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0f48a2c406396ae52a090765d5be5b6b1c404523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 17:28:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5850ce3fdc9f407e92af6bfb54761131\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b72578f9f48842d7a1d1fdadd31bae4a : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/f8c7151c5eaf4ef1861a474503e0011a : SUCCESS in 1h 04m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db5fa20a7b3746e99e44de5d3acf5f44 : SUCCESS in 52m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1499a3f3b101473fb704ad8219539cce : SUCCESS in 24m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f9473a62cd74d1a88ea0f925a995cb4 : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0184ae0cc5284209af692a5886eb81ef : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89437bf02b3946ea93435862c235048f : SUCCESS in 17m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd1aed21124f4e24bcb61ca4618d1b64 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ecd75e82429c404a84277ec01a3796a4 : SUCCESS in 21m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95712aaf0d3645d388fa312545c0fe3e : SUCCESS in 18m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1cf4ccfb3e514b11a655d84cc8ff325c : SUCCESS in 11m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/eba9903e3891464da0df52f9a8d7e975 : SUCCESS in 21m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7a00a94302974958a0bbf7e10fdb51f5 : SUCCESS in 22m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9fa5409567d24fd0b44f50ee14853c82 : SUCCESS in 20m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e83b0da3db85443c99435e78aa9c58d5 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5af73af597ff447f8c95b8123679cdf2 : SUCCESS in 11m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8044348bed264fd196d53230cf208eb4 : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9283f34b50704824bdd7f1b7de7c0989 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/532abdf5f92747e4a7da97ff463faaa7 : SUCCESS in 19m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08f11f8a32014819b2abc695b40f6084 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e982480977b44a5b0247ec1ddb9ead3 : SUCCESS in 14m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4a80cdc0daa64b0d8b38bdebe78f5e8b : SUCCESS in 4m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a978a7d004b84486a5879855d32f7710 : SUCCESS in 50m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c2f9ece6566b4ae8ba9ac3ff8ec18fc5 : SUCCESS in 53m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/63b445fbd44a4f7b965e3611952ea0d7 : SUCCESS in 40m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b41e058583164ade9c04d6b01b40292b : SUCCESS in 26m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c845cb1f5c7b4f38821e18a8a6430757 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f6ed3cb8c564dac915f78774abd82bb : SUCCESS in 19m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30cbd46f8d8c46b8a49c7960639fb5b6 : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":2},{"id":"af88ed55b55893b249005b98cf1df594b24396b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-23 17:48:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ee41c4a1616f0d13e8c24f4716848f107d618322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-23 18:23:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"71f9ca222eb2887025a271343cef150ad745ae7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-23 18:26:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c6653d5a485cd2725778f7588179a98c1ed35e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 19:24:57.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c63e1a980224ba3b5a47a521a9dea48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2753f4d2b04a18b1bc2c1921d19f31 : SUCCESS in 6m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/60f97e6aeb9a4999a3667f4ecd2e54ee : SUCCESS in 41m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ef32ae4af284399a41cc56b53664a31 : SUCCESS in 42m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee51b54ec19f4bac897e90a06974db3d : SUCCESS in 23m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbd767a6679f416984eb55923f3e3066 : SUCCESS in 13m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af76d39e5f4144e8849eb4f5d6b78bd4 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58bad20657c5440f8bd4f4e651ebc5fd : SUCCESS in 13m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b01c550dc164c32862aa353a6ccb94b : SUCCESS in 7m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/329624b7e1f5455fb7b91eae60f2c015 : SUCCESS in 13m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c6141f1d3f1491f9c12004b9dfdad6d : SUCCESS in 19m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4e153a46a5b74a3fa183d91e75ecac39 : SUCCESS in 13m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fe6bc623a0c44dc4a8a19d672c1274d0 : SUCCESS in 19m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e12ac04d930249c2b1f7480bc6614b8c : SUCCESS in 29m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf56423d6c084bf3b21413d3d873efc5 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b7ff44a94814b22b35feb9fdcb077b3 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b787dd18cf2d41278df1355f96d7c12d : SUCCESS in 19m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8eddd59903d74a5d92a7ff3b4a130006 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ac1980879055484ebf7cf9f6357f74ba : SUCCESS in 18m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/74eac6c8b48740ceaae7919bfe29bcb4 : SUCCESS in 13m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/133626b487254e25a2619ed553a8bac8 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb7cd4c6e2b842828171a4927de95ae1 : SUCCESS in 15m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/33d176e1c9d746ce9b32cc44de56dc6c : SUCCESS in 6m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fc8a77dd2cf4790badd831574c2fae7 : SUCCESS in 53m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5691d15e6c4d4056a981f0e8fb3ed451 : SUCCESS in 37m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40f4ecdcb46e4f118368c6c7095b5580 : SUCCESS in 39m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94e08f443eb8459eb8448b4159fcd2b1 : SUCCESS in 22m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e522bbbbf0c4149887fefcc327a8463 : FAILURE in 3m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e2120cef7f484efe85e3076d4f0a1f0f : SUCCESS in 14m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a2699db36f543e1bb5e83d5b100afa8 : SUCCESS in 49m 31s","accounts_in_message":[],"_revision_number":5},{"id":"217770018988fda5c5bfc3a7f64d773dc7984694","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-23 20:26:11.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0afd9a93864127d01dc30ea01d892d9f0efd591b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 21:19:59.000000000","message":"Patch Set 6: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd5bb1f01efa4a759c1e6654b5dce997\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/376e560c07154ea0bbc85f5a7c37e032 : SUCCESS in 7m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/95662760aa39496499c28ebe110108c2 : SUCCESS in 42m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/122bdb4769474f4694cad2a658bd77e2 : SUCCESS in 38m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee72a07496ae44c28dbe4f1308c341c5 : SUCCESS in 18m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ea210bd493841fea0cb18b8d30acce8 : SUCCESS in 11m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37a003ca4458403a9da7018978651712 : SUCCESS in 16m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd963ca252ea4bff995df90c8e279e49 : SUCCESS in 18m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/906ae576253b49d0874ab4d1d99ba74f : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f59ea80cee4b4aec9318b391ca3ec58b : SUCCESS in 11m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f70db298cc914097bb904bb6766ac2f0 : SUCCESS in 10m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/78b472a0e83648309ef31a85e3c6fa60 : SUCCESS in 16m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/84d1db5e31964f87872ba7f264ceb3e7 : SUCCESS in 22m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f68d35d230ed481f987041a946cf2d9b : SUCCESS in 15m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6088b86e81d84e62902180015ba77aa5 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ef8ffe67b31478dbe6c0a719d5b3a4f : SUCCESS in 11m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/be3d452996764e578ec4de7429f059a4 : SUCCESS in 19m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d5db801506f84532a319c342ca959e67 : SUCCESS in 8m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b98d15e3bee543d593f1f37b0237f260 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ed10a3762b1e46f99b00e2d8b3b451f7 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/497f3afbd09d49dd9cff4dbdeb6d6584 : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7054d21ebfa841bc95d9968aa8213711 : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/522915004ba248c8ba164ae2b28cd465 : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67608a102abb4581a5cb53d9c6597391 : SUCCESS in 46m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04434c21e5594b3988f06f960e0af3a7 : SUCCESS in 38m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e341c59f95774e2d89658de8dd6a2d7a : SUCCESS in 36m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c79f5406f49b4dd89635d41e1a926587 : SUCCESS in 17m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7d1dadff7a74246b5beeb39c743958f : FAILURE in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0b8be0c4156444c906d7dae999c67bd : SUCCESS in 17m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa842132e82b444dae1096fb8d31c075 : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":6},{"id":"4499cdbeb0596faf53699372a3c2f1f4cb83d217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-23 21:28:29.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"637c689e97dc1c4675187167197ee7bbb1b3f121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 22:32:46.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b25e81194854b368ae79fdcdb10793a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b2d80e6d2b4510ab867df0d01c8fee : SUCCESS in 5m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/389903f8a2d044358a1f4809abc2e139 : SUCCESS in 39m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4e5af936e1134f76bc3fcdfcdea5c993 : SUCCESS in 37m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6df7e92381144e2c9360c5c09e7a00fc : SUCCESS in 17m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4592854cb24471bb434377f781bcd16 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2694d655b3ef4a03ba19997eea8f12c0 : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5880d9cb648d42fa98cbbddc0ccc5327 : SUCCESS in 14m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd509a617e4d479984d3bd95dced322a : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2de4725bf6d544daa0578434f00d4c16 : SUCCESS in 16m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3da04b162e114eb0a789482b54a4c518 : SUCCESS in 15m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5fcdf69b98dd40d489ef8ba761e1ffec : SUCCESS in 16m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0ce1f1f21fa94dc08e4a3fbb655e86ce : SUCCESS in 10m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ea70fa678ac64b8eb76e5f057663d4b1 : SUCCESS in 26m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b5cd9a69f854e19bd685938b75ae93c : SUCCESS in 10m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef8b38be5dfa499b99ec6fa9572d1cc8 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/462ff6e656d849b88223ff101e0c8941 : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4bf76b83374843d2aab1f60079a4c281 : SUCCESS in 8m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/61dabed105e04f7d859556da3ad50b4a : SUCCESS in 18m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/939a1520c7bf4bd0be0d28e3a2371590 : SUCCESS in 17m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3186686eec0048eeb1581ccfdcdd51d5 : SUCCESS in 4m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab76497f7900471d9b8543a6ccb7eba9 : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a76ab12420746aa9a072662d935d412 : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/89d804a842764bbe8f6e52f091b119f1 : SUCCESS in 36m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c975c6297e24a898606704ad394818e : SUCCESS in 49m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c5dea6b4e18b49338e971b7f4e9b2dd8 : SUCCESS in 52m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/947b03b801a04e2d806eaa05bd0c246e : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01532e0f40af41b7bc631b75ac4bb669 : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac5b4f64d12545d1a3e3aacf7152170d : SUCCESS in 17m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4da1ea79ea9c4e3f8cb3af872b4dff78 : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":7},{"id":"e8ca1c8b2a5b3c20cd3f3782473a93ae1671e8c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-23 22:41:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"0f06e7376f901acd885b32d60b1b552d0c1df6de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 23:52:39.000000000","message":"Patch Set 8: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e5526fd917f4d0b8d0391cb72184d29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82dcb417dbb941dabf177e0e12024497 : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/c480504326cf4c08b2bb3e639fa4368e : SUCCESS in 1h 05m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8cc3505811eb4d2d9c1b4cefe18f2725 : SUCCESS in 47m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9897bfad7f4344dcae71b6e7d30cd9c9 : SUCCESS in 31m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25ad4c24e3ea453184da59070d7c2e1e : SUCCESS in 12m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f9affc69cfd4b5caa901477c9dc9ce9 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46d71022cc4d4ed4a627d0a61cfbe252 : SUCCESS in 19m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25854a7a6ed546a19eca8b6f81359c2a : SUCCESS in 4m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9be7bbac7f7416eaad6b6a653d677b8 : SUCCESS in 20m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/feb4ffca47864993b13309f31f24d45c : SUCCESS in 13m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/69d735d8702245e78d428dffb3515ad4 : SUCCESS in 20m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d172334bdca34eafa65a6d291b9737da : SUCCESS in 14m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8c725af9eb7348529506a56d8891b20e : SUCCESS in 18m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc6bfc5cb9ab48128783c27e1b112cf6 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1791a6b0af7347aab23e4e967f80d2aa : SUCCESS in 13m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc4933b44ffa4f12ba092a57578e7ae3 : SUCCESS in 17m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f7bb9f09d10d4cbea061bfb430b2a719 : SUCCESS in 12m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/198ec04b338f4531990b1097a5e8c7f8 : SUCCESS in 14m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7ace82b3532f4a01a70e5d9c7d4c36f2 : SUCCESS in 12m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1e29963c56d34c93a8d5c8285425f487 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8fad1ffe29cb406aba0da6b1f66252f2 : SUCCESS in 13m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c9385ec287f493ba353af9f1826a143 : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d98de7c2bf94980a90d97d126dd9dd7 : SUCCESS in 53m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a8e27633fbe40a2af0a9cd1453e6dc3 : SUCCESS in 33m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/27899e4a73a84d5cb3e55ed19e74d9c5 : SUCCESS in 56m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d0bedd79b644e97b614ad102340a2f2 : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46e9e3a95e6e406191875684e3859c5b : FAILURE in 3m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6ad0a2b0b3094e83b7bd78c385ea1f74 : SUCCESS in 17m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9d24acb677240cbaed5aa0c2d6b7845 : SUCCESS in 34m 42s","accounts_in_message":[],"_revision_number":8},{"id":"699d8232dba07a981bff6d955e849d6df4da3907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-24 00:36:39.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"fa287e335096e90bf5a681ee3a8e597a237e5838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 01:39:01.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7487841df8784803aa8544d0d3fd9fff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18cc009489e54fb8b3da91c66c1dedec : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/cf15808b79d44b52aa440382a6110060 : SUCCESS in 52m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc4fd39c8cd44474a85c790db81efc47 : SUCCESS in 40m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e37bb6a042ca42ac94a053f3ff4a7611 : SUCCESS in 24m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0eb3b14cfc1f4507a49ee36ddb748d85 : SUCCESS in 12m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91ee7175866248429d2fa8f472d8826b : SUCCESS in 13m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a3b5d58bcc442b98bd27b0a81682b95 : SUCCESS in 13m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed56f4ac4748461882e25513c24685cf : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b32f64e93424317bd27a6c19885df14 : SUCCESS in 21m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/22dc28c6315b471ca38589130d7670ba : SUCCESS in 10m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9b636b557ed54bc79290f56441cbcbd2 : SUCCESS in 13m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/769ebe38371144ca96e126649b29c2ad : SUCCESS in 22m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ea9491cf0a2147e2bc07f6bf75b9d088 : SUCCESS in 15m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12786d8273a8410f83b08214755aee59 : SUCCESS in 12m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d65e12f585af4e5fb860c9b5278d31bc : SUCCESS in 17m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d60832e7400f46e8946e973e7f0719c5 : SUCCESS in 11m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ee8145b4cb944bf7a090d72becb7b5b0 : SUCCESS in 11m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1cc428df3f0c44ffaace6adde4f11777 : SUCCESS in 13m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e8be59ace45948be9a5be10eae27d1e8 : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e2265664a06f48bb9d2e7e14b3f91704 : SUCCESS in 5m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a0a886abd43411b90f7ced1d76e8dc7 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/628203ee51394fcda6648dd8bdc2ee8e : SUCCESS in 5m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/28215d5642194e9eb9b1b51c00ad22af : SUCCESS in 49m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d752220c25ef4b588f95b848edb3ca82 : SUCCESS in 56m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/467eccd73ab14f9cbf8e1ea9dc8312a5 : SUCCESS in 47m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b9453abbae24d76a9b68beeabb63e61 : SUCCESS in 24m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b33cf5c5e0944db7992a90643bd7dc5a : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91286fd6f6444a109ed7e2faf7ed55ad : SUCCESS in 17m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6f6dd114dfc4df9bf509878022d7730 : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":9},{"id":"717bb30f0ddd92241cff7a66ff8e3fd0655ea538","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-24 03:29:34.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2ac1582306e1364bcafd7203bae1d0143b7e0994","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 04:38:34.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26b15e82e3df40898aaabb5e4fa7c8df\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d3ac337c2d746e38e3639fe40fb5e5a : SUCCESS in 1h 12m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b803441940944a85b723701dd5676604 : SUCCESS in 1h 31m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/63be69ef33764eb0ada0fd2528170e2f : SUCCESS in 1h 33m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e8c2ede6189431f82144208809544f7 : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":9},{"id":"67f7b01d7659ff605ec64cf393b9f37314bd952b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-24 10:28:04.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5d23e2d4ed958504836ed50d526d9375fc56f304","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-24 10:49:34.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1f9c586b6012c46d8bf60f0ee6f4d8974548a778","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-26 20:50:09.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"8b165406e7fb752f5dbd7ed01e6775c08505eacf","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-26 20:52:44.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7a9cef0da1cd05e991e625d6a2dcf70217f71a33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-26 20:54:28.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"8d36ffb35339b7e294e1c3891eb0dff948e4ae36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:55:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8af01f9108f249fab1009e999f8c8f0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0deeb28b1f4a6db71f25a13d88d524 : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/a112ee7ce8714cc8acb0f5b5466cff52 : SUCCESS in 38m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/18d636634b9848d8a0b4a7a86e9179af : SUCCESS in 38m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7de6987c865b4f5588a88e61d33d366d : SUCCESS in 16m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d6052ce7f644687910de9fed6097f6e : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afc72d0c35924c1da72fc5bd82a4a98c : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0318a0eb439e409dbbb1fd3e3bb9eb1e : SUCCESS in 19m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f0c357dcdb842059a99ceae99c99b0f : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d277142b572c4fd59b6f045b68ffd6f8 : SUCCESS in 10m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cce77ce5f0e44801ade31c085fa9aef1 : SUCCESS in 19m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1562fdfc27304ee99c28779db2132661 : SUCCESS in 10m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5d4b537834634047b8861b546fe0d2e4 : SUCCESS in 22m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a03abada91f4456daf4b39228260eaa4 : SUCCESS in 22m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c75e38da369148dfa7c4b76269273eb2 : SUCCESS in 10m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a5bbac173ca4471fbb88e28ecab601d1 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c59eabfe7c5c4b41b90fb50071653f21 : SUCCESS in 15m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/da0dd53fd76543d8a6d7180edf2371ca : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/33560974d2624227af01d2a8090a4de6 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3240baacfc4411ea31ab4a60412b198 : SUCCESS in 14m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1eb97ef1dcc44c75aad1164ee3f11cec : SUCCESS in 7m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f931d28d17d4b1b846438d37c998d80 : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef66eff11a58413db22bc7d7ce23724d : SUCCESS in 7m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4078225b5dd24a0daf8889f0a1af82b9 : SUCCESS in 49m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a07ea3c5799467d8cdb4eb1dafae5a9 : SUCCESS in 58m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/583afc984d5b49feb533a2f1ecdcbc3b : SUCCESS in 32m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22e6a7865f4d442daa1bdc45b661e38a : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66791fcedaff44699195c76e12ab381e : SUCCESS in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7cb3ae4a5c5f429fb8b05c87a43b93c2 : SUCCESS in 15m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f706a6ec0284b40b88f484783ebd20e : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":11},{"id":"6c8b2711f1b3f5ae6ba35ee6d26b2f1d4ad04c68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-26 22:07:40.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"2ded9e23c06ea2961b28981e751cb20939e9d7af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 23:12:21.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a5296b104e140b68ed18ec0ff0eff88\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3eebfe7a95440949621731e32d2cafe : SUCCESS in 55m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4b3926d01284411836c1c0b88e5cfe7 : SUCCESS in 1h 03m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eccd7454ea7947e599435cc2a080cc11 : SUCCESS in 40m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d3a8bf32a5f4ff88c5201e903fdec56 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":12},{"id":"3cca85f02722cea071d74a9b914fb24d5b3d0e7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 00:14:53.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/362e2904db854fb09bdf13ce0255f5c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34aaa7ba01c54eadac9f7486ace36441 : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/6d3fb77190dd41f6a5b570c6cd3430ec : SUCCESS in 39m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ba9ee92a77645f9a89346e3e94f15c8 : SUCCESS in 36m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5460d79234541728aaeab4c2d5cb5b1 : SUCCESS in 17m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e7554eccf9a454a8e7aaed8074e87a2 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23d343c291d848859315a4aacae57800 : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e17cad49c1b4de5a2457207d445e6bb : SUCCESS in 17m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e8b79887dc64676baabfead1695552d : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47155782ecc9436795b19271303df02d : SUCCESS in 21m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a5f3bce04114e5591b259ea7cae25a1 : SUCCESS in 15m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8060c8f1dc1843e0ad9eb134755e1626 : SUCCESS in 21m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/571358ccb109464c9ec5dc4476b843c6 : SUCCESS in 10m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/45898235961c482caf7d8a737bf21458 : SUCCESS in 29m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a072a366be064e62ac8570221c3b06bd : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3ff2a7faad6e402aa26cc1f8694f764c : SUCCESS in 10m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5121a1864aa440f3a269f6232790d77a : SUCCESS in 17m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f5a0b29bf0b946859133529c6b16e130 : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5eba680404b64921b31da8005281fe93 : SUCCESS in 14m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c2e096e92fbb4afcab5c0ee544d8fdaf : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e04897de55d643f9af1fa8399fb27679 : SUCCESS in 4m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ade6d0d7d5e481c916f11b5bcb0e242 : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c964ac672cee4cdbb35e071a9b6fab2f : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d4a7a7f3c014abc8522bfad9b44a78d : TIMED_OUT in 2h 00m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2743036f44ce4ccebe476ec6306918fe : SUCCESS in 28m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9092e04de3604e6892150e98de071467 : SUCCESS in 49m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca8b09d12e2345a1b7a810fb84471e39 : SUCCESS in 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc97038a2c6448279004269efae1698c : SUCCESS in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d23ba5310ed4350be40a23bdce72101 : SUCCESS in 17m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c92c45a2ea844b1eab342ed060d987e8 : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":12},{"id":"e1f2ac6f4590c8e0a52c97894543206220d8e671","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-27 12:25:13.000000000","message":"Patch Set 12: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":12},{"id":"1b0a185ba4990485218a68bc17b82bec5e62f0ed","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-09-27 16:07:00.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"9e6a53c108d12b9df45efde01a92e9be5c411a6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-27 17:33:03.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"d0806f9c100e12a7700097092a8441e66a1c89f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-27 17:36:34.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"736364f25cc3a10e9b4dba74db88ed1964235a52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-27 17:41:11.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"2934cb88f074a6a75eb74637170be205f7b47677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-27 17:48:39.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"ade81ac11755896ee52818e913102a75ca22b5cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-27 17:53:16.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"dfdc230d190875b4c467e25312476f5d15100b47","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-09-27 18:02:15.000000000","message":"Patch Set 17:\n\n(11 comments)","accounts_in_message":[],"_revision_number":17},{"id":"2856b246459a2bbce0e585f74aa774ab4bc80c9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 19:10:01.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f227a503ba29411abccf9e63d18613db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858bdbc3f3b34d03b95cbe276fef10b1 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/822cf8c1c29440eba2bc2e1e4685fae0 : SUCCESS in 1h 09m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8514fede901d48529e4145802a48d815 : SUCCESS in 1h 06m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/35222eaa217c4054b8c01dbf2d95e853 : SUCCESS in 51m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12293d48c7c240e488cb6164e5d89e45 : SUCCESS in 27m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12a2929da03e4dc0929f60f13c953be6 : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4f22c0fcfea4ed782eb516073c68e88 : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d62bdefffc7f4b20a69793a2c0c447fa : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40550ce3da074fb6aec5145c2de3828b : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/190702595fbb492fa58f210bb0052dcd : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/729e2be1e6dd4a98aa6b27793ff47af1 : SUCCESS in 18m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a9969b8ebe714418ac136f1f46bf9879 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/266d8dfa69f643bc982665c442e79ce0 : SUCCESS in 10m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a74c686cb6ba4364be211d00294e2359 : SUCCESS in 26m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c4f0f64b9d254295a6c954c4bddae75f : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3e55090d35949c49a68ca64b0612769 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4e4fa6c6e61145e4a35f95e1c5e15c6d : SUCCESS in 16m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/38b07feb74f74ba0abd6a1c2c88bec49 : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6400150b5a2b461586f12244cd431c7d : SUCCESS in 13m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f273ec1e303547cfab4deb895fab8a5c : SUCCESS in 14m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2333b122b0fd458ab50bac23e7c8c717 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e8ca745b53534a4cbc6730d2f2b28c0f : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f4090b2ffb74bf3949a69dad29c9185 : SUCCESS in 6m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0611bbf9ff184ae4be9f30ab1b0cc490 : SUCCESS in 50m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/20472699b69f4b64a7c77bc3569eb452 : SUCCESS in 44m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/52c9a7c8e206469ea97fb5b60b28d165 : SUCCESS in 52m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a5219f850734b99986ee7cb1e8530f5 : SUCCESS in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4313bcb6eb6e452ea9fb2adf4fee444f : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3448b15bdeea4de799ebb522ddfe7214 : SUCCESS in 20m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/921ebc9b64bc4f3e918e0f3950915208 : SUCCESS in 43m 50s","accounts_in_message":[],"_revision_number":17},{"id":"7fbea031fb34848634d32db09d2c03ddf8fe7b69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 23:11:40.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d268c358c7e2417489180b76d2aea9e0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bd44e0f0e3a43059fed5b7248c71e39 : SUCCESS in 1h 44m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/548b28d579b24217b4ae91af6310f043 : SUCCESS in 1h 19m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c08eec1654c408983e1580548381b6b : SUCCESS in 1h 35m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dff5e20029b4f88b0b9cc1241026af6 : FAILURE in 1h 40m 57s","accounts_in_message":[],"_revision_number":17},{"id":"b2e45d6678459d4c2f43a6be94fe584f37cd7b55","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-09-30 20:39:32.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"dd003f54511dcb3731ca5048fada86fdb257f871","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-10-01 16:23:04.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"2346f9d623217e31cb7439c10b8cdab3da3a1b3f","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-10-01 16:26:13.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"4a64e783a9f56804900abec399f1ce4228a556f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 17:36:54.000000000","message":"Patch Set 18: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdf1627e92374da9926647028329aab5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62153f9103bc4597a278954374cd5317 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/84c20270dee743cdac4f0375f0cd79b3 : SUCCESS in 36m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b99328c664b047faabd4e24ecd798c8b : SUCCESS in 1h 12m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4c10f7b748c4e4bab1e8b4afad57a10 : SUCCESS in 1h 01m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a45a145d942647319b1d30f321493513 : SUCCESS in 20m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ef9f3fb74594e3f8916446f46c7b286 : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d68805c5375439da4babbbfdbd75ea0 : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c03dd52ed0e747298f4c395626402f84 : SUCCESS in 15m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd819eb17038419f9301ed43bf587545 : SUCCESS in 8m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/749893622bdc42feb7c5a94f9e329fb2 : SUCCESS in 24m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c589ddd310a4040b111bcb8b5bdab24 : SUCCESS in 22m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a08b195ca82143af91047d24e9f6b6a4 : SUCCESS in 20m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c405ad007975400c9edf9c4e5b3a54fa : SUCCESS in 23m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2ddc940bcc7b4781bbe5c53da7007a1f : SUCCESS in 27m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f23e4313b10e48bd89a940665214c74a : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3be164b7042b4e92b246b8352ab9acd9 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fbb485d2f40746138b4694c78a022bf0 : SUCCESS in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb1c185fca6e4cc48c923d1d7ea2f7be : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/648dce9897084f5d9af8729994edbd4a : SUCCESS in 21m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/02081753087d4d3b9e5e011bc517fc57 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/356f1a8a23cd44459db8d50a69b96bde : SUCCESS in 4m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed9bd322e45a43bcb18d2fa7f327637c : SUCCESS in 15m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d99e2deddb634f5dac025c1fa1db494e : SUCCESS in 7m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9df8b2739744cabbf9b322869e4d412 : SUCCESS in 52m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2f80a7e15284c85bd7a82f2e92cbf82 : SUCCESS in 28m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c959468d4b2c495aa7267020dd8935b5 : SUCCESS in 49m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f928380694e4278ac839b27b1b4ea52 : SUCCESS in 22m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b507fb4cca46cfb67fdaa37793715c : FAILURE in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/20550ab4106043e481a42eb5e7169ef5 : SUCCESS in 22m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85b539acd91d4c45881f334e5b766d0c : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":18},{"id":"18d20402535007e2c61c153ff6848a65afc52d5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-10-01 17:49:23.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"27d7c1d7d909a25a2101c7e92191727d0f869674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 19:17:52.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a5a75d0a7f44d3985cda456f005ca25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe89c8ee2424897ab95e83832cf3a75 : SUCCESS in 5m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/d6308189f0bc47e8a71eb4aa20ca6135 : SUCCESS in 1h 20m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1eb39398a6674d58b7377f8724faced1 : SUCCESS in 1h 26m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/689f0cf2aab4438d974f0829b579cca2 : SUCCESS in 40m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/451e9ca01ab649a1b2eb4d810e667211 : SUCCESS in 20m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73fa6cb9b4bd4411bdd1c4538b5ed15e : SUCCESS in 14m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49ccf41c983f40528476cc2caadf9246 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afe1cd85276f416098dd9d5de20eb8ae : SUCCESS in 16m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1fdd25b3276d4eb9b0336ed21fc7de36 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be889788734a445f911d029e82c761a3 : SUCCESS in 20m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/087cb4f3f7b24f2cb3c19f4f4aad483d : SUCCESS in 18m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8f126d1ec4cc45a58e38333d75a381b4 : SUCCESS in 21m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/906e17e509494075aa47557524f24058 : SUCCESS in 25m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/74dbc852b01043f0a9f5fda64245d05c : SUCCESS in 23m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4fb5fd71af54bf296eb744eededb2dd : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e208d9e32b5b449cad213986eea7f4d3 : SUCCESS in 19m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5219c4ab559b4a7daea8bb3b70430f33 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5950d27f17f14f03a6a0526ba63c6187 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/816a9788b5414565add83e31ad719139 : SUCCESS in 14m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3b92bdbb311749ea8cdef038c80a675a : SUCCESS in 18m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e1660faf20014d9e9cfdaacdacd00a3b : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ff1a5501fcd46e7b97fbabf4691aa5d : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6f94c14d14649de939502c16c5cfef9 : SUCCESS in 5m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d915a371c39f43ceb33a9da23da546fe : SUCCESS in 54m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab7fc3cb90a24ddaa5c16f43ecc8f5de : SUCCESS in 44m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0db5191472f43609c8013adea4aa6cc : SUCCESS in 47m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27447a88b6104283a30739b07af81c16 : SUCCESS in 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa19aab2e97406c9a43f2a361e0345f : SUCCESS in 3m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f8f29df95f8049d9922706c3cda641c4 : SUCCESS in 19m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c910a4d906a4748a21924c2bd9c843e : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":19},{"id":"431a0f877a29686e0d8d9e7cebcb8b5d01f8b8f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 20:10:20.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baad524aa1af4c6388bbb1c7835e037f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac2c89fc074b4505884d455a711967b0 : TIMED_OUT in 2h 01m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2776480be41444f29c2b7700b33bfc4c : SUCCESS in 1h 28m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/83884b44a041401ba4c4edc61ab55c11 : TIMED_OUT in 2h 00m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/508aa3d1df4941ac8a7385b31d5ecac2 : SUCCESS in 1h 37m 36s","accounts_in_message":[],"_revision_number":19},{"id":"2517485fba0ad75bd8aa5a1873304fac7a7ef4d0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 08:36:32.000000000","message":"Patch Set 19:\n\n(6 comments)","accounts_in_message":[],"_revision_number":19},{"id":"2a2257c04fdffb906c3152ff6fda72078b522503","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 08:38:01.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"629986bf8f3743eef8b83fbf9e2b24e977281ce7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-10-02 15:58:10.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"0d1a32c32613358739035d8f2a41212155ac4e72","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-10-02 16:02:02.000000000","message":"Patch Set 20:\n\n(5 comments)","accounts_in_message":[],"_revision_number":20},{"id":"a9764e51d0ae875fcde3d86613d2489ebd54532c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 16:29:05.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"b00005159ddbb96e0c9b731190e4832c5ebbf124","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-02 16:40:29.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"a553af74de52e95a189d155a1e641ceee90d2e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 17:09:58.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf9adc1f4ca6460a83273d8ebfe63e52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41bfb12eea394a0a810632937a854e83 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/352a3d1948f84344b19a3c565ac8a8dc : SUCCESS in 57m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/480ec28e8d1e47b78c95ef7df68339ed : SUCCESS in 39m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aa69c3221f7547af9100e5c23daedda5 : SUCCESS in 52m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/866703d4ce0b4a22a6820ede0e2b110f : FAILURE in 25m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64cbd2fd38eb4881bc1ee4e77a326d23 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/486507b5ae7f461c9cc22a72df4b10d0 : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb9561635d4f4f2e8aa668d139725f4f : SUCCESS in 11m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12aef8b5bce94aafa3cb68443ca1bbf3 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1395d794e5214f649851158b4b3de445 : SUCCESS in 10m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86f18ccb725340e4849bed60cca5b284 : SUCCESS in 18m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/293535b6ffc94cfc877ffeeccfda73ba : SUCCESS in 20m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b116cf5a676e44f7b84475d437e8925d : SUCCESS in 23m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/80484cc8edc84eee8af2003be4adff39 : SUCCESS in 26m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/043b150cd86f4f35bc257a5d53ba5e1f : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/150c60a6d6a849849ccddd0ded0e229c : SUCCESS in 20m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65e3de6f5a594476a2ee552e419f5bea : SUCCESS in 18m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d3bde2c6f2fa4f2b94e18aa0cc2f2755 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/339618b4405346f9894a94e4e13c986f : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e825dc6c12942f58e3aca0711eaf612 : SUCCESS in 14m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/daa9169710954e849b94a1a8ffb35b95 : SUCCESS in 8m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9021f76e973148b691b86756596d0423 : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0144ac39a7a94b3996dcf8a1bd4edc5c : SUCCESS in 5m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e60c78ae9e9430c97e9a95ac8405bfd : SUCCESS in 47m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f251402953f04fd9b3a57501bb04d0ff : SUCCESS in 46m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14b17c5e8f5c4cec9cccfe9035b3b9b5 : SUCCESS in 53m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfd72f6c646c453b97b79ee0afc317cd : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a09785923347ddbeee6cc895cf5fa8 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d7e3f3bd75148d49471c5483cd82744 : SUCCESS in 20m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/288256fd339a4eedbe65969506629b98 : SUCCESS in 48m 47s","accounts_in_message":[],"_revision_number":20},{"id":"3e735725d6c56be518fa4ffd8d7fe2cd7503f982","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 22:25:33.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20565c8949f74140acad6ae56ea0ded9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b4732556e934ba593740972183c1624 : SUCCESS in 59m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50d7cc572869428280a445b8d0bd5bac : SUCCESS in 52m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee18ef0c73ee4becb8e28b92f072bfeb : SUCCESS in 49m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d3aaf5cc72745a38e38c5bb24a70496 : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":20},{"id":"9b37800ab48f909b97a0c71dd037c191cd99a9db","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-03 08:58:26.000000000","message":"Patch Set 20: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"9e17532e399323597d55f4c377c5cd07560c4ced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 10:15:52.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb39276a6534f20bcff33020a33d475\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02fa9378c3649498420b19839378c75 : SUCCESS in 12m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/0c1c1f6821504d03b1c0dc032f741f25 : SUCCESS in 1h 03m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0bf1456c65840afb54001095ce9fe0f : SUCCESS in 40m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8056ab6a74f34f0d8018542b276efe29 : SUCCESS in 40m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/333b5d302e86431898847c79764481a5 : SUCCESS in 30m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99666d07e5204f7b96a70269d9b56dea : SUCCESS in 11m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/460bc93485c946debdf39f1d79faa4aa : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad30e800ebe8486b975965d0b9de6f7d : SUCCESS in 11m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a208cc3b2314e31bacbc3f8a0bbe404 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d4701f6a272f44ecb2f81ac443477a3e : SUCCESS in 20m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8065adf0f471440a9c9842cac4fece18 : SUCCESS in 22m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ab68867a383447b1b7e87568c24222c3 : SUCCESS in 21m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d0a59bfe30114a5a8688f2a517e94c4b : SUCCESS in 22m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4178a06feb7e4860be0a73be7ad8be98 : SUCCESS in 30m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a06552c1d3ef4860928855d1a02ae2e5 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/90f5f10a09b249d7a379f0150af7dad7 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/524bd2de2a9d4f58afb6d8a894d240cc : SUCCESS in 22m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/459df31beb144ea9bece23dcf2b54a86 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/85a78700dae748e89f4d4774c07a78b2 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/acae601383374ad0828839efb2f2ef3e : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/97b6a1ce127a43d89dbfbc10df1f34a2 : SUCCESS in 8m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/42056ca4ed0448f495c6c4e5f73e1136 : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aff463fca42f4b55be563681f4d465ae : SUCCESS in 9m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e60dea9d39674900bd75c31272b1e108 : SUCCESS in 56m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3e5ec20f97148559f94251704e929fe : SUCCESS in 51m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bdbc95596f78440ea337ef24b711474b : SUCCESS in 44m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e5128e7adeb48d9a9e19d6598271e19 : SUCCESS in 21m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd332b9b0da434fa6c0b6f39dc24f37 : SUCCESS in 7m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8b84053258a841629f72457996b57d0b : SUCCESS in 17m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f114cc5dbf14cfca9ecb5ad1b9a9c87 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":20},{"id":"1dfc87ab2411464bcab2b5d0ddab102986208e64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 10:16:25.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"d2b0e50eee2923d51b1f0de2df6b26f01081ab47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 11:22:21.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9733c4ff4fd3407ea71ade393a06623a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e1fb059ef4043b0ab28ea48004de9fd : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/4cc7a847fc2749a78e51ece64b1dee62 : SUCCESS in 58m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff91d72689e74467a1c93912124bed6b : SUCCESS in 1h 01m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0bcd7870b59c4f888e8fb0e9a955cd47 : SUCCESS in 1h 01m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/756fd59a6cb7436cb8ba0a0072386b2a : SUCCESS in 28m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0598571e794946a3bfe2083b1de21f70 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/53cd32ff1d6a4da09493a455d25a770e : SUCCESS in 19m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e26febbb9d6b4197afcd4d4a27402942 : SUCCESS in 18m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/60b1a8113f5b488f8444fb60646bb5c5 : SUCCESS in 20m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8d242a580a9e403582bc48a9909c1b26 : SUCCESS in 27m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a5cdc385d82749b8b1c2a4aedb16c764 : SUCCESS in 29m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/40fe3833ce6243c682e8c4ff29203d53 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/432063f55b79451bbce3ce932ea0b935 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c169c6055a8f4d8aae80e05585d33d7b : SUCCESS in 16m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dfdd5256e6b54bfa83fc8774174a6d34 : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/65361b3cdbfb403b8788f04250b2574f : SUCCESS in 11m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f299785900004f8ebab0635857b453bd : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/23c95379b3d24e8794ecabecb9c35ef4 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1365ab99569499eb42c13e9ad0a71c2 : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bd8531611f54c51bd79f0e99f59105b : SUCCESS in 52m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d03912805a16491bbdf8e12a328bf22f : SUCCESS in 38m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed2b6d0cfc764986b0d541fbb44a3ed4 : SUCCESS in 51m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08d1a20e74d74072b950921e8420cef9 : SUCCESS in 17m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38e66a0df8604188959b0b2237ef269d : SUCCESS in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e495eb5c62342fe9779da41f963a206 : SUCCESS in 17m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8625a92c0df04cf4938c48f4b03d951b : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":20},{"id":"3117b1e5b11f523d5c49c846af4ef51765773054","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 11:22:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"c05d134f9684e50f79e2bfb8091fac98c6422534","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 11:23:29.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fb986ae41c547de964a9cc7d9d6930d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51f01078eb814d91ba4aa0b0afe70f0a : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/deb9308b425f4bb9a30205b522f99b38 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"dfb69d15cce94782531b3d17c25a31a0da1fc5dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 11:51:21.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af981eeb55a3432b8d22cdf8a5c129b8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/776e002532b84586b4b21303ffcc03b0 : TIMED_OUT in 2h 00m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba6edc4f71bf4a5697c5c42085a24a6a : SUCCESS in 1h 55m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e60e58ebf5b4d188567a33b9748f5e8 : SUCCESS in 1h 30m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9875ca2c16c14eadac48334a952ccf47 : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"ee80f99aec8719fc3ad1d86bf8cd6bd48fc1bd33","revisions":{"c56a68d75d22fdf7b96dff1c6511ee1a7a4df341":{"kind":"REWORK","_number":1,"created":"2024-09-20 22:36:10.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/1"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c56a68d75d22fdf7b96dff1c6511ee1a7a4df341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c56a68d75d22fdf7b96dff1c6511ee1a7a4df341"}]},"branch":"refs/heads/master"},"89cf22439b98fed4c494da2cb9b70d150f40cbbe":{"kind":"REWORK","_number":2,"created":"2024-09-23 16:18:02.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/2"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-23 16:16:39.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89cf22439b98fed4c494da2cb9b70d150f40cbbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89cf22439b98fed4c494da2cb9b70d150f40cbbe"}]},"branch":"refs/heads/master"},"90063b748d80e45912a1ebda612c5aea7bae54fb":{"kind":"REWORK","_number":3,"created":"2024-09-23 17:48:16.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/3"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-23 17:48:03.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90063b748d80e45912a1ebda612c5aea7bae54fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90063b748d80e45912a1ebda612c5aea7bae54fb"}]},"branch":"refs/heads/master"},"af0f0c032195f5d3fe6daa4cb72a0c5a7fb1033f":{"kind":"REWORK","_number":4,"created":"2024-09-23 18:23:13.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/4"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-23 18:23:01.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0f0c032195f5d3fe6daa4cb72a0c5a7fb1033f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0f0c032195f5d3fe6daa4cb72a0c5a7fb1033f"}]},"branch":"refs/heads/master"},"d1cd00816e04ed21ab1545a576e56063499fed23":{"kind":"REWORK","_number":5,"created":"2024-09-23 18:26:03.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/5"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-23 18:25:50.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1cd00816e04ed21ab1545a576e56063499fed23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1cd00816e04ed21ab1545a576e56063499fed23"}]},"branch":"refs/heads/master"},"3c8b2135d4d77c4c154c69b187380d68624e7a78":{"kind":"REWORK","_number":6,"created":"2024-09-23 20:26:11.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/6"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-23 20:26:01.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c8b2135d4d77c4c154c69b187380d68624e7a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c8b2135d4d77c4c154c69b187380d68624e7a78"}]},"branch":"refs/heads/master"},"93eebc2afc0602a088727a65d9af40772ba94465":{"kind":"REWORK","_number":7,"created":"2024-09-23 21:28:29.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/7"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-23 21:26:56.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93eebc2afc0602a088727a65d9af40772ba94465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93eebc2afc0602a088727a65d9af40772ba94465"}]},"branch":"refs/heads/master"},"427a16deb600dc8b3069eca02e35175ee790684b":{"kind":"REWORK","_number":8,"created":"2024-09-23 22:41:34.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/8"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-23 22:41:20.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/427a16deb600dc8b3069eca02e35175ee790684b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/427a16deb600dc8b3069eca02e35175ee790684b"}]},"branch":"refs/heads/master"},"0e2391df63617c540d5c870ee00a6db070268bda":{"kind":"REWORK","_number":9,"created":"2024-09-24 00:36:39.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/9"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-24 00:36:17.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2391df63617c540d5c870ee00a6db070268bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2391df63617c540d5c870ee00a6db070268bda"}]},"branch":"refs/heads/master"},"73f9ca8ed98db7b39fbc3ee47fe4a85f273c981f":{"kind":"REWORK","_number":10,"created":"2024-09-26 20:50:09.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/10"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-26 20:49:57.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nWhen the object-expirer decides it was loaded from a \"legacy_conf\" it won\u0027t respect the internal_client_conf_path.\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73f9ca8ed98db7b39fbc3ee47fe4a85f273c981f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73f9ca8ed98db7b39fbc3ee47fe4a85f273c981f"}]},"branch":"refs/heads/master"},"a853902ced05b52080164db90aa4730bb42766d4":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2024-09-26 20:54:28.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/11"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-26 20:54:14.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a853902ced05b52080164db90aa4730bb42766d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a853902ced05b52080164db90aa4730bb42766d4"}]},"branch":"refs/heads/master"},"b4ad01b3c4f0376c15ca8f6fb88cb7d88ab0de01":{"kind":"REWORK","_number":12,"created":"2024-09-26 22:07:40.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/12"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-26 22:06:58.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4ad01b3c4f0376c15ca8f6fb88cb7d88ab0de01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4ad01b3c4f0376c15ca8f6fb88cb7d88ab0de01"}]},"branch":"refs/heads/master"},"8b94d2eb253f56ad52335d8a37fc76a34877539f":{"kind":"REWORK","_number":13,"created":"2024-09-27 17:33:03.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/13"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-27 17:22:22.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b94d2eb253f56ad52335d8a37fc76a34877539f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b94d2eb253f56ad52335d8a37fc76a34877539f"}]},"branch":"refs/heads/master"},"b7938b1fdf26149781e1f5d74fd8ac59fb58adea":{"kind":"REWORK","_number":14,"created":"2024-09-27 17:36:34.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/14"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-27 17:36:28.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7938b1fdf26149781e1f5d74fd8ac59fb58adea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7938b1fdf26149781e1f5d74fd8ac59fb58adea"}]},"branch":"refs/heads/master"},"dab07a1d376ce7bfd9a0fbaaa4f3786dfdc9bc12":{"kind":"REWORK","_number":15,"created":"2024-09-27 17:41:11.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/15"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-27 17:39:12.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dab07a1d376ce7bfd9a0fbaaa4f3786dfdc9bc12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dab07a1d376ce7bfd9a0fbaaa4f3786dfdc9bc12"}]},"branch":"refs/heads/master"},"14d411fbde9c4385bdb41d25a896671a85ec800d":{"kind":"REWORK","_number":16,"created":"2024-09-27 17:48:39.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/16"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-27 17:48:25.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14d411fbde9c4385bdb41d25a896671a85ec800d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14d411fbde9c4385bdb41d25a896671a85ec800d"}]},"branch":"refs/heads/master"},"318f455aedc9d0768c1d44d9b1e6564d8ffc4b8e":{"kind":"REWORK","_number":17,"created":"2024-09-27 17:53:16.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/17"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-27 17:53:04.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/318f455aedc9d0768c1d44d9b1e6564d8ffc4b8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/318f455aedc9d0768c1d44d9b1e6564d8ffc4b8e"}]},"branch":"refs/heads/master"},"dd518d9d8aa880f3da9e0de9fb54ba03701fc64b":{"kind":"REWORK","_number":18,"created":"2024-10-01 16:23:04.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/18"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-10-01 16:17:19.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd518d9d8aa880f3da9e0de9fb54ba03701fc64b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd518d9d8aa880f3da9e0de9fb54ba03701fc64b"}]},"branch":"refs/heads/master"},"29dfe2b875e4b63e03d63e782d0962b12d2f30cd":{"kind":"REWORK","_number":19,"created":"2024-10-01 17:49:23.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/19"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-10-01 17:49:11.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29dfe2b875e4b63e03d63e782d0962b12d2f30cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29dfe2b875e4b63e03d63e782d0962b12d2f30cd"}]},"branch":"refs/heads/master"},"ee80f99aec8719fc3ad1d86bf8cd6bd48fc1bd33":{"kind":"REWORK","_number":20,"created":"2024-10-02 15:58:10.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/95/930095/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/930095/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/930095/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/930095/20"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-09-20 22:28:04.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-10-02 15:57:33.000000000","tz":-420},"subject":"Make object-expirer respect internal_client_conf_path","message":"Make object-expirer respect internal_client_conf_path\n\nPreviously the object-expirer would not respect an internal_client_conf_path option if its config was loaded from a \"legacy\" config filepath (e.g. **/object-expirer.conf). Legacy object-expirer config expected config sections for an internal-client to be included in the same file. However, \"modern\" object-expirer config files (e.g. **/object-server.conf) expect internal-client config to be loaded from a separate file specified by internal_client_conf_path and defaulting to \u0027internal_client_conf_path\u0027.\n\n\"Legacy\" expirer config is still used in production clusters but it is useful to use a shared internal-client config. This patch therefore makes the internal_client_conf_path option always be respected regardless of the path of the object-expirer conf file. If \u0027internal_client_conf_path\u0027 is not specified, \"modern\" config will continue to use the default \u0027/etc/swift/internal-client.conf\u0027, but \"legacy\" config will default to the path of the expirer conf file (i.e. the same as the previous behavior).\n\nRelated-Change: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\nChange-Id: I24ec702cd2ed074ca9df084cefc896418cece394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee80f99aec8719fc3ad1d86bf8cd6bd48fc1bd33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee80f99aec8719fc3ad1d86bf8cd6bd48fc1bd33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
