)]}'
{"id":"openstack%2Fswift~930480","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ic4a617869abfbfbdcde1fb48bdac7f2067a594a2","project":"openstack/swift","branch":"feature/mpu","topic":"merge-from-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-26 21:30:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic4a617869abfbfbdcde1fb48bdac7f2067a594a2","subject":"merge master to feature/mpu","status":"MERGED","created":"2024-09-25 14:44:38.000000000","updated":"2024-09-26 22:40:41.000000000","submitted":"2024-09-26 21:30:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"930480-merge-from-master","meta_rev_id":"525c57828df1276d47030303c38e912e9b6be682","_number":930480,"virtual_id_number":930480,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-26 21:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-09-26 18:41:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-26 20:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-26 20:45:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-25 15:56:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 20:45:47.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"b0ed48915632c599aab10743d134674889af6ead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-25 14:44:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c36169efd66cddf7942d51c0355d3a9d9f38b2a9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-25 14:45:24.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d741b8386b96a0efdfc15e6c8920295633da6834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 15:56:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5552f28477d4b1788cee711bf05e666\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f8089e6be7154f6297c8143b21fa5f9b : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ca5ff2bf394407b85638b52c81b8a9 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b9c7dc078994d99941ebe1a0d4aa5eb : SUCCESS in 6m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/622d8242d5d94fab9e296a12fff3304e : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a745057c16c74afd8f568b6a006f4560 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42f0566f4e234e3eb48b9b4f60daa260 : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e57be783a9e04ddfa480599ffcbcc181 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c45bfd804c5142d3b9a0285b4a6bebb6 : SUCCESS in 20m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d4dfc39a0f442f9b15bc4c107f10205 : SUCCESS in 19m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5033c5e3dbe843f0bfd025a5ac07521d : SUCCESS in 21m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bcaa212c1f1d4a268b14b71194538856 : SUCCESS in 22m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0c498e4612a04e9c9950a8deda915055 : SUCCESS in 14m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2baf7b48a80a49f7a300a4afc1862b86 : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1188aa8df1034d49a75fa81c7d48812f : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc8611e25e1c4c598b6c23c41040280e : SUCCESS in 20m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f7575a2da2e148cfaacb949ab8cdceee : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/54bcbb0bbd6449fa89e530a353232c65 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/155612ee3b274c16836373a59e45851a : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a85b04a787444a74be860040e3156e77 : SUCCESS in 7m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/85a0d15ee5954de6bf676575a348de76 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d25b29943954dc28bcbc736ab82343c : SUCCESS in 6m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3f9498012e04d159f026b6d2001bc00 : SUCCESS in 52m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f0e265a20b441f5b7422ff58669d1b8 : SUCCESS in 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec63ed51c26d46a69244bee9aff668c6 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12fb53146fae45908e7b67471cd54855 : SUCCESS in 22m 32s","accounts_in_message":[],"_revision_number":1},{"id":"21cdcd81637285c2aa554c042913f5074d03ac05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 16:47:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221a269e26014658af1c288d538ea24f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/930725b2db334fb3bc8b3330396c25df : SUCCESS in 59m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c69dbf53be642e19c432a1ea2ead72a : SUCCESS in 1h 19m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/425c9fc418e640128354dc36c9b945df : SUCCESS in 1h 17m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b30b6af8ce3d465e888e4956909a55b2 : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":1},{"id":"845876e945e78c654f0adddb6ef894629ea7bae8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-26 18:39:20.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f38b665cf02603a25270e60bc669ddd16bd2a7fa","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-26 18:41:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4985e8b27cfaf61d7e305f4c4d485d3b4f6936d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 19:36:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5a94c2e32d3469da87f11e28458a373\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/210b7e2ecb3b4b68a0bd6753a5311aba : SUCCESS in 1m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88006fa1384a4575904ad36a32290ae4 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22bb579df606455a9d4eae2ff62b574e : SUCCESS in 6m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f3a6a09f19641af8871e303b812dc03 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19ea1d978ef14a2ea878bcdb1dfec8dc : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d00530b139a14f27a2675ba29845c9b0 : SUCCESS in 20m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dc94e7ba83cb4b41ade2f03f27991c8c : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba1935813a3a4e299801003bd2322fb8 : SUCCESS in 17m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad9c4d62e253480fb7a1e9c1420607d4 : SUCCESS in 15m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eea4f2c5b4f74a4eaeff4a67590ebff3 : SUCCESS in 22m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f60403efb2444d4593d301729f683e1c : SUCCESS in 25m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/04252fc05588432aa48282438aaaa4b7 : SUCCESS in 29m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c42278a59b814ee8b5ff7f58fc8356ae : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/829273b0a6c048f0bad4db17d2006bba : SUCCESS in 19m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/696b4332d5c546649ed3809cf99e93bb : SUCCESS in 17m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/60ae623e9c594ca799fe97879a8cc1a7 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/516592d601a448c091ab478dd109da07 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4b03389a159146f3bfa39b2475fb6157 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5fb24c35359341c5b566bb661f84ded9 : SUCCESS in 8m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/279968ee3f7041bd99dc7a2aeccefbd6 : SUCCESS in 13m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce60530464a2458eb5e132bcd5f2d6ef : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/230722c1be7344a985b1e10be01abb6a : SUCCESS in 52m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ed5209e2c2f42e49527f68644b1c20a : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f94626d1158040d9a84c86dcf93dc9d7 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2148106385ef493086e81a3fbf6b004c : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1cc0f6a30dc5541c9533304bf87b5234c4c2034e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-26 20:45:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"700bc5c196a9f6fb7ea2bd1405786333f1f5356f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 20:46:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14dbd47dacf17a5eedd55a9f80144781147a3fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:30:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2932a0c30674aa48bada8e753aed186\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a98f16381e1c4d71b70cab6b8e5a6f7c : SUCCESS in 1m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76887847c430443e9c41f0b311c74089 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff90970d7c0e42dab048d1043e303e8f : SUCCESS in 6m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e6bd7ca9188548d78da49bcd9fdea1bd : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28d4cf06d08643a08803d3680575ece4 : SUCCESS in 20m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b824797a10dc4bef8d9e7d638c749e48 : SUCCESS in 15m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/36e7082eec3b4d3b8183b2f0d0209760 : SUCCESS in 21m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/408eebad1d1f48db8494b80ac6190cb1 : SUCCESS in 16m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a179ba0bbf804e70a76d8cd8e4fa593d : SUCCESS in 34m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d64e761bb27c4593b98ca84d96081515 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1a0b3868f7494e109480583ecb3dd561 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ebad0be1ebe40a0b16d98e23c24cb9f : SUCCESS in 11m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/805d66abfde348b4b90ca777e7ad2baf : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/830e12bc86274dd6b025d0c141a179e8 : SUCCESS in 9m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f4e99d35a35c4926b26598a6bc4f4fef : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f7a1d6958610468888f9ac705a3393fb : SUCCESS in 5m 10s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/98752d40a99e419ca2487a7d191639fa : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1faa36f3da194d17a39faa6f885e6de3 : SUCCESS in 37m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81821f6d7e0e42869251f8fc8e583016 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df9abd4606fb4fbda7afb5a28b7e7f53 : SUCCESS in 3m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/004d730eef9c4ea39428d19f917ca91b : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":2},{"id":"9c2ad9f541f90f158a3f4a2b859865048c56d3e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:30:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"944ce0c73e6afc9de3e80fb1eb108ca9bb0ee017","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:32:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f599139d514677a95abd943b64bd44\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2c3d016a254e4ea27186243fac17ca : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9058a994d111463b9c5ed68cb7a0476a : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1f32f8a420914fb29b5ba29be3141cf1 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"525c57828df1276d47030303c38e912e9b6be682","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 22:40:41.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07d63834fb4b4bbd84a453cc4cc3c2f0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c800693f8c8245ca8d425de671240ba0 : FAILURE in 57m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e22bc041137493ab4e6f0e8ee159c64 : SUCCESS in 57m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bf522b1a8c1438c9bfd01b8336266a0 : SUCCESS in 53m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4a9163487734b638984d9d9f2d018ff : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce13529df8981e27fef363a29b34f7b052c6bb47","revisions":{"738e682735569597dcc907a768461b342b0ff00a":{"kind":"REWORK","_number":1,"created":"2024-09-25 14:44:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/80/930480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/930480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/930480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/930480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/930480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/930480/1"}}},"commit":{"parents":[{"commit":"31b3823148bbda17d97a34b1fd7dd5cb389a91bb","subject":"Merge branch \u0027master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31b3823148bbda17d97a34b1fd7dd5cb389a91bb"}]},{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-25 14:43:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-25 14:44:00.000000000","tz":60},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: Ic4a617869abfbfbdcde1fb48bdac7f2067a594a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/738e682735569597dcc907a768461b342b0ff00a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/738e682735569597dcc907a768461b342b0ff00a"}]},"branch":"refs/heads/feature/mpu"},"ce13529df8981e27fef363a29b34f7b052c6bb47":{"kind":"REWORK","_number":2,"created":"2024-09-26 18:39:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/80/930480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/930480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/930480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/930480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/930480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/930480/2"}}},"commit":{"parents":[{"commit":"31b3823148bbda17d97a34b1fd7dd5cb389a91bb","subject":"Merge branch \u0027master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31b3823148bbda17d97a34b1fd7dd5cb389a91bb"}]},{"commit":"6e4ecfc5dcf71d415a425e067a17d8a5723cced2","subject":"proxy: fix is_useful_response for py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e4ecfc5dcf71d415a425e067a17d8a5723cced2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-26 18:38:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-26 18:38:21.000000000","tz":60},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: Ic4a617869abfbfbdcde1fb48bdac7f2067a594a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce13529df8981e27fef363a29b34f7b052c6bb47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce13529df8981e27fef363a29b34f7b052c6bb47"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
