)]}'
{"id":"openstack%2Fswift~930530","triplet_id":"openstack%2Fswift~stable%2F2024.2~Id5b0bf6efcb2c16e10334d2edb268194fa1ec008","project":"openstack/swift","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-09-26 17:16:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5b0bf6efcb2c16e10334d2edb268194fa1ec008","subject":"CI: Include --domain in more openstack commands","status":"MERGED","created":"2024-09-26 04:39:47.000000000","updated":"2024-09-26 17:17:49.000000000","submitted":"2024-09-26 17:16:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930530","cherry_pick_of_change":929688,"cherry_pick_of_patch_set":4,"meta_rev_id":"613cb77f7f11084738abdda204437ceec0019c7a","_number":930530,"virtual_id_number":930530,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-26 17:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-26 15:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-26 15:55:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 04:39:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-09-26 05:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 15:55:43.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"ecf492fe0cd83d0ba6ab260c12fd73d2c317e5b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-26 04:39:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0721fd10f1a7d9c4308251537aff9e248fb11023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 05:43:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84920dbdc3f6482b912ae38de7a0c893\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b726ed12812349a99b3fe3724f9a5ac6 : SUCCESS in 5m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/47617a4895664d54bcf42c28d146c9c4 : POST_FAILURE in 38m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b6591341f374972b7c1f4597218a739 : SUCCESS in 54m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3676a7e44f0f4e8688580ef6b481a56c : SUCCESS in 30m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e84c312ef6740b49a6b1723fb52c943 : SUCCESS in 21m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e65ad49366144b5b23bbf1e2c66a751 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d75aaf5bfa24182a314e4dd99e0aaf8 : SUCCESS in 22m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6ce91328a234690aefcf7a2f0d286d0 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/826fdd02fcfb487791b0a75765f68860 : SUCCESS in 16m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dee05cdb10f04c328c136dbcce67ffdc : SUCCESS in 15m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c4abf3b80d25450799c42acb065b3137 : SUCCESS in 16m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a716e604860b421dbcc16784994e2bd5 : SUCCESS in 16m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7240dec67b8b481390fee72ad81d62f4 : SUCCESS in 19m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8247faf9caa04b4cab3ca7ddb1a3ed0a : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5396a7ddc865436fb7129d8e13e2e8e4 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d1aa113dd7c24ae09f242f9eb91dd02e : SUCCESS in 18m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/896ee9bcfabb4a1fb25461c4210e2fc6 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/625805c09e684356a0cb9f508c6c8d23 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e7a342a1144a4032a0af87628caf7657 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b2e0273d08374b2d9f7ba3701e7c9755 : SUCCESS in 5m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/280322f0cfeb42e4866019ee79523afb : SUCCESS in 13m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f52c61f6fc2b4f25baaf22a1d1b23dec : SUCCESS in 4m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ac42399b79943d8a2f86e1e6e31017a : SUCCESS in 48m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f44363b4a9ca47a9876755228192a689 : SUCCESS in 28m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89f3c45847074843aaed2244e98e782e : SUCCESS in 51m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a90d599159674bda8c0791bee9fdb137 : SUCCESS in 19m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/215c8161f1ab41e9ba0ed01ea688f641 : SUCCESS in 3m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/881fd52755ce4104b1b43fd9d2cbc372 : SUCCESS in 17m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa0214a75064408d97f134260ae1d8ab : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8a47396f05bc505aa83272cd933e1f06d1a3f3f6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-26 05:59:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27a9f5e3a33b3f177e34a290b27a03fb3bcc71bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 06:30:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c0d8c4267fd4c9c8ed994d540389b47\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8d524d0b88b45edaa3da3022e597ae0 : TIMED_OUT in 1h 31m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d671cf6ec7a0404fbca96bef4b6ac65a : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6df0630e068c41d9a7bbc5db7ad07b3e : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d4b6fc811a44594ac7c0b7d4b80e599 : TIMED_OUT in 1h 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2c420d4d97fdf07643bc9f8dd34a5186aed76893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 07:02:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11895e20cca94e96b53c4f96c28d5479\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7039b1656cc244d79b5f81e1a7949357 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/9871682f2b524512b250461c4c48ada8 : SUCCESS in 1h 02m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94c94f5ebf3b406ea39574c26972cad8 : SUCCESS in 39m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9858b93d828147d8b8a58a5f6168a4a7 : SUCCESS in 32m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21ebdc2b7a6040bd8d73b7d11cbdecd4 : SUCCESS in 12m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24ea8c5e85aa4437bca5f63b9cd2cc79 : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfe38eda4c1b49c19d1c22acb63bbcbd : SUCCESS in 12m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/202aa41a4156455e9558c83fb49945ee : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/35c1ba41b9c44b56be82d387f348acd3 : SUCCESS in 16m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b48effa2b2f4d90adbc444222faaa14 : SUCCESS in 19m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/be4006cb2f1347c889c763f859144301 : SUCCESS in 11m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/53225b93fb5649c6bb1b4561cb730aad : SUCCESS in 15m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c9a38e96b30c468d8ae9dfc1d760620f : SUCCESS in 34m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/53dc8a80b8f04468b8bd65ff5c04ac5d : SUCCESS in 12m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44e807f7306141b48b361df0d144b0cc : SUCCESS in 17m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8276fb73be0e46ccbca2e6eb174374de : SUCCESS in 18m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2e47ada3ad684c0b8cd38772bff204f9 : SUCCESS in 11m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c0ec178727654466981b1a07ff0caa50 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/187f863e5b0d4905b324fd0887efe8bd : SUCCESS in 20m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3122874e523745cf96921c0b4205a242 : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e21199077abc4ad0967297dd71cb8279 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1cbb5984576f432e97e9277e8f1a979c : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7ca38e2d9224129ab5e6aeabfca2e41 : SUCCESS in 49m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ad6fdd81ca94b92bbc75fc6e92496ba : SUCCESS in 47m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/847d8b0d20e545809c4e72fa840054d5 : SUCCESS in 48m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df9e6fd60d7a44629a5747171a06d808 : SUCCESS in 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6988924d68a4e0ea2fef8eba6c113f4 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ad9a5e5974a4be2b3e226023cc04292 : SUCCESS in 18m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8997bfb2b8645afa460431776fb6267 : SUCCESS in 36m 48s","accounts_in_message":[],"_revision_number":1},{"id":"30c3dd67fc14e4048f44ef29d4ebb67ec5067306","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-26 15:55:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccf49eab7e0715b60e5be614ba7cdc1918e987c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 15:56:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"145bf5dd639ea7ba9b1faadd4a62d55901fa60a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 17:16:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65316d8e27684e5f9fbf874a93fe9bd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fdbdb060b954620bca822452af9fc22 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/127f2e1939fa4033840b88be349c97d1 : SUCCESS in 1h 08m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d8f0fbbb1d54ccb84d21c46c86de7d1 : SUCCESS in 40m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ba145dd3175476c97ea3f106f101160 : SUCCESS in 21m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/65729fb3b53a4b748124190e5423578c : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c543dc3a2cc6434490a1ced85c2ddaf4 : SUCCESS in 21m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28b50852294c447ca5a26bf7dbce602d : SUCCESS in 19m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3b292f6e632148e39f79cc68cfb7338a : SUCCESS in 18m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6287070cd68548429a6bd7ba8f8b5bc5 : SUCCESS in 10m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6adab469eeca46e3b1a8cedf0f2c7598 : SUCCESS in 29m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c862613730f45b99983171d6f2e2e7f : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8da47cb0c0544463a9a21493ac4e3e2c : SUCCESS in 23m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b8f7c5dccd3c4162b36d0c45993a5d44 : SUCCESS in 18m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4a51e776c1fb4f9d9ab0eb1df665feb1 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/818dce81e33941b393ac343a02a9897b : SUCCESS in 22m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38488fed74a7419d9c2750b56f37e36e : SUCCESS in 17m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fe4420731e6425b8ac7e0cebb46191f : SUCCESS in 5m 59s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5010b7f801d94a1fbf66e040fa8de097 : SUCCESS in 7m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/736b349d647e482ba3644a029f1ba6e8 : SUCCESS in 45m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b24eb77df70d48f3b4e81fed9f66906f : SUCCESS in 44m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d557910884a04beea752620d2b0be47c : SUCCESS in 52m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c865ca5e172a488593dfdfd07f143617 : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/462baeab05124c91b621dc785b98f6ea : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ddfb742d9489472caa5fc9aaa9307dfd : SUCCESS in 22m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7f7a73ad0734a25bc28384860ecbc44 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0a8f6762362516e7e18014915e4c87625c94bcfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 17:16:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"613cb77f7f11084738abdda204437ceec0019c7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 17:17:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d87634a91c3444768144ee6c32cc8260\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96d38a55466427489a22e6504ff1b90 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fff177dcd30d454abf2f821bfe0be2fc : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27503ac2df5cb77b5fdf29196e555178ddbfdef7","revisions":{"27503ac2df5cb77b5fdf29196e555178ddbfdef7":{"kind":"REWORK","_number":1,"created":"2024-09-26 04:39:47.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/30/930530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/930530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/930530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/930530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/930530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/930530/1"}}},"commit":{"parents":[{"commit":"45e746f5184562ba0a0b981b09089b88d7f776e1","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45e746f5184562ba0a0b981b09089b88d7f776e1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-17 17:23:32.000000000","tz":-420},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2024-09-26 04:39:47.000000000","tz":0},"subject":"CI: Include --domain in more openstack commands","message":"CI: Include --domain in more openstack commands\n\nRelated-Bug: #2080600\nChange-Id: Id5b0bf6efcb2c16e10334d2edb268194fa1ec008\n(cherry picked from commit 95b7e16c7f9db6efbf3abe9305289a10774ee4d6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27503ac2df5cb77b5fdf29196e555178ddbfdef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27503ac2df5cb77b5fdf29196e555178ddbfdef7"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
