)]}'
{"id":"openstack%2Fswift~930569","triplet_id":"openstack%2Fswift~master~I0cccb960b51446a3aa58be95c8c9d06cc6b6eada","project":"openstack/swift","branch":"master","topic":"p-fix-in-process-func-test-etag-quoter","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-26 19:33:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0cccb960b51446a3aa58be95c8c9d06cc6b6eada","subject":"trivial: remove print from functional tests","status":"MERGED","created":"2024-09-26 12:32:33.000000000","updated":"2024-09-26 19:35:05.000000000","submitted":"2024-09-26 19:33:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930569-p-fix-in-process-func-test-etag-quoter","meta_rev_id":"87aca9da23bcf50e442eb8ea0d2a96f4a51e775a","_number":930569,"virtual_id_number":930569,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-26 19:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-26 18:30:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-26 18:30:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 13:17:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 18:30:43.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"1408bbc1c24cca0389755f1904e21dea169c8055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-26 12:32:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"279d7a8713c35d0b515afb8f3d67f3a1bebd5258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 13:17:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a615f65781c4fa2bae55e45a3871a31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6158369565ed4100920f80d901cdf6ec : SUCCESS in 9m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/747e340b603141a5bc01aeedd3e67243 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80d55f2fd633408cba6da4d1ffc864c4 : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66feff46b0ad45f59f61ad0e8f365161 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/95abd8eb22484df99057b13672755837 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c87baf5289d437a9d2398eba6a722ec : SUCCESS in 13m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6083891d5acd4cd8978e6b7d4b71848e : SUCCESS in 18m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/567c1fa7dd844e03b377ddb03bb0ec21 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/949bfd602ead4cf288dc1b3462d4ab92 : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cbbfe115686c4bc78289c1ab917a5ddf : SUCCESS in 16m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ef8b423cc9e4f11a6081a8ed7fa0890 : SUCCESS in 38m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4b13bf4f583149fea448cf5bcda0318e : SUCCESS in 44m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/121321ee50664cc3a53f8f109963a73a : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e80bae7fb2054bcbb1167fc923462530 : SUCCESS in 17m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ed2721b28e7917e3831272284ccca2b5ab13e4c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 14:55:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/100bfccf40a943bab946d6b7a9b58536\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e893ba82530441ed95679123cc18ecb8 : SUCCESS in 1h 22m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da4f993637744102a85ef58c9997429a : SUCCESS in 1h 32m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/018dfe131e5b4e4f812931a9a2ff5531 : SUCCESS in 1h 09m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cccb08d2504d457cb8787d03be6cb6b8 : TIMED_OUT in 2h 01m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0f616e3efc9951d3a195ef528796eb99d7b0f6c5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-26 18:30:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f2d3e73758b74b31828f53518e1a374d61ff237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 18:32:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d2f7b8fc73fdf830a32efa3ef33668f09dcadc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 19:33:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa6c90c243a14df296937d46db0e1338\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acccbb7c1ded42f5a667358c867afba0 : SUCCESS in 9m 06s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b30072e051a14f7aba2f1cd61fb0cef1 : SUCCESS in 21m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/46c4c82b7a9e405a8f2dea29445018d4 : SUCCESS in 19m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4e82c03a4d7745249f014c7757c5bc10 : SUCCESS in 21m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/504772ddcaee4574b837802d293fff6d : SUCCESS in 20m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ce79c00172c149b9961c7ebfb8c1428b : SUCCESS in 22m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/20715697f062432ab59c98df7cc1ae50 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/daf53629654f40e09f89364c860deaa7 : SUCCESS in 20m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0f3df1a5bc34d1789b0508e824af605 : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eecfce6dd1554ba98b02916834c89b08 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3f154100722643ae996faa5e07e6879c : SUCCESS in 14m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a309c8db4022471b87f88b42f17401a9 : SUCCESS in 18m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c275331f257d4ed19dcb4c7ce18eea2d : SUCCESS in 6m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9385a2c2404b4f8f87f2102ff92a58a8 : SUCCESS in 7m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bb0980e9b3649a1bcdf5caea5d3fe26 : SUCCESS in 31m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/358bb0607eb447ba9aefc6bf295d3733 : SUCCESS in 52m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/848a8335fe50460abbaf1d8885ab6c30 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a670b4404bd44a9b36403381cc33960 : SUCCESS in 20m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0079fe573aef3daf9e4e8f843807c1583ad5ba11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 19:33:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87aca9da23bcf50e442eb8ea0d2a96f4a51e775a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 19:35:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b280082dc5cf4f3a94141c0fea042264\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4cced2eb07c48f3a194521cbd78d069 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d6da6d7398274b66b38bf757294bc04d : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81547cb67d204e27d4c121d24f4ece5bb338e9ad","revisions":{"81547cb67d204e27d4c121d24f4ece5bb338e9ad":{"kind":"REWORK","_number":1,"created":"2024-09-26 12:32:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/69/930569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/930569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/930569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/930569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/930569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/930569/1"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-26 12:29:13.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-26 12:31:30.000000000","tz":60},"subject":"trivial: remove print from functional tests","message":"trivial: remove print from functional tests\n\nA print statement, presumably for debugging, crept in with the\nRelated-Change.\n\nChange-Id: I0cccb960b51446a3aa58be95c8c9d06cc6b6eada\nRelated-Change: I69974cc8a39731c980b54137b799a36b2e63a44a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81547cb67d204e27d4c121d24f4ece5bb338e9ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81547cb67d204e27d4c121d24f4ece5bb338e9ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
