)]}'
{"id":"openstack%2Fswift~930570","triplet_id":"openstack%2Fswift~master~I640837a4da985bcbe48f9ae5eeb56385b11034eb","project":"openstack/swift","branch":"master","topic":"p-fix-in-process-func-test-etag-quoter","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-26 21:20:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I640837a4da985bcbe48f9ae5eeb56385b11034eb","subject":"func tests: fix etag-quoter insertion in pipeline","status":"MERGED","created":"2024-09-26 12:32:33.000000000","updated":"2024-09-26 21:22:39.000000000","submitted":"2024-09-26 21:20:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930570-p-fix-in-process-func-test-etag-quoter","meta_rev_id":"cd6bd3292c6e7240d79c4ae287938bd65f18b344","_number":930570,"virtual_id_number":930570,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-26 21:20:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-26 20:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-26 20:19:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 13:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 20:19:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"366d8d4b4db4050d7ab5dfda342ac4811ee8cea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-26 12:32:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a02f865cf9a5e05d9e6d6f78abd664cd0a27513f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 13:13:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95e4f368c9124a7aa762aca0c150588d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f17398247e4e5aad800d8ad1eb1d3c : SUCCESS in 8m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebc62b64b20745e399afb79711a9247e : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00874f3097d7409986032d33f5576eb4 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af601cb5bc5740fdb1eb7c854c17c783 : SUCCESS in 23m 25s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1701c93a43a0423b8422ca08319949e0 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/13e6b11e88784dff8213638b4affbf6a : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a20f7abce0554dcba1aea26046aeeebe : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/096452edf7454a309bb1a3f4ed17c49e : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/22f5a506276941b19dbf775f6d74d79e : SUCCESS in 19m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e84c418a0a2e4be39c48094ffdc7fd89 : SUCCESS in 16m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2a798d578574fe08f59c89284950edd : SUCCESS in 39m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33e14c34dc34452f8c969ea3f0969f55 : SUCCESS in 38m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60d33625ef9b4d33a60717ebc95b76f0 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bc59bb1bd4b641c3a4e5deed6676d5ec : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4fc37cfdf41ea8cffc3229f02171784d1c57d664","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 15:00:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deca7d5a9b8a43b7ac01294ec55f98f5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/414897abc36b44c881b5695a0ee5e41b : FAILURE in 2h 00m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91b0fcb0def24d7784e892f10bbdbd24 : SUCCESS in 1h 07m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b6d0baaca204619b0da6f2b9931a3be : FAILURE in 1h 50m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d596784b8f794d478b070dbc4f85cb44 : SUCCESS in 1h 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0e0c543a7f08e6bcf62e533b9dfff9fc9ae30c6f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-26 20:19:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9601e166fbefebb47b16239b2ed0786d4b5d5a50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 20:19:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a446cc085d2067ca2f2a0fa87a0ac8229c1cca88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:20:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb87c5e13084a1cbfc536acfff21a8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce276b025774a9590b0cecd7bcff94e : SUCCESS in 8m 30s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/03dd4bcedaae451494913da45f1797d6 : SUCCESS in 19m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ae9fb3161cc40aba893bc3db84bcad3 : SUCCESS in 17m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/43df32808f924a66b6adb61c45166885 : SUCCESS in 11m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0169fcfa3f264b91ac0d65b4e8a42203 : SUCCESS in 16m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8f55592797c24d27a2e407341b1c9f51 : SUCCESS in 23m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1a5683a411004229bfb6a978cedf7da9 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/03af7504570a48cf9bdd0d2cd816aa44 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bca9539a8dc74459a3bba89355bb3edb : SUCCESS in 16m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d2d590962b014646a9fec90b596cd2d5 : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f5fcb3f6721a4f5bbf73177fafc19c73 : SUCCESS in 15m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/90816a49b9864a20b32aaebc5688c1df : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8634b4c30be14aac9c4a2fd98b73f079 : SUCCESS in 4m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1932227eaf084957a6d6908776eab689 : SUCCESS in 6m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c2bfdb12e17b4250a599920f7fd71616 : SUCCESS in 36m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8484ae25ade544afba9a058fa75a30ef : SUCCESS in 30m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9b55ed5ef44ce8bfea8e600344bea2 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/abc057b8ad084e43bfa0dd6580f7e322 : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"63e0468ad80dd053d66ecf48738fa2a707cda9ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:20:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cd6bd3292c6e7240d79c4ae287938bd65f18b344","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:22:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee56d0ff4b7f456ab879de608719ede9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a917fd89019418aa09aaf4933e9a3ac : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/467de82645dc470690ef71180c18c6ea : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86081b7822bbc003a10f9dd99d63d1ca21fdf194","revisions":{"86081b7822bbc003a10f9dd99d63d1ca21fdf194":{"kind":"REWORK","_number":1,"created":"2024-09-26 12:32:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/70/930570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/930570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/930570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/930570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/930570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/930570/1"}}},"commit":{"parents":[{"commit":"81547cb67d204e27d4c121d24f4ece5bb338e9ad","subject":"trivial: remove print from functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81547cb67d204e27d4c121d24f4ece5bb338e9ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-25 11:57:37.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-26 12:31:30.000000000","tz":60},"subject":"func tests: fix etag-quoter insertion in pipeline","message":"func tests: fix etag-quoter insertion in pipeline\n\nPreviously the in-process functional test setup attempted to insert\netag-quoter in the proxy pipeline by replacing the substring \u0027cache\nlisting_formats\u0027 with \u0027cache etag-quoter listing_formats\u0027. However,\nthe pipeline had already been modified to have \u0027cache domain-remap\nlisting_formats\u0027, so the etag-quoter replacement failed silently.\n\nThis patch modifies the etag-quoter replacement to match just\n\u0027cache\u0027. Some helper functions are also introduced to make the\npipeline modification more uniform and to consistently verify the\nintended modifications have been achieved, or otherwise raise an\nexception.\n\nChange-Id: I640837a4da985bcbe48f9ae5eeb56385b11034eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86081b7822bbc003a10f9dd99d63d1ca21fdf194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86081b7822bbc003a10f9dd99d63d1ca21fdf194"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
