)]}'
{"id":"openstack%2Fswift~930918","triplet_id":"openstack%2Fswift~master~Ideecd0aa58ddf091c9f25f15022a9066088f532b","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics_bytes","attention_set":{"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2025-11-03 16:59:31.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-11-03 04:33:51.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-10-31 10:08:54.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-11-03 05:43:36.000000000","reason":"Change was submitted"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2025-11-03 05:43:36.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-11-03 05:43:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ideecd0aa58ddf091c9f25f15022a9066088f532b","subject":"proxy-logging: Add real-time transfer bytes counters","status":"MERGED","created":"2024-09-30 13:59:57.000000000","updated":"2025-11-03 16:59:31.000000000","submitted":"2025-11-03 05:43:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":717,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"930918-tmp_labeled-metrics_bytes","meta_rev_id":"8ae5e38e20ef7737af2a6b2864c877fa68c5a26e","_number":930918,"virtual_id_number":930918,"owner":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-03 05:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-11-03 04:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2025-10-31 10:08:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-11-03 04:33:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-30 15:47:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-07 17:15:03.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-11-14 14:41:23.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-11-29 16:33:28.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-05-13 15:33:56.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"},{"updated":"2025-07-09 19:42:53.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"},{"updated":"2025-07-25 17:44:50.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-08-25 06:44:46.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"3abab044e95674cfb7d045c523d0d210177d8cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-09-30 13:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2225826f0eff4455ed0aeca042a5b15fa323681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:47:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21cafcf7ad4e41d48277845edc11bfa5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/398b4e388a63436cbad8cead8103ec15 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/0b771c658765481ba380b5bcfd522b3a : SUCCESS in 52m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00fe4f8c58f84cc6914e9918aee88ff3 : SUCCESS in 1h 06m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ffa39804f507462ab6853b64f157cd55 : SUCCESS in 1h 03m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f8f7c2680e9486cab18e0235adcfb9f : SUCCESS in 34m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/951046a0915e4027af6043c96c399ebe : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8dcdb2e571746ab8093f1f87c4a6fbb : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b88d78b36cfd4799bf58a4e6cee0c696 : SUCCESS in 22m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/659c4bc5224c4b36a556bd984520e202 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17d05a7371cb4649ab7528793d37d380 : SUCCESS in 23m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8ca0234ded04f6d9cde3c8cb3c46c1b : SUCCESS in 18m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dd6b1a7897934f49b88f60cb44d0e283 : SUCCESS in 20m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/39df72cc8b5144a2b527a15eece2609f : SUCCESS in 26m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c0763f9b5f3045e6aa13b5d0fb0a9a1f : SUCCESS in 36m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4da6c7dbbad944ae858a634e29412864 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/189287c1a6d94a86b2308c963f632675 : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2af861d5206b4addbcca4125c22d775d : SUCCESS in 17m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2812536cbd934dac8a08f6bd45d72bb9 : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1255692c7183401882088cbf984c3775 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1407f3f04aa14a1ab52768c6d410148f : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74284941017c42128afe2668faf84b06 : SUCCESS in 4m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/939a016e6e9848e3bd32eaf0b787ec8c : SUCCESS in 14m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a3422f1463f4ed4bc4ea37432874a6f : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d971c06d39dc476e8b870c3fc4e7343d : SUCCESS in 46m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e8ee787aecb14db48a88dcb37b0dc842 : SUCCESS in 50m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/504185fcbd6741b684a65277936ecbab : SUCCESS in 54m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a1d145998464220ab5b113d1f7daaed : SUCCESS in 19m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6298d30dcec42779415c80731d9997d : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8883649410354805b7bf543439e6895c : SUCCESS in 11m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c28c9c13ac154aeeaa93ecf06d02e865 : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2515761854c1cddf771d89e0331f471ebe4a6410","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 17:50:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56fbdad273b64cda8a94af657d0c1526\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7d672baecfe40d9956fbc426b1f851e : SUCCESS in 1h 38m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa6bda9d1cec45b393fded6ea91ac566 : SUCCESS in 1h 29m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff6cab7710124e6d983389a2bc109759 : SUCCESS in 1h 43m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6b8f3b1790d4f4cb2054e98b44e5a0e : SUCCESS in 1h 28m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1d736367b21bd63d3ddad617524b98a32299c73e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-02 16:59:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"62b4bd234c9b541a942a72ce3aeb4fd643b0553a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:30:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d09d4f01b93b4aa0bf5dfa1568b87e6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41d8bdf42d14fdeaaad04666d9c8344 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/0d2d8146a9b94760b4e4a798c754f195 : SUCCESS in 1h 10m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e3da21e3ba440cd928724e6a56f144c : SUCCESS in 1h 00m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb9ece3201b84310bab05e231892925b : SUCCESS in 43m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddd1b9bdd06f43429cce46f1481bcda9 : SUCCESS in 33m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b40b168179c947ec82ddcf67983744f6 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54f440278f8643959ecc39d80674b0f5 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7378f5ee77ba459fa52825a1bc6e1dea : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2e6cd23f98b40178781e4678e3cb606 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b727d479997d4aefb1587c2ad175450f : SUCCESS in 24m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/20e6d3befb56404fb09f79ce63a328a9 : SUCCESS in 22m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/35a304c1892046bebc141ad239306d2d : SUCCESS in 18m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2ac6f8a7785348cc8a96db7b267c9163 : SUCCESS in 17m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4f2c479a10a342a6bc6cf36413adf4df : SUCCESS in 29m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/882ce621114d407694b65f29df68152a : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/38f0a5a5b7d34b3796962fe75fbf8e27 : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4709c3df5e8e44b2bcc4739705cf9f54 : SUCCESS in 19m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/471ca026cfb543a28e2afdf107c82881 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/df6cfd97304142f2bb9067c3c6c32515 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38d69d316f184e508856963cfb23512c : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b7b65aae7804aee8a32e9e4f27da467 : SUCCESS in 4m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7f66ec2375b44777b6bbe92f878c356e : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6954547bb0647499779b8150fa86e2f : SUCCESS in 5m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b449963736843bb9df00858859358be : SUCCESS in 46m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f95f7f26f9c47878001274f4f6f8360 : FAILURE in 35m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d82eca57b36d4bf5aace29b013ed075e : FAILURE in 36m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed072b3bf5234e14a944a143c50326fd : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7da88ba7a99444eda8882457bddb0b03 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f31ff835a14f4ce9ad3006114e73d2f7 : SUCCESS in 25m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a6f9b6f56f746e69265224d7a1c0cd8 : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d883096137f3a8a16e0fbe6771edb02b6cdc1766","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 22:25:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7439a785bd24493950d54763321a921\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8a6c6325beb49e2bc9136d37d942f54 : SUCCESS in 1h 00m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa908711490748c99f45a7b3b22897ca : SUCCESS in 1h 01m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/faea6b6527704f7ba9b7b736265d914f : SUCCESS in 55m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a782f99f5754d0db42629705c4a2d96 : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4482de8bc114592c02fcfd43902415a818c55b47","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-07 15:23:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb7dac9cfed35ef8cab3fe72d708e2ecf692336d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 16:33:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eab08f21ca2b49028cabd5470d52b732\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/835ad236a0844d8c8eeb2020fc38b1ec : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/d19d076fa1db417e899355e46456591a : SUCCESS in 1h 04m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0b8729435934c3c96d4be929bbe7b8d : SUCCESS in 59m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/61a58b4894bf4a6185da248dddfa2801 : SUCCESS in 55m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eabe1ca88b6a47fb8a912efdb4d4c1ba : SUCCESS in 1h 00m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98b2c6ef70c1429fab1f32e591e0bf9d : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/376074a663b64735b4d4218e21991fb6 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4497746be7447669646b535e319681a : SUCCESS in 16m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/460b69fa653445aeb693989aa7a91d75 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5623794ab923414096596d608dbb493e : SUCCESS in 19m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19e7b2e565d5448faf037e83f2548ac7 : SUCCESS in 15m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b98e77f6ace943e48c18911898b0377b : SUCCESS in 19m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d69e26e7890b430fba5d3fd85fb895fb : SUCCESS in 20m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/62c6328fd51740848229a7b5008a053f : SUCCESS in 24m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5b011e557c5d46f996a6e0bc14889d76 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cdc55885da0f4edc89756f2eb7f2f8d2 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/27697d374da1472092829a3e05698315 : SUCCESS in 15m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/71dbb47c8d8f4e45989c2e9d37215688 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/93f8cdfa54cf4057bbc595546183d0b6 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/337d6762ba2e446fbc9d5aa7463d073c : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e76e26391d5145029f91d8ce20d32204 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce63e777a9b9428584c8195fb93a6ec3 : SUCCESS in 15m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0460a29c72a74e5794e36bae3c8a5ffd : SUCCESS in 9m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/64de8df38b24468db9dde4b862583e28 : SUCCESS in 55m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc2898a4b0cf4c9180ed8407dd6d0023 : FAILURE in 38m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d06b95650ce40529df864dac92af5a5 : FAILURE in 29m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21c00ea240dc49bf8c43bbe9e5acce8a : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/542657d4b0e640468e377e81511d18cb : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c175d86ba0ee400593827752e27b7478 : SUCCESS in 24m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe3e90c6246a4893bf0d86273af93daa : SUCCESS in 44m 57s","accounts_in_message":[],"_revision_number":2},{"id":"735abc7928e73bf5b3f3464946fe04633df6fb42","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-07 17:15:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b5908aafadacf061710dbe2e97d47e0a382ce3a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 18:22:18.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd05605155fa44378f0034dec801f7d5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72a43a4b0975468a83dacae99ab1e9a6 : SUCCESS in 1h 52m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3b777a3a3734c648148da3994d391e7 : SUCCESS in 1h 35m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4a9912e642f47fca3c561652f641275 : SUCCESS in 1h 55m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/229d34247b5247a098a795955c3001c6 : TIMED_OUT in 2h 01m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d0b8bde792eb3a58aaa7c34f87f1f16e75215544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-11 13:59:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"119b2607338c5b0fbae77421dfb9582ab8355918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 15:13:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22b524680346446a871d3dfb91b822ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d33e0f7825f44539bce3eb4deb023cf0 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/07513e37a1844ee68ea65a3bbf2d5d84 : SUCCESS in 1h 03m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f58403b412814c19b4df8105680b0569 : SUCCESS in 1h 06m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e015bc377314427f81845d6921797f85 : SUCCESS in 52m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ef9735114a54c108e7a65a6ae06c525 : SUCCESS in 1h 06m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16fc3abbb0cc46d59e72e588db8e865d : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41f883fc74e248818cf37b647fbbad52 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6d5a4b385414f3d8311ef33ce41bdc2 : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba40604fe0bc4219a03d6a6addc8565c : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55eb2d8e280e4011915127aac1daa897 : SUCCESS in 17m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c20fa9b7752e45fb8981e8e8f9f7022f : SUCCESS in 16m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42ca9f81f97d43e29b79ed59f1c85726 : SUCCESS in 20m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fe4697fb99794ce4844c9a32e8ba583c : SUCCESS in 20m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e990c775ad2947ef962104580ece4ccc : SUCCESS in 20m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/40e3b048cd0746f687d375c81305aecf : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a8f3ea1fb6941f9b6ee400b86dadd26 : SUCCESS in 20m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3d6c4bb3e1fa4d0caba9daef25f4f6cb : SUCCESS in 22m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a358caee5714408db4f9300e37db00cc : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6f94a374cae74c559c612867dbfa02d3 : SUCCESS in 21m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6f29b05ede814c418342953d1ee88815 : SUCCESS in 19m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/25176d49f06641d2a43aa2720954a27a : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5c68d4a9c12e48e4af5df636f989e4af : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a3dea07710a4eebb9a7ce35d4530106 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2fcf963701c47a6a7489b1ebb9a8ef7 : SUCCESS in 56m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac328da516f041759d96a2ae8342751a : SUCCESS in 51m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf95b8e43088450db639841e8333ce08 : SUCCESS in 55m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fdc1983adcb440fa5bbb98e9aeb1453 : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c65bf4d312584eabbf1dd0b9a1bf76da : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5738db1bcba3453f8925febb8dd1bdff : SUCCESS in 18m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bb8c8b4e9254a419aa31c1169e7e75f : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":3},{"id":"cf087b0546a525a7c0b6b9144552345ff5068d98","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-11 15:22:05.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1d94feffdf6fe6c7f2285bef3ae5781d1c9e30dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 15:30:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5f1396308964addafcd87aba992206d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc273a3f943f42368c020066d3c2022c : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":3},{"id":"6578bf38dcfbbb449948bb15e330a17d5db773cf","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-11 15:33:16.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"779e00f5a93cd747e7fc63bcfefb3440df7f5c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-25 13:59:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"cdf6df1a72ae5ee749a1155d540452364729dd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 15:16:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a55cb8ba5164bc085d6f313a640dc0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091a677069a24584983db859b6830c55 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/5ec573fe9b5f4b24ba8caec55c828303 : SUCCESS in 1h 02m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9db056ac156e45b385cd8a5440aff214 : SUCCESS in 1h 02m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4428837ba13941ddb25e4456a290432f : SUCCESS in 50m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/273cf6f926d043e7b4046b97dbbe9498 : SUCCESS in 58m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4344baa1033241f596a3ab0cc7029550 : SUCCESS in 17m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d839bc6da834692b694a1fd43bf96c8 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92a454368d054349aa0ad0180e4dbd7e : SUCCESS in 20m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bc328424725649c380945a3555c34862 : POST_FAILURE in 36m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dcd52722d3434143b16686bf6748757b : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/057a79735b004d0e914e483f7bc91854 : SUCCESS in 18m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8c2ab377772142cd9dd24f8ab04751c8 : SUCCESS in 8m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/68c54d0dd4cf46389124c2a609fb4f8c : SUCCESS in 19m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2a255b6028b34091887eceaadfd78559 : SUCCESS in 22m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee6aea125b5e4c9088d45b915a186339 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/00a3f70277a14354bae6298bf591a5a4 : SUCCESS in 10m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ed71924a6b6543248a65b5b93e47f544 : SUCCESS in 18m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d79ca9f1367d47d297fa5183a71ea0ba : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8f4d1b455f374d0da0b1b73f91b6632b : SUCCESS in 17m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc602939c2764a78ae440207d94304ed : SUCCESS in 13m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4d4b12cc0764f9d9ab26ec550a38e29 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3be6bdfd32f841208d7f40dd98e5a639 : SUCCESS in 9m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c3bc923150674e769dc78282567b7095 : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0111d0e6086d43378e178e01e4ba98dc : SUCCESS in 45m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97073b7a3958476681776911d3c2c493 : SUCCESS in 41m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0eb83a254b3844a6a23b42ef7d54eccb : SUCCESS in 28m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04c0e5d8c14a497b8105c7d1ea642e3d : SUCCESS in 21m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf5eeb8ca85455a9c8d1304a26f71fb : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c49522c5c4cc4fb7a00d16998b05c708 : SUCCESS in 17m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/919d6b3234d2459a8ad388129314685f : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":4},{"id":"15c81a70efeef62d21e0aea498d29aada62c6fec","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-25 15:18:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ec08075b84c680f31994d36584447ec4d3761af0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 16:06:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a7689069594c17ac12972ed1948dd0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8fe9752c0f0d4e01bb9a0ee902b7319b : SUCCESS in 2h 05m 17s","accounts_in_message":[],"_revision_number":4},{"id":"88385b4cb0bf4ad1b773fc44366c408fde5ed0b7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-29 17:29:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a7c52999f7cfca1879b4ca42d346ebabe5e22a1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-31 19:36:37.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f0a6fa590149b78a5fe6013a0c379d183e9adddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 20:50:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2963c9415aba41dfa604a7524b7d88f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab1258fac6b4a398f319499538c4bcf : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/07f37e7f684840199c9e42aa0935966c : SUCCESS in 38m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a6ed093ff834a589b6712542bb595a2 : SUCCESS in 39m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/465f3fd109af406a8ad1c95568548baf : SUCCESS in 58m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5095ba7c8d1f4c51a5bd08f020b0a6d6 : SUCCESS in 44m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/929176c2721e4551bcf63b0b02ca44d1 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62279b19e018472683349199faa3eb31 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/901f04116e5c40aa8b2c2eaf50447a1e : SUCCESS in 18m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3fb95f2b16934427823b66f7ac46abe6 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c4f63b48ce441a48166e642ec6dc07a : SUCCESS in 23m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23a692f15bcc4015a80eed0c3b1ea1ff : SUCCESS in 19m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d94986a605a0405a986fbb34ab29c69f : SUCCESS in 9m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d5122ccf1cee40a1bcb3a37c2ea3438a : SUCCESS in 15m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/543817ecdad948fd8a84d52a5185d34a : SUCCESS in 34m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30d5bee750134b23ad237f27377af0a5 : SUCCESS in 11m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3eed53da48a043cdbac9a4903fcff145 : SUCCESS in 17m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9676d99549264cc3a461df7b016191eb : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3889b01cb4104fb8811c47438bcafd08 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/849dbba0c49c4412951f13800dbea95e : SUCCESS in 15m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d231aece29a046a18455acc0f015af6c : SUCCESS in 13m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9481010569e64fb6ad6ba073b45b93e4 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3520bd50f59421cad2bbc9d1a61315b : SUCCESS in 13m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c60d34b8ddb9476894d44800854f6256 : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/302aa8947df542a68fb9f580711f9cdc : SUCCESS in 42m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b94252507a5f4e218c123890740e522d : SUCCESS in 48m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/80e5f87a4ce84cb89bdde8e947549f87 : SUCCESS in 34m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51c3608f138e40689e4926886339a7f4 : SUCCESS in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c701888ff7654810952b9977fc09c9ad : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4626d30f22d749d69968f0ea85b55999 : SUCCESS in 17m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/802f2a10f3c94522938d76ad4a24461e : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":5},{"id":"13f7b7684c6fde92de16ec51e9beded45cae9782","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 10:54:01.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13dc8c00ffc244e6802e3ce15a7615a4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28ef1bd3fd8341978fa4a0e5e774df16 : TIMED_OUT in 3h 00m 57s","accounts_in_message":[],"_revision_number":5},{"id":"770fc5ba81f79fd485837068bcefe7f8e5d99a74","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-11-01 15:35:26.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cb15c832481ae6c0f7b946721addefada72f84b3","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-11-01 15:35:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e5aae7fa129404ef30d9045a3774711b5c558385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-11-06 14:59:35.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"036f399cd699e042639faa044533d2b725239f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 16:12:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/965650f77d8e45539a3eb95bd5c9f6d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25d1387aa16416e9f8d9b4aba1d9ad1 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/2fe48cafc24c45659f2be984c1ab18f1 : SUCCESS in 54m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b72bc517f97148609c0cf4fa5cbcc526 : SUCCESS in 1h 00m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d439d372e2ec450ab715c9d999b43031 : SUCCESS in 57m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff7246d8d4d64e97b8904275cbf9cfeb : SUCCESS in 1h 06m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6042f22592ea4948b47890420ad253b0 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e64e588e1ed4d99a4d23c41e391aadf : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1fc2f3061c9c482bad13ca37af23756a : SUCCESS in 26m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d462ea144a24441f90e4f3d58501e982 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a8f128763294683984df9907c718736 : SUCCESS in 25m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29f851e309d04c2e88b246da485668a5 : SUCCESS in 23m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6bf01e541e354373b29a73897ba6da6b : SUCCESS in 10m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/adc17c1115a84c1a98218cb9e871f0e2 : SUCCESS in 22m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ee27443d1e04a19a214cfc44aeaae7c : SUCCESS in 10m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/27e5cd8868fe45a1988528874d67bb62 : SUCCESS in 10m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3ad276294b3e4343b5ee342d91df8cba : SUCCESS in 17m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4ef7cf19d61e415881414c0abd639f9a : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/84201325a9544f578eb3ba0b15d9c87e : SUCCESS in 15m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1c242b5449ab4e11bde053024770320b : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a1a120fe8d1b47479b77fb0e2a987fd8 : SUCCESS in 4m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c9774626f77440d1b3dfde46ac1f7352 : SUCCESS in 16m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/92b6aaed5ba54991885bcd085db4502f : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce98d04d18024b1bae8edd19dfd1d9a6 : SUCCESS in 39m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a77877374be140fb8292341e39ecb1be : SUCCESS in 31m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58f1d5db00b64a6089f4c0bc0352135d : SUCCESS in 33m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21f3e65c5af4429daea3ab5c6af704ca : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2200fd14c55648048fab50836ee5712b : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e12777d0c5b544839545f2b0a3ef2aa4 : SUCCESS in 16m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/840ccc9a1c64405a83fd5b3a4014876c : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":6},{"id":"1615185d3b04dcf2165b4ba4210a94420e0cf382","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 17:45:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f83f2ae54214c0580a75702a4ce721d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e39a67f1d51042b4947d6c519f9e82ed : SUCCESS in 2h 06m 44s","accounts_in_message":[],"_revision_number":6},{"id":"34996bc75ca6f593d928381298d4af325195443c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-14 14:41:23.000000000","message":"Patch Set 6:\n\n(20 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d21278b9a94cd7c6739dd5897408c28bc9120615","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-15 19:40:40.000000000","message":"Patch Set 6:\n\n(14 comments)","accounts_in_message":[],"_revision_number":6},{"id":"50cfce9a3e0bacd0e5a159d88d1a76c3079ff742","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-19 17:50:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"83aaeaa742cdf60cda754faeb378c1f60824bdd3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-19 18:48:36.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e6df311bac282d96055b00037a8324cedb49d04a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-19 18:58:22.000000000","message":"Patch Set 6:\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"775ad9dd119d52fe9927ca2020317205a6299bcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-11-22 16:56:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"b2a4c731faad97da1e3d69b7ff25e07dad1ad9e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 18:59:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9dd73d2ec24244e4b034323c2f25478d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a537741ba56468bbb10ed4f8d7cf197 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/93489744403242e7ab33d4b1edff2c5c : SUCCESS in 50m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c3ea0c3672943c5ba5f4c7bfd644a27 : SUCCESS in 46m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/16fdc134cf9b473bbe4d1d3bc3f78486 : SUCCESS in 41m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cb3f26e674549918116429dce0e5a34 : SUCCESS in 58m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c939888275e5457e8443027a3b0773e6 : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07f59124780445949a8835c49b4999f0 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/882ead69d73343a5bbd230054934875b : SUCCESS in 23m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2af0ee4e6de34bd994e6e49f8f89d358 : FAILURE in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9fde316893c419ebdfa38d82e7bcab8 : SUCCESS in 21m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2a5762418e94c8085fd753f3476c70c : SUCCESS in 11m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c5efb629986748559c7b6da98dd5f8c1 : SUCCESS in 20m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/39659a6c5e0c4c7ea909a3908a10a9a3 : SUCCESS in 36m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/27ae03578c234bc5805402ac85aa2ade : SUCCESS in 19m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dd867de805cd4b51b497a155f087d619 : SUCCESS in 11m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f3f15df48805448ebbaffdca796a2cb0 : SUCCESS in 17m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e916edcc4af546258997e96978d86c3e : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0832b4276977463aacdb2641d1d67f60 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/82b8acf43ec347368c4c5ca8c55a9d96 : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6913bb68c8945319519093d5299e91e : FAILURE in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f8cd52bd1e4425ca186bcf60d35639d : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b117a2cd3d764cb38e594a92e96fae99 : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cda9ec5ea7e4f50a2bff11cc3b29848 : SUCCESS in 52m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a18663bf833f4f5f96163e554f8ea6c1 : SUCCESS in 55m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de4e7562900c42f1b9de4c1182a8065c : SUCCESS in 42m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8acdca92f88f45cc8c39387e9c4031c2 : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56682bd033b84087b39756067b69894f : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eb9c32137af1447a85c3ce76a0f80d65 : SUCCESS in 21m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d48e2859a3645a7a7b256ecc7f83d6b : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":7},{"id":"4b9ee0b911947baa95e252f353be2c07d374e263","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 19:36:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad1ce655c1904a80a77bdbc7e2cb7561\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc976602d3e04043a37566c163803cdf : SUCCESS in 1h 51m 03s","accounts_in_message":[],"_revision_number":7},{"id":"8062d7d788300517ff28f1945ed9fe676e1def9e","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-11-22 20:30:38.000000000","message":"Patch Set 7:\n\n(39 comments)","accounts_in_message":[],"_revision_number":7},{"id":"846ecd0929d36ec418f5702a6e550371bdc62543","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-11-26 14:59:37.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"546b1650b2fe77cbd801e0e67172aa411ee14bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:07:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfaa00d6cce545ce87735af531d0bbdc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e5e7cfa3d04fe1be55046bce81023d : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/f240cb6b2a504c1f8a7006f119e976bc : SUCCESS in 54m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1176c55a88ab4f29a3da5ddbcee50b1f : SUCCESS in 44m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1ba3fea0542c4efa9b882f4f7f5f3fe0 : SUCCESS in 48m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb3ae60ea3b74b13ab908d1729a869f2 : SUCCESS in 1h 02m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82ff689612be4d65b93239fded3760b7 : SUCCESS in 24m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bfd02e9683a4c5eab5a5ff5cfb4f04f : SUCCESS in 25m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdbb06c19fae45388c5d97ede735d475 : SUCCESS in 22m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/256da888b7684d7ebbce4375a1ab9167 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b37aed3a9886484c8f3ba80b7bad58c0 : SUCCESS in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/87147de8f98d422caffd089103d71e58 : SUCCESS in 9m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/63622cfa33eb4ce4abdf9efbbd53b1ba : SUCCESS in 23m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/38b3c690567745038534d55503b88b77 : SUCCESS in 15m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a4fb321497bd4a4a93af1f1aa67838e8 : SUCCESS in 20m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/23dea18b43d84b84b71dc463460cdafa : SUCCESS in 10m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a48daedf730c48feb3a5570262547ab7 : SUCCESS in 22m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d2e1843bdbd84ee0a9cdb664a973b548 : SUCCESS in 9m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e58c8ff2bb2c42468f342d01bda6815d : SUCCESS in 9m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c7d7d4de7d0148cea5153a2906ad630e : SUCCESS in 9m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab9ffaea5ecd47ba8795380572f88042 : SUCCESS in 4m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d9b989b907cc4b678fcf3fa6d7260b07 : SUCCESS in 13m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0274d661262c4e08b366d6fed9dea598 : SUCCESS in 7m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6d624fec5584373a8500234c8777257 : SUCCESS in 48m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cee37ac6f5e044f183c1e092a57b2ec6 : SUCCESS in 45m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bdae545f91a340ecb65cac5b7ab74815 : SUCCESS in 48m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d80d445f0c44241aaa713f838d2b661 : SUCCESS in 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab6233ce5354b109d5b8c1bff7ca072 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e132a26867504660a76d88649ac28d23 : SUCCESS in 18m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66982b79c73442d4a876581223a6ca96 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":8},{"id":"fe2a4989e2a7207ce09b0b39e2fb2a2cd9dcc053","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:31:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf1c0a73a8c542de9008c6e6fecc0300\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b63121c54fe949ed8aa08b2d3239e615 : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":8},{"id":"c61ab70d547693eed14abc17d5d34c4e0b37cfdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-11-26 17:22:35.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"0d45a2927a87b8ed5fa56ece11c129e78dce9179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 18:29:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f839a7b35f524d0782baa5e648adfa80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c43ef20fa4d14502a6051bc6b2f4514e : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/fcbd32ebea4a4300a3a625b61d09f3dc : SUCCESS in 54m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e586698788a47f1ab0a171251f68887 : SUCCESS in 53m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c8893bbda69641a9937cbbf3ee087d9f : SUCCESS in 49m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd8206f537894e109afcb0ff48ec8557 : SUCCESS in 57m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5340441425d04606a90220118b3a4057 : SUCCESS in 24m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1410636b07ef490fba0455e8080b81e9 : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0bcac8757e64bdbbf8a4696852ff92a : SUCCESS in 27m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/caea27effd95494eb00e952a7ace554d : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51644da7207c4b7c831ca17eb439cbcd : SUCCESS in 30m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/870f19786a5f45c9a2edac91b4217202 : SUCCESS in 12m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bed580448e1d4ef0a180f871a28132c2 : SUCCESS in 30m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/66d1366e5a864fbe985633e302fd510e : SUCCESS in 17m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/55c52940034343389739ef4702f1540e : SUCCESS in 11m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d98e32e4284344b3a8a4e89491cb73c2 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/869eff5399d44ac0bb985772b59214dc : SUCCESS in 19m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e14fb15d755c485381596df341cc5db9 : SUCCESS in 12m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8eb42583105b4efba4f572bc365aba66 : SUCCESS in 13m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cf4a7d87b07f4d8b946f703e71d35613 : SUCCESS in 10m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8968a2fd767944f594846da0921ac821 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f37b1cb323f64575928072e55dc2053d : SUCCESS in 13m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b5bf74b005b24e448663b76b6e9a3a59 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c56dae8daefe4aca99c593426b2b1d70 : SUCCESS in 58m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0aa2bd1d44d494fad07fc97cc3f7645 : SUCCESS in 31m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8bc6326d76f04598b50a335eae904d43 : SUCCESS in 26m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6160694643944f9ab3a4bac92aa8804e : SUCCESS in 28m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe06437456543a1bbcf34bbfe146920 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf879534b55845dea8fe5560312673d6 : SUCCESS in 18m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/718b167b6de54f8d9e3f04195bf06f39 : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":9},{"id":"18a517346ce7ca236c5e5fb8ccf828b6356588db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:00:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fae84ae2c1a848089baa298223eb489d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db0ba6e1432943f89f8ba012a25c4e2d : SUCCESS in 1h 28m 23s","accounts_in_message":[],"_revision_number":9},{"id":"3c9db4b292eacdc99e65c8934ed5e7d5148bdd8b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 16:33:28.000000000","message":"Patch Set 9: Code-Review-1\n\n(19 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e449e2c41988c5d33619a80189c6af75101cc38e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-12-03 20:56:37.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"b82ada662d43a87a2cae2b3e801ec1b9bc83f50d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 22:02:58.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fabf42d393f04ec5971ea435c9b5a4ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/377c7dd8c2dc491cad95fec87afc9700 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/b5600d2720f24a54a05c8fca28bb394d : SUCCESS in 33m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eedc2c28f46047acb9d76803315d571d : SUCCESS in 1h 00m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fc6e11d6eb004ea58fe573e51b5aa03d : SUCCESS in 36m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74caf6cd15d945f8a2ae5e4855066818 : SUCCESS in 40m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ac297e7a6ec42faa7fa8e3ab533e5c5 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20168a9e04bd40468bab33d200ffb57b : SUCCESS in 11m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5271f3e69ba4517ae33485d7c71bf17 : SUCCESS in 19m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/74d8282fbcc64a7eabc46a88b521bbdc : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d953292b7492498a9610469ba52ecfb5 : SUCCESS in 9m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f607d9e49c4e58b5444193811e03c4 : SUCCESS in 18m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0c35183bd1a34d1382a53749b3ea0dec : SUCCESS in 22m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5201ee115b3641eba4191b6687d4a7a3 : SUCCESS in 33m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b19aed3deb3747ed9bab4ba019a66cb8 : SUCCESS in 20m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/929d4d1edd4944c9ae2f24e75dc2794a : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e9b82d417f74624bee622155f67b074 : SUCCESS in 21m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/979efdd1b3d447538f7a4f896a51e4dd : SUCCESS in 13m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b8737b29c724432d93d162d174c7983d : SUCCESS in 17m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ff896f1eac3042458bb6abed009ea754 : SUCCESS in 17m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/71d6b5c94d3a4866942ce8a5325072f6 : SUCCESS in 6m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8eedc3555d3f4172a64d70230567b2ad : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8c7fb784244749b8bcd7eadb36d7329a : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9723865c65a4f44a3d6c4b9f4c6cd7a : SUCCESS in 48m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e74e9ff205647718bbc464622e5860b : SUCCESS in 51m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70e004ecf84a480e983bde409a312ac7 : SUCCESS in 44m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfba4e0e85bb46f5a24c61cecf838b80 : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97be538f978b4bdb93b26ee1c764355d : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aaa161b5c8ea4e6398b70a613707f940 : SUCCESS in 16m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/baec5e806aa243fe82845e2cb835d04a : SUCCESS in 53m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/867c51a485154c7a89edb59a30aaf2ac : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":10},{"id":"aed453fb3b0d7917e82ad9e7fb06062f95252901","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 22:27:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c6a302557b54691b6dbc614c7bcbc57\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c8435926063c4858960c3dd920c5432a : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":10},{"id":"468bc49b0fe77526e30486f813edd2285d279fa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-12-10 14:59:34.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"40a66b1bea46086b5071e45f3bcd63ade74db750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 16:15:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbaaa81750f04a88b0f8e68d01398223\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354a797c53d44d7f9dffca0a99d91a2b : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/ccd2580fb59645e3bbd7bb7990bd0693 : SUCCESS in 1h 01m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1e9da2e5b4ed48f68a465213d242c8ab : SUCCESS in 55m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8ce33cc9084e44a5bc3e2a7342edfbfa : SUCCESS in 49m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/156031c5579a4c71904e68cab74b27bd : SUCCESS in 1h 09m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d41358ba7f4849dc859de872c21ddb71 : SUCCESS in 14m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b60cc73916a24069ae70c892d91c89d3 : SUCCESS in 22m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/616664e51006461ca243c4ea48129a54 : SUCCESS in 21m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c09340d293dc4cea870aeb705bc62f5d : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ebe256ce94e4a70b45c067af7b22020 : SUCCESS in 18m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ccc0c05696cd4dd88c3d0606f7fa8d6b : SUCCESS in 18m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4f3c343e47e542dca89381e115e96a00 : SUCCESS in 20m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3a86fe59936d4884b7f1a01931d80d49 : SUCCESS in 32m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e69ad25162fd46078dcd084f23096a6b : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d0f278ce96114f44afb1b31e89635419 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd2c510bd22b49479ab0336462ca233c : SUCCESS in 18m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/748c838ab1ad4246856cad30c7fa112e : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a99cccf6584741c8ac6a2f2c7f290d67 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dcc293eb77a64159a604e0c226e55281 : SUCCESS in 20m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83de8766618d4277a4d7790d32435da7 : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d4d2b19b6b2c4b8abd5377e445b1f94d : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf7858a9452f4ad1afe2b643fef092f4 : SUCCESS in 7m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a25e95b97e647c6bf09bcc1593ecd25 : SUCCESS in 1h 03m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a7b15b542f947fcb4da0fd2635dc73a : SUCCESS in 41m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e5136977c3b483ba3ed24f90b391799 : SUCCESS in 41m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f52101fe5a34567854eb6fe66eee9d9 : SUCCESS in 21m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd44e8dceaa940e89fb8f4e29a439c55 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e68d7a55631245d28c8433183ec56cf3 : SUCCESS in 23m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89283a621aff4cdab1bc009fae511191 : SUCCESS in 1h 01m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eae87b25818644f3ba913ff5b4bbf726 : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":11},{"id":"ec749a90b62e4614dfbbd965ec77307a011e0994","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-12-10 16:29:15.000000000","message":"Patch Set 11:\n\n(17 comments)","accounts_in_message":[],"_revision_number":11},{"id":"799b543a6dc7574ea6bea54d5e985ad26c581044","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 17:03:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30047521e5d04f2aba55497c61d38283\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fddbf86b14ce40e58233c913550f89c5 : SUCCESS in 1h 53m 58s","accounts_in_message":[],"_revision_number":11},{"id":"c836907958e81fc6029007ffa68113095eb48b82","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-11 15:35:43.000000000","message":"Patch Set 11: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"fbcc66532469f04126e83924f4d66a757c3580c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-01-03 14:59:41.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"0d58c41c0e70e567c0adb7b491dd6f590336c71b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 16:22:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0016bbcdbb284e4682a2f49b79493a58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d40b846edc418d82ab05d482d785cd : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/b610b4475f1d48cf9fbc14550c5d2806 : SUCCESS in 1h 10m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35fade46fa0d491aa12450a64fcaa61c : SUCCESS in 1h 00m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8e7834d22b2a482e87808561701ef493 : SUCCESS in 1h 08m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c3b68fa455241f5a1c5c79f044d8328 : SUCCESS in 1h 05m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43a3913c11564d378e5bb6be263048ac : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa533db28c874dbfa639e5b7a3fd394f : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a988b6575c1479780c30001eaec6eda : SUCCESS in 22m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1128dffc3684281add31e7cfd8e5b1e : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a8c31302f544684a668afe4514149b7 : SUCCESS in 12m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2bda6d30c7874e68805488f7a73bbf65 : SUCCESS in 21m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/227c8efa2f1b492b8492189faf8ff01e : SUCCESS in 18m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/82bda1b2a7ac4626b39027ddd4feb26f : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/143e079fc62843179fbaf3eda2dd1ba6 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/584036b3c44d460da690d1df9f2fee28 : SUCCESS in 19m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6d1e65df77734fb6a5d45ec367dd18dd : SUCCESS in 8m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0608467a3c1d437f84698d42568ca038 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f3403c919f744e5844f5b0df880bf41 : SUCCESS in 4m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c364aa807af64046b15f6276a886d8a1 : SUCCESS in 57m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa58e56d51c5435885b0febaf8220170 : SUCCESS in 38m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/86fdb99c0dcf4894a4585a01790b7a68 : SUCCESS in 29m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/527a39ad37004b309b0432070ddc5754 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e88b01194b94fdb8fdc76e74dda9835 : SUCCESS in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb686e94add84befb28c7b7cbe008d2b : SUCCESS in 12m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae9a6630386c4850aab003f39ab0e92a : SUCCESS in 1h 02m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57e5ed57134743bdae0981511ce0e238 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":12},{"id":"95af8f10f32e2f8430669825672d1b9701c15cf6","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-01-03 16:33:48.000000000","message":"Patch Set 12:\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"8d2c2393cbf499613e267722eb65b83a16789c97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 17:46:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/944db7d6daf9484ba80c38143cfcf759\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f7bcc0011cb4a9e81ee1d7580ab1e3b : SUCCESS in 1h 59m 35s","accounts_in_message":[],"_revision_number":12},{"id":"36536f9d541c82ab2ec7358f82e3a74bcdd824a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-01-10 20:58:36.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"7a4a3d9d55fcd1e0b1bb34c771d3e95b1f842b37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 22:23:33.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49f6f1adf7cc4b948c8af2bb14c6da9f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8471defa25e44927bab11daf369df507 : SUCCESS in 1h 23m 07s","accounts_in_message":[],"_revision_number":13},{"id":"c5135841dc63dc6fb4f4edcd267738dd6a3d66c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 22:48:03.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f18a1d804f4c47c6995f7009ecbf5d92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5346f185215469f84e21eaece16402d : SUCCESS in 4m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/9b4c059ae75e459783a7baab1047dbed : SUCCESS in 34m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0983f185ae942e4a7c3c423820c453f : SUCCESS in 32m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cbd824e7383549d98f426f11d07fad3e : SUCCESS in 41m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32e573ef8c7b4eeea5b0fefa1c183fa8 : SUCCESS in 38m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/458500e6270841fe8ec074dbd0dc4904 : SUCCESS in 15m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b82aedb97914da182bbfa35205aadb1 : SUCCESS in 11m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91171ef24cb74529b0d23d3d6f9e5a35 : SUCCESS in 15m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f49e80b01ec46f0b052c24a8a3b7cb4 : SUCCESS in 4m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a452987e605641e7a304bb8cc44e4eb4 : SUCCESS in 13m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e797b56d2c9d4191ab0373dd05e11d57 : SUCCESS in 15m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/92893b069c564d9d88c9250d2b4cca89 : SUCCESS in 33m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b7470d5283b64cd7a163dc20d79bbeb2 : SUCCESS in 11m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2efb5d139d95401eaaafa46aa467dcf1 : SUCCESS in 14m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c98a8a57228041e5897316bab3a76e64 : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9522c622bd442b199c124d2b6d7c423 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/31fe26fe5d834d848c7c9290c3de656b : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b9b10b8221b44238ab4025b3ae0cd8bd : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b72b10e3121470abab78a647a5e4257 : SUCCESS in 56m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/114715f73a5644fbb4f7a0590a0da2a7 : SUCCESS in 29m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1c3a31fb60849c6bd34ee977e27cc63 : SUCCESS in 29m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4333195987814eb1b1a232f9fb993dc5 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9cf3b17351c4915bedd896f5df98f81 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e98661409d724b43a8ed68a30941990f : SUCCESS in 16m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38a10dd9a4924187ab90ae617828a35d : SUCCESS in 55m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9955b595f978407ca6f3577736dd213e : SUCCESS in 1h 37m 25s","accounts_in_message":[],"_revision_number":13},{"id":"14cf8a0f380dfd1b62c64e3ae94d323a6c4d23ef","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-15 15:01:49.000000000","message":"Patch Set 13: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":13},{"id":"6dc23d89aa71584c07c81b7b2de5d803576085ca","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-17 21:20:47.000000000","message":"Patch Set 13: Code-Review-1\n\n(32 comments)","accounts_in_message":[],"_revision_number":13},{"id":"aed47e8ce42d6c936ec27c70fe84d5e74816fd1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-11 14:59:07.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"47feeeebf75d6bc9a2ea00494a028397803e1f9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 16:19:24.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a931f9cc483407d8b8239ac18ecbc94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e414892f234b476faa712d0b46749a7b : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/4019b60da38d4804b1db65c4bd766eae : SUCCESS in 1h 12m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/854d017f7ac14d25adec743890cef2f4 : SUCCESS in 1h 04m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45720126509c48818721c5f5d752299b : SUCCESS in 1h 07m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b209900775d4e5f8c05bb703a7cff37 : SUCCESS in 1h 14m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47176059b5484ed3bce83e817839e2bc : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40763ac52654441c86a76d3bd84591be : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fed731a8812d4c828f799dfafac9a0aa : SUCCESS in 23m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/22f6473aefa84330bb3d901a1072bde2 : FAILURE in 4m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62b482270a2f43ce8292d4dc0537d775 : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a69ccb1932ce4b23b035218d32b2265f : SUCCESS in 24m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b07b6f24be7046328e316b76e470cbf4 : SUCCESS in 22m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5383d02e87e34084b63a73a366e4deb9 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/61f53e7c9d48470ca978c5bcb47060eb : SUCCESS in 8m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b14af44e078e4cdba4f65e7b5ad8a4f0 : SUCCESS in 18m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3a4e28d0492c4f2f9d8420e189b03eaf : SUCCESS in 4m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13f94abab64749c1b03cf0e1f18a71db : FAILURE in 10m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6467a88a9f7e43b9b2c2c35d43249c96 : SUCCESS in 5m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/36492954ec794e45bc3c866bc2a15860 : SUCCESS in 44m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d191733a242e46d58abf559d75e23abe : SUCCESS in 36m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c1717463e66b41bbb00a0106cddc573f : SUCCESS in 55m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1206f719fe7b401b94674d29d9d24403 : SUCCESS in 21m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40344174850a407d9892f8a4b58525b7 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/141b1e5ccfcb4d18804c755e96685c35 : SUCCESS in 19m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f9d093f66b04abca8610d990b32d8c8 : SUCCESS in 55m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21e8a426d2e84e8cb98d5acb004aec89 : SUCCESS in 52m 59s","accounts_in_message":[],"_revision_number":14},{"id":"4af806be4d2999ac00110ee0d20bb40ce9af6ccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:22:03.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd3f115892924925952430738b87ab4d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f866622a7a114feaab310716ecff1b42 : SUCCESS in 1h 47m 40s","accounts_in_message":[],"_revision_number":14},{"id":"576f7aa1851788b769331c8c466ad3f6ec18fb59","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-11 20:22:12.000000000","message":"Patch Set 14:\n\n(14 comments)","accounts_in_message":[],"_revision_number":14},{"id":"91bf6bebd51e5e74fa5ba2f19ab233f5c391317e","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-11 20:34:03.000000000","message":"Patch Set 14:\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"c497484c0b6f27f6f3ebb0e997780f911912c04c","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-11 20:34:26.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"682cd0ddb3f0abfe2550e1b62d33071034626b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-21 22:23:15.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"e130b8b5a022aa8406570e034d6aeb1f459c5076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:32:47.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e108045af0014cf18ab2b5929e32a0f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2387d0c5eba846009c116f602085ab78 : FAILURE in 6m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/09b06e17564c4997a0e6da5fc386e576 : SUCCESS in 1h 00m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e280c28725d0475b9b6d43be77000228 : SUCCESS in 1h 02m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7f952d2dcb04341a0f94bca5ec95056 : SUCCESS in 58m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0cee77480fb49409ee6a2469235adb7 : SUCCESS in 1h 06m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee3b697a1b1e4a19a89518e9ba87eb0d : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee0a05d2cce0409296225ec87c61f488 : SUCCESS in 23m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a146209706ac49818b03cc82dce46a9e : SUCCESS in 18m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/961ba505c05a4b90a1f74ad2026c3a33 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ca6e557e06ad4187b0c907cc68054f77 : SUCCESS in 14m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9b8ade6d42594675a76ac98e5c034b69 : SUCCESS in 17m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ecb194d28afe4a8784d4c6121568a334 : SUCCESS in 22m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4ae6465bb7bd4d41aa487b4ccfd98eb4 : SUCCESS in 8m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cbcf9230136f4ac8a365ecdac32da2b3 : SUCCESS in 8m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bfbc4d151d17498ab914f57d00497399 : SUCCESS in 16m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b039e5639c543eeb854dd6c51157653 : SUCCESS in 4m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/656eba12d3e74a9c9df2d7924ebc325b : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d7553fc0b84245688eee9df47db0d234 : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/063680fc04824e808dffe2521a690f90 : SUCCESS in 52m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9955de4053bb403788e2f5e7be88eaf9 : SUCCESS in 45m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4139d6b698fd46ea8ad0194f04f533f0 : SUCCESS in 45m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e776c48b7aa4c89bcf21376f2affebc : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd1f85705f0642e3a0df5a005af0063e : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4fca586e9a884b4491cd233a36ec33e3 : SUCCESS in 21m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35fe019e3277446db656ecf8e2d8c778 : SUCCESS in 50m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/353f04346dd54901b1d5d8d980bbdbd8 : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":15},{"id":"32ed5d5185f8c19a79749ba7ca41a3230ecd9702","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 00:36:57.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03764b472e224088a4674a95578d7bb9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3034e77c63274568a8501feebd301ae0 : SUCCESS in 2h 07m 26s","accounts_in_message":[],"_revision_number":15},{"id":"2cf84e04dade1b637fc97428392dddc14d0eff30","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-24 16:22:27.000000000","message":"Patch Set 15:\n\n(22 comments)","accounts_in_message":[],"_revision_number":15},{"id":"e01e17500490502d806f024774292f04ba9b2533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-25 15:01:08.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"bb454340fdd8fd5f8a17907cc6c379f781002673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 16:46:11.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e67a07879844ce0a2c0ddefa53f0185\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628b03d2262841d6928edbde98da6f8f : FAILURE in 5m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/5fcd14ef7c374e5b8596382f25487a6d : SUCCESS in 1h 11m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c446bd0e651421a93e9367221842c9d : SUCCESS in 1h 03m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/add1d500e89d4e529e4d8eab245b94a5 : SUCCESS in 1h 01m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a940eb15ae7454781b27d0d201d65ad : SUCCESS in 1h 08m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d11f14ee3ad4b3b9cf138d51e7b2c4d : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1aa2176c7e924b2aa72a3a190f533950 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fae111189714ffaad514cd39f6aaaba : SUCCESS in 23m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ba86d493a4d45f6a9b4b5ede65482f4 : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d235e58612944f58bf13bc5b05b34876 : SUCCESS in 19m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2f84a50ebd3415da471004be683fdf0 : SUCCESS in 17m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4b0575d8d0c14d22a0b4c1e076860e8e : SUCCESS in 20m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6d46c3c959c8479e8c2e7d0ef8ac2437 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9c1fa47c462a4ad4b3de5590ec04a995 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7603679b1b8d4bb7bb480822f7279edb : SUCCESS in 16m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/050b7d9c160640e69d1e0fcd6d905145 : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1d8195e904cc453c96aaddbd18cc00e8 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bfeb1e3e4a554f299397b1e6ef6b767f : SUCCESS in 10m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffa209064e8849f695b40d14ffa16bde : SUCCESS in 50m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a32e2f156ec4568a32905e5f3a48d33 : RETRY_LIMIT in 23m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd933df8316f47fc969183fa5448fc70 : RETRY_LIMIT in 10m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2728dc975e0d451b8a4a3899c7d20514 : SUCCESS in 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a94ec13b67425881f32916b389f054 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d49f87b7b1a4b80a67dfe94a7df8cd6 : SUCCESS in 18m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57794c1bd29b43f6ad02844889f1e1b2 : SUCCESS in 47m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46947c3d44574de1851b22891e34024d : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":16},{"id":"7de50f07bb6fb70694c0bdc2cbbaa8c7cfa418e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 18:55:48.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ebc7ecc449a4caab1c231a512681fe2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee19b40bd2df41e28ac0a1c4811e5223 : TIMED_OUT in 3h 01m 01s","accounts_in_message":[],"_revision_number":16},{"id":"ebfa19cdafc981316881875fa2f92939b6d233d5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-26 16:28:18.000000000","message":"Patch Set 16:\n\n(6 comments)","accounts_in_message":[],"_revision_number":16},{"id":"50c87e2d19199cf73e8821c5764a8fb6a635db4d","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-28 16:18:39.000000000","message":"Patch Set 16:\n\n(6 comments)","accounts_in_message":[],"_revision_number":16},{"id":"05c924bc7c4ccb763e897e1e107180eb1b7ab731","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-28 17:23:21.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"9afadd122e6531b4aecf3dcb2919cbc171083673","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 19:07:33.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ddd826c01724627bbd0f5910639309b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5fb2dcf942de4403a842f2c5427c3b20 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":17},{"id":"9b21ce122a1c0ef8207c281a79a82c43d78d53a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 19:30:00.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ff192daddc64713a36f78b55b21b6b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3533ef55014f26aad607c214668239 : FAILURE in 6m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/55a1958b51df46d8a5de0958bb5d0911 : SUCCESS in 1h 23m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b1556663eb54ba39147d5b3b199e9c2 : SUCCESS in 55m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6a4d64364886416d9b580fb27b51b507 : SUCCESS in 49m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2f5a8d672d94b8f9849255c530d06e6 : SUCCESS in 1h 03m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5fce6e7ec8ae4874b385a101edead757 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd7c7247d78a4692ad53e29fa72d6759 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/862528984ed449e0b55687b2f55bd069 : SUCCESS in 23m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67b18731416341b799fbee1e66b0c916 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa9bf246cfbb438da0f470fef138c5d5 : SUCCESS in 22m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3cfa8d4e3ae148ce9229a38a77d7d05c : SUCCESS in 19m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d4454a3b712e486c87b4ea7dec3edb69 : SUCCESS in 35m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ea8999a922354944b466e430414dc75c : SUCCESS in 19m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/84ebc4d19f4240fd90215adde3014f02 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/af155ebb627643f78ddb5949addaef21 : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf567e5e93d148d982290132ebdd551e : SUCCESS in 17m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e45ee2897498405a90601fa643d3f1ef : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0fb7ae51878142348b73b274eb7af9c2 : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/187d23ff472f452c8a225b76ba7c1eac : SUCCESS in 52m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/624acc7f100a4ab8bd7f04d71d21a0d8 : SUCCESS in 56m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1dd6eb6cfead49aaac3e1c821dbda672 : SUCCESS in 50m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf4554ae5c304b129d68c8bfafd03e32 : SUCCESS in 27m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c11271c76954511b3d143dba55e2893 : SUCCESS in 9m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b2dbc0c9a2824051a6e5986aaa960fbd : SUCCESS in 23m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efe60889966f405f953e1f8ec255e5f7 : SUCCESS in 52m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a76b86536bb14d3da5272a7123844582 : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":17},{"id":"09e61ced8cc882345e63a40c46d56d3e9dd20d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-05 20:58:23.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"917eb2753bf5b621b4971ed26d7ad874b59c1ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 22:04:44.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9ae0d19e2bc44378a579c724087520d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929800f262dd493cba7df8ee08970734 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/6a1aa4f96d554469ab5656b3aad63465 : SUCCESS in 59m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86bc4bd5130f4cc496274acd7690f825 : SUCCESS in 56m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b488e8ddd1cf47188ea2467dc16bacd8 : SUCCESS in 48m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bc646b0c0764a70a23e65aed3154035 : SUCCESS in 43m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24e7afe1442e409bbc12a9c4906666bd : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7abde8517fa54bc3955a0656d2cc7c37 : SUCCESS in 21m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9321eb61c3e54794923ef697ff8b5193 : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd389973dc5a4112b933d19a78ab8f60 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f6988a193d84009ac49628e8f1b327c : SUCCESS in 16m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d086305beadd42b9bc56f3faea4d18ec : SUCCESS in 22m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8c0992de0e054c82a34f572796aac405 : SUCCESS in 21m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a780f251322445cabacde46af692cbf4 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ce22fa903a4411cae37e95b28674bb7 : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/43fb47fcc93244dd9e339c84c937d565 : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4d56e8f3dd94e67b0d4db2425c82e5b : SUCCESS in 8m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a8f46fd41ce49249f7f822faf9e3504 : SUCCESS in 16m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38fd2b42ecd441118c194ce8bd091da0 : SUCCESS in 7m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bf59a23e8de48bd9d5c15c73a1d0f99 : SUCCESS in 41m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9507412cf7964810b9e31722a713fb18 : SUCCESS in 34m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de522197016a4550b4c61a4abf2bb773 : SUCCESS in 49m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d576def75f846daaa03057ec10a7ea6 : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d16d90c856f44a769e2f76dffa3fbf20 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e2ce7fdce074404cb40a5df510a91a65 : SUCCESS in 19m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02195800e249498f8df896b8e0c819f5 : SUCCESS in 35m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c9fbcf9aa854ef1b5e9a0810dafa2bf : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":18},{"id":"f50ae7493e254069c40ae8ba323fd77d8e22d118","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 22:29:36.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffb7d74e501342c18b4957a6053f73e9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/38665d6e3f78488aad2b49e8637dfa91 : SUCCESS in 1h 29m 29s","accounts_in_message":[],"_revision_number":18},{"id":"c18d9096100c33d4b60c2528df7faba203435a4f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-06 18:38:33.000000000","message":"Patch Set 18:\n\n(16 comments)","accounts_in_message":[],"_revision_number":18},{"id":"465975153a2d5e48bcaf0fcd2346f44438a9edcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 11:58:42.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"c08213424b4cec83182a67366e68896d4f5a67cf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 12:39:36.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"09df477f35f14a0fc941fbc804a61f254c2cfe32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 13:12:08.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"8c4aaf4438ac7f36542913ee5de63d00656059c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:29:46.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18e35ff0cceb46119eeb7e14f32df337\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2386ac08c56e4287b5192c4af8b16098 : SUCCESS in 1h 15m 09s","accounts_in_message":[],"_revision_number":20},{"id":"f1738912a82c8ccf410439fe18332c93142417d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:34:00.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d31c779030f472fa38a16f36f64c6f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf51099357ad46e8a79ca629f93b1381 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/6edcdf14fe254eda85b73b0990475107 : SUCCESS in 45m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1531014dce7a494d96f21f09b179df7f : SUCCESS in 1h 03m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dd979011b57f4a77af092409184bd0fd : SUCCESS in 46m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10ae394d62c443f9912a2ac3f11bc371 : SUCCESS in 1h 09m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e23577de2c4404f8c95eb75ef067373 : SUCCESS in 21m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c2e71fc410442dbbf215f0d4f93b909 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0e70e4b3f7e4f98846df4f76745f680 : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0661557004fd4d9cbd97cd73f8a82dbe : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9eac7540560d4170a06037c42ee1946f : SUCCESS in 9m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bfe259852d004ed6bffa6a5ad3587ca8 : SUCCESS in 17m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/15373e29225e474d88dca82dbda68ada : SUCCESS in 19m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/69e4bc2771c747fb8b276972eeff06fc : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/95bce481c5e54b74b64b487b6d408385 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9cce2066483845069fa9ac5ecfc52af3 : SUCCESS in 19m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/943060ecc8f649458e6e760a1d958fe6 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e6f001e8cb94e7783292b69fab9310e : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1b7aaa8678da4cc982609ca7518a2917 : SUCCESS in 5m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7bc1ec4a4804698b99d7520e85a7694 : SUCCESS in 50m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44cbbf9c23be4b1ba7d1a699cda706c0 : SUCCESS in 50m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/211f59f19e544061b6527b57d8576ecd : SUCCESS in 1h 10m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a3320ac741c4e758dbf75eaa9070802 : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db68323b4a694dddb49aa3fd35a6daea : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5415e4629bbe4ebb96e554046ccd1317 : SUCCESS in 22m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c214724a276d44deba275afbe7894f98 : SUCCESS in 1h 01m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71658b36752a46f5ab44acfdee84c0d4 : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":20},{"id":"7ba6da5e7244c7ac095d73a88b59c340feb284e1","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-10 15:23:19.000000000","message":"Patch Set 20:\n\n(16 comments)","accounts_in_message":[],"_revision_number":20},{"id":"bb135078e1bcafb3cc0cd0588952dd3d3187a310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-10 19:58:26.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"49e7d1bba259a79a3b16450f8ee628c13e0f340b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 20:40:25.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dec0673603ba42de8679a7618ae1771f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8cfb59870c3d4c51a0b65ef679488450 : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":21},{"id":"807c6bf97ed5ff157d04bd83c101180885e2fe4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 21:06:28.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5fc47cfa1614697831b362363b08853\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f7dbdb00174047b1709ab8ec6544a0 : SUCCESS in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/8986cec966694988ab73f52ca52eb4bd : SUCCESS in 49m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f2ef06f6a73845dca042802100a17318 : SUCCESS in 29m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5ea2514381e45e787ca21e1cf695a21 : SUCCESS in 41m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d015b0d408ad4a1fa27107f040a34d82 : SUCCESS in 34m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2add1f8aac34c3c883c275f084ecffa : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/821d5d6f20d2437e86e36ca2a902ebef : SUCCESS in 11m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe6e04d63569448c84d1718b864a48e3 : SUCCESS in 10m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cea18274dc7e48cbb97e9fec6d1d4cd6 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e916c8e288924ee2b5c5cf09aa1ee4c0 : SUCCESS in 18m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/855f61d49a964a50bb9a0c5a82a1d94f : SUCCESS in 15m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/34af351099eb4a338c2d86481ffef0f9 : SUCCESS in 12m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9fa954b1066946e18c40b5e3a0e47874 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/28293132025d4e438deaaee09b61dd0a : SUCCESS in 7m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/84846fd1bd714e65a9277783a0c859e3 : SUCCESS in 12m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/792b23000a99407083905057eb46b7f4 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe363d2a6c554cf8b3d14df6f59747d1 : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e470d709098c4826a0dc14ca31385847 : SUCCESS in 8m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc8a5bed8f564181b53429f49b8719b3 : SUCCESS in 33m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/823b5f84dfa146a4bfe5e844ee1f4749 : SUCCESS in 26m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64a3e14e0049410693b55a7f824c2a68 : SUCCESS in 33m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15e916af16cc463fb7b0da6d9c41238e : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f041af8abf4901bffedd362bb1ca88 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/09210f50de5d46f5ace4098fa9038e08 : SUCCESS in 10m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c88348853ab4a98a363e1d23b53deb1 : SUCCESS in 37m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/050197d8fa1e46e385025a958ec51e69 : SUCCESS in 1h 07m 17s","accounts_in_message":[],"_revision_number":21},{"id":"476839049bd894500a66e8c12f900d35f71d671f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-19 20:16:52.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"eeff3c83809505cbc3fe188b0419e53625048209","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:52:16.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4518348c2ea424189b3b01cc9d49568\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/34d47a0ae6c34900b4462268de346655 : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":22},{"id":"f046408361b69066377a98246ec1384f0a1e93ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 21:25:08.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b8ab8a52d94b3b9ef1e963ec7410d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b95c645b7945f1aa284e968524c71d : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c097a6526da644bbba13b4e4d5d07f43 : SUCCESS in 54m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b4bea49ccdd4f68ae0b04c33cd26b6a : SUCCESS in 52m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2384cd9505734d3c809330fb9430fea6 : SUCCESS in 1h 01m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b82ea4245fcb4f4ab77fe21c8426b50d : SUCCESS in 57m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fe779172f3142da9e70728049cdf595 : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13dbd8a4a74441c8a344cd50df23f024 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75bea4cbe57041ca8bb2410e01a857cb : SUCCESS in 18m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aed87dc068d3433e804057db62dd38d5 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c7519e6a799d463893bf38156f5c23b6 : SUCCESS in 16m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/97811bfd6cda4a809b8bf939f816dd35 : SUCCESS in 22m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cdf67719a91c4497b289ae83953c81d1 : SUCCESS in 19m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ff14f60bf08f4863983b3eb5511b7d91 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/98542cbea37944bd8e10e2f20ff68221 : SUCCESS in 11m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0e748e771b1d4eedbeaa3ea4d9dffbcf : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83e2f388661f461b8279bbad28c35667 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6212b401f12047cba72cb691d082eef8 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2b6d3cc750a64fe994c0707c41b2467a : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d5eed9134b74935be7b47d0455c6869 : SUCCESS in 1h 00m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55155dba19a84db78e86d21d34c71467 : SUCCESS in 43m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c73cd4dc59a4cbb92dfea4c7091e5c9 : SUCCESS in 45m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/862e7c45f3354e8aa2d928c25bbb3309 : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07212da9af2b4f0b8391f97f29806780 : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/701a28faf1d340f1a2d2e8846ffb0e2d : SUCCESS in 22m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/512639d836314455a0e32e09f4350ef1 : SUCCESS in 59m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4185715c1ea44f15ab28f4268b71ea48 : SUCCESS in 41m 04s","accounts_in_message":[],"_revision_number":22},{"id":"2cfa6b6492dc3701d6674f7cd64619368d121689","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-23 15:58:25.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/proxy_logging.py\n* test/unit/common/middleware/test_proxy_logging.py\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"8a615f2c3531a2fbbf4a780886a1702bb524fc68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 15:59:18.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 940791,10","accounts_in_message":[],"_revision_number":23},{"id":"6552b884c8c89b31ecdb84ac9c19a24f10ca0393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 15:59:19.000000000","message":"Patch Set 23: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 940791,10","accounts_in_message":[],"_revision_number":23},{"id":"89a4a41a09b2f28f67e5cc33da4ed6033fc9b695","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-28 01:18:35.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"bf02eafd0d8b4ec6c263c2e4fde5b8a9ff3dd82d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 02:15:08.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e85e8a3d9074d2db5c11670904c9732\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3bf9b989737e4395b85b492bdee2faaf : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":24},{"id":"4ced8688da47cd463b031558135796fe98b836eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 02:25:33.000000000","message":"Patch Set 24: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25952b7260154f138d8eb5400e452ec3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9656f8f52e949c79bdb28c16410ee9a : SUCCESS in 5m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/3517ba4845174f7fa8d4a2ed15244e2d : SUCCESS in 59m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eeb4659527f7400c9fcfb46390253d91 : SUCCESS in 29m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6d15975f9eb546edbc39d884ed33b792 : SUCCESS in 48m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b7aa540f6cf42439936720849fa0b42 : SUCCESS in 59m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b948248c66124000b8b04142e513cd44 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e511e7b92b93457db5ada6cae53aab9e : SUCCESS in 11m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e435d98c57f14eea8881790a368248b3 : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5241a700717c42678d27d9fa69c975a1 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8cb75cf6dc42f2889269b8d6614d86 : SUCCESS in 12m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c5e071be97b0420a8b238dd885b82de4 : SUCCESS in 20m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d2a758ae671c48f7824be26b4fc766f6 : SUCCESS in 27m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1b91471841094e44978bfd51c5755a5c : SUCCESS in 10m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/914d407a5c374f0398a68d3782705c30 : SUCCESS in 11m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fa527fa33e474928a973d8ba1f3988ca : SUCCESS in 14m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d39b00bd21a247d4acf2626340d17ddd : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/daf69b515b3046508278afb899096dd2 : SUCCESS in 4m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/14fcaad139f34238af3a5629fa0dd3de : SUCCESS in 11m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4268e203fca4dc5b1cebf234a2f6e3c : SUCCESS in 6m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e95ec5bab1b4144b3ff4690ceb27b3a : SUCCESS in 43m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/35b1416b86fe4945ba1acd5c55e8025b : SUCCESS in 30m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/84193392f3f54456aa5bd0488e74343b : SUCCESS in 17m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5630cffc3c9a4dc098fc182ddb0099e7 : SUCCESS in 20m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b776f8fdda834b91ab6e1963aa1f372c : FAILURE in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a47baf4376054f4a90101a0c96d0cf7d : SUCCESS in 21m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bae33e64e2e34f30b23c593895c7d817 : SUCCESS in 35m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b0ec43a21a4a5e8a93d68455d0775e : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":24},{"id":"88235017037ff36f8385cb97e5808d8a30990f68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-29 13:17:18.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"6a347d3d619fd0efbda506b6a2d4df8d5565345a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:05:29.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ba120fe2e424781b9f20f8e0f73b080\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a1d3fe927b2415a845121513bec8e8c : FAILURE in 46m 01s","accounts_in_message":[],"_revision_number":25},{"id":"298585edf018e20045f2127233ff229b9938929a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:25:11.000000000","message":"Patch Set 25: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/370382aa3a9b42db95ff163cc42e961c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a62a59fb3ce4cb6a8f9cb2608a448aa : SUCCESS in 10m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/66d93018f4fc42708a86ad97c054977d : SUCCESS in 52m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67fa4970a53c42949fbec6fdaae94176 : SUCCESS in 53m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d713efcd0d54e77bf86b8423eca6e5f : SUCCESS in 47m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/022b9e3c8eb94dfcb044c502b692c4e2 : SUCCESS in 1h 02m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e1babfdc8b446699b1e6a36225e4c33 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35387b89f31141ac99a49131d287796c : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89ce9dfaccdd439a9dcd6e0ca8af02c0 : SUCCESS in 23m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/906018f48ad24d1db6cd6cbfda06b175 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ccbd74b03ea64217991b33ac7cdd95c4 : FAILURE in 13m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/705c71b6802f4b8f9ec963633fec552e : FAILURE in 22m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/330e8ebeb9584e989a627306b129da33 : FAILURE in 26m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/932bf69259dc47bc8eb96de60f26e992 : FAILURE in 18m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c5d68a58f34345e6b8c62e98423e219d : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/93fdd4d4942b4eca84e9378061b04fba : SUCCESS in 15m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fe7003b0202a4dbfb85efd3a97373b85 : SUCCESS in 12m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/647c728f9cfb47f4b940f5a2621ed975 : SUCCESS in 5m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/664885d2a01741da86d8ccd2b3387e5c : SUCCESS in 12m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/170c90f325a54e98921fd03a7e6a1e7d : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b45c724d8d294a8597f0439bbda7ca3f : SUCCESS in 44m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ca0f095b54a44e4840781cfb5d2f21b : SUCCESS in 27m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b9843d2f689d4d6a92b9c141ca9c206b : SUCCESS in 45m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7211d145e55c4b0cb5d41e2e8c8fa7d5 : FAILURE in 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05060e1561b42b88e1288f7a41a5206 : FAILURE in 3m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88ee951a953444f582bd4e163522a89e : SUCCESS in 20m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1cac2620f1648379092d7e6ee02ed93 : SUCCESS in 41m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/512b888f7fc147f49aa3bd8211d46572 : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":25},{"id":"cdab240f04647f08e8430bd9a70e694401b8fc0e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-02 13:52:03.000000000","message":"Patch Set 26: Patch Set 25 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/proxy_logging.py\n* test/unit/common/middleware/test_proxy_logging.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"b52b3e31b76626d8d3fcb905783009afa3462205","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 14:05:38.000000000","message":"Patch Set 26:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf44733c144f43988454544d416324e4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22dba2b6fc29462a9ecefeef6c3b1bf8 : FAILURE in 11m 12s","accounts_in_message":[],"_revision_number":26},{"id":"1692ac006c3dd7a4f9892bf5b6386e8e82885a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 15:53:56.000000000","message":"Patch Set 26: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/405f44175e144b298ce2ffbd85cb3c54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/263c953a04d9496098dc2ef8289332b5 : FAILURE in 5m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/063e67a85614453694dc344ffd5f29b1 : FAILURE in 31m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b630f7ab95244ad4990dd5c02d2afcc2 : FAILURE in 32m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/010e9e7403694c688e3408718caa9717 : FAILURE in 14m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b7e4bb68d944756ba0f1d8368049459 : POST_FAILURE in 15m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d3b119ee5cb4028a81de38bdb6a211c : FAILURE in 6m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67a63ed5e7544d2ba3303356b36456c8 : FAILURE in 5m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f149469427a34d74824dff9317d6d980 : FAILURE in 5m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49fdd642abd645dc9842389dc86fca26 : FAILURE in 3m 27s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/741ecddd8b7b4412999cbecf7130e1f3 : FAILURE in 4m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb2db25eff0440b1aebd51166637c66f : FAILURE in 4m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fe909fe4c1424afbaa305cde998349f3 : FAILURE in 7m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/209d0c4a84f74b22ba4b2c107c43de1e : FAILURE in 7m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/08b0ab273a1142e2bf1856ebf5902df8 : FAILURE in 3m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/30ad6f94cc3b48ad9c64ca96b63d9e29 : FAILURE in 4m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7e96daa3e6ad493aa9728c86dfbae8fe : FAILURE in 3m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2f348d316d243ad8e350b676e741873 : FAILURE in 5m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46cb1f689aa24e859628074ebbd9b89f : RETRY_LIMIT in 4m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/86a6c4416f5b45fda03eefe03ba83cd7 : RETRY_LIMIT in 3m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f483d3d93ba74f2b8d0f91029e202250 : TIMED_OUT in 2h 00m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15a2f83f8c9c41bd923fa4699c87f0d2 : RETRY_LIMIT in 7m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/323f06465a5c42c288a6de1b5d761985 : RETRY_LIMIT in 14m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f09385c14c54d988ff52a60403492e5 : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aff1b32a3db3456085075f039a3ca6f3 : FAILURE in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d916865b364a476a974fa48c699b5111 : SUCCESS in 21m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b31d1731bf61452eb915a370eae90863 : FAILURE in 14m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2384f632da74787be3a9a40ed7c5363 : FAILURE in 14m 48s","accounts_in_message":[],"_revision_number":26},{"id":"fa860b049ed6ecfd3b8cbd25575db42147884e55","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-05 12:21:38.000000000","message":"Patch Set 27: Patch Set 26 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"9c641855aa0c4cff055098185644fafeec1443ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 12:37:59.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba4fd5b0cd7f439b8740525177150dfe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/669b565216674c6191dbdd002806515b : FAILURE in 14m 18s","accounts_in_message":[],"_revision_number":27},{"id":"f7d1067b4d94c718798da363370a4998fc3b7992","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-05 14:00:40.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"0ae78a95599fbc6cd6d61989460c328916f22a76","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-05 14:11:08.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"03ea6a1c850eb5adb00d9b2dcf9a478068b1c322","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-05 15:08:44.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"91112e17c1148cb6ec11d359a1eb71e711474039","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-05-05 16:16:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":30},{"id":"3496ab5a20d50373825018f75d0a3d0b7a64b083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 16:57:05.000000000","message":"Patch Set 30: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a0017604b154f2584dcc4baf9d3b426\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7541f3d8a4fb4fc9b7e9974e7e1412ed : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/71483595d6be49648930d0a0b974c531 : SUCCESS in 26m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4cb970c7b3c4fdb927d14ed407e3466 : SUCCESS in 25m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3dc8187e13a0430ca9e5f6375350d406 : SUCCESS in 29m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/685c27e2822542519527fbbf49a8d4df : SUCCESS in 1h 10m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d73dff377cf44389b27527502a11434 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15aec4c205e8412c824e0326950d91bc : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22af21d753ac4e17aefee8cda587beef : SUCCESS in 22m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ef731d5eab14f71adf0c676e69190cd : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51cae7a2fbf64cba973bb139925983b7 : SUCCESS in 23m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd9d00b3763d423eb3a200048b39ee3e : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b9971758d9d74c3aa04d6b900b090685 : SUCCESS in 32m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c9189bddbf464e9fa44e54a1bf8d6f41 : SUCCESS in 26m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2bc8ca1189b24194a587e7adb5a79634 : SUCCESS in 13m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3bb4b75ca6504630a4b563efa0ea685b : SUCCESS in 15m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/52398bae6e8a4964b1c6c842cef82aef : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/217e79bf250045d48b4f3b033453b8f2 : SUCCESS in 6m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3af48f8f64384cf195cc516ef8304a2b : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d5f81bc40dd6455c982c7d585f46e290 : SUCCESS in 9m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8118b497f045498183740ed4ad59ab8a : SUCCESS in 56m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e331738fe9714c19af6689d622f74fe7 : SUCCESS in 38m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4878f91e3edb4352bd7924a248635f30 : SUCCESS in 1h 02m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd5fdff627d24afc826cd85f6d78cc5c : SUCCESS in 18m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb5c448a4b34f28840085012b59ae83 : FAILURE in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d78471f0c6d4a12a07bd562e0db7ee0 : SUCCESS in 21m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/310b788a51344138b52f0abc73d61561 : SUCCESS in 38m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae5e570eb03f4b28814b394710009e71 : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":30},{"id":"5973f813649c6b216d957987c4de0ed68abff858","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 18:51:40.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b90c08bb304f4964857275ac875db43b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f9d555c70bbd4d83b300983cdbacc8ed : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":30},{"id":"1e9beab458ab5653b196981aab7f70b72adca8ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-06 18:59:06.000000000","message":"Patch Set 31: Patch Set 30 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/proxy_logging.py\n* test/unit/common/middleware/test_proxy_logging.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"d33fea1d16f822b57917c6585138c6c67510da67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 19:09:38.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd52ff4d8af47eeafd151118b9e1f34\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7a17afac61214411944f63fe461300c6 : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":31},{"id":"beafd2dfd8441105f5da07962c0884493a115379","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-06 20:38:17.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"2f84932fae87d8d4c8de089b7f282bf66fc62a21","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-06 20:55:31.000000000","message":"Patch Set 33: Patch Set 32 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/s3api/s3request.py\n","accounts_in_message":[],"_revision_number":33},{"id":"731a536f5339849a7f38f68d2a21044540bc1b62","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-06 21:06:19.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"f7e39d4f67193e4e281e3316dfbc6f885cca091b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 21:58:55.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d21244b0f1344b0ba5124af649c600d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0aa5acc03fac4affa35a560b0a8c20c6 : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":34},{"id":"3f4cc29d05b4407b7e1e96edb3550f7f2538d089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 22:15:45.000000000","message":"Patch Set 34: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7263204913494baa9259dac760d13579\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1663dd1a6ab4cbcad37541b80eb2471 : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/2146e3b975464629b505729571ac3e60 : SUCCESS in 44m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ee87d39855340fa9e823c91d7e37714 : SUCCESS in 53m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ed2854ca6fd46e99e26239213e6f644 : SUCCESS in 37m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c0db1cc2db94a6da09e094caf50a1b5 : SUCCESS in 1h 04m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d78e2d1c627a468b82611ea6404405bd : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1d10dae2f0a4c068ae94024a56b6b34 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/539d04d227884ba6b20e63f8e848a5c2 : SUCCESS in 18m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b15880f8fc464b10b498622204b0ea43 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/75a488debcc946358896114631a85014 : SUCCESS in 18m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/977968f8dce547cc8afc043a8e50fbb4 : SUCCESS in 16m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a35b5bf4acaf40dab14d739e298664df : SUCCESS in 28m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0fa0660ee9484d11a117cdea449d826f : SUCCESS in 23m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6855402a4c3c48c1b48c09ea7bff606b : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/26a45d8e343a47648f46aac51132c2f9 : SUCCESS in 14m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/475f0a41208b48a9be5ffe4789a01ad9 : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac618d56eb8d49f3aadd636fe658681a : SUCCESS in 3m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/845c41733cd1413cae6a4b3a2d5cbb3e : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e07eff49cde449cb8025dd9862750e6 : SUCCESS in 3m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/94ca5ec10a7442b59a51a0a62cdacb38 : SUCCESS in 56m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b3db2b2824f4db286e08ea00181f182 : SUCCESS in 17m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54e77f48d9a2470d89d6ecdc7b803715 : SUCCESS in 47m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8102830cd62a458b95a15a31bfa49cdd : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3458b6488d69487a8e2c56e9df5e06ee : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83805caf361540359fb92aa663736bf0 : SUCCESS in 8m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4abe9ad00f740df825d6acd996dd8b3 : SUCCESS in 55m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17f505c441684eebba958a83f26ec06e : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":34},{"id":"16a636e3836148fca12df0b6a18c19c48d795519","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-05-08 14:23:16.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":34},{"id":"2b961fb9ae5653179e8337913d00942717f3bf7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-08 19:56:43.000000000","message":"Patch Set 35: Patch Set 34 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"ea0c65b54a5df757ebbe1085d38358636949d74a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 20:52:21.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a1a6564ce44f3ea000e1f6e653fb72\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/773dd0332ea343d28dd3acb2ea9c58b4 : SUCCESS in 39m 42s","accounts_in_message":[],"_revision_number":35},{"id":"a0d9db18e7cbf3ba537ffe14bd1cdba78d5e0052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 21:05:06.000000000","message":"Patch Set 35: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a064c1294c534ec68e5f58f85268460b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e84c525e04984e94a2fa8e9d1ebaff26 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/e6f9b5560837460a9907f20a91ce30f4 : SUCCESS in 1h 00m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4bf2005d2aea4ddc82ce1dd235f0d586 : SUCCESS in 49m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bdf03c06d5244c35970531bee2b3d148 : SUCCESS in 50m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60df6cbcb0a44c31af590596ac67614d : SUCCESS in 1h 00m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c2192eb7812496c8a7d83301c4c8fa3 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2852e98381c34697a0e8c7c894fbc1ad : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6facc263901545eeb859162515c4b86a : SUCCESS in 18m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b5e1e0b53a164439b1f7f364922fa4c2 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f482535c20b648f68336a3dc3aafac61 : SUCCESS in 10m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8c7c84902ae040d0b025a2aad89e8107 : SUCCESS in 15m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5076efad5d6a446795b10ab9049704b6 : SUCCESS in 18m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f6370dbfc195414ca95d3d029f5621f1 : SUCCESS in 27m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/250ea19e5f914c149a4709d3c893f7a9 : SUCCESS in 13m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c76caec684e643b18390b6efac5a05b1 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d633931a0d7449e6a9c744199031a8c4 : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b66ec25a1ff4218b12995468f4d77d2 : SUCCESS in 6m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1791e834971049b0979d315279c3b4d1 : SUCCESS in 13m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fd2f97377b5344d8bcc33f6ffac0b2a0 : SUCCESS in 6m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/daf7f9ea1259479499154a0e2339a1d5 : SUCCESS in 50m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1cb25816be174a82acca6e86973f8b96 : SUCCESS in 32m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9de900d192b4cca93fabf78a5f944c3 : SUCCESS in 45m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78558ea992e04ba69ac1d0986992e280 : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9739ec28798040648f270d6196bcb2fb : FAILURE in 2m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aaf71d8069ed4425b3889379a5d606ca : SUCCESS in 21m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db14a583e4054c0eaf37b6239e1501c9 : SUCCESS in 40m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/102bb96bc7f64004be108433fab95d97 : SUCCESS in 47m 57s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/utils.py","accounts_in_message":[],"_revision_number":35},{"id":"37dd399e2067b32ae1b03b3bc176aef82675aa56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-13 14:29:28.000000000","message":"Patch Set 36: Patch Set 35 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"1e8c6eeb2cb91d1a0daa12bfe3c969c21fa1e5d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-13 14:36:20.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"f0911cf71c5cc9a15735dd29c759e86744afdb44","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-13 15:33:56.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"89d3d1c8ec956d9ff4765dab08a83b0ddee5fb63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:39:13.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fa988a7986444c9b06f87a35285754d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ab6c52495843e5aec5a2f8daf28594 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/815680687511481794bf70f99ae7044f : SUCCESS in 59m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a48f1aabb2a04e11af6a07a290c6b310 : SUCCESS in 54m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1306045dba14d4286efa3369263e1eb : SUCCESS in 48m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da8887e894694562a8e02198f30f453f : SUCCESS in 54m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bea133c92a324bf7bcbe1606bedf9d58 : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/842a7ed13d7f40338267a6ccda1b1e78 : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e0730a726ff473386e802b75c70c589 : SUCCESS in 23m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eaaf694b51894877908fc4fdd584c053 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43409b03e68945888d00cf4d725d64ea : SUCCESS in 19m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/841d510469c0481aa7173ab5ccd64fa5 : SUCCESS in 16m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f53b191aa4e044b7927d105eecdeed5b : SUCCESS in 24m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d11426c953d5453095a6f3a2b275e206 : SUCCESS in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/68925ea92ddb4de2a4352dfd9532158d : SUCCESS in 18m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c41bb5973c29493699cf98a3e9503131 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/acaddddc597b4cfa8df926f5031681e9 : SUCCESS in 17m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e7ec25a663ed422aa00533a0e8a2b69d : SUCCESS in 5m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a320371f2b84c3a969a443809e03383 : SUCCESS in 13m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ccb27c52182a429eae243d89ef085891 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/43aec41010004162844bae450d4e9d54 : SUCCESS in 58m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eca6eef84e67461da48e8c88693e566e : SUCCESS in 34m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7119f5c673b46e4abcdb89e99b8d015 : SUCCESS in 38m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a2ae621408a49748021c919cb2974f8 : SUCCESS in 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b691f54c48433bbe78d08666b6d5a6 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dee98edeba0b41fbae434eed33e3ec7a : SUCCESS in 23m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5840ac3081cf416683cd06e4cc36d68e : SUCCESS in 46m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33cd3f75f9644135b947437b7f8ffb94 : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":37},{"id":"d1bba9d4cd46f564f8ee499bf3331e7644ac6f0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:53:54.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99801b2bbfab4f1e9014665e31d845f5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f53b387a280649e083ab48e6f6eaed03 : SUCCESS in 1h 14m 26s","accounts_in_message":[],"_revision_number":37},{"id":"9b1e0a569dfc5bb761a9ce9ecf0441224defe3af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-15 03:02:16.000000000","message":"Patch Set 38: Patch Set 37 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"0c620fecfca59f56121ac165632ef5b41964989f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:42:59.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eff9148d72c483a8fe3f1c77fc7737a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/136ea6aec94c401ba470a189d69eb3f9 : SUCCESS in 38m 40s","accounts_in_message":[],"_revision_number":38},{"id":"3503a3668ec6c20bf0aefe7b2849d6b581ff1320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:10:26.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07dd676b5a974c0886a58d3e066d919c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a6f21066604406588bd4876d51d8cae : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/ca9c8883758b468fb0686ac933dcc2c7 : SUCCESS in 43m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f91bf705529542c3830dcf7166d27f90 : SUCCESS in 1h 03m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d0d8fa7f4644bcf828b7a36c465aeb4 : SUCCESS in 47m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4283cf38afdd4887a136e839a1196bff : SUCCESS in 50m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f209fae828a64ad8984097c6806d649a : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b472ff02015841dca22c2ce3fb619f2a : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3594a03afe0541908d46c10172f3dfa6 : SUCCESS in 23m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d3afc40d027d439a9153c8743c16729a : FAILURE in 5m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4cce598244a242fea1daf2280b08d464 : SUCCESS in 15m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6b0a1fb3d694402e9fb7229800904ed7 : SUCCESS in 21m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bb58e0d97d7c426a9fb88189098e234e : SUCCESS in 34m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/51c50af8309142ac89a04a017d86667f : SUCCESS in 30m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dc582b07c78141f19c0ba267c6507954 : SUCCESS in 12m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b31c3a10f3d54ebc859fc0ced9732ed8 : SUCCESS in 9m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fad954f6a8704b8ab32893de0e5e0164 : SUCCESS in 7m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1b610ff406844319081b67fbbd3d83c : SUCCESS in 6m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83397d08c3a14e3daf711faf160b1387 : SUCCESS in 13m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/95432fbc73ec48fc90c586615470ff9d : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/be53770b6c694cffa900d848d45a64bc : SUCCESS in 1h 00m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1a06e764c1147e78e3ac7d60a1d9811 : SUCCESS in 18m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3a0a8106ae5a49ae9d896efd083533f8 : SUCCESS in 49m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e98dc0740434860b740cf67826b5d4e : SUCCESS in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36e359cc447481eb2cfee016f885ab2 : FAILURE in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dcb2a5d9a79a482eae2edb9345498be9 : SUCCESS in 15m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/575a8ca5ca7240d380a7248e230d6ad5 : SUCCESS in 45m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2544efa23a5425d9eddc1f5d5b4295b : SUCCESS in 57m 43s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/s3request.py\n  Comments left for invalid file test/unit/common/middleware/s3api/test_utils.py","accounts_in_message":[],"_revision_number":38},{"id":"572a5dcc9c9a7c25f5e3a88db1a1ab685fa7893a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-19 15:40:26.000000000","message":"Patch Set 38: Code-Review-1\n\n(30 comments)","accounts_in_message":[],"_revision_number":38},{"id":"dcd997e3b245733996c412ebf4596c944db29e9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-20 12:40:37.000000000","message":"Patch Set 39: Patch Set 38 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"788f0dd72382393fdbbf90f1a68e3779be244559","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 13:57:49.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd2d6246873c4b66b25a760ff91f0ad5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0260276c15814bf59c5bd7c966e3ef5b : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":39},{"id":"869fab8de1acfa77f7616402958951d2b4edef60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 14:19:13.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de78540b090d42f786fc04fd572ba574\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0b81191d6fb4a0fbbe1f01e5a042082 : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/9e0d6a41fb9740f6b298ed0ebd59ab23 : SUCCESS in 28m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f318534fdeb4bf98f4b8aad9e11cf23 : SUCCESS in 1h 02m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a43e3372e7e34f02b4dfdce8bcbbb1ed : SUCCESS in 33m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3cffd8d829b46628fcc001e439f7a64 : SUCCESS in 1h 03m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61736630c1e14c77aa8874b4d8778e77 : SUCCESS in 10m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ae3cb445b00475c9d0a7d761568dfdb : SUCCESS in 24m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0dbe576268304e9d8bf8393c290fbe5d : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85101fd0a97e4000aa25b684b3224843 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/801faf33e6334d22aae342b685455b3c : SUCCESS in 18m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1ced281367294f64908f717f1b82324f : SUCCESS in 15m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0586ec6fc1ed46feb3e0920cabdd1d36 : SUCCESS in 21m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a9e428c9ec394d3eac1d05c1df47222e : SUCCESS in 20m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4f91c2e7297a4799837947b297d36448 : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/62a36235f44b4fa88974efee1d7750c2 : SUCCESS in 13m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bfe9206c9f5a4049a589b2764b232b37 : SUCCESS in 8m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd4a03cf8bea476aa6da8944b6d69bbb : SUCCESS in 5m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9b9e8eb557c74189a50c84b11dc2634b : SUCCESS in 9m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9a925b57c23489cbbc40abb9641baf8 : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7392836bf38641a7b83f2f08da5aff0d : SUCCESS in 53m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3205a50d421240c59e617a8cfaacbb73 : SUCCESS in 43m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/238388d24ff4476ea2b6f632a4365042 : SUCCESS in 35m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf04c69cf20f4461884262d4521b4934 : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f5a7dde1d84fc4b32dce4ad7b4301d : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a732f6b68754a538a66ff1417811fac : SUCCESS in 17m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/948e99dbf75f4beab06bf9928d926668 : SUCCESS in 52m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61415ee9c4c64d5e801ae91f7b0917f3 : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":39},{"id":"e4e0a6fc9d7d441739e81faaadc47f012f9afc20","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-20 20:58:18.000000000","message":"Patch Set 39:\n\n(8 comments)","accounts_in_message":[],"_revision_number":39},{"id":"1313ffec141015879c6e116a2e3dd2902eb4311a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-20 21:00:25.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"32ae30a71d34143d937378fbf06b84f19ea34ef7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-21 11:38:08.000000000","message":"Patch Set 40:\n\n(3 comments)","accounts_in_message":[],"_revision_number":40},{"id":"87623fa8c85eb962481585a35f74f0a3fbc6752f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-21 20:45:27.000000000","message":"Patch Set 41: Patch Set 40 was rebased","accounts_in_message":[],"_revision_number":41},{"id":"b0af8699c4f7b7d73227471581caa4d072297ad2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:31:57.000000000","message":"Patch Set 41:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27823cb3ac7c411dbc02cec9969f0912\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f1d27ca5dd1484eb771b307912a7a71 : FAILURE in 44m 37s","accounts_in_message":[],"_revision_number":41},{"id":"104d0a7a57f7eec446c16319191f575320ca7a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 22:01:55.000000000","message":"Patch Set 41: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bce9a85296d42848476c00bba8ab182\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f724133e74bf401c9624f585963f6b89 : SUCCESS in 10m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/6de84cd6c4164634a178b8497e8209ec : SUCCESS in 52m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5bcb7256f2734230a600c82bf8446156 : SUCCESS in 1h 10m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cac0aeba89ca4fd1857c6bdf1f08d8f7 : SUCCESS in 44m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b25789d2f3c4c99850e4a9e2ebeb19d : SUCCESS in 56m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/366de871b025487583120335929fed24 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8e21f18d8c04681a2a1efd303cb6284 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e1ea4376324486c8b0f75b76ab31bdc : SUCCESS in 19m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18abd0cdb5e346129393f67eec96298c : SUCCESS in 4m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff73ea5a2a384df78271679b3c4d333b : FAILURE in 17m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35ed75f73cec43c5b9865f94c642f53a : FAILURE in 8m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/574b48d1a962474d8bf981aa634eeffe : FAILURE in 22m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5f55074fa9ab4b20ae46541a437f7bc8 : FAILURE in 28m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f58fa913a669488c9b95391e0b90d736 : SUCCESS in 13m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ee59b3ed54774d84baa32a343b5eeddc : SUCCESS in 18m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/41fc3848767d44c7a83b71e8a1adecbd : SUCCESS in 19m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5c11f66562f4aa3a82fc55d3dd70a1a : SUCCESS in 6m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/94240675e2a7402080ac9e437a5ee5a3 : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/35cf7f10a177447f94d8b7f9025a9bd7 : SUCCESS in 3m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a95e82be5ae64d6da2ad1defa076de9e : SUCCESS in 52m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9381ac6d50dd4b538726fcf2a5c74ae0 : SUCCESS in 37m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f11fb5f74c134c92b5e1d6e8e6b64c6d : SUCCESS in 40m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3390e140800a4c6cae75248b65945ad9 : FAILURE in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c0d3cf3e4f4012a67efd5648a2045c : FAILURE in 3m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a9615773edb7428eb699efd34ab4584a : SUCCESS in 9m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2937306c7c68479880cadd73627e5099 : SUCCESS in 48m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9b5da1bb98b47e6ad765307566aadef : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":41},{"id":"31e0fb568a380319c15ece5acc47007595b0cbcb","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-28 12:47:46.000000000","message":"Patch Set 41:\n\n(14 comments)","accounts_in_message":[],"_revision_number":41},{"id":"9568f430ad3334ab11fdc928e83b0ca16607b86b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-28 15:21:12.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"647c63912629da7b26772262a54113d71d6a93ce","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-28 16:01:48.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"77a4b82cd52e66a3816afb903eab96aecfa45459","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 16:40:40.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e18bc09195ac425a8ff0a139687e3562\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3b93a184736f41f2b837bb36ed2e39a2 : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":42},{"id":"85d6bec3c65f7f9d87aa7c0dbedae5af774acab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 17:02:17.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32ceb38361f649f898337df6bede1d74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5225bfe3116c4c12be8f14c75a9570d9 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/80dd9a91f63b4608a4b1f4bf67b87cdc : SUCCESS in 49m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/734db99217644d3499b406dba4fa1492 : SUCCESS in 56m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8448e61de7e44b49bc79710fec0b6bc1 : SUCCESS in 52m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7704d338b68a4ed68edad1e3773f079c : SUCCESS in 1h 00m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae15a981855449188baaafb02fb7f5cf : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c263381ccfd4db489d9d0cd84b7fc9c : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06f633e5a66c404cbf337909ce153357 : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0b5c61476fe4e829a28b77a7f67ce4f : SUCCESS in 4m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c455c287b65b400aab4944dfa12ed4d7 : SUCCESS in 25m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb0afb767aa34e51af4f6068bd1275d5 : SUCCESS in 20m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/aafd34d0bf764b3aa93173576c7699cf : SUCCESS in 31m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0dcd594fa6b740caaee14c2d9ae6d2d6 : SUCCESS in 28m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7ac2cb9a0c3544219e5e600de70b23e7 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0a20cf7a8fde4c1fa858f0e2a7b541c7 : SUCCESS in 15m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ea33ab90f39b4708932c8b236c7dd49c : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/26caf3c1871c45ddb70cadcfc6a28091 : SUCCESS in 10m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2933e46e98de432b9ca6d79223addafc : SUCCESS in 13m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2688f9b316c941d4896c111cd48393f0 : SUCCESS in 7m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/03d39a4e61a34f8d8c77ed5c4f352243 : SUCCESS in 1h 02m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aca9ed6469824b7597c9a7db5f59430c : SUCCESS in 38m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c2676e92d1e4b5990069adae423310e : SUCCESS in 38m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58b37ca1311d4ee4985f1a1799b2234b : SUCCESS in 23m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aad3b42cc6e4283b931ba336ebbbc7d : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec7ffbd1648146f9bed6dd8e8c5220dd : SUCCESS in 27m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8363011874564aa4bb539217d0ac2e3f : SUCCESS in 48m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44e3767cb89f4656bb4a9630e483b617 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":42},{"id":"fad47fde9e8413872e9248916c63d06d52e3f4d1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-29 14:04:02.000000000","message":"Patch Set 42:\n\n(3 comments)","accounts_in_message":[],"_revision_number":42},{"id":"b341a2ec64623dd1d758b6b7c37d19846e3bcc8c","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-03 11:34:30.000000000","message":"Patch Set 42:\n\n(8 comments)","accounts_in_message":[],"_revision_number":42},{"id":"2d0a6878d391e32dde30fe587602fb0dd8121bb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-03 11:56:17.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"0d247f75213e4580bbaf4e04b593d429354f2dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 13:36:44.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/023b9be42a0f47c385ba3010437b31e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f5eb2229e14234821aa8ed9b25c8e7 : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/67e816cde4294be5b3aa24dc449d6201 : SUCCESS in 1h 04m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d0bb8a2ffe6f41f7b8e134577fa60b6d : SUCCESS in 58m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75fc964c46a945afba16b305672dec80 : SUCCESS in 54m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bee670e96f6241c787a24f4be9671316 : SUCCESS in 58m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a1113c5eb7345659f39eb0b5de8635a : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95942812362b42239873c21769f15d92 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff30559e31f5480faeab72f31c4e60bb : SUCCESS in 19m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b1f79dd5ee6c4560a173652284c8a7d9 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7128934a38c64a168b17e0457283ab5d : SUCCESS in 20m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2b183b459314409b9acae15260f08244 : SUCCESS in 20m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4f786fb9904f45bbbf92bad59ee3e431 : SUCCESS in 18m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/497b9c10ff5441b7a7df346a7b2985b6 : SUCCESS in 22m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1931b8afc10640538bc5248c9801e2b6 : SUCCESS in 12m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/55a6cbcc22f04f29b47e37880db9e369 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/80ec289b6be34f0b8f569d1d5ee26010 : SUCCESS in 19m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93e26efeafdf4f89a1ec23553e4d311b : SUCCESS in 5m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f1a435e59ab44d38e9c15bc3426b22d : SUCCESS in 16m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b82f5df598a743be8e9b26d1e8506eb8 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b1bdb1b21e9490f8d2356ff6d4aba86 : SUCCESS in 58m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b6b3d71c66e495fb6e2af53074069bf : SUCCESS in 50m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a2e9af833faa4a1293870457a9f7f73a : SUCCESS in 38m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1c4dbe29bc94b5bb91b6a1c30a1ea6f : SUCCESS in 24m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c1e2bac60440d8966c3def3483b82f : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4dc4b24c42a64a9b99828e2590f437c4 : SUCCESS in 24m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0303a9cdc4dd4fa29c4e28bffeeadac8 : SUCCESS in 40m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fef3a2031c4147d7ab8aa2ff7096a169 : SUCCESS in 49m 10s","accounts_in_message":[],"_revision_number":43},{"id":"71f69c29570f9c72793bbc2901ea45f7a5427428","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:01:24.000000000","message":"Patch Set 43:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f65e131736534d5dbae25abc7d710998\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b1305a9a412f4030a95912d4c10d1db2 : TIMED_OUT in 3h 00m 56s","accounts_in_message":[],"_revision_number":43},{"id":"c57ad89fac0b2e65583f5f342cc167aff89d8139","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-04 15:04:37.000000000","message":"Patch Set 43:\n\n(2 comments)","accounts_in_message":[],"_revision_number":43},{"id":"70c348870f7adbeed93f123b1c2347dbaa74bee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-04 15:10:15.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"2752425c717e32c8aa2e03c53013fd4e8723f023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 17:28:24.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80c94b58fac840b99533f372ae1ac8cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa94953ec1d4977a2eadbbb9c052669 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/0c1fefca62a74721948874002d1fbb5b : SUCCESS in 59m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e693b4c5686346098fa621d059abdbaf : SUCCESS in 1h 02m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bae9bbfab6e747868fc4fbe349946808 : SUCCESS in 52m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71274ba506844cf2ac3fa90862595f78 : SUCCESS in 54m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b985e3f865148d38fe5c7bb55c2c2e5 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d4e91f1cfae42ffa036fc9dc17bdef9 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ead0843194554020901a1c58eed9d9da : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69cf2b4d377d481fa2b516b175363b2e : SUCCESS in 4m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7bc952a8847e4a02b6b5dfc64294dc09 : SUCCESS in 22m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1a584b3fbcfd48e18d26c665ca43745a : SUCCESS in 17m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c130c9a4f0664507a883223ba94bbb7f : SUCCESS in 19m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f1d1789e0d174216bf75e562d07d728c : FAILURE in 24m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dfd63df1f0b24ed1bf37981af09c7ce2 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/32afaca529444436a4f2b8ca9ef32c64 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5b0dcfeb566944e197a82729e7feea78 : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c5caaf709ce247fd99110776ea0f9153 : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/15f325ba58f349c68c178cef8a67fc42 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/838ef87b11e947d994885c3f0a3ccc07 : SUCCESS in 8m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2aa61e79273c458497d28f05d1d82f18 : SUCCESS in 52m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e60169d885e542f5b812bdbad826c284 : SUCCESS in 48m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54d6aa1ad56d4f2c97964b13b3f6c92a : SUCCESS in 40m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2c2f1a333c4413ba3a43ae52567d0c1 : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32857658eac54b9ba7b18d61d68475ef : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25dbbbde28fa4d38bf2d7e8b0e86662a : SUCCESS in 21m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e181973ae23342e59d73cb9e2de4be30 : SUCCESS in 45m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/057140572a584bdb8b80df542acc8ecf : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":44},{"id":"1a43829fedd2046b7c12451b92dc4a1f3d342a29","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-04 18:08:30.000000000","message":"Patch Set 43:\n\n(36 comments)","accounts_in_message":[],"_revision_number":43},{"id":"f3ce9d72fdeef6e34b742f3b73a6959f25a2cbe4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-04 18:10:43.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"c915d4c9422e967a8c5282ddc313241e4eea2cb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 20:58:01.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1055b0b92a5a42858661afc7b31a06b4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4ad34cb538b74a6ca4c24d5921b945b8 : SUCCESS in 1h 34m 52s","accounts_in_message":[],"_revision_number":44},{"id":"0365e0865edf899fcf27d47143e0f9b0d59ec422","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-05 11:46:42.000000000","message":"Patch Set 44: Code-Review-1\n\n(32 comments)","accounts_in_message":[],"_revision_number":44},{"id":"24c5d44d6b8c78a50f40cc5981234abb91f0d136","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-12 20:05:52.000000000","message":"Patch Set 44:\n\n(8 comments)","accounts_in_message":[],"_revision_number":44},{"id":"ed794e50a11a2dffa3b209ca547e234631790c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-14 04:09:33.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"882046edbfa151e01a295bbd60d141e128e3c8ff","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-14 04:17:00.000000000","message":"Patch Set 44:\n\n(5 comments)","accounts_in_message":[],"_revision_number":44},{"id":"caa061c5f31025d3e13fe969190d137af4e36c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 05:04:41.000000000","message":"Patch Set 45: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a35283cb6cec4d24ba3e4236e079bfef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b6029b70454ba2bbf3278cbf093eb8 : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/a7a4dc923a694274ad95051925bb8ca2 : SUCCESS in 42m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/110fb79ea922403bae5282c52e0fdf31 : FAILURE in 46m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/172cecf1374543e2ab1d6151e3f43196 : SUCCESS in 38m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/394db8293a374db9bf9389fb6e436790 : SUCCESS in 40m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6583fd0ad35484b8160d2d03c259f11 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37c658e4db02495583d31c71db99c7e6 : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a76a3e48bcfb4d8cb6725ca479f20c0e : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7755a63b77841839ceb93a0cdcef932 : FAILURE in 3m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a738d609518e45148a45c1c32a095dc7 : SUCCESS in 20m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5684830cae9d4a849ab1ff36e64bf07b : SUCCESS in 18m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bc82973bc49d4e6089e97a1e4bfe7bce : SUCCESS in 9m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8a7b397caa4448b290b18bc9054b5df1 : SUCCESS in 28m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/41fdc26d3a704675941c6abd92b8fd37 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8792397b577a4632bb47cf52240d5a63 : SUCCESS in 7m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/05593b5389344feb87ba0dcab97186d2 : SUCCESS in 12m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5bac51f961c547e6b71983b01946b62e : SUCCESS in 5m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf949b21a5504c3b9958fd6a166800f0 : SUCCESS in 12m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fffc411d98204e9082b875b1c5446152 : SUCCESS in 6m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffb11814e0ca4faaac70cbcd609eba83 : SUCCESS in 43m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82cf1c0c0a3147e6b200adeb16365a03 : SUCCESS in 31m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c2f446c778a454ab34d8f3988afa339 : SUCCESS in 32m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/444b0d5f5ad442c48a75e85b376be7a3 : SUCCESS in 15m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a9a6fd461f446ca73b9a81427a6133 : FAILURE in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b7fec47f408445d787d7993ceda12bf9 : SUCCESS in 15m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0335f0e15834984aa0a6a1390bc1ca2 : SUCCESS in 34m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ff09feba15c4d9cb5ad22dc38b54ce7 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":45},{"id":"82a84672e1c1d99fafc8957555c14a08f6c03cc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 05:04:48.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e85e677b8fe417585b4f4b9a193b4a9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d8a0a0c10044338bf3f9bdeaceba53c : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":45},{"id":"9237818e7872731319b58e2ec2b6f7976543e67d","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-25 20:41:54.000000000","message":"Patch Set 45:\n\n(15 comments)","accounts_in_message":[],"_revision_number":45},{"id":"1ee8fcf35e5f5809c83d05dbabde12d6b329d701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-25 20:43:49.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":46},{"id":"ec15f927a515daecb1506ebc08ff5ce09937c37c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 21:41:45.000000000","message":"Patch Set 46: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03b2632d904f4b7da7ad04530270be5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ddb5a050324993b997ad86053f7329 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/9fa39488d45546f0b9e12d0503158725 : SUCCESS in 55m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee699ced86b7458299624b4bd5d33adb : SUCCESS in 48m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/677ab8d427e64710ab9bf6a4842e5ef1 : SUCCESS in 49m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa4239a4cef24827ad84d194eb682578 : SUCCESS in 49m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5fb54e8f0d546c193a800755140c28d : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c281e0cef654d93a7be42927bc37f41 : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8fa384b44b5543689d380746767353c8 : SUCCESS in 21m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87e61b3b5d304ca6b1bc373a699dbe6d : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bed4bfce3de746fe9db552df756599b6 : SUCCESS in 22m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/977808b6e6a04af39b821641f9b73bf1 : SUCCESS in 19m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fbf8a517561a489b8d13be18dd975527 : SUCCESS in 20m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ea4acf7c73854ccb9f9ddde1c77d678b : SUCCESS in 22m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/83ccf3f8d49b4c4cb46004c12bce268f : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ece2ab7b824c4e0aa7ada339ee01e17d : SUCCESS in 15m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b7c00921a9704b3aaac086b204abf457 : SUCCESS in 8m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3980a9aae8e64d4d845f182fbe4fc66d : SUCCESS in 6m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5e850cd9d9a4c158be96694a6cfa09d : SUCCESS in 12m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5b0f18337494ab19f01e21453613fd2 : SUCCESS in 3m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/798ecd41a68f406eb789dae340330630 : SUCCESS in 38m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91f199ac144e42228e6a6c26fb0d2396 : SUCCESS in 21m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a24b5d3d5cf5478cbb16f25aa097c695 : SUCCESS in 36m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00b09523ad7543c298e108fbbf6b3261 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97edaf3f38e640f4a1228ac89738923c : FAILURE in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/af7c924ea9b44470b9f11ce81aa72223 : SUCCESS in 9m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64527d484f7f440eafe03a6f8c4a08e6 : SUCCESS in 39m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bb98b7848e14958870020bebbc1c5c6 : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":46},{"id":"7f899aa6a4f29b755488e43b203299b0920057c4","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-25 21:43:07.000000000","message":"Patch Set 46:\n\n(4 comments)","accounts_in_message":[],"_revision_number":46},{"id":"a45685bda3d2ab61fda1f43318276b1ad26ac3d5","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-25 22:20:53.000000000","message":"Patch Set 46:\n\n(13 comments)","accounts_in_message":[],"_revision_number":46},{"id":"6d9d5055f53292a5ed69ef72dec7f7a32f05228e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-25 22:33:59.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":47},{"id":"9a9a2b236563efd73984418f5910b4df7dd4f6e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 01:14:53.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d3856b331f748b0943697ae94c78f18\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5204b909c1d444c286601d005d7494bb : SUCCESS in 1h 17m 37s","accounts_in_message":[],"_revision_number":47},{"id":"75f0d1c8971403e86c768451b284fc1d2ccb615f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 01:43:57.000000000","message":"Patch Set 47: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/218d321a85ed417d9d4277479c17b8bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f54e05b16844317898a608d380ea599 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/114236f9f868407fa599c73586dba13c : SUCCESS in 55m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5012522dfcc940b99213573f2f8e32cc : TIMED_OUT in 3h 01m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1c9fbddea8bc428382a1e5b21fefbb97 : SUCCESS in 49m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e46b164770494fbda94e325b5a860c0b : SUCCESS in 58m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ab7c2d5a31048f0b15ffe308048889e : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39e764dabf094e6a81e3b86d1517e142 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11e44ff9eb064ac98d37af4df6c93f75 : SUCCESS in 19m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e7da3e1a2c3c4fa88011ea6d5dabafe8 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/978c8621311146898c5dd51331afb4d9 : SUCCESS in 15m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4883a7d3ea764657ae7cf41c77cc457b : SUCCESS in 9m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/50f2361287364b73a05f63a99416c2a0 : SUCCESS in 26m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e64b14c05a0549d588b3f747015dbc1d : SUCCESS in 12m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/51c98eea1a484a87be8be35214214e4c : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0e2d87e92b3b4fafa87a04b202054cdf : SUCCESS in 13m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/371f880b3cf64a72bce7022e45b4f2da : SUCCESS in 9m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/230a83b0cf1f4466b18ac93fec21ee60 : SUCCESS in 6m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e32b57aaf51e4d6e88614bdddcb3da12 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/70cbb23f9d6745188ca064c3369f7f3d : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3fbd3f37b574be2872c290b1c986cea : SUCCESS in 32m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0244552d3ae46879b210f7f14a6bfbf : SUCCESS in 57m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/426abf4b5f6c4124ba6ec32bb99b6ed1 : SUCCESS in 34m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c901d24903b74e0ba9e883c2b144b6c6 : SUCCESS in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a42cb6d0dc41a3b27709098e2b4aae : FAILURE in 2m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e71a00c8ec16409d97b07c5d77763a2a : SUCCESS in 17m 14s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e1baaf255ef416ba5573efd8db2a5f2 : SUCCESS in 51m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50864e3999d840999217e99529a0065e : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":47},{"id":"c6325f932e7ffec038c4ebe1de235cbda134d44f","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-26 16:13:49.000000000","message":"Patch Set 45:\n\n(7 comments)","accounts_in_message":[],"_revision_number":45},{"id":"cd45a581820bc1bee129daf0e9221cbf54e58829","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-27 14:49:57.000000000","message":"Patch Set 47:\n\n(4 comments)","accounts_in_message":[],"_revision_number":47},{"id":"c086f64da9a868fd447c686fcaf99d14dcd834c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-27 14:53:07.000000000","message":"Uploaded patch set 48.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":48},{"id":"4b8b2a8d5dc5960c69f5cfbdcd2e05b666c34558","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-27 15:25:37.000000000","message":"Patch Set 48:\n\n(4 comments)","accounts_in_message":[],"_revision_number":48},{"id":"77dddec74ba550cd0683b09f26d64e09d3626a3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-06-27 15:27:04.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"a8462b9715332984927c9ec8d2a8ef960d5b31e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:47:24.000000000","message":"Patch Set 49: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8db873a3d67244d683146316f15da00a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49dd18988ea546ad8b24fab4ddf7e0fe : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/e55be62246c84ab7879ed2d24e7e2312 : SUCCESS in 29m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a3e75a6d91454c3a8670b8602167d7d5 : SUCCESS in 52m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/be96f30edaaa421a99f7ae8227aaec01 : SUCCESS in 33m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77e5a7c2cfc742bd83f04100a04f8eae : SUCCESS in 48m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01b97a82b06b45ea833b265b00b55e16 : SUCCESS in 10m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ff8cda006ad4e9992c73febd5d012c6 : SUCCESS in 10m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb196d258fa74d39810fda0e14d143f6 : SUCCESS in 19m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/af5a9a865a7f4deeb350fa16f3bd41f8 : SUCCESS in 3m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5c00a951e2c9423ca9739eb0a0a04551 : SUCCESS in 21m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c57ff22e17fe498e8ef232d5d36eb641 : SUCCESS in 10m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e5fa8ec5f08e4db088a2ba5279c00a75 : SUCCESS in 26m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e901833e10e340bba6159bbe8f6ff71f : SUCCESS in 27m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/128d85c3c8d5410e91f3982bba850d26 : SUCCESS in 9m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/51b3627a711d4b0dbfeb02029f39a7f9 : SUCCESS in 15m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/237770846f024bfb93be9c9e5d7fcceb : SUCCESS in 9m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10bd97df0e9a424295ab740e8fc070e2 : SUCCESS in 6m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc5f25a0e4604706bb99ffe805c7f42d : SUCCESS in 10m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd2d58a584874d23b3b293fba7067e05 : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/45050d316232481b96c8d402cd26f5bf : SUCCESS in 39m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f3230c5ba38e4383b3da2c5e5d8e845c : SUCCESS in 52m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31da6301d1e4406e82ef14317584bc56 : SUCCESS in 34m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4a30472b3dc4bff89ffa41374322f7d : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab6cb2d636464f729c7bf4b418dffcaa : FAILURE in 2m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/567b6d0e886d4d0b87b30a9ac8f89d92 : SUCCESS in 17m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d847a15022004cd8ae9e1b8152557e01 : SUCCESS in 35m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dab62988daf494aba4775404b94f018 : SUCCESS in 33m 39s","accounts_in_message":[],"_revision_number":49},{"id":"31f5ddc310147f644173e9eaa6ca40a1b1d01a9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:58:29.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a633a6e56614fc9a5633a94ccef6c48\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23967893edbc4412863361f3dce64595 : SUCCESS in 1h 29m 35s","accounts_in_message":[],"_revision_number":49},{"id":"8d850855ef6a8d8c3488d4b31a3fc82df66487f3","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-01 17:33:41.000000000","message":"Patch Set 49:\n\n(3 comments)","accounts_in_message":[],"_revision_number":49},{"id":"5a6a291f5a616b4ba8f14fea825d14cf68656b81","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-09 16:30:59.000000000","message":"Patch Set 49:\n\n(6 comments)","accounts_in_message":[],"_revision_number":49},{"id":"4285f95ca2779f63518524faeb2bb6c1b1ae649a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-09 19:40:06.000000000","message":"Patch Set 50: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":50},{"id":"8854b98f6a3d43f77c197adee625881b8c19f56d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-09 19:42:53.000000000","message":"Uploaded patch set 51.","accounts_in_message":[],"_revision_number":51},{"id":"0c6560ebbbc2218048d18e3c42932d3d12f8d9e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 20:46:33.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b69677aabd4e63a3049ea764d37d3a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0890d2aa56474cc18500683e8ff68863 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":51},{"id":"2e7a246109fbb2969c6fa2e544ffcb49b8044abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 23:06:15.000000000","message":"Patch Set 51: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54262cec4dfe4dc59f63551d38791c92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f365cbf51c4fc385d868d6915a6a6b : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/7e9b2125e9cb41b2a3cd899494e4e792 : SUCCESS in 58m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41cd33ecec714e348704e09aad5cd9e0 : SUCCESS in 47m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/13215709ea484c84b00e5d261ee7d0e0 : SUCCESS in 57m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/434c374641ed49c68181fb5be2326ec3 : SUCCESS in 57m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/895321310fbd4f41aabf09e8ba6dc2a8 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93e60aa6cf5a4639bed913b6fcfb620f : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07a9e7ac15c84ea5882a4ccc637a6747 : SUCCESS in 17m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e661b0bc0d3b41629c5a6ebd177aab2d : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b1076c16a6d54179b86e926f34d351c0 : SUCCESS in 16m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a766adca2a0b435d934cdc41b55f5419 : SUCCESS in 16m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/02fbbf07177b4da5918e4e2b7f867e3e : SUCCESS in 20m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d999738893b7454cae001deab475e240 : SUCCESS in 21m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5402d993fd644e3685613d25fa86fdb6 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/83b1e4dc383d43cab060b92456afc233 : SUCCESS in 13m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/58f1c348a2764496bc58210cf65742db : SUCCESS in 14m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7cf69ce655e54ae984be56e631552036 : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/233e49eefadc43f7b82b4f810c293ea5 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d46addfba6847628d9abecccf17a624 : SUCCESS in 10m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fdb00f484af4760b4a0a339edccb039 : SUCCESS in 51m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a603b9cf4d54f349de4c5f715308c9c : SUCCESS in 35m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9251ed3f873f4e849f3345bd012b5781 : SUCCESS in 37m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5655b3271534f88860be3000f44d797 : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31f0576077f445c1a1975f9c7f9eb523 : FAILURE in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f167f6b392ba4af39fd34d6fc9a90f15 : TIMED_OUT in 1h 31m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b21e1e5f09ff4abc8e6cce5bc422a75b : SUCCESS in 40m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce8e2a230a9447b6990be08c7f0c01cf : SUCCESS in 42m 44s","accounts_in_message":[],"_revision_number":51},{"id":"c7d75b78058c82c68629366a28418b0f567898a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-11 18:59:24.000000000","message":"Uploaded patch set 52.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":52},{"id":"2f7bc6b3c74f6d4444ed1751d613db9b69c665e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-11 19:41:49.000000000","message":"Uploaded patch set 53.","accounts_in_message":[],"_revision_number":53},{"id":"10b9971a338bf501dac283cf9af2e6055bdfd30c","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-15 16:23:29.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"4afaae044bc746e53603593f09a1672d7421495e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-16 15:48:21.000000000","message":"Patch Set 54: Patch Set 53 was rebased","accounts_in_message":[],"_revision_number":54},{"id":"067b79e7f8c9577b9d5a18c22a68e493f5fe0ba3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 15:48:30.000000000","message":"Patch Set 54:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 940791,23","accounts_in_message":[],"_revision_number":54},{"id":"4af0854abf3cb8fdf606a6f4f4ae859f9bbea25c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 15:48:52.000000000","message":"Patch Set 54: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 940791,23","accounts_in_message":[],"_revision_number":54},{"id":"066d48170cf2a325d9cc387bd5437a5f8d13acef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-16 20:47:47.000000000","message":"Patch Set 55: Patch Set 54 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":55},{"id":"dfeb47c6dcf70affb33b15bedd666be9abd61143","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-16 20:48:39.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"9a6e0468f74d52d6b52d5f04ec36c81010501ac6","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-16 20:48:56.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"2156296f6816d0409c1387a17a2e06e8d3f48576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 22:03:25.000000000","message":"Patch Set 55: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbb1f6ad5c49414a96854b7a7e8aeff3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a00c2f8cabb4909a5e84020499a318a : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/6da8e2af178f4a3db18b365ea8dd8b17 : SUCCESS in 59m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f35b2f0dfa574345b86b5c98695b5e61 : SUCCESS in 1h 00m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49ff0f68cbf84efc8db9741e192d8f28 : SUCCESS in 57m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aef78f796d04420481cf56d372da7a6a : SUCCESS in 56m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6e2274b2d2c49df89a3fe3181895094 : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9e983d973a54d36b97f4996fc51303f : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d56348fd176348acb912cfe87ca1acca : SUCCESS in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1621f91f1ecf4a1a934ff5a11f6e7060 : SUCCESS in 4m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/be5f023d25fa46af9d76f754d187f3e8 : SUCCESS in 16m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d127288f591a42f5a90316e83e207887 : SUCCESS in 18m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e1656260f3fa4b99a29f5829f4608ea2 : SUCCESS in 23m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3fe04d189cab47388daba0225b216a8c : SUCCESS in 38m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/859367bc0666472c8bf5e71bed514b6a : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/879fe6de0fc34339aab0f8cb8419ba4b : SUCCESS in 13m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4756a2cafeb942b596e2bc4deed3bc76 : SUCCESS in 15m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3a27ee08c28c453794908e1e50b674b9 : SUCCESS in 5m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c96cf2c189e4d5990b3bf2e95ea1547 : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c935f4c0c030442db61d08a88b25e8a1 : SUCCESS in 4m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b64d473367149a39fc2f2d423a58e92 : SUCCESS in 52m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76c5d80b05de4687824c3732cab60743 : SUCCESS in 43m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/903922fe48b14158b7819e78cfeb9880 : SUCCESS in 40m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8087903255e044dd8399d173cac83300 : SUCCESS in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51b30de7495c4e33852fc43ff855206d : FAILURE in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5667fda6e2d466b89b7084a61cc7fd7 : SUCCESS in 22m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/231cef7cd5d1419f9097e1c95b343967 : SUCCESS in 49m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa94e5ee53884605a84abc7b3f607d8e : SUCCESS in 44m 38s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/s3request.py","accounts_in_message":[],"_revision_number":55},{"id":"d40bf9fab2409dd1b88b0416860a872b06de7598","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 22:04:57.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f66fadcc9f4829867ad19a914ae92c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/414cb4f1eeaa4e98aadbd426b8668d6e : SUCCESS in 1h 08m 07s","accounts_in_message":[],"_revision_number":55},{"id":"0e8a537bf3d1af448beae0437acb406e61f639a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-17 02:19:21.000000000","message":"Patch Set 56: Patch Set 55 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":56},{"id":"1496088a2225a7b00fe0168be1588579ec42d0b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-17 02:46:52.000000000","message":"Uploaded patch set 57.","accounts_in_message":[],"_revision_number":57},{"id":"9b224bda0033880fc3b64640abbbb148e7f75605","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 04:30:43.000000000","message":"Patch Set 57:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfbabcb647bc4ab7be63a55905d8c3f6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/54782a5c7c834273b363ac08e8e57af7 : FAILURE in 1h 20m 17s","accounts_in_message":[],"_revision_number":57},{"id":"6e95db521ca6e7b603e2decf68ef6b771574c576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 05:51:37.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9420e21cd36f49a096724af0798d15e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8862a835b5764b4eb99bcf726414e38e : SUCCESS in 10m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/4acd452bf9f54a688619e415b2ed0811 : FAILURE in 49m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57871bfcc07a4626b5a11a2c41a6c14b : TIMED_OUT in 3h 02m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9369b16fdd047669f0326cee5102f98 : SUCCESS in 49m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5dbba87bb6440cdb9c6e03c02620906 : SUCCESS in 1h 04m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35d1579994354426be38f797c7a94107 : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbdc7298af29464aac1780888b6245be : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf1e77421a844235b9130e23ac9098be : SUCCESS in 17m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5be187aafda4fb1ab6fcf68d134ce9c : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4364505022fd41ad92fd46a323343894 : FAILURE in 21m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4080a1060f64474bb7d9124e7cea7fc : FAILURE in 17m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e79adeab684e4e63aafd84dad57b4d38 : FAILURE in 26m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7e1693f2be54451a97e51b01329834c3 : FAILURE in 23m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a4f20b44dcdb469e8cf403f910957a77 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/81ae920754ce41e7accd30ef871f9f6b : SUCCESS in 15m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ba823565a5e4458681987eb6f7d0fd89 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d9cac0f82bc47be8fc520812cf7bf2d : SUCCESS in 6m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/11df6297a3a942ed9096437274bbe7c2 : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0e81dc1efbbc4fb49d0a773948bbde6b : SUCCESS in 10m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7bcf1af65d74f01991d1f65295c955d : SUCCESS in 47m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2a2f440805f4bf3b769651c7b020c6c : SUCCESS in 52m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3fa0c4f4008945d3b9305f2c0d3e5c14 : SUCCESS in 51m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a088936454e8477baf367e13b64654b8 : FAILURE in 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf55121a1f33428e8344096732a9da07 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01d609a383d64727bc84113ee3855ba4 : SUCCESS in 19m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5df2bae97a7a4d2c81d2db65675d1244 : SUCCESS in 41m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ee57a2c84f848ec8e46fc03aa416cdf : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":57},{"id":"1432c12767bfa249bb2d1d29e769cd306b817b23","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-17 14:31:02.000000000","message":"Patch Set 57:\n\n(25 comments)","accounts_in_message":[],"_revision_number":57},{"id":"9c175fca54bd4eec0f0d2e996f5cb2bb4fce3870","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-21 21:19:09.000000000","message":"Patch Set 57:\n\n(21 comments)","accounts_in_message":[],"_revision_number":57},{"id":"a39c0feabac6f56fb66a4ceb2f85c8ec54bd4753","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-22 21:26:46.000000000","message":"Patch Set 57:\n\n(2 comments)","accounts_in_message":[],"_revision_number":57},{"id":"0db17671d6943738520718e212eb15e008a094ea","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-22 21:45:11.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"8c2a19bbda3bdcd0e5880e10e26a46764be3ae0a","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-22 22:35:04.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"5dad91ff5d9ef085b0f3b19647985ebdc080845d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-22 22:37:13.000000000","message":"Uploaded patch set 58.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":58},{"id":"63c9fefe70c18f75982d93eddc57f2da52731a23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 23:31:28.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/130033e6dfae45f085d1a7a6a86e7ad2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ad88a7686884d62a4c0c5f7d12c831b : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":58},{"id":"9d003d00122575a32289efc474a709a3565b6423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 23:43:07.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60ec96f6de464685809ebacb2efea400\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5cbe357af104a58a04e598938ce3e29 : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/5a04f3d4a19948849103bd0277a52651 : SUCCESS in 56m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ae7df022e68e41a7ae973b25c530f9e4 : SUCCESS in 46m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d2119bbda33c4b01829011e3ed51a840 : SUCCESS in 32m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea0626ee271240e699d8e2400497a0ad : SUCCESS in 57m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/161ddd428c794c4fbc60bb17b83da3de : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d939fc5be89435880f8e2987dc1480d : SUCCESS in 23m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb4f564597ba483aabd66ababdbcd92e : SUCCESS in 17m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/64ea6f633c7248b3bed8655030e7d69d : SUCCESS in 4m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/42d8293e1cfa4e54bc8b57188edf089f : SUCCESS in 20m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/44d0b87395b54aec8dfe91f4f3f857e7 : SUCCESS in 21m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c121b18b676b4e32b17a11c65b644db0 : SUCCESS in 27m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/00181bdc6b2a4ed4839aef7401ab7a33 : SUCCESS in 28m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d943a0265c94499eb30e4456ec093c4f : SUCCESS in 13m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c42ceeecc95a45cf9869e94a742efcee : SUCCESS in 13m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d7418586e3e04903a13b06f58814b844 : SUCCESS in 13m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9383792d30fe4d3ca147965f1da613ef : SUCCESS in 6m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/96d23d43bb0b465bb401cce58a1050d9 : SUCCESS in 13m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fd33a073e8c14fe8bd63530df1f54fb8 : SUCCESS in 8m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dd04336c6bb4d10a7bd1e82535db6a2 : SUCCESS in 53m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f81d74ec0e1242c5b942f4120287da7f : SUCCESS in 32m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4aa9feffd4db4e51bcbe01f565877e79 : SUCCESS in 37m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fccaafa2a9743b1a7cbedda38288f04 : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c54261640f41688b2000b58783c921 : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46a6b5b7730447d5a982e838641f654a : SUCCESS in 19m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e174c8661a3b40768c15034b4bc0212f : SUCCESS in 46m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7f403bfd0344f71938eaa1247e72d23 : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":58},{"id":"ed52e6f601100413157c728b3379848ae27d2538","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-24 09:38:58.000000000","message":"Patch Set 58: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":58},{"id":"a37da6eb6c70194e35ae4422536f75ee9e681e35","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-24 22:24:06.000000000","message":"Patch Set 58:\n\n(8 comments)","accounts_in_message":[],"_revision_number":58},{"id":"6a5a0956c0eac0af7b027419eb6e467741c2c15c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-24 22:25:37.000000000","message":"Uploaded patch set 59.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":59},{"id":"454029505d4ebdd5495b27e9852a95b95773ef4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 23:11:42.000000000","message":"Patch Set 59:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ec275fc00a34e5186b9c94783309828\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c935f74e713b448588ad2a97f4579cdb : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":59},{"id":"619a8b82a3335c2909c7bab65c13ace609ed1255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 23:27:52.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/608186ddea184eefada5593378ed8f62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe1fa1aae6145cb8fe44c626c7dfb6c : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/364b0b1edbad40d0b346fbc268937356 : SUCCESS in 55m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99ed9722de50404ab805e8599a461b1b : SUCCESS in 44m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89ed00b41d5c4c0f8e9fea4ff2fbb34a : SUCCESS in 48m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f03411f1122541d49d3c8f39ebfe74fc : SUCCESS in 44m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd13cee6535d41b893a01d6c1a91638d : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42c00491190448d287e940cec132c912 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4cbf7de1e2cc4813b05496653d5fb8e8 : SUCCESS in 21m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f8c64f7b8a241399a0d8b8935ddc580 : SUCCESS in 3m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/18cfb742532e48c28decebcbe3fca2bc : SUCCESS in 10m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/98cf2cb992b64835955c1233f584e92b : SUCCESS in 16m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fe25c14d0ec644dd95eadd6168340bcb : SUCCESS in 28m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/84073c428bdb471da84a8f76a5fac08d : SUCCESS in 28m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1d12cdefe9054d16a7c7cb18305264c5 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a9a286a3dd1944cbb4d0e0dd06cc410a : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c77169c0575c4a39baa3b5395e246b25 : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05a3c8df881445898177bf72ecd8b854 : SUCCESS in 7m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0781e873c0e54120bea71cb795729066 : SUCCESS in 13m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1452fc496c9a4bfa832985b2b6f36ae6 : SUCCESS in 10m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea85a94ea7b0453abbfbcbd74d1d63a9 : SUCCESS in 57m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/133faaa75688473fb81e8a3119c997a6 : SUCCESS in 48m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c466d2e2cc114867930b263542449ab3 : SUCCESS in 52m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a426fb037da4b1b87c7e8beb0da6ba3 : SUCCESS in 18m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2cd71733944c008ebb014f08cd8d28 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a095e4f15b644a1bfcd6123d8b8d844 : SUCCESS in 19m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94394719ca2a4837a3cd1eb7b8f5948e : SUCCESS in 48m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8eed4d4e8361421084e1051d8ee51361 : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":59},{"id":"ee1220b30696d45d05e3da2553a5336a50f467d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-25 02:47:28.000000000","message":"Uploaded patch set 60.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":60},{"id":"e216461cd51b52d9e50c004deacf336a8c4e22fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 04:01:11.000000000","message":"Patch Set 60: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78c5fc0971b04875906eff5ed4e0f830\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba8f68f814449538678517747c2a5e0 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/fe22b64ab789462fb1724252573c5222 : SUCCESS in 59m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b04449f18144fd5a619795e73b9b2e2 : SUCCESS in 1h 01m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/27e3bede5d6b40638993e59048ab59a7 : SUCCESS in 32m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77448b3c4f0e4a8284d7619b3d482620 : SUCCESS in 1h 02m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68c76215c92b44af9451df19333bc51f : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74fa7f9002b446db803ec5da7a97c260 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b4fd26dc9094a3ea2fe8ca687b2d527 : SUCCESS in 19m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac080fddd9ed4ac89a2eea5c3175fdd0 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/17a52738616447d18c005d48076092fd : SUCCESS in 21m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/445400a1e9ff4fa69e5a37faf4dd2667 : SUCCESS in 20m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ed39d23de3994b298d05570f0a0e3e63 : SUCCESS in 21m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/80e03886360448c3bb8952f24909cfcc : SUCCESS in 23m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b1e5a89ae278409ead4c736ca742fd6a : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/75cc6baab8f04dc69096a859d2f1fc57 : SUCCESS in 13m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/432bbd6d8d4345e19eee7935eb6cfe60 : SUCCESS in 10m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ed749a3a5c4404499140e64dfdd0fe5 : SUCCESS in 3m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f4abf3062c14786bf257ee5bedf66bd : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/67e1adc064ae4aeab846cfc25760a4a3 : SUCCESS in 10m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/43efb5b10f654b3b9dc3725194a62eab : SUCCESS in 33m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/69c99dc1e80e4ac1ba6c31d3a00a3bf0 : SUCCESS in 50m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/870286583df9445998eeb044df3e80d3 : SUCCESS in 49m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc6d756ff0cb4344a725bad0e8faf527 : SUCCESS in 22m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a2a61272e60417aa0bd3133abe2ac36 : FAILURE in 2m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a04f3ec9c05b4538b2296327dc38cfe7 : SUCCESS in 18m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f346f913eb714930be752b478ba72541 : SUCCESS in 53m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82bb88d27d2044b7ae7952ee0aef8245 : SUCCESS in 44m 55s","accounts_in_message":[],"_revision_number":60},{"id":"7fd87c43c162228a203ef50cdcc77662e546542f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 04:54:22.000000000","message":"Patch Set 60:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5f5e26f224b4af69b48736d6a2e6497\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/39473e0a7d494552a7991d62033a76be : NODE_FAILURE Node(set) request 3839f0a42cc440f5b2111a019262e89c failed in 0s","accounts_in_message":[],"_revision_number":60},{"id":"46125e0af1f69af1dabc959a2622a817da27f1e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-25 14:15:41.000000000","message":"Uploaded patch set 61.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":61},{"id":"68e35953e4d89b6a18ac8f568db7d9ad1d9cef4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 15:25:09.000000000","message":"Patch Set 61: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e5fc24fd58540dca2fbbb9ac76a1e66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49e71ed151b46c7a5b6b4cc1c2bbcf0 : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/97bb3a61ea8641349de46c16e9deb0a2 : SUCCESS in 57m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/919ae86a911f461792808edf94ca6f1f : FAILURE in 1h 04m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0aaceec2332f44df832bbd54a6d067bb : SUCCESS in 49m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbea116de742453db853ccae2ac8750d : SUCCESS in 1h 04m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffc3199fe3a341abb2ce1e92ecd85da4 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bc047104df745c88cb48ebd18280f4d : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46c0e7e925cb4e8abc4220f5d2868bbe : SUCCESS in 21m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0abbdccf7f5415bbd851fbbf7f67849 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/49da2b67d4884cf2b8b3d06045fe1c67 : SUCCESS in 17m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dad61dd4d44a4eb899248299aae03275 : SUCCESS in 18m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/acfd1c0cb0ab4ff7bae0e85d71cc800b : SUCCESS in 24m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/01f0769635f84f14b5fa407042ea3984 : SUCCESS in 34m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/494e3488226d4f5c95f4bb428c5c331a : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aaaebdb6886e4383bb847469fe20950b : SUCCESS in 13m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5b4a7dfe007143da81ac0662dec65901 : SUCCESS in 13m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fc3416eaa21499ebc5fc7fb0cf2b524 : SUCCESS in 5m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d60f3cb4b35435ba36f18735eb42163 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b17123cb4f81477fa83ad286a5bd8fbd : SUCCESS in 7m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67c8a24e4eff4a988edd054b23783e57 : SUCCESS in 56m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76c0cded47ec4c318fcf85e40bdbadd2 : SUCCESS in 37m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7edc7aad6abb4c93a0d15a76a0fb51fc : SUCCESS in 51m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92d262a2ffb74d36a5e389bccc4f89d8 : SUCCESS in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6faa684f4da14ffd9e539d9be31222c9 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2232008f44ac448c955213bf53387648 : SUCCESS in 24m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f16ee7b5c38e49b2830189ef46ebd17e : SUCCESS in 49m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21ff142b0a5f43d39526b30fe57438bd : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":61},{"id":"621f9368f5ff187878728375e366e1292bd2f335","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 16:26:48.000000000","message":"Patch Set 61:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d24c5f37a6404c0eac9e8503d1386a9a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3403292f418440d898597e904a533060 : SUCCESS in 1h 36m 43s","accounts_in_message":[],"_revision_number":61},{"id":"144e64fa36b2a06eef9c238f2ce7f42c529efe6f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-25 17:44:50.000000000","message":"Patch Set 61:\n\n(4 comments)","accounts_in_message":[],"_revision_number":61},{"id":"198b984c59b8a6fd4ba03b46d765c424b2ef8250","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-28 14:00:11.000000000","message":"Patch Set 61:\n\n(6 comments)","accounts_in_message":[],"_revision_number":61},{"id":"5c3cbe9af5998b3faaa3ddf429e35e646af854e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-01 15:48:02.000000000","message":"Uploaded patch set 62.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":62},{"id":"6cd17935952a72154c46b0576c8d1e4c49521cdc","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-01 15:49:03.000000000","message":"Patch Set 62:\n\n(6 comments)","accounts_in_message":[],"_revision_number":62},{"id":"28ee66864f14fc57e21e89c76273b2fe42fad4be","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-01 16:02:27.000000000","message":"Patch Set 62:\n\n(1 comment)","accounts_in_message":[],"_revision_number":62},{"id":"01341d13c1d5b2edecb18f3ac4d9b0e51692bb8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 17:09:30.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d323e347bf447388faec377b119f34e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23551b8ba9d9439f8a516bdb0269c8c0 : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/dab62093904d4e2683eea1787073683a : SUCCESS in 1h 00m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/371e5fa6c76e4750914aaefa3ed60c47 : SUCCESS in 46m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad2a83b2133d46d6bce1bf2db9c6fd90 : SUCCESS in 1h 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/329aba0a0596475dbae54ea485fbeb1a : FAILURE in 1h 01m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e36b79045444a188ffed315c6b119b3 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b1b5218ddf646a0a4ac83f19d30ab38 : SUCCESS in 11m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c59403a436b4c27a31bbb9da205f363 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/913f9256edff4276bb5c62ab26f7d725 : SUCCESS in 4m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b719c8fbe21548788f585df0df754809 : SUCCESS in 21m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9db9fd47e064c56b6acc5e0b9dc0918 : SUCCESS in 19m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7e3063f1528a4154b5783a8dfa877ebd : SUCCESS in 20m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/89dcbbdaef82428095f6684fc4a529e7 : SUCCESS in 23m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/61a70c869f5c4f3d8b332753abb144bf : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2b6cb67a84fa4f7b9a7281688a80e42b : SUCCESS in 18m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/af2b12139ff64cb5ab70441bbcbcef35 : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d21b7c6601b04057a612726298cb523d : SUCCESS in 5m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6f9bc4a6e7664321b5b287945be89367 : SUCCESS in 14m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8383f9c6c2064c549353ab49f33d5a61 : SUCCESS in 9m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8530e069a690483b982216699d1a5356 : FAILURE in 55m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae68fae3e15a4c73b9a71e726d88ae0e : SUCCESS in 36m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7bc08f24406246a3a53ad110f138ed4a : SUCCESS in 37m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a5df7657ff44cd3948cf2027b551de6 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0af9ea81b3fa4ad3b5271110e21c8d78 : SUCCESS in 7m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7052ad72123a480da54b0f0ebe19004e : SUCCESS in 24m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdcc63a7f3ca47159e22d6f70acbb4cd : SUCCESS in 40m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b211b704a7648099bf64b0762ce480a : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":62},{"id":"99f4162636c918fb75e2608b1f38469490144426","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 18:22:28.000000000","message":"Patch Set 62:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a40bfc18621048758e56d38888b3e1f6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eebb43ffbf7e4fd1a9bcefb719456a93 : SUCCESS in 1h 10m 42s","accounts_in_message":[],"_revision_number":62},{"id":"718fe7ff32d084b0bcfa35aab8949d5d353da5f2","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-01 20:30:51.000000000","message":"Patch Set 62:\n\n(1 comment)","accounts_in_message":[],"_revision_number":62},{"id":"16bcd5220429d5c908070afac60f2b6f93b23277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-01 20:33:53.000000000","message":"Uploaded patch set 63.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":63},{"id":"4d412b4e58be26494785848a285c0e4b75a21cd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 21:32:37.000000000","message":"Patch Set 63:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aef520002c4d4160a00a3dfaaed80a3c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc7fddcba39047dd9e2c2ed9d8b529de : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":63},{"id":"a9c8058405b8c8e8356d0dc33f8f506031c196da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 21:44:54.000000000","message":"Patch Set 63: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/717876c78efb45a38af517715f40966f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08a6ef29b4241eaa6af407a473c324a : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/0c3b6fca4b7b40299e1a15e2e1d0711f : SUCCESS in 58m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2efac9e50554bc694563b267f544db2 : SUCCESS in 1h 06m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c39ce7eb253f4ec3938d2a451479c1a3 : SUCCESS in 34m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee372e07d57d4e7e91520371b475cb07 : SUCCESS in 36m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7871d99df88941d2a46e8d570391c6ae : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b093202876a4ab1a5a2b1e1362cda57 : SUCCESS in 12m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31405ba407914c3f9198e4d27645800b : SUCCESS in 12m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3af761c7b6a4b0a906ae80fb3c90862 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/374243d634154dc495ec04fb8a4a71d9 : SUCCESS in 21m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f50d36af66c24898960eb5fd8848fa99 : SUCCESS in 23m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b7c27330739a4f75b45eee1b51f3099b : SUCCESS in 19m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d34b764c626e47c19d48ed2301a41177 : SUCCESS in 20m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ab9472f5c3474bf59b17762892a58728 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2cb5d1be0b2649b3b28bf1e7cf7734f5 : SUCCESS in 13m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a2df778202ec4a36be04997f79c286e0 : SUCCESS in 14m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b60eeba72ae749489a424236845fe5ee : SUCCESS in 6m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3d29dce55bf04c09ba0c9c4b75d51a63 : SUCCESS in 13m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a0e1deff0e6c49eeb149d29616644ec9 : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c39e9a6cf285454ba71a7ece88ac5538 : SUCCESS in 33m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f448f1bd0c8b42afb8a01a86d705eec3 : SUCCESS in 1h 03m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/32eea196413f4123b480f2f8c755ebb5 : SUCCESS in 22m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c694927bd264136a4722b00544a60bc : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f91e867ed04b119f833ef663b6986d : FAILURE in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/984a8e80dd6c49c8a79ab402b154c32d : SUCCESS in 26m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d41dd626da5497484f17be9034302b0 : SUCCESS in 51m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70f92e3e30d64a3796f9a7c52f3304da : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":63},{"id":"131fbcc57b66ee676477aeb1b6bee5785ffcb7d5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-04 09:55:17.000000000","message":"Patch Set 63:\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"203d2e18ce30234f4d9fc616b4fa97ca519a0fdc","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-04 20:35:53.000000000","message":"Patch Set 63:\n\n(2 comments)","accounts_in_message":[],"_revision_number":63},{"id":"c3b1410050cf484f06d37b6cfa39716651282487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-04 20:37:29.000000000","message":"Uploaded patch set 64.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":64},{"id":"2ca83f0c904ff62f7d278e27c88ced5e798fe883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 22:00:53.000000000","message":"Patch Set 64: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb686273f9c0445eaa2335a9a175c5a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e04ab579d8409b908f3ddc04b791e4 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/ffb117c3a956472cac6465122561bfef : SUCCESS in 33m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea91aea909fc4be096a69cfb335257a4 : SUCCESS in 1h 00m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/571f074492f3475985a60e37d0fd4db8 : SUCCESS in 30m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71dbdf33df344bd2a08e394d83daf244 : SUCCESS in 57m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1941f440b12648308286dd79710b4ce2 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bfb28609b3743f1bf556131d04db04e : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0939a4b718248b4ab98cf0179594e73 : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3fafba67902c4c6897fef166726c7775 : SUCCESS in 3m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1992f1fbba3a4665984590fd3bff97ce : SUCCESS in 9m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/87443f2b14654d2f8e891f7b1771626c : SUCCESS in 19m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f6481fdcf1a54a8a9fbe781465199043 : SUCCESS in 10m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/64ad66b35d8149269b03abef2c3eb777 : SUCCESS in 19m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9ccf882d2ba946a69e7f13c389f7216d : SUCCESS in 7m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/326c48dcc36e42d19448e4599f9fdfe5 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76088dbf763a4b31b7755b8b97e1a890 : SUCCESS in 7m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f12dd00de5f47eeaecf452248922b4f : SUCCESS in 6m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5fea6ce54b8843fb84c42a8df2bf9022 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/256b08dda9294131b496a17bfbf7a706 : SUCCESS in 8m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2955a5315d44def83116279c21c02fa : SUCCESS in 46m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1218142613e2435188a8bfa472cba0af : SUCCESS in 21m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30241e1ccac840a5ac976240ec23e62a : SUCCESS in 39m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68bd6ba8576e4eb986dbb77cd9564089 : SUCCESS in 18m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6b9b02718cf42beafc8aa6e594dfbb0 : FAILURE in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9fcd6c296b6407fbeda38d1f155a0c2 : SUCCESS in 8m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/680b4cd72b524b5fad18c06a9417b922 : SUCCESS in 50m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c49bea3954f84b9fb04fb6f0e77127e8 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":64},{"id":"8e33f118fc6ad8a29e0a95fc0ca16501d8be74a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 22:02:01.000000000","message":"Patch Set 64:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fed9e376ed8c49258de0ad14a1bdaad4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba0ae29a34d54f4f874162cdb9fbe7a2 : SUCCESS in 1h 18m 36s","accounts_in_message":[],"_revision_number":64},{"id":"39f2ccece4cf04b118f66a5b48358098d758c9b7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-06 15:02:22.000000000","message":"Patch Set 64:\n\n(6 comments)","accounts_in_message":[],"_revision_number":64},{"id":"5a2be1b55ca7dfdee0efbda8033b4833e0b2c6bc","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-06 20:31:35.000000000","message":"Patch Set 64:\n\n(5 comments)","accounts_in_message":[],"_revision_number":64},{"id":"5c4dc8e5c19f0545238b05d0edac900a849444e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-06 20:54:28.000000000","message":"Uploaded patch set 65.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":65},{"id":"67a93add2bbb015d21c8facfa5c6f72979d47eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 22:00:52.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f65f8e43579c4f5eab99fb61ebd6987f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb20adf76c9d4f7f82939147b7095b66 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/9798a846949140e090216da9659b5bd0 : SUCCESS in 56m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b94a058eefb4683adb9decd349b3d77 : SUCCESS in 40m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dbb49b01db024e87926cb30b8e4a5c3c : SUCCESS in 43m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13bbbdc8681c4f9f8f944e9886768ef1 : SUCCESS in 1h 01m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/027b57ea57574c82a0ed8fe54b52a343 : SUCCESS in 11m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56ffcaf32b9b404a99abd691b4e17900 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef12cff3563c448297201c34dde354cc : SUCCESS in 10m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b72403f1e621415798944033e76c8809 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/230d0e579dcf4ed9863cd8201c3c4217 : SUCCESS in 21m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4785c029ac534d519830acc24554ac2a : SUCCESS in 18m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ab019257c8fa42ed94420517ce355ba5 : SUCCESS in 29m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/072cefc2361b41f38ddca80e429102d1 : SUCCESS in 23m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e4e4583998714cf38b2c3da16edde2b9 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3dcb9fd2d55d40ac9324200ae7b9314b : SUCCESS in 15m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5cd08ece1a7e4dc5a80307a69bd83a2e : SUCCESS in 15m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b8d909f5baa2437fbfdcf893cdfe0857 : SUCCESS in 8m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/647812392581424091c950c755630215 : SUCCESS in 7m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aeb03377d6b4484db838509ba0f2cf99 : SUCCESS in 5m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9938489b4c6c407db93974a5699ef741 : SUCCESS in 32m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bb9c33849a7e4e5ea44cea5e9bdd6d52 : SUCCESS in 21m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a09048adba46427bb877e98468188a14 : SUCCESS in 48m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/849ca4c07e6141d29286ff2ee5dfdffe : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3154b2d7dee2438a9c89394475f329ad : SUCCESS in 2m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14a86617b68e4658a0a52ac0948168a2 : SUCCESS in 18m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/917e439ced7c4544beea7d856511527b : SUCCESS in 32m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a76d09f5602d4c50b87a63370b3a5e31 : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":65},{"id":"7d6fdde4444e1be2cf45523a920ea0cf125037fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 01:48:54.000000000","message":"Patch Set 65:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eeccb13f0b445c684cdf7889b2f6b23\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66526695dfe748179667cb7d0e1d8775 : SUCCESS in 1h 54m 00s","accounts_in_message":[],"_revision_number":65},{"id":"fbfb2a73e3030a8ddeecea098b1d17d6043d82a1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-07 12:33:31.000000000","message":"Patch Set 65: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":65},{"id":"d6a4ea68434d00c67f23c3c16c93fbe87ebac2d4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-07 16:39:59.000000000","message":"Patch Set 65:\n\n(2 comments)","accounts_in_message":[],"_revision_number":65},{"id":"42336a8cd9e9c574e3e1b551220f2baf412a2b0d","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-07 22:46:58.000000000","message":"Patch Set 65:\n\n(6 comments)","accounts_in_message":[],"_revision_number":65},{"id":"f3494c7b208d2afdcb090662a7aa7ea91e84108c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-08 01:34:54.000000000","message":"Uploaded patch set 66.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":66},{"id":"ffbd97ab6caccc8dc5dc156adc61accfbb0750df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 02:46:06.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f59dfc465c39463d8a6e51e77c9b722b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5663f62c30ef460b877fc04d4bb93120 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/639a61bab94d42308acf4ea6131e9a3e : SUCCESS in 57m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2a44ab71fc8d4c7cb340bd36eb627857 : SUCCESS in 1h 05m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d72bf7fa6114f3cbfdc3b9cb151382d : SUCCESS in 42m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/825d307a63d94342a88d02465fa45e63 : SUCCESS in 56m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a288a89d9cb04337b9d1d72ae434a0c3 : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/138e5ce0ce484120aeab24055d04847b : SUCCESS in 12m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32bdc60358234069a03bd3c1b15a1744 : SUCCESS in 10m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7d3a1abe26a41a5927c882a064ee91f : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/822abc756c9c417aaf5a7dff3974c8fe : SUCCESS in 20m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f400f9b30bb2471db13ca5290750e5f2 : SUCCESS in 20m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/62ed311aac2147dabd2d88ba36670e89 : SUCCESS in 27m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c18f1030edc044729003c6854899ffe7 : SUCCESS in 22m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a3f3302d08d94cf99eed5deea674c5b5 : SUCCESS in 11m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/20f7be508d974f44bb03edc7a11cf5ff : SUCCESS in 14m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fa285769b1b045d0b46078ecd266c52d : SUCCESS in 14m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c8fb954789264c0fba54d491bcb40ad6 : SUCCESS in 5m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/deb550503ad141558eb6ba17c013bdea : SUCCESS in 9m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d037ccad41a94ebbb1c1a3f5dbb47644 : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dd3512f4d5f4847b748d9885e3036f6 : SUCCESS in 37m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04d35730ac2e4956bc37c73d7368508d : SUCCESS in 53m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c344d30dec164bf5aecbe0409e8ad13c : SUCCESS in 50m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74f2d8c62ad44df5b0e8ecced1e96459 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/899705df0f084867922dfda5fa828c7d : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74064566f1fe4a2aabc0f3c6afc233dd : SUCCESS in 18m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/673c60b5e1d34641a75b0083864b3a0e : SUCCESS in 33m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33a59bf7e1a54c7aae97675ad14426de : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":66},{"id":"ef77fd114bf7dda24f0ee7173c2d342172abb66d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 03:18:26.000000000","message":"Patch Set 66:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e738a29cc85496d96732a1b28eb09a1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fac1d8b4080c440f81c21b11c0cf13a0 : SUCCESS in 1h 37m 11s","accounts_in_message":[],"_revision_number":66},{"id":"94af435e5e843a4093f95790290a679681a01ea5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-08 17:38:52.000000000","message":"Patch Set 66: Code-Review-1\n\n(16 comments)","accounts_in_message":[],"_revision_number":66},{"id":"4bcbfbcbc06a15be10d54b1253f8bd6a8e3bbcc3","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-08 20:17:53.000000000","message":"Patch Set 66:\n\n(15 comments)","accounts_in_message":[],"_revision_number":66},{"id":"810163012cf0bdd97f747a64ca207e20c1699446","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-08 20:37:31.000000000","message":"Uploaded patch set 67.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":67},{"id":"ed342bcfb58e39e5b8dc8196bb9c142bcb41186b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 21:48:03.000000000","message":"Patch Set 67: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d33bce84208b4a4b85ce431228e2ff23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c2b0aacc4e4b79a016ff944e7f73fb : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/0d9e525433c645ee85d172fb2e2ed170 : SUCCESS in 53m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ac56076d64548ebbcd5dd48f136425f : SUCCESS in 52m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/18cc4bf45df54ace8bd4ff019287dfbb : SUCCESS in 47m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbc7dc8be8e84572ac059d0927fed7f6 : SUCCESS in 51m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbec0d13428f4730bed8f970ad97d6e6 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bbb499364384cc38db1a2ac31b3f1e9 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db0749bf75bb4a6d8ceb4272514e122f : SUCCESS in 17m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d0aa39fb1d4845ab992a2caaf4ffbb8e : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/93792e9b3bb949dd89b3cc9a7e3f4365 : SUCCESS in 19m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d437ab00d1ab40568465812d44cc6f0d : SUCCESS in 17m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/24f7a8fce683469e8813e34e54b1d331 : SUCCESS in 20m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d2eb0fecfda6477697067eba16a376b6 : SUCCESS in 10m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a7b81bd267244fa58db1a7b57f05626b : SUCCESS in 12m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3da62af3f53e4ad68f0f573ce8a92770 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ac1bbe18552e41a1954b2985277d62ab : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/982f194346e84a109c05280b7e96e9a3 : SUCCESS in 8m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f26d48abc314fd49ccad3ef82483435 : SUCCESS in 13m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a9f0c21c2e8400f8a2e8039a8faf6c2 : FAILURE in 11m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b74d82a26971478ba29902970586c6bc : SUCCESS in 39m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29b52af2d1ca420a922723aee8c5d0d7 : SUCCESS in 48m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6bd43d26f6a84ec58c17a2d0a1d1c166 : SUCCESS in 1h 04m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6934f01083c24f5dbd7ce4de09b213d5 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03c9382d195b41d49c6cbc355db68e22 : SUCCESS in 2m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68988d675cee4412a6edbe9ea723d468 : SUCCESS in 28m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9c8d6be2d9747af9b4a8f8a7d013c11 : SUCCESS in 48m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55ef46483df94993a7d76adac3ebf14c : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":67},{"id":"0ccd85fbd57debb5574279126d03975101729931","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 21:59:49.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cb163f28184483f96d46c37ba7fe636\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/00c315bd57c547dc9f45d6a23b8e8caf : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":67},{"id":"a5da64bad6adc61d078a6df64c1eba017f139f03","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-11 02:02:28.000000000","message":"Patch Set 67:\n\n(1 comment)","accounts_in_message":[],"_revision_number":67},{"id":"9d9c3b7692fa39cb1d59f75c848eb6da1637a6d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 03:17:35.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c7f60f8f514c6d9489184dffdf4646\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1c9bab29cc47a8951f46f2e601684f : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/c6ac0ff53a674bce88a867954ad5bf0b : SUCCESS in 45m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a05be63df746463c93ffca0ded4a5d16 : SUCCESS in 40m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1b434ef9993e48aeb7bd8f0f654024d2 : SUCCESS in 51m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3cb06f52fda4af6869e8736c04ba9e5 : SUCCESS in 53m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a23215325b74677b47bf7e7e031cc88 : SUCCESS in 12m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46354c739865448fb337f9e5ecccbe53 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d06ad85e60d347eaafeb281f7d2b7dc4 : SUCCESS in 21m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba73bf586c9a4e58bee783295fc3b704 : SUCCESS in 3m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0d346baaec8439f97e619473a9491d4 : SUCCESS in 19m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b4df050bc69b4f2992327bf182817c15 : SUCCESS in 14m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3b3e5abd457b4219aedd5e732b68a19f : SUCCESS in 29m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9f412b7811904c06b0cc552badf11e00 : SUCCESS in 26m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9149b3c9dd87482a94e6eaad9a5aff75 : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/324606a59fd94385a63774b4b737fbc2 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3bdf08b419cf485b8af28b79c985275f : SUCCESS in 13m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7907309e602d428aa29e77728bd7e7f0 : SUCCESS in 7m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8c20db9757124423aaae50d2ce26858f : SUCCESS in 10m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a748d42667341fca93d9e5edbbb95f9 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4257f11d10ed4e71b09d496317413e6f : SUCCESS in 52m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa54fc61db854680a923cb1cbde5cbb5 : SUCCESS in 49m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d5525d783384911a30faed97cf74b34 : SUCCESS in 47m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e59c37863b4344c1bf190c90a7347132 : SUCCESS in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1650e703a39e4177b1c25189692dbe7c : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e0d07a55d96477586b6441a50fb21e1 : SUCCESS in 19m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b60b71a5c6f4177a74fd3039ef83ef1 : SUCCESS in 44m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89107281e5634846955580dcda591a32 : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":67},{"id":"6bf910f91edf16184d5bd55e31a5b86944179493","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 03:30:40.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb01b6fcc77d4a3199b935ee44b084cc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b640524c9b34ac7a8c744c06f88e4e1 : SUCCESS in 1h 24m 14s","accounts_in_message":[],"_revision_number":67},{"id":"c21f8ce79fd1904a2a8ea1a8f3a5eeedd48c8ff8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-11 13:37:44.000000000","message":"Patch Set 67:\n\n(5 comments)","accounts_in_message":[],"_revision_number":67},{"id":"ed17a4c7ab656b8c1515a91dc93db37242d8d334","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-11 14:17:00.000000000","message":"Patch Set 67:\n\n(1 comment)","accounts_in_message":[],"_revision_number":67},{"id":"61ae00f41d0471a942b50efc4484f8a283c956e2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-20 14:02:28.000000000","message":"Patch Set 67: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":67},{"id":"b13aaec32653c7cb6d0b7c00ede2b860498b0144","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-08-25 06:44:46.000000000","message":"Patch Set 67: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":67},{"id":"ea59bdc7df092684f0fd45d3d9f3f8bd2eceb3f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-25 15:48:19.000000000","message":"Patch Set 67:\n\n(1 comment)","accounts_in_message":[],"_revision_number":67},{"id":"abf5bab7587bcd5660fa493fffa174e2ce662b54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-27 20:33:39.000000000","message":"Patch Set 68: Patch Set 67 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":68},{"id":"41ef765bc394ae091dc21abf19ed666830963746","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 20:33:57.000000000","message":"Patch Set 68:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 930918,68","accounts_in_message":[],"_revision_number":68},{"id":"f2f2bdb3045edf665f48efd7168374b3863e72d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 20:34:02.000000000","message":"Patch Set 68: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 930918,68","accounts_in_message":[],"_revision_number":68},{"id":"37a14d43fdbd6d6681d7e180153e2f52cc8b7404","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-08-29 08:08:08.000000000","message":"Patch Set 68: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":68},{"id":"afe44c6c740249fa71bddf18e82c1f9fba3b20a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-02 19:19:59.000000000","message":"Patch Set 69: Patch Set 68 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":69},{"id":"9522243d807ecefed49b0ab193aa184c53637e14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 19:20:13.000000000","message":"Patch Set 69:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 930918,69","accounts_in_message":[],"_revision_number":69},{"id":"4d1ce6c406747dc7d18a72d4d3053a69bbbde435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 19:20:32.000000000","message":"Patch Set 69: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 930918,69","accounts_in_message":[],"_revision_number":69},{"id":"aeadf4d2aa88ff6c71eb5fa84ed1cf091e0a9e75","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-02 19:57:59.000000000","message":"Patch Set 70: Patch Set 69 was rebased\n\nThe following files contain Git conflicts:\n* etc/proxy-server.conf-sample\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":70},{"id":"4bbb96cb532bddb2a69b9d3cd3dfb944ce2d360f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-09-02 20:19:10.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":70},{"id":"06f7960d403f7ab5bac00a15336ee3290aac8288","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-02 20:38:32.000000000","message":"Patch Set 70:\n\n(1 comment)","accounts_in_message":[],"_revision_number":70},{"id":"58d0e64285f3c60a6af276cc9efbd9c817153a6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 20:58:29.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8ebff4c282741e19289c3f4e0f6c296\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b2f446d607d454fb1d562da59d3aa85 : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/a97d8aa56ec34a54a1850a1cbcef1012 : SUCCESS in 58m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd94796d623943a88d8f97722a6c6b98 : SUCCESS in 33m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ec2bdfa554443cd9af983a77a8bd1c8 : FAILURE in 17m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7d6dcac19b54eac8a7b6fb06462f63b : POST_FAILURE in 22m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc98b5ec4dbd448ab58f9cedbc992605 : FAILURE in 4m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a77f049161a746b5bb122b036be84eb2 : FAILURE in 6m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47a16c95fe8049a09060bb6a4ce54976 : FAILURE in 6m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c8c8d7cf914407e8269f7e42156ca42 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/65fa12bd629d4f38950beceadb70951b : FAILURE in 18m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d7d0cb0304642f3b6615688e3a67631 : FAILURE in 14m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5977d343c94241469028a5163cab37ad : FAILURE in 29m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/16066c950c87415e88ef33f70b05ace9 : FAILURE in 21m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5c0a404ccb744157a247cb418a1b51bf : FAILURE in 4m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1044fa65f9e447e6b52f352a02bc3a9a : FAILURE in 5m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e3ad99bb80d04accba41ccc8cad49177 : FAILURE in 4m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/422e73ba0504484987081a325290317f : SUCCESS in 7m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf7387ae84fe424b866a4c9f46ad1e22 : SUCCESS in 12m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6c1e26e80b0f42a48b7e9b77b7f4e6f0 : SUCCESS in 8m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/99a6004d1b7c4481a7229a6063b03a1e : SUCCESS in 34m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76d757f75f05410aacb22e70e38c6f0b : RETRY_LIMIT in 8m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a22fd52b4fe4266a7eb393902e2d05b : RETRY_LIMIT in 14m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea59f5839c1742c083f3062cdebc877d : FAILURE in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94dc21e8d74e49138e179b9234b54415 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3307b1057dbc4b149385461866f6d90f : SUCCESS in 18m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30cf2ef712d14a4eb2a5c30460093f67 : FAILURE in 20m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a1c7e1f668845db8790e1edc374d78a : FAILURE in 24m 03s","accounts_in_message":[],"_revision_number":70},{"id":"87bed46ce8464d6ca1cb66101b3caaa24803fc1c","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-02 21:39:53.000000000","message":"Patch Set 70:\n\n(1 comment)","accounts_in_message":[],"_revision_number":70},{"id":"d6db863df05f6fb11865c85e4fca523cc1792916","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 21:54:01.000000000","message":"Patch Set 70:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cae6dd5921724e14a15b782b6a85f653\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0977e9cb7a83490798d6a92fd4008248 : FAILURE in 1h 00m 12s","accounts_in_message":[],"_revision_number":70},{"id":"7100f1e19dae4fa829cb0d58eaf7493f534dec0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 22:48:20.000000000","message":"Patch Set 70:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1bf2561059544009a0d3b6f4c9ba608\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab998a4b9d1645b191b536af760ed820 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/7b04d0c3344c44a0adee299f9d88d061 : SUCCESS in 1h 03m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff2ce053422f4ff89f8cdc10170d9245 : SUCCESS in 30m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7bf00b890734edaa56b9857c1b2a5dc : FAILURE in 14m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b0fd1751c624568bffa9f96dae6f17b : POST_FAILURE in 26m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17973d750966427082d192b77caf47a4 : FAILURE in 5m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac25f7c298d14c69bd3a6d2c5733c093 : FAILURE in 8m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9a0ef4a5e2944eb87660fd85da2eee3 : FAILURE in 4m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/76d564a5a75c4c049b85a307984c14c3 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/afd36eb3636e402a878bb8446e3a9139 : FAILURE in 24m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/68929e7dd4d54521837bb821bb3b0c85 : FAILURE in 8m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5e458a884b4c4a7a8950d694757b5a20 : FAILURE in 10m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6868a4c0e97f4eaf9c5258d426af6516 : FAILURE in 31m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f1d9130e28c74ec1b2ab436c1ea24290 : FAILURE in 3m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/78a9c1f08f824ffd9bbcb856bcab0525 : FAILURE in 4m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b5c84cfa8a7a4a0ca1b720e080233265 : FAILURE in 4m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/41f0e25d696447408471151932020428 : SUCCESS in 6m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/255d680550b74108a08183afc99539bb : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7387288552c140cda0c295fb35f971c3 : SUCCESS in 4m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/285b3bacf96642cf9013b45c7e806768 : SUCCESS in 57m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3929650c74e14a14b439da94e3772c05 : RETRY_LIMIT in 16m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ca27907f3e54927abba34043992f0a1 : RETRY_LIMIT in 9m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03bb39cad0da46d49d1a1f74b8def9de : FAILURE in 32m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28d5d7d32614e14aae6cafa432100a9 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/94a3c18431224095a7ac271c108d42eb : SUCCESS in 19m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/753c3bc054504f31a0dc7052065f6561 : FAILURE in 24m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5643ebf6d1d4edab27dd9e883dc8f07 : FAILURE in 19m 26s","accounts_in_message":[],"_revision_number":70},{"id":"df4f4ce15898bcf8fefe4bed4c3f528f447938ba","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 01:41:42.000000000","message":"Patch Set 70:\n\n(1 comment)","accounts_in_message":[],"_revision_number":70},{"id":"6875bbbeb721d5f0d34cc3cea047d95167204086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 02:45:33.000000000","message":"Patch Set 70:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/062582f6200b44bc8cf4c173e51cbc00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3deba1578e77455baf36707c5ced1e52 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/04e03f88882d4a57a29878c73ab27018 : SUCCESS in 1h 01m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c1008c2067443c08ab776a5f72fb067 : FAILURE in 1h 00m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b4c572b49e44b4ea1df8cde1f26074c : FAILURE in 24m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f92279ecd9e84bdb9da05b803bdb111d : POST_FAILURE in 27m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdf63c0c745f469e9afa0c8d22f34ed8 : FAILURE in 4m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03b9ef05a39a465898327aaa1a2c31f9 : FAILURE in 6m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bfaa48bde8448ff84d09feffcf58215 : FAILURE in 4m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6c6d72a96b745119f61cf875814a048 : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/745f9f9c411144b5872cc04a77d9577b : FAILURE in 23m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2f20d165da34c95b3b0943304446f2c : FAILURE in 8m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/617d5afcff1948d992f7bcb46cd8072d : FAILURE in 21m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/10dcc87b6b1b4711a9889cf2382f9276 : FAILURE in 22m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/443903dc144840449c62ef8d7b7fee42 : FAILURE in 3m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ca29749fed57415b84aabf929e6d2ccf : FAILURE in 2m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/20107a13e9984757bc6616dd8405f818 : FAILURE in 5m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3251f41f01b64a2d9bba9d3175c0f286 : SUCCESS in 6m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a303d0d693d14d4e87d73f44803b9a66 : SUCCESS in 14m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/859915330f194b1d88b62d960abe2992 : SUCCESS in 11m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/92cdcb000eba41a6983a6c46f4dcc145 : SUCCESS in 57m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5680e759f07748f6855509d32ffd9c37 : RETRY_LIMIT in 10m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/62a772799c604347b6a8b1c4bb03f7b5 : RETRY_LIMIT in 13m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d32fcbd156342b396468711177a1536 : FAILURE in 26m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174cd67690d543f29d76013ab4119e05 : SUCCESS in 2m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dfb1a3c74bd14caa999608e6c528f36c : SUCCESS in 18m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd7e94a26b7b480394038bf073c91b01 : FAILURE in 26m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b685a0d81464c4593cb5d6e58b59938 : FAILURE in 22m 01s","accounts_in_message":[],"_revision_number":70},{"id":"3bb11c451048731a7614b86502b56db38019e949","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 02:47:00.000000000","message":"Patch Set 70:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd807d054d8b4db6b827ae33a62ed12b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9e6bae4c6886474f9dccbf8ff8ceb4eb : FAILURE in 1h 03m 11s","accounts_in_message":[],"_revision_number":70},{"id":"dd22723e4051ebb7784730a82b60fb9630d3e2ba","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-03 16:14:08.000000000","message":"Patch Set 70: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":70},{"id":"a2ae2394ab6b929e952e2ba1a189726817a8419c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 18:46:49.000000000","message":"Patch Set 71: Published edit on patch set 70.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":71},{"id":"6d745b77fd9800f68f106f15c401aa0c05eee08b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 18:49:11.000000000","message":"Patch Set 72: Published edit on patch set 71.","accounts_in_message":[],"_revision_number":72},{"id":"c508fad486daafb491a681d4a06784779fc36512","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 18:49:18.000000000","message":"Patch Set 72:\n\n(1 comment)","accounts_in_message":[],"_revision_number":72},{"id":"5afbe5f80e500e63c1f1ffcf22caa6c803b0d01b","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 18:50:02.000000000","message":"Patch Set 72:\n\n(1 comment)","accounts_in_message":[],"_revision_number":72},{"id":"504f1547325aa77e6e44f5dea300a57dda4ba0e4","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 19:00:05.000000000","message":"Patch Set 72:\n\n(6 comments)","accounts_in_message":[],"_revision_number":72},{"id":"2c3723b2c9b1ec93eb1b0687a9b90ab275ef67e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 19:01:42.000000000","message":"Uploaded patch set 73.","accounts_in_message":[],"_revision_number":73},{"id":"06de57e4e278bc2766218473e42874f611d63f8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 19:28:26.000000000","message":"Patch Set 74: Patch Set 73 was rebased","accounts_in_message":[],"_revision_number":74},{"id":"cf7c6e298905660fd8de85fad8aceb076d0284d5","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 19:28:42.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"9cd1f28fb1057467eb692a51834a50df90f25db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:34:26.000000000","message":"Patch Set 74: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/723f9277aa334ca492b7cf76214c1f2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b8dd9372274ff28541f34820f2068f : SUCCESS in 5m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/f1f93d2849a949a39ce337f8b4ad0869 : SUCCESS in 26m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f7079f982474213ac011b2259fa7fad : SUCCESS in 41m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7e9c804479e46d89ac067ed9e92e03b : SUCCESS in 45m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/893aa59611ba4066a4c65e15a291ff5d : SUCCESS in 1h 03m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8435937410c4e54a5118ec904aafec7 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2c8023c5b6147fca8c5a7cf11f2130b : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee1524bc9de94058a151378fcea69000 : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2a4c8bd293c4f6d8c9d7ebbe2790a2a : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/71518dfe6cd347899e9e139c4d1538ea : SUCCESS in 13m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6e0a9c4fa918453ea17e75f7829e52a6 : SUCCESS in 22m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/93bf843835c74acf970a1fe926a03994 : SUCCESS in 14m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c5160ec691824bcabc0b5cb2c5ea2e76 : SUCCESS in 11m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/260c444f03c94407bbd7c812589b39e3 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eb75ae8707ce4246b072c3abdc48acbf : SUCCESS in 17m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fdc6f9ba7f8448e5a5ad8a456acdf3e4 : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ed6b6702c5e2499ca571212f1d06e8c9 : SUCCESS in 6m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/53bd3c8e592b428a8798af72119c7c28 : SUCCESS in 15m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/17c82ba885a04d5594a1c57362800794 : SUCCESS in 5m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5194155816d54f3896da3ffe890cb9cb : SUCCESS in 57m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2643f45a47b4038ac389b07ac4f12b0 : SUCCESS in 39m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1824c5d79ddf47a7925da7272dabf651 : SUCCESS in 49m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a00b91f7f9a94988aefb77fed14aaa5e : SUCCESS in 22m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83f9468266914a06bb53367a528fd23f : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55249014c2304f029e797565379c865a : SUCCESS in 9m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b33a60dfe2a4a579f7265488785d0bb : SUCCESS in 29m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0746dee57bfe4b26822c1a34d24d67e2 : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":74},{"id":"1b531a8fc6d5aa5bdac5fcb5b5e213e806bf11d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 20:45:03.000000000","message":"Uploaded patch set 75.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":75},{"id":"48d792f9cecc76402d4e76367a5338fab8aaf4de","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 20:46:15.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"3ee3bb13f641c7fc043d8bf79eb3652b31449c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 21:52:41.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8189f563be6549a6b58ed5911a8a384e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8e8d6175934ac19d50dc995cc5486a : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/e8ad269a2080481aac16845b06ce724e : SUCCESS in 31m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/144170c39e364b6eba9754eae67833d2 : SUCCESS in 1h 02m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fce7d507c8a3428f96c6a7a881291428 : SUCCESS in 35m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c443423a1aab4a3485525d3fbbb692e9 : SUCCESS in 43m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f00fb68de5c4af9baf1d3678dcd6f9e : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5ee278c5eee4f8999309d6346500cc8 : SUCCESS in 26m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abd043056b6e47109b54885212a9b0af : SUCCESS in 18m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d0478e6c14c94af0b30608bbc1f0c25e : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6087a071c0a14ce48ce2246a37dfeca1 : SUCCESS in 18m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1ac6dc20b7a9420e88c6150e015ba885 : SUCCESS in 20m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4956d51c5fed417cb85926191b7db588 : SUCCESS in 31m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c5dd8d5adba74afc8493bdcf8cbb5f40 : SUCCESS in 28m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/09211455fbfa41f28647dc1f3f560fd7 : SUCCESS in 9m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6253ae4fb5b14344b5cac45545c3619a : SUCCESS in 12m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fd3944c04d374a0c899ea96be73cf888 : SUCCESS in 21m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0c22bb58a4e42f6bc6a186e6ae24552 : SUCCESS in 6m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dbcb7b0230f34f7890c316a21582893b : SUCCESS in 13m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2afbb87065304bdd9929233a5ee9a340 : SUCCESS in 5m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/326f9db83fce4627b6c1bbb417d1d431 : SUCCESS in 54m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ecfb60a7394f4ade8e485cb7b8731454 : SUCCESS in 21m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d4b3f6787eb4cc9b9503b2e50365187 : SUCCESS in 28m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/343d1a33113f46c5aab8200c76e5d0ea : SUCCESS in 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21d3372bb00b43458eb24f5ff870504b : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92eb70944c1140aaa9543b51ab83fc01 : SUCCESS in 18m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/882d64ea53d240efb8400e46d44a478d : SUCCESS in 53m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5423e6fb25c46a6b3bce044dc2ea674 : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":75},{"id":"222399c48dbff02a7cbb954990b471f7b79e0aa9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 22:12:34.000000000","message":"Patch Set 75:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fc5195cfb904061a7f1d4c4985d7539\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/750352e2229a4dc38a2546336333f83b : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":75},{"id":"89eae2049bfa57ea342ec9e9cd4e9019aeb6abdc","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-09-08 16:59:48.000000000","message":"Patch Set 75: Code-Review+1","accounts_in_message":[],"_revision_number":75},{"id":"f719d84280a8c3c93378a84805a52e9720b34b4e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-24 22:04:31.000000000","message":"Patch Set 75:\n\n(7 comments)","accounts_in_message":[],"_revision_number":75},{"id":"05aee98297ec68fdd1ed87d7f659544dc3a299e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-02 19:45:01.000000000","message":"Patch Set 76: Patch Set 75 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":76},{"id":"c0548152e616ffd6171598cf914629aa5dad7bcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 20:53:33.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc29d49efd234fb2b67a41db5ef51960\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad3b829a5de44cd9a5be64145bcf2f7 : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f0a2be4602484d1ba138200cf468f724 : SUCCESS in 57m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ccf0b0db40344b369e4a63a1ac2f8f99 : SUCCESS in 58m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2414d0e1202149f1a4048d67bd181823 : SUCCESS in 53m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9ba81a3378c41278335a05436cdbbc2 : SUCCESS in 1h 01m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/478fd1844fe342c3a21e74120c3a365f : SUCCESS in 24m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6cdb35171a7404aac670702c830780f : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21e306a03d0b41179515037fad5f1164 : SUCCESS in 20m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43d6b6eff13746cb86eff6312eb72ee8 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f5850b0ceeae45aeb8095d91948d63b7 : SUCCESS in 19m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3cae6baebfa142feb0df610afe146e29 : SUCCESS in 19m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/01bf826b068249ec85695e88303526d3 : SUCCESS in 28m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c7b2147767b0464c957128d6e3484e2b : SUCCESS in 23m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/71754ed5f4084736a9ab062fb319f14f : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cd870aa8b6004128b35f8d875163b2c6 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c10c6aee0e104bf1b9b41b6ea4897bee : SUCCESS in 8m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1238a26bc0ca48dd9a645a2dd035f9ad : SUCCESS in 6m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a176382959aa49b3a18936337b70fe5e : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bbac7036844b4bfa8684bb773353f58f : SUCCESS in 4m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cba286d68810462d8b78dc94d94b4c8a : SUCCESS in 58m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3edfb9e0b7324a508da7a38b2ed3916a : SUCCESS in 37m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8a3a7a30496544fe9d465536c7a033f4 : SUCCESS in 34m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e97c19bb9674c1e91c88f6e58a4a8ca : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b1c9880d234a45a8dc9af99d4a9a31 : SUCCESS in 2m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3dde71265d84b63b676fb25dc2539b7 : SUCCESS in 24m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4263c2c394d9467aa7a99f93bf0722c3 : SUCCESS in 35m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8a5d6ba4c4c4b6dbeee20d3c95bf91a : SUCCESS in 43m 52s","accounts_in_message":[],"_revision_number":76},{"id":"820133932ca9acbd4c3d9272a74f719a7ad868cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:45:06.000000000","message":"Patch Set 76:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b6ddcbd30b349fc89aa5f246acd3ef9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40860461a50f460b94ba6b6693cf55b8 : SUCCESS in 1h 15m 53s","accounts_in_message":[],"_revision_number":76},{"id":"79e53f5d982a5df59c1b5b89ec7abbb7901bfdf1","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-07 15:36:59.000000000","message":"Patch Set 76:\n\n(4 comments)","accounts_in_message":[],"_revision_number":76},{"id":"c509fe3a4a8d2eace59e9eae36036f7a6e2b3693","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-07 15:40:14.000000000","message":"Uploaded patch set 77.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":77},{"id":"d64a997e706ad41d93ecbed4a1aa08a42ea3a29d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 16:20:26.000000000","message":"Patch Set 77:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/262217c0b336406282760bb5d8c1403a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e81d7cc4db0c48c4a5d9ea298c807982 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":77},{"id":"2579ae56587180a6efe3f8876eede1fd3cf98d4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 16:40:56.000000000","message":"Patch Set 77: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5b6f4d3a37549e2a829eaed9fa3e25a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5834f1982252497a8c1216fc926b49d7 : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/1d1856b5383f49bdaceb5cff0582de15 : SUCCESS in 33m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4db8728cb2ef49df9e6ce21f869ebe62 : SUCCESS in 37m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0cb65bb2cf2b4e728673e1a23a1378b7 : SUCCESS in 50m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47920b03e6ba44568c884870b2253470 : SUCCESS in 37m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9ccf8d9f9374ce0a49a96038313941d : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/299c177a30184124a6a177d8675c90f7 : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffb9d71826844e5f8756b712a6ae5adf : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be2e7ee5a58d4570808013bd83bc436e : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/509e908992634e8189d5c2eb693569ff : SUCCESS in 28m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bea44d8c8f9d44188cd803271ffc8613 : SUCCESS in 20m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a75d01546954f0b9ba469f52018b501 : SUCCESS in 10m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2b72857e73f442a2b8fab6a776887c93 : SUCCESS in 13m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/28efdf198ce3471ebd0cb8e016233c99 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/412f12a693f747b79ca8cf9c7115f4c1 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/30dbe0da52364825b7fb0e156c61876c : SUCCESS in 15m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c204a068e90b4190806ddf6ff2637fa7 : SUCCESS in 6m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/56e12b5c605d4e56b4c4d6fc1c3aab02 : SUCCESS in 15m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/760ad410a1ff4ac79021193993564500 : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8aec4aefdb9409885ec119d4834fa54 : SUCCESS in 54m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91c7bd923e5c408bbe947ab525d2f0b8 : SUCCESS in 50m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9000bbebe9c14a57b18300ccf7e1faba : SUCCESS in 47m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f83287759504696849185d759d981d9 : SUCCESS in 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd7dc7fc33645079606a05880811a8f : FAILURE in 2m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/99c1fc017e4643a1b02ed303cfc96a5a : SUCCESS in 18m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c50431974e434c6d8bcc67e42ee37db7 : SUCCESS in 48m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf0373d8d3844b458679432b1c120e42 : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":77},{"id":"7fccf0956d3bcafc6964ef526936f2fbf7fe77ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-07 17:44:33.000000000","message":"Patch Set 78: Patch Set 77 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":78},{"id":"5802449013ed7ad06366fde5e3388532d08a7997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:04:05.000000000","message":"Patch Set 78: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53798a736b6340e4abc52c017b447da6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe17b7cd1f74c7896bf9d9dded4cdff : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/7b251afef8424061ba1bac27a478d1ad : SUCCESS in 1h 11m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/39258eb7fa5441ca97f0534b5f5acd87 : SUCCESS in 1h 04m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1603f43ae09f48f6bf2a25e8bbf187a2 : SUCCESS in 54m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b796e8ac1a7348f2a24ea5ca8c82918d : SUCCESS in 48m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26edef0f72cf4f29979704be3fd02dad : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e000866006554a4a8687e983878e221b : SUCCESS in 13m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f95e6ee250f84f0ab91ce6727a90f0c7 : SUCCESS in 20m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae49cedab32a46e3b7f31895d2b4f9e5 : FAILURE in 4m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eec2f41c6a3e43f2b946e33e38007a5b : SUCCESS in 19m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/42e652df31284dc2baf831e4f98eb504 : SUCCESS in 15m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d36f893df1ff466ca21d263279ce5410 : SUCCESS in 19m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/53fafcb7401b46d1990b49b746374662 : SUCCESS in 25m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/13daab318f6346fcba71c1e8a2745c37 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cac1ae4b25ea4635a286f47ebbb5db2a : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fd150ec234ed4124a6b34e9a852c1d82 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8dbe8fc8631046709bab7a349478daa7 : SUCCESS in 5m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5ee847dec1e483b97a95ea7b0731b47 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/17b14c2866cf4ce49bacb8a42fd80fa4 : SUCCESS in 8m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0efe6883b7164a90b3325e2300e02918 : SUCCESS in 58m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/347805eb907a437a83bc8dc2d261f9a7 : SUCCESS in 49m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd47ccd5e705454fb7c3be864672026d : SUCCESS in 37m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/884de38ffb39464d8bc463de144b95c7 : SUCCESS in 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e79f2122f446436994722a291a9a9030 : FAILURE in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7053b4d37d1846bc8f7ade99dabf2800 : SUCCESS in 19m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6b814c0f6fd49a590a4bbef4c1dacf9 : SUCCESS in 47m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4124234b96645c89af0c5e98fb73265 : SUCCESS in 1h 02m 54s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/utils.py","accounts_in_message":[],"_revision_number":78},{"id":"bf760056ee5b2a4b219930ca5fbc28e89bd59a7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:17:42.000000000","message":"Patch Set 78:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8af7de36d38499eafd40c8f68093420\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/efb2d3a95e20437a814f6e6c942ca3db : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":78},{"id":"f27f3a4d33da023c98a6f1f9c25ad3e7722bf4e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-08 20:14:34.000000000","message":"Patch Set 79: Patch Set 78 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":79},{"id":"9e2f528c10cd1d41cb1c167cae8601ff4beaf822","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-08 21:08:25.000000000","message":"Uploaded patch set 80.","accounts_in_message":[],"_revision_number":80},{"id":"8b3c3c2174c1f853a1096f055ddd7b0c2bee19af","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-08 21:15:06.000000000","message":"Patch Set 80:\n\n(1 comment)","accounts_in_message":[],"_revision_number":80},{"id":"3e66e12c271f6a8d33c2142bb01991c52217dd2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 22:23:40.000000000","message":"Patch Set 80: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d23edc4fdab4a3fa2510934f0879a10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0762e3220347bbb5e90c891cf9175b : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/636a2980d10e4abaaa6051669353c9ee : SUCCESS in 32m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df122233a9a54954b015a5324aa9595f : SUCCESS in 1h 04m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a76bed14b6ac4516a79afe7cb0c4f8fc : SUCCESS in 38m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11d155de18fc4ef3b2f2ef509e005993 : SUCCESS in 46m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ff300e1805c4dd9b5019ca2ae0b17d5 : SUCCESS in 29m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/856c9ba32db54927b52a4772459d5a5d : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1656cc28a39940b29f5712fb30218a21 : SUCCESS in 12m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a74cf95cdee641bfadf7ae0490255834 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e2b704e5f3f440282803560bd5cfcd8 : SUCCESS in 25m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab418db1b74540f3ae4af33b6d9b7b47 : SUCCESS in 17m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/87006398fc7145ecb81799020d489f0d : SUCCESS in 35m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4f769d328eb547be94f19bdf0c480ad6 : SUCCESS in 20m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/52ff762c3af44670be06c2ff2235edcd : SUCCESS in 9m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/646ea24f87294428a5f719fb677d0a12 : SUCCESS in 15m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4cc7f5352913445689576aaac249689f : SUCCESS in 9m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0784860224204272bbd854cbdbee7bfa : SUCCESS in 6m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8d389625deec43cd94531fbe51b06b2d : SUCCESS in 13m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7215bfdedeac4532b20d5c36b286e935 : SUCCESS in 5m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8be9dcca1d19446895246e9b0938b288 : SUCCESS in 34m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c739cf4cd784d56860379c65155397f : SUCCESS in 22m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b1c00dabd56948ff9c27af2e06fabc99 : SUCCESS in 52m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba899736e5224ab796d527bc00b724ce : SUCCESS in 27m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4516a034dafb400bb13f8861a1989fff : SUCCESS in 2m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef6198be00ea461b8d8b2b3f93e7b496 : SUCCESS in 17m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff17210c0e7f4f358f9fc6db11450109 : SUCCESS in 50m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98a355d3b5fd41ea9219074de14702c1 : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":80},{"id":"aeedf055c1653a360278c1ca554aa435f12a3240","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 00:33:53.000000000","message":"Patch Set 80:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e78924bfa2d43d1a22d0519ec9da19d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c22b8eff741f4dab95725b6aace5ef30 : SUCCESS in 1h 18m 28s","accounts_in_message":[],"_revision_number":80},{"id":"2e1de4e1078d779a18a1fd69377c7628c63b884e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-09 11:46:15.000000000","message":"Patch Set 80:\n\n(3 comments)","accounts_in_message":[],"_revision_number":80},{"id":"64391880f56abe6f5493ccf88bbbdb439576fd6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-10 18:26:51.000000000","message":"Patch Set 81: Patch Set 80 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":81},{"id":"bb73cfefbee08032ad27edd00b6f5c0e9da82f79","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-10 18:27:29.000000000","message":"Patch Set 81:\n\n(5 comments)","accounts_in_message":[],"_revision_number":81},{"id":"363a24163f7adbae2f2f731d924d345af043ad4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-10 18:30:25.000000000","message":"Uploaded patch set 82.","accounts_in_message":[],"_revision_number":82},{"id":"0b9da612c9ca4ec6f95d7739d22ccf23472e9e7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 18:59:05.000000000","message":"Patch Set 82:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdb5e642eae74fd786ba1d2c43bf173f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff8cf1c281a546bcbd8eadbc65e2d93b : SUCCESS in 26m 47s","accounts_in_message":[],"_revision_number":82},{"id":"d3b1702dc86293afc4182fb9ca329a5c0b570cf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 19:34:37.000000000","message":"Patch Set 82: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24fbdfcf9cde49ba9b502cf4d5b1b3da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fd449c83314527a0418e82a2f2cce1 : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/37e4a63e9fac477ab3b7aab0e3ebe9b1 : SUCCESS in 1h 01m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d50aa0682c9749debb03303cc1465cc8 : SUCCESS in 1h 02m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e32fb15d5b72488d8e1acd2cb74b2147 : SUCCESS in 54m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1905614819ae4a20935b6e6100fa6aca : SUCCESS in 1h 00m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d6831804bc045a7ad6151a9833fcd0f : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b87bac698ee44783b9d641cc5b97de9f : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b33a522543646f4a5dea6ed3f5416de : SUCCESS in 12m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b2d639714e874c4695a43b437372e573 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/61229ea5e0ec4aebbbd350a5d592e796 : SUCCESS in 26m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31fc54ea5be345f098afa0ddf15e0ae7 : SUCCESS in 21m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5f3641fb796b42fd8c639f5fdbc2112d : SUCCESS in 29m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b966a448f15f47de8ce04891c9e3cb77 : SUCCESS in 12m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7431dbc8c9904aa2a99879f8a341ed7b : SUCCESS in 7m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6c285c91bb9b49de999a1298e276bf5f : SUCCESS in 14m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/053fc19ab85d4c649028b16160266974 : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9c868d3c7ec8445fa694f0c24ef56494 : SUCCESS in 6m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b31e0d51a1c43109c2f80c9a8fdc7ca : SUCCESS in 7m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7620901f62ae4f38be82bfa5e69f3012 : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/89b6bff2a65b498aa94ebf9cb44cee73 : SUCCESS in 57m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c6a8bb6464a4039a3a2bc12a1836793 : SUCCESS in 53m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1abfefd6c8464f269a2bbcbbe37e3769 : SUCCESS in 22m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf74fa748b6541e2906ef218a46efc99 : SUCCESS in 34m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/461ebded472c44ec9400e135e8f8f1f2 : FAILURE in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c478a85ffd6746599cb2a760af86c313 : SUCCESS in 18m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc34a2fb8e654494a84f9974eff6aa23 : SUCCESS in 54m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f5e3178b3cb468f8e36199312344428 : SUCCESS in 1h 02m 29s\n\nWarning:\n  Comments left for invalid file test/unit/common/middleware/s3api/test_utils.py","accounts_in_message":[],"_revision_number":82},{"id":"fabd5c45dcfbd0c60cb01778dc302eba80e2e074","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-13 14:25:43.000000000","message":"Patch Set 83: Patch Set 82 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":83},{"id":"9be5ba021b8d8eef99fd40aa7b0d9e4e388590bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 14:45:50.000000000","message":"Patch Set 83:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a70ed0877883455198b9a670dd8a54a5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63e079814a464b8ba059f8274a5abe61 : FAILURE in 13m 01s","accounts_in_message":[],"_revision_number":83},{"id":"87d2da81fc751e14b53a0866fcf40dc812e15cd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-13 15:29:29.000000000","message":"Patch Set 84: Patch Set 83 was rebased","accounts_in_message":[],"_revision_number":84},{"id":"76643d87c6df6f4e4763013d011e0eb171b67d10","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-13 15:54:33.000000000","message":"Patch Set 84: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":84},{"id":"2ba8c51180f945f208f94b8d71df17cad0b907dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 16:15:51.000000000","message":"Patch Set 84:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69c7abfadff4ea1a742a62518ec0033\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a897ffc88fb748f9ae401214f271832f : FAILURE in 17m 58s","accounts_in_message":[],"_revision_number":84},{"id":"bf6fcd6af598f6f2171b61e9ee5ae1889d99cfc0","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-13 16:29:50.000000000","message":"Patch Set 84:\n\n(4 comments)","accounts_in_message":[],"_revision_number":84},{"id":"d11ea8f08f0e895fdff91a64f34f98bbdb8e5137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-13 16:38:28.000000000","message":"Uploaded patch set 85.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":85},{"id":"ae3cfeb983e350d48a0a2849a43fb4445136bbb2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-13 16:43:36.000000000","message":"Patch Set 85: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":85},{"id":"2784a2e03b3080d781213d906c50992e47691cdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 17:41:59.000000000","message":"Patch Set 85:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16c82a7cad642a5903e7fcadaa1399f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23a980d7f31c4075902e96c86714b3c9 : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":85},{"id":"797878cbac88d30f19d22349c140115484fb8eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 17:51:58.000000000","message":"Patch Set 85: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f0ef47514584c3f8003b3dda02c6dbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fdad204b2a2482ca38edd58c7cfe9e2 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/88c246ef13b84e2692850dcff21d8d7a : SUCCESS in 56m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/365e0eabeb934c90a4615363a85487ce : SUCCESS in 1h 01m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a421c533df454211bf3910bd2ad314e6 : SUCCESS in 58m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55a486d158b64a5892e41571dace324b : SUCCESS in 48m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31a72780769844f3afb73a430d722a3c : SUCCESS in 21m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/549df27e7aff4280a18012e363cf638d : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd1336ea5d874b26b8a89a3c88fcfa4a : SUCCESS in 11m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e400c8a87c2431894845d75e2603354 : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/54de28c9bf1b44a096b9179c3f8eaf8c : SUCCESS in 21m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f005a9ae9434f599dde28250edf6a4e : SUCCESS in 10m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f34c4972c1ee440fa8b3785bf1650528 : SUCCESS in 19m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ab6f528a5e0549ada033f3b3b20dd58d : SUCCESS in 10m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/34751f41aa764fffbec83928594b78b6 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/03a6cc5a3e724f259587f26419e6bb35 : SUCCESS in 12m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f3c39c4517434935bfae1641bf6357f5 : SUCCESS in 9m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f09511306774aac8e23b4ce73eab1f4 : SUCCESS in 6m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ef706736d5346df85be581a6061cc5f : SUCCESS in 8m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c9601418db8f4d17a8bc5ed0163feb68 : SUCCESS in 8m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a92994bad5d843d4b2638b91143752ae : SUCCESS in 56m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/359ca0d3375a48efb94ccbb41d79b485 : SUCCESS in 55m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c6798b5af3aa4d97bcfb17f37793b603 : SUCCESS in 21m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96a00c3ce0fc4ec4a322fbf8d6ad3ee0 : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05170d72416148ddaeb51aea29b23228 : SUCCESS in 3m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a962b46c008e4a649844721e514efc45 : SUCCESS in 16m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f906a7101f774dc8851c0a0c919e3ec8 : SUCCESS in 50m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50a4be0a18f146d8830d0bc04000e705 : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":85},{"id":"9c32b7f9afbe7985eb4b1204107317370889ac2b","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-10-14 13:55:33.000000000","message":"Patch Set 85: Code-Review+1","accounts_in_message":[],"_revision_number":85},{"id":"4ee5721ac4059eae9271b68be96b40876d97e95e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-14 13:58:19.000000000","message":"Patch Set 86: Patch Set 85 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":86},{"id":"03bc071e083416125df77e02059efc1bf5dc0d75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 15:07:55.000000000","message":"Patch Set 86: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1166043e48a24bd09cf7eb7e55f08500\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b1bd74b91642079a1b036e9145574b : SUCCESS in 5m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/281983e889b34bfe9a64fc7f118019b1 : SUCCESS in 1h 03m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dab841a34ce845fba02eb2eb07321a34 : SUCCESS in 51m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdb0186e4a3e4f9daf560e2260589cb2 : SUCCESS in 53m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d35c8b3e44e478aaa59d1663dbae4fe : SUCCESS in 49m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f49d3caba3f4e0e99cc920dd03b07d2 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31557adc0a874180b102e09e965b43cf : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3d7b32de6fd42438f6cb5e6734ba1bc : SUCCESS in 17m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ba3e4a44d9440138a4abe02364fc63e : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/89d43caf4e2148d6988915076e76fb08 : SUCCESS in 25m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2324f51a3a0469d8c512208b01536f5 : SUCCESS in 9m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ea3036172b0c4feab442462cdba1342c : SUCCESS in 28m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/511c7088997146aa8db8899185cee4e4 : SUCCESS in 28m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a41b3c01c9544722bcb602e4ed0381d6 : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4759fd65a2394275aca83f66b31dc69a : SUCCESS in 14m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/304e81cea04d44b49568102fc5c943e8 : SUCCESS in 8m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee60862400f0412bb7664b684ae0bb1f : SUCCESS in 6m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/075fe8be700447acb70684e1023516d8 : SUCCESS in 7m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/65e0154de8ec458c9497e3cfcff505aa : SUCCESS in 5m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/24fc4492d26c49bb864b958ce57a9066 : SUCCESS in 34m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/787661c9901b443199670416f8783811 : SUCCESS in 54m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4b8cf6b93984889bb257286d3e495f4 : SUCCESS in 22m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d537bd8fcb04e13907952a327d12fd0 : SUCCESS in 28m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a375ad7f8049678a116d24155f824e : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7c6558810b5452c86dcfb797dfa89ed : SUCCESS in 19m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c6869005f2b4122a90d44d26835e92d : SUCCESS in 39m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73c6534bc7044f4d8437a70503397c2a : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":86},{"id":"22bd338c8e952e3485cdde1dd056aeda31c35523","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 16:17:34.000000000","message":"Patch Set 86:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6de47c49dfc7450880ae4bccdb492e78\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/09f0b0f15c2249feacabf35ee175cd4b : SUCCESS in 1h 18m 35s","accounts_in_message":[],"_revision_number":86},{"id":"9a60c6b09c949ba1aebc3f4ccb0189d2ae644a8b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-28 08:26:30.000000000","message":"Patch Set 86: Code-Review-1\n\n(2 comments)\n\nThis is basically ready to go. Just think there is a bit of debug code still in a test. Once that one line is removed this is good to go! Nice one everyone!","accounts_in_message":[],"_revision_number":86},{"id":"851483495da99afc66fa59549c3dbe4822ebc57c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-28 08:30:10.000000000","message":"Patch Set 86:\n\n(1 comment)","accounts_in_message":[],"_revision_number":86},{"id":"7b2531c7f28d0fb2ec9be24e13cb165840e1ade0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-28 08:31:04.000000000","message":"Patch Set 86:\n\n(1 comment)","accounts_in_message":[],"_revision_number":86},{"id":"f1ff6fe03b84a9459586a4e497449e251685f176","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-28 10:48:24.000000000","message":"Patch Set 86: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":86},{"id":"d4114294b47d51df3029c6fc487fc881318e724f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-10-28 21:23:19.000000000","message":"Uploaded patch set 87.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":87},{"id":"28e694101f4c34119d225ed95d41eeda3a786d68","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-10-28 21:29:27.000000000","message":"Patch Set 87:\n\n(3 comments)","accounts_in_message":[],"_revision_number":87},{"id":"a26c965f93aee214728b5afc9ae796fa039809f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:14:45.000000000","message":"Patch Set 87:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85e2a18ee6ca4d31b22d7c0f6c56aa18\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57414417d8dc40bf9e0132e58fe31cc7 : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":87},{"id":"7839ca277fac8ba18ddb80aef1320827e0aa5b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:37:25.000000000","message":"Patch Set 87: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/578f74dd1cf44bacb016d8f644463be7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab45fb31bffb49709a29ffe54c2e6614 : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/122036d0ce2441e58c684bbfda0c2b64 : SUCCESS in 28m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a5eb6cbed4a4999a80b0199f4bc9f48 : SUCCESS in 28m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5ea949eca0245df92516b831c84ad85 : SUCCESS in 51m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5ea5cab5835435fb8c742832c374ab0 : SUCCESS in 51m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80a563c4f36a4ded96499330e83b5cc2 : SUCCESS in 10m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9204ac12cc9d449991c9b2c57d9029c7 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27bd6c296f7244778368e4311630d5f3 : SUCCESS in 17m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a30975b1541f4f5e836cdb29545f761d : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c5bf62648f3e4698aba7af6fa5cea81c : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d5aec2d3f914cab8f932c8e02af72cf : SUCCESS in 23m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4ce0881e87a04e68b68ac70ec4f7941d : SUCCESS in 21m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cfe923945a654f8cb57585de47cf07df : SUCCESS in 20m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/98a61c8d510440c59b4949973d8cda1f : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b8f3422e16af491ead1bd0e0391268bc : SUCCESS in 15m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d19f9a9e742a4646aafceb67b54247e3 : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d27d464736e4ea88526ff6e0f01997c : SUCCESS in 6m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1205ff7a73984f5987ecac07ea8914f8 : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f07ff046a6cd4b5395220db4fc6ff0d2 : SUCCESS in 10m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8813f0343574c559278c8ddb2d2cdc1 : SUCCESS in 1h 02m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af28782974674079b85e8d2c2ef283ef : SUCCESS in 50m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d9e5e66a0034d05804b3301f52185ba : SUCCESS in 22m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07d0b12e75f74bc9ab2885f9e1bd19ac : SUCCESS in 27m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5e6b493ba740c19c613c0e1391f60a : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a12ac1ea05d4d5d906727d64f9a7a5e : SUCCESS in 18m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cd588d9b9cc47268661e2f51fc1a917 : SUCCESS in 51m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b9649120f94470e8b6c646f6620b49a : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":87},{"id":"90e6c6e5d07b8a95f517189f46dbb253f3b2ceae","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-29 16:11:35.000000000","message":"Patch Set 87: Code-Review+1","accounts_in_message":[],"_revision_number":87},{"id":"e99ee6c186ad046b85c27f34e827f6d22079c4ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-10-30 19:31:21.000000000","message":"Uploaded patch set 88.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":88},{"id":"8a44bbdb057de66bd23bb282dac760ad8789d16a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:07:48.000000000","message":"Patch Set 88:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45849b9350124a31acffd1757e294628\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a58d2492b234470a1d3506d7216ad35 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":88},{"id":"aba77e2d22f7a29bac4e839d4e12c5d722f37e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:41:28.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f293bbbd09cb4084b86f9d179d3f0e8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4484d5c73c2403994bf9d267417a52e : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c5ac71a034e94e65baa7155fd65b24b0 : SUCCESS in 34m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/60a8b867e2464feb913667e9312365d9 : SUCCESS in 1h 00m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53168414bf5044b68661249fc0d6edc8 : SUCCESS in 51m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dfa185b9da04571bb66035b211ac9e7 : SUCCESS in 1h 04m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dcdd6f0ac9764a7e897cc2202c4403fa : SUCCESS in 12m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9751451c234046c8b7b2798478f75a3d : SUCCESS in 13m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/330eec4675ba40799d0ab256820e1f99 : SUCCESS in 18m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d125ef690f54d95a1056e0755b059cd : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8907752e8f3e4063bdc1594c792f87b9 : SUCCESS in 12m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/352d4eca499945b3bec4b2a347317a03 : SUCCESS in 10m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8c2adcdf110a4b569b6600c9c33a3e67 : SUCCESS in 13m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3902c638ca7b4ab685c9f8c40de48827 : SUCCESS in 36m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e129ed72721c409b8d7a429393120267 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6f8ad9c6a8494303801f95f2d14c1d67 : SUCCESS in 15m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/67a0b84bfd6d45dcb5580a21d795f97e : SUCCESS in 8m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c68c01c36d84c379a571015fc4af894 : SUCCESS in 4m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/baed1e5930cc4bcfa5c7d42580de7e00 : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e94b60d2ade84c96be8916479ea863e8 : SUCCESS in 8m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f093a349a0714071bffacfae43bc7ac8 : SUCCESS in 51m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/101f817faf7f4dd8ac71ea0171730b01 : SUCCESS in 55m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/09040e72751647ed9fb2337df55e31bf : SUCCESS in 55m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/334983a245d948fe8eea314a6cd28e08 : SUCCESS in 27m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d804bbcea50048cc81515690c2f1fd15 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/988dcfb7f30442ba94c3667b9ad714ef : SUCCESS in 19m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30d6c71205a4474f9cbfce0c25fde68a : SUCCESS in 40m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1023fcd2bdfc4700aba6d633becdac16 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":88},{"id":"9aaab042ff299a3e5f4f5cf495e39f84cb15147f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-31 10:08:54.000000000","message":"Patch Set 88: Code-Review+1","accounts_in_message":[],"_revision_number":88},{"id":"fd08a6b9075fc5125bb65152f0ac612e74508e53","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-03 04:33:51.000000000","message":"Patch Set 88: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":88},{"id":"73d42a085665d616d834d261fa025107e7c2ab73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 04:34:17.000000000","message":"Patch Set 88: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":88},{"id":"e4b846a73660174c3db524894d0b637193545569","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 05:43:32.000000000","message":"Patch Set 88: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/687437fc265e4ba1a2c08b1ff9eb42ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ffdffd9695944b2a9960072953d0d0b : SUCCESS in 7m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/f06a861147de4811a75eac52bf3c8cd9 : SUCCESS in 29m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29ba87732b914142a7aa0b554d503fdb : SUCCESS in 39m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae9d032239fa4640a7d5c7cce8a019fa : SUCCESS in 54m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bef8e92a24545c09afd2c6c5bdd83e2 : SUCCESS in 58m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/53a40680d0e5414c993ee787ace3e6dc : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7a9196642da44a3ab8234b7159a1d295 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/664fd01489f0457189bee7a39542e2ad : SUCCESS in 9m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f921eb1651c24391b2ba1b1a71203ee8 : SUCCESS in 27m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2e6cf5c73d9e48f4a270a3d4780dff38 : SUCCESS in 32m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a2a1761137234728843158a261353c38 : SUCCESS in 9m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cc23d11b54274e218d102d28c33de9ce : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/75912dbfe33843f3a5139a93925d07c0 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/035eb390d16144889b3c44d90c0d10ff : SUCCESS in 4m 33s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b6670e1778e9409db75d4b8b483af1b3 : SUCCESS in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c583815662f46288326171e2d5d6cef : SUCCESS in 54m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa2c913839fb45008667b034af28543e : SUCCESS in 38m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/52e5b5e5f68e40a38b8d6e1ac3d6bcb5 : SUCCESS in 50m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1699ccc68fb447f9a64eefd5cbcd6d1 : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4257f95cbb4c3e9b252d318134ee5c : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d41fceaeb6c04136805a4de1ebfb3f28 : SUCCESS in 10m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/461befd9f8304546abb2b0cd03a3e5d8 : SUCCESS in 1h 04m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b4497e03e2d45d7b5a78288e1b78a51 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":88},{"id":"8df5c490e0a8a43e4d19b8dd5b3ef305020e464a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 05:43:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":88},{"id":"062855e986140c0df7be31a35dd77b5db1d9743c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 05:44:25.000000000","message":"Patch Set 88:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b33793938334f4db87b5bbcc747e99c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce677aa770dd4a62982d853ae14e68b8 : SUCCESS in 38s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/79396bce550347929c93408c47e5ae7e : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":88},{"id":"6806342e03965f5ec7301292ef7c3f255797b395","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-11-03 16:23:39.000000000","message":"Patch Set 88:\n\n(1 comment)","accounts_in_message":[],"_revision_number":88},{"id":"8ae5e38e20ef7737af2a6b2864c877fa68c5a26e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-03 16:59:31.000000000","message":"Patch Set 88:\n\n(1 comment)","accounts_in_message":[],"_revision_number":88}],"current_revision_number":88,"current_revision":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","revisions":{"8943b92269f32cfa016254e858c0ffd26b3aafda":{"kind":"REWORK","_number":1,"created":"2024-09-30 13:59:57.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/1"}}},"commit":{"parents":[{"commit":"6fefda865f61bef3e117b89bd65d1d31bcbf676b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fefda865f61bef3e117b89bd65d1d31bcbf676b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-30 13:36:52.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8943b92269f32cfa016254e858c0ffd26b3aafda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8943b92269f32cfa016254e858c0ffd26b3aafda"}]},"branch":"refs/heads/master"},"93971c11c36809e4f15deb33704f4929c44ec3ab":{"kind":"REWORK","_number":2,"created":"2024-10-02 16:59:38.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/2"}}},"commit":{"parents":[{"commit":"947277842cbc0893b4e8790da611ad96a28fce17","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/947277842cbc0893b4e8790da611ad96a28fce17"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-10-02 16:33:43.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93971c11c36809e4f15deb33704f4929c44ec3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93971c11c36809e4f15deb33704f4929c44ec3ab"}]},"branch":"refs/heads/master"},"8b9721904322d1cbdeff1b6bc7e6bbbc4c8e1b9d":{"kind":"REWORK","_number":3,"created":"2024-10-11 13:59:35.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/3"}}},"commit":{"parents":[{"commit":"947277842cbc0893b4e8790da611ad96a28fce17","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/947277842cbc0893b4e8790da611ad96a28fce17"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-10-11 13:23:19.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b9721904322d1cbdeff1b6bc7e6bbbc4c8e1b9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b9721904322d1cbdeff1b6bc7e6bbbc4c8e1b9d"}]},"branch":"refs/heads/master"},"cdb1cdf7df5bdc8bfcf782b7efa622976eb0df1b":{"kind":"REWORK","_number":4,"created":"2024-10-25 13:59:37.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/4"}}},"commit":{"parents":[{"commit":"09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-10-25 13:35:53.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the duration of the upload/download. We want another stat coming out of proxy-logging: something that gets emitted periodically as bytes are actually sent/received so we can get reasonably accurate point-in-time breakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdb1cdf7df5bdc8bfcf782b7efa622976eb0df1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdb1cdf7df5bdc8bfcf782b7efa622976eb0df1b"}]},"branch":"refs/heads/master"},"c02f2d0f04efab4c4e102091c9bf6d5766aae53b":{"kind":"REWORK","_number":5,"created":"2024-10-31 19:36:37.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/5"}}},"commit":{"parents":[{"commit":"09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-10-31 19:34:20.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the duration of the upload/download. We want another stat coming out of proxy-logging: something that gets emitted periodically as bytes are actually sent/received so we can get reasonably accurate point-in-time breakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c02f2d0f04efab4c4e102091c9bf6d5766aae53b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c02f2d0f04efab4c4e102091c9bf6d5766aae53b"}]},"branch":"refs/heads/master"},"0b792d8917bc1503d3e5d3fe89b07a605fad415d":{"kind":"REWORK","_number":6,"created":"2024-11-06 14:59:35.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/6"}}},"commit":{"parents":[{"commit":"09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-11-06 14:36:18.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the duration of the upload/download. We want another stat coming out of proxy-logging: something that gets emitted periodically as bytes are actually sent/received so we can get reasonably accurate point-in-time breakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b792d8917bc1503d3e5d3fe89b07a605fad415d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b792d8917bc1503d3e5d3fe89b07a605fad415d"}]},"branch":"refs/heads/master"},"e85411a080155fcebadec29dbb61c9fbfffa283a":{"kind":"REWORK","_number":7,"created":"2024-11-22 16:56:39.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/7"}}},"commit":{"parents":[{"commit":"09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-11-22 16:39:10.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e85411a080155fcebadec29dbb61c9fbfffa283a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e85411a080155fcebadec29dbb61c9fbfffa283a"}]},"branch":"refs/heads/master"},"d6f4b5f79bea55585c1ec567df1905cc137d7cee":{"kind":"REWORK","_number":8,"created":"2024-11-26 14:59:37.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/8"}}},"commit":{"parents":[{"commit":"09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-11-26 14:35:37.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6f4b5f79bea55585c1ec567df1905cc137d7cee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6f4b5f79bea55585c1ec567df1905cc137d7cee"}]},"branch":"refs/heads/master"},"d85b9d0e6a50c0dfb36e67c68e7e9a9e098cead3":{"kind":"REWORK","_number":9,"created":"2024-11-26 17:22:35.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/9"}}},"commit":{"parents":[{"commit":"09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-11-26 17:16:38.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d85b9d0e6a50c0dfb36e67c68e7e9a9e098cead3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d85b9d0e6a50c0dfb36e67c68e7e9a9e098cead3"}]},"branch":"refs/heads/master"},"1ad1eb45e3bc7aabcb675ecf51b3ca401fb4c80b":{"kind":"REWORK","_number":10,"created":"2024-12-03 20:56:37.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/10"}}},"commit":{"parents":[{"commit":"ee01a3293167ef6faa2bb323815210fd6a6bd915","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee01a3293167ef6faa2bb323815210fd6a6bd915"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-03 20:21:30.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ad1eb45e3bc7aabcb675ecf51b3ca401fb4c80b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ad1eb45e3bc7aabcb675ecf51b3ca401fb4c80b"}]},"branch":"refs/heads/master"},"1af3d28b70a02ca2ed62f9252b527b1efb018b60":{"kind":"REWORK","_number":11,"created":"2024-12-10 14:59:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/11"}}},"commit":{"parents":[{"commit":"ee01a3293167ef6faa2bb323815210fd6a6bd915","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee01a3293167ef6faa2bb323815210fd6a6bd915"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-10 14:36:32.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1af3d28b70a02ca2ed62f9252b527b1efb018b60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1af3d28b70a02ca2ed62f9252b527b1efb018b60"}]},"branch":"refs/heads/master"},"a5b71d2dabc796d3f7a7266a2abb9513856530be":{"kind":"REWORK","_number":12,"created":"2025-01-03 14:59:41.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/12"}}},"commit":{"parents":[{"commit":"ee01a3293167ef6faa2bb323815210fd6a6bd915","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee01a3293167ef6faa2bb323815210fd6a6bd915"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-03 14:31:50.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5b71d2dabc796d3f7a7266a2abb9513856530be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5b71d2dabc796d3f7a7266a2abb9513856530be"}]},"branch":"refs/heads/master"},"e5dd8769e64eb93e3bef81043321b0ea655df85b":{"kind":"REWORK","_number":13,"created":"2025-01-10 20:58:36.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/13"}}},"commit":{"parents":[{"commit":"a2748293d30a3795835c76ee75becddfd20196e2","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2748293d30a3795835c76ee75becddfd20196e2"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-10 20:56:49.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5dd8769e64eb93e3bef81043321b0ea655df85b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5dd8769e64eb93e3bef81043321b0ea655df85b"}]},"branch":"refs/heads/master"},"c0aa3a8c9f6926aba617f36616cdffec998a2d55":{"kind":"REWORK","_number":14,"created":"2025-02-11 14:59:07.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/14"}}},"commit":{"parents":[{"commit":"0540a79255c98653ba1f53aee8294cacaf33d4e5","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0540a79255c98653ba1f53aee8294cacaf33d4e5"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-02-10 22:16:57.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0aa3a8c9f6926aba617f36616cdffec998a2d55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0aa3a8c9f6926aba617f36616cdffec998a2d55"}]},"branch":"refs/heads/master"},"a1d245440619d08ed1997222ac5004e6a802a19b":{"kind":"REWORK","_number":15,"created":"2025-02-21 22:23:15.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/15"}}},"commit":{"parents":[{"commit":"dc2812e2aab06c8082ae921261941360f7a7cc6c","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc2812e2aab06c8082ae921261941360f7a7cc6c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-02-21 20:59:47.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1d245440619d08ed1997222ac5004e6a802a19b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1d245440619d08ed1997222ac5004e6a802a19b"}]},"branch":"refs/heads/master"},"0631f7a7db1413bbabc646fb2e56720a03596b5c":{"kind":"REWORK","_number":16,"created":"2025-02-25 15:01:08.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/16"}}},"commit":{"parents":[{"commit":"dc2812e2aab06c8082ae921261941360f7a7cc6c","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc2812e2aab06c8082ae921261941360f7a7cc6c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-02-25 14:55:32.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0631f7a7db1413bbabc646fb2e56720a03596b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0631f7a7db1413bbabc646fb2e56720a03596b5c"}]},"branch":"refs/heads/master"},"5c17a3f35fc517227b68343c44db0c944faac975":{"kind":"REWORK","_number":17,"created":"2025-02-28 17:23:21.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/17"}}},"commit":{"parents":[{"commit":"3ae4fb5a88f17c5bcd5a6a3208e8fd990c15198d","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ae4fb5a88f17c5bcd5a6a3208e8fd990c15198d"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-02-28 14:59:45.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c17a3f35fc517227b68343c44db0c944faac975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c17a3f35fc517227b68343c44db0c944faac975"}]},"branch":"refs/heads/master"},"2a9da8ea1d57c0de57d5b6d82df9764f09e990e3":{"kind":"REWORK","_number":18,"created":"2025-03-05 20:58:23.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/18"}}},"commit":{"parents":[{"commit":"af0326340dd917f38277eb23a1f576d6e8cbd3bc","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0326340dd917f38277eb23a1f576d6e8cbd3bc"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-05 20:22:44.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a9da8ea1d57c0de57d5b6d82df9764f09e990e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a9da8ea1d57c0de57d5b6d82df9764f09e990e3"}]},"branch":"refs/heads/master"},"a5565e3eeae8ee9fd424f63e52b32587a0efd322":{"kind":"REWORK","_number":19,"created":"2025-03-07 11:58:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/930918/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/19"}}},"commit":{"parents":[{"commit":"f4fa65eb9bd2da286313aa0a1f1840f15d1fcf69","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4fa65eb9bd2da286313aa0a1f1840f15d1fcf69"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 11:57:57.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5565e3eeae8ee9fd424f63e52b32587a0efd322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5565e3eeae8ee9fd424f63e52b32587a0efd322"}]},"branch":"refs/heads/master"},"046bed0e70e78380e2b5e609cc0b1eacef607252":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2025-03-07 13:12:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/930918/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/20"}}},"commit":{"parents":[{"commit":"a215fefd5502c2ce78eec66b502678cdd64f9cad","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a215fefd5502c2ce78eec66b502678cdd64f9cad"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 12:53:16.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/046bed0e70e78380e2b5e609cc0b1eacef607252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/046bed0e70e78380e2b5e609cc0b1eacef607252"}]},"branch":"refs/heads/master"},"94abccacdcc560c80b59f79c6d08d3d04a1a12b3":{"kind":"REWORK","_number":21,"created":"2025-03-10 19:58:26.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/21"}}},"commit":{"parents":[{"commit":"a215fefd5502c2ce78eec66b502678cdd64f9cad","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a215fefd5502c2ce78eec66b502678cdd64f9cad"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-10 19:36:44.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94abccacdcc560c80b59f79c6d08d3d04a1a12b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94abccacdcc560c80b59f79c6d08d3d04a1a12b3"}]},"branch":"refs/heads/master"},"be65cb38fb36dfed10aecd319239221762ccca2a":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2025-03-19 20:16:52.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/22"}}},"commit":{"parents":[{"commit":"c2e806bf13053be39a9a5be1f1fdc000d950c80e","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2e806bf13053be39a9a5be1f1fdc000d950c80e"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-19 19:36:04.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be65cb38fb36dfed10aecd319239221762ccca2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be65cb38fb36dfed10aecd319239221762ccca2a"}]},"branch":"refs/heads/master"},"f3107dcdd6807b24511a6a0922efde7bb707e8d0":{"kind":"REWORK","_number":23,"created":"2025-04-23 15:58:25.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/23"}}},"commit":{"parents":[{"commit":"c3c7eeec4a0c0dc7f4ed6b0e03942e49cfbebefc","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3c7eeec4a0c0dc7f4ed6b0e03942e49cfbebefc"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-23 15:58:25.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3107dcdd6807b24511a6a0922efde7bb707e8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3107dcdd6807b24511a6a0922efde7bb707e8d0"}]},"branch":"refs/heads/master","description":"Rebase"},"806b5b403d8bdfce695c728082231cdf8fb36d85":{"kind":"REWORK","_number":24,"created":"2025-04-28 01:18:35.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/24"}}},"commit":{"parents":[{"commit":"432d31e9d30307005c25e85cd7845984d7b991f3","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/432d31e9d30307005c25e85cd7845984d7b991f3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-28 01:13:40.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/806b5b403d8bdfce695c728082231cdf8fb36d85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/806b5b403d8bdfce695c728082231cdf8fb36d85"}]},"branch":"refs/heads/master"},"a713a9335b9edde16ba5387ebd2022decdc1c753":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2025-04-29 13:17:18.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/25"}}},"commit":{"parents":[{"commit":"0c9c04902043ff2f0f0a57b99dcc81bc3b00de65","subject":"stats: Add path_parts environ for s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c9c04902043ff2f0f0a57b99dcc81bc3b00de65"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-29 13:17:18.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a713a9335b9edde16ba5387ebd2022decdc1c753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a713a9335b9edde16ba5387ebd2022decdc1c753"}]},"branch":"refs/heads/master","description":"Rebase"},"e79cb3e147bc245146fb0111237c4df8665ff537":{"kind":"REWORK","_number":26,"created":"2025-05-02 13:52:03.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/26"}}},"commit":{"parents":[{"commit":"f5d205ac2ce0b30beaba243c5a866446cb202790","subject":"stats: s3 helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5d205ac2ce0b30beaba243c5a866446cb202790"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-02 13:52:03.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e79cb3e147bc245146fb0111237c4df8665ff537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e79cb3e147bc245146fb0111237c4df8665ff537"}]},"branch":"refs/heads/master","description":"Rebase"},"6a672196ac9c8c7694faa1ff60434b9d31849f1a":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-05-05 12:21:38.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/27"}}},"commit":{"parents":[{"commit":"e300cfbdc8ffd07edac65c9aac2f94bc2ff16801","subject":"stats: s3 helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e300cfbdc8ffd07edac65c9aac2f94bc2ff16801"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-05 12:21:38.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a672196ac9c8c7694faa1ff60434b9d31849f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a672196ac9c8c7694faa1ff60434b9d31849f1a"}]},"branch":"refs/heads/master","description":"Rebase"},"cbfe05e1d9264a69da4933a78c48fd4e28f23dc8":{"kind":"REWORK","_number":28,"created":"2025-05-05 14:00:40.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/28"}}},"commit":{"parents":[{"commit":"e300cfbdc8ffd07edac65c9aac2f94bc2ff16801","subject":"stats: s3 helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e300cfbdc8ffd07edac65c9aac2f94bc2ff16801"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-05 13:59:53.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbfe05e1d9264a69da4933a78c48fd4e28f23dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbfe05e1d9264a69da4933a78c48fd4e28f23dc8"}]},"branch":"refs/heads/master"},"7ba70424df600b3d5dac53f035c04d63277fbfd5":{"kind":"REWORK","_number":29,"created":"2025-05-05 14:11:08.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/29"}}},"commit":{"parents":[{"commit":"e300cfbdc8ffd07edac65c9aac2f94bc2ff16801","subject":"stats: s3 helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e300cfbdc8ffd07edac65c9aac2f94bc2ff16801"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-05 14:09:33.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ba70424df600b3d5dac53f035c04d63277fbfd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ba70424df600b3d5dac53f035c04d63277fbfd5"}]},"branch":"refs/heads/master"},"cc4b0952cc9d9a1ab9dc73f57aef814faa82e8bf":{"kind":"REWORK","_number":30,"created":"2025-05-05 15:08:44.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/30"}}},"commit":{"parents":[{"commit":"e300cfbdc8ffd07edac65c9aac2f94bc2ff16801","subject":"stats: s3 helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e300cfbdc8ffd07edac65c9aac2f94bc2ff16801"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-05 15:08:26.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc4b0952cc9d9a1ab9dc73f57aef814faa82e8bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc4b0952cc9d9a1ab9dc73f57aef814faa82e8bf"}]},"branch":"refs/heads/master"},"7a9f556d1a24991a8d466f80e6083d5fd4fe9b35":{"kind":"REWORK","_number":31,"created":"2025-05-06 18:59:06.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/31"}}},"commit":{"parents":[{"commit":"5885da7c26456ecbb4015384085ae84822b7be57","subject":"stats: s3 helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5885da7c26456ecbb4015384085ae84822b7be57"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-06 18:59:06.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a9f556d1a24991a8d466f80e6083d5fd4fe9b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a9f556d1a24991a8d466f80e6083d5fd4fe9b35"}]},"branch":"refs/heads/master","description":"Rebase"},"8e7d4b413998a9225d1edaea87903fa00cd25b7c":{"kind":"REWORK","_number":32,"created":"2025-05-06 20:38:17.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/32"}}},"commit":{"parents":[{"commit":"5885da7c26456ecbb4015384085ae84822b7be57","subject":"stats: s3 helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5885da7c26456ecbb4015384085ae84822b7be57"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-06 20:37:32.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7d4b413998a9225d1edaea87903fa00cd25b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7d4b413998a9225d1edaea87903fa00cd25b7c"}]},"branch":"refs/heads/master"},"89e606f41165d9298ef59173797bf1df6c1c91fc":{"kind":"REWORK","_number":33,"created":"2025-05-06 20:55:31.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/33"}}},"commit":{"parents":[{"commit":"70bd16d5bb983854951e1dba4badc8738f2f0d6c","subject":"stats: s3 helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70bd16d5bb983854951e1dba4badc8738f2f0d6c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-06 20:55:31.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89e606f41165d9298ef59173797bf1df6c1c91fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89e606f41165d9298ef59173797bf1df6c1c91fc"}]},"branch":"refs/heads/master","description":"Rebase"},"5380d3127caf58b370fc4816abd97d35dff948e2":{"kind":"REWORK","_number":34,"created":"2025-05-06 21:06:19.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/34"}}},"commit":{"parents":[{"commit":"70bd16d5bb983854951e1dba4badc8738f2f0d6c","subject":"stats: s3 helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70bd16d5bb983854951e1dba4badc8738f2f0d6c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-06 21:02:53.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5380d3127caf58b370fc4816abd97d35dff948e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5380d3127caf58b370fc4816abd97d35dff948e2"}]},"branch":"refs/heads/master"},"90298d005aa8a28120f38ea151a349c4bbd556c5":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2025-05-08 19:56:43.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/35"}}},"commit":{"parents":[{"commit":"228dd4fffd6c88aa7a3b82e44af78866c0dd3fbf","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/228dd4fffd6c88aa7a3b82e44af78866c0dd3fbf"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-08 19:56:43.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90298d005aa8a28120f38ea151a349c4bbd556c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90298d005aa8a28120f38ea151a349c4bbd556c5"}]},"branch":"refs/heads/master","description":"Rebase"},"231c06cb30196f5958f46902ae0ce01cfe56412e":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2025-05-13 14:29:28.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/36"}}},"commit":{"parents":[{"commit":"5c4c62660acfd671c38966d9301ca78830b26f45","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c4c62660acfd671c38966d9301ca78830b26f45"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-13 14:29:28.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/231c06cb30196f5958f46902ae0ce01cfe56412e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/231c06cb30196f5958f46902ae0ce01cfe56412e"}]},"branch":"refs/heads/master","description":"Rebase"},"9b76f1ea04924cd50bb5538a4f70f9dc905fd09c":{"kind":"REWORK","_number":37,"created":"2025-05-13 14:36:20.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/37"}}},"commit":{"parents":[{"commit":"5c4c62660acfd671c38966d9301ca78830b26f45","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c4c62660acfd671c38966d9301ca78830b26f45"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-13 14:35:52.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b76f1ea04924cd50bb5538a4f70f9dc905fd09c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b76f1ea04924cd50bb5538a4f70f9dc905fd09c"}]},"branch":"refs/heads/master"},"89a6330b92cfc3e8c99e564ca3b96778fc9016ea":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2025-05-15 03:02:16.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/38"}}},"commit":{"parents":[{"commit":"46b31ab3172a8c4c1a28545172157f4f46a38023","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46b31ab3172a8c4c1a28545172157f4f46a38023"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-15 03:02:16.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89a6330b92cfc3e8c99e564ca3b96778fc9016ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89a6330b92cfc3e8c99e564ca3b96778fc9016ea"}]},"branch":"refs/heads/master","description":"Rebase"},"400e87f46063c9cab8980618a2e959e114ea8136":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2025-05-20 12:40:37.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/39"}}},"commit":{"parents":[{"commit":"e21838a53a7aebf5ab16a33bd1b21bfe877b3d27","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e21838a53a7aebf5ab16a33bd1b21bfe877b3d27"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-20 12:40:37.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/400e87f46063c9cab8980618a2e959e114ea8136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/400e87f46063c9cab8980618a2e959e114ea8136"}]},"branch":"refs/heads/master","description":"Rebase"},"b70738f11f1c24ba30dfeb0a822a191be1122ef2":{"kind":"REWORK","_number":40,"created":"2025-05-20 21:00:25.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/40"}}},"commit":{"parents":[{"commit":"46b31ab3172a8c4c1a28545172157f4f46a38023","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46b31ab3172a8c4c1a28545172157f4f46a38023"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-20 21:00:05.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b70738f11f1c24ba30dfeb0a822a191be1122ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b70738f11f1c24ba30dfeb0a822a191be1122ef2"}]},"branch":"refs/heads/master"},"e153431521b46b0c2603fdc3303b8191f8f898d3":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2025-05-21 20:45:27.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/41"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-21 20:45:27.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e153431521b46b0c2603fdc3303b8191f8f898d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e153431521b46b0c2603fdc3303b8191f8f898d3"}]},"branch":"refs/heads/master","description":"Rebase"},"860d8978fa781cffafb3c9eae18849083aaf8548":{"kind":"REWORK","_number":42,"created":"2025-05-28 15:21:12.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/42"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-28 15:20:45.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/860d8978fa781cffafb3c9eae18849083aaf8548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/860d8978fa781cffafb3c9eae18849083aaf8548"}]},"branch":"refs/heads/master"},"eb858c8aff531e84c487b01f06bc7d6e0e3c6c3c":{"kind":"REWORK","_number":43,"created":"2025-06-03 11:56:17.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/43"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-06-03 11:55:44.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb858c8aff531e84c487b01f06bc7d6e0e3c6c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb858c8aff531e84c487b01f06bc7d6e0e3c6c3c"}]},"branch":"refs/heads/master"},"51d715ab0656cf64b2283618fbc7624195e285d8":{"kind":"REWORK","_number":44,"created":"2025-06-04 15:10:15.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/44"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-06-04 15:09:42.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51d715ab0656cf64b2283618fbc7624195e285d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51d715ab0656cf64b2283618fbc7624195e285d8"}]},"branch":"refs/heads/master"},"96024f893ae6a21875cb6c00b4b8a87415ac1d8f":{"kind":"REWORK","_number":45,"created":"2025-06-14 04:09:33.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/45"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-06-14 04:08:53.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96024f893ae6a21875cb6c00b4b8a87415ac1d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96024f893ae6a21875cb6c00b4b8a87415ac1d8f"}]},"branch":"refs/heads/master"},"b50e091123037ffa2652b57f2cc14544b465eb90":{"kind":"REWORK","_number":46,"created":"2025-06-25 20:43:49.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/46"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-06-25 20:42:39.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b50e091123037ffa2652b57f2cc14544b465eb90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b50e091123037ffa2652b57f2cc14544b465eb90"}]},"branch":"refs/heads/master"},"c0433964d3e5c6d0fb67aed705defdb7fb482999":{"kind":"REWORK","_number":47,"created":"2025-06-25 22:33:59.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/47"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-06-25 22:33:22.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0433964d3e5c6d0fb67aed705defdb7fb482999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0433964d3e5c6d0fb67aed705defdb7fb482999"}]},"branch":"refs/heads/master"},"c7ef3f4a9b0fa591ce93aa2e453df3f4d6bdba59":{"kind":"REWORK","_number":48,"created":"2025-06-27 14:53:07.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/48"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-06-27 14:52:41.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7ef3f4a9b0fa591ce93aa2e453df3f4d6bdba59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7ef3f4a9b0fa591ce93aa2e453df3f4d6bdba59"}]},"branch":"refs/heads/master"},"94630db0c954a38588604d7c636e8efb0c666103":{"kind":"REWORK","_number":49,"created":"2025-06-27 15:27:04.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/49"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-06-27 15:26:42.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94630db0c954a38588604d7c636e8efb0c666103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94630db0c954a38588604d7c636e8efb0c666103"}]},"branch":"refs/heads/master"},"080d24a65f8d3efea384046c09b04052b98eecc3":{"kind":"NO_CODE_CHANGE","_number":50,"created":"2025-07-09 19:40:06.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/50"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-09 19:40:06.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/080d24a65f8d3efea384046c09b04052b98eecc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/080d24a65f8d3efea384046c09b04052b98eecc3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"df38dee5c966c360ec432624bd31d4a442d2c6d1":{"kind":"REWORK","_number":51,"created":"2025-07-09 19:42:53.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/51"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-09 19:42:29.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df38dee5c966c360ec432624bd31d4a442d2c6d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df38dee5c966c360ec432624bd31d4a442d2c6d1"}]},"branch":"refs/heads/master"},"bdcefb2bde09d88ae308815838804d7d28660d2c":{"kind":"REWORK","_number":52,"created":"2025-07-11 18:59:24.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/52"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-11 18:58:50.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdcefb2bde09d88ae308815838804d7d28660d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdcefb2bde09d88ae308815838804d7d28660d2c"}]},"branch":"refs/heads/master"},"b7ff6ff1646c1a19f3e7054bd597c97db6449ae7":{"kind":"REWORK","_number":53,"created":"2025-07-11 19:41:49.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/53"}}},"commit":{"parents":[{"commit":"f5546e62ad2ca87422c4898d0f556e43a38f129a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-11 19:41:23.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7ff6ff1646c1a19f3e7054bd597c97db6449ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7ff6ff1646c1a19f3e7054bd597c97db6449ae7"}]},"branch":"refs/heads/master"},"b6fc7a9b37d8d8cc7e8473fbda029ba3bac2cca7":{"kind":"NO_CHANGE","_number":54,"created":"2025-07-16 15:48:21.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/54"}}},"commit":{"parents":[{"commit":"5ef3533ba64da5510a249e464dce2f9bfc19172e","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ef3533ba64da5510a249e464dce2f9bfc19172e"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-16 15:48:21.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6fc7a9b37d8d8cc7e8473fbda029ba3bac2cca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6fc7a9b37d8d8cc7e8473fbda029ba3bac2cca7"}]},"branch":"refs/heads/master","description":"Rebase"},"8aaf69b82928a4f55cb48f2b87a2563258e8e6ea":{"kind":"TRIVIAL_REBASE","_number":55,"created":"2025-07-16 20:47:47.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/55"}}},"commit":{"parents":[{"commit":"237178f76cd2785c88b614edacf371db8e654c09","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/237178f76cd2785c88b614edacf371db8e654c09"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-16 20:47:47.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8aaf69b82928a4f55cb48f2b87a2563258e8e6ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8aaf69b82928a4f55cb48f2b87a2563258e8e6ea"}]},"branch":"refs/heads/master","description":"Rebase"},"12e0281be8b02d7b001895aee4c70e916164faa0":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2025-07-17 02:19:21.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/56"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-17 02:19:21.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12e0281be8b02d7b001895aee4c70e916164faa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12e0281be8b02d7b001895aee4c70e916164faa0"}]},"branch":"refs/heads/master","description":"Rebase"},"4646c1774c0d41974b0389d518c90922a62fe3d3":{"kind":"REWORK","_number":57,"created":"2025-07-17 02:46:52.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/57"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-17 02:46:22.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4646c1774c0d41974b0389d518c90922a62fe3d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4646c1774c0d41974b0389d518c90922a62fe3d3"}]},"branch":"refs/heads/master"},"4641e0ca4aa3481f0bd66d4af9ff4a61e5267946":{"kind":"REWORK","_number":58,"created":"2025-07-22 22:37:13.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/58"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-22 22:36:56.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4641e0ca4aa3481f0bd66d4af9ff4a61e5267946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4641e0ca4aa3481f0bd66d4af9ff4a61e5267946"}]},"branch":"refs/heads/master"},"593cca1a4217fbeb40d456f61be0ba5569b610fd":{"kind":"REWORK","_number":59,"created":"2025-07-24 22:25:37.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/59"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-24 22:25:11.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/593cca1a4217fbeb40d456f61be0ba5569b610fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/593cca1a4217fbeb40d456f61be0ba5569b610fd"}]},"branch":"refs/heads/master"},"2117e189d6ed7c7a044da6daeeec983f07592cce":{"kind":"REWORK","_number":60,"created":"2025-07-25 02:47:28.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/60"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-25 02:33:48.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2117e189d6ed7c7a044da6daeeec983f07592cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2117e189d6ed7c7a044da6daeeec983f07592cce"}]},"branch":"refs/heads/master"},"74e88a1633605f28d0d5eb95d1c889616245a54f":{"kind":"REWORK","_number":61,"created":"2025-07-25 14:15:41.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/61"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-25 14:10:43.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74e88a1633605f28d0d5eb95d1c889616245a54f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74e88a1633605f28d0d5eb95d1c889616245a54f"}]},"branch":"refs/heads/master"},"f412121956cfdd939089254deb7fd518bd06bda9":{"kind":"REWORK","_number":62,"created":"2025-08-01 15:48:02.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/62"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-01 15:47:44.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f412121956cfdd939089254deb7fd518bd06bda9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f412121956cfdd939089254deb7fd518bd06bda9"}]},"branch":"refs/heads/master"},"06a569be3dd39ceec72aa75a3061100daa5c7c8a":{"kind":"REWORK","_number":63,"created":"2025-08-01 20:33:53.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/63"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-01 20:33:22.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06a569be3dd39ceec72aa75a3061100daa5c7c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06a569be3dd39ceec72aa75a3061100daa5c7c8a"}]},"branch":"refs/heads/master"},"89f2c7a2e176b641d2b0d2c5cd17c1c0e47747bb":{"kind":"REWORK","_number":64,"created":"2025-08-04 20:37:29.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/64"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-04 20:36:57.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89f2c7a2e176b641d2b0d2c5cd17c1c0e47747bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89f2c7a2e176b641d2b0d2c5cd17c1c0e47747bb"}]},"branch":"refs/heads/master"},"d8a8d7e051e290790b5e8df57fca28166015a6c8":{"kind":"REWORK","_number":65,"created":"2025-08-06 20:54:28.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/65"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-06 20:54:10.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8a8d7e051e290790b5e8df57fca28166015a6c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8a8d7e051e290790b5e8df57fca28166015a6c8"}]},"branch":"refs/heads/master"},"decefdbe008aca089bb1da28ca05181d53d55a87":{"kind":"REWORK","_number":66,"created":"2025-08-08 01:34:54.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/66"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-08 01:33:50.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/decefdbe008aca089bb1da28ca05181d53d55a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/decefdbe008aca089bb1da28ca05181d53d55a87"}]},"branch":"refs/heads/master"},"752531a59abaad6978e1e3ed6e0f0c47abd10472":{"kind":"REWORK","_number":67,"created":"2025-08-08 20:37:31.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/67"}}},"commit":{"parents":[{"commit":"f7c69412ad30693b81983366b2d173f85296d6c7","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-08 20:37:07.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/752531a59abaad6978e1e3ed6e0f0c47abd10472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/752531a59abaad6978e1e3ed6e0f0c47abd10472"}]},"branch":"refs/heads/master"},"9a96b6ebd540c99922dadbb74af70e67827b3d6a":{"kind":"TRIVIAL_REBASE","_number":68,"created":"2025-08-27 20:33:39.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/68"}}},"commit":{"parents":[{"commit":"120e31e182803b97cbc2908e9786297ca5775770","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/120e31e182803b97cbc2908e9786297ca5775770"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-27 20:33:39.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a96b6ebd540c99922dadbb74af70e67827b3d6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a96b6ebd540c99922dadbb74af70e67827b3d6a"}]},"branch":"refs/heads/master","description":"Rebase"},"30df983babb59bc370dcac3c66dadd29bb4a10f7":{"kind":"TRIVIAL_REBASE","_number":69,"created":"2025-09-02 19:19:59.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/69"}}},"commit":{"parents":[{"commit":"dc25a4c63297a9a5fbc9392364e63e3f7c147762","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc25a4c63297a9a5fbc9392364e63e3f7c147762"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-02 19:19:59.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30df983babb59bc370dcac3c66dadd29bb4a10f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30df983babb59bc370dcac3c66dadd29bb4a10f7"}]},"branch":"refs/heads/master","description":"Rebase"},"ee8afcbbe64c15b5bf236f6fcbaa62df8c880319":{"kind":"REWORK","_number":70,"created":"2025-09-02 19:57:59.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/70"}}},"commit":{"parents":[{"commit":"9b57c987181266a52a7a420694592b07856d4884","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b57c987181266a52a7a420694592b07856d4884"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-02 19:57:59.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee8afcbbe64c15b5bf236f6fcbaa62df8c880319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee8afcbbe64c15b5bf236f6fcbaa62df8c880319"}]},"branch":"refs/heads/master","description":"Rebase"},"fd5f69db7fca4860cb70bdc88b966bde89f7cced":{"kind":"REWORK","_number":71,"created":"2025-09-03 18:46:49.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/71"}}},"commit":{"parents":[{"commit":"9b57c987181266a52a7a420694592b07856d4884","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b57c987181266a52a7a420694592b07856d4884"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-03 18:46:49.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd5f69db7fca4860cb70bdc88b966bde89f7cced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd5f69db7fca4860cb70bdc88b966bde89f7cced"}]},"branch":"refs/heads/master"},"c0005760b717a22f5fb5e6f8de0201936b651254":{"kind":"REWORK","_number":72,"created":"2025-09-03 18:49:11.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/72"}}},"commit":{"parents":[{"commit":"9b57c987181266a52a7a420694592b07856d4884","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b57c987181266a52a7a420694592b07856d4884"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-03 18:49:11.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0005760b717a22f5fb5e6f8de0201936b651254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0005760b717a22f5fb5e6f8de0201936b651254"}]},"branch":"refs/heads/master"},"58c629bb3bb9be9271441c4b8153a666e284d007":{"kind":"REWORK","_number":73,"created":"2025-09-03 19:01:42.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/73"}}},"commit":{"parents":[{"commit":"9b57c987181266a52a7a420694592b07856d4884","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b57c987181266a52a7a420694592b07856d4884"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-03 19:01:13.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58c629bb3bb9be9271441c4b8153a666e284d007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58c629bb3bb9be9271441c4b8153a666e284d007"}]},"branch":"refs/heads/master"},"7075a5885497cfd1115b58457bffb292b10a8aee":{"kind":"TRIVIAL_REBASE","_number":74,"created":"2025-09-03 19:28:26.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/74"}}},"commit":{"parents":[{"commit":"d4a254ddbfade515959995043f1a577d5e8f8715","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4a254ddbfade515959995043f1a577d5e8f8715"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-03 19:28:26.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7075a5885497cfd1115b58457bffb292b10a8aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7075a5885497cfd1115b58457bffb292b10a8aee"}]},"branch":"refs/heads/master","description":"Rebase"},"e4bac03afdc5b67284407a8a585eea20d624cddb":{"kind":"REWORK","_number":75,"created":"2025-09-03 20:45:03.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/75"}}},"commit":{"parents":[{"commit":"d4a254ddbfade515959995043f1a577d5e8f8715","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4a254ddbfade515959995043f1a577d5e8f8715"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-03 20:44:43.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4bac03afdc5b67284407a8a585eea20d624cddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4bac03afdc5b67284407a8a585eea20d624cddb"}]},"branch":"refs/heads/master"},"1e61e565c14700ee7df5490fb6ebbed672045479":{"kind":"TRIVIAL_REBASE","_number":76,"created":"2025-10-02 19:45:01.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/76"}}},"commit":{"parents":[{"commit":"615393f49ea53cccbc36321399a5d85caeca16d1","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/615393f49ea53cccbc36321399a5d85caeca16d1"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-02 19:45:01.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e61e565c14700ee7df5490fb6ebbed672045479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e61e565c14700ee7df5490fb6ebbed672045479"}]},"branch":"refs/heads/master","description":"Rebase"},"3738bf93b3091aa67ecee4cefc01b68180904c46":{"kind":"REWORK","_number":77,"created":"2025-10-07 15:40:14.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/77"}}},"commit":{"parents":[{"commit":"615393f49ea53cccbc36321399a5d85caeca16d1","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/615393f49ea53cccbc36321399a5d85caeca16d1"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-07 15:39:54.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3738bf93b3091aa67ecee4cefc01b68180904c46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3738bf93b3091aa67ecee4cefc01b68180904c46"}]},"branch":"refs/heads/master"},"0f7719469d92e067c30c5f64e77f2c6aeb0abae7":{"kind":"TRIVIAL_REBASE","_number":78,"created":"2025-10-07 17:44:33.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/78"}}},"commit":{"parents":[{"commit":"656c5709c8025732ef3b84e24b17d7f24d36257f","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/656c5709c8025732ef3b84e24b17d7f24d36257f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-07 17:44:33.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f7719469d92e067c30c5f64e77f2c6aeb0abae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f7719469d92e067c30c5f64e77f2c6aeb0abae7"}]},"branch":"refs/heads/master","description":"Rebase"},"3d5648e06f43a125073f4b34fc3ace371e392492":{"kind":"TRIVIAL_REBASE","_number":79,"created":"2025-10-08 20:14:34.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/79"}}},"commit":{"parents":[{"commit":"17111ccac98839d9eb8a0144ca3cd97c6a418f2a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17111ccac98839d9eb8a0144ca3cd97c6a418f2a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-08 20:14:34.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d5648e06f43a125073f4b34fc3ace371e392492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d5648e06f43a125073f4b34fc3ace371e392492"}]},"branch":"refs/heads/master","description":"Rebase"},"0a724c02fbdf5c065f6d0a72b6ed2753379ee095":{"kind":"REWORK","_number":80,"created":"2025-10-08 21:08:25.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/80"}}},"commit":{"parents":[{"commit":"17111ccac98839d9eb8a0144ca3cd97c6a418f2a","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17111ccac98839d9eb8a0144ca3cd97c6a418f2a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-08 21:08:03.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a724c02fbdf5c065f6d0a72b6ed2753379ee095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a724c02fbdf5c065f6d0a72b6ed2753379ee095"}]},"branch":"refs/heads/master"},"76e805a3f02e549caf3c9d5bddf1c071b2a26844":{"kind":"TRIVIAL_REBASE","_number":81,"created":"2025-10-10 18:26:51.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/81"}}},"commit":{"parents":[{"commit":"ddccb990f55e264ebfda4442346668a99302df5e","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddccb990f55e264ebfda4442346668a99302df5e"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-10 18:26:51.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76e805a3f02e549caf3c9d5bddf1c071b2a26844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76e805a3f02e549caf3c9d5bddf1c071b2a26844"}]},"branch":"refs/heads/master","description":"Rebase"},"630c47a8f554f3ed7b903effead5bce3ce82abcf":{"kind":"REWORK","_number":82,"created":"2025-10-10 18:30:25.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/82"}}},"commit":{"parents":[{"commit":"ddccb990f55e264ebfda4442346668a99302df5e","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddccb990f55e264ebfda4442346668a99302df5e"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-10 18:30:02.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/630c47a8f554f3ed7b903effead5bce3ce82abcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/630c47a8f554f3ed7b903effead5bce3ce82abcf"}]},"branch":"refs/heads/master"},"e58aafd5c8fdd450062e76e47bd0222ebdbb466f":{"kind":"TRIVIAL_REBASE","_number":83,"created":"2025-10-13 14:25:43.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/83"}}},"commit":{"parents":[{"commit":"5b1052c7bbab23750a4ab487e9d4ad5ee2869ad0","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b1052c7bbab23750a4ab487e9d4ad5ee2869ad0"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-13 14:25:43.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e58aafd5c8fdd450062e76e47bd0222ebdbb466f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e58aafd5c8fdd450062e76e47bd0222ebdbb466f"}]},"branch":"refs/heads/master","description":"Rebase"},"042e31fd0d5e282c830fd7ca2186e4a7aec31533":{"kind":"TRIVIAL_REBASE","_number":84,"created":"2025-10-13 15:29:29.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/84"}}},"commit":{"parents":[{"commit":"a3f8b0ab4573bf1588744d568fc49bc08c40f5df","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3f8b0ab4573bf1588744d568fc49bc08c40f5df"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-13 15:29:29.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/042e31fd0d5e282c830fd7ca2186e4a7aec31533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/042e31fd0d5e282c830fd7ca2186e4a7aec31533"}]},"branch":"refs/heads/master","description":"Rebase"},"eb6137934faf92bb74fe2607c6ea52fbb87b72be":{"kind":"REWORK","_number":85,"created":"2025-10-13 16:38:28.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/85"}}},"commit":{"parents":[{"commit":"a3f8b0ab4573bf1588744d568fc49bc08c40f5df","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3f8b0ab4573bf1588744d568fc49bc08c40f5df"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-13 16:38:05.000000000","tz":-300},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb6137934faf92bb74fe2607c6ea52fbb87b72be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb6137934faf92bb74fe2607c6ea52fbb87b72be"}]},"branch":"refs/heads/master"},"1a081d4e3f01893c0bd898b1b877b9a95b1339b3":{"kind":"TRIVIAL_REBASE","_number":86,"created":"2025-10-14 13:58:19.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/930918/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/86"}}},"commit":{"parents":[{"commit":"9d7e7e27a5114ee7390519324514e1bd26124ffc","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d7e7e27a5114ee7390519324514e1bd26124ffc"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-14 13:58:19.000000000","tz":0},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a081d4e3f01893c0bd898b1b877b9a95b1339b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a081d4e3f01893c0bd898b1b877b9a95b1339b3"}]},"branch":"refs/heads/master","description":"Rebase"},"93a95f90af65ad1d3f4fe8efc8a1f9db25e1ed0a":{"kind":"REWORK","_number":87,"created":"2025-10-28 21:23:19.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/87"}}},"commit":{"parents":[{"commit":"9d7e7e27a5114ee7390519324514e1bd26124ffc","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d7e7e27a5114ee7390519324514e1bd26124ffc"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-10-28 19:55:19.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93a95f90af65ad1d3f4fe8efc8a1f9db25e1ed0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93a95f90af65ad1d3f4fe8efc8a1f9db25e1ed0a"}]},"branch":"refs/heads/master"},"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3":{"kind":"REWORK","_number":88,"created":"2025-10-30 19:31:21.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/18/930918/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/930918/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/930918/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/930918/88"}}},"commit":{"parents":[{"commit":"9d7e7e27a5114ee7390519324514e1bd26124ffc","subject":"Provide some s3 helper methods for other middlewares to use.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d7e7e27a5114ee7390519324514e1bd26124ffc"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-26 13:35:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-10-30 19:30:37.000000000","tz":-240},"subject":"proxy-logging: Add real-time transfer bytes counters","message":"proxy-logging: Add real-time transfer bytes counters\n\nCurrently we can get one proxy-logging transfer stat emission over the\nduration of the upload/download. We want another stat coming out of\nproxy-logging: something that gets emitted periodically as bytes are\nactually sent/received so we can get reasonably accurate point-in-time\nbreakdowns of bandwidth usage.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Ideecd0aa58ddf091c9f25f15022a9066088f532b\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
