)]}'
{"id":"openstack%2Fswift~930954","triplet_id":"openstack%2Fswift~master~I536ca0908733b148457bd48a119d1054cabf717d","project":"openstack/swift","branch":"master","topic":"mpu-expire-randomized-chain","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-10-01 13:43:22.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-10-01 13:43:22.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I536ca0908733b148457bd48a119d1054cabf717d","subject":"wip: rotate cycle and skip after awhile","status":"NEW","created":"2024-09-30 22:18:03.000000000","updated":"2024-10-01 13:43:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"b2d02bedbdb42324090b386129edab31c431b57f","_number":930954,"virtual_id_number":930954,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-30 23:29:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":-1,"date":"2024-10-01 04:20:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-30 22:43:38.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2024-09-30 23:29:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-01 04:20:12.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"a8e6a8b015fbfe4f3a9392dcea1186acf014372f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-09-30 22:18:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfcc08f6bae85285dee21fe0baea7b5f93ca5e66","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-09-30 22:43:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"719d36a5e55c89dc16020e50576459ee932c3916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 23:29:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5c6668bc82e40e89159071143afb56b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e060ddc73268440aa391d5b3dab579d6 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/2de6ed3ce0c04458aa699f2610708008 : SUCCESS in 1h 00m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed38043711584e3687afddb33bf14506 : SUCCESS in 1h 05m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/15e3a2845dfb47f2942ad8ddf8ec359a : SUCCESS in 58m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85e4674729074cdda4b1b27f637baeeb : SUCCESS in 24m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4b49d571db44d748079afa3878a27c0 : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/162041f43ad64867bf76fe6afdb7b3c5 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f524b0115504525b58d0b0f8ea55cfb : SUCCESS in 18m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a482c8df6e3d42b0b95179a71c0eedbf : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/117f977bd58e4a4994e3b418f16dc308 : FAILURE in 7m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c04912d454f4f89a415f0fa8b330505 : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/30e3c7aa97e04ed7943b41c7d03d314c : SUCCESS in 15m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6d8f2c1aa6b04e89bca12f0b2b3e2ded : SUCCESS in 14m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/043f696d2dd1455d895ba192cc837362 : SUCCESS in 19m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb3701b1e2fe45f49efd3f50d861c9df : FAILURE in 3m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a34f4f6af10e42d5a6624645251c604d : FAILURE in 3m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47f32b8705bf47fcacaaabf02cc20fbb : FAILURE in 3m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ded56158bdd64df6a30915e2154ac463 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/37ecea4d1a634134b38638771a28e090 : SUCCESS in 8m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c4ca0733130b40fb8887e00e15cf4359 : SUCCESS in 14m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d0c39eb5952c402eba84867f63f0fedf : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b66769717c9f43bbb34e2cd70a61de41 : SUCCESS in 14m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7280890f7b4241239f2d632418ee8afa : SUCCESS in 5m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/42df915cc4b04f53a356a18525a2dfc0 : SUCCESS in 46m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdb3e4ffd8fc42a0bf1852ffe3bd4195 : SUCCESS in 59m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb6e45386b084b039b398aa6a98afc77 : SUCCESS in 52m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/218acd87e7534536a0df835f86bacf24 : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32eed377caf7454691dd34a3c956f40d : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c9f781b084f4157aa6c32283c58c6a8 : SUCCESS in 19m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1b78bc5c1ae422db36ba455e08c2cea : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e00767d2e16defa80997412a791d26ac9aebc803","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 00:22:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a03ea774f67944d0920d126e1da849d5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/366b25389a544542aff8156a0c8d729c : TIMED_OUT in 2h 01m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/354b28c07dcd44e98f2f5a390044aec1 : SUCCESS in 1h 05m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ea47a6d51f44b22a39d3f0bb58b01e0 : SUCCESS in 1h 18m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f75ec6ee67b54b4c8f19f43214b5b0bd : SUCCESS in 1h 14m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3c75b7a7a368d7c53948bd97cd684b148cc00126","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-01 04:20:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b2d02bedbdb42324090b386129edab31c431b57f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-01 13:43:22.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1fea093331c62ada553f210cd28582a8519533d","revisions":{"a1fea093331c62ada553f210cd28582a8519533d":{"kind":"REWORK","_number":1,"created":"2024-09-30 22:18:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/54/930954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/930954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/930954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/930954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/930954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/930954/1"}}},"commit":{"parents":[{"commit":"4112dd0c7b9df84b387bf785ab8dc93d033b4c1b","subject":"sq? expirer: don\u0027t try to delete non-existent task containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4112dd0c7b9df84b387bf785ab8dc93d033b4c1b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-09-30 22:17:50.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-09-30 22:17:50.000000000","tz":-300},"subject":"wip: rotate cycle and skip after awhile","message":"wip: rotate cycle and skip after awhile\n\nChange-Id: I536ca0908733b148457bd48a119d1054cabf717d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1fea093331c62ada553f210cd28582a8519533d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1fea093331c62ada553f210cd28582a8519533d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
