)]}'
{"id":"openstack%2Fswift~931011","triplet_id":"openstack%2Fswift~feature%2Fmpu~I806302c798b67489f540387796e778c29001afdb","project":"openstack/swift","branch":"feature/mpu","topic":"p-content-type","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-02 16:54:19.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-02 18:38:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I806302c798b67489f540387796e778c29001afdb","subject":"mpu: minor middleware fixes","status":"MERGED","created":"2024-10-01 13:14:59.000000000","updated":"2024-10-03 03:43:35.000000000","submitted":"2024-10-02 18:38:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"931011-p-content-type","meta_rev_id":"082f3eab6f4944677f0cac1c34f9ca11e4afbee7","_number":931011,"virtual_id_number":931011,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-02 18:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-10-02 16:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-10-02 16:54:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 14:16:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-01 15:51:36.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"c1a960e218056479fce166e420b6991db655dc36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-01 13:14:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af61425348ab02c20447da79c5d3cbf65adc2dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 14:16:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68a2baba6b1a42c9ba7280ced0c33414\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9b53efe2cd44614a206ee359bca83a9 : SUCCESS in 6m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/137db9e6995c484ca7e04e9702992ba2 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a34a2b52342490da69d60de83defc68 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbb097a7cd404423a78618d1ae9f937d : SUCCESS in 14m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/369fabc4112b4731b0b9a37c6c56aec8 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/509d8798a3ee4e438d855bba2148b00b : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c5e6c04dbd242099c6b607240424f92 : SUCCESS in 17m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a68be962db5a455083b804f6bba9f652 : SUCCESS in 23m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9b62fa83b64f4731a06d023468a02fcd : SUCCESS in 10m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ac50b273a91a45e8aad65d68ffd64138 : SUCCESS in 23m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/18422910a3be4ebcb78886b157c7304a : SUCCESS in 10m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3b9b93e94534176b76554742d934510 : SUCCESS in 10m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7d91819cf2a4f5eb1c0b1140addd851 : SUCCESS in 10m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4628b49004084e4994b62de4b0f779a8 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eccea6cc91fa4d29a4693eaac0095e7a : SUCCESS in 20m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b961c8970a1f478a81bedc8e0fe911c0 : SUCCESS in 9m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/949b888437fc474eb9fb1573c5d236e2 : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c8455d4141145ca8dc6c55894679f44 : SUCCESS in 14m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2577ac33cdeb44ecbf74562c61046b0d : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/737991be51c74f1282741e4392bb3ab9 : SUCCESS in 56m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6773e9a2d73346ba863eaba8effa3f9b : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097aa35b376241389beee0ae687e154c : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f0ffcae278d24de09a2faa4663a5764a : SUCCESS in 18m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b0055df24f374eba01826ceb5483036d7967b79b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-01 15:51:36.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4bf735abb28de4d85a57c333d2fc5275dd8291f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 15:54:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e48cf035b7ea4e71a1b0aa110cee7126\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbc36b23ee4744789ec5ef6a376f4a5f : TIMED_OUT in 2h 01m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f2394b56ba04517a3a40c625f508b85 : TIMED_OUT in 2h 01m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ae3acede1864f48bee108ec108d3a55 : SUCCESS in 1h 53m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4279bf9965d4246a284883eba8f94ee : TIMED_OUT in 2h 01m 03s","accounts_in_message":[],"_revision_number":1},{"id":"13f5325539321393d9d385294bf5c7260c6d586e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 13:10:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0f20081749a1596533977ab0e5f50fb26c7498dc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 13:11:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eb83df0efceea53d60188fc95cf44c8ecc77f794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 13:14:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"92a75b3bb8cfda9f61f2e1b513489c5ecb17454f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 14:10:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49de6d3f164948018ec6d6729ab7f8bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db3b4a7ac774456caed19b9e53f6942f : SUCCESS in 8m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9178f3c493f84c308d2937032716089b : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18af1d3535a448c18424e407a595b424 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0112660d6258469cb778d2a702c0cb21 : SUCCESS in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c455af0d123440149e8d10a4fd41249e : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/97cf70c3ff0e47dc9734a86daa9d7259 : SUCCESS in 16m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2558864ba3b4c0d90753a30ad3949b0 : SUCCESS in 18m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3cc3f3998547437a860b15cc6b99e5b9 : SUCCESS in 21m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/45040173184c4c3faf480a9c937b9b2a : SUCCESS in 22m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b2b4be7c9d524eeea88184fda4b45613 : SUCCESS in 25m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/df21b6ea7a1a4257a62e379c16955cc4 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/532e805e61894c2d94a891399985a5ad : SUCCESS in 22m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e65b52d335c46d0beecf58154d92b2b : SUCCESS in 17m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b21578fbe53d4b75813f86f089f96790 : SUCCESS in 8m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dbdeea414b3645689491a33cfbe200c1 : SUCCESS in 21m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/76c5861f763a4b0baf426126ffe1cbb8 : SUCCESS in 17m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2f3f8814d1b43d4bcb655b79c827769 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f424f47242ab4fa28c3ddbfbac781982 : SUCCESS in 14m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b029f049f63940a78c2a8a3cbd766825 : SUCCESS in 7m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/83afd68fa5014d8b9d9d836b8cf9fece : SUCCESS in 50m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/daa200952b6d40b29506d675de19ab14 : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc07834370045d988d3aed6e06bc6c7 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52ecdd85245747f8adeaf3f65a2c23e9 : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":3},{"id":"ec2db4d9992bf66e0a3d5b57fd006afb026423a5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 14:27:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ea8880b45525591a3f438c9cdd63f4869e709470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 14:28:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"532f18fc3522559e9dc36643f0c013dcaf67cebc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 14:58:46.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"66a4b7a1c3e41a32fc51122dd281926cd840481e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 16:32:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e823ac05fe46b9b7e51e38ca2386bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36720fd2fdf645ec9626933a03b6bd02 : SUCCESS in 8m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c142c276aae4ac39445f46287b1e572 : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae6e4547b3174f78aa5e792e82e8556c : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26e75e33c23449e4834101be3e4bb23b : SUCCESS in 14m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9511d603772e45dd8388aa3d0db59900 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dac6538cc8144213afdfa113722a8284 : SUCCESS in 21m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33bdca32b55a4adb86513382c402e2b9 : SUCCESS in 18m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7cd3be0db05d4c2c939a7cf650e0eb91 : SUCCESS in 17m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/da14f33c341b403d8039bccccccb068b : SUCCESS in 27m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7d988dfbff77469b9d9a09df6978a5d1 : SUCCESS in 28m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6965386212154d56a8a8bbf255303f7b : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a180c1aa8f814c8ca1ca600c70bd0d66 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ea9a01d1b75244778ebb34334b8a4850 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/82fe912692a0449fb0d047047d738cf8 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7868426b6c5e48c59460b97b61a0136b : SUCCESS in 14m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/382effea5e38468885b796fbf8172cd0 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/76c0023c55ca493ab0f7baee38674e64 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/625a45769e2a4c1e8c5aaf49363df09b : SUCCESS in 14m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f3b440539c2c4a91985b8daa2ee3a283 : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/121d77ff33044e04934d2e3a6b598449 : SUCCESS in 53m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3da04b8186584598ae12934847873da7 : SUCCESS in 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d05640a62fe46f98efffc42211d75b7 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a79e2d0d511140dbae5db6fa1dc3ea64 : SUCCESS in 21m 17s","accounts_in_message":[],"_revision_number":4},{"id":"353f08a26463b29cdaaf7095933a31cf81d382f4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 16:54:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fb91fac009c1f96a09d9c23f5fa19b8905e78898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 16:55:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0be85d70a9fcaeb0b97f82be5d27e5e2a2fb3438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:38:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c36c58ff8354186a2b998e2bad9c3b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e6503452d14ca3a70b9900c857a91d : SUCCESS in 8m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/29ed005027f24af6a85bdea170078509 : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/57f587bc16b24397b5c5d0c2a37fe2e3 : SUCCESS in 18m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/99e5998904784effa4c3a95b5647e76b : SUCCESS in 18m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8c73b99611cf4d4d98b40c2725dd6e8e : SUCCESS in 20m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1271339c1f9844c687b78c04cf7c981f : SUCCESS in 22m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/82b12b528bed4efa87a248f60f0aacb4 : SUCCESS in 14m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1c4bf1e02a7947eb9ac27e9e8f4dfae7 : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa9806b00af94bf6b7d8920610ba5268 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/39ceb23e272a4268994aa4039658b501 : SUCCESS in 17m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/98d3bbca6be14535bee3711ae3e959f5 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9599784887f048189bf24e9c31c2d473 : SUCCESS in 17m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7714d6e4ee1f4627a341a69bc7c5deb6 : SUCCESS in 19m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9cdbd9e8e5f4dd382b7c48d5273d17f : SUCCESS in 6m 38s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/84179601404f41eda142d177d0f38603 : SUCCESS in 7m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4b039a0be4b4cd09206566035f9918d : SUCCESS in 46m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f850fb3db8346949cfbc944a9c7523a : SUCCESS in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe204b685a934ae6b377d443fd012cdf : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ceae7b0ed1584047b08da2363002a5dc : SUCCESS in 12m 09s","accounts_in_message":[],"_revision_number":4},{"id":"dc0d50b169d6c2054e411066087d407401715ab3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:38:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d942154c6f81697e2f4edab2872f2068ff654e82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:39:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e562bc1b1ec44ab299db41b4f7bcf47f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8735afab5fc4e96b3ef1e4c571b2aae : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6cb6f321fa254faabbbc48be46c62ec7 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"082f3eab6f4944677f0cac1c34f9ca11e4afbee7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 03:43:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/165ca93e773b46358ff439c32cd0dde0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/19028c8c04604e748a5b06f29946f1e1 : SUCCESS in 1h 09m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d80c1d551b3d42468d68bdc59132f897 : SUCCESS in 1h 47m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b167b4161df4a43aba9d9655b0762b9 : SUCCESS in 1h 06m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a1d108dd39340168577fb10904b32f5 : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b53dd8e8eb32e6347e74f98a9079eec7d1c4ed7d","revisions":{"4028a4858a3f9ea0dc1e8d22e9e6c11e736b6866":{"kind":"REWORK","_number":1,"created":"2024-10-01 13:14:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/931011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/931011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/931011/1"}}},"commit":{"parents":[{"commit":"b37f212e34ab8c73969916e4d1ee3cff2f811c90","subject":"Merge \"Refuse part upload to aborted or completing session\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b37f212e34ab8c73969916e4d1ee3cff2f811c90"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-01 13:13:57.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-01 13:13:57.000000000","tz":60},"subject":"mpu: minor middleware fixes","message":"mpu: minor middleware fixes\n\nChange-Id: I806302c798b67489f540387796e778c29001afdb\nRelated-Change: I835a029d0fb366852c83cb3fc17be4eb51274efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4028a4858a3f9ea0dc1e8d22e9e6c11e736b6866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4028a4858a3f9ea0dc1e8d22e9e6c11e736b6866"}]},"branch":"refs/heads/feature/mpu"},"d2f1810db7136c719afe2c56987776a377d48fa2":{"kind":"REWORK","_number":2,"created":"2024-10-02 13:10:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/931011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/931011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/931011/2"}}},"commit":{"parents":[{"commit":"b37f212e34ab8c73969916e4d1ee3cff2f811c90","subject":"Merge \"Refuse part upload to aborted or completing session\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b37f212e34ab8c73969916e4d1ee3cff2f811c90"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 11:58:00.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 11:58:00.000000000","tz":60},"subject":"mpu: minor middleware fixes","message":"mpu: minor middleware fixes\n\nChange-Id: I806302c798b67489f540387796e778c29001afdb\nRelated-Change: I835a029d0fb366852c83cb3fc17be4eb51274efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2f1810db7136c719afe2c56987776a377d48fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2f1810db7136c719afe2c56987776a377d48fa2"}]},"branch":"refs/heads/feature/mpu"},"9efdbf62752ef3ebe0493d3d43fb3029fcce4d3c":{"kind":"REWORK","_number":3,"created":"2024-10-02 13:14:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/931011/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/931011/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/931011/3"}}},"commit":{"parents":[{"commit":"b37f212e34ab8c73969916e4d1ee3cff2f811c90","subject":"Merge \"Refuse part upload to aborted or completing session\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b37f212e34ab8c73969916e4d1ee3cff2f811c90"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 11:58:00.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 13:14:11.000000000","tz":60},"subject":"mpu: minor middleware fixes","message":"mpu: minor middleware fixes\n\nChange-Id: I806302c798b67489f540387796e778c29001afdb\nRelated-Change: I835a029d0fb366852c83cb3fc17be4eb51274efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9efdbf62752ef3ebe0493d3d43fb3029fcce4d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9efdbf62752ef3ebe0493d3d43fb3029fcce4d3c"}]},"branch":"refs/heads/feature/mpu"},"b53dd8e8eb32e6347e74f98a9079eec7d1c4ed7d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-02 14:58:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/931011/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/931011/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/931011/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/931011/4"}}},"commit":{"parents":[{"commit":"8bfb6b4c0bf8138ff080856a3a912e26d24aab71","subject":"mpu: make MPUSession.is_aborted a property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bfb6b4c0bf8138ff080856a3a912e26d24aab71"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 11:58:00.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 14:57:52.000000000","tz":60},"subject":"mpu: minor middleware fixes","message":"mpu: minor middleware fixes\n\nChange-Id: I806302c798b67489f540387796e778c29001afdb\nRelated-Change: I835a029d0fb366852c83cb3fc17be4eb51274efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53dd8e8eb32e6347e74f98a9079eec7d1c4ed7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53dd8e8eb32e6347e74f98a9079eec7d1c4ed7d"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
