)]}'
{"id":"openstack%2Fswift~931023","triplet_id":"openstack%2Fswift~feature%2Fmpu~If908d3dff98629cd98d6f816efda69959eba999e","project":"openstack/swift","branch":"feature/mpu","topic":"p-aborted-property","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-02 13:09:41.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-02 14:14:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If908d3dff98629cd98d6f816efda69959eba999e","subject":"mpu: make MPUSession.is_aborted a property","status":"MERGED","created":"2024-10-01 14:22:27.000000000","updated":"2024-10-02 14:15:38.000000000","submitted":"2024-10-02 14:14:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931023-p-aborted-property","meta_rev_id":"99709a9ca5de42038719480b85a26f7c4a74a889","_number":931023,"virtual_id_number":931023,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-02 14:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-10-02 13:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-02 10:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-10-02 13:09:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 15:00:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-01 17:54:35.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"a90c8ad663ff84f7cd1dd414154930bbd36d52eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-01 14:22:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63bdd5afb7c71d5be0c09520fe481640fa5ce9a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 15:00:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da951749c2c54f0788f7aabaab76876e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d02d76076e940fc98f2b743a6d01685 : SUCCESS in 7m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e2be905c0924be497460f9578185540 : SUCCESS in 11m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f935ab91ad64be0bd51f596e431ce62 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73f8efd123544ec6b218742e0a8979dc : SUCCESS in 21m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ffe39a264136494cb84e4d1efc642189 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ad6ab3e12ee41f79697b9300506435c : SUCCESS in 22m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/afe917639cce4e56a953d6cc5590c947 : SUCCESS in 20m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/194104268122430b91cc9774924220b2 : SUCCESS in 22m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ba4bfde87aef4576b80283c755ca7062 : SUCCESS in 21m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d1209ad076f348b1a84ce8e9334457ab : SUCCESS in 22m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/40d276b33b214d12889cfc27e791cefa : SUCCESS in 20m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9e35f3307f9e46aabcbe090ad97b2440 : SUCCESS in 20m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4684b0cca97248e6a4cedeb632481ad9 : SUCCESS in 10m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e1fcf6ebf7194ad292393ab8f7712c41 : SUCCESS in 8m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8166dae425ef4a7f8f46d2d07751f069 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/34989065e6954d75b48beaee415e04b5 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ac6e2987a4544bba66889a87061c9e0 : SUCCESS in 4m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0fd234645f624c4381c85dd64ebd77bd : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/12b30ca2d076489ab2aa930f82b2a40c : SUCCESS in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5f9fdcec9a144b8b90257091e567518 : SUCCESS in 37m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dabb6adaf6e443e5bdd1d890edbbf4d0 : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69b5cb5ff0cb4986873b261cdc9dbfff : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b21552b3b224424e8353256c025a368f : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f2b2de28ed79b87615d9cd11864bf28967d71754","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 16:35:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9ab50449de4415b00c1da86bfc7d99\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b8139f2155a44abbcb89eee7811764c : SUCCESS in 1h 22m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0173615b76da4118bcf39575c9f47dc1 : SUCCESS in 1h 20m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e6596e5eedf42deafce5137e02db82f : SUCCESS in 1h 29m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ee36219dd914192a7e6904b10d28e91 : SUCCESS in 1h 39m 52s","accounts_in_message":[],"_revision_number":1},{"id":"14a2e7be268f5d901134157d970f597e1a8e06b3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-01 17:54:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a321f190e83b6689f35b8e7982cf522e107615e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 10:26:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6016c87588723464b2bafbef7f745cc5ab79441b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 11:19:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97fe6024729649fab33ea831048f06f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42dde795434f4c929ccce6befd0e7371 : SUCCESS in 8m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd43d6615d014e6a81f368b5f9ff318b : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48db87e2f2c64c758b2f7a1ab1a00b0c : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c027d94dfc154584a45b37c1dfdba66c : SUCCESS in 16m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca4371fbe296420998e523b6be182a02 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2c75aa567dbe4f589a15588b542a5866 : SUCCESS in 25m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3ce6978d4be46758f7c788974ba2c33 : SUCCESS in 21m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/09670ba25d554d4fb7532bb26a1080c2 : SUCCESS in 20m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/19db31babf41448195dc2ffe9bb45ad8 : SUCCESS in 24m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8c39ac4305cc4a53ab7cca229f2706a2 : SUCCESS in 29m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de8916dfdf2948769b05cb9852d69f13 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d65e1d563fa7420aa954a306d35a30d9 : SUCCESS in 21m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/03a3c45201a94273ade6810df39b0333 : SUCCESS in 16m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/92085cc867234eeead94cda67ca09b9c : SUCCESS in 8m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0b4cee555c9641dcb31acef7774d7654 : SUCCESS in 8m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/feb5c9a509c7443bb223e82cc5d8e5f0 : SUCCESS in 22m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b2943b476804ecd8c07f0c93c0fd4a7 : SUCCESS in 8m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84c18849ad8849f1a6a18f15d902facf : SUCCESS in 15m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a24f131d5ab4e5dae73c246d8136a8d : SUCCESS in 7m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/17ea9f93ecf742ca852283a380b5e356 : SUCCESS in 37m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d247f597bb7546c5a6a73c248d8e4088 : SUCCESS in 25m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35d548ff3ce840859c962e731f9ff073 : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8d3b5af400fb417e91dc2ba45aabd778 : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5b625fd1ca5bf6f7c9388bbcde99298466696dce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 12:48:17.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1283523ddd744e58a543c27a1955aee1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d86945a132e54c38a4ea2b14dbd54437 : SUCCESS in 1h 28m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/152f3b7f8dc243aea6c60a1ce2332924 : SUCCESS in 1h 27m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6e506b9dbcc402183651b6f168fb331 : SUCCESS in 2h 00m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1f7eef48e0f4d56989d71f7a6fc8357 : TIMED_OUT in 2h 00m 50s","accounts_in_message":[],"_revision_number":2},{"id":"044167dc55c61a776c4873041903954d48b60786","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 13:09:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c7997dc12b80dd5fe4a572cb375c5786003f8af3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 13:09:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d2cc69e76c297c12665c0fd4abb97f51a225b686","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 14:13:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ffd55135b0478fae198c0f25cb14a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a59a9033a7496e997fac5a64b0a020 : SUCCESS in 10m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/034973bfddbf4701b5982f3c8c792a28 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15e277fa77d0458687c9e869e5960329 : SUCCESS in 19m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aaebae49d50d40908fd49c6289b2080b : SUCCESS in 18m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/360e28355e124cf9adf373a801637040 : SUCCESS in 11m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9b4a8bec55904c458cd1b9490a2c7ef5 : SUCCESS in 17m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/31d1a3d88670488a8a69ff31e73595c5 : SUCCESS in 30m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b8db0b31da2d4354b74e78bca0921d9c : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/52574d0c7ecf4b2796ecabbbcb6f0a2c : SUCCESS in 11m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c347ead287a54a98aa2c359b105e969c : SUCCESS in 20m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/538b56a5728040e78914523e0a384d82 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a6d9f2a393bd4d0c9f21bcc02c356ad7 : SUCCESS in 19m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae7c520a99cf4efb994fff899abee43d : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6508220963bd428ab34a2217f40c605d : SUCCESS in 9m 47s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c8af26c03fdc49e09f30b94f62e83db6 : SUCCESS in 7m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/018fba2b1f954de0b2b3f0654bc6ae69 : SUCCESS in 58m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35bb84b48fe44da19d36dcef91d82cf8 : SUCCESS in 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80afb042e9c542c0999e1c166510bb88 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d662e96ef58452abbb81cc49ed15a1b : SUCCESS in 18m 45s","accounts_in_message":[],"_revision_number":2},{"id":"fd2475856c2086cc0fcee28a919afda463e8d830","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 14:14:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"99709a9ca5de42038719480b85a26f7c4a74a889","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 14:15:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c2a568d830e4792a555aee21fa3d6b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e52fd9b7893a488d9a904f2d21be4ede : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/469df56e6ff54dc7b7428ee750d80a99 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8bfb6b4c0bf8138ff080856a3a912e26d24aab71","revisions":{"7e0c14470e89e8fb4914592e613b5f57872feea9":{"kind":"REWORK","_number":1,"created":"2024-10-01 14:22:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/23/931023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931023/1"}}},"commit":{"parents":[{"commit":"d117fa1dd6b79c2b36e2557cb863e54b5d1462fc","subject":"mpu: use \u0027:\u0027 delimiter for MPUId parts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d117fa1dd6b79c2b36e2557cb863e54b5d1462fc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-01 14:21:18.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-01 14:21:18.000000000","tz":60},"subject":"mpu: make MPUSession.is_aborted a property","message":"mpu: make MPUSession.is_aborted a property\n\nAlso, delete the unused  abort() method.\n\nChange-Id: If908d3dff98629cd98d6f816efda69959eba999e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e0c14470e89e8fb4914592e613b5f57872feea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e0c14470e89e8fb4914592e613b5f57872feea9"}]},"branch":"refs/heads/feature/mpu"},"8bfb6b4c0bf8138ff080856a3a912e26d24aab71":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-02 10:26:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/23/931023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931023/2"}}},"commit":{"parents":[{"commit":"b37f212e34ab8c73969916e4d1ee3cff2f811c90","subject":"Merge \"Refuse part upload to aborted or completing session\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b37f212e34ab8c73969916e4d1ee3cff2f811c90"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-01 14:21:18.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 10:26:21.000000000","tz":60},"subject":"mpu: make MPUSession.is_aborted a property","message":"mpu: make MPUSession.is_aborted a property\n\nAlso, delete the unused  abort() method.\n\nChange-Id: If908d3dff98629cd98d6f816efda69959eba999e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bfb6b4c0bf8138ff080856a3a912e26d24aab71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bfb6b4c0bf8138ff080856a3a912e26d24aab71"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
