)]}'
{"id":"openstack%2Fswift~931060","triplet_id":"openstack%2Fswift~master~I8988c0add6bb4a65cc8be38f0bf527f141aac48a","project":"openstack/swift","branch":"master","topic":"removeLoggers","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-15 22:47:09.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e mentioned you in a comment","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-11-15 22:47:09.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2024-11-15 22:47:09.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-15 22:47:09.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"removed_from_attention_set":{"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2024-11-18 13:11:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35790\u003e replied on the change","reason_account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}}},"hashtags":[],"change_id":"I8988c0add6bb4a65cc8be38f0bf527f141aac48a","subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","status":"MERGED","created":"2024-10-01 20:15:46.000000000","updated":"2024-11-18 13:11:07.000000000","submitted":"2024-11-14 12:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":264,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"931060-removeLoggers","meta_rev_id":"862c89beb71cb12bc204e9217f62da9dc188f01e","_number":931060,"virtual_id_number":931060,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 12:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-14 11:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":2,"date":"2024-11-13 23:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-14 11:21:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"CC":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 21:26:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-02 03:43:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-10-02 08:56:06.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-10-03 18:34:32.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-10-03 22:27:31.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-10-25 21:18:58.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"CC"},{"updated":"2024-10-28 17:59:02.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"REVIEWER"},{"updated":"2024-11-15 22:47:09.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"}],"messages":[{"id":"53bafd77d54cba82c90e0fcd8d15e84e32cea5b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-01 20:15:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"793ef10c152c280d34e6d41f59bd19b287d956b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 21:26:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84928d68c8ec44888e3deb306ff4ec51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2163a1379b48529844147007c70734 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/84c0ff0f3bc8415894696a4a135423d5 : SUCCESS in 1h 04m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ba4fc8b4a45461fbd827e97d08c56ba : SUCCESS in 56m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1ed7896819f478ea3e0b1d6f7b8eb2b : SUCCESS in 55m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7241fa3f43f4c56b751c92d119e7182 : SUCCESS in 33m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04f1b233a0e547eab6a46da1b7d3f0cf : SUCCESS in 13m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c58f33235634f6fbd84341f8dedea76 : SUCCESS in 23m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05832797a4bf4649b9d5618f77b44df1 : SUCCESS in 24m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/158dc1457b074b2dafa40b0d5037d25b : SUCCESS in 4m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a283080dc2d742798381b1bdc51fceaf : SUCCESS in 19m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db4cdec92e354c2aa9b023a2347b5aa1 : SUCCESS in 19m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0fe882009d8f4b41be25a25f11635e1b : SUCCESS in 10m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f4e79fbdfd1144fda9ab615211edaf10 : SUCCESS in 22m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bdada7c5fee64e1fb2d088405de58aa3 : SUCCESS in 34m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2084dcb99a5346ce92be01f6ffaf343c : SUCCESS in 10m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4bfc753fa937423fbf1e5f854a06e0f7 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b80f5ec00abb4db5be43edb02fffcddf : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4239da29ef8b4fd0b4ea4e347d9e13d8 : SUCCESS in 8m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/52817a79c0a9420ea68adc8862ffac10 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/06e4993c4f8241c999ea4f673b1cc152 : SUCCESS in 12m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0b7b94184224ba09c4f9dfd8df9efc5 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26183b4c94a74adf8e669251f2196bfc : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b115c929299d4c4bb3b3d73f8b74ef56 : SUCCESS in 8m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/40b9ff700bb444fca43d3bc52a46c391 : FAILURE in 44m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d76ef6f5cb404d6db827391f56d2fb3d : SUCCESS in 48m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4ed7b66cb60a4935ace83a3e463f4ec5 : SUCCESS in 40m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17e9b005ef7145dbbf3da4d2a58b2a7b : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302fd9f796b9497c942d2383fc56b8f6 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/457dc1c0efef41f2a031540aceae2945 : SUCCESS in 16m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8de90bd247894cdf912ed7f0156d3326 : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4d7a39a0e80dc93fd709d4904ae4c4b73b793858","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 00:07:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b82064cdf445ec9cbead70dec5e292\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26dbe77459d44ca1ac41a6416f67c507 : TIMED_OUT in 2h 00m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fbce101a550450d964e64ed5d078b7e : TIMED_OUT in 2h 00m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2bbbf52c05064878b2bc0a3bfb9e7d41 : TIMED_OUT in 2h 00m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c48ce51ef6b4523bfd5f33a7c269154 : TIMED_OUT in 2h 01m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c8bc87bd474765ec26e9a79a4bd4f5ce5d4a64b4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-02 03:43:33.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bfee5eb3b134b9b440b2dcce4f1a644b2562f12a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 08:56:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e0032febfe8d39883b8726c170a61704f197f5b0","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-03 12:37:22.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1846638ff49da27d9e763e23db8d01fbe7ce5710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-03 12:37:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"00e006d25945c17a6c975df45652b99f94bcc7a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:49:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02782828d4a34213a52b83c208becdc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49c53943e1245779608064127700e03 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/a4d28b2662f7464cb1f0a9cbd7624a2d : SUCCESS in 38m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9aae8b82cf854050960ab83b8f9445d2 : SUCCESS in 1h 04m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d77f97da2d724a77b65e344b003aa089 : SUCCESS in 58m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1f033d30a7b4bd986716ffb3546647d : SUCCESS in 23m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3018072332e4b7a84a93742b9f6e4db : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5dc8700af6d74966bac3ff56454c6522 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/235beaa22fe6449e9f90c24ee36c12fd : SUCCESS in 21m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26e360a07fe5422682e34af73319d2a2 : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a99127ca8b04531971bb9ca4676f075 : SUCCESS in 23m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff2a9bf14b0b4f4ebae3887a9e73c48d : SUCCESS in 20m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ebd6675d854342eca8bc6a80de827bde : SUCCESS in 23m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f147052e4d4f4a4992e7c54f7af1c584 : SUCCESS in 21m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/99e295aaabf04cce92b28b5ed11e6b15 : SUCCESS in 21m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/326146734b3d405c995817b05500e881 : SUCCESS in 34m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31780f51d52d458f900e95219e1728f0 : SUCCESS in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b2bf50ccab9d4373aa5d1ed235312a85 : SUCCESS in 18m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8ae7a17af1154c59bcd760c644e42b2b : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/81de2900c8cd4ee29190a084574c8334 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e5fc6863299148898284803218548e9f : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/602cc458a4654c51aa947ae52bccd8e4 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cfc86cb08d574e3c962c0a7053844ea7 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d774b5b76d464882887660fe392055a2 : SUCCESS in 5m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/711f92c7cf504bfa97b99ebc52690924 : FAILURE in 53m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8dd2b9796d44f0bbfdc417bb364b522 : SUCCESS in 55m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/af0cec07d73d4eb0a59f0ae94b23fdce : SUCCESS in 59m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28603a3d1bf24c6aa1fc8d2e60700931 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964c4164efc340a2b631dd2ddbeb8889 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90a24cb946004d5582c492b3fbf6bbf1 : SUCCESS in 17m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adc470f9f5f94b869cca185e20ea2813 : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d068076f9a65951d8a985bb20d4d79e971d7fdd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-03 14:00:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5870f8739edd4a53da57af7c11d1ca646180306b","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-03 14:01:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ae1901efb102c13707d01b6cf0254ec2d097fd41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 15:06:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de07b118c76942c6b93026ee0e036800\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65df369950624bbea73e022caf333330 : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/14dc3a528b4a458884aa0664f2ab0e6d : SUCCESS in 1h 00m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e52fb80caa34752adb39c6f02383895 : SUCCESS in 57m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4e2290dbe58a4d589a4d271c2d309542 : SUCCESS in 52m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/196826f636b24b60b493c967b4251d71 : SUCCESS in 26m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d80a3c9d7436439396e5fcee39e898be : SUCCESS in 15m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4fdea9a468f469085bc07ed83f95b7d : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57c76e5096d44f8daeb7e067499f11be : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2dd02c6ba62f4709a4d1c798f959627a : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/56a75590c1134cd09e61126ab8ae13cb : SUCCESS in 19m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82c90f7e343b4512bbbe7c03953c7afb : SUCCESS in 17m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fdac1b3318d04af0b011898d94e93383 : SUCCESS in 20m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ea104a35250141c291a43d7898db3778 : SUCCESS in 27m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6f2f97ed92f14c3fba445d4c3750c1b1 : SUCCESS in 32m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e925df587a748fd895a11577aff964b : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1bf77c171ec4413ca27ecd830c3d3338 : SUCCESS in 19m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/00041379921f4510a1ffdb5ba969e5b5 : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bcb39d700fb04410ab413dd99a39da27 : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0a7fb8d164b9461aaac29c2e90fc2500 : SUCCESS in 8m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5d9e9a8ac0c443fca2456002b0c6adc0 : SUCCESS in 17m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bfb433b2858d444db86f171ab0a18166 : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3da786ea2b734da3a722657cae919b7d : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/522dfbea6ed542c9855f07883eecf79d : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e59abff6df394a4b8d317f5158e5bde6 : SUCCESS in 46m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9fb69d044c5d43f09590ed4538f2d59d : SUCCESS in 46m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/efe1e2218ebd41a295e030a50487d199 : SUCCESS in 44m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dedc6cc2bfb444b8f5e677634c14dae : SUCCESS in 23m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba28a22688c14dc48184bf00da261f37 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b53c4a203eab4e09b6814aeafde4e7b0 : SUCCESS in 19m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af8a2a454ed942eb83afe6463bb9869a : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":3},{"id":"de895a8b43607351d596136b37b775b9dc6b5ccf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-03 17:51:38.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7d47515a19c4dbd15075ca5da8bea7097020349f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 17:52:24.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b7096831dfa45a69aa15de04dfd9017\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/782f1c5ca65c44cd9d1695b0253d5c00 : FAILURE in 1h 30m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b03cd03916b94f7197e22a97d71c0350 : SUCCESS in 1h 03m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fee43f6c4f124a11b4b2d8877a6f6a2b : TIMED_OUT in 2h 01m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/41bd15a0055a4e0c809a790b3024e753 : TIMED_OUT in 2h 00m 55s","accounts_in_message":[],"_revision_number":3},{"id":"8c143258150618eeb78a49662e3c9452d1368e8e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-03 18:34:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b2fb5052354a78f4a98c479fadaa1ea13bd3af7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-03 19:23:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"390ba1c04da5a774bee003df96162d8844de0a60","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-03 19:23:11.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"128a277e71c730b9fb4b1fa23cc1bf36e6f3fb9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:29:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5ecdd347da14ff5b578f94bf78e4b8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c524a49c9414a048a88639635684ae4 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/031ab3522a0547738bc5dd501c6155fc : SUCCESS in 51m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f266073ec4646659ba57e764394ade0 : SUCCESS in 52m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a21df7e4aba84d93980b646d66b2562a : SUCCESS in 41m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7dc7b7e1a0d045b2b3089c3080011368 : SUCCESS in 56m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2b9ade1dbad47768e96879a0c2d4fff : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b269d80d38c4d428ccc69c2f1c9175a : SUCCESS in 24m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79595d434f444961ad01b280800067b7 : SUCCESS in 17m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46dfeb1be4824ce282a4670a0223488c : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/11a69840f85d41ccb8944dfc17115575 : SUCCESS in 16m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/617b2628536b4925a9212fc8b4070972 : SUCCESS in 19m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e490ddda8cf40ae9461656314d8e247 : SUCCESS in 13m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9a808b6cc6a9480e90bf0c55a2caae6f : SUCCESS in 15m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/717649a54e3b4da29d2dd95282ef95e7 : SUCCESS in 26m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e8afe83335d43e581786c3225b71036 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bfa6b9347ffd4a8585f836acf765c507 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ea075b850f674f7785a02945edb5a9f1 : SUCCESS in 15m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/73a0c5abc39443f5873918e7fad8e0a3 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4ec08283571841f78e401ce1c701a2d6 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc6b2050b03140a8bb7dabbde0d0e931 : SUCCESS in 15m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/226862d2e0bb49578f32141171974175 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dab72bd34d8d49bab84ece497da51a9a : SUCCESS in 8m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7628efa98dec4e5ab4360a1a06de86e8 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/417302748ef44cef93c9799198fbd2c5 : SUCCESS in 32m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/beb350189233452bb5601228e39eb306 : SUCCESS in 58m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b86d68bb8df47949d4748d159eae4c8 : SUCCESS in 26m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75eb9c4fbabf44f6b51b704a9240a1fe : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab50c952b86444b7a152040c0a5ac89b : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f21b47525249404799b0844bea29f696 : SUCCESS in 22m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b974e9944a6b407cb6112d924ae5f366 : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":4},{"id":"9bf59170455a5961165b53d4fef70c10a134df73","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-03 22:27:31.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"49aea680ddfbc28e77e76e9dbbcb08aeb2743dac","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-03 22:47:17.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2c337ce4796bce3d746c56c64fe74c612fd04b3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 00:03:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a74fdc18124349368afa3bedd578aa4e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27a20ece4b714669b7905c7b29a92701 : SUCCESS in 1h 53m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bf07fb115644d35a70f66a200a86d3a : SUCCESS in 1h 46m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9032bf5164a94989ac06f348dbef2b8c : SUCCESS in 1h 53m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/784c3002921e4ec595cff2347d119f8e : SUCCESS in 1h 28m 31s","accounts_in_message":[],"_revision_number":4},{"id":"cc45dbb4a19f910340ba78e6d6c4eacb6b847753","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-04 09:42:46.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4bc4c5e0c254b492bb1a81b58ad056c9bb1d189d","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-07 14:20:26.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e1aa8b7d5854f8f2769c969a1ed8b8329d2ba8a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-07 14:22:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"bb223b07cb14c6bc8e23128837e705a12ac189a1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-07 14:31:17.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"330a56ce1920aaf61ab9fc2f397d997f1ab879a3","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-07 14:47:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4aad665d9f089618a82e328562edb1c948675247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 15:22:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4d0888513a847b696f7ef93cb25da8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f968a0f7525c4c5da2cf861159a0a2fc : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/ed0b6a925491492b8cdd08163b540a2f : SUCCESS in 36m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecbdf82139434b7b84a455ea1ad198ed : SUCCESS in 57m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f793f08c24d94d928f14b708fd276b34 : SUCCESS in 40m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d1e9ee60a044e63b0e64a81644101bb : SUCCESS in 53m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a663e3509d054d78a8f8eb8ab47354dc : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a4647612d134f718ccc6f781589a0bc : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9c9bbda85ba4441bf4d62c79740c89f : SUCCESS in 14m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/60641efd59cf4c6b97f45986b6a7084f : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a88bfe7b2d4248dc835ca796082e82e5 : SUCCESS in 15m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14265eae0add41e98810358c098e3738 : SUCCESS in 15m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/090c98b932f544b1a5bf59b4c1038c9e : SUCCESS in 25m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7d3177971fef41b1ae5606c564b7c654 : SUCCESS in 20m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/665e2182eb5b453385f5faa9db23a2c8 : SUCCESS in 32m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eefd6148a9a24d9d987881b9f6ad66c1 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6829f6309ed64dbb8dc3efe533400d96 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7783b309fb6c4092a4f18eedb003596e : SUCCESS in 16m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/809dd72fdf8548168268cc0c23c4dbc2 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/25f4894c46e14aae90bf134b21562c41 : SUCCESS in 13m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/73056cda398e445e8a992297218b6374 : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c3d8f607fb414b0bb39f3f408b74336a : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ac41f5c2ca14af78cf2d6568a550e39 : SUCCESS in 14m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e64a15a527e4ab2bf0ea6372d82c3ce : SUCCESS in 6m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fdd86ae5be944907be1a97839d5eaacf : FAILURE in 36m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f91dc225f194442eac59c181789fe277 : SUCCESS in 46m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f277c027bf54190aaf9125a31f58fc0 : SUCCESS in 54m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/185b865f31a84adca34729dc93186d00 : SUCCESS in 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f78436694347bdabf19268a6e37787 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c959d52a86f4427fae6687866d3cc8f3 : SUCCESS in 23m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61d082b142b542149b607c0167377857 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":5},{"id":"1c67d14a345d2ad41e28b681c11a4a9b2c07360d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 16:08:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e9f2e6b849a42d38e27208921027e39\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ffc7e7ecfb74337a9ced233dbaae7f8 : SUCCESS in 1h 15m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc705d683c5e45779a7adb60386b97ed : SUCCESS in 1h 29m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/853539266a3b4613bbf069249b2dd712 : SUCCESS in 1h 44m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/314c8b8f7d9c4bb89db501498dd8cc48 : SUCCESS in 1h 27m 21s","accounts_in_message":[],"_revision_number":5},{"id":"83b357faee4474502bcfba8af96c1e0defea1c05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-07 17:34:05.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5b11f11b6cd26a8595c5664fe22a22103affbfff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 18:43:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b58e2bf9b44a67a1c64b541e7460cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12d855e1b6c34213bd3a7200b2c941fc : SUCCESS in 5m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/9ec8afce8695452dbc84b5c8cba43784 : SUCCESS in 38m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22510605d4ab4a8e9c65a1520339989c : SUCCESS in 1h 02m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1729f7217d0c4ffcbf89199bb663d895 : SUCCESS in 57m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3a6e553e9a24d6c8cb0b5764dec89d8 : SUCCESS in 49m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/578da7aa4061435abad68c8d02171f56 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56f4edb4845c4864b1d46e24d65c086a : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a213bfb018ff425c877a230cbcc5d85c : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1955ad7693ac4149ac2142e8345da6f3 : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4329a13836174f8a97790bee5809c320 : SUCCESS in 20m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc28a6d95b194562b714dc0a5b25963b : SUCCESS in 20m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03709f80d10740bd94956b949e1d3d89 : SUCCESS in 16m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/56e481c7c98649fda6c52a08e6c65144 : SUCCESS in 29m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8c0c0dd9ec6a42c08dac8c8099e66116 : SUCCESS in 34m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24e55591f5404cefa051e9772e37e654 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05e9f5a7a62142ba9a873e2fed2a4ff7 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9a51e03ecb64c75939c69825722c493 : SUCCESS in 17m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2d03a11dade648e7bef68d9f62f9f477 : SUCCESS in 12m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ce89489affdf4bbab8ca032d371acbee : SUCCESS in 15m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/97d5286ff7e845bc9e9954afae66634f : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/80cf312938784a38b6c647548d851daf : SUCCESS in 7m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/82c64a9935794cb79d1990f71984bd3a : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/86c3f6a2bafa4592a162c02c98e677db : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfdb7cc9b18d4a59abb30e8bd87c4ef9 : SUCCESS in 40m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f906bc47c1414ad58fa7b09472e90bc5 : SUCCESS in 28m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b3c251c3d0ae4b889db42c1e82e322f2 : SUCCESS in 44m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/454a9e82881d4fbda23b8c833a60f84c : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9456f2ef5746cfbba627296900ad8d : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b1fb5cfc3b643f28198cce686d3aedd : SUCCESS in 11m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5aae91306cc14ef8b7a9745bd823d594 : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":6},{"id":"df3c941044d833d4dcdcfe5e5d63a7918abadea8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 20:31:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56a2d79ea8b4bc89c3a4cb2815e7943\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69389a0a2966468893f65a7bef711d62 : SUCCESS in 1h 22m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b9c0590885d436ca52cb81dc2d26d62 : SUCCESS in 1h 45m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/689503d9a0c04596a9bfd33dc62d4acb : SUCCESS in 1h 19m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f74a53b53a834d9a9b387328d200831b : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":6},{"id":"0184677abe9dfeb10ea5e70fe150ba9b39444516","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-07 23:09:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9f649bbad26133c22d764df65f8420e36e3119b4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-08 08:38:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"efb4a7d7164ebad5cbce70f483b8f088f744c9e7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-08 18:10:43.000000000","message":"Patch Set 6: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":6},{"id":"65457acc410fc6ee1609e8af4f4de88b08c4e53d","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-09 21:24:58.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"57e6948849f35adde0fd99ecf4caa2bb06ea72ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-10 14:37:00.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"92f1cbabf545e50aaed23729d3c2e42e0c1e6b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:51:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77e7a3a9932c4e6aa0fcc9d339865ef4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2045685e83ad4deda2943ebf8a943a4d : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/2a0c84615e864533aee7f6937bc5cc04 : SUCCESS in 1h 03m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b994ec7312974615828e355104d3a311 : SUCCESS in 1h 04m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b5667a1d743f438e8e730ce5ad115059 : SUCCESS in 56m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9b67c4ceb6941b2a1d801c22d7e5230 : SUCCESS in 1h 02m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0af7fbfb2544f129b303043523b2c47 : SUCCESS in 15m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3c0e8437d3f4d9789a69a5d9eb2abff : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c208638130bb4c8ba084375f98b66f22 : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b876a78ef1f2410cb6ff3f63463aea8f : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4358d5959e0f463d886903d8246b0590 : SUCCESS in 18m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb38b4ddcd444abf9dbb8438b45b7f9b : SUCCESS in 18m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d861cb4e6f604a9d967c4816febc66cf : SUCCESS in 19m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/854c978b89034121ac896781062f0d44 : SUCCESS in 27m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d2e1926638ca4de6ab0558333d8cfb96 : SUCCESS in 27m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c265eef908fa41e1975e8d131e0ccae4 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f605b523bafa40fab5268540f95a5ccd : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dff78c97c231497fbbf002b05d77d6a5 : SUCCESS in 13m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/98121460b2a3406ebefd9168553a524b : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/505098800da24e3d8927009137722452 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2a98b7611a874656b596566900d7603c : SUCCESS in 15m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/effba13d5a874b06a3ee04ea59fb9f57 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/494da2c6845b4ba6ae3507d4af639198 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c639466afae14e8686f8aaca3e4c871e : SUCCESS in 5m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce36cf3bc3164f38a8d27a0be161ae8f : SUCCESS in 47m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de26d1eefb344758b7c07dd0f9efa297 : SUCCESS in 53m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ec6417385284a85826174a55fa1e1da : SUCCESS in 45m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2887960072c243db8c109621535a8955 : SUCCESS in 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff681c7d0564fdf9124afeec393c692 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/566cb3aa75154398ab2149183ecef27a : SUCCESS in 17m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30a364ccb5c340fab5b858fa72f64c21 : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":7},{"id":"2499d4c96758035383a9227acc728ddbd1131e55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 20:35:21.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b7bf681a7748d08b270ddbb9771994\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04b30020a92046eabe3bca8ee31952f2 : TIMED_OUT in 2h 00m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f34270165f3410795eeeecccc544277 : TIMED_OUT in 2h 00m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e65b502396564282b66832ede06118e4 : TIMED_OUT in 2h 01m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc5db4b034904a99b20cd259a462639a : TIMED_OUT in 2h 00m 57s","accounts_in_message":[],"_revision_number":7},{"id":"ff8d902cb225a263ab8f03ed4ec6407519614240","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-14 13:33:44.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"de202f7659ca92b94e1ad584ec672e1bc1e7716b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-14 15:56:38.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"09a586aefce0548bb7e3791300f5baf19b15f379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 17:05:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba056e9f0c6415a8fd0c5cc00df4d87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aeeb0353168462a8761c5e825d98fa9 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/be460a56401347a19a5263f4615bc16f : SUCCESS in 59m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15415786f1394ede8b14d64953bbfb83 : SUCCESS in 1h 02m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f27592f71d5a4336a566a22abb0faa94 : SUCCESS in 44m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/240787177a914242ae7da1d58d99d80c : SUCCESS in 41m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a34dfb36af9f4ee88751f13c86573ec3 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66df7003081a48cf80e880112df500b9 : SUCCESS in 13m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3397bb9245b8411997a6ade495945042 : SUCCESS in 14m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7f93ab6477ef4f71b4001726da51192e : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e135447fcaf143bf9e305d7106f332cb : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bdf17597ea2d4caf99e0ec4d5a48d06d : SUCCESS in 20m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ccdf45d6f5dd47b2b75f81a0b618ad31 : SUCCESS in 17m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4ff9889010e048a680365c33a43e634a : SUCCESS in 17m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/61fd0ac2717b4620a57f2d17c82e0c34 : SUCCESS in 29m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/44d7f5040dc34ca08c17ffc0635ae827 : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b817b5f60f884c1bb545c9f85fb9b9ec : SUCCESS in 11m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c61b40bc26642c69021f1d771381b13 : SUCCESS in 15m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/07663780e05446e0bed91bcd7d6aaae4 : SUCCESS in 13m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d8e95f83ae5549219d96ba648ec6f258 : SUCCESS in 12m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1d830357c6f942a5947269fa28557f53 : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/203075302e8e45fcb6c3727dbae7646f : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a1955fe09154b10afe6359903276f93 : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7fa2998fe9d9423fa109a1f2f4d048f1 : SUCCESS in 6m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/670e2ed39a204732b10767c44b096cd7 : SUCCESS in 37m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c1fbc997d35b4cb49c37e6f7b3bff6c3 : SUCCESS in 46m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc37dae93f084b63856886aaee0ef8d2 : SUCCESS in 56m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f521cf7de096441f90b3206dd7037209 : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c366e470cd4059909936b884dba7db : SUCCESS in 3m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bae58005085a4f0f85ae19e97a139aed : SUCCESS in 11m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89f24a99e8294d52a4f1532ddf16b2cd : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":8},{"id":"7947dfe58796ba0f84dff50b387f4289f58e83cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 17:35:32.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4188d13a1c3941438ecf49d20980ca84\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc1ceaf930394937906448bdbedc2097 : SUCCESS in 1h 38m 24s","accounts_in_message":[],"_revision_number":8},{"id":"8d8c2d3e81ec63e4f67c5a327f6aaaab18c468c8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-14 18:06:35.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"47a051b0471acec5ccc9b1108466a8cfc9897e64","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 10:55:19.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"cd12a340fc191306330d5c053bcd7159f2e87a9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-16 15:37:23.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"e1423cffbf683069a71b2a752187f51b908703b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 17:04:19.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae5e41ac9252420ebddbc4461de3f2c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31531248c2e448b88f4016dcdd79450 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/d8c22fb882d44b348e628ab0774683fc : SUCCESS in 52m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20bef79bb4254d839b1c01d20bc1c0dd : SUCCESS in 41m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5793ccaec1e7437dabd7b096432cbe66 : SUCCESS in 42m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3d29d52e69c4ca3b725229ea2cbb537 : SUCCESS in 40m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bacd97480bf04e97bc0c16acada80fbd : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/440996161ecf4f798c74fd8a16bdb667 : SUCCESS in 12m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/255e8a90ad1f4c6f93e437b42c526e2a : SUCCESS in 17m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a0273996206043adbd57c057f3d9b5d4 : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14a1e19bd17d43119fb9ad370332c3c7 : SUCCESS in 17m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f22660323e5f4f988f6914332efb8449 : SUCCESS in 18m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d2cd60b05b594532b820f9339ef0b906 : SUCCESS in 20m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9b50dbaeebf14f53acd16d599543f06f : SUCCESS in 20m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3fee1733cc2942b482439a2baf33cbff : SUCCESS in 24m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59cf7025b82e433c9c0c11e18440488a : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/570724fedc1c4933a4546052f1cfc6b4 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/479bfad9b3d74554af1eee6decda5092 : SUCCESS in 18m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/886ba58c34dc414a884a4f0bcd5c4786 : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f029a3868e514b56a4a04b5f7be3efc2 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0f553cffc8d048f4b5de810bc90bf744 : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ecffcdd1293c4a098de1f7bb139a3ca2 : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/33f8d72900054b51845dd30499c500d2 : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/23b490edc04146a69f2ca4b369aeb47e : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e15084f61f454c67940025bb1924e1e2 : SUCCESS in 56m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0501285488164fdea8c5397a41598e84 : SUCCESS in 56m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4faa7ce857274cd08f3f1306b8c0f38b : SUCCESS in 45m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a7eb81db30e429cb89cd2f53ff0a474 : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f83b60867b4c738371aecda6fbe401 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b02bba46d6f4dad9f4c7b6bfabe831e : SUCCESS in 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9d3eaa3100f46e0b8cef70533bd3c83 : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":9},{"id":"6093d4025a16d26aad276c5de1f099d588ebd2d2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-16 18:04:35.000000000","message":"Patch Set 9:\n\n(10 comments)","accounts_in_message":[],"_revision_number":9},{"id":"052c084296422c5a05571589bff6fba3890f2441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 21:51:34.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56521629d0ed406b8d66e58b3dd1f9fa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e48afb50d6f4b0e8b4a5ef8163ec48e : SUCCESS in 2h 44m 45s","accounts_in_message":[],"_revision_number":9},{"id":"0f855a69abc83e0ab5e4471750f85d211752a067","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-22 13:28:01.000000000","message":"Patch Set 9:\n\n(14 comments)","accounts_in_message":[],"_revision_number":9},{"id":"07176a8e98209a54b4d717eaad6255361f4e741e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-22 14:24:33.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"c735fb858d1872929c7892014e5fa5c93738b697","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-22 14:40:09.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"825dda63b88dff8ee11ec36ed8b063e053a966a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 15:34:51.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc54eea9f8bc4bd1a0b03ef10ff6be43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff41327a29244f8ab5177467294b3836 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/180a84fcaeb54d43beb4e097d1d1664b : SUCCESS in 1h 03m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6bcea4aa61a1467aafebbdb6b8cb9d54 : SUCCESS in 52m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/64d9e0cbdcee4f16a7d04a286296defc : SUCCESS in 50m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a93c174bb1e4862af7ad7b3c78f3f6c : SUCCESS in 53m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/161836368a184f369b04d0499b06b2d4 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5dbe279fdecf4482999891d3df465b16 : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/790b1295e34d49d980d24e552e9a38cf : SUCCESS in 11m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a5cbafc82fae4854902210bb0ef20618 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c6f5266f7fc14de1bdd083035fb3d357 : SUCCESS in 20m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/895bf6d2415c4b8ca6288d81bbc48c86 : SUCCESS in 18m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/adba7ac240184a9fb60d584af194f64f : SUCCESS in 10m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/656aacdeb41d479e9571a51e6bd44527 : SUCCESS in 17m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c5966d2d7b714cefbcda193aca8528f1 : SUCCESS in 23m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a16d141224664accbf6cd1ba9cda564b : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4573a141dd0b4c0fa4f35e4cf3e6c967 : SUCCESS in 23m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0acc0f9bc3494d1ca694039cdc2c52a3 : SUCCESS in 22m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/673e2dae44bc4543a099b702fd877d86 : SUCCESS in 12m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/229ec96624394c969b524ea45cac3397 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b8ee8bc006be49ceabfd690b9e6da21d : SUCCESS in 19m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ec73f720c57d44eb980218de285c4b71 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/697b2aa13f3c430dad43b4cd2f0235e8 : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/209eb6dc8620443190eb1389bf74d03b : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b147e976708c49d79ac5af04058cb278 : SUCCESS in 52m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6df72e00cedd4e7b899543099fd70508 : SUCCESS in 46m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0839905532d447c5a7c71cd3bcf8964c : SUCCESS in 29m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7236d6e187d24a85aa3bb2705eda6896 : SUCCESS in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc89c3dd7a0422ebe30565c714f7923 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41b166d399ab49b89abecc33d7cafc07 : SUCCESS in 16m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4e9dfca21094c8cae47bbbaa67bdf17 : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":10},{"id":"db0a4ab4db1bd13e80507b2dacb373362d189805","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 15:47:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30dfb52b882240579bf1cce6192004e2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/13eae9e908c84bfea13823b50b18ddc8 : SUCCESS in 1h 21m 35s","accounts_in_message":[],"_revision_number":10},{"id":"8151d0d939fdb7de27b23a5b1fbd78041aeefe88","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-24 10:57:30.000000000","message":"Patch Set 10: Code-Review-1\n\n(25 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6b69ae71f85d2365ab1349d926b06d2a76cf709a","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-24 15:37:33.000000000","message":"Patch Set 10:\n\n(23 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a3d24227f7455c1de6919b38c913bb06db74eb3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-24 17:04:57.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"2aa5718fbe88a4a94cde5d8db0e1e1c5884937e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:19:15.000000000","message":"Patch Set 11: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b343f35a52f64f0eb1242b9928786b4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c88ac595847c4bc8b446eca8d9d76f7e : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/8f76025a8b9c408d90d43ca7f96f6746 : SUCCESS in 57m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/364876127b704cf6b2a8b3d021f4578f : SUCCESS in 40m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/428de27817244d1d8c50b075792cedb2 : SUCCESS in 43m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/065dd4645e234e70b26691ef837d951c : SUCCESS in 59m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd999765f5b94b608c080880855644a6 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1aef3601a8614f678459cd85571f3d0c : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1706f5c7bcc04cbc84b83c17d43c5da7 : SUCCESS in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b1d1efad5c14edcb4e9e44594f0463a : SUCCESS in 8m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/385664690f034e9f849cd2ce81dd54b0 : SUCCESS in 25m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c5d23115293447daa997deb174e3fe0 : SUCCESS in 17m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3737ea91f7294ad5bc38b0ad46183d72 : SUCCESS in 18m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4d300da0a9e64f808a5ae1e872772f97 : SUCCESS in 22m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/546a930416cd4908abfb484387e92703 : SUCCESS in 21m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70893aebe6f046febc8c28809af5a4fc : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d909c4c463c44250b046f5d55aadbac6 : SUCCESS in 11m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/769b53772aca4983a03c73e87a707c87 : SUCCESS in 16m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0bdc193464eb4372a6e973a097e740da : SUCCESS in 9m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/19026f7b482b467ea6b2836431112864 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a7881c05dc684f8baad06ece49c8f275 : SUCCESS in 9m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/be571c8cf41640faa744542832711cf9 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93969985e3f84543871fceb13b38b04a : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0e556af8c524759b9e594fc696e79f8 : SUCCESS in 6m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4157cd90e0194ebe8182a0cab56cd4d0 : TIMED_OUT in 2h 00m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f57562a4ced24376a585c35e2e64c5f8 : SUCCESS in 30m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f8ef097f9edb4e7da934547258f2ae1d : SUCCESS in 45m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/450e6413617049d2b2657fc674b9d4d7 : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc68a11f7e64c24ab082706a4f0a3ca : FAILURE in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/842bba90492c4220a3fb5a21e0ddb60a : SUCCESS in 17m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2d6a146fe7945caa284a06e386e7ab1 : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":11},{"id":"ca4dd45031663ca2f9d09396b0c24e9e381f0bc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:42:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bb5ecbd95f44870be783ee27c6ac5e5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/38eb84560ea644b7bdcbad4f8722dfad : SUCCESS in 2h 09m 47s","accounts_in_message":[],"_revision_number":11},{"id":"0687676e9a94df080739a4b2a16128be82f50efd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-24 20:40:17.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"4e54754c0f2689bb65c6a0b05e9bf1e9179af42f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 21:56:33.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42c6dc279c554d2987c6f62489499bae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09871367d551476da0f31d4f224879c9 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/6d308cf79a8b4e09b86f6827015a5eeb : SUCCESS in 1h 02m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd39cc0dd1394f70aca759f84f801616 : SUCCESS in 58m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/30d0a0be4a084c92b8be82aa4a227b20 : SUCCESS in 54m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/679219f5e3ca45b38f28be6752a91ff1 : SUCCESS in 1h 06m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c001c1a57d443999fafde3f30e50e2d : SUCCESS in 10m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a6f6f2d48384177a4f309a28350c26e : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cef35455b8d4893b382b02ea1375fab : SUCCESS in 15m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/581cf64c557746cda3589d19f2b6f810 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a5005822d2742329f393629bd2f57a7 : SUCCESS in 14m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d920bfc246e74b29b822c716e4d37e39 : SUCCESS in 17m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7c613af27e9a4a0d94e93992f9081ad3 : FAILURE in 2m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/df2ba1e0efe0427bb6dd726de36bbc50 : SUCCESS in 10m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/457f99a82fca49f293f390428357ca0c : SUCCESS in 20m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2779c984cec04f18b10e4990445bebf6 : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5670186116ac4fbea880422cf2241c1c : SUCCESS in 14m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56a16d4307184cdea9dab013a839968a : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3aef18c247b44402a8ab4af9e925197c : SUCCESS in 19m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0a035e5e9e894f2f9705416055218460 : SUCCESS in 11m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/452231cdaa664b3e80934541c829dcca : SUCCESS in 19m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eaff4e8af3d240a1887575ffd451c6cc : SUCCESS in 8m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6049e8312d744ffbaa248b707931adc : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1d24e0cb11704de0b4ac7c676b902182 : SUCCESS in 10m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/105c38cfa82442c99acb73affcd18e2b : SUCCESS in 50m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b853e4bacf154be5b494f0582d178eeb : SUCCESS in 51m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c0f9e53c24c84e26b129654b2ddf146d : SUCCESS in 55m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afd907bb2f684089b25c01800b4e5358 : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06528394028e402199c111b9bd5d6295 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a6c30f0126494afea808f713e2b86905 : SUCCESS in 17m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/670cc83dac8a476d9f868b1d87c88825 : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":12},{"id":"563fdc97c9567f1c7c848380ec4811089bb3b6c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:19:20.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e27cea41c074879829b86deb06b873c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/899a8d6b52af47848ba01da25ead100e : FAILURE in 1h 37m 04s","accounts_in_message":[],"_revision_number":12},{"id":"648b6d0afa334718bd89043f8fde15f97224f9f7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-25 02:36:18.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f2e23e219ed55f2dee7156caa8e39f5e885389af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 04:00:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40ef60f0a05f49eab2dcda3915f50a29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1acb6c9227a4a348aadba3df499819d : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/03c7464a83d84b52aa1c549a04143c94 : SUCCESS in 1h 02m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af77bf6310c44e748fe855753f4fb9d2 : SUCCESS in 1h 07m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da46709d21664acfa8ca9d055499a9f4 : SUCCESS in 55m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d288dc0e12b5460cbe78231d1307774d : SUCCESS in 1h 01m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d83652e2bec4fb3b3ddc4b3bddafd91 : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c4ace8550a94afa8c77a63b9d01972f : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d61dafd564ec40858e2729e2ed5644c3 : SUCCESS in 16m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba220fcf820149c78c1d94dda4202e81 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7af1f8aafbd489cb1ace7b203bc3dd7 : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/814365abf31c42798b4b418348417ffb : SUCCESS in 18m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/764b878a445d4df481c06a7ff86b2a34 : SUCCESS in 20m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e97ea010c97046c387cdc077569f7bf9 : SUCCESS in 21m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/001f202413ad4645890a06a2ea525b85 : SUCCESS in 28m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/860f7b63dd41429195ceea17e00df39d : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cdb5a4443b5d4d7885e6f16758f7887a : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/680b9eea38c44879a446998654cb3163 : SUCCESS in 16m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/53a6ef0960da48789579e84c3b8ceaca : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/16d1596729e3461b843f5b6a9d12a7eb : SUCCESS in 13m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6370e03ebbe248f9b43d11f3ef6e704a : SUCCESS in 15m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5579e2337e8f4cf59bedb51392e055cf : SUCCESS in 9m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/172c0c516b2b4a9e994e7ca2a7318679 : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/20bd8143b36d46e49fa268f938be4032 : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0d57608158d4ecb9b94c9b1423003fd : SUCCESS in 51m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efdeda35e6c249589ab4895236b82829 : SUCCESS in 48m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5932d2aba10644bc8717e6735165eb02 : SUCCESS in 49m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e4888d48c294214aa06ae4b47f7f500 : SUCCESS in 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252edb435f904af59d847770527ead19 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/544a2d8614b74ca8bb0cbab38aafbe0b : SUCCESS in 17m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81ef3f749b1b4b3b959efb64f0739b3b : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":12},{"id":"c7d0fad0e4fb5799187fc4f2b0a2830ca385b951","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 05:40:34.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f403549b65b74e68ac5a38d32b33e50c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8733d83deccb442c9327a5bc12628451 : TIMED_OUT in 3h 01m 00s","accounts_in_message":[],"_revision_number":12},{"id":"a62e5efa6ba33e5c8876e76c86a27f4721c6c39b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-25 11:33:20.000000000","message":"Patch Set 12: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"2822f9b32077fc5dd14d450c4e1e2719140e5594","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-25 13:26:13.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"389b7fff0403eaaf27019935f751b1c38a35f6d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-25 13:30:38.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"422846babc0a85dcf3fee7d347fff4ccd8f83c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 14:34:32.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45e23dd0317b42fdb6cd9dc2f91870a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/128c77f764e543da84457067b8a8ccb4 : SUCCESS in 5m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/3a07ee3cbccb499bba0d9b7c982b2042 : SUCCESS in 34m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1cc54014cc404dd2b58e8e0e145f03ff : SUCCESS in 32m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/081cbf96e9864ef09bbe10eefdef7cfc : SUCCESS in 39m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/999d4970452440308ed6d3fbbbff4153 : SUCCESS in 48m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/033d371363d3439f8b10e391658319ce : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3815ba6e5fd4fecaf8fae7602ee6f6d : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8181256992cd48f78ae4150d3d9bfcd0 : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba9a7a372f6b4dd08a7ffac4db0b3fdd : POST_FAILURE in 35m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba005002e86d4563b06ac8bae00d08cc : SUCCESS in 20m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2dacd0b5028045e8b820786753efb2fd : SUCCESS in 18m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/66b255d83dc944e8931cdf9421563ea9 : SUCCESS in 19m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/57a7353e76904c24889ea5d8d0c16478 : SUCCESS in 28m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1f34d0ba463e4a579f80e8bc8940d6f5 : SUCCESS in 33m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c8e9c1e428294a8eb4e9f2e22cab9d82 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/579e10ad900f47de96780f4e6661187d : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ecc49cfe17e242d39febd6cc3357e0f3 : SUCCESS in 10m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/830c32e5187e4040a51ca067698e3991 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6384f2e6e32b4f809f2039e98231bc3a : SUCCESS in 17m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/57d5dbc2c3074cdab6e85bbe0cc6e476 : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58046667fba244a8b3a83cc41ce487aa : SUCCESS in 4m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2fef7be6ad4347d2a355ea435f902607 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0744cfa225c4742bfe1eb44999e079c : SUCCESS in 6m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e69342c1f17a4cb195fc682e38a5e8e9 : FAILURE in 38m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/adcd68b4d9ce4815badf19efac504173 : SUCCESS in 38m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ee57a45299041b3a2b03425ff3a7cca : SUCCESS in 50m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e02e83bc9da246059342df4fb3303956 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c226d6a5da54905a244339e6e0b9a68 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0c9f2aabe3f469cbd7999aae65b3e10 : SUCCESS in 17m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75d8de918eb341e3a64877782615de39 : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":13},{"id":"7708ebc7680e39d2a40c8cdf88f8de4d2618c71f","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-25 15:18:44.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"2df313bdaf1a52f957b0972ef41f56b550f47790","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 15:26:37.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65c5da2b6ef14d79a810dc5f45974010\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/92286341480e4839ac5f91ad38b4dd10 : SUCCESS in 1h 53m 52s","accounts_in_message":[],"_revision_number":13},{"id":"b46fe3a29ca40faf2e18186f21986d19042cc105","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-25 16:16:35.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c358c3339a7a7a67300d035c2395fc35e790d614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 16:31:14.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd6e1d1e56e14803aa000260f344adb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a927bf822d458bbbf6988e37b4221a : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/f600df68c6494e1f952cbc441ba96ca8 : SUCCESS in 1h 03m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b122958c6f541e69e2a46be6a65ead1 : SUCCESS in 56m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a702e53558a4dde965fc83921e010d3 : SUCCESS in 50m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f94784fc4230480888ca8d3bd741c473 : FAILURE in 35m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4157067264d4cdf9fb758fb37359938 : SUCCESS in 14m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7857edc81e464596a24f22895ad2b106 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70c40513f6f34e80b0ba5177e6c7c5c1 : SUCCESS in 18m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bc19b525c24b4c32ba6cc45b3e797b14 : POST_FAILURE in 34m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d7c374bc96a649f39b899058d5838940 : SUCCESS in 14m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a2f6d17b06be43679f673b9d310c2ef7 : SUCCESS in 14m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0d5324a190e7415090787c203f9ae041 : SUCCESS in 18m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0ef659c3d9184ef6862ef302c37c1c33 : SUCCESS in 15m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3fa4f9f2bf1d47df8841bb400af5abcc : SUCCESS in 31m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f308fb6eaa542eca2199238b38bd4ec : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b530b9e3c4b43ad9c7e6513ac6bb2af : SUCCESS in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6de0f949da804e38be681b492a131b72 : SUCCESS in 16m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7ca3e4f8b889442da32010ad4995ec12 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2cf963b46a4f4333930cb7764bf7d806 : SUCCESS in 16m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/362c657bf3bd40e0b2494f0730c6872d : SUCCESS in 13m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d7b17d7802a4f839723bb084ddc9515 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6005a928090442ef894fed4e161c1a77 : SUCCESS in 16m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/67b5ae0eeae7440bb22ee973439ed4d7 : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3da4ccf450ff49289573ffbbfc66301e : SUCCESS in 52m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/42d92c22b6324b98911e4c37b186b781 : SUCCESS in 57m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d246c514c7934f43a73eee061f8fb024 : SUCCESS in 47m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b3c9904c08147898f6160eb9b5179fc : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d228b0daa563420bb81834f70b0936ba : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/81de1392f64c44ef89ac6cedd79865fa : SUCCESS in 17m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6abbb26b02b04a48b98154dd8bc74427 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":13},{"id":"b481ff37e95b2cf7013d43bb77ab96e04372d819","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-25 17:09:38.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4bb722f0d5364f4e6bdf7403af04c06a78bf5c06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 18:38:06.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a13a2d5de0d403a871536f57decd23b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23b6f8af47454d93b9ba95c181466ce4 : SUCCESS in 1h 26m 37s","accounts_in_message":[],"_revision_number":13},{"id":"9c8bb1e1c5fef0cb32b2d5f73651050ba1c18aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 19:17:14.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/957436f7a8cc4731a6d2bdbac22b5139\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71b4f61ff2ed455d997fc9faf7da90d2 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/7ed2f33d5bf041f88707eaa06d176573 : SUCCESS in 47m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d8c390e91f5849b6b801932f75b929fb : SUCCESS in 44m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5893b4877c0b4f97abcf16a482914582 : SUCCESS in 37m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c47fdbb69fb84892874a30f305511672 : SUCCESS in 45m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68641155978149aeb90933d8030bbfa6 : SUCCESS in 12m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f024654ac1c54ee4bfe34213c2e45df3 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0d30fc7b77a46459e6ee20971072671 : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19f92820daf440e8b63063692690247f : POST_FAILURE in 34m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb5557960a1f42ca8f2943fe232a6df0 : SUCCESS in 21m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7cf7aa95d8d14b959ef1f84b4b8b2dbe : SUCCESS in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/46da66677ef345b88cbc1aa3b4a476ff : SUCCESS in 19m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c8a261214fe8489d96b94ac3591ec93e : SUCCESS in 22m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7586bab95b1145dda9e573d7c41e5320 : SUCCESS in 28m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a267b0aae0245869762c14085ce6b48 : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9e1247f317324bedbb6723ce6a9f27d6 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/33451db08f8b4a679b3c1a34510e167a : SUCCESS in 17m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bb2cee6cc8994d47a52065f455cd9d6b : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e4cd9ee3ac4740faa103798f16f826b8 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/adbf240add464f4b9b368ae36fe08399 : SUCCESS in 18m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f398055ce019480b8c1e5b93c1aad775 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad1263a00b4443ac89c22b6294728c3c : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/581e5dedef87458598a176d494a5967f : SUCCESS in 6m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/40260074b0da47ac8525103e76e44226 : TIMED_OUT in 2h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b8d55fe8bbb4b4294e27b2e66a9b483 : SUCCESS in 30m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1adaaa9df5cb4e89a00089e38087384d : SUCCESS in 40m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c93782e44ad04e368166fb04828b2b2b : SUCCESS in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04014ddb96ad4a78abf466d1da90748c : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/350df30d25ec467a8107c33f66daf0db : SUCCESS in 25m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c212b4556a854706994f0f9b9e2ce062 : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":13},{"id":"34cfe9f459aa9a5538d7e69eb563ee86b4ac04aa","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-25 21:18:58.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"e241617ea8ac994b5ee56f9c2a099d0ea44640d9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-28 09:46:22.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"8f4b0daee54ec28ec3cd61d28db985323ca1ab75","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-28 15:50:18.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1358df8d9d551c516e170f7132ebfaa64b81ce19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 17:00:42.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170556a82b7442919b9f4f1a5a1f0226\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72fc990bc7014d6ebd3aeeb43a836f2d : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7a55711818dc41a798956acc62bef565 : SUCCESS in 1h 02m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee1194dccec04242ac194b230405d5af : SUCCESS in 37m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/efe1f19be9c44378985cee286d6b3cdb : SUCCESS in 59m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03d6dd32079f4816a04e32cc6e4351c6 : SUCCESS in 53m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fce733d3bc864df8a19f6d9ef2a7c909 : SUCCESS in 14m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c028b21eafb43c4b0ca0094de0d8efe : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/747522518d9a466ab5c3f00fcfe854a6 : SUCCESS in 21m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ad9e2f1690d94174981cd2c82ea2eee5 : POST_FAILURE in 33m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/39d81fe9fcf9432ea5d97a3698047ebd : SUCCESS in 19m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0928c853940648bbb357ae6092b54eec : SUCCESS in 13m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3246e0fa66a344119252ad1bfc262e80 : SUCCESS in 17m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/af84200ec2714ed0a661baf2a702f707 : SUCCESS in 20m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bd54b158bf3947c4920492cc71ec0529 : SUCCESS in 22m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ad0618eef4f4724adf1f1314c320dad : SUCCESS in 12m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/17e1b9ad032d46c190d9163940ccf595 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/627d654b37fc467a9505ce5253cc25e6 : SUCCESS in 11m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e8f8b8440fca479e9a65417978c73d90 : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ad48f60f98d0464a9a813eeb87120c8b : SUCCESS in 13m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2ee779eddc004371a9e138d428978ce5 : SUCCESS in 18m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8bbd6962c63d49b3903eed3f8d013da1 : SUCCESS in 7m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/17e16cf82029464e91c131dbb775f4a1 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/102cd57c92544f3fbca77f1e0e2373a9 : SUCCESS in 9m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6295e9ce20b2464c9ca087c75e19c411 : SUCCESS in 44m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d3746cf870f4485af9ff252962f0f5e : SUCCESS in 38m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7d5a05d336642198653444ef7681b1a : SUCCESS in 50m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3d1786f345a42558cdb283e1044d018 : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4996b0d4c2ae4c40acfc04788535a6a2 : SUCCESS in 3m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0bcb2064dab849cc97f46562ce9d465d : SUCCESS in 17m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b70228a8a854e9fa6b8babef2761c26 : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":13},{"id":"603c9f3685c134623e2d40fb7d387e217db83f0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 17:34:19.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee13626506bf4e4bbd6cca841779440e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fac4d8aa60854120a52c2e7912149ce4 : SUCCESS in 1h 41m 48s","accounts_in_message":[],"_revision_number":13},{"id":"0fc72b6e41337e6983b1eac485092a71292bdd24","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-28 17:59:02.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"66a002840b8d5fdc523f5270af9f39416e79367e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-29 12:47:56.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"cdabf4eff0e948d569b2a333802381b951444eaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 14:05:38.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4820460a35c949bf940537cd2f2b5a2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/095d50d271014fb38cb5f24eedc47c1b : SUCCESS in 6m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/6a13696a49c44f1789f8fd80f11da059 : SUCCESS in 1h 07m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/04bf7cbb216e4ed89fe4aebe94be5e96 : SUCCESS in 58m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a215e7f8f4346e4a63cc485035e5b2a : SUCCESS in 44m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dae86a251fb3421c9fd0b7b7d8b0e4e3 : SUCCESS in 56m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f7138601d4f461f8c0606f8fa9567d6 : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de16a82c5b19489b9d0881d0e8712219 : SUCCESS in 13m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c84b2d753d464229ac89b59f007e256f : SUCCESS in 17m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2e814222a0b486fa820a000c0d12c9f : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da0c357015a74e1eb59ad6ac818f372c : SUCCESS in 16m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56a99717b6be46d4b16321a919b34ce8 : SUCCESS in 14m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bce293f9364b4b9495152e580b04edcd : SUCCESS in 15m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6438f2106fba4ad3803ca28cc95d172a : SUCCESS in 17m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a7049e298cb147dd8aed5312c146d0ef : SUCCESS in 21m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1fdeb38ecaef497b981c9c46b9195018 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dbec9d963a00491689e9287982ad0fb8 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ec6ac081bdbe44f990b67523024e7f6f : SUCCESS in 18m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/93be1c08f64f4b32822ad49128fac835 : SUCCESS in 13m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0232cb5ef41049d5bd932d07b8202f3d : SUCCESS in 15m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/154bd97a6c4640478089d4c7b5aca50b : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6596d76798146049d645bea0abe34ff : SUCCESS in 5m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a022d8547d364416814e1d8fcad10714 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7855b86445f34ab3b6bd80248e29b6ea : SUCCESS in 8m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e165ed1c1474462a263b8fe6f6aaf97 : SUCCESS in 48m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f2ea43adbf5464ab2ddfc0dfd11a378 : SUCCESS in 34m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a9787dbde6b48e980307b798b61bf64 : SUCCESS in 45m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5b9c89c61ac4f9ba4090f455e146f2b : SUCCESS in 21m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad912178c23946dc95a290ba1ef4bc74 : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d2bd7b60345643af84b3cee9bdd3f6e4 : SUCCESS in 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01477678370849e0992222a3e8f4bc53 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":14},{"id":"75e7932723ddae9be7aac4c91e1406f2bea986e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 14:53:47.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1298af97356f49df951b3f5f5f2cc04f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3409944b67224877bcd91a2baf6b2539 : SUCCESS in 2h 03m 05s","accounts_in_message":[],"_revision_number":14},{"id":"4726f9a49a315186c6bfb3696128e32e0794aaa9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-05 18:34:53.000000000","message":"Patch Set 14: Code-Review+2\n\n(11 comments)","accounts_in_message":[],"_revision_number":14},{"id":"57eff4c67738ea98538f330a81511d0899a02cff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-05 19:58:13.000000000","message":"Patch Set 14:\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"0f6c085325df06dd873d0d882fa85f646a76791b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-06 11:18:31.000000000","message":"Patch Set 14: -Code-Review\n\n(11 comments)","accounts_in_message":[],"_revision_number":14},{"id":"dc63e755a33aa2fbc252e63f63c9e819717bdbea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-06 22:22:54.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"7e20fb398204d712e47907f015959d28d11bc62a","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-06 22:23:14.000000000","message":"Patch Set 14:\n\n(14 comments)","accounts_in_message":[],"_revision_number":14},{"id":"30a32c61ce9e1b33717e6332f84816e8829f241e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 23:25:25.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/833d5ff642ba4da88d3d778440377481\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1223395bc5734b62ba90fc466fcffb11 : SUCCESS in 7m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/91e724f492784a939917b6778b8ad2c3 : SUCCESS in 45m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6b5cc8736e154287a1d072e8d48ef54c : SUCCESS in 50m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8a5790d85434c469415589c266d52b2 : SUCCESS in 54m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f682287d9634464a8bb0a6dc2876ada : SUCCESS in 36m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47a339672c2041bc94a18bd7753088cf : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4141b26563164634bd61e84ea4318ee8 : SUCCESS in 11m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b91dee218e94ccc8fc4da3acc5e7437 : SUCCESS in 21m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/619d4c8bad3f4bf0b1149d365b20e921 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d760b0cc8236439184f30c0b64aa91c8 : SUCCESS in 21m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/91fc17131f394852b2c57445ef9fd38b : SUCCESS in 8m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/51ec4bed16e847ad8ff2bcdc5978824e : SUCCESS in 22m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ab2bcb34889c4480aee84f90aa12d9e7 : SUCCESS in 29m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28e146f248044807aa35c511844864a9 : SUCCESS in 20m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef554aca7a8d4534a1574becfd759738 : SUCCESS in 17m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/300d74e2bf7a41999d5f968a433b3c59 : SUCCESS in 9m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cfadd3e6abf24fb997fd437da833372f : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1abeb62e0ad549699fc914f65f2a054f : SUCCESS in 14m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f6880fc3b975497386e70d2e5a68362e : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/90d56cc7067d43c5ba67218494bf3976 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/acff3f57dc6040b1a74fe8dde428bc8c : FAILURE in 13m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a3312c35745848ceb64c76a87d66ad43 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e959102458af4b23af4776c3cb317a38 : SUCCESS in 51m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/981ea489ba634ea08039e34952c51164 : SUCCESS in 52m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1bbb381c87c49eabd0ba7bf7b827e1b : SUCCESS in 34m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de45ccdce8044753b19c06fe72eec569 : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4058e5582cfa4f9fa7953c7f2f5ff452 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cb6926ac2db4053ad24fae588108af0 : SUCCESS in 25m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b17f9410882440088b4a87f8f9f0c852 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":15},{"id":"a29ad043e81df71badf28097f33279f037ace830","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 23:43:14.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0874a73d984c432e8d512abe5788c6a0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/685a4a4aac274861891d6671cba9310f : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":15},{"id":"585cbbdd51ab2e389f32967aaa58b92b7649c41f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-07 15:26:50.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"8530c53a922584874b4f82a20f620204d89a763b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 16:45:50.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a599eabd3fda471a80b73883dd3971b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c0d4adf974475487479a433bc98f60 : SUCCESS in 5m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/d35eb72014d543a684d531b4bac85044 : SUCCESS in 49m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e9f705259b84e0c93def4c2656836bd : POST_FAILURE in 58m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/41c96d121bc540db854ec3d62857387e : SUCCESS in 51m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44c29585229f4a0a8cb44a377c36d324 : SUCCESS in 59m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/acf6007895fb4d008f000678e2f792ff : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f4adefa503c44589b904fb1b7b6dab4 : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a386983741743fdb84d1a47bb5a0ce2 : SUCCESS in 19m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b7b6052f2f9645ecab5e68791c94b254 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb66e7e9443e4da68b353bc02b3ea3f3 : SUCCESS in 20m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/575b6e964c804fe29815c8441f1a1808 : SUCCESS in 9m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3991a2edcfeb48159c50d715bd124ae5 : SUCCESS in 10m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8acc18d8fbc74ff797573eb9691ac3b7 : SUCCESS in 13m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/875bff1ae6d64a38b828c6f0b7d2701c : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f36dfd81de824c6bba74bf6b55b3bf0f : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57244bf7bbef4f46ac588d595e889db6 : SUCCESS in 16m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/382c3313bd5e42bd9b47a49b20f4e7a7 : SUCCESS in 9m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f04eeba69d2f412490cab48ba03bf441 : SUCCESS in 9m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/97dd285117124a51a2477773b86a3135 : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/591d959f47d7492f96eabc6a156e0bbd : SUCCESS in 5m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f49b178869954a9d8b2697df98aefc65 : SUCCESS in 13m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d5a97de40f4341fc8523314fd1324cf1 : SUCCESS in 4m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/16c767dd75e2425381ae9d3b79813098 : POST_FAILURE in 56m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e85c9696b9e240b69b57775910fc1f3c : SUCCESS in 49m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f8fae3331b84a4a81bf70ced6295958 : SUCCESS in 56m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de9ccc1d0ab547189fdc16b989b6b44f : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb8cf4b1aa96410c9dd709257868f55a : FAILURE in 3m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b6c06b958fc4eb4b716039f65d50509 : SUCCESS in 11m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b11ba3d9cc2493eac585c7609425ec9 : POST_FAILURE in 53m 45s","accounts_in_message":[],"_revision_number":16},{"id":"8964640b29bd9cc57e469eced7b6cd86a7afd0e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-07 16:47:19.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"84f837f754e460ffd7ebe2860b2c21575e921777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 17:58:03.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd977889d4874a4d8e2d906e1b5601a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9cab77ec8374a2f9b58b5562d1461cc : POST_FAILURE in 5m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/09a0a076d5d44fd4a3c1fce416314559 : SUCCESS in 35m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6ddfddcf8094be4ac08a75c2b371f19 : SUCCESS in 38m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5cc645afb43446149cf1f036ce315370 : SUCCESS in 45m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79608aba9c284ca1a4fb9a1186c49a5d : SUCCESS in 50m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe5c57a909634282ae14e63e3ce891eb : SUCCESS in 11m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20f8e3c061794c6f881c61461cddc39d : POST_FAILURE in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c91404bb94d348148d7d420d2a1142f1 : POST_FAILURE in 13m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/145e8099fc8c4d4ba6e6b49627d38a1d : POST_FAILURE in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17dc3d590cb24cc59a364c5596b87149 : SUCCESS in 17m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f573ecfc6fff430db2b4314a496e79d7 : SUCCESS in 15m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd2f40f146fe49c493d79913811bc65b : SUCCESS in 17m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2dbbcf77e0324131b0bd7a3355509262 : SUCCESS in 31m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/546923a3c22b4024a89d2dc9e4a4115f : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb9d997380ac47f9a3be28c7122bc035 : POST_FAILURE in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3bce86c6651a453e9d33ddd3c38bd1f6 : POST_FAILURE in 17m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b60537429d81477cbf01cd001960bb2d : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f7eea115f2454dd7987a8da1d58cad09 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a80429ba10984614995509a7d48259fa : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e33b682bfef4de5b242d99243add551 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6981b43bd2724d0c899f45247ccb1ca2 : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1822ff33b47341b191453e8275830164 : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7db18c21b3d49b8b69a5ea6b2a24451 : SUCCESS in 46m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac5bcb9782ab40d2a09ad65d6f2330cb : SUCCESS in 52m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/276fa2c1fea84fc787685a01aaf84c21 : SUCCESS in 52m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69ee9920ee7c422b938c846531d0313e : POST_FAILURE in 19m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a585b9a849e4556b574cf8c5c7c8753 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f6d41de800ce4bda8e1572d757484d79 : POST_FAILURE in 19m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c13fa5b5bc7049268ea8e62c8531a1ed : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":17},{"id":"6f1156231fa8456a6189ed57da87ef28e9260a6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 20:38:39.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63a4d22f3f1846e38ed704ccbb20398d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d191909dc8754eb8a426799ae0a702d6 : SUCCESS in 1h 57m 47s","accounts_in_message":[],"_revision_number":17},{"id":"310e12fb86c4b2b624b75acdade54ec71f23e214","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-07 21:07:26.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"145708d2651ca0734c7e3bf7ce87350a657c427e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 04:35:46.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"e8140b6e6236892242d5f5ff83277f6e99c7059e","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-08 15:00:24.000000000","message":"Patch Set 17:\n\n(5 comments)","accounts_in_message":[],"_revision_number":17},{"id":"9bdae3963c2417ea6cb98d76944908eb7f55a1e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-08 15:12:47.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"e0eef949f5cf394f589a0e0702e091fdc6c35575","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 16:18:15.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"990cded9c6ffc9f5277ec252230a31352a3fb8da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 16:28:43.000000000","message":"Patch Set 18: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b85774ab6ae74fa298d044539c344107\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e0cc56c85184fb086e6db867157a2dd : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0eeecb05710146c88c61385542afd19a : SUCCESS in 59m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6abc381dd7b5431daece2f66978972d4 : SUCCESS in 52m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8a9b44b89444644a8fe48e66920d58b : SUCCESS in 43m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e73ea5c4e07d4e5e8f6f7b1dc2ff953e : SUCCESS in 1h 08m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7b8a30db06c43ff856a849405cf8e9d : SUCCESS in 12m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d189c21dca443b8b80dbb3d14825d80 : SUCCESS in 25m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28b451e500cf408ca71e8a10ee82c644 : SUCCESS in 22m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87d18b594af64932b2cbf39efd972d01 : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c7356bb3c6b94f7a8f60a199fdc1c9d4 : SUCCESS in 16m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/163ab40cae5341d0bca4156eb3782187 : SUCCESS in 18m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7851a571373142a2b3eec634ff2e9263 : SUCCESS in 20m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/91d70891ecb946458536dfa4fa329a67 : SUCCESS in 24m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dce650e088b240baaa6be8be3d4be3f2 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b08d5fba695b46ceb64d95c69c1b598a : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d31a64274e5a40eaa5d3c3f13e3ef3e2 : SUCCESS in 19m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ac42043b3310484faab6cbef681b8981 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/24eae8005b3e425795658501cabae708 : SUCCESS in 13m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/82e61dd3c39f430fa6ea7c707e097092 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e1abd75acc847de9ad7579e0633600b : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3229e18a61f494faf1e575a97785b3d : SUCCESS in 13m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/070b04f343dc475294b81f7272553621 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/38b71de094524b48a51246a200bf90f3 : SUCCESS in 52m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/827a5717e46a4a09a80cf1fcb5b45a98 : SUCCESS in 41m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff5bd3026cba418fa2cbfcdfdc259a39 : SUCCESS in 42m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/081568ea3e9942589967f75d319aaa89 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79747b72adc645c888c76fcf66a5c20f : FAILURE in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/caf001a2edec4d0c94fa1ec66a3adda1 : SUCCESS in 21m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27962afba8904208ab953aed6a02b7c3 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":18},{"id":"893cf1bd91c89a58ca4fa870810078988271d32f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 18:08:47.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/603203a1b7ab425181f9368f21eca0fe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/478d2fbdeaab45ef889da7de9ff2522a : SUCCESS in 2h 18m 34s","accounts_in_message":[],"_revision_number":18},{"id":"e6fc6d0063a746022c39b6d1752e599dae18998d","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-08 20:10:40.000000000","message":"Patch Set 18:\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"5f9ca65b597c49db4f8d566e4bdc32abfb9251d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-08 20:15:52.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"633ab2a316386f8b17607f899a28a2bcbe44fc65","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 21:11:46.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"64410adf83188972ffaeaa03cfe20c4e95e15991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 21:29:22.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5912ce538acc4bc6909a5788daa5e169\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99555ee73ed141659f20c00da758a47b : SUCCESS in 6m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/08d73e68313c47e8b2581f7b88445b34 : SUCCESS in 1h 07m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5e4776d89b4b4f669c01ea102611295a : SUCCESS in 46m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48bbf8c87f404726ba5bc2faf06778c3 : SUCCESS in 44m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0bb32bb163a4029b15135e76b1f75a9 : SUCCESS in 1h 01m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e89195cb1c784fc087157d9151866cf0 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0345b6a1d32a4aab94393e0a212d55f6 : SUCCESS in 16m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b06c145c1754498ebc041ef55fc83c1d : SUCCESS in 22m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bdf48ca0357e41bea860b193852b8655 : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/41ba1f3e6e964a5491f206708477d980 : SUCCESS in 21m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/788e70cf09c4469bb3b8721b22aa6251 : SUCCESS in 17m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0d2c8f032a141d1bee6b9a813d991be : SUCCESS in 18m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e539239f92b741bd9d2c81bdbb422dbf : SUCCESS in 23m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fa1aafbee560480fae4466ed2b2274b3 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa1fee80a95b436798a116143f29e029 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0a689088838d447585a435a633deab2f : SUCCESS in 19m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/65e2faff9c9047038813154a9bed79ed : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/34d6a87bb78e42f8b8fdcb9cd01ec856 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/26420f9d9e34446a89bd87c0f4eb8125 : SUCCESS in 16m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/49624c6f132c42ca8bb6270dfd9ab3fb : SUCCESS in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f4de3e1f7a9407f8ee457a7850fc492 : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f40d2c01c3ff40c59f7d7baa6b15cc6d : SUCCESS in 5m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3dc621f35424fa598617e974466e6a4 : SUCCESS in 40m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9297d3328ea478799dda92af3304ea9 : SUCCESS in 33m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60592f63ca4a4d119ad4eac2b86c5d9b : SUCCESS in 33m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fc4ff963ff5493f88400754e719ca7f : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0127a10b8d894e1aa101065ef36473a4 : SUCCESS in 3m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23f70d460c544a799ab291b03c85934e : SUCCESS in 17m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e85d5c6fc34e49099698ff4122a64fc3 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":19},{"id":"0c06f08086c8fe2c584452f250913b7598244b6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 23:04:48.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a637020817f347e6a6431a9db1b5abda\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d9b4251eb5b4733b460493269918234 : SUCCESS in 2h 40m 40s","accounts_in_message":[],"_revision_number":19},{"id":"4cac1d9db8cca233f8e5b0719b89b26b8f40f1e7","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-12 20:40:57.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"ef000013d2bee0a7e194bb9cea627e63b8c48549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-12 21:09:20.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"50344b49dc44d928599c05c3e25c44fed650660a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 22:02:15.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7c6553047b441c5a95b9ffc39aab273\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/555e64ecb5cb41af8a0ce2eb8d678939 : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/2846dbed76f648d49380228edf38e9a6 : SUCCESS in 47m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b83d55594dd8433d9782fdcf0c2e872e : SUCCESS in 46m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9cbb8ef868ed43028610b920239c2971 : SUCCESS in 36m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba93d6a3d0ab463fbbe37b735d468e87 : SUCCESS in 51m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdd9eddd85c742dabcd599399d50bcd7 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/405e39c514224cd9a00a2043ec51ac5d : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/122ed5534e3944cbbe8a414301963e64 : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f58b37959ba48099bf96c0eddafbefa : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/46379e4fbb634f5098776dabc930cff9 : SUCCESS in 18m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/39624eba7b9a449c8a5755d57fa8df00 : SUCCESS in 17m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fdfa98ae9f7c407ab9cdaed6e1ccaa5e : SUCCESS in 20m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2d3628358a364ac284dc657cf52a5fa4 : SUCCESS in 25m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/42f5e95a0d644399b882841c5fa9ec79 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dcf3cefca5c0445085e419bb982bceb7 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/20c6e06376034a08b2fbea6b1cbe24cf : SUCCESS in 10m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8c369399543b457db66a40b1567ae954 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7553bb50959e4b5baa29f77f4bb4b92a : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/487a4bf1486a47578b0437b1e516f974 : SUCCESS in 16m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/27c1d07d9a9343d8ab4aac78ae579635 : SUCCESS in 9m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6cce883821fe44ccb2c35b91eeb72a8c : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1aee0e7057f048b9b823a0d6870a5b5b : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd37ab40588046aa80ace67718fe7709 : SUCCESS in 46m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/01aba95826794a4a989ca03669abd1db : SUCCESS in 46m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dac5d8313b58441485368dc691372050 : SUCCESS in 38m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c649ee958354e0d9aed32cc2420fe22 : SUCCESS in 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a351bed3e19496c9b76c045f0ee036a : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b16125d3b2594fc9b9245a59bbd51c2c : SUCCESS in 16m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f0557ff9759470bbe3559eaaeec4d27 : SUCCESS in 41m 33s","accounts_in_message":[],"_revision_number":20},{"id":"21d5ba0e54d5405e4b3f91049e44433cc8e95075","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 22:20:20.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1387c29c53c648b4a8fa4f3e5b100035\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fa480731ce2b4c1aa5603231cae1bc5e : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":20},{"id":"05eb8bc68bbc391d9bb147e603f1d329e013f7d7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-13 12:47:40.000000000","message":"Patch Set 20: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":20},{"id":"eb09a1ec757c9c5c93da5fb759e8acc587bd73ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-13 20:41:56.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"0633dd218d70a3d4706277d0a3b0388feae82578","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-13 20:55:25.000000000","message":"Patch Set 21:\n\n(12 comments)","accounts_in_message":[],"_revision_number":21},{"id":"7421ca3480a3b7b14734c91edd7510419ee8276b","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-13 20:55:38.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"6bd439a462e5c46b66e533e7eb8a3ce6267b5f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 21:45:15.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18178638ffbb448ab389bd23e230c9ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b3bad186b1e4aef83c9d55a977ab715 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/1d8ab04d0c714c73a4f8a33709b0e872 : SUCCESS in 56m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ab2911b611649f1bfc709945c33ec2e : SUCCESS in 54m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/abcc00085ec54636840bf4975a5f6e1c : SUCCESS in 43m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41aa36e9db21490e99de746eb5b655e8 : SUCCESS in 1h 02m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7eacf092731464ca105389b18c7e1ae : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c46ee34b4e34ce0b28ed387bd6af37b : SUCCESS in 33m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5569406b4c3d44f8b2a9f7e748fc780f : SUCCESS in 23m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6269ccc03b304117aaa2e6623a5de889 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da6a29f7f4ad4bc5b0e51fb23c835a3a : SUCCESS in 21m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5e165d7778724006806607370f72fcbd : SUCCESS in 22m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/24fe65f11d23466580b99ebca477b561 : SUCCESS in 20m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/42bf105632844935993fda8c74321493 : SUCCESS in 30m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a5085e06de624bbd9c886d9f5058a828 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/359cc315994544c5943a1e17e6f90aa8 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df053c964bd141ddb15c8da9d4cff8b2 : SUCCESS in 12m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2a941a0365da4665b8e3ef2f05698a33 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f7b087c6cb5342e382e30255cf8b10fa : SUCCESS in 14m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fba07d6099d44ec89a610e0288b1b3f6 : SUCCESS in 10m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c72a128031645808c567009698ccee3 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/553d90fc311b4ebdb8155d9d1d6f245f : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce7637c3c9c94fcbbd7a00665e04f4ee : SUCCESS in 4m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/289805c03dee4cc5901f9265ac1c0754 : SUCCESS in 39m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a79704c0ff354e0caaca9fc284b7615a : SUCCESS in 39m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1480098bfc534bef9befc81d01ff91fe : SUCCESS in 39m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c8a16809b8a49e69132d34410481844 : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c27531c8bc445a2b1831bccc403cc6f : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/adbac7a6f6484677ad12e875057d4a0f : SUCCESS in 18m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1473eb79bab643e0b698bea314a379dd : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":21},{"id":"0a7ab655016a18b5c3f263d5b19f52dd47c59610","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 22:35:34.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d898adce02304ed3995a416827aaef21\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a9a3968398524f1b9da91de6437484b6 : SUCCESS in 1h 50m 55s","accounts_in_message":[],"_revision_number":21},{"id":"07d3e67a68b0e36e262b55b34d17af6b3154d16f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-13 23:14:58.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"303d906739b7e2d21df9768a56a16912b3c6f79d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-14 11:21:48.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"77acedd46311c78a27a9cfcd95be322b6094a143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:22:31.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"160eaf985752615c64471287484a9bd214b2fd94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:51:08.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e66cc24a659d4ffaba61b2371f116a59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec825421f66406eab73dc67e056f1fc : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/192a57bb22244e709ad59779e8460efe : SUCCESS in 1h 07m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8e4bcb94d4ed437ca2f52a0f93c058c2 : SUCCESS in 1h 04m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/186cda75b0c447e4af16772f63be2425 : SUCCESS in 56m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08674646dac141699f60096c56dfd24e : SUCCESS in 1h 15m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8948c922799941e286854afa18057ffb : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3442f9326c5b4e9c8f97ce531367bcdd : SUCCESS in 21m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f255c540cc148a0ad393b31d2fc1d1e : SUCCESS in 18m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1e3aceac930e435783d2263515712a45 : SUCCESS in 21m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ccf9b9aa07f64e178d04957abefcab65 : SUCCESS in 22m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bb5aa97d03604ff1b3bba1e451c7d229 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3d4c4b6e36214b89a2c0f4bae8ac07de : SUCCESS in 16m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f16f0a31a2ec4306af1aa853b014348d : SUCCESS in 18m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cb87047250b84ec7be8e92ae5dbcb109 : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a30ae60189c94b178b2ff57dd9da69a8 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eca004bfb9ed4f9fae6f7c83c2c8f3c2 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/43d0502f751e4f0eb5074fdbc1353f3c : SUCCESS in 6m 42s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/104fdbbc33c242b082c086c7fbd6574e : SUCCESS in 8m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9910a02d32ab428899a079df4bd40c0c : SUCCESS in 41m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee5492dc92d84b8a8d27bf0dc83ca21e : SUCCESS in 52m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db69f8172dcc4531abd076ca43d3d194 : SUCCESS in 33m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dceff3bd8ba540798cc286026c8ad0b7 : SUCCESS in 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce89446441045c9af8d3d5718d5461e : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d145089a6ad45acbd9f73077908ce45 : SUCCESS in 11m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15b6a5ec7b2f4cc39ff8fb98900fd298 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":21},{"id":"b8ea5aab42e1f6456e6b7bdce844d0aba4f52843","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:51:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"28c9f5118796b3cd1c2efdb84c83f3283ae5208d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:52:11.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa504dfd8710432699d16e69df50c03e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5420e4b524914d0c86fe7afebe6e7f73 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0efdc1dde032430e940c386bb43cc236 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"c81de27f54ec69477cb11ee9f3af3dd0aaa264ca","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-15 22:47:09.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"862c89beb71cb12bc204e9217f62da9dc188f01e","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-18 13:11:07.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"f88efdb4df37577c687f68b2cc3712465955c20e","revisions":{"39f1105849e4efbd83bcc2a63c22453e14689185":{"kind":"REWORK","_number":1,"created":"2024-10-01 20:15:46.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/1"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"subject":"Kill inheritance","message":"Kill inheritance\n\n- Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n- SwiftLogAdapter is now Swift\u0027s LogAdapter\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39f1105849e4efbd83bcc2a63c22453e14689185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39f1105849e4efbd83bcc2a63c22453e14689185"}]},"branch":"refs/heads/master"},"7f72ae7d7c4940bdf86ede590c2805820724beed":{"kind":"REWORK","_number":2,"created":"2024-10-03 12:37:42.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/2"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-03 12:35:34.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter Rename LogAdapter to SwiftLogAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\nRename LogAdapter to SwiftLogAdapter\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f72ae7d7c4940bdf86ede590c2805820724beed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f72ae7d7c4940bdf86ede590c2805820724beed"}]},"branch":"refs/heads/master"},"0c414a8411793fd95cdc92bf6e8196a49bff159b":{"kind":"REWORK","_number":3,"created":"2024-10-03 14:00:25.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/3"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-03 13:59:59.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nRename LogAdapter to SwiftLogAdapter\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c414a8411793fd95cdc92bf6e8196a49bff159b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c414a8411793fd95cdc92bf6e8196a49bff159b"}]},"branch":"refs/heads/master"},"ee6f1d15d08ff1e12e7a86ef2eebc76b9481aa69":{"kind":"REWORK","_number":4,"created":"2024-10-03 19:23:10.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/4"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-03 19:21:36.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee6f1d15d08ff1e12e7a86ef2eebc76b9481aa69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee6f1d15d08ff1e12e7a86ef2eebc76b9481aa69"}]},"branch":"refs/heads/master"},"efa13a89207a094f4dd6d3ba4dea7acababadfea":{"kind":"REWORK","_number":5,"created":"2024-10-07 14:22:30.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/5"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-07 14:21:57.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/efa13a89207a094f4dd6d3ba4dea7acababadfea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/efa13a89207a094f4dd6d3ba4dea7acababadfea"}]},"branch":"refs/heads/master"},"f81c1cda3c4210ac8443f0fa391c8ed3a1cd9861":{"kind":"REWORK","_number":6,"created":"2024-10-07 17:34:05.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/6"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-07 17:33:00.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f81c1cda3c4210ac8443f0fa391c8ed3a1cd9861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f81c1cda3c4210ac8443f0fa391c8ed3a1cd9861"}]},"branch":"refs/heads/master"},"41c5d740d4271920ec3daaf29cd76d86399a604d":{"kind":"REWORK","_number":7,"created":"2024-10-10 14:37:00.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/7"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-10 14:36:34.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41c5d740d4271920ec3daaf29cd76d86399a604d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41c5d740d4271920ec3daaf29cd76d86399a604d"}]},"branch":"refs/heads/master"},"8bd333ba690887cda6da623c35356e75d82030be":{"kind":"REWORK","_number":8,"created":"2024-10-14 15:56:38.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/8"}}},"commit":{"parents":[{"commit":"a427d2754f2b5ff48d0fd72e08824369e035c84e","subject":"Merge \"trivial: Default value for EUCLEAN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a427d2754f2b5ff48d0fd72e08824369e035c84e"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-14 15:56:07.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bd333ba690887cda6da623c35356e75d82030be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bd333ba690887cda6da623c35356e75d82030be"}]},"branch":"refs/heads/master"},"c2d5e5c36c1fedc6093c9c47e8a6353f85292391":{"kind":"REWORK","_number":9,"created":"2024-10-16 15:37:23.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/9"}}},"commit":{"parents":[{"commit":"a427d2754f2b5ff48d0fd72e08824369e035c84e","subject":"Merge \"trivial: Default value for EUCLEAN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a427d2754f2b5ff48d0fd72e08824369e035c84e"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-16 15:36:04.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nAlso adding support and tests for custom log handlers\nRelated-Change: Ib04b12939dcac7e4ad6453dea9795682044c6ae0\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2d5e5c36c1fedc6093c9c47e8a6353f85292391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2d5e5c36c1fedc6093c9c47e8a6353f85292391"}]},"branch":"refs/heads/master"},"f89e468a9161fd031c4c7e0fd0e42cf02b13084c":{"kind":"REWORK","_number":10,"created":"2024-10-22 14:24:33.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/10"}}},"commit":{"parents":[{"commit":"a427d2754f2b5ff48d0fd72e08824369e035c84e","subject":"Merge \"trivial: Default value for EUCLEAN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a427d2754f2b5ff48d0fd72e08824369e035c84e"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-22 14:22:33.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f89e468a9161fd031c4c7e0fd0e42cf02b13084c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f89e468a9161fd031c4c7e0fd0e42cf02b13084c"}]},"branch":"refs/heads/master"},"78392a927e3e9068178d527d4113d402400a7a60":{"kind":"REWORK","_number":11,"created":"2024-10-24 17:04:57.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/11"}}},"commit":{"parents":[{"commit":"a427d2754f2b5ff48d0fd72e08824369e035c84e","subject":"Merge \"trivial: Default value for EUCLEAN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a427d2754f2b5ff48d0fd72e08824369e035c84e"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-24 17:03:43.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78392a927e3e9068178d527d4113d402400a7a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78392a927e3e9068178d527d4113d402400a7a60"}]},"branch":"refs/heads/master"},"791c831dbc0bb3ab29a527caee0d59a65c6ee331":{"kind":"REWORK","_number":12,"created":"2024-10-24 20:40:17.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/12"}}},"commit":{"parents":[{"commit":"a427d2754f2b5ff48d0fd72e08824369e035c84e","subject":"Merge \"trivial: Default value for EUCLEAN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a427d2754f2b5ff48d0fd72e08824369e035c84e"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-24 20:40:00.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/791c831dbc0bb3ab29a527caee0d59a65c6ee331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/791c831dbc0bb3ab29a527caee0d59a65c6ee331"}]},"branch":"refs/heads/master"},"eb39ebd2a6b5fedb11fdbe6f39823ed972707d18":{"kind":"REWORK","_number":13,"created":"2024-10-25 13:30:38.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/13"}}},"commit":{"parents":[{"commit":"a427d2754f2b5ff48d0fd72e08824369e035c84e","subject":"Merge \"trivial: Default value for EUCLEAN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a427d2754f2b5ff48d0fd72e08824369e035c84e"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-25 13:30:22.000000000","tz":-240},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb39ebd2a6b5fedb11fdbe6f39823ed972707d18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb39ebd2a6b5fedb11fdbe6f39823ed972707d18"}]},"branch":"refs/heads/master"},"ec8b868bb1368cfaaaaa8d0bf8699161e26738eb":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-10-29 12:47:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/931060/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/14"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-29 11:22:46.000000000","tz":0},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter which supports all the\nfeatures swift\u0027s servers/daemons need from our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8b868bb1368cfaaaaa8d0bf8699161e26738eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8b868bb1368cfaaaaa8d0bf8699161e26738eb"}]},"branch":"refs/heads/master"},"8e518168eff200d7e860fe949d990ed202c330a8":{"kind":"REWORK","_number":15,"created":"2024-11-06 22:22:54.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/15"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-11-06 22:21:15.000000000","tz":-300},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter (previously LogAdapter)\nwhich supports all the features swift\u0027s servers/daemons need\nfrom our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e518168eff200d7e860fe949d990ed202c330a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e518168eff200d7e860fe949d990ed202c330a8"}]},"branch":"refs/heads/master"},"56223e7208865f95ccc534ed7e7e5710a9d4c734":{"kind":"REWORK","_number":16,"created":"2024-11-07 15:26:50.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/16"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-11-07 15:26:31.000000000","tz":-300},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter (previously LogAdapter)\nwhich supports all the features swift\u0027s servers/daemons need\nfrom our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56223e7208865f95ccc534ed7e7e5710a9d4c734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56223e7208865f95ccc534ed7e7e5710a9d4c734"}]},"branch":"refs/heads/master"},"96892c27a673bb88caa8f28d30d6f0cb4cb07ff4":{"kind":"REWORK","_number":17,"created":"2024-11-07 16:47:19.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/17"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-11-07 16:46:56.000000000","tz":-300},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter (previously LogAdapter)\nwhich supports all the features swift\u0027s servers/daemons need\nfrom our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96892c27a673bb88caa8f28d30d6f0cb4cb07ff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96892c27a673bb88caa8f28d30d6f0cb4cb07ff4"}]},"branch":"refs/heads/master"},"a68d7df386fd5b781c6902c2803679e12e921da4":{"kind":"REWORK","_number":18,"created":"2024-11-08 15:12:47.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/18"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-11-08 15:12:26.000000000","tz":-300},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter (previously LogAdapter)\nwhich supports all the features swift\u0027s servers/daemons need\nfrom our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a68d7df386fd5b781c6902c2803679e12e921da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a68d7df386fd5b781c6902c2803679e12e921da4"}]},"branch":"refs/heads/master"},"2d6e9be2064f74254a9123e7011755963f6c5ae5":{"kind":"REWORK","_number":19,"created":"2024-11-08 20:15:52.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/19"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-11-08 20:15:38.000000000","tz":-300},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter (previously LogAdapter)\nwhich supports all the features swift\u0027s servers/daemons need\nfrom our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d6e9be2064f74254a9123e7011755963f6c5ae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d6e9be2064f74254a9123e7011755963f6c5ae5"}]},"branch":"refs/heads/master"},"44fea4420e517b398bf02b9afc55fbc948a0da67":{"kind":"REWORK","_number":20,"created":"2024-11-12 21:09:20.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/20"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-11-12 21:05:22.000000000","tz":-300},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter (previously LogAdapter)\nwhich supports all the features swift\u0027s servers/daemons need\nfrom our logger instance interface.\n\nRelated-Change-Id: I408ddb69e7f9122905c66b94b6b63ba9628eb4d9\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44fea4420e517b398bf02b9afc55fbc948a0da67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44fea4420e517b398bf02b9afc55fbc948a0da67"}]},"branch":"refs/heads/master"},"f88efdb4df37577c687f68b2cc3712465955c20e":{"kind":"REWORK","_number":21,"created":"2024-11-13 20:41:56.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/60/931060/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/931060/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/931060/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/931060/21"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-10-01 20:14:09.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-11-13 20:40:41.000000000","tz":-300},"subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","message":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter\n\nIn order to modernize swift\u0027s statsd configuration we\u0027re working to\nseparate it from logging. This change is a pre-requisite for the\nRelated-Change in order to simplfy the stdlib base logger instance\nwrapping in a single extended SwiftLogAdapter (previously LogAdapter)\nwhich supports all the features swift\u0027s servers/daemons need\nfrom our logger instance interface.\n\nRelated-Change-Id: I44694b92264066ca427bb96456d6f944e09b31c0\n\nChange-Id: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f88efdb4df37577c687f68b2cc3712465955c20e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f88efdb4df37577c687f68b2cc3712465955c20e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
