)]}'
{"id":"openstack%2Fswift~931165","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ib3083a7e987c0e6ed921c68b44f074aebd4af5c3","project":"openstack/swift","branch":"feature/mpu","topic":"p-content-type","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-03 22:01:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3083a7e987c0e6ed921c68b44f074aebd4af5c3","subject":"mpu: complete mpu with user content-type when available","status":"MERGED","created":"2024-10-02 13:10:14.000000000","updated":"2024-10-03 22:02:38.000000000","submitted":"2024-10-03 22:01:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931165-p-content-type","meta_rev_id":"ddbc5a99f1926e33f16815cd598a7faedde0cd85","_number":931165,"virtual_id_number":931165,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 22:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-03 21:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-03 21:05:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 14:10:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-03 21:05:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"35e11e8ebee7aaf522980b1073592570cd2ad0b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 13:10:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04fb7766796b12273a8a69231e18df1565f6629a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 13:14:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f1e5f984b89291f0ac089f2e8bfdf0187993d433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 14:10:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a0b8c02585f453e922526c3cd3f1cbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d6645145ef4432aa6e01cc2c98d590 : SUCCESS in 8m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/808dcb348fa24bccb78a377b6b67a27d : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e9500246581499f9d280bbf525ef758 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/827b3289fa144257a01604d34c73b14d : SUCCESS in 18m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bad051e83bf94950a8ed5c22014e9b26 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/86283b172f7b45e89574cc566de37b0b : SUCCESS in 19m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb45e269de494bc39b3389a93faf1825 : SUCCESS in 17m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a2d1e8d0692848258702caef7076dbc3 : SUCCESS in 19m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8b88a17d46154707b9d116a65ad76f0f : SUCCESS in 24m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/beaa8e6b132f4a518c7eeb5cc5914a5d : SUCCESS in 25m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5eab944d7eb8428f868719cec0641362 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce360662af6d4a88a55b78635860b5ef : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe784f0f9fe344dbb33495954f950e22 : SUCCESS in 16m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cbf464bf6010462f96b0a0fabe722596 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f4f7b08ba61f49adb9ec975b00cc88ff : SUCCESS in 13m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/86948a5670af487093c1f40858972c99 : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/81e46b2a23ee4cda8555a9aae1041a62 : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cbbc970fb6064ca8a0f5f1d04989eeb2 : SUCCESS in 9m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2052b3437714073aa215f16207adf2d : SUCCESS in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fcc79afc7744722ac048ef3c5d8c855 : SUCCESS in 52m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75e65dcb807e4875aaf434e4e37e8155 : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d5c9f7799443bc88df54371831dafe : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/48b8f9148f984130bde8ddc27f1b7ee0 : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9317ff3ba279456a3b88a6950b96146697059c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 14:58:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7f3ccfeec36341654ea22def042ff817b8e03bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 16:14:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8164529c58d34deebe00d1b3bc004b25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4afbdea9273442baff2dfcc949b722f : SUCCESS in 8m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4cf14256b30a428498e3eab2c9c6856a : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c1b07b605184a68b4307bfdb1646e23 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbcf8c969f2c4e42b27304b18cf657bb : SUCCESS in 17m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6712342d85d14dfcb8921f9c0d5766f2 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ec4b80e0d57b440f8faab0f70fd62fd1 : SUCCESS in 18m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b59f46848911412caa2d77817c2bbf4f : SUCCESS in 13m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7e0d0ca45eac4a42b6803130c44187ed : SUCCESS in 20m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d3b8187c236e4fc39bc198d08078c36d : SUCCESS in 21m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/85680d9c7db144c9b33c296d02999448 : SUCCESS in 25m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7e5b48607cd4a4d9025e951f97d22ca : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ff9b9a753d74c9fa851da1b8f2ea3c1 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8f15d6aa0e6b488c8f1f5c3789f55041 : SUCCESS in 18m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/915583e3baa84934b4363c2e0596d418 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/36025addc2fa4f54ad587935b06516ac : SUCCESS in 9m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b544abc6cfca4ec79e9a52f678dc7458 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/027ad6f4888448babe7f98668900ee35 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7aa0abe5997445f1a4aa8ea4476556b9 : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb48e0f8ffca46969eb818cabc200011 : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f996d3574b954d85b102845148f32c0c : SUCCESS in 48m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82f5b0e091ba4e8ba0fdf5046b269bde : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d10a3348ff24445b8b8dad0f17ecac6d : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57b4c2d128bc4f06ba4edf649edb7d06 : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":3},{"id":"a0ae54810eeebd3568f7a9b66c561adc53b42fb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 05:52:55.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42d507a9a7d04be4821071b59cdbc5f4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a7004721f714799b202888a629e5265 : SUCCESS in 1h 17m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83a7b65cb3104374b25cbc370bfe9a80 : SUCCESS in 1h 39m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b564b6f0c4b342638ee18600a2dd65e4 : TIMED_OUT in 2h 00m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9251c4e6881244a1906cc507fed0668d : TIMED_OUT in 2h 01m 05s","accounts_in_message":[],"_revision_number":3},{"id":"15eae44add91c875cb6a37c319f47f134eadf917","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-03 21:05:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5e902ea01d3fdbdd42cbaa556d8b30ffed23ec42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 21:05:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f5a6581c18cd9fcf577684c7f99ebee5733c5b72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 22:01:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/195c286fe0074ca188a2198914db31eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a37c0f06464055b427a05e0a11809c : SUCCESS in 8m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e3dc8165bfcc4d208149262f1c5a2222 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d667b0dfb124da79faebe2440ff4ab7 : SUCCESS in 18m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6e607147005e41649e390350b83039bd : SUCCESS in 18m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/db635117b0ff4b489477bb8397f583f0 : SUCCESS in 20m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/36171bfdb2b842bf8506d2a40ffd088b : SUCCESS in 21m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bcf6c3525ea740099e6409afa0fd0e57 : SUCCESS in 31m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f47b21849d614847b90dca1a249451c6 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2129911b6b4a44d9b0b4f3f2facb689e : SUCCESS in 18m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a96c749fca24ed4a23456ed2a43d88f : SUCCESS in 10m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3c8121071a1b4725a90e731ff06b9544 : SUCCESS in 9m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c958602a0e834c55b6edc63132bab496 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f87f51e752ce4ea7aefabb038c35726f : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b334a240c8d942e7a9f78b23cd319070 : SUCCESS in 5m 50s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/824781c37edd4faeae65401f4fbad5cb : SUCCESS in 4m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a542cdbc60c4cbb82043588164134cf : SUCCESS in 50m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a463dfb46c1640bdb41b8dabb91632f8 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edfa708c5cb444548194875e0dd650cc : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb95962f74d34079bbec309477c2a704 : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":3},{"id":"0711e100b74a721e173d114e4900857f2bebff30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 22:01:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ddbc5a99f1926e33f16815cd598a7faedde0cd85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 22:02:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b63110489bc4a16a960ce53fd5759af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e83336b2f2945448a980b09da2a3e60 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bead9021f923468da933f4ba5e58f60f : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fa209bd1b18ead0ea8cc0670aeac458d97b93537","revisions":{"a940d730fb373c665a7b01481080f2bda4fe98f2":{"kind":"REWORK","_number":1,"created":"2024-10-02 13:10:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/931165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/931165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/931165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/931165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/931165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/931165/1"}}},"commit":{"parents":[{"commit":"d2f1810db7136c719afe2c56987776a377d48fa2","subject":"mpu: minor middleware fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2f1810db7136c719afe2c56987776a377d48fa2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 11:58:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 13:06:53.000000000","tz":60},"subject":"mpu: complete mpu with user content-type when available","message":"mpu: complete mpu with user content-type when available\n\nIf the user sent a content-type header with a createUpload request\nthen, during completeUpload, MPU middleware sends that with the\nmanifest PUT and the symlink PUT.\n\nIf the user did not send a content-type header with a createUpload\nrequest then, during completeUpload, MPU middleware sends a default\n\u0027application/x-mpu\u0027 with the manifest PUT and sends no content-type\nwith the symlink PUT. The proxy app will guess content-type for the\nsymlink based on the user namespace path.\n\nAlso clarify some module constant names.\n\nChange-Id: Ib3083a7e987c0e6ed921c68b44f074aebd4af5c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a940d730fb373c665a7b01481080f2bda4fe98f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a940d730fb373c665a7b01481080f2bda4fe98f2"}]},"branch":"refs/heads/feature/mpu"},"f368e86a88d9e7db90c7e00ca96aa1e0b5257ed6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-02 13:14:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/931165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/931165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/931165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/931165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/931165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/931165/2"}}},"commit":{"parents":[{"commit":"9efdbf62752ef3ebe0493d3d43fb3029fcce4d3c","subject":"mpu: minor middleware fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9efdbf62752ef3ebe0493d3d43fb3029fcce4d3c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 11:58:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 13:14:16.000000000","tz":60},"subject":"mpu: complete mpu with user content-type when available","message":"mpu: complete mpu with user content-type when available\n\nIf the user sent a content-type header with a createUpload request\nthen, during completeUpload, MPU middleware sends that with the\nmanifest PUT and the symlink PUT.\n\nIf the user did not send a content-type header with a createUpload\nrequest then, during completeUpload, MPU middleware sends a default\n\u0027application/x-mpu\u0027 with the manifest PUT and sends no content-type\nwith the symlink PUT. The proxy app will guess content-type for the\nsymlink based on the user namespace path.\n\nAlso clarify some module constant names.\n\nChange-Id: Ib3083a7e987c0e6ed921c68b44f074aebd4af5c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f368e86a88d9e7db90c7e00ca96aa1e0b5257ed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f368e86a88d9e7db90c7e00ca96aa1e0b5257ed6"}]},"branch":"refs/heads/feature/mpu"},"fa209bd1b18ead0ea8cc0670aeac458d97b93537":{"kind":"REWORK","_number":3,"created":"2024-10-02 14:58:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/931165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/931165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/931165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/931165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/931165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/931165/3"}}},"commit":{"parents":[{"commit":"b53dd8e8eb32e6347e74f98a9079eec7d1c4ed7d","subject":"mpu: minor middleware fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53dd8e8eb32e6347e74f98a9079eec7d1c4ed7d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 11:58:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 14:58:30.000000000","tz":60},"subject":"mpu: complete mpu with user content-type when available","message":"mpu: complete mpu with user content-type when available\n\nIf the user sent a content-type header with a createUpload request\nthen, during completeUpload, MPU middleware sends that with the\nmanifest PUT and the symlink PUT.\n\nIf the user did not send a content-type header with a createUpload\nrequest then, during completeUpload, MPU middleware sends a default\n\u0027application/x-mpu\u0027 with the manifest PUT and sends no content-type\nwith the symlink PUT. The proxy app will guess content-type for the\nsymlink based on the user namespace path.\n\nAlso clarify some module constant names.\n\nChange-Id: Ib3083a7e987c0e6ed921c68b44f074aebd4af5c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa209bd1b18ead0ea8cc0670aeac458d97b93537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa209bd1b18ead0ea8cc0670aeac458d97b93537"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
