)]}'
{"id":"openstack%2Fswift~931187","triplet_id":"openstack%2Fswift~feature%2Fmpu~I917edf6b8ac77285c4b925ade20e97916b4ba50a","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-cross-compat-tests","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-28 10:09:21.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I917edf6b8ac77285c4b925ade20e97916b4ba50a","subject":"s3api: add more MPU cross-compat tests","status":"MERGED","created":"2024-10-02 15:07:16.000000000","updated":"2024-11-28 12:32:54.000000000","submitted":"2024-11-28 11:43:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931187-p-mpu-cross-compat-tests","meta_rev_id":"1b9636c0aaa103bb3c36475b42ec7a627edb1213","_number":931187,"virtual_id_number":931187,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-28 11:43:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-27 14:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-27 19:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 15:23:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e265099add8f06440cb59348af12946be1857228","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 15:07:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed45acb77b6dcef3be9103749d181a976bf64579","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 15:07:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"951fe91df7293e8e170d83c6206127aa14741318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 15:23:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1b2e0b70fd2483dafecb4f6233f19c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed924fea950247878fa90ed897ea216a : SUCCESS in 8m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/39dd182312b84a78b5aa8713bd79e04d : SUCCESS in 6m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0997cb0d9785484996d2ae384abeae42 : FAILURE in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30eaeea24fb542e39f2d093eafe09d9a : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1310695ad2c0689752b224db49839de2a33b8991","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 22:26:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab6ec7ce48fb4a6baa76a97cfe55646f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fc11c501f744491bd91f6d18c776a96 : SUCCESS in 1h 33m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cfb799667e4b49d4b18bd123e4e3ced6 : SUCCESS in 1h 36m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec8eb07049354bf8802f6ba8d4182031 : SUCCESS in 1h 51m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/476068375a4b457cbf006991193684f3 : SUCCESS in 1h 35m 58s","accounts_in_message":[],"_revision_number":1},{"id":"112628ae51b17223ff2a642a5aba68ed59990896","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 15:44:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e8361585c592463524b8ba764a545e194ae50d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 16:37:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cce814b435ab4800927623eb43b075c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad138c736bb44909520093d7ba95ca0 : SUCCESS in 7m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df370c6557fb4095851df4c6723c1863 : FAILURE in 3m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5460f293b4164faab1898b1edc205e02 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc0d7baa1734ce881c5ebd80b032d25 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"bccca45452477ede269b0544dc3d1c84191c8a12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 21:37:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b4b6bda6efa4cc99da359e5441e096f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/efb7de0206ac4a74b694236d4d53facc : SUCCESS in 1h 38m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f576861d0d6a528518da15854f237c57d42c0589","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 14:02:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"74e2f44c56494801e199b09931ebaf2db7f59eba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 14:53:18.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/933935 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"05178fd5661864ee6980d0cf503c3d612660d972","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 19:07:32.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"438c737679471972f0018fd54e3898c1ab17ece7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 19:07:42.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f48a21bd4d7d8b0de3eceeef7756c675e1edf37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 19:23:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8d2085f10da4be8ad1c48378c859322\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f99bbbf5371b401480679a61d2562e91 : SUCCESS in 8m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86c6030986c040349799ea3639fa7d80 : SUCCESS in 6m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ad235521ec54e03a75c2b294a1d0b5b : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37b05c233a344caa7db9953d0f6199a : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"4b438bd97f42ce9effadf186a975eac334e9ddd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 19:24:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"92707102dd0d25cfab2d71aee2885b4e5451a32a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 20:17:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94a0584c46ae46aa80de20e4a4b8acea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83bd12efa32940d992177830c3fa5c28 : SUCCESS in 8m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8de595be1f604531a58d47e2ccb37c2a : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/504e0cde19fc4902a471eada8bd6c6f8 : SUCCESS in 21m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd3f8e60ab024fdf953a166af6647bf5 : SUCCESS in 23m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f0be5fcedf2d48108b4e914cc20108f6 : SUCCESS in 20m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/faf36fee00ee49889194ec4b94c0901b : SUCCESS in 33m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bdc2b4fab5eb40759e2d5b64507e0268 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa9258736a444fdf91ab3087e278431a : FAILURE in 19m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/089abe8f6e1e49bfa305ea805a5661b9 : SUCCESS in 22m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1d93da35c6db4019bd781faec26ba4d2 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ef076b9f24a24cc48435f652fffaed67 : SUCCESS in 14m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d944ca1ac24b4057b2ec2f256205cd68 : SUCCESS in 18m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ff46cae4b94b47b3b98e84ae720db742 : SUCCESS in 7m 15s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2cbe42515c9445228532fd0821f2f89e : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ded3219b0e48038fc2905b5cb04305 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"327a6bc6986495c9804a508473e0f2d4a59870a1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-28 10:09:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24c9302a22cb03151830d042abeeee11704e86c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 10:29:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dc630b2ef1247d9a5641b225e5a2004\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8a41486ae94968876ff6ed68983956 : SUCCESS in 7m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08c6fa832ea543569a3d6a3745ca2b17 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/90f7137554644bb7bcc42793509e5e9c : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c49db40758f4fa3b48a9663a1037bc7 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"26ee97a4b3cd3bea0ed8b870ad95b14259b5cbc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 10:30:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4c7309cb19953252184c7ae443ec87f6301a22f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 11:43:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2cf245ac0744a2689010407448436dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425d2a4b0b4a4a1d8b767190aab86883 : SUCCESS in 9m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e8b951c0de8e45a8bc1da11529949262 : POST_FAILURE in 6m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/122c38fda21849c4861b79b850b368db : SUCCESS in 15m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e0b7da2cf374e6c866532e550550e17 : SUCCESS in 14m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b8604f5fab7c4c42a2b1396dd35ee276 : SUCCESS in 21m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eb353455f07e4a9d9f72a93a6aaf6715 : SUCCESS in 19m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/464d39490f5847ed9898ab1640836c8f : SUCCESS in 13m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c33eb4fe897a476e8e73cc98f61b3d3a : SUCCESS in 15m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/80f55a4208b54f87b0f733e17bbf9d14 : SUCCESS in 18m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f907beb5dc7341bbafe3d91bbf4cbd8b : SUCCESS in 9m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/acff82a40f2c483bbafe7094ded96757 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dc84252d02bf430ca5daf2444197ce16 : SUCCESS in 9m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/518481a7f63648e8bbd8e1e2122a1b41 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c8e48887fecb48e99facf0d00a578d4e : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70096450e8294d73be1620c4150b085c : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a997db48f3816e0640a8a7f6081f381361849d02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 11:43:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8d6bc5ab7e128296be18ad574f1fbbf30cd7acfb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 11:45:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f2dab8a99d46b79a95ace7bea50a76\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6405dbcbb94c4bba57fe561ca549ab : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6b1853bdfbc44862bec1f80cf545fb33 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b9636c0aaa103bb3c36475b42ec7a627edb1213","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 12:32:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/859ac96a8d684bfe8bb55a04fddc0c5e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d7ac0077b18c4d709592d9ae640314dc : SUCCESS in 1h 50m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c8fd8d41dd46fbeb2f555052802b3891baf730a","revisions":{"c500d34ce44011b61ac80bb7b6ff8393426f4d77":{"kind":"REWORK","_number":1,"created":"2024-10-02 15:07:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/87/931187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/931187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/931187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/931187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/931187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/931187/1"}}},"commit":{"parents":[{"commit":"8bfb6b4c0bf8138ff080856a3a912e26d24aab71","subject":"mpu: make MPUSession.is_aborted a property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bfb6b4c0bf8138ff080856a3a912e26d24aab71"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-05 11:05:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-02 15:06:34.000000000","tz":60},"subject":"WIP s3api: add more MPU cross-compat  tests","message":"WIP s3api: add more MPU cross-compat  tests\n\n9/26 tests (XXX) fail with Swift s3api.\nAll tests pass with AWS S3.\n\nChange-Id: I917edf6b8ac77285c4b925ade20e97916b4ba50a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c500d34ce44011b61ac80bb7b6ff8393426f4d77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c500d34ce44011b61ac80bb7b6ff8393426f4d77"}]},"branch":"refs/heads/feature/mpu"},"8c8fd8d41dd46fbeb2f555052802b3891baf730a":{"kind":"REWORK","_number":2,"created":"2024-11-22 15:44:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/87/931187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/931187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/931187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/931187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/931187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/931187/2"}}},"commit":{"parents":[{"commit":"819b63b76a8553668466fd699f5e234f14597ea0","subject":"mpu: migrate s3api multi_upload to use native mpu middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/819b63b76a8553668466fd699f5e234f14597ea0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-05 11:05:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-22 15:44:11.000000000","tz":0},"subject":"s3api: add more MPU cross-compat tests","message":"s3api: add more MPU cross-compat tests\n\nChange-Id: I917edf6b8ac77285c4b925ade20e97916b4ba50a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c8fd8d41dd46fbeb2f555052802b3891baf730a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c8fd8d41dd46fbeb2f555052802b3891baf730a"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
