)]}'
{"id":"openstack%2Fswift~931223","triplet_id":"openstack%2Fswift~master~If4029e9d3a6ec59382bac879f71853fc59b4fc82","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-10-25 20:28:22.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-15 17:17:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If4029e9d3a6ec59382bac879f71853fc59b4fc82","subject":"s3api: refactor testing for checksum errors","status":"ABANDONED","created":"2024-10-02 19:56:48.000000000","updated":"2024-11-15 17:17:19.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9f9594a4cb04fd33a93f5d5e45b613e3001d4544","_number":931223,"virtual_id_number":931223,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-04 20:50:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-04 19:28:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 21:31:51.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-10-02 21:31:51.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-10-02 21:57:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a9d4d96bc60959eb1cd3b710ab854ff7b3f0ea2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-02 19:56:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29239e8a14654928c2cbf28258ed2b8a6fb04471","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-02 20:13:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8cf9320c5caf31f713059931ecdf0f3bd9f92471","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-02 20:14:45.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"551907d0d42a1cdaac0aefd8ef2fb2617825ed5d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-02 20:52:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bb915c7457b4050d3957737ee19bcd62095f335c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-02 20:53:13.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"8f73e0ffa52ede918a774b64930c9b4efcf03db9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-02 20:53:19.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"8ad78c9536bf5f509d46e489825f1d5ab402de5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 21:57:40.000000000","message":"Patch Set 4: Verified-1\n\n(18 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3aa8ad405d1c4e098f21d17b694711c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c288ad5b9724a1e8128be8a8e560bf4 : SUCCESS in 5m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9763ff1b0f9d42dcae702a848638e0af : SUCCESS in 55m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be217f82408040f38589a27b83ff682f : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47bc388dfa4b4ecabcacc72bbdc04344 : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b513e418de2449d486d571686cf5f164 : SUCCESS in 16m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f063604f1d14dec88ede4f7b3615a1a : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/08fd3d6167424f3b872d4600f1f0727e : FAILURE in 16m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e5f90fc1079440abb7b8c9b0662806e4 : FAILURE in 16m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b25e59f2971847d4b0219602b381bd34 : FAILURE in 21m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0c17c59b4fef443f8bd5d63dc5e30b5e : FAILURE in 25m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3e92d12ffaa5467eb7306c0b97c1d266 : FAILURE in 30m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c7e4731f67f43b7bc2ca1e242def341 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/468e9dcb285544acb7e5d14c8e1f0e75 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25b61e942bc348e6906f1c7694f60053 : SUCCESS in 17m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4fe365a1e1f84ed6ac3cc67c4856b874 : SUCCESS in 8m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a8c96d37752c438e861d71ebd19a1b34 : SUCCESS in 13m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/df396584a42f4a2996fcc0298eaa20b0 : SUCCESS in 16m 34s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d3684ed0a64043f0a93c96e0ab907f1f : SUCCESS in 7m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b153c4346e24b528aadce882fd9362d : SUCCESS in 29m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20d6535b9c6649e097abe3761aab857b : SUCCESS in 52m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eec229d26fb9499899da27bc3028272c : FAILURE in 18m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c321db0de1a14e5aa3568803cd63fb3e : FAILURE in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e40ec11ab3484943ab5886239a7ad6ac : SUCCESS in 19m 50s","accounts_in_message":[],"_revision_number":4},{"id":"ad4b125c4388cf72d07e5a4dda7eef049626776f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-03 00:01:27.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d3dba2b628948ea97939ffe87a7f27223b4d1c73","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-03 00:06:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4c44582e16530af77cc1c788900bc352f1db43e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-03 00:08:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d2dd344f2d808699cce00681facbf6d0d7891704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 01:16:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f86a316280f848e6a0cd0e3992869dfa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7c55ba4959491b9a29a4a6d920c192 : SUCCESS in 8m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1e17a680e74445d8bafcb24fcb971eff : SUCCESS in 1h 06m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2f795fbabb4470c8801d921da17ea04 : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9dd2427c1ad542559c65dd252a7a6b38 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a89d610e9d3a4e9fb0dfe4be8b561fdb : SUCCESS in 17m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/957f1a129f0b4f43880552ef252979ec : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0fb592b1b824fe8adf177334b7ba07a : SUCCESS in 20m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef91c7811f0a4d8daa803c5912a1a9a7 : SUCCESS in 15m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6527ddcbde694d54a8dcf58baa4266cb : SUCCESS in 16m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1c6297a214e94b169f1a51a204e57212 : SUCCESS in 22m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ffef102552ac4b979dbd2ba2021b542c : SUCCESS in 33m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31edc3b8f8fa4fec84312c4f97a99234 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4539a7f316354307817022196778abab : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56973fd4974b4d9e93d83c8c90553899 : SUCCESS in 15m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b7b51a773da44d5da86d3d46dfc4f212 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/883a3b42139f42b083ee80494af6e258 : SUCCESS in 8m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3d5f11fe02db430887447d99312b9fbe : SUCCESS in 14m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/669bccb9110d4d4682b123dee3f02097 : SUCCESS in 5m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/98b9cce78d7b4638bd7e9ec9b875d1cd : SUCCESS in 34m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b8920340e0544450913d52b310514952 : SUCCESS in 53m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adda1066e80644ad8f4f18d8c0237878 : SUCCESS in 14m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8d3e1d80484d4d80a68f928433c45e : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c66cd32c4caa44b1b973c18f8b1634c1 : SUCCESS in 15m 29s","accounts_in_message":[],"_revision_number":6},{"id":"cb4335afda0675211b03415ade5a81016a3bf35a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 03:18:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29a5cbe815a24033b32dfe32de8edfab\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/734eebc033a54741b5b91ae9840cb83a : SUCCESS in 1h 19m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a4aee3854714a9c834dd970a14b44b6 : SUCCESS in 1h 31m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f01285dd4947471eb762f8542fbc3f13 : SUCCESS in 1h 00m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b506d6738dee4fc686e4cd0ea6baa463 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":6},{"id":"73114bc3fc66f69d01b04b7f0821744e28d98ff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-09 12:26:48.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5950876ae7a5300d7e800d31dd66c5e3f399f0b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 13:26:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65f948632ca94c6d9b9012d7497c1ede\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1028332e7d84d0dbc67fc7b2889a573 : SUCCESS in 8m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7d9c575f342496b982147044fa8abd2 : SUCCESS in 54m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b48b4ab2dfa24ba7a4b63881134a816c : FAILURE in 12m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88a09cd60e8a4da49a08d65238464100 : FAILURE in 21m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9955be8c4b7146b09fe264456c875f3d : FAILURE in 12m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e3e13d9539e47f993dbe6c959b31657 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a35cba5e059a4e0e82d21b86c1dbe821 : SUCCESS in 11m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df313c0290414f13bc4189603c742ec1 : SUCCESS in 21m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bb89f796c22c4251be26ed09add3ae50 : SUCCESS in 10m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5016aa12ca414f0793422cfd61202e73 : SUCCESS in 28m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d70d4448eb0646f58881b0ba0c3e1739 : SUCCESS in 30m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9a7a96e94c745f8af26383cb431c611 : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f13791e9e5424716af56c8799f3c5d1c : SUCCESS in 16m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a325a4d7b3e54c1e826e5f9987c1ea09 : SUCCESS in 18m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3b9f25de30be41a1a9f37c90173e5d88 : FAILURE in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/223e86f4918b4a62b3042c5aae8113ee : FAILURE in 13m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d357e2fe3b8f4f6f9ff3217880bb7606 : FAILURE in 14m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dfa8421884a642d9868738aa7b02a839 : SUCCESS in 7m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f95320d3daf40e0875d3dc905d95f6e : SUCCESS in 53m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4276a50830a345bda103774699d09860 : SUCCESS in 43m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4763973c59984e4fa84890ee5b5aea2d : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d2895caf114008a3f0764dcae4576c : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3c21322d15c34add8f8dd460ba438900 : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":7},{"id":"00c91c659d9722119a6c91522d5d2c0e7de5fc8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 15:32:43.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ac19813212b4eab8692046c72bdcac1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/551f52de43744ccea3758970eacd9041 : TIMED_OUT in 2h 01m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41dc8760de3c46e788ef1dbb863d7859 : SUCCESS in 2h 00m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8609bde8936043e7ae024a6537681c38 : FAILURE in 2h 01m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/567c77fcebac45ca96acf3b527ca802f : FAILURE in 1h 58m 28s","accounts_in_message":[],"_revision_number":7},{"id":"05242193e6d75fc8f5e25e01454190b6b908758a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-09 20:29:58.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"fb2a819f4a42b45ba9af25d463571070e59b4dcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 21:55:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/756048bd2ea742a6b02be8a27667d5ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/686a7df276db40da9077790bc88d1599 : SUCCESS in 8m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/476531c04bc34865a81f4a15a012f598 : SUCCESS in 1h 09m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3ab6c1bd8984360a69be00820d6e2cc : FAILURE in 17m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d36ae9b976444bb8caf17ac95159aa5 : FAILURE in 14m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04f68993171745339b73528340645e1d : FAILURE in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c04dcfd62b924695a22a87fb1f681e94 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a9b057f6879845719169b9e75b59654c : SUCCESS in 18m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9468a2b195448f3b28d622b790fe711 : SUCCESS in 16m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6129b8e0169c490d87de5ea1f45bcf44 : SUCCESS in 20m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a25bf3d048424855a57d6e6fb3b61b15 : SUCCESS in 27m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a9817b4d05184fc4a5d775c5ea7832ee : SUCCESS in 29m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f8cc492c2ce4abaaeb9e78dfacff2ca : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2ec0a634f58840e4967bac49fe358a27 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ff778e0552d948b78588a7944f36b4f5 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9f46c0167cd04915a4cc5b5d84f6a4b4 : FAILURE in 14m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/466c95226072417f85b87d398800b64e : FAILURE in 14m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/db55b54baf1d431c8d3eea7a0c0a4663 : FAILURE in 12m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8881cfc232a14efe9abc29edc8e5ee7f : SUCCESS in 7m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b1ca250001334f0898259264fe59e58e : SUCCESS in 36m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f000b6cea8f4b09b4ed41057063406d : SUCCESS in 53m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2aa21d90e4f349e5be15180066c0de27 : SUCCESS in 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c323d43e7b4b454ba7908c3682176586 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63e671bb5cf34ad7b8ca9789a8069543 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":8},{"id":"c163a45499a853570eee497f9830838d1ace0c73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 13:05:09.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8984bc1b374c4a08a59f74859bb08b81\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/576f464b88a6452ab8f533f1cb8d0caa : TIMED_OUT in 2h 00m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ace58a08412d408ebba45d0c33b25193 : TIMED_OUT in 2h 01m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b15a535537f4d769048a114bb42d28b : TIMED_OUT in 2h 01m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9e7bf2528df40c6b2070e2e68917b2b : FAILURE in 1h 19m 21s","accounts_in_message":[],"_revision_number":8},{"id":"01809c8650262e44d6225ce72fd29e84a7aa019f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-10 21:54:21.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"2269aa7f55df022400045450f9a410afb3aa50e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 22:58:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c01dee701965477fbe6b77fcd9aa9a7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8005c291a7044aca0400cc56a32c425 : SUCCESS in 8m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5340df1dd0df43b7b8d3365e2dd9e473 : SUCCESS in 52m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aba660e7501f4b8d82b5b0f12bd7ec49 : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9337064aa40740c4b928e7d3bf2b742b : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b479d2dec194dbf8d34805e4bef3534 : SUCCESS in 18m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/287054f8afa84e65ba06fa4b55648d68 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b4a81df1d0fd46c780ffa7dc0a153d05 : SUCCESS in 18m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9bab489e899544338657c54e85eff7e1 : SUCCESS in 13m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b29677946d754eecbc189801c1b0951a : SUCCESS in 16m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/adc132ac054e483784850e6041a9b168 : SUCCESS in 20m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/483733873cbc4fc2b2b15bf3d5a75915 : SUCCESS in 22m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d763d2f566241c1a112d635aed352b2 : SUCCESS in 20m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/294fa28c69454b6998b3a4c443c507de : SUCCESS in 19m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/01c4ac3fb8f242d39541459220fbe518 : SUCCESS in 14m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/50ec2ea888ae41f485068027a52387d7 : SUCCESS in 12m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9a7301a5a0e64404b01f8e9c9a5aab77 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c1f9d61088a2468487d81c30f359ccb7 : SUCCESS in 15m 38s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/76ee0604d3b64d4b9ee469fd9a624d21 : SUCCESS in 6m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c3edccca7ea64a0fb6f5d53487e88ceb : SUCCESS in 57m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b2c1c1bd85545b4b7252f0505370f53 : SUCCESS in 32m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c38f44f1010f45099a40122bf5e5dbe4 : SUCCESS in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159438532bd246cb8683ed03dd044afe : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b39d8b6ca82e4e69bdf0e36d02845288 : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":9},{"id":"b826bf753ce07eaedae058f43ca70cf4971f975f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 00:49:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aabfede05754d87ad24545a818d9562\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/21f8855e75c84b7f82c7bf191e496bec : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":9},{"id":"05da0b3408ebb6ea70f4e61a17e5b7a7c92ccdad","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 20:28:22.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"db67370e846e7e6bb23a2330e30be53c6f84e7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 19:28:42.000000000","message":"Patch Set 10: Patch Set 9 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":10},{"id":"bb7d721fee21fbcb14d8dccdb72ad1c53681d56e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 20:50:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31a67ffa6b244e7fa1477f54aa457c1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8832b13f5733491985ea67e1e0d18e70 : SUCCESS in 8m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e4a32be939a4a8dbd506b0b7ab65c4a : SUCCESS in 1h 04m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f825cf28d75d40b4b197205a2580ae5a : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a617b6752b744e4aabd454a7b96d1902 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d0f7b927d3f4bd5b14f1b1a61f9f654 : SUCCESS in 21m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ec9abfe88cc4648bce36514137c1a28 : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e35cb0833bcc4203b4c624550d5f30f0 : SUCCESS in 20m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80ddf065d19a4122ae35305a66718d2d : SUCCESS in 17m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0a81322770ce4ed4ac703155c0c253a1 : SUCCESS in 19m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c6742eeb51a24d07820d769442edf82e : SUCCESS in 39m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de73840aeaea4d4dbe6c519a4fb42326 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf34c2c040d54c3c8be4589f3c96aad2 : SUCCESS in 20m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6721f356ec404eb89ca06660426ce82c : SUCCESS in 15m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4349bfc2e76a4264856bda20b2d588fa : SUCCESS in 11m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a0ebfa1cfb2f4036893150c64e407201 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6dab9e8148754067a4e71035418f9238 : SUCCESS in 14m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/896a476be79a4ec3a338c053fb7607e1 : SUCCESS in 7m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6871e0423050473aaad2dfdc7803d50c : SUCCESS in 54m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9521cd802785468fbf16cf66172f6aa3 : SUCCESS in 50m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbb7920df93041c69e1670a4afd35533 : SUCCESS in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c008c30dd754e1891c5935a8f782823 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5883050b5a24b53bca713e09fbeb99d : SUCCESS in 16m 53s","accounts_in_message":[],"_revision_number":10},{"id":"bbc35c620c4e7844a98dc802040ae11ebbf761bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 21:26:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d04f05f1d5442590ae64d5f53f68ca\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd436b12c5cd4b33bab6ac58ca501325 : SUCCESS in 1h 31m 25s","accounts_in_message":[],"_revision_number":10},{"id":"9f9594a4cb04fd33a93f5d5e45b613e3001d4544","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-15 17:17:19.000000000","message":"Abandoned\n\nBetter test changes have been incorporated in  https://review.opendev.org/c/openstack/swift/+/909801/32..33 and modifying unit-test to verify the correct error message being emitted by the server, ref: https://review.opendev.org/c/openstack/swift/+/935387","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a4fd1241111e93dd6770899db7d589eb32f693ab","revisions":{"22a5dfbe0bb764c902857e239cbdf03057cdd058":{"kind":"REWORK","_number":1,"created":"2024-10-02 19:56:48.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/23/931223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/1"}}},"commit":{"parents":[{"commit":"79c60c8f04da7649a5959646877b587a28fb52f6","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79c60c8f04da7649a5959646877b587a28fb52f6"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 19:55:28.000000000","tz":-420},"subject":"s3api: testing for md5 mismatch errors","message":"s3api: testing for md5 mismatch errors\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22a5dfbe0bb764c902857e239cbdf03057cdd058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22a5dfbe0bb764c902857e239cbdf03057cdd058"}]},"branch":"refs/heads/master"},"55610d565dd0c776b9143305bbd9b35be4740c4c":{"kind":"REWORK","_number":2,"created":"2024-10-02 20:13:17.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/23/931223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/2"}}},"commit":{"parents":[{"commit":"79c60c8f04da7649a5959646877b587a28fb52f6","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79c60c8f04da7649a5959646877b587a28fb52f6"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 20:13:08.000000000","tz":-420},"subject":"s3api: testing for md5 mismatch errors","message":"s3api: testing for md5 mismatch errors\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55610d565dd0c776b9143305bbd9b35be4740c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55610d565dd0c776b9143305bbd9b35be4740c4c"}]},"branch":"refs/heads/master"},"85d8d525dccffcdfd761eace3f620addc41c6b9e":{"kind":"REWORK","_number":3,"created":"2024-10-02 20:52:13.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/23/931223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/3"}}},"commit":{"parents":[{"commit":"79c60c8f04da7649a5959646877b587a28fb52f6","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79c60c8f04da7649a5959646877b587a28fb52f6"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 20:49:05.000000000","tz":-420},"subject":"s3api: testing for md5 mismatch errors","message":"s3api: testing for md5 mismatch errors\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d8d525dccffcdfd761eace3f620addc41c6b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d8d525dccffcdfd761eace3f620addc41c6b9e"}]},"branch":"refs/heads/master"},"467ac720ff405e0b5aebb1df579aa3ee8d226018":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-10-02 20:53:13.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/23/931223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/4"}}},"commit":{"parents":[{"commit":"79c60c8f04da7649a5959646877b587a28fb52f6","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79c60c8f04da7649a5959646877b587a28fb52f6"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 20:52:18.000000000","tz":-420},"subject":"s3api: refactor testing for checksum errors","message":"s3api: refactor testing for checksum errors\n\nThis is just to refactor the helper method check_algo\nwe are using to validate all these new checksums we are\nintroducing.\n\nDrive-By: Added an extra assert statement to validate\nour md5-mismatch error reporting\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/467ac720ff405e0b5aebb1df579aa3ee8d226018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/467ac720ff405e0b5aebb1df579aa3ee8d226018"}]},"branch":"refs/heads/master"},"5eea24ed316a8498fe9e82fcb0c190b339095627":{"kind":"REWORK","_number":5,"created":"2024-10-03 00:01:27.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/23/931223/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/5"}}},"commit":{"parents":[{"commit":"79c60c8f04da7649a5959646877b587a28fb52f6","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79c60c8f04da7649a5959646877b587a28fb52f6"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-03 00:01:11.000000000","tz":-420},"subject":"s3api: refactor testing for checksum errors","message":"s3api: refactor testing for checksum errors\n\nThis is just to refactor the helper method check_algo\nwe are using to validate all these new checksums we are\nintroducing.\n\nDrive-By: Added an extra assert statement to validate\nour md5-mismatch error reporting\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5eea24ed316a8498fe9e82fcb0c190b339095627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5eea24ed316a8498fe9e82fcb0c190b339095627"}]},"branch":"refs/heads/master"},"1f5b14a7c74138e303fd3dbd4a4596f0d1456196":{"kind":"REWORK","_number":6,"created":"2024-10-03 00:08:46.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/23/931223/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/6"}}},"commit":{"parents":[{"commit":"79c60c8f04da7649a5959646877b587a28fb52f6","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79c60c8f04da7649a5959646877b587a28fb52f6"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-03 00:08:25.000000000","tz":-420},"subject":"s3api: refactor testing for checksum errors","message":"s3api: refactor testing for checksum errors\n\nThis is just to refactor the helper method check_algo\nwe are using to validate all these new checksums we are\nintroducing.\n\nDrive-By: Added an extra assert statement to validate\nour md5-mismatch error reporting\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f5b14a7c74138e303fd3dbd4a4596f0d1456196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f5b14a7c74138e303fd3dbd4a4596f0d1456196"}]},"branch":"refs/heads/master"},"f2829a7daa3a5e157a89b831f649e318cebe3702":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-10-09 12:26:48.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/23/931223/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/7"}}},"commit":{"parents":[{"commit":"d6194628d0cc0188e9039a5b89c1e056b1f14d5d","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6194628d0cc0188e9039a5b89c1e056b1f14d5d"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-10-09 12:26:48.000000000","tz":0},"subject":"s3api: refactor testing for checksum errors","message":"s3api: refactor testing for checksum errors\n\nThis is just to refactor the helper method check_algo\nwe are using to validate all these new checksums we are\nintroducing.\n\nDrive-By: Added an extra assert statement to validate\nour md5-mismatch error reporting\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2829a7daa3a5e157a89b831f649e318cebe3702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2829a7daa3a5e157a89b831f649e318cebe3702"}]},"branch":"refs/heads/master","description":"Rebase"},"bae6ab1679b6b796eda8f47ad8c5ddfb54a6c466":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-10-09 20:29:58.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/23/931223/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/8"}}},"commit":{"parents":[{"commit":"458172a9a9baf265780feb7e8340853b1d633fd1","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/458172a9a9baf265780feb7e8340853b1d633fd1"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-10-09 20:29:58.000000000","tz":0},"subject":"s3api: refactor testing for checksum errors","message":"s3api: refactor testing for checksum errors\n\nThis is just to refactor the helper method check_algo\nwe are using to validate all these new checksums we are\nintroducing.\n\nDrive-By: Added an extra assert statement to validate\nour md5-mismatch error reporting\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bae6ab1679b6b796eda8f47ad8c5ddfb54a6c466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bae6ab1679b6b796eda8f47ad8c5ddfb54a6c466"}]},"branch":"refs/heads/master","description":"Rebase"},"073ed59ce87851ee6e71de9a29eaab079a48c6fd":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-10-10 21:54:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/23/931223/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/9"}}},"commit":{"parents":[{"commit":"ee0c1a1da16f6c6386341275fba9f14129061080","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee0c1a1da16f6c6386341275fba9f14129061080"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-10 21:39:33.000000000","tz":-420},"subject":"s3api: refactor testing for checksum errors","message":"s3api: refactor testing for checksum errors\n\nThis is just to refactor the helper method check_algo\nwe are using to validate all these new checksums we are\nintroducing.\n\nDrive-By: Added an extra assert statement to validate\nour md5-mismatch error reporting\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/073ed59ce87851ee6e71de9a29eaab079a48c6fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/073ed59ce87851ee6e71de9a29eaab079a48c6fd"}]},"branch":"refs/heads/master"},"a4fd1241111e93dd6770899db7d589eb32f693ab":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-11-04 19:28:42.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/23/931223/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/931223/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/931223/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/931223/10"}}},"commit":{"parents":[{"commit":"40e5c890a2b099c153dd9af09ed3443948d19dc0","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40e5c890a2b099c153dd9af09ed3443948d19dc0"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-02 17:24:05.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-04 19:28:42.000000000","tz":0},"subject":"s3api: refactor testing for checksum errors","message":"s3api: refactor testing for checksum errors\n\nThis is just to refactor the helper method check_algo\nwe are using to validate all these new checksums we are\nintroducing.\n\nDrive-By: Added an extra assert statement to validate\nour md5-mismatch error reporting\n\nChange-Id: If4029e9d3a6ec59382bac879f71853fc59b4fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4fd1241111e93dd6770899db7d589eb32f693ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4fd1241111e93dd6770899db7d589eb32f693ab"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
