)]}'
{"id":"openstack%2Fswift~931325","triplet_id":"openstack%2Fswift~master~I56c1ecc4edb953c0bade8744e4bed584099f29c7","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I56c1ecc4edb953c0bade8744e4bed584099f29c7","subject":"versioning: 411 PUTs with neither content-length nor transfer-encoding","status":"MERGED","created":"2024-10-03 17:00:12.000000000","updated":"2025-02-12 22:10:37.000000000","submitted":"2025-02-12 22:09:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931325","meta_rev_id":"605a424613e669ece2d184447ae942c2bbfea34d","_number":931325,"virtual_id_number":931325,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-12 22:09:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-12 20:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-12 20:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 19:01:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 16:10:11.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"2cf68ad6c2f6fb7ed71fd1223fddad7b4f020a5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-03 17:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e23e4bff745aac3b8b18a53bc507488be628424e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-03 17:25:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b216ef6929e1fcd49246fd99a0d97a674cd3db63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 19:01:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/228648f5022b4cc99b712139afbab5df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb53cba05ce14d24add6101d8e045221 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/3c75fa462f3d4a318110dc673766a1b2 : SUCCESS in 1h 03m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/10eff26b6b3046dfa27700c1f992cc83 : SUCCESS in 1h 27m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a3b4e27230cb42f58bea8b7d596d3dc4 : SUCCESS in 58m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cdb203a157647b3a3f6e52f375c9676 : SUCCESS in 51m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/defe793fc84f44c083a765ca4af42f10 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4fefdda1ea24506a0e5ad4eb3df8740 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/288961d44cf64984aea9da26a350133f : SUCCESS in 18m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2590d43dcce34993bef26c0079bd2009 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c2400ded85f4b5a93d1d52ddba06b15 : SUCCESS in 9m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eac328aefffe48029d9d0d5f7803f1c2 : SUCCESS in 19m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2a6113beabae4a54b95a50695ae0b820 : SUCCESS in 9m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7b6769ad048c4c28a16b2216decc94f5 : SUCCESS in 19m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/17dff9dee46f43edbcbfae978a89a8e0 : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6528f2ed89784f4c82e1361b1571a3a9 : SUCCESS in 9m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/90a4d5c4ddd54271a8d0b7a8042679d2 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/483a5a577ac844208720fcdb4c76492a : SUCCESS in 10m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/42d32fda07f347afa4a5fedcad0c7910 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6c5b3401261a4c87978abb8241b7acd2 : SUCCESS in 7m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ccc9d686d824881b5186db54301c49f : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cb338a60bfb54f84ab8cc9630fdadf81 : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c8f3c895c832418096a4516c23825459 : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b399175cb8dd40a2863295ed5808964e : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/50aae51a3b344a0690dc32ffc1e6f1e0 : SUCCESS in 57m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aba41880ce6c4bd9b3b317461cf69e47 : SUCCESS in 51m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf282e9130b84278ae116883a9e832ea : SUCCESS in 34m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5de283234a2b47bdb0550d868d99b2a0 : SUCCESS in 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e45f65bd324157bfb1b6277ee9612d : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/31f6ca1ceb324e5881d53aab0bbff7f0 : SUCCESS in 18m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c3c435bc3d44351bd4a15a61576ecc5 : SUCCESS in 40m 27s","accounts_in_message":[],"_revision_number":2},{"id":"17c86e9d4b01cac8c8a2a2a76e1b23febff57fd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 21:49:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaddbfabcd194af1b8deadfd4e0e55cd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b8ca3b51291421ab13bada21e9e1687 : SUCCESS in 51m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61f6e49e1c0642ba881b968f168b749b : TIMED_OUT in 2h 01m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0493391a1b84466ada7af6832ebfff9 : SUCCESS in 1h 40m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a5e230cf2b5481396681cf3541b5720 : TIMED_OUT in 2h 00m 59s","accounts_in_message":[],"_revision_number":2},{"id":"0091473ad5366127aa36d4f57d319d4a452d9a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-15 20:52:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d18b887173cdb55740823b0a38874f3ad2cdce0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 22:20:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e52c551cc2fc412c9e544bdadc8096cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdc01a9bb2ab4307a677dfaff939aed8 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/3d8466b8047d4698b48c6244dae98964 : SUCCESS in 1h 14m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1317298ee79c4d33abb4889a15a1b811 : SUCCESS in 1h 03m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9602c2a6b0674d02968d7904f58f8cd4 : SUCCESS in 46m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/058db69156f54ed69cc9e40ca31a7258 : SUCCESS in 1h 04m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4602afeb307245d295f91f87458253be : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5679138d8dcc44da801c304c1c73c71d : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/248a3b18fe124c15a17ef1dfca17cc42 : SUCCESS in 23m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbea360f1dfd4c0cb491a0ac04218e39 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/897f7e0966b048349846914189092542 : SUCCESS in 14m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32164076f09e4d61b1b5b75021fd5c37 : SUCCESS in 23m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/33238a38635b4cb3a775a2507ef926fd : SUCCESS in 30m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bdafca5266f4432a8cdc689d105ba6d0 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c14814a50a8c4918b84017dfcd7e84ed : SUCCESS in 12m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/343861b222784a299505651d910a2040 : SUCCESS in 14m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/84efd0e8bb3842f3a1fa1a27276f1cce : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ccd8ab7a0d614b9a88bee4a19192d9e1 : SUCCESS in 9m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4de2f2462fef4258a985f344f1995385 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f307da032ae54d2984873467905ec6b8 : SUCCESS in 56m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cd5b373c932048bba2f5ec17a0b242d1 : SUCCESS in 41m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ba04ec240b542c0ab908557c71de238 : SUCCESS in 25m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8343c147dfb84233bc9bad93912af16a : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3487a4b26fd64692906d5ca3213ab7bd : SUCCESS in 2m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9199ad7850674dacb211c995d672dd14 : SUCCESS in 20m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9745be0ceffe449ba30e1d14e16bdee2 : SUCCESS in 35m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2e3bf735710482d8b34e3517a566570 : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":3},{"id":"3add12851d6973e8333c81bffb6490f040b383d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 22:44:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e94bdc5f7cfb401babf069136ed1e19f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8db744e324e4dafa1c4314ba0abe6c1 : SUCCESS in 1h 50m 20s","accounts_in_message":[],"_revision_number":3},{"id":"a8b4b035552ff65ff3737163d3a11acd8b41703c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-12 20:53:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bf9fcadb90c000807febe44de06f08c4fc18d666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 20:54:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fa19c3505a0fcebf33b2584ea30c7b6ef0d5f045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 22:09:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2fc302d7c424907be36cd8a62089160\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66b2d65dfe64dd0ad17c411dcb81d15 : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/8d11c91173e945e8ba36b8797741a0e5 : SUCCESS in 49m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d262f147bdb44dbbd7d16a1cce2485c : SUCCESS in 39m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/16fd822aa69d408f8cc38b386fd69cb0 : SUCCESS in 45m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a44403e09ac74a45b2dae8b9caca08f6 : SUCCESS in 1h 09m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/df4177024fcf4a66a89d3cfe117a5f97 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/22b5c81562c34924932d5c5f8bbc5b38 : SUCCESS in 9m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/09557c7624344215bbef067b647a7a92 : SUCCESS in 19m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/aee426191f7e4b3692bd7d6dfe6c475b : SUCCESS in 27m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cbd1867f02e14428aed8e78e9a4571f3 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ee196a79b5c9436b859186ada0e94dfd : SUCCESS in 14m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1283759cebe24479a165ea5bea7c3103 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9749b3ae4ebc4172a1644a54ec5be4e3 : SUCCESS in 4m 56s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/69ceeb8330574669abb32756a1ff3afe : SUCCESS in 4m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5b5170a01ff4b94a07fa46e4977a378 : SUCCESS in 39m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5c88b6055674493a324442ec6a15881 : SUCCESS in 39m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d29400d9e8a45879cd18522a477078e : SUCCESS in 25m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba7a1f4cc10a4a8796fb51b64b258178 : SUCCESS in 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22f541ea235b47ce99f0af1f4a916995 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d00a7413bf0499bade6faf2c6778110 : SUCCESS in 19m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf19aeee3dac475ea5a487a39c64f5e9 : SUCCESS in 53m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4157d7db3b944ab59659817f4649c0ab : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":3},{"id":"dfe61ff302ad55e1564837a8946e45cbbc8b8b24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 22:09:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"605a424613e669ece2d184447ae942c2bbfea34d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 22:10:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/956813a8cf9d4720979409496da34eaa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37312f8881d24b629ce4615962a9bbc3 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7c98003614594051b56aa4665cbe72b5 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a630e76d6c2d09a3a47465b61471fff86dd4521b","revisions":{"472923affba62319dba9c29a1cdab5d66c3cc879":{"kind":"REWORK","_number":1,"created":"2024-10-03 17:00:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/931325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/931325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/931325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/931325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/931325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/931325/1"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-03 16:22:09.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-03 16:38:00.000000000","tz":-420},"subject":"versioning: 411 PUTs with neither content-length nor transfer-encoding","message":"versioning: 411 PUTs with neither content-length nor transfer-encoding\n\n... just like we would do in a normal container. Previously, we\u0027d try to\nread a byte from the client which, due to a bug in eventlet HTTP framing,\nwould either hang until we hit a timeout or worse read from the next\npipelined request.\n\nThis required that we reset a (repeatedly-reused!) request in s3api to\nhave an empty body, or it would start triggering 411s, too.\n\nSee also: https://github.com/eventlet/eventlet/pull/985\n\nCloses-Bug: #2081103\nChange-Id: I56c1ecc4edb953c0bade8744e4bed584099f29c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/472923affba62319dba9c29a1cdab5d66c3cc879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/472923affba62319dba9c29a1cdab5d66c3cc879"}]},"branch":"refs/heads/master"},"81fee737c2212ded8137fdce63b62ecb57261431":{"kind":"REWORK","_number":2,"created":"2024-10-03 17:25:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/931325/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/931325/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/931325/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/931325/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/931325/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/931325/2"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-03 16:22:09.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-03 17:25:32.000000000","tz":-420},"subject":"versioning: 411 PUTs with neither content-length nor transfer-encoding","message":"versioning: 411 PUTs with neither content-length nor transfer-encoding\n\n... just like we would do in a normal container. Previously, we\u0027d try to\nread a byte from the client which, due to a bug in eventlet HTTP framing,\nwould either hang until we hit a timeout or worse read from the next\npipelined request.\n\nThis required that we reset a (repeatedly-reused!) request in s3api to\nhave an empty body, or it would start triggering 411s, too.\n\nSee also: https://github.com/eventlet/eventlet/pull/985\n\nCloses-Bug: #2081103\nChange-Id: I56c1ecc4edb953c0bade8744e4bed584099f29c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81fee737c2212ded8137fdce63b62ecb57261431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81fee737c2212ded8137fdce63b62ecb57261431"}]},"branch":"refs/heads/master"},"a630e76d6c2d09a3a47465b61471fff86dd4521b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-01-15 20:52:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/931325/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/931325/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/931325/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/931325/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/931325/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/931325/3"}}},"commit":{"parents":[{"commit":"f95315b7111746fc54e4ab17c96ee1c1909ee256","subject":"CI: Remove B320 and B410 bandit skips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f95315b7111746fc54e4ab17c96ee1c1909ee256"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-03 16:22:09.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-15 20:52:09.000000000","tz":-480},"subject":"versioning: 411 PUTs with neither content-length nor transfer-encoding","message":"versioning: 411 PUTs with neither content-length nor transfer-encoding\n\n... just like we would do in a normal container. Previously, we\u0027d try to\nread a byte from the client which, due to a bug in eventlet HTTP framing,\nwould either hang until we hit a timeout or worse read from the next\npipelined request.\n\nThis required that we reset a (repeatedly-reused!) request in s3api to\nhave an empty body, or it would start triggering 411s, too.\n\nSee also: https://github.com/eventlet/eventlet/pull/985\n\nCloses-Bug: #2081103\nChange-Id: I56c1ecc4edb953c0bade8744e4bed584099f29c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a630e76d6c2d09a3a47465b61471fff86dd4521b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a630e76d6c2d09a3a47465b61471fff86dd4521b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
