)]}'
{"id":"openstack%2Fswift~931374","triplet_id":"openstack%2Fswift~master~Ifa1cd498df6c56d363342d16eb6e9fc6009b454a","project":"openstack/swift","branch":"master","topic":"removeLoggerAdapters","hashtags":[],"change_id":"Ifa1cd498df6c56d363342d16eb6e9fc6009b454a","subject":"sq? be more defensive with log prefix","status":"ABANDONED","created":"2024-10-04 09:17:20.000000000","updated":"2024-10-14 16:50:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"310d576c96a1aa796815f29c35585ebda0fe18a6","_number":931374,"virtual_id_number":931374,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-04 10:19:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-04 10:19:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01b25c7caa3e234073028cb98995df0d6b500e0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-04 09:17:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea6612d1bea870aa1cd5e5d3ba1dfbd84ee18616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:19:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb7f5152cd514e1f8013fd864af89afb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9ad01221a84414a14bc7103a2ef94b : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/d99e4f1880244c8ab68fe321e93657ef : SUCCESS in 51m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0c7a9d05e3b4f44941eb714568e3008 : SUCCESS in 53m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75f25c957970443c9fef16ce0f91ca56 : SUCCESS in 44m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73095bcb08d84a9888c29a07120280e4 : SUCCESS in 45m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6dc9d783631476ea88d1703c3c5c452 : SUCCESS in 17m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ac028162d874f6781dfe1d5eb01ec1a : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d93f380dc0684cafa183fe45543855c7 : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b4adb271134434eacc420fd401b6a61 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/683d9717140540aab8726c41b45a8952 : SUCCESS in 23m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/94c0092a919e4d07abe90b8677c752fb : SUCCESS in 19m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1b53d3b182434e23a18617aec26a7844 : SUCCESS in 21m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4f5cabbd84f6492b9003938321d14e23 : SUCCESS in 24m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/53a2670b9f99419ab158103426d37c3c : SUCCESS in 30m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a38c2aa91d614a579b9949ab2820c4e7 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/77d1842a0ca94d87a013523cd44c72f2 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/107764a257434360b2c3043b60cd4110 : SUCCESS in 24m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3483ad652bbd4a4b92818b8065db4dd7 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6be03b344f6f4814bab4121f3652e505 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/33b5e7bc7bde403eac86384480eedf5e : SUCCESS in 14m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba0772db1ce34e5b9cfcc0d1000eaf8c : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/54a825ba8d1e4f57ab3c6cbe5e37ce91 : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38e6128d61984dabbc44bd5f0b01a6eb : SUCCESS in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1165306adc84392a725746cfa71112e : SUCCESS in 54m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5740675167234491a4fcb86e5376ab57 : SUCCESS in 37m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5a8374180f04cb4a085534351658915 : SUCCESS in 48m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5834e7921e2a49e89e16e6835df1e81d : SUCCESS in 22m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ec09c51f2a4ae8a9356437d0b0dcd8 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/140c2b6a8d134907862cce9e272c6a62 : SUCCESS in 24m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51868c58ea144975a64eeb0a96c50c21 : SUCCESS in 47m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d62d52a6bef81f49f2667827df012b11207cdfce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 11:24:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2be976ff7e148558c15aea6bb035aa8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fb511bbdada4ae4afb30719c58194e3 : SUCCESS in 1h 39m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/caffe9ce87254f6d88f625958de5e13e : SUCCESS in 1h 22m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e20a9c4d5164a9990f4d2d2a3df026f : TIMED_OUT in 2h 01m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/07103e1cb3ce471faf0d8f88174e925b : TIMED_OUT in 2h 00m 55s","accounts_in_message":[],"_revision_number":1},{"id":"310d576c96a1aa796815f29c35585ebda0fe18a6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-14 16:50:21.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3acbef9655827b8f81557fcf2f459331b0b5bfff","revisions":{"3acbef9655827b8f81557fcf2f459331b0b5bfff":{"kind":"REWORK","_number":1,"created":"2024-10-04 09:17:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/931374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/931374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/931374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/931374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/931374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/931374/1"}}},"commit":{"parents":[{"commit":"ee6f1d15d08ff1e12e7a86ef2eebc76b9481aa69","subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee6f1d15d08ff1e12e7a86ef2eebc76b9481aa69"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 09:02:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 09:15:53.000000000","tz":60},"subject":"sq? be more defensive with log prefix","message":"sq? be more defensive with log prefix\n\nWhen concatenating a prefix with a log message, cast both to strings.\n\nDefault the prefix to the empty string.\n\nChange-Id: Ifa1cd498df6c56d363342d16eb6e9fc6009b454a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3acbef9655827b8f81557fcf2f459331b0b5bfff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3acbef9655827b8f81557fcf2f459331b0b5bfff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
