)]}'
{"id":"openstack%2Fswift~931375","triplet_id":"openstack%2Fswift~master~Ia6234876725981cb21eb00df341ef702b13eb535","project":"openstack/swift","branch":"master","topic":"removeLoggerAdapters","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-14 16:50:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia6234876725981cb21eb00df341ef702b13eb535","subject":"Pass string messages to logger","status":"ABANDONED","created":"2024-10-04 09:17:20.000000000","updated":"2024-10-14 16:50:52.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"29da6eaf0318083365a7b7486d7bd795a1bca9bf","_number":931375,"virtual_id_number":931375,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-04 10:36:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-07 13:56:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-04 10:36:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-07 13:56:00.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"},{"updated":"2024-10-07 14:30:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"f0e37d2d10be812bb99c52873fb661d58b2d8c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-04 09:17:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e36c8a24ea89637d3c9f9bf408b04a248bab6a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:36:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6da4ac8933b44b94b5b5848c04aa0424\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4749ef801f422ea5e4950c74c31ab0 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/5399a4a307a347b7a8a97de42df9d6b3 : SUCCESS in 1h 06m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0efcc59b164a42e199507af61e6e8198 : SUCCESS in 1h 01m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/276c4428c3864799ab7e283dc8beb97b : SUCCESS in 48m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8daa2dd81ff3486daf5ec4003f4cf084 : SUCCESS in 51m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a56604543bae47b7b0adee31b0fd900b : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4322235ffc384ab49f4eae9f3f27ac23 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b54e45f0ff247cf851d13ec64ccc83c : SUCCESS in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba5d72d1f042453aa77155edfc255318 : SUCCESS in 8m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/949bc76b848248f4b64cabe2c6735738 : SUCCESS in 22m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7d197f144aab448da23a140b124ee9a4 : SUCCESS in 20m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7f14ff609e9e4a28a73b4339ddb844e5 : SUCCESS in 24m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/092ea667ce5248398215fa7e1264bb3d : SUCCESS in 23m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/df40f8bcb7b7481787c418f7965b07fe : SUCCESS in 29m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/37828881afff40b5bd8a4f1d3d3340c9 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49159968a4f24eee8c51954ad7bb6c8c : SUCCESS in 24m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ebd733a58e04297a32c228ddbec18f7 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eaa24a5dee4f4e17bb8bccb80c2f5a5b : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db778947f0da4705a75348caf0f003a0 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3b2e60e4a7044844821514d2898a0d19 : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d44b6f052c144cf985b47a08f51ce74 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/948acd4561a547cdaf2c00b36459310e : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5388527145a48b5b2c8d383d3e61f8f : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e623362a84814dc58dce81556d864a01 : SUCCESS in 49m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fee38f5ab35244a3b970930ce0a23032 : SUCCESS in 1h 00m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30029472ef6a42c89deb2f2c3feef281 : SUCCESS in 1h 08m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11644c35c8fa45a39652a6469986b88d : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d6ef8728f0479bbd4079766b7d73a0 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8de43ca19d984a2ab940592477f82678 : SUCCESS in 18m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3d53dedf7634efea91e495b00922639 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":1},{"id":"dd38df5e9a9ae6b5e3b5b9510b9a4d862399ce8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 11:46:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cfdfbe0c57f45968a0c049afce1e67a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce2fe6f30e0d4da8be725c94e7b19eb4 : SUCCESS in 1h 35m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c58ff2baf714cd6abf1804e0f5e4449 : SUCCESS in 1h 36m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e45a7143d18a41fea14cf2da5cd0d6d7 : SUCCESS in 1h 11m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e84932afaeb94b6681405b26ec7f6f44 : SUCCESS in 1h 55m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6b35f6cc0a907a31620dcd26e4a67f5d2a2684bc","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-10-07 13:56:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fce957cd6526604cf082d41eaf084c4cf03cbadd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-07 14:30:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29da6eaf0318083365a7b7486d7bd795a1bca9bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-14 16:50:52.000000000","message":"Abandoned\n\nSwiftLogAdapter fixed to handle these cases","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5752ffb2c634b143e2b4bb6e1c8737ece3feb35e","revisions":{"5752ffb2c634b143e2b4bb6e1c8737ece3feb35e":{"kind":"REWORK","_number":1,"created":"2024-10-04 09:17:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/931375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/931375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/931375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/931375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/931375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/931375/1"}}},"commit":{"parents":[{"commit":"3acbef9655827b8f81557fcf2f459331b0b5bfff","subject":"sq? be more defensive with log prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3acbef9655827b8f81557fcf2f459331b0b5bfff"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:53:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 09:16:35.000000000","tz":60},"subject":"Pass string messages to logger","message":"Pass string messages to logger\n\nThere were a few places where the logger is passed a type other than a\nstring (for example an instance Exception). The stdlib Logger seems to\nbe ok with that, but it custom LogAdapter code might not be. This\npatch therefore modifies some log calls to cast the arg to a string.\n\nChange-Id: Ia6234876725981cb21eb00df341ef702b13eb535\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5752ffb2c634b143e2b4bb6e1c8737ece3feb35e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5752ffb2c634b143e2b4bb6e1c8737ece3feb35e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
