)]}'
{"id":"openstack%2Fswift~931516","triplet_id":"openstack%2Fswift~feature%2Fmpu~I7a828af6e25848d03bc3da256f2d60e4cce6e02f","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-post","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-18 16:41:10.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-18 15:42:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7a828af6e25848d03bc3da256f2d60e4cce6e02f","subject":"mpu: Cleanup MPUSession class","status":"MERGED","created":"2024-10-04 18:10:18.000000000","updated":"2024-10-18 16:48:47.000000000","submitted":"2024-10-18 16:41:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"931516-p-mpu-post","meta_rev_id":"09cff5604ecc5774d80dbe3741e357cf6d9d3554","_number":931516,"virtual_id_number":931516,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-18 16:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-18 15:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-18 15:42:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-04 19:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-17 21:18:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-10-18 15:42:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f6c8e27abbd8c94e63c13d3bfe226ef095c4a98c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-04 18:10:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea44f93227661167a4ebf042d848a269c4b6f68b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 19:11:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d57bf1867f5482fa6c2a6afc74dcdf4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2e0256b5b549409aa9950715fea267 : SUCCESS in 5m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7eb4535f7de74d4493d5127d390f472f : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ea089e3b2bf43c79c254a0da1375f44 : SUCCESS in 11m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc49c23f716149da943446cad61670a6 : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b588a91388884a01a0b01049423905c9 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/783dd3c1c4c54a368bcd38753e256e0b : SUCCESS in 18m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f47e238d79714fda8bd4dd84946cb2aa : SUCCESS in 18m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6060cf21beed48a393302dfbb56921a1 : SUCCESS in 20m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d1197e06b32a4495b398ba926d76bd64 : SUCCESS in 20m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1ff5c644d6c14f61a72915038b04c517 : SUCCESS in 19m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b4ef5c6166004c0f9d5be62e7b1cf4aa : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f71de0a43494475f9c7a995cb6dcb4c6 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/450b216eaeaf448cb770701241321d3a : SUCCESS in 17m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/33174ce62bc443f7953e560077150c2f : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/154d994acb0e4d969d48e97bc03c01fa : SUCCESS in 20m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b637f558fbb84a2fa26d3804abacb511 : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e3eecbc889f4bd1935c7a86eb358607 : SUCCESS in 8m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a05b1d58f027469f9d49dce35fbfa9ef : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d8b7f6815f194741af8756538e821a70 : SUCCESS in 8m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fe2def51ad4411ba6df4091119edb02 : SUCCESS in 58m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9084bcc254748c2bfdb874bc18722e7 : SUCCESS in 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3cc10883094fcda9d3820458996381 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a2247b84274b4ca484490531a7973fef : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"24b8ff9cb4d52e7bb5524381ddcd723157f65634","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 20:38:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3853e5bc814347d1acf96251f7e555c4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/10d524d27d09458fb138cc923548ce89 : SUCCESS in 1h 37m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c8166ea938d4321be979482ec306310 : SUCCESS in 1h 14m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c60d371e36534d2fb7e241ce44e07ade : SUCCESS in 1h 11m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd19c83d0ff64af2ac3a46e9007bf45c : SUCCESS in 1h 25m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fbbb5b47b3a063e0ad276272eb46a1af78e64089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-09 18:31:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d3406f7c85962b4189810b8b6d0a9ae39ca54821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 19:27:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4d61d0f66c449aa3384aa6ae387ecc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e8c3a9154294cfd8f4374bd7f64d14c : SUCCESS in 8m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a0ba19746464a53928ccd01f83db942 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5c852a76e7d4af5b0ed49181ec49f7d : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecce0d86aef742c693d00fc7814d65ad : SUCCESS in 19m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8156758a1d924562a42939d26f5a0fe5 : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/090bf7afab424b929429dbaaebd716cf : SUCCESS in 20m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/49d4543a80d44de283c6d3c71a59a162 : SUCCESS in 10m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3a7740a16b8149adbb6a24644d6d2dd6 : SUCCESS in 23m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cc321cc51c98430492ccacbbfdeb4ab0 : SUCCESS in 11m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/21fa57498d8d49168edf4b11695e7732 : SUCCESS in 34m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41a90cdf835d46a78ef3de7d5da10abc : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0cf74023d3c6479e88fafafbaff0ebbf : SUCCESS in 16m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7bc863249cc402384a19d9e0547ed55 : SUCCESS in 16m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/249872147809488299cbc89cfca31717 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a9e5908a846244ff8ce4996ca87cc171 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2ffd57f985074d5f91d975cec444cde9 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/49c5f56a205c41f1ba05230a9aa4e18c : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bbfe4ab9ce1e4cd1a2dbbd09a08bef06 : SUCCESS in 14m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/57bcd2ff42dd4ebdada2eac3a840b34f : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/968305fe22ff46f9b5a110281a20810d : SUCCESS in 51m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9e4d39cdc574d78906c740f0b09c096 : SUCCESS in 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49b46e42c9748d589ffc8c403e1c6e5 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a91cccb750494d1a873fa197bd1e1e27 : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d283dd38d9431c78435aba92eaf4402a0368004f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 13:21:29.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74ebfbeea7c7433491808cbeeee93ab0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92f42a478fa646ceb6c2b2050d51b72d : SUCCESS in 1h 36m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd8bd156780d475e87e11468cab3443a : SUCCESS in 1h 32m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2bb31e4326dc43638bf7f98f8ebc95f7 : SUCCESS in 1h 13m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c36582c029b4952b7027b39e46b8d54 : TIMED_OUT in 2h 01m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9d2abee65c267d2f04459e32d2a85b1384c132bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-11 15:32:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8839de4799d49f30c613c51c03e5147772eba4a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 16:45:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5ffc208cb924f6bb8d14a180c120861\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40db01e3ec714da5ae43a7ac00674195 : SUCCESS in 9m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bcf6caa92c24690b47bfdb0f5316c3b : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d38c7fff7b94400dbcdfbced0ae17ecc : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20cfb0cfb99941cab175e8ec9afe9c5f : SUCCESS in 20m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42dc25e44f054456b5e13076fcb1208b : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cdb67b765d474e8ebfd7c35ba759f9e8 : SUCCESS in 19m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ad2238fdea641bea0ec91150c596718 : SUCCESS in 16m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/703bdca4994c47539c67fbc9ac815504 : SUCCESS in 21m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c536cc50640448d7ae03e43440514925 : SUCCESS in 22m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5328e13b00a44b2980050418df97271b : SUCCESS in 27m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fea40ec1bb0e4a858618180e1f18e3f0 : SUCCESS in 21m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1a00d0b2f189405da4f3cbc176493ee9 : SUCCESS in 14m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b39e5f9b8e748f08adc866e21f821df : SUCCESS in 19m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0b5264a64134452287e4021b0ba32ab7 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db981bfbec4f440a9e3fa2a5d14238ba : SUCCESS in 14m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/df3cad30d5584583967b08b7674c4280 : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4dd2f34c4ddf46688b8c354ae5baa61e : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/94977ac562ee4de8b23847ed8a96677a : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b99179f2a73747998dbc65f035b4923a : SUCCESS in 7m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0863f72f7d23408b8a580ac701b1c239 : SUCCESS in 46m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c742f89e6a44231af3be3058cfb38f6 : SUCCESS in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce232684bd3c4eda8fc1f886e783a097 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0abd568e30fc4b7fbd11e0a4eb3c42da : SUCCESS in 25m 36s","accounts_in_message":[],"_revision_number":3},{"id":"5a9b19e98e0b7b67e16b06fcfda2c287a1db4bbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 17:22:38.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b666478031c541f39866ee8ba8ec2427\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4476f413b19e432abd8b0e06dfe62a00 : SUCCESS in 1h 18m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c047cb12f4fb448d852642067a78a988 : SUCCESS in 1h 04m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dab57ce3fed54d8fb075edfb1f9f8721 : FAILURE in 1h 42m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3dc3685479c47e298793dda17fdff7d : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":3},{"id":"533ae658681ef80702a7e8ab5247edf1a4959c20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 13:09:46.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2e3438d13eea62d4d082b70281e8600d626f4157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 14:25:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9e6d2dfe6dc4e26996194bac0e9e36d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f24e4fee87c492182f9342aa68383a2 : SUCCESS in 9m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0b715c2ee6d44598d1cb994d208810f : SUCCESS in 21m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/389af3414e5f4efeb19216cd68720e8f : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57cbe1be92464f02a583c8b385264c1a : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5a186f2e71d44eab671d0a744d7fe71 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f7de3c60f97546cc8f364387921795dc : SUCCESS in 23m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d61e08685a7c4d1b8730c2ae9a7b96a7 : SUCCESS in 19m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/370cedfe1fe74dd8bd153da47b78ec9a : SUCCESS in 23m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/05ee571375bf471baa3d37cd152a19ca : SUCCESS in 25m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f6a63774bded4060bbee802e9e82eda8 : SUCCESS in 29m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28a1db6a4fa0456f9354bff6cd20e356 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49d32534e6d1420e91d80078b840088e : SUCCESS in 19m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2f6f523366e34edda2a2244c7fd6c16a : SUCCESS in 16m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1d01b2cf735749e0bce6711820748e2f : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5408d2b706ca4ab2860b57a41ccea550 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/78561a16d4cc4eebbf56ad9f2fc15ed3 : SUCCESS in 16m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85ac85d9f3684e949936dead9238e88d : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7fa5dd471b054cbc9502b1d17850c8ee : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4c9f39297098433386b4e27e8c15fe66 : SUCCESS in 6m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f11233a392954d34a54b7fc6b255fad4 : SUCCESS in 1h 06m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bb9ec756eb44eab92b3ddcf3f53233d : SUCCESS in 25m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b798da56bb5749bd8e4f85b6fecab35f : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad4514f57b114b0f8f075ef2391e1405 : SUCCESS in 18m 22s","accounts_in_message":[],"_revision_number":4},{"id":"3a7c3d26c7fb428be651fdc43614e3854e0f3152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 15:25:43.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6be0b2191dae3831310a43be02a70cc9f9456a60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 16:26:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4d0ffc340164a5bac8beacd21cf2d76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a207b891bfcc4d4c8202e9d084a7ffee : SUCCESS in 7m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d20d0abe30ec414693cf92cd59a8ea8c : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a4830eae1d248cabfb63661cf925894 : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e096452060440e29194c9d9ba5ae335 : SUCCESS in 18m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/381deb3f0f6c4a319f88e859ff759f8c : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbde3edeba9146b68c159f0919d156ff : SUCCESS in 18m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27a9f0dafa284e8aa25c6cce9bed02b9 : SUCCESS in 15m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7338e3e6f731408386dce0423257731c : SUCCESS in 20m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/55b7aa6761554b9c9a6e8ebd27afa254 : SUCCESS in 22m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a2d9ad9a17294810a63ddbec0e0e1752 : SUCCESS in 24m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/285940c94fda467a8b385e66c270eed8 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da1c0fda20ea4801a428cfb6ef2b6173 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9b4129fe7e14ba8a70afa49b7e980ed : SUCCESS in 16m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6c0d98a097bb4b2da3ccd04b2ebdb00e : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/51b815c4994f4344b71f75b5c4a9c84d : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e63b7241d80b46df81fcba769c83a185 : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/12000a860bcb4b55afb81965baaecf95 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb7b38ccc3d743fb82ed88f427fbe4ea : SUCCESS in 13m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ddbdc27cef634e6786b309d07b2da541 : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ddd6ab863874b7a9b9b36ae85dec509 : SUCCESS in 42m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dab1809b5414549829a64e35e0b349c : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a187da10b54fa39505fb3810e48d45 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/246875249e0e413ea15b033222dcb323 : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":5},{"id":"5aada9ded2b7dedaf9eeb9c09a11952369f08ad2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:29:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2018bc5af2489abf27c77dccc54a07\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d79d2b4526784630bbb1e09455b21d26 : SUCCESS in 1h 34m 32s","accounts_in_message":[],"_revision_number":5},{"id":"bb89b7ec8ca1c181e6dc2e36f577da224b1c07a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-16 15:34:07.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ee35d6d23e63aa9b27dc3e4a2118a9b573eac3b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 16:44:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1a14323f6954d18a8160fabdf19cc4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a850766b4fa54118974d94e863fc54d0 : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0ef7ea1b91f4f358403d84da670f4fa : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1eaf0c1bec4e44f8a78c3f1e6f81e6b8 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/553161ad1ca34a59b18d0a9e6de11e4d : SUCCESS in 17m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/abb98c09d810401090e6e66694dc5af1 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7762b955e9dd46dc929207c825128ce3 : SUCCESS in 19m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6e9ccb35eacf455abb8d56c7acc998e6 : SUCCESS in 18m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0214590fc3c0470aa4b48b43d5b70aa8 : SUCCESS in 19m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7eb96c5e5407426a8738837df7aa510a : SUCCESS in 11m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3bdf8f2306024b9cb7d5c500187b4742 : SUCCESS in 14m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba9d0eecad2a468e9ebf4b463ff87ed4 : SUCCESS in 10m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e3a14ec73b443dfad7c3c8ccc4092e5 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/79bebd1026354747b9616beb9f9ab06d : SUCCESS in 16m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aff07c1cd1d9432590dea46b3940695f : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0b023abd0bb74bddbd3998c482a15c6a : SUCCESS in 13m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2d333505978c4477b943f68f014d8e0b : SUCCESS in 14m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6b95f0eec4a48e091648adb3f80315a : SUCCESS in 7m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fb388d8a0b194c2596a0ccf5ad4949e6 : SUCCESS in 12m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8aec8d1d5c42467690842a72de0ae2e9 : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f33baf0af69142e1807ce6c05cfdc9ba : SUCCESS in 52m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56710b4ab3be43729738a4f042047b51 : SUCCESS in 20m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd0cc7c4d4db4ba5a831d03fba1cabcf : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/86d7cef061d34049a7a4161be0f2f4aa : SUCCESS in 17m 13s","accounts_in_message":[],"_revision_number":6},{"id":"31e345fe0ee9975ee71ed2f7a5f255b9268cdd7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 18:46:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c836729845b44bc5a94a4b692b9a5abe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c2056314e6745efa5d1128412c72d41 : SUCCESS in 2h 06m 26s","accounts_in_message":[],"_revision_number":6},{"id":"2d779515c9c917c2ef0b0c8962c5684b0ebc487e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-17 21:18:54.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"17639650a287f58f004b0360c84c116f62252272","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-18 14:50:01.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d2dabef5f0e298d6c5e9946905ad760929b08dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-18 14:50:10.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"cfd11313c31a232ada19f653d6ed325ceb54219d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-18 14:52:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6d3f52da559c3be2780f49b5311782ca142d9c6d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-18 15:42:59.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2107de6293b9ea6726bdbaeb90c4d6fe7c35c5a3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-18 15:43:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2e083ec5746254cb819867ab15338e4c1bcc837e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 15:43:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c051b1c38e6a47c0bb184aa49f1aea1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbcd64bd2d0040cc881c4a96ae4b1497 : SUCCESS in 8m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5058603813744d2db78eb7c79f03165e : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4841980304264945a832aeb8c24eae40 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/faa03f95add34c4c847830affa377159 : SUCCESS in 15m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56ac6dc4d8df41348623e4b3c17600e3 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12557df84d6647e8b3632599c9c9453f : SUCCESS in 15m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ab08202db9041be9ad8374cab68a21f : SUCCESS in 15m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8388be019e034cc1a388271830dd0565 : SUCCESS in 17m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9dc61ff203374baca5f1760d837b120c : SUCCESS in 20m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db427b532ed14a6abfe1b1bf090e75ce : SUCCESS in 22m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ed29034988848d8987b9cceeaf5f804 : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c88355f8828d4d60b6b0136afbbb6ac8 : SUCCESS in 20m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d389926abce406783b42093160cea19 : SUCCESS in 12m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c4cd217f10ce4b6bbd0d887af3fee73c : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3d7c47b20c0c49568ff941659f97b4e5 : SUCCESS in 18m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/faa344d38b604c818066f88f1975bd73 : SUCCESS in 9m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1628c952eb974be7a2fba13e11c5c2f6 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9024bca6df154e61a3b27c600afdbace : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/97a541eaa8e7410d90ea5d3ecb3521c8 : SUCCESS in 5m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/097d9c96459d4004863d377d4acb3a2f : SUCCESS in 52m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27878dca5da94adfb6a8aee09a0c90b4 : SUCCESS in 19m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf282d6ed2748dea3b9bab10480e846 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b87e3d03a7b049c3ab3e724e08a9789e : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":7},{"id":"e6e3565f316ebda9fa2ebacdf7a136b1b0eaad61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 15:43:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6b0f19ec39abad634884a4223e98b8d95db89d86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:41:10.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b2d57327f2c4c10823bddbc917834d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f088336500464b83a52ecadf336d82d3 : SUCCESS in 8m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/65988407f7a9411885ca7464399e7603 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4630e3b2106f45ff9945ff6358865235 : SUCCESS in 10m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/025ab33fc3fe47558504157bdd256e56 : SUCCESS in 20m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/85f2f026f74e4383ac7222b0ba72472e : SUCCESS in 11m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/891a904f13824275a67bf99830d73169 : SUCCESS in 20m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a1481a89da724a78a801d7c0aec37f94 : SUCCESS in 34m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28aa186910a840c182c8948c9415d04e : SUCCESS in 11m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5957b1027dde44769b49bacfdca1e35c : SUCCESS in 19m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c1b27eab22c459d91526b07e82e6339 : SUCCESS in 19m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/feda5511c55e4d00badd8fb9cc52a82b : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d68d900151c84dd8a57a59548fcfa33d : SUCCESS in 17m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/83b99dd50010402d9285c84319ec47fe : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/284635fd36e84e62807723ff4f8bf47b : SUCCESS in 6m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/25f4d85800cf4b91874251d1972d5523 : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/85408191c24c45a5b833c0f2dc215954 : SUCCESS in 39m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b9120c4f34b4268a6e5d64a3ad6e20c : SUCCESS in 21m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b93712562fd411eab2ef76bb8776f6b : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74826bb46f9e4be090f496fbd329ff4f : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":7},{"id":"2a9db5df4dbf6d4e1f02d6437f465983a1349bf2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:41:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"03d6d9ef946229ce965eb658b59e9c169bf1238e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:42:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/168c64543a2c4d0dbdf58bbedb807085\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9174ba32d2044d37be2e856771dbe70e : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dca631230ba64a14a60387f953ba898b : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"09cff5604ecc5774d80dbe3741e357cf6d9d3554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:48:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d502a2471a64c5697c2ca50ad4bd3ac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d14d6596e424a61886bbc52ceddb4ae : SUCCESS in 1h 56m 45s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"564b57835f81a5726a5ecda92abad1d3a14917ff","revisions":{"156e69090e7f8207db43bc56cbb93b9bc749d85a":{"kind":"REWORK","_number":1,"created":"2024-10-04 18:10:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/931516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/931516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/931516/1"}}},"commit":{"parents":[{"commit":"3124a89e0969bcd43a3ef8f636eb506b9e583201","subject":"WIP: func test symlink to MPU","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3124a89e0969bcd43a3ef8f636eb506b9e583201"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 15:36:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 17:14:57.000000000","tz":60},"subject":"mpu: Cleanup MPUSession class","message":"mpu: Cleanup MPUSession class\n\nMPUSession is meant to act as an abstraction of the MPU session object\nmetadata.  Mkae it clearer that an instance can be constructed from\nbackend response headers or client request headers during\ncreateUpload.\n\nChange-Id: I7a828af6e25848d03bc3da256f2d60e4cce6e02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/156e69090e7f8207db43bc56cbb93b9bc749d85a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/156e69090e7f8207db43bc56cbb93b9bc749d85a"}]},"branch":"refs/heads/feature/mpu"},"f6abec4163a7c2d9e9748e1060e483d1ab73ec7f":{"kind":"REWORK","_number":2,"created":"2024-10-09 18:31:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/931516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/931516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/931516/2"}}},"commit":{"parents":[{"commit":"3124a89e0969bcd43a3ef8f636eb506b9e583201","subject":"WIP: func test symlink to MPU","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3124a89e0969bcd43a3ef8f636eb506b9e583201"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 15:36:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-09 11:19:50.000000000","tz":60},"subject":"mpu: Cleanup MPUSession class","message":"mpu: Cleanup MPUSession class\n\nMPUSession is meant to act as an abstraction of the MPU session object\nmetadata.  Make it clearer that an instance can be constructed from\nbackend response headers or client request headers during\ncreateUpload.\n\nAlso add property helper methods.\n\nChange-Id: I7a828af6e25848d03bc3da256f2d60e4cce6e02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6abec4163a7c2d9e9748e1060e483d1ab73ec7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6abec4163a7c2d9e9748e1060e483d1ab73ec7f"}]},"branch":"refs/heads/feature/mpu"},"4c4465d6e6c2801a69e381e5cdb8041e5042dc5c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-11 15:32:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/931516/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/931516/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/931516/3"}}},"commit":{"parents":[{"commit":"4cbcfd68e455ee3cc20900dfb16f1063e4003765","subject":"FakeSwift: capture unexpected calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cbcfd68e455ee3cc20900dfb16f1063e4003765"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 15:36:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-11 13:12:11.000000000","tz":60},"subject":"mpu: Cleanup MPUSession class","message":"mpu: Cleanup MPUSession class\n\nMPUSession is meant to act as an abstraction of the MPU session object\nmetadata.  Make it clearer that an instance can be constructed from\nbackend response headers or client request headers during\ncreateUpload.\n\nAlso add property helper methods.\n\nChange-Id: I7a828af6e25848d03bc3da256f2d60e4cce6e02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c4465d6e6c2801a69e381e5cdb8041e5042dc5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c4465d6e6c2801a69e381e5cdb8041e5042dc5c"}]},"branch":"refs/heads/feature/mpu"},"55545e638fcf974f560df6a4fea93a7e7dea9459":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-15 13:09:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/931516/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/931516/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/931516/4"}}},"commit":{"parents":[{"commit":"2b0b6553051a0441cff0d6d93bac88c3fe647e6b","subject":"WIP: func test symlink to MPU","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b0b6553051a0441cff0d6d93bac88c3fe647e6b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 15:36:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-15 13:05:20.000000000","tz":60},"subject":"mpu: Cleanup MPUSession class","message":"mpu: Cleanup MPUSession class\n\nMPUSession is meant to act as an abstraction of the MPU session object\nmetadata.  Make it clearer that an instance can be constructed from\nbackend response headers or client request headers during\ncreateUpload.\n\nAlso add property helper methods.\n\nChange-Id: I7a828af6e25848d03bc3da256f2d60e4cce6e02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55545e638fcf974f560df6a4fea93a7e7dea9459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55545e638fcf974f560df6a4fea93a7e7dea9459"}]},"branch":"refs/heads/feature/mpu"},"35539f52caf6816ead510a9d6605252c776238d3":{"kind":"NO_CHANGE","_number":5,"created":"2024-10-15 15:25:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/931516/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/931516/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/931516/5"}}},"commit":{"parents":[{"commit":"53f828250b5a9e4b1caecffbd7bb4371ca4a7354","subject":"WIP: func test symlink to MPU","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53f828250b5a9e4b1caecffbd7bb4371ca4a7354"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 15:36:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-15 15:25:28.000000000","tz":60},"subject":"mpu: Cleanup MPUSession class","message":"mpu: Cleanup MPUSession class\n\nMPUSession is meant to act as an abstraction of the MPU session object\nmetadata.  Make it clearer that an instance can be constructed from\nbackend response headers or client request headers during\ncreateUpload.\n\nAlso add property helper methods.\n\nChange-Id: I7a828af6e25848d03bc3da256f2d60e4cce6e02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35539f52caf6816ead510a9d6605252c776238d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35539f52caf6816ead510a9d6605252c776238d3"}]},"branch":"refs/heads/feature/mpu"},"aeee0e3540542b56f191644eb6284217e7522de5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-16 15:34:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/931516/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/931516/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/931516/6"}}},"commit":{"parents":[{"commit":"7328310e157128afa3a0daab2453b4e203f6d40b","subject":"mpu: pad part-number in part names to 6 digits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7328310e157128afa3a0daab2453b4e203f6d40b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 15:36:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-16 15:33:53.000000000","tz":60},"subject":"mpu: Cleanup MPUSession class","message":"mpu: Cleanup MPUSession class\n\nMPUSession is meant to act as an abstraction of the MPU session object\nmetadata.  Make it clearer that an instance can be constructed from\nbackend response headers or client request headers during\ncreateUpload.\n\nAlso add property helper methods.\n\nChange-Id: I7a828af6e25848d03bc3da256f2d60e4cce6e02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aeee0e3540542b56f191644eb6284217e7522de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aeee0e3540542b56f191644eb6284217e7522de5"}]},"branch":"refs/heads/feature/mpu"},"564b57835f81a5726a5ecda92abad1d3a14917ff":{"kind":"REWORK","_number":7,"created":"2024-10-18 14:50:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/931516/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/931516/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/931516/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/931516/7"}}},"commit":{"parents":[{"commit":"7328310e157128afa3a0daab2453b4e203f6d40b","subject":"mpu: pad part-number in part names to 6 digits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7328310e157128afa3a0daab2453b4e203f6d40b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 15:36:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-18 14:49:38.000000000","tz":60},"subject":"mpu: Cleanup MPUSession class","message":"mpu: Cleanup MPUSession class\n\nMPUSession is meant to act as an abstraction of the MPU session object\nmetadata.  Make it clearer that an instance can be constructed from\nbackend response headers or client request headers during\ncreateUpload.\n\nAlso add property helper methods.\n\nChange-Id: I7a828af6e25848d03bc3da256f2d60e4cce6e02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/564b57835f81a5726a5ecda92abad1d3a14917ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/564b57835f81a5726a5ecda92abad1d3a14917ff"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
