)]}'
{"id":"openstack%2Fswift~931517","triplet_id":"openstack%2Fswift~feature%2Fmpu~I376df764041943c982d71c4e20f4af37e69de498","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-audit","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-29 09:58:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-29 14:47:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I376df764041943c982d71c4e20f4af37e69de498","subject":"mpu-auditor: audit sessions","status":"MERGED","created":"2024-10-04 18:10:18.000000000","updated":"2024-11-29 14:48:39.000000000","submitted":"2024-11-29 14:47:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931517-p-mpu-audit","meta_rev_id":"d1e6445e7b1131c401c23ed8605749e3d14e739e","_number":931517,"virtual_id_number":931517,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-29 14:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-29 09:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-29 09:58:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-04 19:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 19:12:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"8d73cbbbbdc2f11c21e15d24b7a9718938cc5311","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-04 18:10:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e20dfec74eb6ae81899605e279bfba084994e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 19:19:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65338430f3d04fd885e375155e916ea6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c1de8d64a5447aa159965842cb5f09 : SUCCESS in 8m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f42c3f5f87e44209b02c4eb80b92e85 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ad2bc56f5184828bb9084e1c06dc63b : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d12b52558f44305877e65763daef007 : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/da0f6b16c3ee4c7a814d4fdc36a65458 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6e3896c945724ea09441c384b97f6fa4 : SUCCESS in 20m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f6f781bb43e4edb9a9c14ff3de4894e : SUCCESS in 21m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2b74b069d8514608b1cf29d5e9a93243 : SUCCESS in 20m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a628dc33e62043a198bf5a76d071f949 : SUCCESS in 24m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/178ea70d901d469b97b85514eb960ffd : SUCCESS in 28m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae1ef8a46df643a6936b56d7c90f2686 : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ee1bfb9d6d184bd19a76e039ff977ab6 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0fe511efc9f843d18b37b1c59d14d0b0 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3bf08159882b44839447525dc51640ae : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c68639393e8d4ef5a1479404fe8a01f2 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bf6a901fb2da4f8fb950fb78be0b9b9a : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8bdc5be7f30a403dae671563f2d41b53 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d253708c0ffa4329abda0c476d923d73 : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ec17a703b384c5fb42bd15d1a59f331 : SUCCESS in 6m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d0630f7814d466c8d2d6f299d00ef06 : SUCCESS in 37m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f8934cdeca644ac885d6991e85e1cc7 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64215541ed9f4d959d63afb1f44598a1 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32648ded325444f69e494499d733859b : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a16edd21938794c85c9e8c2e54e91fd49a856fa8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 20:37:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7b01e9b8ea94133b536926ff766b702\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4e401e94a054c188ae5f9ee11a2e222 : SUCCESS in 1h 03m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44faa337f9c94e16b09a62369edea2c9 : SUCCESS in 1h 25m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a8e006461b04d6daad135e1fc1df02a : SUCCESS in 1h 17m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcbc0b61ed684e0fa7006360fc5ac154 : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"00d39f9e054e9267ce9428f2517dd20f008d1c2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-09 18:31:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f0f68740de9cc5e223d347639861b3e82357d447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 21:03:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02c2e881af5945638eca47a505313448\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ab9c5f48624fc6be05eaacc0b4c89b : SUCCESS in 9m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d347d3b3e2446f186e99485f8efabf3 : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5087c2fabb954c2b8f49e95739494bf9 : SUCCESS in 15m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24ca189963bb4bb9ad2ab396c7554707 : SUCCESS in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aec4bd0bfeee4d3792923975e0cc1018 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/036d83f581da4a8384efb6a94eea998b : FAILURE in 15m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f257476d83c74058a3c929b83f268532 : FAILURE in 10m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b8df7954a2eb4238b1209176949c17ac : FAILURE in 20m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/af2ff768d90d4053b0a413eba1a9aa84 : FAILURE in 17m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6d4d3766a2a44c4e888d19a8d5b930e7 : FAILURE in 29m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/27283bb28c9044978f756502b823a567 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6c0615f544d400e8b1f6f3bb5b59a39 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d3916137d29438f82cb5123791313d9 : SUCCESS in 16m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b9f291b1a8474a3ebd830eedcf64ac73 : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a624ae4d924a4d068254927ad3721416 : SUCCESS in 13m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b462bcbeac6c456484a2c3049947664d : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2de3894f710840469d70c88059a039c0 : SUCCESS in 8m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db118198929f4405903bb7ccff5b75dd : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a3442f264ad4dfdb06ab2316825de0d : SUCCESS in 9m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2363355ec0842faa3bacb5cc306c711 : TIMED_OUT in 2h 01m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d84ed1cf07fc4009afa0ea3d3f4b4719 : FAILURE in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6222386b72524012afd63d46925956b7 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/306fa5ede0d44565899872bff2a9cf55 : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"937e4f82201c18b6f324dc1db07cdbf14264d86d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 13:28:47.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1951e69123084141a5794768ff77c3e2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0063edc7424a4cd38235dd7ef98eba8a : FAILURE in 1h 57m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/849692f9ebbe49f381db137f18e80db5 : TIMED_OUT in 2h 00m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ca91f1798a74f8590a48172c6256708 : TIMED_OUT in 2h 01m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b8aa3b548784b2da0ae11fe58556636 : TIMED_OUT in 2h 01m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d24f35fa3c02ee46faeaaaa9115a0a8176964944","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-10 18:24:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6978b288189a9804374800b9f9769bb0132184b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 20:30:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5e99dc323be43e4ab7db057ea431fa4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a08a58dcc6d417f965a66a2eeb65ad2 : SUCCESS in 8m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ae33b8533824fcaa14ce26bcaf45cfd : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0d3a22992f648a1879db364c10cf2c1 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0dead0aac64845d79452b7836c4715cc : SUCCESS in 15m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67f1da60519a4a18b901d1ddf6151d6b : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2574bb368f8b4e11bd0b3922de428ea7 : SUCCESS in 19m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00c0e784c56741429efa1eb4756a7c61 : SUCCESS in 12m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5348822178b442d7a1a2840e832391a1 : SUCCESS in 20m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8049224b1a5e4caeac25058f2edf7c51 : SUCCESS in 13m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0798a730246248a6819d714abc4bb721 : SUCCESS in 29m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/69d043181ee44fab9a311470d244251f : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2d27c1cb7e2f4961bb49d912a2df3328 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/563d5ae3264c4d4c88da18364c059d5c : SUCCESS in 18m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81509ae81a4a417bae5e557fd0f0fd0b : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/21565fbf531f49eba2e826d2af3807b4 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dfa05026d7554c95a496c1087aab1039 : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4129807272bb44ed9752bd9484bc0473 : SUCCESS in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3d2a07ece4ff450bbd7b40746011c009 : SUCCESS in 11m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/353fd0076db24821b2abb0b279d38a1c : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1b49735a0a44ce7862185b928f67df4 : TIMED_OUT in 2h 00m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a502eece1d6140a99fd71d65955a10c5 : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca101fe2112c4cbdb3cc3ef40ca74e82 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6428d0b1ac4f4fcabdcc5796de69a0c6 : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":3},{"id":"85626f4b241952bc2c930f8d61a258acabc4e888","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 23:09:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8a03cc2d0f0429396c0ced71fbeb597\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24f73e0446d14a1892cb2829b0f4fe26 : SUCCESS in 1h 39m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6f64afd244543b98e61ce8057018dc7 : SUCCESS in 1h 55m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/63cf264209584f848e86031391969257 : SUCCESS in 1h 38m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/db6b695651e8489fba101bd39b64e19f : SUCCESS in 1h 30m 32s","accounts_in_message":[],"_revision_number":3},{"id":"b62c8df962357576f9975a92d5c21b9aaf040f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-11 15:32:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c35d35e8eae26cdc9b2557e1bbb23bd336d8f68c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 17:27:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99f2b73490cb4c5d8dc598d84b8ded52\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d1622fae6614ac68b6c5124c1777c0e : SUCCESS in 1h 23m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b3f86b9ae554e72b8b3928a2dbf09b3 : SUCCESS in 1h 26m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/615fef97dd864641a889b56df8f1f165 : SUCCESS in 1h 15m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f14b442732c4285b01b0eb90cdc3a85 : SUCCESS in 1h 08m 32s","accounts_in_message":[],"_revision_number":4},{"id":"0d9dcdd70f0b41d80cf44ba37d95f3b72b57ed46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 17:40:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61b885c53be14de0a97d0c385e5c4ab9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9587b0a1c1624b4cbd4e5bcd05f8d3af : SUCCESS in 7m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b49c811bbe8141b489898c7e3a317fb0 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d099ca90431646708cbc4b9fb04c7732 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4027cdc12dbf4a258c34d96fb61def62 : SUCCESS in 19m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a749201751ee4bcb8d1a6b2c8d6c5cba : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a310518f5f664d008be30ae1827e393b : SUCCESS in 20m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d2b2eedc917494f983c7b0a4c433d51 : SUCCESS in 17m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/37b251fc544f4310b3f12397bcf7f495 : SUCCESS in 23m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/91a2384d499946f89466183358209aff : SUCCESS in 20m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/24096293d371439f909d7ed06b597452 : SUCCESS in 34m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b3c3dfbac31454fba9d1ce2287a510a : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7238dff37b9413b897ceb0a27021a82 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d7785e1c57d346ab839cf957f3a6f2b1 : SUCCESS in 19m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0dcc51eae73b41079f142173ed41cbba : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a8553c3ce274483ea392304d4a1364e7 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f77ce29285cf4fea9f10044d26ae929a : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89fb0e4dbddc4c31a66b94f921fc6bf5 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04dd4da671734c939a0dec89dd864c53 : SUCCESS in 14m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/96a3a4247d5047ea9e75ff0d0717954c : SUCCESS in 8m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/61ba4ece676549a4b8a0219aabab7fa0 : TIMED_OUT in 2h 00m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4d75988b8f54398982165dbc9b4e15f : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a252eca7e1cd4129a4e0702f639f3ca6 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f8fa29566eda48d6a99ce962f446d48c : SUCCESS in 19m 36s","accounts_in_message":[],"_revision_number":4},{"id":"26d0df5a6bc2261e0c73337e8377d94e9e07cb84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-14 16:42:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0b95a6ad4feee6862e94b8136daba12ef6083909","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 18:24:01.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/766ac29b04464c4abfb7ce6c2ab9d844\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03e2796263bb4b969b9d684c6fc857be : SUCCESS in 1h 27m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4e868be45cc46f4b5cfef35c184891f : SUCCESS in 1h 20m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8787a414882d4613a69bb8930364688a : SUCCESS in 1h 33m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d04dd6afc8f4c14a3cfd1e43e680493 : FAILURE in 1h 39m 20s","accounts_in_message":[],"_revision_number":5},{"id":"aeae0924d605e6a3fcd1e4e0aa45d9b93e2db1d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 18:49:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70dfc798cfb44d03983b37e838b14746\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03fd59fba4942a79d39711c59e3f3cb : SUCCESS in 10m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fb687ae65984e58aca320ba6ff54e54 : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1582d647192b4dceb6fbefc5a94a569c : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/575cd20ac2194098b829381b69000425 : SUCCESS in 18m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/535ed7e824f2463897b356e4f98d1347 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b9c0f620bdb4e749332157135abdc29 : FAILURE in 3m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84981493635d418a893e982cfe523e9d : SUCCESS in 19m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e48ffd5abe5d4798977c52951ebc2a2c : SUCCESS in 20m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d804c11ea8604c748634a9fd85c88f6a : SUCCESS in 14m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/785d0b89d1c74e93869e3bf75b3a7e38 : SUCCESS in 20m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3736d439f870402d9a1663dc4ccca82c : FAILURE in 3m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92a31e12ce234f838da7251c33ac831f : FAILURE in 3m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ed513b62dc648d690b90b530574a5d8 : FAILURE in 4m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/839762838e6246ba9b1cc2d18c0473e7 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aa90e90a4d864cdab1a38026e176eecb : SUCCESS in 17m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f85f61dda8994978bce395257ab9b0fc : SUCCESS in 13m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4db4d6cce05d422eaa970dbc26de149a : SUCCESS in 7m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/819f49888d764d4b8adfe98c6ea0b0c5 : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/276b53c09baa42c9999e1f93967cea10 : SUCCESS in 4m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7d2823299534f5d9bb498f3a440f7af : TIMED_OUT in 2h 01m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b1e048b135b447788bb71031dd5a324 : FAILURE in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60cdd849b3674a839cd89d5d4bc10465 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/747ec7a249cb490fb6027fcbbeb6b6e7 : SUCCESS in 21m 46s","accounts_in_message":[],"_revision_number":5},{"id":"551d9bd4705bc05ffd9ed3eb2877cdefb15acdca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 13:09:46.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c4ec508b7a42c188eb0d926dcce5a00bd529bb16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:19:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a2a36d1cef4412780f2ed0db02a560d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54f4f13a32644483ac1c393aceb6e5b3 : SUCCESS in 13m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d230e7a0846461f973e4d94df5537ad : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c80184e8e7a147ada8a0084d1d84e41e : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f0c160de3c44b5bac9e3635c74f37ee : SUCCESS in 17m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f69787b3f9cd44cf9bc253c86ebf68ec : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6f8067cf01446aaaab60027c2178ec5 : FAILURE in 4m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7adde452b1444bcda59ce051bed81fc5 : SUCCESS in 20m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5dd4fc9a79a4459fae5fc2b048d513ba : SUCCESS in 26m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/140dc7284ca941f19fd87d65f7276f33 : SUCCESS in 28m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/aac717908f2d45e48975f0208226820a : SUCCESS in 29m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9503a3fb42a44997b3824561aaaa653e : FAILURE in 3m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ec53260dd3e42099755b03ac544e8cb : FAILURE in 4m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e7b2fff4fa3a4d83934622e3f9f24e56 : FAILURE in 4m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/af7a0f93c85a4a3a922dc1a488080de2 : SUCCESS in 12m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/661808ae6ace46f386c04f3cbc4314f2 : SUCCESS in 20m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/81218c8ab0ef424aa1f54a01c9ec9846 : SUCCESS in 19m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d8bafe506734dc386f5f220ee6f6f18 : SUCCESS in 9m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ffa3ac3654b4d108dd9bfd160ee0bd0 : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac5858e9b987475e9e874d5a74ad074f : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc827517b7a14fc8accc0e1adf71dbc7 : TIMED_OUT in 2h 01m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/693c75314e7e4c9a95b3f00145e0c062 : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ac13389ef849bc8ee739bbe0cf0667 : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d28001b8ab742249cb8024fc865276e : SUCCESS in 17m 34s","accounts_in_message":[],"_revision_number":6},{"id":"d9ebbffdadcaf32cb64c45aba7929300db86caf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 15:25:43.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"684e517220cdddd2e8c4f4fe535d7df86edb79a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:28:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc34ff31ac2403b9d8174eb45d89a71\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/084e7575a5ad495c9c38eb8794001d07 : SUCCESS in 1h 33m 12s","accounts_in_message":[],"_revision_number":7},{"id":"c4615e986e5ef8d4ed664135414815cd96893986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:34:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe3e45242fe046a2aede480d6e768ec0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebfeba2605c846e1a7f783202b24c5b2 : SUCCESS in 9m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e141bfabe4274278ba4f1272f850424a : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0382d41c639b49f38bdb9d84179399d3 : SUCCESS in 18m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75f812419e574072899e121465fa2458 : SUCCESS in 16m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7c5f08175a924e0bbf2c64cf5679bdf9 : SUCCESS in 7m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2f9981acfcac4ec496506c77d3c5fb20 : FAILURE in 4m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac34b1aa7f8a4a98890a02588b99633e : SUCCESS in 15m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f0121576340b4e1e9e974fee51c290e2 : SUCCESS in 20m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b4bea0fc078f4974aee741f0bf70f896 : SUCCESS in 11m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4495194869574b8b840f20e4903e705b : SUCCESS in 38m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/890c8e62592b41cdbebfb2711be55501 : FAILURE in 4m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce4218c450d6455795b386acc39c05f5 : FAILURE in 4m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76eaa7b3437548e5b16bf5d00bf03d43 : FAILURE in 3m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5eff207061af4995b9a0d4657cda9de6 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5776622cc2624000b7c852c25d09cd19 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/334b782936da4c85a37faefc2f2bad41 : SUCCESS in 13m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d863bd89d8f0474093ed938d96c5c400 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f32c30d8a4b4aed807c8062de7bc4a5 : SUCCESS in 14m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e2b4ddd821648408427b447735088d3 : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/548ecf705bc7422a9c586ae9746bff2e : TIMED_OUT in 2h 00m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebf872af961245559ca5197e0f0b245a : FAILURE in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93b2141483e45f98bc55469dd826fe6 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa47f602a909427cb6a384602d037116 : SUCCESS in 22m 59s","accounts_in_message":[],"_revision_number":7},{"id":"3a394aa4264f50301c8934f1c8ee34b7e3523a56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-16 15:34:07.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"7677a70442f52c35c8884793bbdc42402fc43259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 17:54:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b71a93ffa6cf4233b0282e6b5c395fe0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2df04331794e299d5729540aa716ff : SUCCESS in 8m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59b2d466c4da4cd2bd6ade2bb3571dc9 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df8273119b4d48758caa1d3f6161de27 : SUCCESS in 13m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f88d9d22a7ef4cd3a89130845d6a9509 : SUCCESS in 16m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee64d2c8a4f24fe188bef709f3b03a77 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d793a8fd55049ecbd50681dcd3fddb9 : SUCCESS in 20m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf43ca326a54da8a5ded5e451c8e119 : SUCCESS in 17m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9b0f1b5d1cc1422ca4e300b1e4868f5f : SUCCESS in 13m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3f1626f2c8b44aee89f5e2dc94145bc9 : SUCCESS in 13m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/30ef1ff440aa454c86cf2e968477098e : SUCCESS in 27m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d3dceadb6ba14176b7645435a0fbaaa9 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/99e1cdc839b749d9bfe09c978aa394e7 : SUCCESS in 15m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d365b5205fc1426692bfa21ffa0bbfc9 : SUCCESS in 14m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9c3b0014e8db4330af5a3b729c607527 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a2312b2e02ee4c9fafb20f288ed1bed0 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/deb5f29b33ba4ffc909d116cabfc9dfe : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2ad0841a3674fe5a701745f3fa74e21 : SUCCESS in 7m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bf67f19f85441aeb257a23f307e9f6e : SUCCESS in 9m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/81f2a8e8ac584c34be328c832cf78435 : SUCCESS in 5m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/701f2929c2894077af34ebba902ccf55 : TIMED_OUT in 2h 01m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c8f2c931a284e6590719eefe1aad8a1 : SUCCESS in 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934364fe386c4d9fa470c3049ab858b1 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5bd9048682345528bb743e381f5601f : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":8},{"id":"91326ca628510c8166a272709e0877964798e32c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 18:42:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a1f8fde5334fbb9c21e17c8d28da94\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d63a84819d734a2883bbbdf77adc6d19 : SUCCESS in 2h 00m 23s","accounts_in_message":[],"_revision_number":8},{"id":"7ece7d9b78ad3897ae047b138396c60da3bcb99f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-18 15:34:37.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"796ed76faef683e4c062a78be4953a225adb0523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 17:36:27.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9358b958c79e4cf386bfbd4d31766f8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d8fe2e184d4e5ab8029e77eb46f583 : FAILURE in 7m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f387baf2a394aa98f300a84da0798c8 : SUCCESS in 11m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e00fb26f28e0424290c433a254c3a596 : SUCCESS in 23m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0abc8a50b5914a939ac5954bd053a1f6 : SUCCESS in 18m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97e86ee620f744009b03dede2d4bebf3 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7efbf89c62f544b195e90667152c02f4 : SUCCESS in 17m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3bcc8df71954876b837f167db8ddfe7 : SUCCESS in 16m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3ac3a57ca8f5435bb7b5d5bc4781694b : SUCCESS in 20m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8d6a8f82caeb409bb2925c48e315d263 : SUCCESS in 15m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/687e27880055488887422ecf813eff54 : SUCCESS in 24m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6af6b221f56049f98e17118961cadd68 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b28df91f4e44060bb62dc5b81710221 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56cbe8eccd2142f985446926f17ced79 : SUCCESS in 16m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/94aaf2a4119c489f9bb50d723a8d25c4 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dbeb7273a92b4d0ba9eba7978f2fcc6f : SUCCESS in 13m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f724842d61ae454cb2d08525e464e7e8 : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/023549a49ff04b58a146e04a244b46d8 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b801f74fac264ad6a2455a5394f0d1b6 : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f33772990cd49c29bcd76daf530f0bd : SUCCESS in 4m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2494344ae1c40ccabc09a920380e720 : TIMED_OUT in 2h 00m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e03be1a74b94ccb84717f3cbd47d402 : SUCCESS in 19m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcd0b5b752d447c5bbdd3f4261a2ee75 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7aaf98b1196a49bf90b62166605fe212 : SUCCESS in 18m 36s","accounts_in_message":[],"_revision_number":9},{"id":"f12a5aab33a216d47f6bcded17f79996a0156adb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 17:36:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c3c14381f243a680327e395999d422\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/24348f361947469db463b47c4414289a : SUCCESS in 2h 00m 10s","accounts_in_message":[],"_revision_number":9},{"id":"0fde9e82932ca32cc9a2e5e31658a00b7965cccc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-21 12:14:57.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"3af1a8d13e99689ae60bfdbec93654c37905d325","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 13:27:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afa42e0b3889486d914fb3fec11c9705\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d5d4f57882f04f48837fd8503e6ba2f8 : SUCCESS in 1h 11m 15s","accounts_in_message":[],"_revision_number":10},{"id":"8ea58834eff5cf958017cb3e8392c8beff04ede8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 14:46:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af69eaafcb82483a94d95e821c3a6b08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa7890e38e04cda965affb090077bcc : FAILURE in 5m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06505da2d186486abc90353bc5466e3a : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d352889db974c878742c6c8093d8641 : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c0dfa3b9a504b3a966bb54cdbf4dd44 : SUCCESS in 21m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa8e680fa9654516b07b35faf8a8614b : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/170e19cf987e46fba6ca083446eb50ef : SUCCESS in 20m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ff52e320a514c0588ae654f84eeb45b : SUCCESS in 18m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/675275264a0549f9af4343c090c18d4f : SUCCESS in 20m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/11c5526f7bf743c4a7819dbd63231bcf : SUCCESS in 20m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3aff19007f554e7da0d0fd1f8e1c042c : SUCCESS in 34m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e81560a567194bf7be35d03a8155aa89 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f78093bf05ac408bb1d6231ce60b410f : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b165983a60b24f6ab2615f90162a63c1 : SUCCESS in 18m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9006aecb2c144a32baad2efc27c1838e : SUCCESS in 12m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/832425b6f9884cc1aedf9e64efabfe43 : SUCCESS in 18m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/10142594241b4840818cc9c0ad6e4d03 : SUCCESS in 8m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2a53b5bc09954559ac762ad0d3d7bfcf : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65eabf15ce1b4279800979da1da31ef6 : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/15536984e65344668f182552f9c21b59 : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/31eeb17097c546bdb51e3775d4434449 : TIMED_OUT in 2h 00m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed5d94d2594846e0a179c8d6e222d2be : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017cb8a165b74185b60298bca6fe899e : SUCCESS in 3m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1655c975a2b46ef9bcb22b8ef9e7e7a : SUCCESS in 24m 20s","accounts_in_message":[],"_revision_number":10},{"id":"b2d336587a2ab241a3ef48408460457f0d010b94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-23 16:15:59.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"6f02acaeadea313b7e0117283138443fb472e3bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 17:48:51.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2b4054bd4445d185926e4a645e28ef\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea573a73a8764028b98af3895482ec4c : SUCCESS in 1h 29m 10s","accounts_in_message":[],"_revision_number":11},{"id":"b1e6e702c256d9cdf178c2f1abbccd2a20062f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:19:47.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c67cc20bbadb45bea38ccc75bfc4a808\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe48a832cb942008836d510a167caa0 : FAILURE in 4m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf7303b312c64dc5bed7deaef354fee7 : SUCCESS in 15m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c960b92dfc04c07b4b8b48419236f31 : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e9b726463564cfbb3097b8563f88304 : SUCCESS in 11m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/587f885f075f413981e8750a79a766b4 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e67a0fff5f224d92b6dc8c9bbd7b9cd1 : SUCCESS in 18m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/71835922e1a54da1a2ad4658a08b8b35 : SUCCESS in 10m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/57db03e514504186a97511097f566507 : SUCCESS in 18m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ae87fb61439c4d0f9ab6269d7dfd7517 : SUCCESS in 22m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/722bedff87fa44ad9e2d967039cee7aa : SUCCESS in 19m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/752ecfe6474647fca5931601c5b82877 : SUCCESS in 20m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/690b37784e9d4b028dfe0f747875a348 : SUCCESS in 19m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c002b3389f314377b1c03de7b7ca7836 : SUCCESS in 18m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/391c6925401c418b8ac73516dba12b2e : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3709bfbe189b470da70e47d86cdd2907 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0706f687d46042ba8ffa2a686b26ef8f : SUCCESS in 15m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d375e090f9a84775a52a0478e5981b70 : SUCCESS in 7m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4656a3ca83b541d2b7a19aa9ce932022 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d242858ecab046889c2ca15801048211 : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a92fe1f78404f96a619e3b1709a753a : TIMED_OUT in 2h 01m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37d228ab1c694192a9f3fb21074e97ba : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b30f6b71edf4d27803875f6b988e18c : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e58519cd5b442cf978fea560c39a692 : SUCCESS in 17m 42s","accounts_in_message":[],"_revision_number":11},{"id":"8554316494ccbde1322d873b420080a06e701ad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-25 12:01:03.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"925491bf78efd32ad1cd04b6de24506e6e53491a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 14:08:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea7c712ac978485fa2dd1027d02e6b31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093e6a46cf774e9688ffbc8725aad806 : FAILURE in 5m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f4f0f08651f47609af0f8cc3abe46d4 : SUCCESS in 11m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e5646959974454a9786cbc430c33452 : SUCCESS in 22m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e005b629b3ee490ca7d5a1068489df47 : SUCCESS in 20m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e22156d1e46c453890b14f0b4e08b3ee : POST_FAILURE in 33m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bddb4ed83d7f40668d856d7d66282a43 : SUCCESS in 18m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3aa7c8901951462eaa15de355f8fe7d0 : SUCCESS in 24m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/44ed08a8588c4f18a5f7de62089bb07f : FAILURE in 3m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/61db111975d545d1b5c56fd1a0bbd3a0 : SUCCESS in 18m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e56265234f42410ea84d6a226f2c5945 : SUCCESS in 28m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/088e944746d940599c18fc1333ef311a : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/86340575f14d4cab95d7ced1f876c76f : SUCCESS in 19m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/426fee59094a4f8ea1ee3a0981ce54e3 : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c165310ce054417ebeb07026c0d8b7f5 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8ea0f8a469ec440eabf975710f62a5e9 : SUCCESS in 14m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6e3c855ee5d046f5bd31eed59f2656e8 : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c3f0002a30314541aae64cd321a067c6 : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c47775f7f28144f189add4be8939dfa2 : SUCCESS in 11m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/182cd909761946419afcbe6832704256 : SUCCESS in 6m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/45069777c41e4a9ea3cf954877e923e2 : TIMED_OUT in 2h 00m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56e5b885906d49e4a739dce8f8a386b3 : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d2e95ede4345208a19811ab3275f95 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fbd613fe00044bd7871ca8f450dd613d : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":12},{"id":"8b0ba0366bc458284af40f23e1070dd0ffc95e6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 14:12:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16e9a6f6ecb4415cadd8b56c20ef1f1e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d0d6a29d86c4019b0aa573ae43ff358 : SUCCESS in 2h 09m 22s","accounts_in_message":[],"_revision_number":12},{"id":"d828b84b1ee6732bcb865bd36c8854c8a178dc5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-25 18:10:08.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"1edf8e862b8dd79b17de77c7b1e4dcee86e094da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 19:52:53.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c089918f3dd4d0e886e91d58dd78822\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb051e225c64405bb9bf1d34bfd9e5e3 : SUCCESS in 1h 40m 11s","accounts_in_message":[],"_revision_number":13},{"id":"3966b5889a7466b608e1cc046fe86c5b6e6163d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 20:12:32.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b09ec8d7518140919a011fbe66d2f257\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f047c2b1988a4ee18c2b788e3c8e00ae : FAILURE in 6m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f26fc235b6dc4d3081136e7922b19bd3 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a671e2f5866c4d0b9813c174cd39c00f : SUCCESS in 14m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0e00f5153ee45e8ad90fb7503c9e057 : SUCCESS in 12m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd00d76473c54729a99ea423abbbd6d5 : POST_FAILURE in 35m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/603c516a1f9d4f699a45eef05b10dda2 : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2a4c3fb4838405997be31a590f2601c : SUCCESS in 10m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/abf5b14cdaf84e9796ce9db1a4516911 : SUCCESS in 20m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7ff9fbbeb6db42caa50ec20df5343729 : SUCCESS in 18m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/555cda9720f748fda4d497b71c700517 : SUCCESS in 21m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5369063072d940849f323bf7539e712f : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/162a52afada2407e807cd44bd9b70d61 : SUCCESS in 10m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a9e9e4f417744994bc190c4c1d28c041 : SUCCESS in 20m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8f70e5e9bc1442a6814dd842a92d4118 : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/978595e961fc4120a597ad086aa1a82d : SUCCESS in 16m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/36df901be1ff4c7189746ac230815fc9 : SUCCESS in 20m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/838ef4a4d48d43a3942b408599b11a78 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2dae02f0a2714408915a48647fb20919 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b48c20e78b17401c8a7e2f1c06d66caf : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c128f065c7404dc5a072aff49ba51c16 : TIMED_OUT in 2h 00m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af2135eb241b49bbaca848e9c992f1f2 : SUCCESS in 22m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc38c8a48970406e8815eb0b75ac8ad8 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/99bcfb254828429c98bca5ddd05f5b8a : SUCCESS in 20m 11s","accounts_in_message":[],"_revision_number":13},{"id":"fb129edb6d325b413759c94b529af17812847be0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-28 10:27:28.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"f2dc15b611e98a74237d4a3c2d080a5898701e6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 11:51:30.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faeaa8fee9364541b81a42bc22be6ef8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/865c6796a4e64658a365f89eee68f791 : SUCCESS in 1h 22m 04s","accounts_in_message":[],"_revision_number":14},{"id":"5968fb64dc9e16d523d81807beeb6eee957cf95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:37:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c286daab982a4d059d2b4d98ef45ec72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e952e113d0684cc6bb9cb7e12111eddb : SUCCESS in 7m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4f80d1c51154a24b592a95097801855 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0c0899d38634eca82289e8ee97eee94 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/490cefdc82c34fb8b5a062246ffb3bf6 : SUCCESS in 24m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef1b005448324d3f9c59c5909954733c : POST_FAILURE in 33m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4bd9285ca6ef4fd5bbddb11a7c693e65 : SUCCESS in 15m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/523ec612d03140589527b5eb1822ca7f : SUCCESS in 25m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8fd1ab3548fe4a7bb46a4fb8cd8e4428 : SUCCESS in 25m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/14ae2730e35849098cde0eeb57063a3b : SUCCESS in 17m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/00771fe0e966464c81435d5b58a02b19 : SUCCESS in 13m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dfaee4b4deb046598fbcbf090794fbed : SUCCESS in 23m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/00405bdf8b14421ba3f4e45f4b236a08 : SUCCESS in 10m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b9650c09ae6498dbb2432ba66929097 : SUCCESS in 10m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1ec6cc73d9d94dfdac09531ad6e8672b : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/31a90b4871b8483e9e68b2b9115b6154 : SUCCESS in 8m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2667d819eadd411bb41ece381074afba : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/43960c5a7d8846b593b6f66d50bce584 : SUCCESS in 5m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab476a8853014c50855b78a9e1bfe688 : SUCCESS in 14m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d8efda8f1ab44e2aa40ecb73a2a2028a : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec20cdb233264e4f8481fe7e0a0a6edc : TIMED_OUT in 2h 00m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fce09a76a794542bcb9a5b3d9295272 : SUCCESS in 20m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86602c541e63473fa5e710506a3cf15d : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe9f4a1b274940a0a888b252609da118 : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":14},{"id":"53b5ec89e6d37e52462ff3c71193ad2852705f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-28 15:14:01.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"b24f9a756552aa2e6865d8c82103107a5c065a89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:51:23.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29fb7e670c074104b3a20e1aecaf3adb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4331a099c9b247739471f249b4860b48 : SUCCESS in 1h 35m 40s","accounts_in_message":[],"_revision_number":15},{"id":"d297569f424e676c72e246db85f30150b8ad0149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 17:17:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a07d89d251bd47da8c39865488e1df4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e269bfc7a1f484ea45255d5baf87a95 : SUCCESS in 8m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b477499e487f4e6587c2c376b99601b5 : SUCCESS in 12m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/632475f7b28744a3b030b82132ec58dc : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b5d79bb60494b24a125fb7d57162008 : SUCCESS in 12m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c135f3e34e434a2f8ec665acaf35b54d : POST_FAILURE in 35m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fad5634ffa8f4822a9d341763a797762 : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3c1eca3c1f47869a08a42f0aae04f2 : SUCCESS in 14m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/24a509acff3e4817910588c21b9df1bc : SUCCESS in 20m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8d785115617a419fa8616d027b6ec05d : SUCCESS in 22m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/28f1c6aaba9c425c86e41fc66d757955 : SUCCESS in 29m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/737244fdcb95422e99790679b09e358a : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b4c36f7d9a54e1dab5080e6bbfbd387 : SUCCESS in 15m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/285b2e08ab0a40e4ae4bbafccfcbce46 : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9fc0afc52f424fdb813c36fdd6518dc0 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5d188f6c3550453781beaddd7162b146 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1dbd9b9271b64bb2997746d1698f80f5 : SUCCESS in 8m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8df8f112ca32467b90b980687f7208f2 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c8f5ca953bbf47498b1b8a2f4b796c41 : SUCCESS in 13m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ebd6676b82e49008afe5ba64235260f : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f128d8ee6acb4446a081ae07fe875eb5 : TIMED_OUT in 2h 01m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ef08ac97e5d4b44b2ed5ff9dcb421de : SUCCESS in 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a8a41f22e86416fa5287592cf536c2d : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/377a799802ce4e22ba1804ae4414c978 : SUCCESS in 23m 11s","accounts_in_message":[],"_revision_number":15},{"id":"ace0ab17e9e75a1680a42ece022903563661493f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-04 12:34:11.000000000","message":"Uploaded patch set 16: New patch set was added with same tree, parent tree, and commit message as Patch Set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"b74e6e7c5ef15e3f9d57565b6ad20e7a4278fc66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 14:44:10.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa0b70899f534c2a8b19191845e8e2d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5c0fe39e03433196f7ed543e94d39b : SUCCESS in 8m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c8824805bac42159470cba964f9c05c : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b490a785d24b40d5b691641b1f8ffd10 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/026ebbb58a7045bb869d1ff5f55ce7fe : SUCCESS in 19m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cbe2576c77d1434bb476473c3c5549ec : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f05b6ff4a7004812aa810eaa575b6a86 : SUCCESS in 9m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43e59be30ff54874a6aecced20a386ef : SUCCESS in 19m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7e096269b6134886bd486eb972998de5 : SUCCESS in 20m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2577a9b5e6584424baebebfc235ffa21 : SUCCESS in 26m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/715b7bd30ac44d2ea88cc06b0653707e : SUCCESS in 35m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/46549033da8244449fe56fa180860192 : SUCCESS in 9m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/27b8f61abbe34e2691d7a8ff939733fe : SUCCESS in 20m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4cb3efbb6235435b920592e04a0f0088 : SUCCESS in 19m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e927beba2f574b0c8ac24d34f597a2a0 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/08d781474b6041699c3a206c13fc3362 : SUCCESS in 12m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/619159deb4dc45af8f1bd23d42c4a901 : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f77d426e93554272aba94ad1e78f8426 : SUCCESS in 5m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/01fd2ee58d144da2bdf2860511119e7a : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e60afbb3eaf143708f30cf202be0b5da : SUCCESS in 6m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad273f3ad79544ebb5d1a61cf6424cbb : TIMED_OUT in 2h 00m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2d56aed236748a2bbb08911a83bb08b : SUCCESS in 21m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0f7ef749dd43f7b7a6daa046687fac : SUCCESS in 3m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49ad39fdf38a465daddcc5124189ec4d : SUCCESS in 24m 38s","accounts_in_message":[],"_revision_number":16},{"id":"2f4c8547f3209f138ee0cbc3f120f2cd874d11ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 15:34:15.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c275b633eba4974871ad52aaafbe2b1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c8b378191fc4d5e95260048f1bba4dc : SUCCESS in 1h 39m 58s","accounts_in_message":[],"_revision_number":16},{"id":"4bf0c8d425147981e6603f8e41588382a4344d02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-06 16:40:22.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"f19c10bfc196be94681e082cd527c4bea1750218","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 18:10:48.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba1a69b69fe6438293570fe6a1fbace3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f07eda995494bfaacb9780b84adb3c5 : SUCCESS in 1h 27m 41s","accounts_in_message":[],"_revision_number":17},{"id":"3cefe2fedfb3e0a49ff34c05e83519b6f29a54ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 18:48:39.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9d43f2590e54d11850b3061fdce0618\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3a357fca79c4223bde679fceaf32204 : SUCCESS in 5m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4327118af680484e8dd540e8f4221638 : SUCCESS in 15m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83e13a5924914aa1b5ccd25b5f711bbe : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65e346ddcf864045b8844d7ebe02d36f : SUCCESS in 17m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55013e1c8842475e82f249d36931f6ff : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c55bdb709a647109ce48feaa99fd4e6 : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c67bb119a0b74550b1f3c0503f18749d : SUCCESS in 16m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fc69d9003f9146fbbec2b62447f97370 : SUCCESS in 17m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2ffc30918a40496ba21a39ee637f1a12 : SUCCESS in 27m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e3451b6f19c46f3bbbb636faf882446 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3d91543a10d94f7f9772fa958affcc3f : SUCCESS in 16m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4d154ff251f44f898de3bf5b0c3c005b : SUCCESS in 18m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/305086cad53d434cae46791c296b5d27 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0730647563f345d2a0c8e1f7cc69cc4d : SUCCESS in 14m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c3e8d6fbe9bd4277b274d9378ec7e4f6 : SUCCESS in 17m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57c1fb6732464d69b01d1da5b22ee9af : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51583f7c38714491a8f05ae2df6f9498 : SUCCESS in 14m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/94c91e410f1f4df1975f68aedca57542 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bf19aa4244241228d48485588e76272 : TIMED_OUT in 2h 01m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36e0f69cd43a41018e03b53ae646901f : SUCCESS in 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c1acc00df341aaa7d1544abf8f59d7 : SUCCESS in 3m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05e1a0bc02ef43f7ad124936fa519020 : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":17},{"id":"f5d126b19a3e06e3dadff5a874bf5faefad1d89f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-12 11:17:49.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"d56817da864074d52849a814bea0304a0f0ee684","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:43:50.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/858101d264ea4a4884e65807caca839a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d2db51e4e744e07a956ecb9f7c23712 : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":18},{"id":"8d5dea48f0f8881c107d11a4f14acbe6bd4f65a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 13:24:34.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af759f8ff16a47dd8ac8ba6c3f45e1d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe1e34b4b8d478c88cd071f196aff88 : SUCCESS in 8m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13fea0c63d304580819b133b635d17f4 : SUCCESS in 13m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8e32684cbe8480c9b77abc2d13f31d1 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6125099b3f94b32aecc0689cd0e6bed : SUCCESS in 16m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d6463d6e07234b298b57e4cd8e3687bf : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69916c1745b24ecb8ec9e0b57b77e65d : SUCCESS in 22m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/55b75cdcdc144e5b9c5091b665097b57 : SUCCESS in 15m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ccbd442572d24d448562173d1f8715d9 : SUCCESS in 18m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c3ab8665a3fd4e80825d35998c444139 : SUCCESS in 14m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/37e1fb4d2bfa4880a6cdbd1985b147aa : SUCCESS in 20m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f1045126ead4e969f4856ff2bccbd92 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73c9f9d454804978995bc620e29d7674 : SUCCESS in 21m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/263ab607ee174ef48bf76feabc008ce1 : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4b63bd6ab7a84de8aeaf355a62ee1c95 : SUCCESS in 13m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b3942e576ad94721bed2d6780d8ee650 : SUCCESS in 17m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fa45d5c818d145b293d276c71dd81b57 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2739236522ba4c2d901a5084d7ee1e5e : SUCCESS in 13m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7b7c0317407f477c97ddeb3fd3db01e8 : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/54eecba37e614c94ae9ad0c163a82c12 : TIMED_OUT in 2h 01m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4f3f81be7c84a90bb6e07069c40a74c : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd4f3f4cdb704efe8c3b53b395747d64 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee90134209604a79b4ed4a26d45a9d62 : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":18},{"id":"8a4ff0f58f3fa49a4c5b0eb9a7e0e8c255392268","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-12 19:12:29.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"3e9b143be9f14c7d2150b1f92deeb8277d541319","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-12 20:09:19.000000000","message":"Patch Set 18:\n\n(7 comments)","accounts_in_message":[],"_revision_number":18},{"id":"a7fde161b62d5e1dba100a99291e815925fe1fba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 20:16:56.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caede3c19c9b4a43949a5c514a810c3c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb8e786df07848398259fbfc006fceaf : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":18},{"id":"eab89de43c9148e2eeaf6c9cdea451f97f26bffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 21:20:26.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaeb9f9713994810ad1b49eb99063fb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd5ca79720c4a86b881b04f6eaa2dff : SUCCESS in 5m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70ebca58240f4626adb21be6a868cec1 : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/202341e3322e4bc091bc7ef060d10589 : SUCCESS in 13m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ae9cb8cbd6449c7b9847dcd8bce883b : SUCCESS in 20m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bfab0ef971224c08b3a78de80d1e8b71 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2925a71f25e94a3192ceddc3fe552201 : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a78e8c9b8434ce3a3213f6783b639ed : SUCCESS in 11m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d9e973993cb24129a7a3b14022285c1f : SUCCESS in 12m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/94abab7f9985472e8fb6726b236d9ab9 : SUCCESS in 14m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/021e0a4f691d4e379154da7ee7b4e268 : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d6802f564ada4a63806f7633d18f2410 : SUCCESS in 14m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d3012644d71c4562b60c1e83699d5184 : SUCCESS in 19m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e3d30904a97944c3a20441014051de64 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7c5154a669984124abfd302c79568de6 : SUCCESS in 13m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ab8355f26a6140c0b678064203d321a7 : SUCCESS in 13m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/65113f0a768f4b788072986fd82f980d : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f605e70174a34a9281aa873ead7a2bf4 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/108ec9e18ad344acb813fc4133a0106c : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1358e40883c64c07b05ecdfd9ff8dcba : TIMED_OUT in 2h 00m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f69b3363b273466887b2884eb2e21beb : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a657656eda97499b89285f59b4591809 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/325abfff81514bdea5bb6e342d3084ce : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":18},{"id":"22076ea354baf32c543ba5c2114ecca5629b5ef2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-13 09:59:54.000000000","message":"Patch Set 18:\n\n(7 comments)","accounts_in_message":[],"_revision_number":18},{"id":"77ffd668f53ccfcb8190b5e5bfea819fb4938ba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-13 10:00:39.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"adf50aeaaf155b8be17dbe794607f426e175eb07","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-13 10:01:41.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"3c47c45fd58889ba1aac48f3b1cca6c74618f090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 10:55:18.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc2212d7f5754ce7bbf897a5a3cca14d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab826aff35f4445bd4fe731bff38b7f : SUCCESS in 9m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45f9fa56e37c460094b37d7afc06a047 : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f1d74e68c9342ec94b8d8234d25a494 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc9f75b7240441e3b2ae086141da3619 : SUCCESS in 18m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f054a91614b745c4a880aef88da18d3b : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5a3850e1d0444181a83823a54c425065 : SUCCESS in 18m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0710ebdf60a43bf9517c05649849053 : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/126c6477d92b4c50a3669c9907680ec8 : SUCCESS in 20m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e387b92c7a4d4653b589224cb0b50835 : SUCCESS in 26m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/57675db64c7c45b08c8bc24989dfb60c : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/24db002ddca74fcd949f37872dafecd3 : SUCCESS in 20m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/637e465e4f274894938b70c6c5be0e1b : SUCCESS in 16m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/11a0071b6b8b4877bc09d711a64d4488 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8cfb8cbee184434caed2b5e18bb03a18 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d12de19abc014d42ad80d2abfa6c0d03 : SUCCESS in 14m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0fb621ce8d5d447da7b88c493437a673 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc17a8bdf6b24b1ca66021ed9e6aa720 : SUCCESS in 13m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/058ff15f9a4844b7bf8401b86d1cc3ac : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbaf155108134b09a6b65c6a9bb108b7 : SUCCESS in 46m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c80f87a3508d4001858cdf09d0df2c66 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a49c0324534b229217833be36d451b : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a88ec9186f08442d8c1f20686d96d3a8 : SUCCESS in 18m 23s","accounts_in_message":[],"_revision_number":19},{"id":"5bf35749dc1244b2a9392271832f4fdb30b7bfdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 11:45:18.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f21fd708f7d4454eb39bc4fd0c1be273\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d5824a35f8704506bad32faabeadd802 : SUCCESS in 1h 20m 54s","accounts_in_message":[],"_revision_number":19},{"id":"969394c1722b220cf6709c302b6060f4059443dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-19 14:38:27.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"b84eac87af822581a47d6a408329c0b87b60dbf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 15:41:40.000000000","message":"Patch Set 20: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb22e845d54b4464abf32fea2bb416f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9233bde5d2d4aea906f22b3afecc84b : SUCCESS in 8m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58b88d0c73744469aece31ef99cc8503 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8bec1efa82b4436ba712f40589deead : SUCCESS in 26m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4719c475f4594108b9ebc4439ad88c67 : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c172990e8f24a7b82d0432a7b9982cc : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14f49c95b6ef44a69501850857229d28 : SUCCESS in 21m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b13629eebf14070a005531886cb55cf : SUCCESS in 22m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/248a24863e4b4281895d72f33487bf18 : SUCCESS in 20m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5721af4b0948475ca1737e402e4c6fa7 : SUCCESS in 34m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cf493c19d5a54cdaa7bbc554e51a8296 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6dab887b0cdd4c29a0439f44b7f952f3 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6e5a4d740d8a4b7eaf5cfc2a7404ed32 : SUCCESS in 18m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8d3ac3f365534b13b22d57f64c73906e : SUCCESS in 19m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/94a93fc2973545c79c789fc508cb6100 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a8a4fdeead114ad3b724438d7a849a8e : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24d945d5803647b48543de0287e77bdd : FAILURE in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d8080e69b224b0aab291d736e3252c3 : SUCCESS in 14m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/109e64bfc8f2489485d006dda716603c : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9eb81d04f9f14620b563c63e2bcd0a60 : FAILURE in 52m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81dc308dd151427a94a0fdde7d57a81a : SUCCESS in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b390565c4954135a7d9b49a451038c8 : FAILURE in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fdac2ec55f0b4893a7da5416f1a92ecb : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":20},{"id":"9a25f941d6869474cece5e7aba58e662280a0395","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:39:27.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/601385ea064a4dcf93dd1eb2b163466b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/505115798666493995974513513fee4b : SUCCESS in 1h 25m 39s","accounts_in_message":[],"_revision_number":20},{"id":"fcdef4a0777ef3a443e689cc945be7beb607fc57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 15:48:47.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"5cedb8693d260fc795a5824623039b8efbb06d0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 16:55:47.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f59c4c61f0874c54883db707e583efee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e29e367d80347fb9bb94af0a04494db : SUCCESS in 9m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70008cf37a92413790c64c162f259d1f : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9db0d09644aa463d88b08fbea6dffb68 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3622dbdb8b1a4d84aacd99043a6a51c8 : SUCCESS in 19m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6743285c6f604c24888ff43f9eb2f559 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/30b521bf444948c296d37ae34299d989 : SUCCESS in 21m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb72cc5c41cd417880d153b41420d6aa : SUCCESS in 22m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d5302864bd2d4f60948591ce9d75673e : SUCCESS in 23m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1d638eb8048e42f1be09e131ddde5b59 : SUCCESS in 22m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/438b0d063d474c3cab6fd6a1ed6ab64c : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/60fd4259cf8b42b399b117db90870dba : FAILURE in 16m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/95b1373915d1439baf7e5a27f47dc374 : SUCCESS in 15m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3df5353b1d4f40cb8a11892b9829b6ca : SUCCESS in 13m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9e90ed38aeb54fb4a02cf9ab8b43c5bc : SUCCESS in 14m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3e95bf362e1b41aaab221d28fd74601a : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b87b0b32a79499e9858ead63a15cdf1 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/098096251a354727914ebfabd722a333 : FAILURE in 16m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/11df7793d3b6423795644188f491f6b9 : FAILURE in 8m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/91707b1c829f4815b535ef72ff283264 : SUCCESS in 58m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2c21e25db00480e93d2a8cfb607245e : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc8e1421d5249e6b3e7cb3738daabd1 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/712c299067b64343b6b99cd1ea83313d : SUCCESS in 18m 45s","accounts_in_message":[],"_revision_number":21},{"id":"07c9c9d8b02f7104440a8a9a10dbe0516875ad0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 03:47:29.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd93eb578fc949e5bfb8f1510db930c6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f0ef4b671b549ebb0e722eb5caabbb1 : SUCCESS in 2h 15m 44s","accounts_in_message":[],"_revision_number":21},{"id":"e4f38a2ae6f749f89f4c9e670d9b57e687434308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-28 18:34:34.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"59c7e50317f89a37df71f8b751a0856b79a46661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 19:34:06.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa8a4b5c59d48bbabdbac71eb011a35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fea4853f1784bd4ad9cccb02ba6b1dc : SUCCESS in 5m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9565fe746a8f4c5bb79f8336e5a778fa : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ce00e2e057844f282a189e01ea04bbc : SUCCESS in 22m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f504331147ce4c7982fb6bbb60e8a58a : SUCCESS in 21m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6dc7e29286b14e76a31eb3c4f8880d62 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f54404b96c64e3399f19c55ad2adde2 : SUCCESS in 15m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1a60e4fc48342d5a002dd2f643ed19a : SUCCESS in 15m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dbf182883742404e859b3f59a6e838b4 : SUCCESS in 23m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/39f3f67d2f1a4c5d811eec0960b206ad : SUCCESS in 30m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f1b57706f3124e0a846c2f5a210787cb : SUCCESS in 24m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05ecf1e518ce42298c8e25b31c35ac24 : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/93e02a0f35eb41db812ad25f0e9abe6d : SUCCESS in 16m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cfcf4b26288346688ad9d64e4be08d5b : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f8a4e6834cf446f7a0e16bf4c882f71c : SUCCESS in 14m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cd892030916242f0aacbd3a5c76d63cd : SUCCESS in 9m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c4cd1b24aa0947b2a9b5aa03171c9822 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d61e849d0234451e9be9c2067d07131a : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dab46413224245fea19eb7e97fe1430d : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1b91a07974d410e9908b7c8d6a9ce4c : SUCCESS in 54m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81832206096345578a73a31007ad16ff : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57a024a91384b89b73e3f2930262fc4 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf294f4c42cd4a9f9c906a4fe8508023 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":22},{"id":"1fa5b2c4e7aa12c23b27e42c0b62473aacefeb74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 22:53:24.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7245ac8e92d54401a276cd7efaf8c577\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62fe3e3f91d74955af52b18a094c6cd4 : SUCCESS in 2h 05m 46s","accounts_in_message":[],"_revision_number":22},{"id":"00e7140c4c23f164901c5ed570d3064d6d82c77d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 09:58:20.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":22},{"id":"86c6cd111468dee0b45bd082dd533e9910f2e721","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 09:59:34.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"42325a737bdbc3c89863bb8bd2950fc4cf0d8361","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:47:08.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f40e990ee5064ec18a96405a92cfe651\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6303f3b26b8247f382b0ff7828d136df : SUCCESS in 7m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c71d88f4043740fdab07bdaafb646acc : POST_FAILURE in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c22ba2c64c2a46cb85a5bb5d8862fe1e : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4707bbcb72844a04b0a5cee0a6db3f56 : SUCCESS in 19m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c7afdc97dadc4320ab3a6aa9caef8c45 : SUCCESS in 24m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d824710914184f388894b83a17c55c80 : SUCCESS in 30m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aef22b1eea3c4d9da4c8ba405c5f1f17 : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d549ba8855644ea83d76541d4795ac4 : SUCCESS in 20m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/217df90b38d84528b80afef626d17505 : SUCCESS in 17m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/11d5601129344f3aad76a744c6db5416 : SUCCESS in 12m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/13548e7c549a4bd6b1c7d7fbfe2423b4 : SUCCESS in 12m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/27f7dc7cba594d6a979c19601baf601d : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/892f4156612d42b59d7c8d6ee9a81401 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d5fcd5cff23448b1a4161995d9daeff0 : SUCCESS in 9m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd052240c6c84c1286a83daca80b25cd : SUCCESS in 51m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/407b08aaf5a24aa796b1202e9802c40d : SUCCESS in 21m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b566c400fc804b128d5fe9f26f3ad38e : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6acc801d841746e599db04e64ee2cff7 : SUCCESS in 20m 52s","accounts_in_message":[],"_revision_number":22},{"id":"abd0ffa9b2b8e9cfa3cb0f3d097138744affa154","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:47:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"d1e6445e7b1131c401c23ed8605749e3d14e739e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:48:39.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e38452544a4a4b25b0355d2bd3bbbb51\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5e87907aa054b9bab1a2178e226a4d8 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6dea9920d0a241bab00eb681eb860d53 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"9636ef0694f2b2966725063e5c79d10169fd5d9f","revisions":{"8fb0bfc57648615c517914e2dfb1354a82468737":{"kind":"REWORK","_number":1,"created":"2024-10-04 18:10:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/1"}}},"commit":{"parents":[{"commit":"156e69090e7f8207db43bc56cbb93b9bc749d85a","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/156e69090e7f8207db43bc56cbb93b9bc749d85a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 18:10:05.000000000","tz":60},"subject":"WIP audit sessions","message":"WIP audit sessions\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fb0bfc57648615c517914e2dfb1354a82468737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fb0bfc57648615c517914e2dfb1354a82468737"}]},"branch":"refs/heads/feature/mpu"},"b6ca9c62c0eddb144e6d6474e7f67eeaa52e93ce":{"kind":"REWORK","_number":2,"created":"2024-10-09 18:31:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/2"}}},"commit":{"parents":[{"commit":"f6abec4163a7c2d9e9748e1060e483d1ab73ec7f","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6abec4163a7c2d9e9748e1060e483d1ab73ec7f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-09 17:23:29.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6ca9c62c0eddb144e6d6474e7f67eeaa52e93ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6ca9c62c0eddb144e6d6474e7f67eeaa52e93ce"}]},"branch":"refs/heads/feature/mpu"},"a30b6b5d5131101292941a1db4937f7e4b5d9c9a":{"kind":"REWORK","_number":3,"created":"2024-10-10 18:24:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/3"}}},"commit":{"parents":[{"commit":"f6abec4163a7c2d9e9748e1060e483d1ab73ec7f","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6abec4163a7c2d9e9748e1060e483d1ab73ec7f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-10 18:19:47.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a30b6b5d5131101292941a1db4937f7e4b5d9c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a30b6b5d5131101292941a1db4937f7e4b5d9c9a"}]},"branch":"refs/heads/feature/mpu"},"77beea36ebd52e21e3c287207b68d126faf5e136":{"kind":"REWORK","_number":4,"created":"2024-10-11 15:32:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/4"}}},"commit":{"parents":[{"commit":"4c4465d6e6c2801a69e381e5cdb8041e5042dc5c","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c4465d6e6c2801a69e381e5cdb8041e5042dc5c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-11 14:42:51.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77beea36ebd52e21e3c287207b68d126faf5e136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77beea36ebd52e21e3c287207b68d126faf5e136"}]},"branch":"refs/heads/feature/mpu"},"10087714e770f06ca71ac8936b1ac9964f3fd996":{"kind":"REWORK","_number":5,"created":"2024-10-14 16:42:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/5"}}},"commit":{"parents":[{"commit":"4c4465d6e6c2801a69e381e5cdb8041e5042dc5c","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c4465d6e6c2801a69e381e5cdb8041e5042dc5c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:57:25.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10087714e770f06ca71ac8936b1ac9964f3fd996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10087714e770f06ca71ac8936b1ac9964f3fd996"}]},"branch":"refs/heads/feature/mpu"},"40e7587cb2c9e8edf55d752c74b26a987f347ccc":{"kind":"REWORK","_number":6,"created":"2024-10-15 13:09:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/6"}}},"commit":{"parents":[{"commit":"55545e638fcf974f560df6a4fea93a7e7dea9459","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55545e638fcf974f560df6a4fea93a7e7dea9459"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-15 13:07:45.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40e7587cb2c9e8edf55d752c74b26a987f347ccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40e7587cb2c9e8edf55d752c74b26a987f347ccc"}]},"branch":"refs/heads/feature/mpu"},"ab634c62768f237e2b5f71ddb63ee1c7384cef6a":{"kind":"NO_CHANGE","_number":7,"created":"2024-10-15 15:25:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/7"}}},"commit":{"parents":[{"commit":"35539f52caf6816ead510a9d6605252c776238d3","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35539f52caf6816ead510a9d6605252c776238d3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-15 15:25:28.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab634c62768f237e2b5f71ddb63ee1c7384cef6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab634c62768f237e2b5f71ddb63ee1c7384cef6a"}]},"branch":"refs/heads/feature/mpu"},"8f4f7e251a32f8c312f868debe7af1c812ba24bf":{"kind":"REWORK","_number":8,"created":"2024-10-16 15:34:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/8"}}},"commit":{"parents":[{"commit":"aeee0e3540542b56f191644eb6284217e7522de5","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aeee0e3540542b56f191644eb6284217e7522de5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-16 15:33:53.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f4f7e251a32f8c312f868debe7af1c812ba24bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f4f7e251a32f8c312f868debe7af1c812ba24bf"}]},"branch":"refs/heads/feature/mpu"},"e544e383ce225af54b9b4783b078940566fc032a":{"kind":"REWORK","_number":9,"created":"2024-10-18 15:34:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/9"}}},"commit":{"parents":[{"commit":"564b57835f81a5726a5ecda92abad1d3a14917ff","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/564b57835f81a5726a5ecda92abad1d3a14917ff"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-18 15:32:02.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e544e383ce225af54b9b4783b078940566fc032a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e544e383ce225af54b9b4783b078940566fc032a"}]},"branch":"refs/heads/feature/mpu"},"c98dfd6014cf666385d5cf977ed116d81be9fed5":{"kind":"REWORK","_number":10,"created":"2024-10-21 12:14:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/10"}}},"commit":{"parents":[{"commit":"564b57835f81a5726a5ecda92abad1d3a14917ff","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/564b57835f81a5726a5ecda92abad1d3a14917ff"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-21 12:13:46.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nAdd MpuAuditorConfig class and encapsulate config in an object that\ncan be passed between auditor classes.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c98dfd6014cf666385d5cf977ed116d81be9fed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c98dfd6014cf666385d5cf977ed116d81be9fed5"}]},"branch":"refs/heads/feature/mpu"},"f5075fb04caa39826a442282829a5887c5ef5a9b":{"kind":"REWORK","_number":11,"created":"2024-10-23 16:15:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/11"}}},"commit":{"parents":[{"commit":"5591aaa7f034ec48830c484b3e85004a0a9c2709","subject":"mpu: probe tests: capture_logs option for run_custom_daemon","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5591aaa7f034ec48830c484b3e85004a0a9c2709"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-04 08:22:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-23 16:14:06.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nAdd MpuAuditorConfig class and encapsulate config in an object that\ncan be passed between auditor classes.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5075fb04caa39826a442282829a5887c5ef5a9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5075fb04caa39826a442282829a5887c5ef5a9b"}]},"branch":"refs/heads/feature/mpu"},"78e1d859aa5b57f1d3ab2f69cb288c36bf9d6c7e":{"kind":"REWORK","_number":12,"created":"2024-10-25 12:01:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/12"}}},"commit":{"parents":[{"commit":"8d997a3082bbd6013cd41b2f4b2e162f37d36825","subject":"mpu: probe tests: add a helper to check mpu resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d997a3082bbd6013cd41b2f4b2e162f37d36825"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 12:00:47.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nAdd MpuAuditorConfig class and encapsulate config in an object that\ncan be passed between auditor classes.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78e1d859aa5b57f1d3ab2f69cb288c36bf9d6c7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78e1d859aa5b57f1d3ab2f69cb288c36bf9d6c7e"}]},"branch":"refs/heads/feature/mpu"},"e8b7f1f4abb7707153ada11d6f57d7ebfcb04059":{"kind":"REWORK","_number":13,"created":"2024-10-25 18:10:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/13"}}},"commit":{"parents":[{"commit":"8d997a3082bbd6013cd41b2f4b2e162f37d36825","subject":"mpu: probe tests: add a helper to check mpu resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d997a3082bbd6013cd41b2f4b2e162f37d36825"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 18:09:49.000000000","tz":60},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nAdd MpuAuditorConfig class and encapsulate config in an object that\ncan be passed between auditor classes.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8b7f1f4abb7707153ada11d6f57d7ebfcb04059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8b7f1f4abb7707153ada11d6f57d7ebfcb04059"}]},"branch":"refs/heads/feature/mpu"},"5d81fda65a95a1f5602aa7c3bc98f5e7ff12c886":{"kind":"REWORK","_number":14,"created":"2024-10-28 10:27:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/14"}}},"commit":{"parents":[{"commit":"ff5b399b4a179f3ba6f3c8fd85cf0286272d2eae","subject":"mpu mw: add POST request handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff5b399b4a179f3ba6f3c8fd85cf0286272d2eae"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-28 10:26:42.000000000","tz":0},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nAdd MpuAuditorConfig class and encapsulate config in an object that\ncan be passed between auditor classes.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d81fda65a95a1f5602aa7c3bc98f5e7ff12c886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d81fda65a95a1f5602aa7c3bc98f5e7ff12c886"}]},"branch":"refs/heads/feature/mpu"},"5a80817be384255c1f3dcb92439e49ecd493d027":{"kind":"REWORK","_number":15,"created":"2024-10-28 15:14:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/15"}}},"commit":{"parents":[{"commit":"69b79002d6f754313d4c75c0e50ea3c6d8dd029a","subject":"mpu-adutior: Add config class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69b79002d6f754313d4c75c0e50ea3c6d8dd029a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-28 15:11:00.000000000","tz":0},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a80817be384255c1f3dcb92439e49ecd493d027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a80817be384255c1f3dcb92439e49ecd493d027"}]},"branch":"refs/heads/feature/mpu"},"f5cf663762fa465606420fca1ddbf4f8d835fbf8":{"kind":"NO_CHANGE","_number":16,"created":"2024-11-04 12:34:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/16"}}},"commit":{"parents":[{"commit":"55a40a816108ef35ea2115d9fc5b1799a600254b","subject":"mpu-auditor: Add config class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55a40a816108ef35ea2115d9fc5b1799a600254b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-04 12:31:41.000000000","tz":0},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5cf663762fa465606420fca1ddbf4f8d835fbf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5cf663762fa465606420fca1ddbf4f8d835fbf8"}]},"branch":"refs/heads/feature/mpu"},"52e6ad181883cbde142e6d9bcb034d28ab250e07":{"kind":"REWORK","_number":17,"created":"2024-11-06 16:40:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/17"}}},"commit":{"parents":[{"commit":"50dfff7db1251aff0b549449b4e41633bacff4b7","subject":"mpu: fix param passing to sub requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50dfff7db1251aff0b549449b4e41633bacff4b7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-06 16:40:06.000000000","tz":0},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52e6ad181883cbde142e6d9bcb034d28ab250e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52e6ad181883cbde142e6d9bcb034d28ab250e07"}]},"branch":"refs/heads/feature/mpu"},"1094057cc396d4f007ce8f3ecc4766d35626cff8":{"kind":"REWORK","_number":18,"created":"2024-11-12 11:17:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/18"}}},"commit":{"parents":[{"commit":"67859418b14a26d54d7178e0ad78e857a4c145a6","subject":"mpu: refactor MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67859418b14a26d54d7178e0ad78e857a4c145a6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-12 11:16:26.000000000","tz":0},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1094057cc396d4f007ce8f3ecc4766d35626cff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1094057cc396d4f007ce8f3ecc4766d35626cff8"}]},"branch":"refs/heads/feature/mpu"},"f38af05239f8435affd89202758afed50ff3563e":{"kind":"REWORK","_number":19,"created":"2024-11-13 10:00:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/19"}}},"commit":{"parents":[{"commit":"67859418b14a26d54d7178e0ad78e857a4c145a6","subject":"mpu: refactor MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67859418b14a26d54d7178e0ad78e857a4c145a6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-13 10:00:25.000000000","tz":0},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f38af05239f8435affd89202758afed50ff3563e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f38af05239f8435affd89202758afed50ff3563e"}]},"branch":"refs/heads/feature/mpu"},"a92222f7c98a79eaa9185a68b7f92cde86b55482":{"kind":"REWORK","_number":20,"created":"2024-11-19 14:38:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/20"}}},"commit":{"parents":[{"commit":"a62892c4db3e337e60849e9a31d6e84bb73c8936","subject":"Merge \"merge master to feature/mpu\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a62892c4db3e337e60849e9a31d6e84bb73c8936"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 14:38:09.000000000","tz":0},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a92222f7c98a79eaa9185a68b7f92cde86b55482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a92222f7c98a79eaa9185a68b7f92cde86b55482"}]},"branch":"refs/heads/feature/mpu"},"c29f5e77484b906cdbfc3d317e5265acb80cf9f7":{"kind":"REWORK","_number":21,"created":"2024-11-27 15:48:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/21"}}},"commit":{"parents":[{"commit":"83df840ec3155bc5b49e474f62d260e3dfcd3c4a","subject":"mpu: repeated completeUploads get unique timestamp offsets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83df840ec3155bc5b49e474f62d260e3dfcd3c4a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-27 15:48:19.000000000","tz":0},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c29f5e77484b906cdbfc3d317e5265acb80cf9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c29f5e77484b906cdbfc3d317e5265acb80cf9f7"}]},"branch":"refs/heads/feature/mpu"},"9636ef0694f2b2966725063e5c79d10169fd5d9f":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2024-11-28 18:34:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/931517/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/931517/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/931517/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/931517/22"}}},"commit":{"parents":[{"commit":"7e20d95f1e04f416353c3a889a60a181818aff5f","subject":"Merge \"mpu: fix last-modified time in UploadPartCopy response\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e20d95f1e04f416353c3a889a60a181818aff5f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 11:59:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-28 18:23:19.000000000","tz":0},"subject":"mpu-auditor: audit sessions","message":"mpu-auditor: audit sessions\n\nAdd mpu session auditing to mpu-auditor.\n\nSimplify session state management:\n- that session state is represented by content-type.\n- sessions can be in progress, completing, completed or aborted.\n- concurrent completeUpload and abortUpload can leave a session\n  in a state that is inconsistent with the mpu e.g. a linked mpu\n  may be in state aborted.\n- the mpu-auditor is responsible for detecting state mismatch\n  async after a delay period.\n\nRows that need to be revisited have their meta-timestamp bumped.\n\nChange-Id: I376df764041943c982d71c4e20f4af37e69de498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9636ef0694f2b2966725063e5c79d10169fd5d9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9636ef0694f2b2966725063e5c79d10169fd5d9f"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
