)]}'
{"id":"openstack%2Fswift~931851","triplet_id":"openstack%2Fswift~master~Icb62ad8b9200b6c8d418b1678a2b6b01eeab021c","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-09 01:02:47.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Icb62ad8b9200b6c8d418b1678a2b6b01eeab021c","subject":"tests: Skip IPv6 tests if IPv6 address not available","status":"NEW","created":"2024-10-08 23:53:31.000000000","updated":"2024-10-17 19:21:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32e4c7f447afa0d2ffa2e4894e661a39e4b17840","_number":931851,"virtual_id_number":931851,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-17 18:44:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 01:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f318e2de95e244127958a51ebb94c0c76bc0b6a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-08 23:53:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54bc82c36af9a81f662642c278cea59060c9e290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 01:02:47.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7485a9967a0d4344a99653aedff1c5f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de430dd373e4fb0a088034bf2137583 : SUCCESS in 8m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/743f451c533545a39dccb76a4387b03e : SUCCESS in 1h 03m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7693ffa344e343d197202d4c68fb7519 : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d59f852d215e432b8d6a47cedbddf3ff : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82ef5bb4d63a4c32a4b1e776691927a4 : SUCCESS in 15m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/208690edeec6495eb49c6d225f37b84d : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0ff4519d4da4495b0f655b09cf38173 : FAILURE in 15m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ebe8cd50ef8743428d1719d454803449 : SUCCESS in 18m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4573c44b536a4b3a93eeea78174a8ece : SUCCESS in 15m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/93016056aa99432b90f6e783b1fc09bb : SUCCESS in 15m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ea86666d1ac04022ba8168bfd908d3b1 : SUCCESS in 31m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c51f4c26a91497ea5868e1681f4b1c8 : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4de9f824b14a401093f0da3a113e127d : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b797491fd9b64e37a960b0e3ce4b7eb5 : SUCCESS in 16m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/305d5ce730a0478e97dc745d604d66be : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d140a48a6587453f8ea145dbe464855a : SUCCESS in 12m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/68cb3c5823ad40efa96a1d2991020bbc : SUCCESS in 13m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/632978e10ad8488bb90864085e56f1f3 : SUCCESS in 52m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3045709e58fa4c6a893777efcc932746 : SUCCESS in 51m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f31b8fdaa23404fb19b346282138f82 : FAILURE in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9d121428e54a44adebb2931b02da32 : FAILURE in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00fe199fea5f48d48f9de871a09505d3 : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"33d24aebb05e3be5bfd18e4f61cae7443742d731","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 01:43:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82ac77cc92b04b7eb8943e0e842345ab\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f46698278e2d4a4e97c66efc82303611 : SUCCESS in 1h 48m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48b2644fd93f49548cf84a201f7db11c : SUCCESS in 1h 14m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d803e303a4d4bb4b3c6066955951397 : SUCCESS in 1h 29m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f2c5cede99243c98418688c67e03476 : SUCCESS in 1h 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"67b102732e1468c18aeb16b3a0a707ca4048c321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-16 20:18:27.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b2978674a6f82b222186dd1eccfcf62e953bddf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 21:11:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/952e5854439a4653bdf2c76a278c3e30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a8c57ebc5649e3a227a5e4a3fd9475 : SUCCESS in 8m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d2ca957aa124b408dee7c2c839360b0 : SUCCESS in 43m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/755b1ff32ddb4750942f8d2be996de03 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/861fc3544ab84f6091d48f2c3efffb86 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd28de669ffb4ef8ba94a442f8752211 : SUCCESS in 17m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b24a402e5de74e78b731f86f04ccecc8 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14c49262b1bb4165aa31f07a74ce7044 : FAILURE in 17m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d48cee031c6245459cbcebc22783d600 : SUCCESS in 16m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f91701a7134a4421bb466d2c2dfefbc2 : SUCCESS in 17m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/49ad2ca0c0ac4be1aafddad4496d42a5 : SUCCESS in 10m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1c1517299e2e4d63a0fb35892db796d9 : SUCCESS in 18m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0ae520f6f852493498c45cc10970cbf2 : SUCCESS in 10m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4bbb36885304cfb9bd01010ee728916 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/77644459dfd4406fbc1b4e6c0f4df442 : SUCCESS in 17m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d7bdfd78ac904d67b5616c64657b0a4e : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/edda19baa03c4130b5d608d944f942cc : SUCCESS in 8m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b08bac3a912347d195ec13aa1f159aa6 : SUCCESS in 15m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/03e98d1a7eb14b849f40c7216c49f4b2 : SUCCESS in 29m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ef2271f464046da833aff58aa7a52b2 : SUCCESS in 47m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3420af271ef6483a885eb71db6e16432 : FAILURE in 20m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edce62291c2d4a9b996e3077031c113b : SUCCESS in 3m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7103c0102c154d51a4066efca6d05ed3 : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":2},{"id":"295f74370838bc3f2959d29f3f267c0b8377d27c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 23:33:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aefb1afe048b4891902fe78140cb93db\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9839ecd69f284f539befd0b2ceb9fd29 : SUCCESS in 2h 30m 11s","accounts_in_message":[],"_revision_number":2},{"id":"bb0c6cf0f3339e928d0a2245d4e181cd2d466ae7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-17 17:36:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d3d94dc76b151502b9795a0562035b8804a77ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 18:44:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e07e13ab6ac49feba789ba334015619\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375257f5bfce429087c85459ed9e4ef3 : SUCCESS in 8m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7f2ded7cf4740f7a071f623e1888e5e : SUCCESS in 1h 02m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfdf634f5ca34ebea3a40927a96f20e5 : SUCCESS in 11m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa9b336f0f244f94897f22c9314d7efb : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/442a2fb39389420c9bd10781ede897ea : SUCCESS in 18m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/644e6368c1a14def817ed303fa947823 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0101a2c7a2b84bddba91a048e41ea7e2 : SUCCESS in 11m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d984829e052f4281a5bd5cd9c62500ff : SUCCESS in 18m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4783143ba5894b3c82c5c3620f755345 : SUCCESS in 19m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5677d622ab2f4e0b8d7c649520dac765 : SUCCESS in 17m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a63291b023164ff6ba36e61be5712ace : SUCCESS in 26m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/678e3995f1664aeeb1817bcb26d9a5cc : SUCCESS in 11m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0f5a0c54bb6f479fb67eeeeca8444a7d : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/53bf25a5ebf24af58d1441b99538d5b8 : SUCCESS in 19m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/420ae5df4a5b45df855262cfa8f6cbf7 : SUCCESS in 9m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/551b0485a23b45fcaafb69c05a38121b : SUCCESS in 15m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3d6b72338e7a4a51aa42525589ab46fa : SUCCESS in 16m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f8cb4ddf9e6466da651cc6b8830c518 : SUCCESS in 30m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b42205c8d15d444e84248eb62b0348e5 : SUCCESS in 46m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52b6dd17787645059e5544a47ea53b65 : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7cad30a547e424681b8abda6ea5951f : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8fbc8c191e9e45b4bc20453cf2573d30 : SUCCESS in 24m 22s","accounts_in_message":[],"_revision_number":3},{"id":"32e4c7f447afa0d2ffa2e4894e661a39e4b17840","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 19:21:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/283b3a0ffa3d4b458c28bbb5f988e272\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40fa4d9419c24890816846f5116bfafa : SUCCESS in 1h 43m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"853e9f7661f52976a222c1f4328cf82df86a6464","revisions":{"0c5e41ac1d73d4651590dc11a3b645c002b394f5":{"kind":"REWORK","_number":1,"created":"2024-10-08 23:53:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/931851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/931851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/931851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/931851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/931851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/931851/1"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-08 23:53:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-08 23:53:17.000000000","tz":-420},"subject":"tests: Skip IPv6 tests if IPv6 address not available","message":"tests: Skip IPv6 tests if IPv6 address not available\n\nChange-Id: Icb62ad8b9200b6c8d418b1678a2b6b01eeab021c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c5e41ac1d73d4651590dc11a3b645c002b394f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c5e41ac1d73d4651590dc11a3b645c002b394f5"}]},"branch":"refs/heads/master"},"25d14a365909ba75aa97959978155324f8b4ff77":{"kind":"REWORK","_number":2,"created":"2024-10-16 20:18:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/931851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/931851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/931851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/931851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/931851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/931851/2"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-08 23:53:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-16 20:18:27.000000000","tz":0},"subject":"tests: Skip IPv6 tests if IPv6 address not available","message":"tests: Skip IPv6 tests if IPv6 address not available\n\nChange-Id: Icb62ad8b9200b6c8d418b1678a2b6b01eeab021c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25d14a365909ba75aa97959978155324f8b4ff77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25d14a365909ba75aa97959978155324f8b4ff77"}]},"branch":"refs/heads/master"},"853e9f7661f52976a222c1f4328cf82df86a6464":{"kind":"REWORK","_number":3,"created":"2024-10-17 17:36:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/931851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/931851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/931851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/931851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/931851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/931851/3"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-08 23:53:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-17 17:36:04.000000000","tz":-420},"subject":"tests: Skip IPv6 tests if IPv6 address not available","message":"tests: Skip IPv6 tests if IPv6 address not available\n\nChange-Id: Icb62ad8b9200b6c8d418b1678a2b6b01eeab021c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/853e9f7661f52976a222c1f4328cf82df86a6464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/853e9f7661f52976a222c1f4328cf82df86a6464"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
