)]}'
{"id":"openstack%2Fswift~932116","triplet_id":"openstack%2Fswift~feature%2Fmpu~Idf89b0a84ae13c4d18e26f1128aa5985c7d8ec63","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-audit-sessions","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-15 15:51:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idf89b0a84ae13c4d18e26f1128aa5985c7d8ec63","subject":"WIP mpu-auditor: bump revisit rows","status":"ABANDONED","created":"2024-10-10 18:24:21.000000000","updated":"2024-10-15 15:51:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b90efbf87e89bebc1ccae339d5f27e31d965cbd9","_number":932116,"virtual_id_number":932116,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-10 20:37:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-10 20:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24f16ab2f9563eeb8b51167561d69cbc6b96d124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-10 18:24:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca889a660f19343120497b03344d1d816e4c4d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 20:37:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8641c371fae4c8297bd6e6e416e8c9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eefddc1d7394b5d8e7965782ad19545 : SUCCESS in 9m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a913ba3fafe34c008fab8f413f2d396f : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62c746c80aef4d22bf1d37ce5d3dc6e5 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/101e9d1d6e2341ffb4c52e66381f973a : SUCCESS in 13m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/80a79a8fc4ef4b5290186c1330cd7af2 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f1832cdccde4d96bdf74c26f135a0b0 : SUCCESS in 19m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/556e3c608e6245d0a05ba75b81750c2d : SUCCESS in 16m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0def28f5346b40208d7e0d39f9b0d7b5 : SUCCESS in 23m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/12bf3496ef0e4823b02a358f9b0d90be : SUCCESS in 28m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d3408588043342aba509d302bfa45cbd : SUCCESS in 19m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71dc353b884046f59aaa7ead7d0007b7 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d4b31c91ae5f46778d30753107145b23 : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab796404f7314084851c4cc3dce66cd3 : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb66a3510395496abf169b48daa2fd83 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7c4fd0badcb54b5c81b6f25ff88f9ef4 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3f46e2441ee34809aa281059739e8108 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2279c75b8ff84d50ac4e86e0db31ed15 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a311b522ac3459db09a011d498e23e5 : FAILURE in 14m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d89660d9e3ea43fdbafde09ddfdbb272 : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b50d3d9fcb7f41d3922c0ff4056f214a : TIMED_OUT in 2h 00m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fe0d6250ac44c5db674eaf035600f29 : SUCCESS in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795c2d985dba46eb957af143f9c6c970 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f58d9413c5a84173b59f478cc2a880d0 : SUCCESS in 20m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a774ee58d33849454a37d72a53673ed949ca2b7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 23:37:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b5e024f69c4fdfb0715500f174f728\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9007cad47a94d4bace98ea26cb40587 : SUCCESS in 1h 41m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdfc6f8433a943fda4f36db17fd0a728 : SUCCESS in 1h 31m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8547f840805a4957b818a5fed743a462 : FAILURE in 1h 24m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eb5d0055d17457ab35b993ca78e739f : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b90efbf87e89bebc1ccae339d5f27e31d965cbd9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 15:51:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b696f7c3dc21b7375ec4d5b639e74c2ea806282d","revisions":{"b696f7c3dc21b7375ec4d5b639e74c2ea806282d":{"kind":"REWORK","_number":1,"created":"2024-10-10 18:24:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/932116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/932116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/932116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/932116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/932116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/932116/1"}}},"commit":{"parents":[{"commit":"cba3533d7371478a4edbe1b5b26a072b51a08d06","subject":"mpu: remove session state from metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cba3533d7371478a4edbe1b5b26a072b51a08d06"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-10 12:53:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-10 18:20:13.000000000","tz":60},"subject":"WIP mpu-auditor: bump revisit rows","message":"WIP mpu-auditor: bump revisit rows\n\nChange-Id: Idf89b0a84ae13c4d18e26f1128aa5985c7d8ec63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b696f7c3dc21b7375ec4d5b639e74c2ea806282d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b696f7c3dc21b7375ec4d5b639e74c2ea806282d"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
