)]}'
{"id":"openstack%2Fswift~932206","triplet_id":"openstack%2Fswift~feature%2Fmpu~Idc6b6b5a2b665538e861700f5d0996fc39368f5b","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-audit-sessions","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-11 18:27:52.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2024-10-15 15:50:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idc6b6b5a2b665538e861700f5d0996fc39368f5b","subject":"FakeSwift: capture unexpected calls","status":"ABANDONED","created":"2024-10-11 15:32:30.000000000","updated":"2024-10-15 15:50:07.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0b07ab2b6f49bddae1db2814eee54353f7e77d72","_number":932206,"virtual_id_number":932206,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-11 16:49:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-11 16:18:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2024-10-11 18:27:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-11 16:18:32.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"CC"},{"updated":"2024-10-11 16:18:40.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"REVIEWER"},{"updated":"2024-10-11 16:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c64bcc0319d57e07776061149e744c50c99a926d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-11 15:32:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4707773cc9dce3ed45a7b373470cb2f92824733","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-11 16:18:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfeb9a382bf6342320b39eb1d148b2d38f7f013d","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-11 16:18:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b6a4594097bc6307e73e2d4df006cffda0a863c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 16:49:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/703818c774aa4e9194de43b096b776ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0249fac82b834fca8fd59dbca2448937 : SUCCESS in 22m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7beb87e1caf1409586bca6284b8147ab : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd07e9ea3bf74436872266d84ad10fc7 : SUCCESS in 16m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c53fe13c03ca460c9d0944a15a4414aa : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/93c209a3acd6411482eb4c7cdf159878 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/67c36cceb0f0401d9d6e8649ce4f9546 : SUCCESS in 22m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f9eef63049a14718b9834947e7c019fd : SUCCESS in 9m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a33c9c6692104e789a45994fae169217 : SUCCESS in 24m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b5461948204a44b58f12f12ecbe63fee : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1379a26599b54817b3cd977b0ca29e05 : SUCCESS in 29m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d9b0d47dc2e4932972429290f0d4777 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/571669ed9f53404d9dc5668da40acc51 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4abcdd8178c3471f9548ada5d24d76dd : SUCCESS in 17m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/18f201cfb7844e00a8503aab9b4abc0f : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fcdcd9a8a4ef48949c802c22d32d8381 : SUCCESS in 19m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3af6524d890c42ff810a894e42309e7a : SUCCESS in 15m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50df3b3760da4e6589f09521d3e1d36c : SUCCESS in 21m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28e3ccdb2f9847a8935918399e9e2cdb : SUCCESS in 9m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6293ae73e3dd481f88d62fdcd2280b38 : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c519c52af9e1f8869a1d561159dfc0615978b876","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 17:32:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6da52a69a123496e94bf18dee22176dc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/971033c74f654d0a98259816e8682a24 : SUCCESS in 1h 58m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21b4a5b5a4714a529715077ec5c4071f : SUCCESS in 1h 16m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/853577b95c564369b482889fe2cd6bd9 : SUCCESS in 1h 28m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/928cc80de0e44dbf8bcfa06342eeabf6 : SUCCESS in 1h 41m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d18ae7b8b5e86437efcad64079cd0d63d7213214","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-11 18:27:52.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b07ab2b6f49bddae1db2814eee54353f7e77d72","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 15:50:07.000000000","message":"Abandoned\n\nfix merged from master","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cbcfd68e455ee3cc20900dfb16f1063e4003765","revisions":{"4cbcfd68e455ee3cc20900dfb16f1063e4003765":{"kind":"REWORK","_number":1,"created":"2024-10-11 15:32:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/932206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/932206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/932206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/932206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/932206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/932206/1"}}},"commit":{"parents":[{"commit":"3124a89e0969bcd43a3ef8f636eb506b9e583201","subject":"WIP: func test symlink to MPU","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3124a89e0969bcd43a3ef8f636eb506b9e583201"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-11 12:01:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-11 13:12:11.000000000","tz":60},"subject":"FakeSwift: capture unexpected calls","message":"FakeSwift: capture unexpected calls\n\nPreviously FakeSwift would raise a KeyError for a call that had no\nmatching registered response *before* capturing the call. This\nprevents tests reliably asserting that only expected calls have been\nmade. In particular, if the code being tested handles the KeyError\ngracefully then it was possible to write quite reasonable test\nassertions that passed despite unexpected calls being made.\n\nTests in test/unit/container/test_reconciler.py seem to rely on this\nbehaviour, so this patch adds a capture_unexpected_calls option for\nFakeSwift which defauts to True. This allows the reconciler tests to\nopt out of the new stricter call capturing.\n\nChange-Id: Idc6b6b5a2b665538e861700f5d0996fc39368f5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cbcfd68e455ee3cc20900dfb16f1063e4003765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cbcfd68e455ee3cc20900dfb16f1063e4003765"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
