)]}'
{"id":"openstack%2Fswift~932352","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-audit","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-29 09:58:53.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988","subject":"mpu-auditor: audit manifests","status":"MERGED","created":"2024-10-14 16:42:04.000000000","updated":"2024-11-29 14:51:56.000000000","submitted":"2024-11-29 14:47:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932352-p-mpu-audit","meta_rev_id":"5651017ad41bf6a2747a38325f59ffa6a6b7ea2f","_number":932352,"virtual_id_number":932352,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-29 14:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-29 09:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-29 09:58:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-14 18:35:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-14 18:44:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ee246521c9e55e78368c2deddd6dee9e1e451d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-14 16:42:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e09f48ad9642fb30c463ea4504f50647b38edb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 18:35:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/837bd503573245dda4e3d2cefac2f921\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1057f676ad84442a1423d4babd7d6b1 : SUCCESS in 1h 12m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab4098d92cd340d083d6dc3403dda9e3 : SUCCESS in 1h 22m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/84f3b928153e4ea4b06779ab533b311b : SUCCESS in 1h 50m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9beb25a924714dd7983ade7467508acf : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f6c9697620655fd623b49aa0e15f725f72c77ca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 18:44:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f033b0864c874761b85e5bd87815a6d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc87e140976f4da4bd70d5c6fa4c0dce : SUCCESS in 7m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cc1991c872740acaddf29ae4c7979f8 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b9b9465c82848959b8dd1678d808cea : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfafe67d6a3e4d069a44eca6500ad939 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1d5050f20b846f9941a5ee439fc1bef : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b308bd520de3492d806abd9b92a65fdf : FAILURE in 3m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c4a7543ce3e4971aaf25a4aabe6c6d2 : SUCCESS in 17m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fd7206d5c15e463382bcb6a0ba593027 : SUCCESS in 10m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cce39bc879864648a9cd255485bb2921 : SUCCESS in 24m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e442469d68b2471396b083023ab39eda : SUCCESS in 19m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/609312b6f5274ed7be356adce4beec32 : FAILURE in 3m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ba1a048f6304dca83ee418c202bb66e : FAILURE in 3m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/13ff01f741d4488084dad6eafe4f2c67 : FAILURE in 3m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f517685d013c45ffba4736c4d3606cc5 : SUCCESS in 13m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c57bde04b24346198a507cc266f98b9c : SUCCESS in 16m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f1f469879e904c6091803a11e11f555a : SUCCESS in 11m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8db609a6f7e5490e952081f1033e0cd9 : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fea6a419ddc045ad8baf23b8bab26e80 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/963da9d440bf46de9dd20a0604430a5d : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac1d7ae5d9344800b7952065d01b6b9d : TIMED_OUT in 2h 01m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/793e06b4dbe8468dbf0911d43ef41251 : FAILURE in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/027b0d8b5a3242ba9bf02508024130f9 : SUCCESS in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25b77cce1d55498d807a99e0f4ffcc46 : SUCCESS in 19m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5b697bfd802b05dad7d51152cd75bc135d603cb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 13:09:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee4d03e3f20a2c4fd1900a9765ea512cec3c5694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:21:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf61d446f55a4ffab118bfa8eae1783f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f79d979bbc4f8aad310cdb063274b0 : SUCCESS in 9m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08b94bf2269743b195957a8f65fdf573 : SUCCESS in 16m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/969fb51d669143c4bdeede18965653a7 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdc66f9324a944af822c1e2f3351a8d7 : SUCCESS in 18m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a49a5a3d562543d18c4c48638caf4062 : SUCCESS in 10m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c06eb5bfeadf4956b8e5d3924d4e5976 : FAILURE in 7m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d743332c5a2c48e99b83d5aae06a2f71 : SUCCESS in 19m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/845252ab9cb940eb83c50f45d56ffdde : SUCCESS in 23m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/75bb1aac2f234bf3bf8fd17b3cf20992 : SUCCESS in 25m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7877d3f3e1c1481d9c39d390ece39943 : SUCCESS in 31m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ff3093c8a0a465c9c025bf47f97b057 : FAILURE in 6m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2fe839eb92a04a7eb9584f6b9e646777 : FAILURE in 3m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/abe649b9da6e4dbda10e938ac4f6c4a5 : FAILURE in 4m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e09d7da19c434243bf3b5c0500d5a262 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e0ac7c56cf1740e7b32700328f166435 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/60b0ab15ff4440409c0c6eb692eff3fb : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93623c2c645d4dab90f1bb0fde130981 : SUCCESS in 9m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e2af06e84754eb4a515e1140025dc7f : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eed5443d4a504b07968939efb2c455bb : SUCCESS in 8m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6385270891eb4fb190a8621bb6d7f9e4 : TIMED_OUT in 2h 01m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2c3c3b9bcb845daa517c08364805931 : FAILURE in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95681bdfca294e9bbb738e7a0d43aba8 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/96cad9c7196b446f81bdbaa1a08f1efe : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3efb6442e6cfb450a49011430a086612dd3718ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 15:25:43.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"79ef13d98beaab9a6923402b3071e58eef767683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:36:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7dbd3ae79c74a108a2a5ec4b93bd12e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c0de87763741a1bb49090ba0728882 : SUCCESS in 11m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/423eec6543f04ec786bb1040c24ee8ba : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d73160484d744c719aef93aeb95cd248 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e560af6a7fff4487a756709611f1d782 : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b203c615f6fb4523904604e66a49d857 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e1fd72c4be70465fae358d6b2d145806 : FAILURE in 6m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b6ebb59a4404d32a86067bb63c52309 : SUCCESS in 21m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8738b1f087f0459caca6e0f208036733 : SUCCESS in 23m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/169ca984449b4872a2571b8cefd36070 : SUCCESS in 29m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c237d79e03a84f318cb694306b9ed006 : SUCCESS in 35m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a4c3f50acbf4a1b95e042532b8dacad : FAILURE in 4m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7e27c26e6b0e4f12822947cf254e6f52 : FAILURE in 4m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/badf91903ebf47d6bb56ef68b9d67c18 : FAILURE in 3m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7ec48caa2e9d4c5fbff0ae25dca34143 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ef634479939146579d2be9c3e8073009 : SUCCESS in 13m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/585556a04ffd4896ba74ee8c4953e60b : SUCCESS in 14m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/859cec682d9644b88fc4526ed39d057d : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89637be3c97a49ecb5c66b3a6106867e : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eddfdb3982e641e8a0bd391f86f941c7 : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/375b8dde5e3f4d869f8c2ffd995e0a9f : TIMED_OUT in 2h 00m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f7e18568ed94ad6bf267a95a59fb655 : FAILURE in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dbcc04d17f04ac98f2e9452cb8a97b6 : SUCCESS in 3m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b7cbcf01fffa402eb1ac40a64dd2a100 : SUCCESS in 16m 02s","accounts_in_message":[],"_revision_number":3},{"id":"36ac3d60ff56d126f82938cf1c9ebd9a10a9dcb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:54:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebc1f6a1114344c58f49dadf86f83df4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f388709bd3c1447a8f7c6f25c26c1719 : SUCCESS in 1h 56m 04s","accounts_in_message":[],"_revision_number":3},{"id":"eedfbcd032fc10e3a1f94014babde6165915bc9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-16 15:34:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2cff63855ea4d1172ad133020fab28080567b0ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 17:57:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d60ba6b84224a8da7c348fc425d9e5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2274b07102444958d046da60df01335 : SUCCESS in 7m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb9182ac7ab54823858dbd64d8812772 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bea6eb71f95b421a818b89e5982a44b1 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cd5ac281d9141be911d75ac783d1f48 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/93908fcdfaf04e56baf4a53e3a0b3523 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6da6ea08ea67432080895b588aea9ecb : SUCCESS in 20m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/45da774af52246a7abf7a36109abac48 : SUCCESS in 19m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/75695a3464e0420a9d1848779935867c : SUCCESS in 21m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/14576f65a9fc4539bef3a0396f37fe51 : SUCCESS in 22m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/aa2c95c96cdb4dd3883f015181a4680d : SUCCESS in 27m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a03ee37fee3402d98471c9157c3edc4 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/26c75ebe97604af4b33ecf80731044cd : SUCCESS in 15m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/38eb3f02ad8b4ba7b54b7c443caff9fd : SUCCESS in 19m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3436e97e2c214237b8b6d04bed0c8be7 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e7391f75e5b24fae9b26ab4564876899 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/51560c26a3cc4df6885eca59ec471fcb : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96aa019ff03c49d1990c9641b00ed861 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b64a18771794a749df64ab19f4b3a53 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/400f183208ed4faf81c7f1c81ec20dae : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bfe193e81a04ed5aeb4e179626c3932 : TIMED_OUT in 2h 00m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64b97f16908947bc915bf9c992879fa1 : SUCCESS in 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9244be41e72849f8b0909769a72600f5 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf50dbc858a24a5390d71a4d719e65da : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":4},{"id":"f319ea8925b05210a9c3f10bf1a7b5f53274130f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 20:24:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79e9db94642a4ff1b2cc002245af3a11\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff8efaff9b114a3484ca88e0d0049264 : SUCCESS in 2h 49m 09s","accounts_in_message":[],"_revision_number":4},{"id":"91c1483b2ea96492eee283d9bb8ff9f060d5a9f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-18 15:34:37.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"17162b5f81d741cd9ec5e17254b32af5a7dbb21e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 17:21:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fc8d29d7ea45799520a8c522b638ac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0cecd34f71304ec8bdb5efdaee1d7020 : SUCCESS in 1h 44m 54s","accounts_in_message":[],"_revision_number":5},{"id":"adf7dd9188f17fabb1d44bdf3cad84b027c18b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 17:40:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f286cc6f6e34b199a1118fe1fcdc344\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0589c28014634dd797ca09ef81dfe38d : FAILURE in 5m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/159ea765515d455fa33c3fb4b0118727 : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c32e07efe53a44098f1d9ee1fe42f4e0 : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b412ba7d2a4b4a1bbaf45754c6a199e7 : SUCCESS in 23m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1b244a4cabd64a67a0aa4a3771b11b09 : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be9d091959134ff1b575ac6b9fa95c83 : SUCCESS in 21m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/90c682f0195b41568c247e110d9fad60 : SUCCESS in 9m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ea1a457de2b148f8b7649627a8637864 : SUCCESS in 21m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/269f27093d2747d08db49a7a54a3fdcb : SUCCESS in 23m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/66131b5778a54d2c832a9dbe0033efb9 : SUCCESS in 29m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be4feee7a09e4f5b95e4a38a8994430d : SUCCESS in 11m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ab968d7f3caa4aa3870661b1686c050c : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/49eb8ea97e5344c58ac5786854f6e990 : SUCCESS in 22m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7a5cbe2f74d04f1e984421a5c0c62686 : SUCCESS in 8m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ad8646d526974bd3bd41ca158cf574c3 : SUCCESS in 14m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/64a42588c6ca4ccd8633b23028bf1f61 : SUCCESS in 19m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/587da7685d3748f4ab250146695841d2 : SUCCESS in 4m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ee700e57993a477a97cdef639d147e93 : SUCCESS in 13m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9e3ffdb8609d4052b8d393568d1fba83 : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/78956f4bdc3a41c0b6636b0e816d2404 : TIMED_OUT in 2h 00m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49f9c8ee40114657855acc9df46d134a : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66af6ca1b5a34079815b343ccc37530c : SUCCESS in 3m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60b2952877584216b3d85ae9fc17f75c : SUCCESS in 24m 20s","accounts_in_message":[],"_revision_number":5},{"id":"b2652224045e8bfdef1032162b9fdb69ed4e8d9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-21 10:30:53.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"de9d85bac2c3d8a526eb84caaf25e806d6fec101","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 11:43:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e3343abb92c4079ad4865c98fe07c28\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9cc1b98f05064e33bd61d7e68f756d93 : SUCCESS in 1h 10m 38s","accounts_in_message":[],"_revision_number":6},{"id":"dc888f03fc05d149856913ff7ceece7372a91d9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-21 12:14:57.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"9210d4b8a99906bbcfa5690f4ee52271e68b8f1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 13:16:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e773c8e5a97d4b34a8dcfa5c1c6d90ca\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63c58e3289f945ffa272505d77ba2e1b : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":7},{"id":"8aad66b8094b33c510af9db3682b594f6e813d93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 14:22:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac912d7f7d054fd7837dac74bed2275b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f32e5b385c744e77b3f994d65c2b1395 : FAILURE in 6m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d4c97a0cc8f4db789008f82b5235fad : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6150d019a2f45ab8a4655f48ef128bf : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05ec8597e9ab483e8e201826f282be83 : SUCCESS in 18m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd276749424143c4b0dc868276bbc02e : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e15261ef2d5e43179d64ff5c4ba2a4bd : SUCCESS in 21m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a509090eda43489dba00bcb00cf12ea0 : SUCCESS in 17m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/94d66f9e570d47aeb0eb83abd71a5098 : SUCCESS in 21m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2fae586490de490c857cca4504aab4d4 : SUCCESS in 21m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8e4a37e150dd4665a071c7f2a1541bb5 : SUCCESS in 29m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5f397cf00dfe47eb88655989e54296d5 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a5f5d3fc69ef4e6d8a517f0c799b458c : SUCCESS in 17m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d87ace3ddccc4d1bbdbd2b02cfce6142 : SUCCESS in 16m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/87f089a292044b819d899679fb3efb7c : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9a1ab11332f3473aaac1e32685a2ce10 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc1f43285a6a46459823ebe22d56c6db : SUCCESS in 17m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0aafb4d5c4114a21972b123c076a4a4f : SUCCESS in 5m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b924dc85d06041bba2afa96073f0eb10 : SUCCESS in 13m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/426a1ee35ebe4be4a67756e02f7f4d86 : SUCCESS in 6m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/298d5626682749f2a977808834426634 : TIMED_OUT in 2h 00m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1eb9265427ed4da1ba265985fe0c84b9 : SUCCESS in 19m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deaed298bf7c466a85943d920f599819 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5902d031da14854bbbf46c5a91dc0e9 : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":7},{"id":"a34facc8ee98bb02c219c207ed919709c7264d6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-23 16:15:59.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2ac3506b27f1bb1601598483965e20e11e942891","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:10:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6d8cb9a61e94a5aa604229fd4419ac6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7b6d25ba04e4ba2a830a8712234e16e : SUCCESS in 1h 47m 04s","accounts_in_message":[],"_revision_number":8},{"id":"22cec3ab87cf14c5c7bb5f4154cf3e04ce37c4bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:18:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60013a61195a4cfb9fa4fa5b9d3bd0dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fec64ba906f4ce6b4e5c9916ad163ee : FAILURE in 3m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d50c79413b364e52be2a8a375d47d2e6 : SUCCESS in 13m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5802e03fa1d40a1b6a7ac8cda0e9026 : SUCCESS in 11m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e23054cff3414bdfa01e42cbb6495dd2 : SUCCESS in 16m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/818af3a896df4966b18af0358b966def : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6915be010c384edb9ec3e664311e2f23 : SUCCESS in 25m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3f469f45e8114292919dcb67eafb3f49 : SUCCESS in 10m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a2bc87fd4658481a8f68d8f8fa86ba34 : SUCCESS in 24m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a1d0274c6377494084d74d229307c317 : SUCCESS in 23m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/29fc99ba6a284322a6ca4223bad2c00a : SUCCESS in 14m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e092bec248c4117a2f370048682d409 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/61a4048d8d32478b88aeefb7fdd828f0 : SUCCESS in 20m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8004af5a3da7420ab5a1a70dc3e43aa0 : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d88ce4148b454e97baaba86a95c7ad96 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b8ff5978c77044b8b3e3f7e1b79c9330 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5519e7a881f3452a94a7c2f46f7eaa2b : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eab6f50183e24b41a97af9a13d591b17 : SUCCESS in 8m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cee5e23d106f4a2fbacb76c6c8ac6e57 : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/781c0f33df2b46d588e3f6cdd41ddb18 : SUCCESS in 8m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1b0cc8a195644e4bfa2bbac3f78e61f : TIMED_OUT in 2h 01m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9112a976a5ca479aa956a48fd0fe9b76 : SUCCESS in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ecb49737fd4e87a4d483e0bcf3acd2 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/98576fac5f72463783f73195cfe4c9aa : SUCCESS in 22m 49s","accounts_in_message":[],"_revision_number":8},{"id":"996e9a22bef5cf8b092e34f0c8f6d8590cdfadbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-25 18:10:08.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"f6f504a4128c9084596e0004f2aba637ccd753c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 19:55:21.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae7de7dd192b4091b4698b5ff97289d6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/61116613430742ea8ea4df4b69610e6e : SUCCESS in 1h 42m 40s","accounts_in_message":[],"_revision_number":9},{"id":"2c59a0f463419b3a8340fc64ce248ae0477d58ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 20:12:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7bf55e955c84cd49d9d798496b897ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7acb164ba54b10aea50703f8c0bee1 : FAILURE in 6m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe5f6b03a8a046f2a9097629ad30e68a : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72f78ff9e6cd4bc1bddad51485bf07b4 : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1456da4dea754111ae1e5f7df23f6b09 : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47e8d7131aed42f0819825ae2066905b : POST_FAILURE in 33m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c939095b383146458ce4bfa53f6a5f7a : SUCCESS in 14m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8199f6a253c04348afa31577b143a890 : SUCCESS in 21m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03cffd9431184cbbaa86279028f43552 : SUCCESS in 19m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0490212d060f4219a1a9d85168e77d0b : SUCCESS in 14m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1438aaa724204d7a854af97fe2701af4 : SUCCESS in 27m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1929e1c78f584fd7a3955d7af9ebed9e : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f64d486e899467f8513797e65d15c25 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba3b9783a433420c9baadf47a9db1e31 : SUCCESS in 16m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/118cb85728294a188106861ebae1705f : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/afdd44ea1c4a4d9aa6b7e3793e01ae33 : SUCCESS in 8m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4814af67777b4826b2906fe7b2580c51 : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/410c08c6e8ae4ef7a71253907d3f3e34 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ebd6f881041247dfb9bdef9804dcc019 : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b56a606c2d254e87b0c0fa25ac79b860 : SUCCESS in 9m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3acd5bba0564fbab42e6be5d0b2e324 : TIMED_OUT in 2h 01m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd4405c85054464583a2254f3109ea5b : SUCCESS in 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae68e8d1f784a80926458fe7cb6bf2d : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26071b65a55d4480b7b0c79f41c49520 : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":9},{"id":"8b8a07277383abef56bd45c5f3092061ccdffd1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-28 18:34:34.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"93020031eebd4d20229c63f2c2b1cef2a8ab9f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 19:38:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7580fe6e30e7411d870501e45cd12e4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec95947343e493b81e331754a1b7fc0 : SUCCESS in 6m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/004ec444827c4185963790120a9ca82b : SUCCESS in 11m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9433c81719649b799c18ec338de18e9 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3297d68d2e414eca8244426805252aed : SUCCESS in 12m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/486339d4cc834829892c0735d7828aa1 : SUCCESS in 4m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/56c24c5df17b4d3cb666ac017bc9b472 : SUCCESS in 20m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b480ad3effc4a0aa5e2163744c708a7 : SUCCESS in 20m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d5884b7e61ee425ba7085efbc23a57cf : SUCCESS in 20m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d5dfb50f14844443b20c455d8b7abacf : SUCCESS in 28m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ab68b55687554f18973c4b0acae02e48 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1918691e817f4209a2e7b204a3414e94 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/770fd2759dcb418e9201032c19e43e1a : SUCCESS in 18m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a1dcb661343e49e2893b6852542cc111 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8b0f233e36494662b21288281d32208d : SUCCESS in 16m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1914d9f74ad841c89a0335c7c55bcbae : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/755cb71ef9a645408ada888b5607c600 : SUCCESS in 8m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1d477f99197d4364a40f2dd70db09181 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a40821f647d142e9b8d2cae8a1193d02 : SUCCESS in 8m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4ce0a0821854d188c722f178c79d456 : SUCCESS in 59m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e9e9192495646f08848718a16b1e8fc : SUCCESS in 22m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d4c23fd7cf24c70a16f0586f9beb5ff : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9344474f70f4a8bbd9a4c105a7a0582 : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":10},{"id":"662485203cf99a684a6ff7475e75634af4537055","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 00:21:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e198603ea1148ac96f539856d8a9cdd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69299a10004140389a872f3809e0258f : SUCCESS in 2h 50m 36s","accounts_in_message":[],"_revision_number":10},{"id":"67d0dde214c2f8e023553b3ca20a5ecf360c3194","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 09:58:53.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"04803f566523a0e023e2a15e4827fe936588afaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 09:59:54.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a4ed4987e977b757d13d32c1bde43497dfc7149f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:47:11.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f593bdb0632d4f9cb974318d23ae2c45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdf45f2490104e0ca4a02fccff3cfa71 : SUCCESS in 9m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d9be79d601bc4fe8973212a3cc5719fe : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c7285541ba3c4333a875d84cec727949 : SUCCESS in 23m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd5f4c652bc54a42be375c7ba35ac11e : SUCCESS in 8m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/25d4b00c0822417d9418f71dd83374da : SUCCESS in 10m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6ff1434bc7474a37a38671b18b32891a : SUCCESS in 19m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1a77b40c86434958a065c999c65159ca : SUCCESS in 9m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3df26cc5896460fae65ae47c0f5a76c : SUCCESS in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d2f3123685f44e2b393d21b6e9202dc : SUCCESS in 16m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f0924a3b72b5479b82e051471116ee5e : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/31b21edb918342a68f0cbc0d6e2f1fab : SUCCESS in 17m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0399fd8fee3e46d48ad3204d2cdbb97d : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0eb5e57f2cb446bfaad7897f8c39c9ab : SUCCESS in 9m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dfccd8d5dc5c49888340b6d9bc3351b7 : SUCCESS in 4m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/11d4d9b999114af9a7e955ddd2488cbd : SUCCESS in 54m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4719240d64ec4fea839e686fd09f5a5c : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ad01e670f24d23ad69ccf68b2f51b5 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d857a2d78b054d97a5b43eab8f6ad2d2 : SUCCESS in 20m 15s","accounts_in_message":[],"_revision_number":10},{"id":"7cef9b9a6999b3990d7a1386e0728bfbf28f3725","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:47:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"5651017ad41bf6a2747a38325f59ffa6a6b7ea2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:51:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0bbad77a62749c28b2d8436bf915280\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4223daafd2348fcb841a38e4ce8499e : SUCCESS in 1m 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/eaad0f807244499eb1930c7746ab034b : SUCCESS in 1m 35s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"09af73cc9807dea3a1c72977387e2bf805e6e3d4","revisions":{"7d5b334e5bf8e8c1717c672ef8b93729518a44c5":{"kind":"REWORK","_number":1,"created":"2024-10-14 16:42:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/1"}}},"commit":{"parents":[{"commit":"079b9d9ad29c45ed8bd5309dbd29627e56861026","subject":"mpu: Add MpuAuditorConfig class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/079b9d9ad29c45ed8bd5309dbd29627e56861026"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 16:41:42.000000000","tz":60},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d5b334e5bf8e8c1717c672ef8b93729518a44c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d5b334e5bf8e8c1717c672ef8b93729518a44c5"}]},"branch":"refs/heads/feature/mpu"},"4734980e8b268c81b9cc13c127124407e22302eb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-15 13:09:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/2"}}},"commit":{"parents":[{"commit":"c226fb20fb2ed49f83a69e16ce112dfa1d2fe63f","subject":"mpu: Add MpuAuditorConfig class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c226fb20fb2ed49f83a69e16ce112dfa1d2fe63f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-15 13:07:52.000000000","tz":60},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4734980e8b268c81b9cc13c127124407e22302eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4734980e8b268c81b9cc13c127124407e22302eb"}]},"branch":"refs/heads/feature/mpu"},"93e3823d284600416a8716e772473322237fe9bf":{"kind":"NO_CHANGE","_number":3,"created":"2024-10-15 15:25:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/3"}}},"commit":{"parents":[{"commit":"311c71c7b4b5b34a6613ecbb87b03c05574fb77d","subject":"mpu: Add MpuAuditorConfig class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/311c71c7b4b5b34a6613ecbb87b03c05574fb77d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-15 15:25:29.000000000","tz":60},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93e3823d284600416a8716e772473322237fe9bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93e3823d284600416a8716e772473322237fe9bf"}]},"branch":"refs/heads/feature/mpu"},"ae475e0e9ec8b5f8b22be7bc95b6b6e8e6ca58d1":{"kind":"REWORK","_number":4,"created":"2024-10-16 15:34:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/4"}}},"commit":{"parents":[{"commit":"0320bfe2ec2fa878df3b3829df0bc598e36fe4a3","subject":"mpu: Add MpuAuditorConfig class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0320bfe2ec2fa878df3b3829df0bc598e36fe4a3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-16 15:33:53.000000000","tz":60},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae475e0e9ec8b5f8b22be7bc95b6b6e8e6ca58d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae475e0e9ec8b5f8b22be7bc95b6b6e8e6ca58d1"}]},"branch":"refs/heads/feature/mpu"},"350754ecc4979292a96cd0e109a46667cc20f829":{"kind":"REWORK","_number":5,"created":"2024-10-18 15:34:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/5"}}},"commit":{"parents":[{"commit":"0d3fcbddd7a1ee9ca572471f2c3112646923df8c","subject":"mpu: Add MpuAuditorConfig class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d3fcbddd7a1ee9ca572471f2c3112646923df8c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-18 15:33:43.000000000","tz":60},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/350754ecc4979292a96cd0e109a46667cc20f829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/350754ecc4979292a96cd0e109a46667cc20f829"}]},"branch":"refs/heads/feature/mpu"},"92dfead11ce3647f3992b450f0249687286bfe30":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-21 10:30:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/6"}}},"commit":{"parents":[{"commit":"511ab799b29bbd436aedc121c5bc271f0d5cf5d4","subject":"mpu: Add MpuAuditorConfig class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/511ab799b29bbd436aedc121c5bc271f0d5cf5d4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-21 10:30:43.000000000","tz":60},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92dfead11ce3647f3992b450f0249687286bfe30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92dfead11ce3647f3992b450f0249687286bfe30"}]},"branch":"refs/heads/feature/mpu"},"c1d8d353faf91ec572d48d23ec333e27be4f6db0":{"kind":"NO_CHANGE","_number":7,"created":"2024-10-21 12:14:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/7"}}},"commit":{"parents":[{"commit":"c98dfd6014cf666385d5cf977ed116d81be9fed5","subject":"mpu-auditor: audit sessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c98dfd6014cf666385d5cf977ed116d81be9fed5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-21 12:14:43.000000000","tz":60},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1d8d353faf91ec572d48d23ec333e27be4f6db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1d8d353faf91ec572d48d23ec333e27be4f6db0"}]},"branch":"refs/heads/feature/mpu"},"381491777c194fb40e0e85e1636ce6ee3151ffcc":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-10-23 16:15:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/8"}}},"commit":{"parents":[{"commit":"f5075fb04caa39826a442282829a5887c5ef5a9b","subject":"mpu-auditor: audit sessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5075fb04caa39826a442282829a5887c5ef5a9b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-23 16:14:11.000000000","tz":60},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/381491777c194fb40e0e85e1636ce6ee3151ffcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/381491777c194fb40e0e85e1636ce6ee3151ffcc"}]},"branch":"refs/heads/feature/mpu"},"63770e3e05e39f03ac1044fb7cf4c3f192cd26fb":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-10-25 18:10:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/9"}}},"commit":{"parents":[{"commit":"7a0f2626dd3338b2666bc3c709cbf64f2021d66e","subject":"mpu mw: add POST request handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a0f2626dd3338b2666bc3c709cbf64f2021d66e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 18:09:54.000000000","tz":60},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63770e3e05e39f03ac1044fb7cf4c3f192cd26fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63770e3e05e39f03ac1044fb7cf4c3f192cd26fb"}]},"branch":"refs/heads/feature/mpu"},"09af73cc9807dea3a1c72977387e2bf805e6e3d4":{"kind":"REWORK","_number":10,"created":"2024-11-28 18:34:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/932352/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/932352/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/932352/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/932352/10"}}},"commit":{"parents":[{"commit":"9636ef0694f2b2966725063e5c79d10169fd5d9f","subject":"mpu-auditor: audit sessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9636ef0694f2b2966725063e5c79d10169fd5d9f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 13:26:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-28 18:23:20.000000000","tz":0},"subject":"mpu-auditor: audit manifests","message":"mpu-auditor: audit manifests\n\nChange-Id: Ia70c8b19ce4cb2c49a5ad4f81fcdc0a7e7095988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09af73cc9807dea3a1c72977387e2bf805e6e3d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09af73cc9807dea3a1c72977387e2bf805e6e3d4"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
