)]}'
{"id":"openstack%2Fswift~932356","triplet_id":"openstack%2Fswift~master~I7213c5e52be0acefff42b8e3845e7dd0378b23b5","project":"openstack/swift","branch":"master","topic":"removeLoggerAdapters","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-25 13:54:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7213c5e52be0acefff42b8e3845e7dd0378b23b5","subject":"sq? add SwiftLogAdapter clone() method","status":"ABANDONED","created":"2024-10-14 18:02:35.000000000","updated":"2024-10-25 13:54:22.000000000","total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"3d8188d79a5ce435731253f6a3c755edf4d3c658","_number":932356,"virtual_id_number":932356,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-15 12:58:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-14 18:57:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2fef06a2e729f1f877143269cd065f8e5b951c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-14 18:02:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa164164c376b7da9c01742858aad16110c1b7a8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-14 18:06:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"836c58d415d33fdb413e292397d1583faf9ec6da","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-14 18:08:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32c9a73a6429307f51c5aeb10586068aec1160fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 18:57:46.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfadca9343f7471cb0ce33e70bbbed87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc66a775751640e1a2994590186cab99 : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/56e6d525af5b4c86a91965b7fb7af534 : SUCCESS in 50m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d38c26d3a13a4521b272ac4889b9203d : SUCCESS in 49m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c1c56a9189c45e5825d8644bdf7a398 : SUCCESS in 36m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b452ab36ec1406dbbace1693deba0a4 : SUCCESS in 52m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1739ae796984f98807de764daaaeed0 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63b367fcabb24f6fa1f232c39a94a8f6 : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f00c72afc2a4f6487114af51ee5a8e9 : SUCCESS in 18m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4fc99abd7104433ab6a30305e5b5287a : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/93ac2a0995624ae98217d5d61bc7d91a : SUCCESS in 11m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35bfa8c958284903901171bb5f86ecc3 : SUCCESS in 18m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/df1a876913684561b28c37db9900be17 : SUCCESS in 17m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/789bc605a8f1455997a0332579bc388a : SUCCESS in 10m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8d35594d5c18438f928966906d0637c2 : SUCCESS in 18m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e339af2ed10d4276b991b8aa7aaa3636 : SUCCESS in 10m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8456ff30145e4b2990fd4f07950dccd4 : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/056e6ccab0734f53a46a243f65826453 : SUCCESS in 11m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1223436efa54461a907e0624a2c67ee7 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4587de9a51694177aef9668f707fd32f : SUCCESS in 9m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ebaa4bb0ba6442f29ff29654f94f2d00 : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96b9a5aa35ff4e90bc2f7eb25e8be7cf : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e85a87e04274373871ca072a5976761 : SUCCESS in 9m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/31a0f9da9b524ca38bf85926633cc6f6 : SUCCESS in 6m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ac721c237b649b3b2692b741e60bed5 : SUCCESS in 52m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77423c224bf542219f85ce7c20655913 : SUCCESS in 54m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58c36f1b81f74e7c81883692e5b7585a : SUCCESS in 30m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eac72d314c1b48d99f8c75838087c27a : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/756686099ebe41f384c298298a808307 : FAILURE in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f8b3314e4764b278f5f6415ec40c7ce : SUCCESS in 11m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f59212dfdb0406baa8cbb3b0b36a570 : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":1},{"id":"11df96109a34cbc71010bbd1ce75da6f81c0f2cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 20:01:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c72c625708481abd75777da1bed0a1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5629dcea74b24a62914a7817d079f411 : SUCCESS in 1h 56m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3141f3d678fcd78725e6e82a37e23cad5d3cb9fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 10:39:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9a4dddfed0481872d926462e15e4e5eb42b11c22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 10:54:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e927a83a626c716613411abda460440b16248cb2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 11:06:24.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8922ab354c8d56246e16ede7b3edf309d4809724","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-15 12:02:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b937fe50a8b5e1c3f7e1948e504bf919004758bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 12:31:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a35782105ded4429a1390458da00b77f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/88584cf258a94de79bb56f793ed6b615 : SUCCESS in 1h 34m 04s","accounts_in_message":[],"_revision_number":3},{"id":"9386308c90b8b672fe55ee871cfbc864de6fa52e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 12:58:20.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee458fc893e54ebbac4cd48351bdd6b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e371f5274a0484a825a5cc1c0b2d83d : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/5e8f267a1d1240d8b4aa0257ee43e3ce : SUCCESS in 58m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6dbeaa038c04d61a236f690f4c4eacb : SUCCESS in 1h 00m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b38ba52789aa4d41af65c2872bac5100 : SUCCESS in 40m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfe87a5646c146c2bdff5e9dee35f640 : SUCCESS in 1h 02m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff6e0996452342bbb32d253ce4140701 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/311e8de27dce49f99607991bf3625255 : SUCCESS in 11m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71bed787e3df42cc8bec83451c853c0e : SUCCESS in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fef2437fb4ea4131ba6e594f4202cc76 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2c648b5299c94d9caf98719274773628 : SUCCESS in 21m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9bf8c74148ab41789eb8edb5814b003a : SUCCESS in 18m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/983d5c2ba5a84238beaef3217f3fb453 : SUCCESS in 19m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/92842443013f4565a8fe8282af5b5bfe : SUCCESS in 19m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c18149a7553243e6acb52bc71d92508f : SUCCESS in 23m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4805fa39352d42b392080938ed1700ec : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ba5ebcf46da43529e87c8632478410f : SUCCESS in 15m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/71c5cf50d49849c48747793ff579aa63 : SUCCESS in 11m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5feff40712fa409ba237a3c1195f46ba : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d5199657ff9d4c8ab93ab043ca6e8457 : SUCCESS in 18m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/abdbfc6a3d5b4158ab8458abed64914c : SUCCESS in 18m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ed6f703dc2b4451a4f702770a94744b : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/99becf2f70694f4ca279d67e3cdf64f9 : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd1b8c23a0c146f7ab91d26095c009fa : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/be0157c29a0a4cc8ac75ffb8e5d56318 : TIMED_OUT in 2h 01m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30389aec10d44298bf4f8d94a1bb4501 : SUCCESS in 48m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9917485f68e49079f96f2e61f197a1f : SUCCESS in 50m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e13a2ee81734acb82cafebdebda71c3 : SUCCESS in 21m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/340f48da133e447899fd8f9d2b330734 : FAILURE in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60415e1c61064ab4b8ff870d9e4af0b9 : SUCCESS in 20m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48aeee9ba8d3405b95c2297d9997d8af : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":3},{"id":"3d8188d79a5ce435731253f6a3c755edf4d3c658","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-25 13:54:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5ecf60606e69c5d544af4745afc5f7b4e956da3e","revisions":{"60e646d7c4f0b575c6f02436548da5f84cf31a86":{"kind":"REWORK","_number":1,"created":"2024-10-14 18:02:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/932356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/932356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/932356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/932356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/932356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/932356/1"}}},"commit":{"parents":[{"commit":"8bd333ba690887cda6da623c35356e75d82030be","subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bd333ba690887cda6da623c35356e75d82030be"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 18:01:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 18:01:31.000000000","tz":60},"subject":"WIP: try to deal with custom log handlers","message":"WIP: try to deal with custom log handlers\n\nThis doesn\u0027t work yet. Oh what a mess get_logger is!!\n\nChange-Id: I7213c5e52be0acefff42b8e3845e7dd0378b23b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60e646d7c4f0b575c6f02436548da5f84cf31a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60e646d7c4f0b575c6f02436548da5f84cf31a86"}]},"branch":"refs/heads/master"},"b56900d5499cfdf8ac56b109454c8c4a270a6808":{"kind":"REWORK","_number":2,"created":"2024-10-15 10:39:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/932356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/932356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/932356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/932356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/932356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/932356/2"}}},"commit":{"parents":[{"commit":"8bd333ba690887cda6da623c35356e75d82030be","subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bd333ba690887cda6da623c35356e75d82030be"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 18:01:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-15 10:39:20.000000000","tz":60},"subject":"sq? add SwiftLogAdapter clone() method","message":"sq? add SwiftLogAdapter clone() method\n\nAlso add test for custom_log_handlers.\n\nChange-Id: I7213c5e52be0acefff42b8e3845e7dd0378b23b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b56900d5499cfdf8ac56b109454c8c4a270a6808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b56900d5499cfdf8ac56b109454c8c4a270a6808"}]},"branch":"refs/heads/master"},"5ecf60606e69c5d544af4745afc5f7b4e956da3e":{"kind":"REWORK","_number":3,"created":"2024-10-15 10:54:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/932356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/932356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/932356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/932356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/932356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/932356/3"}}},"commit":{"parents":[{"commit":"8bd333ba690887cda6da623c35356e75d82030be","subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bd333ba690887cda6da623c35356e75d82030be"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-14 18:01:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-15 10:53:38.000000000","tz":60},"subject":"sq? add SwiftLogAdapter clone() method","message":"sq? add SwiftLogAdapter clone() method\n\nAlso add test for custom_log_handlers.\n\nChange-Id: I7213c5e52be0acefff42b8e3845e7dd0378b23b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ecf60606e69c5d544af4745afc5f7b4e956da3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ecf60606e69c5d544af4745afc5f7b4e956da3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
