)]}'
{"id":"openstack%2Fswift~932447","triplet_id":"openstack%2Fswift~master~I07a82025658eeabf462b8e26b7bc352530ba88b9","project":"openstack/swift","branch":"master","topic":"fix-rebase","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-04-10 23:11:09.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-04-10 23:03:02.000000000","reason":"A robot voted negatively on a label"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-03-18 13:35:59.000000000","reason":"Reviewer was added"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-07-23 01:07:04.000000000","reason":"A robot voted negatively on a label"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-04-11 00:20:25.000000000","reason":"A robot voted negatively on a label"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-01-16 19:39:50.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I07a82025658eeabf462b8e26b7bc352530ba88b9","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","status":"NEW","created":"2024-10-15 20:18:19.000000000","updated":"2026-07-23 11:50:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":84,"unresolved_comment_count":10,"has_review_started":true,"meta_rev_id":"7975aa26ce466bdc1e9f56dc9f2bb563b3f6cd6d","_number":932447,"virtual_id_number":932447,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-23 11:46:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-15 20:45:00.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-10-15 22:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-15 22:22:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-21 14:28:06.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-10-22 18:46:47.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-11-27 17:23:22.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-02-11 03:52:45.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-03-03 16:18:01.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"},{"updated":"2026-03-18 13:35:59.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"be1ac6be5f1d060585391046afe0658caa70ab08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-15 20:18:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54ce94dcfa8e4a56e1c9996eaf5f473c3356b541","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-15 20:18:59.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"5b78fd28f8aa974a4b1d5c706a87109f28b89540","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 22:21:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60473a28766349e28a2f37e0fd9a4d35\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22a352e7f433470394ad35c6a25cff46 : FAILURE in 2h 01m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3d54a3907b160107b2b3fbf5a154679edeaf5b43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 22:22:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bae611683884bdd8289a91a0d23fd9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b559584d6bc4b0aa5984fcddff7ab0e : SUCCESS in 6m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/b4d5df60c3d14af79f53f8fd89f0a879 : SUCCESS in 43m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd19e6a785c84b08913f6ab20d31c9f2 : SUCCESS in 1h 08m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c34cea8b9494db1ab17c510c83f5536 : SUCCESS in 41m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e50269e82d94e609a5a4cdd37b10943 : SUCCESS in 45m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ced4194944fe4dc793837d7e2a6ed074 : SUCCESS in 11m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecd020f410a142e894d230143c1c316c : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/346019f76cb04731bf2cfc935b66fac5 : SUCCESS in 12m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fadcea448d1341f99431b8690c85cf35 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cdc37d01493e4283ac75c1fb656e8bab : FAILURE in 18m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5dbf6e1dd89c4fa5bed676bceae130e7 : FAILURE in 19m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9d17abe5faca449c9c2a09f22cb973a4 : FAILURE in 16m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/55c40c931fc44e52b20e466ee1cc8d85 : FAILURE in 22m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3bd4ab885fc04e55a65c820f7ce3e3e7 : FAILURE in 32m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c07935c9912449c9ec331971a9305a2 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e00041a900cb4213955d3d06a77f0eb9 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5a00adeea5b64465bcfde2971b585dcd : SUCCESS in 18m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d9d54a68fa5f444abe1f997c63b22eef : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ab979061ffa44f0c9444709fa38737fb : SUCCESS in 15m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d54dcd1e4b9248d28770a46a2b83e7df : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ae556bf77154d7987d4bd96ece48f87 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/003dbeb8be4549be856f3f92be7c353d : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/56f8ee178c1d465ba78771625af06efa : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/142a8f844ab240e682ac66da030b6093 : TIMED_OUT in 2h 00m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/83d261e9606d4e31bfa4242ea5ee67fa : SUCCESS in 31m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25d13624cf64444590a91dc40f43ebd4 : SUCCESS in 31m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58a283e84bd24dc4b4fc9e7bbbccb3bb : FAILURE in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98ca2d2a97f04817859a4d1d4e23ec09 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2fc836c36cc34955a60d2c76bd5cd55e : SUCCESS in 20m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b5c1930bd09435dacd3091efab89f27 : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":1},{"id":"60261c02749e0cdf486c37b10ede67e3230637a9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-19 00:27:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d9b3b9bd7918702b1e68d067ea0a688daf528871","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-19 00:27:19.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"5424493ee4c9d0d9c741ad89eef1bcb93c1566b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-19 01:10:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c58f03f1e698453cad9ff5c2e37142f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927e6f4fca404eea82370a61c596fef1 : SUCCESS in 5m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/214068ac7fee476a81bf80c8a620bb4f : SUCCESS in 36m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5bbd01bead04913be86d160afe96075 : SUCCESS in 38m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ea4252fa64c4113b8878cafe47de46f : SUCCESS in 35m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e699c98ace54bfa8792a27106490aaa : SUCCESS in 42m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a403bca46a5649cfa528f86098520158 : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55e466541fb7403987cfb9222bd07f18 : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33cf7dcabdb0402f83236877361cfdb2 : SUCCESS in 13m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2fe894185a7a47dea48edb8ec18b0638 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e6c4259675af4b49a9f0347bd6f8265a : SUCCESS in 14m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23cc459157364f02a46c9831003c4e65 : SUCCESS in 15m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a65ce11c00cb4180afb5837de73c821c : SUCCESS in 14m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7ffc47e8de07464eaca39c468941cf6f : SUCCESS in 16m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a7d7cc0ab78a4caca15edd10a75677ef : SUCCESS in 18m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8fba791b6d29404bb4eb6f7e2d9ecb1b : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/68af001200f44848adb7a41709562840 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25a39477019144fcb77a0ea210b5bb13 : SUCCESS in 16m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0b5361a54b2b4937a4f8911ae1a1920b : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5151ee901e66478fabaebe6d507a381f : SUCCESS in 13m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c12fb435d5ea47a4862fd5638aa444eb : SUCCESS in 14m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/660330da68ff46759dbcc33b945e9926 : SUCCESS in 5m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b54235a524dc452fb380df7ed19e4bc8 : SUCCESS in 9m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9562a6b6f0f4f18bc24be798da08401 : SUCCESS in 5m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/db393b9fede6432e848c784ee0d367a7 : SUCCESS in 38m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/61a2fac56c3a4254bf995c785c50a12b : SUCCESS in 29m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b67b28ebbe94395bd779f9b7736d484 : SUCCESS in 29m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da0661228b79448c89c4561dad731c57 : SUCCESS in 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d39241531bbd41e885c735c62eef15d6 : SUCCESS in 3m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad3a5d814c4449ccbcf4bd97ebee92be : SUCCESS in 17m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d624560932e04a5482bdeec5b11ade55 : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":2},{"id":"55b028d9313593a9750a83e562323b0d13617260","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-19 02:19:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d014f6f62b34c5fbc51dfb227acac3e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8b955fe04ec443f48f411524ea26594e : SUCCESS in 1h 49m 50s","accounts_in_message":[],"_revision_number":2},{"id":"57792449e7e58db0c3deff89d1a97e5a15533be5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-21 14:28:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"93675d47a103e3444f2ef3f1e38966d8855d9069","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-22 18:46:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0849e11ba152bbab881f44ecbe1404fdd9e0b7e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-20 05:56:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"91057478d097c30f459c30200b9e7113c7896696","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-20 05:59:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2a1dea57cba79ccf65a8ce1df7598d8dd9c385e8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-20 06:01:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a008dce7dda969453255e134bdb3096d3e0fb96f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 07:24:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ce4f7f5c92419aa78b6265ec5fa2a8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b3340cb11f0641d7b10135f7ff897c3c : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":3},{"id":"53c19031abd00ae0e965f014e8c0b11cb705a831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 07:40:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f4dd7258d154f148e2ad3d9b2a12896\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a34126feb1845c0bee49e9570c16d40 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/57d5a6be42df442e93e6749fb310f617 : SUCCESS in 1h 06m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2bb9fce96f1e44feb87cbd65a5c3369f : SUCCESS in 45m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0d3af41781e429b8a364abf675a574d : SUCCESS in 52m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/febba01e9ce9415ba7a9772343a607cf : SUCCESS in 1h 05m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e174cd1df1d349ea96a312cae5664f01 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2357376bcdc249fb8cd84bee43c5efca : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f45aa867272245098f6aceb8a68ff151 : SUCCESS in 18m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/80c78c54202b4826a0c7024878c800a0 : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/317f7265b8a0474b84494211fdfe35d7 : SUCCESS in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f6f6245be324f52ba34729525ce49c0 : SUCCESS in 20m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb3ddb5ffd174b0a9a66ba9890681d10 : SUCCESS in 14m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/606b3b1126684b9ea40ce7d5d9c85b8f : SUCCESS in 19m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a1571d3b9fe74e68af3a96d1eb54d54d : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b689295880ea464ab041bd3b6efe8c44 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/adbd70b857d64487acf52bcf719c2026 : SUCCESS in 20m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ade51dab654a4700bdde75aa462adb38 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/88e3539ceb394e6a9817ef1ec42caeec : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d8bc620236254035aea25de40e1ae4c8 : SUCCESS in 18m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca52585488bd4acdaa15c8906d169d58 : FAILURE in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0b3d56252e3b493f8a367b8a49b11e83 : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cbba116de7a94f14a2d7dc9f4b23d1c0 : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aad9ea8e5e024ec98462f9f9c1f51f27 : SUCCESS in 54m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/963a01d3db304c76b7ef8d026f8d63d5 : SUCCESS in 53m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0420e1575cc2400195f38a3031b07655 : SUCCESS in 34m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78e0ad97170846bc8593b0c972b43194 : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88b2dcd97ad84f1182093f282c8cec65 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8fc058447d99401f8fe8ffb770445612 : SUCCESS in 26m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/721d1d7312ca43cb94bfc1c817002659 : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":3},{"id":"2c787f9d074b3b3ab8421d8cc701aeeccedea6ff","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 17:23:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1dd50d9f8b208c04ac6a20d14b77ebecfb84a685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 18:50:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a07b993329864646b3c6a7b79645eb41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806b583d735a45dc97980a253aed0202 : SUCCESS in 5m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/674ed727f4924a1181a68e3046ac3914 : SUCCESS in 37m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/68d191c7e7c347808b750dd9d6c5c8ba : SUCCESS in 1h 02m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2021eaee384d4f2a8656c34177758da4 : SUCCESS in 37m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b780f870e30345e08369b0eca62a24e6 : SUCCESS in 1h 13m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cdab2807a5043e88c201f6a468f3d83 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b940cfc9cf84e63a0a37eb849f57f32 : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3735e01cc9144daad8ed1d144f9b153 : SUCCESS in 26m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be27663cf2d248b489876b9e793a816d : FAILURE in 3m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ee0fc832dd64f1aa7105e83c8cc56ec : SUCCESS in 23m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/21841b93a07142d1b4d37ab97cb8cd51 : SUCCESS in 21m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c233f591929429f97efc82ccbf44932 : SUCCESS in 23m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/052efb92bc78490fa8009710ba55b4c4 : SUCCESS in 21m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/435149e174fe4a4f99524b741e0ccdad : SUCCESS in 21m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2a4aa271eb3d452bab9e78e3542930a8 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/09a64d4710cc456dad20525dd180edf9 : SUCCESS in 17m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/304ebb2a33cf4c75882392712d2cc7df : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/709a0887726a41a787aab91ae2cc925e : SUCCESS in 13m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f3b7a73d5d0341278ea4e38a2110b7d7 : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/72b87c6882454a408df45f64f50c3aec : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e91868bf35894263ae3b4bcf6a16b6ef : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13f35d282c264d71b8c3dc7bbe5c3b24 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6963b596119414ba103e09fc4bd0806 : SUCCESS in 57m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/19255e5f0cbc485d90fb41c8b5caaa97 : SUCCESS in 36m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/207db384304e48179d6d4a216c5cb318 : SUCCESS in 52m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfd85727e39b4fb88da1dba41d935191 : SUCCESS in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ff36a4ff50492890e13c7643b39721 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26e1b07310244d1a81603771519144dc : SUCCESS in 18m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c269de6999b4ba7819e68e5511613ae : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":3},{"id":"71819e46e157a0009635e725982254238b552c84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 20:58:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be3cafc0d1024f0d8c2aecfb9c0b76ed\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/257205b8161848ee9dfb1368a5705ef3 : SUCCESS in 2h 09m 43s","accounts_in_message":[],"_revision_number":3},{"id":"08a6b1aed228f1abfdfffedbd3bfb825f45c4c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-14 00:27:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a677b62de63f8fb8a0d563d0bdac64ad1f3dd04d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-14 00:32:10.000000000","message":"Patch Set 4:\n\n(20 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5616257998ebdf70aef16582bccdc5574cd050ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 01:39:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27245eb387ac43249d933ad9b672f198\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2a2103f58145feb5c62dffca5ef70f : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/8ca52e99c8c844a980966602c9dda7c3 : SUCCESS in 33m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1581a15ad1854844934bb6035e7e5b41 : SUCCESS in 1h 01m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/80c20037bc54419381744643eb8d8237 : SUCCESS in 40m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea0acd975ad74ccea9f5b3a76c9a0a58 : SUCCESS in 46m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7facda4ccfd9443f80c0130bfe025d33 : SUCCESS in 11m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f71ff09a94a43ab87bf9a753edd7d41 : SUCCESS in 21m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/439fe59be47f4091aa36f3a657d9b77b : SUCCESS in 22m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb6f385dd2b248c7a27c3885487e7beb : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f05848a0dc0a4a799cb5719ae6d61fee : SUCCESS in 21m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5bddfb5fbb1c40feb86b766db5340709 : SUCCESS in 19m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c2942ac13a7e4584aa03b831244e15c3 : SUCCESS in 24m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dc76d433144f4b7b83cb981ba876f0b9 : SUCCESS in 22m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e710808dfbca418b8d2a73539ca3ec12 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dfb314f55101434eba79aec7b46641cc : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ea1d706eac14c04a64a69bfbd057c68 : SUCCESS in 18m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/da4e82e4eb674851b965ffdde56b3dd6 : SUCCESS in 8m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/02e5a3bfe4704b8188526fad1fef8007 : SUCCESS in 8m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ece6cae22b7a47059edd2ab23c0942f6 : SUCCESS in 9m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b82f05271bbc454ca73ed5ed8218c6d7 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/723086b57db54cfcb72c6288670144d2 : SUCCESS in 9m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d8971619a5124ecaab82e1da3e250934 : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d017985b7e884c3e998d7f12ace041b7 : SUCCESS in 58m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48c52c9d64d9455495d664b4509f779d : SUCCESS in 24m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/16a9088e98b54bbe9d07e778229e648f : SUCCESS in 50m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/290b0c253db644ac80672e11c9333394 : SUCCESS in 22m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9c46d2e1f34468bb0542882611d93a : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5050523e6e01411b8658db9675b431d2 : SUCCESS in 10m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09bbe8bca7644573a4ead293b1dd560e : SUCCESS in 36m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87a711e725f44d7abbc7db6f1bb8236a : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":4},{"id":"6df4bfb717af63aa05a291ddb348c80004508d91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 01:54:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17bdae8c0a0c44c19910bf156d84546f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46049953b74c4e36b5b630df0f7e423d : SUCCESS in 1h 24m 04s","accounts_in_message":[],"_revision_number":4},{"id":"6db377d65e1c47cb6fce345d26847e9edd5155b3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-15 18:45:37.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"010ca5cea2f3811bb16c1796d73cbf167bff0ad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 12:23:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0e079eb293c38eb70f9f293e8381eddfde4799c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:48:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80fd79088afe408e989dcae03768db3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c158054b0654171ba09536b0124d0c9 : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/8b403ea6ba724dc384e6a3b8130ca3c5 : SUCCESS in 38m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f61ed4f15b3d4487b43deee3f202a5f3 : SUCCESS in 1h 03m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7c498390744742eb85c396875d93f90c : SUCCESS in 47m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17ae325eed294e8fb65043c7599567f5 : SUCCESS in 51m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f9750378ee64723b94628e551d5b99b : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27a0295160b249219a8fa9e911280ef0 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5c106bb1b92479c9ec1de6fd694a9ca : SUCCESS in 18m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/096a393b416e4e269dfab6a529528c35 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f3e20e9ef494733bd9d664da97e5412 : SUCCESS in 15m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c1d9600f3c834098b128b479b95e74c1 : SUCCESS in 18m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/74f97e50f89a41c0895799d6e8c1f708 : SUCCESS in 38m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/32768b4e0f8e4cf0a7853d021f9a9c12 : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/004d49c196cf487a879de431084470e9 : SUCCESS in 13m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/abc2c00942ce475fb971c7ac5aa6f2b3 : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4608fec0033f482fa18cd0431e3efc9e : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f8d35d7974d40ae83b2e5008a57c134 : FAILURE in 15m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ff3d8663dd64adaa246923cfa9c3efd : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ee0f82c3ea746c68c4fa1990aa4bd08 : SUCCESS in 52m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3077abe6bf4470c89663c0d830d1816 : SUCCESS in 57m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/19ff2c1744544baab481f36c1420ad33 : SUCCESS in 49m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c7c0f382bd344d891ec2ac755a32f2a : SUCCESS in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/566e2de0bcbf42efa82f84e056cb5fd4 : FAILURE in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1590022f569149cb8a8c559a13ff4fc6 : SUCCESS in 17m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b64759eefb02420989bcb9b400901e4b : SUCCESS in 41m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97140f5c8c4f4e3ea62e1a9a78b290b2 : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":5},{"id":"c9bc7a0c05e7e5d809c79d605292c7120881ae79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 15:11:24.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cf949afde1acb5b28498575203d17878e2b76520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 17:51:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dada8ca280c5486e95e9b7eb3c33a86a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2872a3e629f4fb1b540e932286ac5ac : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/429e45ae02da4cf5988cad2bfbd997e1 : SUCCESS in 1h 01m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/335bc8f4d0a84a6dbfdb7ebc34457e72 : SUCCESS in 57m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d6ce85c4f00542fe88229177ea790da5 : SUCCESS in 56m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/267970c50a184d2aacc04d218252bb27 : SUCCESS in 1h 08m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d7beea025cd4541b5f5bc6f088283ec : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b90171548d64d6e9d7cd99d3e3aba17 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7e8fd83401641e4940b05eb6b7bbcd3 : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6aac83aaea414ff799df9b611072068b : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/820b78e0251a4673bacee2a938aecae4 : SUCCESS in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/131c5ea33a504c9cb49f2e6124477988 : SUCCESS in 21m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d1b70ad5c51a46afae3576966822652b : SUCCESS in 25m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/04643f33776e4915b6ef2bb4bc406f1a : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aef09c03303f4de39ca92c2fc1d65d1a : SUCCESS in 13m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0191c54b39864d489082640bf2103c16 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/82242a1996da4fe29278952622a3166d : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b17cbb5c27ef4736b8bbdc0ec0a3fd5b : FAILURE in 14m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d0895edfeda4788b7913e4dcc220517 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d896e113471142118cee9d4e3b4c5401 : SUCCESS in 49m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c66f2849d1a40aaa5d6c2f3c30de700 : SUCCESS in 30m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/829b44104c784aac8280eb3e778ec242 : SUCCESS in 36m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f87a12a31f24449b4b8b8d18552e54e : SUCCESS in 23m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1de0b03f25e7450b969d549b1dfc4852 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/042b833a398b41cca90b38429d6d07e6 : SUCCESS in 18m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2609458b7324c25a8d40392563e90fd : SUCCESS in 1h 03m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d8a5779ecc24be3921476c47e092d83 : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":6},{"id":"d7f9caf63bea3b3151dd961b8fd3de45b6d72dd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-16 19:24:19.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"960a50cbe02e73ee1ef0ff5158cab20d8f6111d9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-16 19:39:50.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9a17e6d68fef48731a62c156dc982e8815899b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 20:25:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a95b9bafbc548b3b0234dfdfb678e5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd7b22f328445568a9ae6e6de378bf2 : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/4ec06de7c843410689ebe93ec004644d : SUCCESS in 39m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d77b0f35e8b45bfab7b6addccf3a208 : SUCCESS in 48m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7ed3918a77442499da8fbb4db6061e2 : SUCCESS in 43m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c025b9a62df4a619db9a4d8321d7bf4 : SUCCESS in 55m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a9b4a70f02744f19a186e57b878098c : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9cfc831c3ee4683b5a958e493d33ab6 : SUCCESS in 21m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31c6bda091a146168c370f049cc77a5d : SUCCESS in 23m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e03ad79c8fd4494965b3fbae0ec8898 : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e61f00c4f3514c9dbee45ef6f38806ab : SUCCESS in 19m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c1496ff40aa7497fae55d14db7231c73 : SUCCESS in 20m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/91db36be214543f4bc082813dd7d9088 : SUCCESS in 32m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fb30668b85f6401c9ddfc930a4eb4e94 : SUCCESS in 10m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3727ca22eb1940588cef9d749f4d21bc : SUCCESS in 18m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a68e425a5c77455592ff368931cec69d : SUCCESS in 10m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/87806a8d278f44e896bce67bab438213 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c759f097f4ec43d780678c67325e271d : FAILURE in 17m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8bb618906f81411093d076da60e55b17 : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/55d05662275b488da0739a94c65b23da : SUCCESS in 41m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0092ad0d17274f7aae3de70415d38904 : SUCCESS in 38m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7be85cfcc1564264a17c3778182b0adb : SUCCESS in 53m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7decac975f34e3c810efef97cf9df3c : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b82ea9cf02c4cd09f81b28b870ad09a : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6dc3726db64a4b088665068d6981965d : SUCCESS in 16m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4f1d90b67994124a13cb25ed8d074a0 : SUCCESS in 38m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1f96eb839f3433a8b19dce17c143c71 : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":7},{"id":"7bd0356df30d3d13b5b69219f4e146906fa7f4bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 22:19:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edb22909cd1545be97ea078ee0045d27\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d80a3f1a32ff484aa01e23fdf4291d04 : SUCCESS in 1h 34m 02s","accounts_in_message":[],"_revision_number":7},{"id":"8e128b58601d778450310abd735df618a77f3d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-21 16:47:59.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"95b8299cd722170eb665e3087e16e025a793de02","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-21 16:50:52.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"15a0ffbcff91cadebf5032885ee14c9ed5a7357b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 18:08:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b4f1464b904a94a6731ce5238e51b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef01043bbd340299c8e6f26ad18b896 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/3e0ed8939c584fa5981ac2096532d200 : SUCCESS in 1h 13m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c32092616225487f9b0fae7aba15e2ce : SUCCESS in 52m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/604feec483f04a33b2c2e49a88d41347 : SUCCESS in 37m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a43f2985579b43c6bff852929c6edb49 : SUCCESS in 1h 07m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c292dc6e11d4c279ccd7c5ec8c63923 : SUCCESS in 11m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cde7bc758834e9f83ae04ea07666d1c : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2db52c49850342b7bda435f3408ef091 : SUCCESS in 26m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/15a45ba5d0824d6fa4e324807988a1ca : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c37027358e8843988be3f20f406b4a41 : SUCCESS in 20m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a45970045884386beae0f6442cdee02 : SUCCESS in 10m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1a52fd17c0384d5da85de31d1aa73d69 : SUCCESS in 20m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d83187f4e26b43c7b54c8887c50831fb : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3f49b902dbd746edb2009acca6c712cc : SUCCESS in 13m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76bb9e0acc3a43799bc23f98ad46cba4 : SUCCESS in 13m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/915663facdb04efabfe4d446c6fd34d9 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e69dfce9582f4672b4962ddf018ec8c6 : FAILURE in 11m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5cf6b685c73b43e38db28f4ff1f92dcd : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d3326a123cf42658a85d234f7fa71dd : SUCCESS in 50m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21543b4a2414468583039161e7eace1c : SUCCESS in 30m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92460645ad344381a7f00ddce640a984 : SUCCESS in 46m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cad943a4ddd46e1949a4fb393c58137 : SUCCESS in 19m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38db1236d97e43df9555092f835966ac : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d1ab4708180c4553b38e202d0f1dcb6f : SUCCESS in 24m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af1b8516d1694156baf6661996b2e09b : SUCCESS in 48m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/497b472a46eb4985a60a4e631b00324e : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":8},{"id":"35a20ae148d4543ab76046a480b60494f9d6b9b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-21 18:28:56.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"ed376503266c2adca61fa60c5680c735c1660b45","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-21 18:29:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"952417cf17a34c07fdc57a3216de9d1fc5cefd45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 20:09:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be3fb5a8889d4b1594553a255066cdd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cccb8ba4d3e4ef9bcf52b28761d127b : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/8a2fdbc73b704ea1882eaf90e7183a37 : SUCCESS in 1h 05m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40f89882387246f7ab59f2d27b4684f5 : SUCCESS in 1h 08m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e66db185e8a74c6e80fd3915a7c85b15 : SUCCESS in 40m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1f0746d873f4caaa94192fe052e40d6 : SUCCESS in 50m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/952deba586a14e1ea2283b52e9b0e5a0 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12738a6725a44c88936f54f3bfc87a7b : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b888c6ee2814a9fae96680e0df73c57 : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/973a976622c8495c81c07b9a38cdbe49 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/edb40fff42e2444281406afcbeba59d6 : SUCCESS in 20m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/329cf5501cac486c891e56df9c033698 : SUCCESS in 10m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/39ef5fb9781349e2bbfa2d27c55b4c01 : SUCCESS in 26m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a948c2a327934e558396320114e62e18 : SUCCESS in 12m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b46c022431c34ab3975a63c00ac2576e : SUCCESS in 15m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/65af8cf66fd74ea185d6f1669931d645 : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e07726be9ad2428a9f712b1a2f7efb9b : SUCCESS in 9m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2ccb4f444d3845a0bd7b4fc3302f5059 : FAILURE in 14m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb7d1f5282874311985cd60408ae782b : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/44ce3b3b11e0494fa4b8f7526ccf82b3 : SUCCESS in 1h 01m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/769731c7e5354d78857271d21974c636 : SUCCESS in 34m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c47b65f5ad704c178ba4e77b445c699b : SUCCESS in 54m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d43bf623e4924de5bba5593edc011bc5 : SUCCESS in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb9b0b6f7be44c5d9c76987fc123995d : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6569a9203f514df7ba66cb65d97bc82c : SUCCESS in 23m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa5c00d9ba12454a97d327701aef23ab : SUCCESS in 57m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90c5784be2184d8eae06828ca33c8b77 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":9},{"id":"ea71295832a7b34f07e4104ecfc2a807ad99a70c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 22:06:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4cea837e72c453391406dec0bb8fb60\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/59e652c0d3b846e1b0d3821c7200ab08 : SUCCESS in 1h 35m 02s","accounts_in_message":[],"_revision_number":9},{"id":"fd9ab4918b197f8f1412a969805b06cfe70d8a63","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-22 17:02:02.000000000","message":"Patch Set 9: Code-Review+1\n\n(14 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1dd654bcd4ef95aecd182f884913f28cd899c993","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-22 18:36:16.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8a651faac1f88a0a498ed6b65e774fdaf3bb3b1f","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-27 23:54:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"466f456ed8fb483342d4f5cccb92e43dfd74205d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-31 20:24:33.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5ec463ceb30c64511d91598f6e05c7efcac59798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-04 17:48:57.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"e039f2977c442d7e5d49039522a7af0b9dfa8dc9","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-04 17:51:20.000000000","message":"Patch Set 10:\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4428e9e7f57348e3ecf9a7091393a5ae52b621b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 19:10:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/962f6baa305b4577a4bb4acd7e92227b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3543cff6020946068bd233787d5c557e : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/b60713084afa4de58b63b73504e11907 : SUCCESS in 55m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c964c6c995f845b9ae8cbacd9d622ba2 : SUCCESS in 44m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/925d928c550042b4b1010f129c2ad621 : SUCCESS in 50m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5d6d64848704090bec4d756a4e7d176 : SUCCESS in 1h 03m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddabb44eec22401d9c0c589aa35faf0a : SUCCESS in 23m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3871cf39885f4bf2b9075706e25c3c97 : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8464d68023094d0eb620e964f14071a9 : SUCCESS in 17m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/306270e09ce04c9587bec3e4cbc70f9e : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3f5af947d4314cdf9a94057bcbf8f31c : SUCCESS in 19m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5e935c027a0740d78de32899928c9579 : SUCCESS in 9m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1a3afb551c2d4845ae54a8a7f0ffea9c : SUCCESS in 13m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/100f2944c06542ddb96aea712f2a7fcb : SUCCESS in 19m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7bc48a3e5aaf4aec855c65f518659e8e : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/784dbd669da340f396203d6dff3a2c55 : SUCCESS in 14m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/49868e9aa15341f993368bb7f661e525 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3f6983d87ba846af843cb35f9063b84e : FAILURE in 20m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d603e4c8331417cb3ec2988bdeff5a0 : SUCCESS in 8m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e70fb7965d15478d8e564a50a5bfefe0 : SUCCESS in 57m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/60e6064d832d497b973b042114ca1154 : SUCCESS in 52m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f7f255f273944384bab610d0ba93937f : SUCCESS in 48m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc1a4ccb21454d5d922ee83e8e9c2885 : SUCCESS in 22m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef848092475c409f83eaf33263d6e844 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36140b8bb7104b6a9e78e2e6b317229d : SUCCESS in 24m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8aefddc608c04df8a3c3e320a1dd52ac : SUCCESS in 41m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca2b3bb90b70490696af8de8aa054863 : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":10},{"id":"9a576767aca75b406e0fba296267a7ae42435b1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 19:22:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b3a7fcf684a4267aafbcf74385552c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa2567d23024498bb652f11fe844bacc : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":10},{"id":"d0567c2c6a5c981fc102a498478a116cd5cfdb4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-04 21:45:00.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"ae4e1b83070edc5e60fcdd6ce4ae07275cc168e5","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-04 21:51:11.000000000","message":"Patch Set 11:\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"34a41224b81fcfe001aa255855c0639ec49cecb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 23:31:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71c08d9fc3a54decb432f2465b13ecdb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6dbd812d5047cc8013952549a1b027 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/48e0401ff3a840b3a8414d302ae3fcd4 : SUCCESS in 38m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8808f6f0a1cc47bb867239a016910a59 : SUCCESS in 50m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ca6542680f148c7b81cb1c2a5ec9d36 : SUCCESS in 41m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24a3cbdaad374789b6397a01534b046c : SUCCESS in 1h 40m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79e384fe233846c6a870cce41b6af804 : SUCCESS in 23m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9266895f8f947ed8738ef4a139af678 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b84e84502c14b40a309d785553fb890 : SUCCESS in 21m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/609a0522ca984befb57337e53017edd8 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a7bb7bfe8154a39a17ab01b5cadc86f : SUCCESS in 20m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b2c2a5fb419e4f4fae572cbac0e43a3c : SUCCESS in 20m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/06eb6b0fd2124d51a416a0abca228349 : SUCCESS in 32m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/12722c51b9564118a06cbb2e0046d0b9 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cbaf6087c1124ee0bb8895acecdc95f4 : SUCCESS in 14m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/46f16a286a27421aa0486d27dba64f97 : SUCCESS in 15m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f907eeeb437442c8eb16076297cfa26 : SUCCESS in 7m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81e91805c0ba4ca4942c4532cd8cf1d4 : FAILURE in 14m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac79aa7d45224b3f9ce829fa912c1e6c : SUCCESS in 7m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/32a65042946f4e6293b2f8fc01454e00 : SUCCESS in 55m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fed603ad99f14702bbf9c1074ca4f854 : SUCCESS in 50m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/98ae7cf0b1a84ea6b2d91f2da5e22368 : SUCCESS in 1h 01m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b01ce97129654c838b56e2f65fed883c : SUCCESS in 20m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b3af6c17b0d4f82a0ddd9e4d44d07b7 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/929b66ee765f4637b871f58894b34c20 : SUCCESS in 18m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99dbd05b35bc4d3d825a8a3db2688d75 : SUCCESS in 1h 09m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8ac630abac040dc9ce6546869b3c766 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":11},{"id":"061a3cb88574b34108b806de23515012463e338d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 00:59:05.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ea1845b2fc748ddaa1582aa11c39291\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d30a04f31be2404cb80394e8ce329977 : TIMED_OUT in 3h 01m 02s","accounts_in_message":[],"_revision_number":11},{"id":"155b9ca744db3aa0ffea4469dc4f8737ee81494b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-08 02:01:05.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"fd9628bb9905a7ff18bc8bf0a5eecdf7e4ec85c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 03:59:43.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a97b0a3695a44982a8ad205d7f6a003d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44be970a79cc43e99a623d7c4edba6a7 : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/55fd3b0e0fa24e3c8446d439ffb2b97a : SUCCESS in 1h 02m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/702f9b4177d34d43be624435b4deaabb : SUCCESS in 57m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eeef1c5cf852429cb559898b5a40ecfd : SUCCESS in 53m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/532ba96110834f02af0f13e268c72768 : SUCCESS in 1h 04m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81cfa2b485b548d98d25f8e549a731c6 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38e8d3ee9f1c4f4bb8c3ca44d533ec81 : SUCCESS in 22m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9da64f52db7145fc8abe40e6ac5b6bc3 : SUCCESS in 18m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e8b5c4a9b74455a80f95237cabb6964 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d2eff7293d034616b4ff109bd7d12e4a : SUCCESS in 23m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3b83ca8c64554d888be7e2006d4b7114 : SUCCESS in 28m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/354d25dbd7594cbcb2b094076af3ef38 : SUCCESS in 26m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/13e48a302db84fa4ae56b5afac1150de : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c019555f1fdc46c5ab5eb0fff4f147a5 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9251866ffff94b1cba49268e6976a561 : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aacd9dfa54974002b2d59735b4982f61 : SUCCESS in 11m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9cbb359735de417eb118395b34b97a14 : FAILURE in 14m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f708f7cd94d04e83a1dcd98145334767 : SUCCESS in 7m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8248e1798494feca1dde0892bcad734 : SUCCESS in 51m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef339fb169724e978662fbda5aacdba4 : SUCCESS in 50m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5aaa424e6b5a4bc19f2e286502f9b25d : SUCCESS in 36m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ef8b0e9836d4166b58081735d96784e : SUCCESS in 25m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c32836964e4b1db27cd71f1ab622d6 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad9101a4a4634105800ab483f91212f2 : SUCCESS in 25m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a8bc61631774730a2f217cbee3635b3 : SUCCESS in 43m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/722b8603ca464048b84cd6183d6fb9cf : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":12},{"id":"73e25a405920f31c37137591ea03b05f4d878b13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 05:21:07.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59faecb5a55a4d32a7c44e20b0a5b357\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ba36e772caa4338a94535bbb3c97e3d : SUCCESS in 2h 57m 19s","accounts_in_message":[],"_revision_number":12},{"id":"c59d66e1260a4fa2bafae3882027a7ae6b80c2e4","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-11 03:52:45.000000000","message":"Patch Set 12: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":12},{"id":"4da2285285a98902608fc593af29be5f919a543a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-18 19:17:44.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"3bdb8e39b5327aabaa2c551cdf4cb03ae1eb7416","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 20:44:04.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd11f30775964914b170a4e38d65d21d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5ff4917dc9df45929727c49c35a1ca87 : SUCCESS in 1h 21m 58s","accounts_in_message":[],"_revision_number":13},{"id":"db090cbababd181eca0d5b8d53d0087c0c83a55d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 20:49:25.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f832dc3f7c1842b8a33c4d41f204158f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105aa189778941279557249e2e3316b0 : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/756cd69c6a73432cb373811673d8f49f : SUCCESS in 53m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e938464e3e1a4aa0b5fdaec7dcbc0faf : SUCCESS in 54m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20ce964788b5481d84a1e820a1edd111 : SUCCESS in 44m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0ad5152f09e4f988dab619c3d18a432 : SUCCESS in 52m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4155ade1c134c9d88490261b3219823 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce6a4ec696f947c982c48a43e6d4abe5 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bd0524a7f5a4c7d90464ff52116456c : SUCCESS in 22m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f2224d441c2407f8afbccdea282503b : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/206381bdbab94a99a9a82517b58eb6b1 : SUCCESS in 19m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4ebc7fb0d0c439e86a4295f27c6d608 : SUCCESS in 19m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d5fb645478734be797a7b9a4374e26b4 : SUCCESS in 32m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/665fd8dd5a9841d9afb29b3645d603c9 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/affb515c183e47449fc04777ebbef7b0 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b867a3136ab54ce782a2875a405d60d3 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/414f8ca0006747f0b4b19f1393c7eb49 : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c57e8c0413ab49d38145c41565b9227c : SUCCESS in 15m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/21bb7f356753422a882ff03940344be9 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/66dfcb14b4cc4d6eb6815890b2c2a458 : SUCCESS in 55m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d96d06cd22945ca9ef942c9e73414e1 : SUCCESS in 49m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/73c44800bc5c4f39945e382bc42dd9a6 : SUCCESS in 50m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab49e1b3d9a64bbd9577ef8b3fb968f3 : SUCCESS in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7dd038c3fe840b888b72fa6de4945eb : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df14e6446a1343bcb7b5f3caecbd4c90 : SUCCESS in 18m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b89b256c89049959c2174192f16e07b : SUCCESS in 48m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc12676f193d46aba88dcefc6af81a50 : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":13},{"id":"53b0f0476f48ca2ad18f9a0595bbf46f49f30034","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 21:43:04.000000000","message":"Patch Set 14: Patch Set 13 was rebased on behalf of \u003cGERRIT_ACCOUNT_34930\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"_revision_number":14},{"id":"71f6357d0ccd0f2ea1994918c2b32ce226d0751b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 21:51:22.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f75fca319d4a9a93aec814a6b75965\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8284140c71d34c9795fbe1ee57f35e6c : FAILURE in 6m 20s","accounts_in_message":[],"_revision_number":14},{"id":"384d09faba2045355dd0e0316d0f92586c48611c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 23:03:02.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/418a46d711014943b30ebaaf53905a0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e536b7e6abe4f5b8ca2e7f0f45c3b2c : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/5be89ed349c643debe6d8cc728e4d84f : SUCCESS in 53m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8dcdbeb762554abd8c213e88140fabdd : SUCCESS in 57m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d9f6d1de0c3a46a59670d0720a79d206 : SUCCESS in 48m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/516be28a36cb4d82af3d0ed2037d71c9 : SUCCESS in 1h 00m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/262c44c612d340faa3c3dbaaf3a6d788 : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f055c2c9f9c24f7f841df321d0473382 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/662fa9375ae04a79ac8c28a73bbe5cbe : SUCCESS in 21m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f43ca2baae84465b88933c3c9f359bf0 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c90836303e2448ecab7903ce4be56825 : FAILURE in 2m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a651a79cdee84972aa03b270a3c0afcd : FAILURE in 3m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4aff5ed3c0b847a3b08e16b9d9363f8a : FAILURE in 7m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bcc3f4e2590946c6861f32ebe1e3ffbf : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1e8483545fb74c5a9e288f091423fd2b : SUCCESS in 13m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/acd8931a13dc4a3bb18d7abb5ba86aa9 : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20ca993ebec24ffda26d8c2a98bd4668 : SUCCESS in 3m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d09b55ddd46040bfbfbcd0f7702c72cb : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1324cd11999f404caa404a25fb8f4912 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e0a28875d4e40f5bb7c8a04add74681 : SUCCESS in 51m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f290c5386a34d94b2a1da798bef9c86 : SUCCESS in 45m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/01952a5e73cb418e816460f02d4f98d8 : SUCCESS in 33m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/470aeac9ce4e4dc68124da15f163d603 : FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c5f348e19294dd9b86fbd2ff46b2b4a : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/491d370bfd8e4748ac3212dd9f4ae107 : SUCCESS in 19m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67a402ef31454855be655c6caf440045 : SUCCESS in 26m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/741b621d1e054ab3a5f3acf5ba295d6a : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":14},{"id":"d450ebeaf631e6e96f37c01b2d27b965e1fa7d19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 23:11:09.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"d4c4cb3457b944c6519a77faf2d44fdd07d6575f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 23:13:32.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"54f2113b393e1b65bcb27506db57805663f6e9cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 23:18:47.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"4e355b933ab1591ab0425a6928f7db30b78b49d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 23:29:10.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af0681393afe408f94a962b25e55ec70\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f8e122baed742d4b4c5e0f45b7329be : FAILURE in 8m 06s","accounts_in_message":[],"_revision_number":17},{"id":"5c5be27b949823257a76f142ecfe792a88ba3c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 00:20:25.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12732f30c51e47cdbd910e5248348bb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1add34927be54141a6060ab254116e4a : SUCCESS in 4m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/ff340bea651b4b5db51acf6883474e62 : SUCCESS in 25m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/028d189db0674d978c6b043c6fc6883e : SUCCESS in 39m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49217b05c3a64d6dae8ccef06668290f : SUCCESS in 44m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51b574e777f44066b26e8ccd731e8ab0 : SUCCESS in 45m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de273a1ec0384bfcbc3e03d6894db549 : SUCCESS in 15m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43e9d2e29150411c8ea33d2f0bf679dc : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b1b4aa148d345319387a7fd67986c15 : SUCCESS in 19m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e7920b7f1664a449bcf0bd9938d485a : SUCCESS in 4m 04s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1b015cb92c344ca826829e9a662c1ee : FAILURE in 3m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/18dc610070b34cedbf04d6477707fb6d : FAILURE in 4m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/54487a0c5114452d899ef7e5922c44a5 : FAILURE in 7m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d744b192d5774fa2ac49ed2626c2890a : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d6a9d962e9af49b09fa4b2ccc802c5f5 : SUCCESS in 9m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/71aeece7013b465d9ea722a564b8b740 : SUCCESS in 7m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/28934e59af4b4b68a7063408b739cb74 : SUCCESS in 5m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4aca6b8311ca452288e314f173ee9b82 : SUCCESS in 13m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d6d2ecbeb8224bc189f9d88c92bb4a3e : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3339658da2b64956b8a130c6f5f929e0 : SUCCESS in 53m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bfb804a9f0ab4f8d89f2294fc82f172a : SUCCESS in 36m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c00e1158acc4b9cbdedcf653e8e1c25 : SUCCESS in 23m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adc8d1b88007480b95192a343dc0c97f : FAILURE in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcae1265a361483aadedf5b859299565 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bae2e22ed0c546e6bacd99a997ba9c05 : SUCCESS in 15m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d93b5a020074c619b00e83f88fa7c82 : SUCCESS in 55m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/109b13c335cf4bb4bbf2f1b8d5456645 : SUCCESS in 27m 29s","accounts_in_message":[],"_revision_number":17},{"id":"411d7804852fcc7c83586b2e64293edc9a282c9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-11 09:53:56.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"3d32c45680d1daafe9ace2d989416dbff0a09f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 11:06:19.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cb5051a8d114defacbede39a988d1c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb86b543a5646c28de9a51fc3281e0c : SUCCESS in 4m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/d5e197d968114cdd8766cbb60c505667 : SUCCESS in 53m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf621076804b4f868f722c5eb7fc7375 : SUCCESS in 48m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51c2c55468f3407d88b6e00c6c30166f : SUCCESS in 51m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2070335dcbd94110a4b5fa925561440c : SUCCESS in 57m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff4d75ed5d4446eebd0ccf168e76dffd : SUCCESS in 9m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92fb7d19920745d7a63439c93d097ac2 : SUCCESS in 16m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f62e5ebfc1b44da48ab28477600ec05c : SUCCESS in 20m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2afb5ff01fc9424c817664136641d1cd : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/20081ea3129a4abb808161c1d3e8f28b : SUCCESS in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4ef75afd2a404fa78d612d179e0a4ba1 : SUCCESS in 8m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8af14351a3eb4a32ad508af8bc600415 : SUCCESS in 26m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dd53fd3c69814564abe10d0c01dc3257 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4ef31acdb31a43d38b5755ae02b4a1f7 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/10ec86b5ee844368a54ae7d8f33bff15 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7da47b995904a8882434375ae37122d : SUCCESS in 5m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/847200824bc34d55b1605d6c4f07cd90 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6eb2a482c1474cbea4bce45c1577b8c7 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/31227f25169349159c3f0b635e5a2e27 : SUCCESS in 52m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d457d3d0b4fc4bd48a327e3c10286273 : SUCCESS in 36m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aef598ce5f9744019711dd138269ab03 : SUCCESS in 17m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6acbd5aa467943a3a64fade18e1c5eee : SUCCESS in 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b713455b6f3a4519a1ef7fe3df25c159 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ede399d2b7864e38ac32eb7025d88dba : SUCCESS in 20m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5a1964995a241f4bf8e3ab55f83215f : SUCCESS in 42m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/683b26eb9b0d47b38cca9abb0b9a911e : SUCCESS in 27m 50s","accounts_in_message":[],"_revision_number":18},{"id":"7d4fb1402b6bfcd38102408dea3736debb19a902","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 11:19:28.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d638731b034b6487b858efcea87c0b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/218bf128427c4654b91b27d057d91fd2 : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":18},{"id":"94d629cdf9392930bf21a234b32d35ab669c6c9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-14 18:38:18.000000000","message":"Patch Set 19: Patch Set 18 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":19},{"id":"f4277e7f43eaa2db76d8837a7d3ca00655e8081b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 19:51:27.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3515048f3224479a175a3b5a7cbebd6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/93db7aee069b41c1a1b176770351a5f4 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":19},{"id":"364d8a03a2b204d2c3f12b1dd2b38129e3913589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 21:48:52.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dea1114d4c3f46e3879f2907d644a499\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a15da81b1c43849a8aff295b3b9863 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/7ee20765ef7a4c9db3a176dbc1c7bd24 : SUCCESS in 54m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6c56f7390ff415495a7813bb8fe2699 : TIMED_OUT in 3h 03m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7315d4240c554bd5a57a732e5e939032 : SUCCESS in 29m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/853fef1f47d244bca081ac4b9ae7793b : SUCCESS in 38m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42c8326a06444fe5a2c1b410de9e5792 : SUCCESS in 12m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5259f91df0a4427875096c3c0353e42 : SUCCESS in 24m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b69704fdeb5e460587fff0ec2ef1426f : SUCCESS in 14m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/884c2a02c8c4418ab13f173514ad09bf : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/030993dc84644b81966bff2a2ace4039 : SUCCESS in 21m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/415a091fd7be4c9a9d712d7a280d1065 : SUCCESS in 24m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/31ee6e416e73451cae78adecf3cd1791 : SUCCESS in 27m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ae022610afb94c8893909922d7329995 : SUCCESS in 28m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b906eb699b0849a794ff49b552d86886 : SUCCESS in 13m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/29e333f75fe240068815df9fb9d964a0 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/29e30efc6e0c4aa894c76d8d67cf475b : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/33ece1a44460407b84d86c3596f226c8 : SUCCESS in 6m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75342b32a7f549ae90efe49951da4ada : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/abbf74d38f4f409697990b093f3438f0 : SUCCESS in 3m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eed380ffa24c416eb7a77100a86fbe4c : SUCCESS in 49m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6b74adbe529b4b57b06a71743714a65b : SUCCESS in 34m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e8b14c834af644c3b75e58a69b5c42c7 : SUCCESS in 34m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9922530b48a84a8cbb945e291f724fd8 : SUCCESS in 16m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6b0072fd1b477384148a366fce006d : SUCCESS in 3m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90a41b07b1e443808e4ce5e47f06ec90 : SUCCESS in 21m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83e78ebc771d4b6d8bb85553d867438f : SUCCESS in 28m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93ab78cb3fee440a9bc2f9efa1bbe8b3 : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":19},{"id":"9fbc255873c6d543c1772a5b7d2a57284cd2a8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 01:24:44.000000000","message":"Patch Set 20: Patch Set 19 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":20},{"id":"c5d2603e6c773e75c176183ada477bdc2478f757","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 02:33:15.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c33c55760df4364a5daa89c98ed9556\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e741cd2d5b34323bb900052511938b7 : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":20},{"id":"8fae0668299123e3c8950f3739385181e24cd6e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 02:36:23.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/203ed833a6f5437da8a5f31e05daeb46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9903125a41684fbb880c1b896d6eb243 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/b2fcd0790d084b00a45a40be26a66885 : SUCCESS in 58m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84adc2ef98094460bbc282fccc254e7a : FAILURE in 51m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84652e7778da41e08507b946404feb5e : SUCCESS in 55m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa5dea8bf1484eeca6d3b0f304fabf0d : SUCCESS in 54m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e240556c91f041b5aed9f3979a818fbc : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9317a17c6084d29b0957581c399c66a : SUCCESS in 14m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/945ed781c0ea46878ebe77f368ff7672 : SUCCESS in 20m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53ff71650fda493a935e3b86e45b32c2 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/693f60eddc054d4184e029b8de4023f7 : SUCCESS in 20m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/21f355e40eff4db7add3092188c57b10 : SUCCESS in 21m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4db9ff3b60df463ab747e985d0b5f522 : SUCCESS in 24m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/658be9223e9344e7bf667ca0d24b5e29 : SUCCESS in 27m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/470611eac41a4283bd726fa63a5b46b1 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f828f06c5678485eb42b2daac9b0e4dd : SUCCESS in 19m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f4a0d4ced5a749fca1bda787c99459e8 : SUCCESS in 7m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b2515fd92b5d4a84b4c3a69e986117d3 : SUCCESS in 5m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e401fd80f8674498919aa7c9f4e20fae : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/52c3473e98fa468d9b2db8bdf9ed45a0 : SUCCESS in 7m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/44a513cba980465fa07925a5a76cb87e : SUCCESS in 1h 01m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e02db08a885e4539815f131baf2dbee3 : SUCCESS in 41m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/004a77cd1d384aa79ebe25b9de9a6e02 : SUCCESS in 46m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ae7c374a8c344169eae4e5eb26ed4d3 : SUCCESS in 26m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6604e4830fd245ff9f90b59545a8bc21 : SUCCESS in 3m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec673d6da9a246c989081d8818d1ac2d : SUCCESS in 25m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/905fcd51cffe40a3bc3efe891fb79306 : SUCCESS in 32m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f03e139bac0478c98b552f00a1aa435 : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":20},{"id":"125ca4838a92117faf16d2d4f0b75c0f57b3a196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 14:27:19.000000000","message":"Patch Set 21: Patch Set 20 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":21},{"id":"a75e9f9a39df66cfc8b5ab3ed6bf0c7374ba2b33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 15:56:54.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"b9436088f8846da36d0732d9ce00dcb5990f9fcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-20 16:48:05.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"3317f4f4ee7aec678a709c6b1721ca7d232a013f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:04:27.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a398f45c06cf4431b1cde72a1cc4dfd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc280c3553e4ceda058cda855cae1a5 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/4bfbc4c53bf44ff68c8394131f6b7605 : SUCCESS in 56m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1b8821fd97cc492daec413e929b11cc6 : SUCCESS in 59m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8748fe7164c94357a2f009564a1bf26a : SUCCESS in 46m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0fa20350b8a49e9b5757204933d9efb : SUCCESS in 55m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ae634905c5544d98882b5cef3cf6474 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4e2fe1e32514772a6a758d0bf406477 : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23f9e6aa42224ec384fca6cac826a250 : SUCCESS in 19m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e79d1bc7992b46c69c97329e1f55b307 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/262eb2146ac94963aac16c665363b1e0 : SUCCESS in 21m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c151a0cbaa154579bb0f413a7a00314b : SUCCESS in 21m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/47b82c04c0dd4ed0817ef3d643fa130a : SUCCESS in 10m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/754e9b19a45e4ba98ef82ee1836f47a6 : SUCCESS in 13m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e16f97533525457ea512b355f3c42463 : SUCCESS in 21m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cba22422412545a988efaee365e7ecfa : SUCCESS in 13m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/69987576101c415a9ef2b77dfcb5dc67 : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6a4f238f0c664515864da23d86cfa428 : SUCCESS in 3m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/25cfe34e1b814759b2f84193b6e8abf5 : SUCCESS in 7m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6b4ad7adc56f4ed7a7ffd7b4af71c1c9 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/abd028468d9d4ef9953fa2f2f9fbb2b7 : SUCCESS in 51m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7cb5e22a2ff549158e421ae1e020c624 : SUCCESS in 18m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9eb31f2905047a6984d0b66f884c7aa : SUCCESS in 18m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad47031cb2ec4369a3f92e1f840826f5 : SUCCESS in 23m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793869e828f94524b2458b2bbc17f206 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b2ce6b16845455abf98391d45fbfdf3 : SUCCESS in 18m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5017fc9799ad4f2297e01d7f42be7801 : SUCCESS in 44m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc160d50b16143bba02a0638d00d83de : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":23},{"id":"2b2db0c2d637f73a542b63e2674ae1799bcec39f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:26:27.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f7e22cb2c954c6e902206e612eed8b4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8fe19bcf1aff4d0090314cf10ab94285 : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":23},{"id":"d407b514d3a87d22d5658638a0a62c4caeae4567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-14 18:58:34.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"e7ac18bb322e68320f1687f0043aca5c009f764c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 19:45:26.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6fb8546164a48579ebe9defae42c958\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1bbb7b45cc8a40d09b44d69afd1ced87 : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":24},{"id":"a8d3749e50115538f618b00a0ba49369547f7c47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 20:10:48.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d590da99fb4d10a82d665a1ac5e95e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4368fcfd3b5c44a6bcd9d01645d9661b : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/a5b103b7a47d491c9023e51c5b148b41 : SUCCESS in 1h 02m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b4b65509a7a481dbe1726ec4e0719c5 : SUCCESS in 1h 03m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d1627d95a31a4a798278b762034fad1f : SUCCESS in 44m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b55ad0fd5444ee9abf04201d010ed01 : SUCCESS in 54m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e65b3ece107a4b849d77ab627c8f3708 : SUCCESS in 16m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60219bac86d8414e97df12bb43cf791f : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e42c4f4f8d304e35abe4129e2c59f35e : SUCCESS in 21m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1a06b2afe0b4f47ae295f41cd81f779 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e17820d4f48f41fe857faae89f2bbfc1 : SUCCESS in 21m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a388f4dff87e4dd6a0cf31182ccdb922 : SUCCESS in 17m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0df1a610d007466e961201e0084d5edd : SUCCESS in 18m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/af69b4f54a104f01935070d93faf942d : SUCCESS in 27m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/674c67450d3c45c2802b1ac717c70e6a : SUCCESS in 12m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fe9a19b2064b43e9b360bd16a6a8f040 : SUCCESS in 12m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d51ef29538c04e0f8d3c34e17d454cfa : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f3f4914e53a4ec2b4ded30e3cbeb180 : SUCCESS in 6m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f7542293655412686a9311dbeb42a3f : SUCCESS in 13m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bdf5e9e6f73d47178e909fc24316a3f4 : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b58de6977a04ccfae03aca8277308a6 : SUCCESS in 47m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa9214b0502840c196321a4409a279aa : SUCCESS in 39m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c6838f9b94b545b18b60df2ec19137e4 : SUCCESS in 39m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6aeafb57ec614d22a6e5d71a312eca1b : SUCCESS in 17m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0074db7ce4845559f9ad2ffbbcdaa0b : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a343ded1c96466f8a07b0217ce57e82 : SUCCESS in 15m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a4ee9ba4aa44a2d9d94cc84ef779e3b : SUCCESS in 50m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c45501c114954f9cb01589c1eda6dde4 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":24},{"id":"c964f77e0d05230d4adb32c0a3efa5e08be66089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-20 05:28:06.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"6a550ee14e85f6e86c529ada42d726bbab5e2ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 06:51:42.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c0355f46744bc9a59ca1baa2a06b65\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afba1c2750554475b3a3dcfb885351f5 : SUCCESS in 11m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/3bb4d5d915e54e19aa73023c1a16cc1e : SUCCESS in 28m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a597bbf982e14a9e81610d8129901ba3 : SUCCESS in 33m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a79b756e06344da78430a49339a8b28d : SUCCESS in 54m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cae6cbe7905a480ab9b4964419d7c3e9 : SUCCESS in 1h 08m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a6f9e1d367b47fa9c6bd86d657275cb : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34fb7b4bcfbb40bc85083221c0654b5c : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44d08134e2b6468fa703f67a622d00f1 : SUCCESS in 24m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bffdba7925c847778c3b7bf090ef1a10 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f7d078670a124ed1bff0b582de44bed0 : SUCCESS in 22m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b3ae98ae9e8a434893a00276aefe989b : SUCCESS in 20m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/17a3d2c21818409d9314c00fb7b1580d : SUCCESS in 30m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/44ba7576739242d8b107f0c4e738740a : SUCCESS in 21m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/27b5ab89d1d04ed69f033a0bc964515a : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f063176bbeb743409b4228ae893bb3c2 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8e876b0d08a841138ec35eeedf8a66c6 : SUCCESS in 12m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/54b2ceaa6878433ab3ecaa6e9a4a7680 : SUCCESS in 9m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c32db53e412644dea179062730a3a41f : SUCCESS in 8m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/09d6945ae5d34416ae20abfe2ace46fd : SUCCESS in 8m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/54d9fc9be17542388062180f72566aad : SUCCESS in 53m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0dbf6c87908444c808ebc7dabbae15c : SUCCESS in 38m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b37f3d6dbeb429aa359a21c92100472 : SUCCESS in 45m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d2a702af18b4157a5cb9086999c6142 : SUCCESS in 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d18bb67e6cb4a17a7b7c553a2c1aace : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a658aa4669034bcd86077861d78fa41e : SUCCESS in 16m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76174f96eb434688b3f21f47f7ee45ee : SUCCESS in 57m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c65ddfea4afb4d3dae61d83791293f93 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":25},{"id":"b363bdd1014d4a4188f087a2eb927435649e2ac0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 07:04:10.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60365346fc444bbab25fd91f51598a96\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/394f5e2ca5c24838b1a826ab98f6b145 : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":25},{"id":"3a10771ec063d6defc0136580ead9adb9396ec8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-05 19:46:54.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"556ef583c796834b36e360c178839ae350868092","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 20:59:03.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74524b7b3b334239882c28ece2c18a39\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d1b9532007d461d8a53469da2f1a3fe : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":26},{"id":"b3d9b696cf58b81d4c685dee02ef2541f52ec41e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 21:01:26.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d7ac2301bca4749af9e934f0a841c76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e35704171e43b0912049bb9ecf42e6 : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/68f7b1c16c804d1ca4f0652bdf3694e6 : SUCCESS in 58m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d49a9155707f4537bb64dc0eb2716e6f : SUCCESS in 1h 02m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1119af9515bd463fbcd686cc6456f255 : SUCCESS in 53m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b337ad165a184870800f84e91799bc29 : SUCCESS in 1h 03m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a598375df5b446aa0b357139249ee13 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ea4d55d379b4556861a9a2df5851e76 : SUCCESS in 10m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cfcc833ebbdf4d119636aabfee3757a2 : SUCCESS in 14m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bb3bbabd27ae49e39e7eb27847eb27e4 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/01447880d262423fa3f2248834761953 : SUCCESS in 20m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/615bd26a4334408e9fd5232dfea352ee : SUCCESS in 8m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d538e4ab04ed45bebace774617653bd9 : SUCCESS in 30m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/84be5bc51ef5499f8a423acdeb63073d : SUCCESS in 11m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c0aa045ce4f84375a08ffa813f0a3308 : SUCCESS in 8m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9eb3ad3778aa40c6beab86d91b1e2ea6 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8d84617266d94c398c4c430e75682408 : SUCCESS in 14m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c9726b00e5e4b6e972b24ebc852a8ce : SUCCESS in 6m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a1c8be3202dd4df582c2fcecccb21653 : SUCCESS in 8m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d8dee63e1bb844639ad8f32e6d7503de : SUCCESS in 4m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/79b0e476bd2b4373a76a0dc4a7cc6496 : SUCCESS in 49m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efbb220fb1434abb8a70102e5fb429f2 : SUCCESS in 22m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a0e6eb5fa1747c99c3e446572d8eb38 : SUCCESS in 41m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c73f92b718b4a06afeb77ca5cf34e9d : SUCCESS in 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d872289ba1574c259fe302e72a550426 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef12c3f2c54248dd8e174dcd1aa620bf : SUCCESS in 18m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce411fd590e94b018bdf20d52caf73a0 : SUCCESS in 48m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7cd18d4650b4e6389c33dd3c424ad6d : SUCCESS in 29m 22s","accounts_in_message":[],"_revision_number":26},{"id":"97c10d333db34cc5e6aa765b7fe8dbf5ed75d152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-10 05:18:45.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"ae9e336c0789a7cf9d51e78c57ca73bacb3e91a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 06:09:07.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18eb9ca473b843ed966ef600a4290d86\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/70c18db7db9541c9bb65d4327eca340c : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":27},{"id":"21fc7e36c020324b8dd41d6622f9950e34aaddcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 06:37:37.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f76f1d9cd64897a1a795b84b028d4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56137881511742559c69d990867c05eb : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/e1d003e1951c4baba9a034b80a631920 : SUCCESS in 53m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d51d94507368492ebdf3294ca69e4ea7 : SUCCESS in 58m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4145b2d84bb0457c86cee7a46ee4bb8d : SUCCESS in 54m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30f71e18f0a14e08a483021c845ea93d : SUCCESS in 1h 09m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e272d6c915af4353a36ae6152a51ed4e : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9326bb79e304a1bb4042bcd8533cb15 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b70b4357f1a4a8a9ff32c5a431fba83 : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e72f4689f69d43f1aeeb6f3ba759aff4 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bd80fe7a420e450bba84bf38a765d8c4 : SUCCESS in 25m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aaac526d8132408b8c428aa0baf1c5d2 : SUCCESS in 18m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f6b7fca359fa471985731a9790a5e5c6 : SUCCESS in 38m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0070bdde318f4e958c26e4524220a9be : SUCCESS in 24m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4dad2350b284492c89c070fc31f48aec : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/61975d4896694012989f9f97bb1974b8 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/38e91ee8c5b84120b8101f7e1536877f : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fadf0f2590a747beb4363eefa6c88285 : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a28426a3eb214efea80c52a70ba6578d : SUCCESS in 13m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/53fab2ca73cc49fca8cdd227a0bc970f : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a783fb004bc49b699b9f97891a54934 : SUCCESS in 54m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c10230a9e1054d928d3c3e2de79b37b2 : SUCCESS in 33m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/da68da88bf294981801a4e912ec70eac : SUCCESS in 49m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f0bfd75785b4c3987a1ccf1843902a5 : SUCCESS in 31m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86c890ca01e94d339e56278316750368 : SUCCESS in 3m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ff22173905d426a804a964b2e606422 : SUCCESS in 16m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f701d6ad23944d99bff454c3ea6f74f : SUCCESS in 57m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6570a4c484d546719336bd4f36d8ab88 : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":27},{"id":"60697f669270a7cfc7d8d565e08cc31bc4075eb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-02-10 04:43:10.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"5ffd97c72d1a6a6c616cfd063b30f2a2ec9a3c79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:40:00.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dbc4fe9b97143548c9be17a14a25c77\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c44ac8f7e7e64a0c9cc3c1a99f30951b : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":28},{"id":"3e4a5ce59369ddda43de5b467cfcedf506215c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:54:24.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c1ae2497735442391fe67a14e203d79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b22bf1c35b246a1b8fb632c3f5034e2 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/8e4cdc35a42e4bb1bb8ac710a39d0e44 : FAILURE in 12m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4fc743ba0b2947e880ddfdf1e4ed4c36 : FAILURE in 12m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/28f457bf0b7d4e85a8a947f1629f4cdc : SUCCESS in 41m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/678b7b9fa9054b5aa6b00ee924da8b79 : FAILURE in 1h 03m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35600140e2fd4855b6f5ead0347a30f4 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fccff9d85e7479f99f50a251b310fc3 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9211f40ace5479c9ceddff7e628cb01 : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa5f1476a6f647faad202f1eda1ace5c : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/06f64bb35fac4766a9785d0dc0479cc7 : SUCCESS in 18m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb47d8516f7f47178e77dd6c27c59ecf : SUCCESS in 22m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0f0b4d03adbb400fb89997fd4a4b49d4 : SUCCESS in 22m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d40d7a7a22f24d539f917d3c36395439 : SUCCESS in 11m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9dd14b7bee5a455cb3e3d4cb619cc71d : SUCCESS in 13m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d8a92cc383ca404f84a342f4e88c5f22 : SUCCESS in 7m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/11f9b27983d74e1e9a9e9386b9e270a4 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ce1186f915240a290074970ae508507 : SUCCESS in 3m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1814d1c0d5114205a29fa4636dcfe79a : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/70b8d8799690486c8e4788a18177f014 : SUCCESS in 10m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cb01fb74fc549c38b007b3c378524c9 : SUCCESS in 54m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9b1d915af8eb405f87c718a3483185f4 : SUCCESS in 23m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9351ffd378d34ad6b4f029c2cd63c9be : SUCCESS in 48m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/267b887223e94008a9a0391859ab2f9e : SUCCESS in 27m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b22efdf4ae44ee987acdd62da2a64a : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5b08e2d9b8b1446db9a161a5e624b69f : SUCCESS in 19m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f84676501b3e497982a7e89f1dbe5ac8 : FAILURE in 43m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49465d2a6bfc4e728078b75fab89c6b4 : SUCCESS in 33m 44s","accounts_in_message":[],"_revision_number":28},{"id":"c0fea60ae6ef5d0ec6c6f54926a0fc2429ea8dec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-03 05:39:12.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"8b7bcd95d72f97bff25b898326d910ac30593569","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:08:56.000000000","message":"Patch Set 29:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bb5ca062e7b4000b22cb88141d3e000\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa1603f79f64420e939a1864fc0bd699 : FAILURE in 27m 33s","accounts_in_message":[],"_revision_number":29},{"id":"fab2d91c5d49721eedb6ad90bea0237dadcfb3dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 07:00:27.000000000","message":"Patch Set 29: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc7c8ddeb62e4689951049af264e9669\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84ced3d82834647ac97c8221c94a2a4 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/7cfd2d5304684fedbacb06880fecf0e8 : SUCCESS in 1h 00m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e93c3eda9c8d431cba888674e2e8d7d2 : SUCCESS in 35m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de77c78c05fc4ba7b3ee97f183fc1f16 : SUCCESS in 53m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/560503af61c04f9a8fe87e67fede870b : SUCCESS in 1h 07m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77db330f8c9c4d5aa2abff875023a6b2 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/518003986ad44f099031ae7eb1799973 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b5879c135334d0e95d5000c25137375 : SUCCESS in 16m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b9bbcc5cebb4b57aa8d1a2e33bdb502 : FAILURE in 4m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0ea23426cf7e44d9abb49006abfc074d : FAILURE in 17m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36dc21cffd3b482292310ccdf0f28c90 : FAILURE in 21m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1c0eabe80bfb40c3863e5d6ebd02a851 : FAILURE in 19m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a226aee2390f4bbcbe5b54683210cf2d : FAILURE in 25m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/923b4ba9065b4a9fab03fc4a57db970d : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/35ac1e1d9b694fd78e0b13de5c6e5ca4 : SUCCESS in 15m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9a0f63f31cb94d23a3ece84ff863db0c : SUCCESS in 9m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/13581a7c2c21406ab2f06b1fcaf5534d : SUCCESS in 5m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f123bb737cd4104a86a218fad5f1b6c : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d5ec4a9f2934cb3b7d608403065f208 : SUCCESS in 5m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb7cd6bd32634522ba35d7d43b50ed80 : SUCCESS in 48m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31abd7038857415480a101a5ed78fccf : SUCCESS in 48m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1fc48ce6c75a45fcb542065265bee08c : SUCCESS in 35m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff38ec3679d5458ab1ce1fade07dd805 : FAILURE in 26m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6d065ae04d424f908ec2f6e804e1e3 : FAILURE in 2m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83cc1ce40ee04debab3019087eedc474 : SUCCESS in 10m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f9e6f6484ed455db99b922acacecade : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":29},{"id":"d230a5004524823f893431fbc3ca270605a48d42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-03 16:02:27.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"59e851b888394c1df4a2997a5f9eeea5c6f3c8c3","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-03-03 16:18:01.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"623e36678afad4309690b1a4a0159193d3d6206b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 16:37:58.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eafb6a0148b54a5d80cc95c841d9a662\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94eb87f9a9c14e93956b2edab31ed823 : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":30},{"id":"a5f77212efea5c488b5c229b33e6a0616fcb2e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 17:16:53.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d377e9c62a42caaafafc65d08c0023\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2dca6460dcc47e1a6edd4acc7e76ec9 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/824845a718e44b29981c641256795ddf : SUCCESS in 1h 01m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f35bc422ae84802be790bc6d294fc09 : SUCCESS in 57m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/929ebb357d6247b8ae8783cbfa765be7 : SUCCESS in 54m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6b5aeb447d54da89b0b8540a53e7acb : SUCCESS in 1h 07m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3cfe5476a9504b41b027228c3782278e : SUCCESS in 10m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5474b49793c34597b0fa72e8d7141f94 : SUCCESS in 10m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8277081969a41b6aeb7a7be1959ac41 : SUCCESS in 22m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1c3efcaf2774cfbb9509d55fc3d2f77 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0424ffb8024c47bda05ac50d064b36d9 : SUCCESS in 10m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11dc0620eb874b74be2596542f69609e : SUCCESS in 20m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/01e45f6bdea04a9e9345b393fa0b45dd : SUCCESS in 28m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c92e1e8238c146ca8ed4ade794edf7ac : SUCCESS in 31m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/54c707133fab442aa53a45aeb3c031bb : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/05a45494a350425e97553c3e577df148 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/17b2c75a9a18470789413048962e77fd : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e363aed03e114fbeb677d6fd940f1bbb : SUCCESS in 3m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/73a2b44dfb1c4571a3d63568606752ab : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c6887f4ca704103b1b8211b20678058 : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/acf1888c02f54320945dc0756ce915cc : SUCCESS in 53m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5880e0d40074c6197324563b21669d7 : SUCCESS in 51m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b149dcfe37c046b5890154b26fcc5959 : SUCCESS in 34m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5512056708eb4d4eb7eddcc81556f568 : SUCCESS in 27m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea209d8de03842b6bd6cb905be570e78 : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92274fd076414943b367013f9b473528 : SUCCESS in 10m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdb50b193eae4c2aacd9f60cb0a89d33 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":30},{"id":"666e0cb48f6162489a8213a3501cd781d7e72b7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-18 13:35:59.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"3e2fdee841d58506754204d7c3fb4086e518544a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:14:42.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/198e4c5d6d9e47e2b832260ff2edca82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f27e14d8069f430dbfd10ad95c447867 : SUCCESS in 10m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/14a5a739f27a4498bfb1de332c888d7e : SUCCESS in 1h 00m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/589caed414464790a811e086c83219cb : SUCCESS in 59m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d290e46ebcb467cbf04c977a4098915 : SUCCESS in 55m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc1f709550d646fabc3fe736444a9c61 : SUCCESS in 1h 04m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb3d73751d1b4472b9e6cea86f17ce04 : SUCCESS in 21m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5515fccc42014e03bb6aca478457dd2c : SUCCESS in 13m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/566e00b4be93422daf5d6fd1c0b3c7e1 : SUCCESS in 21m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18baefec02ce4bb69dffb0666c5eac32 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6378892c22734387a6d02447691f4798 : SUCCESS in 25m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/873d70a4b83f4c3fbbf41ef1659636b2 : SUCCESS in 21m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a9ae07fed162454ab2c013d568ea5d9e : SUCCESS in 24m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c85548680de14f8bb0073bb77f0a474a : SUCCESS in 20m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/320af7c569c54c779492856ee9d112d1 : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9dcc09f6c9d148cb91cf0693b1cf0e06 : SUCCESS in 14m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ccd731b7f2a146c8b6a2cd6d3e6e29a7 : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c7c26c73b86840c9b30fe0aa27457439 : SUCCESS in 6m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1df2ed15c52f4eadada1be0052343d7d : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/00f7149b7f9e4649a77eca850a2ef51b : SUCCESS in 5m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b25e863f27f470492c2fa0203f658fa : SUCCESS in 54m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd3b5756f26b43ee850a681160fcc05e : SUCCESS in 53m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0c6bff8d15e4fa98c58c55ee9cb7116 : SUCCESS in 53m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8e2cd91843944559343b8f972b9fb90 : SUCCESS in 36m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c3d07231074645b01dce881919195d : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ce3f7d9785c436ba70a62ebf878531e : SUCCESS in 20m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fda4043d4baf40f09275e18f3468808f : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":31},{"id":"9e1025f75bcea6144c8b7a5cb6b3ee8d21e060e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-18 15:18:47.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"3c0a4bf16145129a319d1c1bc61bcc618be952fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:49:19.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b29fa4c32374e1ab54b61c83fdb437c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac0508a22a347439330b53c3d0bf325 : SUCCESS in 6m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/4e8a73f1d9e442a7acd8c9da6dd8e3a2 : SUCCESS in 48m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/663fb022e1b1429f94346e9252938f23 : SUCCESS in 1h 16m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09f6e7f5bc334a9cac8f38394d85f577 : SUCCESS in 55m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4a8b5291eda46b59eeb2423b5869731 : SUCCESS in 1h 02m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/808c15adea034ed9897a937517d0c83d : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/215cfe7055af4ea5be8cc5e7223efeff : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bc750006acd4392b15f01287a7fc33f : SUCCESS in 14m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/11fc14bf9a6c44769ad2b45d148bc6e9 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/31db294f09344ba4b1ddfed2c94e040e : SUCCESS in 27m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1a58bec6380045a18c097d7e4ffdde4d : SUCCESS in 24m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c355ce613ea8467896b660fbf5324d8e : SUCCESS in 32m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4f49e24ab77940da8d3f19fea1105dad : SUCCESS in 24m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d55c8beea87147c597ea494acb5c5a4c : SUCCESS in 9m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/df23c09dee6d48e683a01c2c92806ad6 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e5a62f905c224f59861ef7a54a69f374 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5854b7cb6b5148709efb94e343e8e2f6 : SUCCESS in 9m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8c9f48c1c0854825b82428f74617ba0c : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f9df0e20fb0b40049bb7011c5a6a14f4 : SUCCESS in 6m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eae0bf799b4644e49e5e907d9427ceb4 : SUCCESS in 52m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91d40f1e105f4679869b722f44f8178e : SUCCESS in 56m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa657165506a40d9a664aa42a31f3eb7 : SUCCESS in 55m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92ece74e5c0742b09d1c29c0e9cdf414 : SUCCESS in 36m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be0367f1b4d43cdbc53a194061ab423 : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc9eaf12b3db428eb707187ea8cc2cca : SUCCESS in 21m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aef557f50a194f29afb3d9e7000d47dc : SUCCESS in 1h 14m 38s","accounts_in_message":[],"_revision_number":32},{"id":"5079df991e7fb01f3a1ecfe8ebaa007989a314b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 18:53:58.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6ec537ec4454e209b143e4d8ea4e66c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/824b60b53a0a42dd8627e7f56b550ce2 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":32},{"id":"862461a154d46f66b21d6882ae259c167a5bbee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-24 17:58:55.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"489748c2b6327b6fee75ca4348a1a2722368e107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 19:11:42.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfde351b19f3408b8bf710adbf0c1e9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd300b8da6a4c8c9c42500fd856a119 : SUCCESS in 5m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/2af5fda47e94405f9efb0b0f5f1727df : SUCCESS in 1h 02m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86b9d3d2e7e747e1a2a086b1e8de8b80 : SUCCESS in 1h 06m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/77efd1810b394c588551c3cebd3bf791 : SUCCESS in 33m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ba53da45d1e4d12a2c52bed470c87cc : SUCCESS in 34m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/229d44c4e7024762b29d878f00acbbfa : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05a64e55adf54f7d85e33606e1e14f1a : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b61aa3e0e944fb3a38b8d3b0b2d36cf : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b5072063ac4455fa4f7aed447b30801 : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c37255b19b834e2c9fd94386418a9701 : SUCCESS in 20m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/15e88a13861846d3a7e410e09b56b75c : SUCCESS in 22m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/80cf7d2144c645eda43c623487af2b5a : SUCCESS in 12m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0b3ab83a863b4942badb32795d87c809 : SUCCESS in 22m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9932992dda22484ba85c19dffedd3e23 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/76d7ba6ad0f940a7877a69d989982dfc : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3e3fcdd5105249ff8bbce6d09a2d6b1a : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/80a48750e1b047129ccd9c9c38336139 : SUCCESS in 7m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b045aa189d84a5f81a708ffd0fc7e5f : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3da8972985cb4314a9133e3a5a47844e : SUCCESS in 11m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b8c09cca5a84e22830d3cdab0614207 : SUCCESS in 59m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/268c62c4391140838f4440b1fdf77612 : SUCCESS in 25m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7acac73451464db8b080654746154920 : SUCCESS in 51m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f96011114d634206a53a142ad1133b07 : SUCCESS in 32m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31dbd905a517403eb935f333d3d0f4f6 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/471871ec7a9a4dce9ba4c31954c1b835 : SUCCESS in 10m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7374d744e41744c0bb4acb9e93eedf21 : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":33},{"id":"5a73c7fac22dfe404824364e483545eb49b828c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 21:12:17.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7907f57fca634b77a59f15590d07cd45\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2b7db2696c234056891e333d62c9f759 : SUCCESS in 40m 28s","accounts_in_message":[],"_revision_number":33},{"id":"808e32dfffac30ea3c87182b532261432158b5a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-25 16:30:13.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"4cea057a1c3da12c1051eb27f6461a423b6754bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:51:11.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b97e7925e8e04d498f4db2ebb2e6ae1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc6be42c267f4710bb3d145a9e1ddb34 : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/c81bb211a7b64d05aae70cff50837e7e : SUCCESS in 36m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f23ee1c8a014c1b8cc8c137da307461 : SUCCESS in 35m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/840c19665aaf46bf9547e60032189a63 : SUCCESS in 39m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4606b17f9592421c835f2ee98989bd2f : SUCCESS in 37m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99f8eb0189734b4ba240209366efd089 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b580a120dc4c40c0be52f3df9a196f2b : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ecb98b8c54c4459b529a90a5cf2df14 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca9ed3bb93d548bea84902989de0bd84 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c947c5e67e80491ea3e433421f4c05ba : SUCCESS in 24m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c5456fe89aba4e5ebeb62e02348a235f : SUCCESS in 27m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/055255f3387f42d2b7e9573ff966970f : SUCCESS in 23m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/71b4901c53ea46f08335d28cf8be118e : SUCCESS in 20m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c79ba5c10db64ca4bfd3796fcc7c822b : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1e916cf922fe417e9ed904d834d8e884 : SUCCESS in 9m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a4aba0522f7f4ccd8a4ddc2da97432ef : SUCCESS in 9m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c3a6d956e5e74d0a9d0902dc84dcba95 : SUCCESS in 4m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eac2b9811f2c45dc9f775f61702d7d92 : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/65e1eba4945c4a44a5b360522221b89d : SUCCESS in 12m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2787fba2b5344758f99a86eff19189c : SUCCESS in 58m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7349cf1070984fdda3aabc33c3c2fbd6 : SUCCESS in 51m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e09b6dee50fd437796b1dfb732c08ba8 : SUCCESS in 1h 05m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/652ceaf9c8884dba83eac6c0d427bb2a : SUCCESS in 28m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd4aebda967742e1b62f223625209f46 : SUCCESS in 3m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee3a36d3f84c4d618800e4e5d4110e3e : SUCCESS in 18m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af201e3877204893ac82127e632ad1c2 : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":34},{"id":"8e94fb67a0846999c726c1de903aa16ac48ef49c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 20:22:50.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c449db02a99243d7a85ad73add067d8d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6711b3bc181c42c89c299f580e29f17e : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":34},{"id":"781b0a46c2ca7a4b088677acd19275135d73524b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-31 14:54:16.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"2e8759ea273a26362cdae94250539ced641d6371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:22:28.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60d0ab1ba5584df19ea71987e935b396\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa0e27fb82b74f91a3ee7df1e922e6ae : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/44485ba410bd43f3b4527d887a9a8be8 : SUCCESS in 34m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38da2db3a3a24e0e9195abdf8dda8473 : SUCCESS in 34m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/129f9f83e2b44d57937f7a923f8d7152 : SUCCESS in 1h 01m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9cafd179b504e97bed8a8ac069ed664 : SUCCESS in 1h 07m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74361100911944fc82555f3a3697c7a3 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d012bf4ea6954691bfaaaa21a46a8482 : SUCCESS in 11m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01a3579931864644b35b6a73dda0ef33 : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f805106922814a339b2a5cdc93911d40 : SUCCESS in 8m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3c0efe61b476453ca492251b26a1d884 : SUCCESS in 25m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cd5187f5e4ae40858d8f210d09a2ab61 : SUCCESS in 24m 27s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2d55b5132ede4623b81e9fbe19d8beb5 : SUCCESS in 25m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/76fbe023cbed4fea95b2261cb43866d4 : SUCCESS in 23m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/762cae3168324f43b61988ccef151c94 : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/45a490def62f48f18a4c3dce01f56b00 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/55286a96dd1447f8b25565dcb685278c : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa8097dba1fc472aa8d97ddf9c0aab15 : SUCCESS in 6m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce90c5bc3a2e4bd28cf4b72c6dde734b : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6b942b0832cc4a7eb9fcee799dd7bddc : SUCCESS in 8m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/347077b60942475790269af58837c1ce : SUCCESS in 49m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9149eb5458754a18b551afbb30d0aa12 : SUCCESS in 58m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ed8adfea7f644dfb67c58ac3ad442ae : SUCCESS in 49m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3489c6015d1423eb6700f65495c4482 : SUCCESS in 27m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57edd9dd13fb4d4f9414ffca79aaddb1 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ffd4d170a2b484db54fb82c31169725 : SUCCESS in 14m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c7d4a07230a4007ade7f52b3d5e050e : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":35},{"id":"854c358d6a3c9b60936d839c9f37f580c95fae26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:17:09.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23616c9b9b7f4c75a5d522b3ce991861\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ff4a8fde00848129de8d8479eb27099 : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":35},{"id":"09e7b82058c77be120055a2d9b76834a4fc27f70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-09 09:21:56.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"69a3d6bd597e39cd9634a6ae29a14ed8d5a15eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:43:40.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79d0ea1b56e041e5b1c71210f8423b2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5bff4003999414b8a12a92a09e952ab : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/0d146f568f524986b1a300a159d12450 : SUCCESS in 1h 09m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6258f7d3496049d5ae6bdb057dffbc08 : SUCCESS in 1h 02m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8226dedb5ce4d3298c3999501bcd4a3 : SUCCESS in 54m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15194db56ab045d887164e84580c0fe7 : SUCCESS in 1h 09m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec4958729b514bb1b00950ab876d56cb : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce2a5cec1c604527ae90b27fd4463262 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba5f3acb8a674ad0b41b4a190b528ee5 : SUCCESS in 20m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c52f988966b54d43a350af205c454a10 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79fa4f57e3404918ae0443444b9ab1d9 : SUCCESS in 23m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/43f0e0b396b644e494f7cca6dce48d69 : SUCCESS in 22m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3be0fcb41d9d4eb28156fe3e57912e52 : SUCCESS in 25m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d03986b6f26c44d294d53e7dac32a5de : SUCCESS in 12m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d19cfeb35bfe48eabd88326b58756f46 : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6b6c11ae24054fda9c6ec7f5cc288a20 : SUCCESS in 16m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5191c6dfabfe48f5a471a2443206b7dc : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/22ae2552b2d648d8ac910dd1febbd14e : SUCCESS in 6m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/53e41de5e8474640b26844f78dd9396c : SUCCESS in 8m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3abbb97e37734723bd927c49195eff9d : SUCCESS in 9m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7af38fc55eeb494dba46ab3b34c53187 : SUCCESS in 53m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/096f029d9d584762b09087ced3243f4e : SUCCESS in 47m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/922beedb49b449f499a3d7a643449574 : SUCCESS in 57m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b82ae146a64a465e8e3394e30ced2a5d : SUCCESS in 24m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffbedf9f16b840c5912c2a3992feb175 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/329249e9d7654336802689e4f3997523 : SUCCESS in 19m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0afdb76be43f46f2bcb835e5611c97f6 : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":36},{"id":"96d92f42aa223f15c8e66ef815e73685a2d68567","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:15:34.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4378526513084ecbb94191e34ffd6955\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/add7697bf888402c9d0c1e86b47a4303 : SUCCESS in 35m 03s","accounts_in_message":[],"_revision_number":36},{"id":"702feed53e505868c76688a882a8edb563d6d3a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-13 09:47:13.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":37},{"id":"92f4cdaff2ef11ef1e14438ac4269663b7ec5945","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:47:51.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f96db67d99a4b49b1d2db3cb1c2fe6c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/87eb1b3b607b443e81c2a3e2588cf914 : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":37},{"id":"4816221f9b58bbfd4c6c59983341cb90a48ee0be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:50:42.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf79081c9c794f3e893373aff49d97d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e28bae1fd2643d9a832938d74750926 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/445b4e3b553e4b00962759c3dc9109f2 : SUCCESS in 1h 05m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/404458705fff4e12a24e6f0cc56e38c9 : SUCCESS in 1h 03m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4bfbf08ba3c74ee9939b9c8fe33f632d : SUCCESS in 35m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2904f25fff39420894fe39168df5703b : SUCCESS in 46m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9877d239148944eebb4c46f650c4ae2c : SUCCESS in 23m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96447f2368304a9e8062771c7edbb08d : SUCCESS in 15m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/625f2ee95f5c483396d90f6106d08bb3 : SUCCESS in 13m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/39f45ab8e7ac44ad8b1f6151102aca8d : SUCCESS in 7m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6beed19e57174c529c2be5368d257ada : SUCCESS in 26m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4adb955524b54fd19959a650d8809c5b : SUCCESS in 24m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5d397f7473ab4446b32b3f2dec245b7e : SUCCESS in 36m 58s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/52a879ada26c4ca194ed462c25ae91b5 : SUCCESS in 22m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/da4d0e43e607471eb7d474dac9f2bff4 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/827e6adb856d4243bd4b09949784bda5 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/73baac7652d54c128733b3732f0e34eb : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de5d0a3b12fa4fbab22b1e3d9429b4eb : SUCCESS in 6m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2bc9e988cd245219de9c09467bf8c2b : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2233b0a902534c10b79268f209ab5df4 : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7c9b093703c41889b120c0bf5d866a7 : SUCCESS in 58m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75cd36f563a34ba6a85a38d3cd21b437 : SUCCESS in 53m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef2987991bb842dfbfc54d2abbc8246c : SUCCESS in 55m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1be8eb18c4534f6db2d1ea84b638f49d : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0afa87557d504fcfa4fe5b1ba77aa785 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7d534d1d39f409dbae0aacb48798b1b : SUCCESS in 19m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2f0c121b6754dfbbe77dd0b4ccf7f29 : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":37},{"id":"72cb74735c07a581d1fb39e6d872a40be3df53ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-14 15:09:20.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":38},{"id":"2f7fe9837e5e9653784c71ffc92b83dcd210cae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 16:28:58.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faf90dbe7c7d4f298cf2827d62a737fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1daa632bde439aa75646e2d6327750 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/0d735b6bf4c447ffb6046d0a60de3f0f : SUCCESS in 37m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b83b98ce4abb49f993acae541f2949e1 : SUCCESS in 54m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdaf6fe9d272475896d433a7ba72b7c2 : SUCCESS in 47m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/012ee8d0e8ab43b39e2b5782d762caae : SUCCESS in 53m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a74d88dbb5cd4eacbe46ad2446a365cc : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7e0d79cbb1f41c08f35326ce5ab10fa : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f03dc5242d384522ad73ae10d69a2894 : SUCCESS in 22m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cacaceffcbb94aa7bb9fc99877b17b15 : SUCCESS in 11m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e23f313207b141c08420c8d89f2280ea : SUCCESS in 28m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/58f41ea8a6ea4984bf8090b2268683a8 : SUCCESS in 17m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e35ecaa23dd240649bad4c5994117dce : SUCCESS in 30m 54s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/64931ea880fe4ee09b4cb8264a2667f7 : SUCCESS in 11m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/39462ce33884440eacf7da76958b76fb : SUCCESS in 9m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9d368f8625034a4e8bbbaad2a0a3d081 : SUCCESS in 13m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/17f260b583c344a6a647231025dee5cd : SUCCESS in 11m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/851c51ec0fdf4f07bc66ae0995757d8c : SUCCESS in 6m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7a0abaa8a94442a2b2a0051b5310be37 : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9e5e53d98ff0452990d5e49980e4c58d : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a85bc98954d43289d57a0f9b6aae2a0 : SUCCESS in 1h 05m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/78bf5b0c476a4b658920da5c743e1472 : SUCCESS in 28m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e1d57711a5f47aebb692d7f7579157a : SUCCESS in 40m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75228942d87c45678c13973bcf13e3ab : SUCCESS in 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d386e3750f4b4eb6073e9301c77545 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ef492fb9b624a0b987195838e05f0c6 : SUCCESS in 26m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f37bc2bd7cdc4460841cac024f16b8c6 : SUCCESS in 49m 54s","accounts_in_message":[],"_revision_number":38},{"id":"e26e84d8c5356dbcfd9df5f99e3d3c6cf2c950bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 21:07:22.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7e1ac81ae9a48bea82f81c38a0ea1db\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d97b5e5155e14199b93de57d8448fb9a : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":38},{"id":"2f630704d2ae3f227e1f3296bbb17aaf8644029b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-06 10:50:40.000000000","message":"Uploaded patch set 39: Patch Set 38 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":39},{"id":"3a6fadefe50522e34568608e5747e68846cb1841","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:41:25.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05e46db0b6d441faab86dd8a16dda1a0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b36b694493b448f8e6a6bef303c2982 : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":39},{"id":"e5b3e2518821ec7b95ec5ebead2ca6e7f1a3f2f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 12:16:34.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24528ed25b1f42318a0177d2d2c2459a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5342416afc148c3a3cdda881b347a28 : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/4628cb83273a4e24b262c923f13f2d37 : SUCCESS in 1h 04m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c466f6086abc4bb6a942691295b0ca6e : SUCCESS in 42m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/23b6a989c5c642ca9e6da8565b9bceff : SUCCESS in 57m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/630514bff98d406bbc6054c45f4a34b4 : SUCCESS in 1h 03m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa627fcdb7944e56aecd92c529c7b185 : SUCCESS in 25m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d636635cd0246ab8cbbf0c448bd3c26 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8132268934a48aaae3fde97dca927b6 : SUCCESS in 22m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0362719f98204a9ea7044954b2ab859a : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/023623c75e7d47cfb24f3ade3913b337 : SUCCESS in 19m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b2300c46a74c4c50a54943249980231f : SUCCESS in 23m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9fb653ce16174f2db1325635c142116d : SUCCESS in 24m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dbbfba168ac0431baabd6d92fb70d3ec : SUCCESS in 20m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8f264e16fd2346db83c24d4963628180 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0648cf5f3dff4a43987fee1f61fef702 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9c17d1cdcf974048b485db0fff51a916 : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05595e004b27482f8db0eb205bdf7119 : SUCCESS in 6m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7224eeca788d41059c7b73675f6bc7ae : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/96348414914e40f2a7baf99252412a9f : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cea80bdd58b4fcaa9629787b62fef18 : SUCCESS in 45m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/268b24061aee4df7b61a9322b5bbe382 : SUCCESS in 55m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9017e827ddb248c183ab9319ee27878d : SUCCESS in 53m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d019b20995814e0eb54dfba80d003d13 : SUCCESS in 30m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf855f3b4a444252937a97d17b3bde6e : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a2187da3d4754bca95ce3267c11bddd2 : SUCCESS in 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5f381881fcb4409a66fbf4ef51088ec : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":39},{"id":"c1b30a18e04e726750c1f419e496f6807bf32c69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-07-22 23:58:01.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":40},{"id":"8e77d0b5c3c54601c9c458b1ea425dd9405f19f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 00:58:46.000000000","message":"Patch Set 40:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b062454306345029a7acd0868c22eda\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b72b606b54cf4f88a8dec8d8bdb93093 : FAILURE in 56m 15s","accounts_in_message":[],"_revision_number":40},{"id":"edb107ce0911ce43be977074e0775e03c13dfca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 01:07:04.000000000","message":"Patch Set 40: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c2274cc3b1e4d8a98d1d4e0b499f538\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a75ce8da8e046d28f1d87ae9f52351d : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1ad464172cb4e26840a805a378e3cd5 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/95aba1d529dc4e85a8423bbbc70108f9 : SUCCESS in 1h 03m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc2f917fe0fd44388eb3355a0ca6133c : SUCCESS in 47m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2430d0d1aab4a498d0affa1debf490e : SUCCESS in 34m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16053777742d44b6be5cc4ca16b8f2ec : SUCCESS in 34m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b883e4f982eb43649fd5f412a972831e : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af47b77c91a24ba78f4552d3661856fc : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ed712d67c474fd58d983ac4799dff6e : SUCCESS in 14m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5fb3aef146f84f8db9f5bd66072a03f7 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dac4a2b45fc244499a149f86e17c8836 : FAILURE in 11m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d02f016ea5b94886b48982ff657e8461 : FAILURE in 24m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1a4d4579e7fd41dcb4fa31e82cd39d9a : FAILURE in 31m 54s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1c536f16df974cb3997319ec0f91c7d3 : FAILURE in 10m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fd9bb16c906c42aeb2e811822b049376 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1d79518d9f5d4f32ae1ad7dd3039b238 : SUCCESS in 12m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8ad93b4e26974a738cfd9e481db5f76a : SUCCESS in 9m 11s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/24cca6fa0a4b452ca76c5a45c25d297e : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a79c5f25ca0e462a9a842db278a69e61 : SUCCESS in 6m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0efda59cabc34553a45861f0e85e0855 : FAILURE in 7m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a5ff26180974c008c00f3cf96cc437c : SUCCESS in 11m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/28f435090c294b76ac6b61f62e41616d : SUCCESS in 37m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab02b9b9278d4d3f984e2726a6af5d42 : SUCCESS in 54m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/029a102f48e8471da8349dc7df2f4bd8 : SUCCESS in 51m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4902e80acdfe483582c41dd2ffdd3fd3 : FAILURE in 27m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73dfa41489ae4342be3712f36b8a7fcc : FAILURE in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5b2846c7a1454820981e8e60cce3925b : SUCCESS in 19m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fa9146f70f242898478954c161cebbb : SUCCESS in 1h 00m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":40},{"id":"2617b75816445ee39d96b11ff1638decb4ee9017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-23 10:27:09.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":41},{"id":"db19699eeb8a012a74b968016f161d81e1a728be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 11:46:20.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0af6343de3204bb99e345cf2b64f6f1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c36bee439f435bbca6b258e1d2fdf2 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb27c7321453446d928cd53de6881f17 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/0e171ef06fba4b4eae45b74b8075453a : SUCCESS in 1h 10m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fe97b52b88e84802806928266f931820 : SUCCESS in 1h 01m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c5bbf07c54fa4061bb5b83412de52ba2 : SUCCESS in 49m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de9a5f79e90846e08c8aa5298bf7d35f : SUCCESS in 59m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50e7ae5668b1439c90ee5fe00c79eb04 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65f6e79167e942dca4c70fe0a7fc9e04 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a88abb9144194430951cad43bebc2b2f : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d45e81c871f428dafc74f4e8af1a7c0 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/154f4ea18f9542cab0f677edb8442a5c : SUCCESS in 30m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0e21472735884ca29dde41199f4307f9 : SUCCESS in 22m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/adda1169aef5483bafdf2068004b32a0 : SUCCESS in 30m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fc9fec37698444f281ee6e7c74622838 : SUCCESS in 16m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5074180c9f83421ca2a4d9c0ee9df071 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2407d570d1c04da79da288414ed46049 : SUCCESS in 15m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6c1c2b5c3b0b43b8b1294938cf68ef4e : SUCCESS in 16m 02s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ecc918d292d4cfa809f5c260fef7fbd : SUCCESS in 20m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ff9a538b479e4baa81c824a35d111fd1 : SUCCESS in 8m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d382e8e9e35a4687b0d1aa32fadc68fb : FAILURE in 8m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9ac8165020c2477e9695389cf4434915 : SUCCESS in 9m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6622d86639d47b7ac40e4afd3a0fc87 : SUCCESS in 52m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/69c984fb30a14a638a374dafe4c4ce65 : SUCCESS in 55m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c142f339fcd40139db1edbc7a91aa4d : SUCCESS in 58m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4092e82fed9e4804960e30c761182721 : SUCCESS in 31m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d2bca0b6534e7c8b9b3981c45af923 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd74c247b7584146a37e6751cd35d287 : SUCCESS in 21m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce193253cad643c78c0e7243048063c8 : SUCCESS in 58m 57s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":41},{"id":"7975aa26ce466bdc1e9f56dc9f2bb563b3f6cd6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 11:50:24.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6fbf69091b84a6ebef3837a8f94baaf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3518d195497f4dbc98370b218e59ba32 : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":41}],"current_revision_number":41,"current_revision":"a65c31894987ad39b2968658ccd59dc56afe3a54","revisions":{"8b44f913fc42e34df5843e5639fc51a53252239e":{"kind":"REWORK","_number":1,"created":"2024-10-15 20:18:19.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/1"}}},"commit":{"parents":[{"commit":"3db5c137d98880ac9e3b7d16a1302384a5930c1a","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3db5c137d98880ac9e3b7d16a1302384a5930c1a"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-10-15 20:14:11.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-10-15 20:14:11.000000000","tz":-420},"subject":"expirer: remove delete task if object returns newer x-delete-at","message":"expirer: remove delete task if object returns newer x-delete-at\n\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b44f913fc42e34df5843e5639fc51a53252239e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b44f913fc42e34df5843e5639fc51a53252239e"}]},"branch":"refs/heads/master"},"af2a9136fdb81fea4765b6267c60e272a0c35443":{"kind":"REWORK","_number":2,"created":"2024-10-19 00:27:10.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/2"}}},"commit":{"parents":[{"commit":"3db5c137d98880ac9e3b7d16a1302384a5930c1a","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3db5c137d98880ac9e3b7d16a1302384a5930c1a"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-10-15 20:14:11.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-10-18 21:59:54.000000000","tz":-420},"subject":"expirer: remove delete task if object returns newer x-delete-at","message":"expirer: remove delete task if object returns newer x-delete-at\n\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af2a9136fdb81fea4765b6267c60e272a0c35443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af2a9136fdb81fea4765b6267c60e272a0c35443"}]},"branch":"refs/heads/master"},"6954afe1adb58957fa7a453dc3771b65c071d69d":{"kind":"REWORK","_number":3,"created":"2024-11-20 05:56:56.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/3"}}},"commit":{"parents":[{"commit":"38814414c775320caa5492e02d4b170ff697ded4","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38814414c775320caa5492e02d4b170ff697ded4"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-20 05:55:22.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6954afe1adb58957fa7a453dc3771b65c071d69d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6954afe1adb58957fa7a453dc3771b65c071d69d"}]},"branch":"refs/heads/master"},"914d21d667c18afcc2490fa1e6fd7a14db4ac60b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-12-14 00:27:42.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/47/932447/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/4"}}},"commit":{"parents":[{"commit":"b01d4baa4aeaa67ed0ff89331da9f0207b7849f9","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b01d4baa4aeaa67ed0ff89331da9f0207b7849f9"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-14 00:26:24.000000000","tz":-360},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/914d21d667c18afcc2490fa1e6fd7a14db4ac60b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/914d21d667c18afcc2490fa1e6fd7a14db4ac60b"}]},"branch":"refs/heads/master"},"ddd91de7a536ab8196942a02dd06f1d6edf02a73":{"kind":"REWORK","_number":5,"created":"2025-01-16 12:23:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/5"}}},"commit":{"parents":[{"commit":"103f8a73a64c11ceffaf97284726c7a7ebe50b22","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/103f8a73a64c11ceffaf97284726c7a7ebe50b22"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-16 12:18:34.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddd91de7a536ab8196942a02dd06f1d6edf02a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddd91de7a536ab8196942a02dd06f1d6edf02a73"}]},"branch":"refs/heads/master"},"4cc7b0c00cd34366781bedd5bc77ffcc3ff86656":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-01-16 15:11:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/6"}}},"commit":{"parents":[{"commit":"49de5c41e561504a9187493f9e38ae6e6ac33412","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49de5c41e561504a9187493f9e38ae6e6ac33412"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-16 15:11:10.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cc7b0c00cd34366781bedd5bc77ffcc3ff86656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cc7b0c00cd34366781bedd5bc77ffcc3ff86656"}]},"branch":"refs/heads/master"},"7dae5df3afe3f5bad9b05a699b24f2667618b2d9":{"kind":"NO_CHANGE","_number":7,"created":"2025-01-16 19:24:19.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/47/932447/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/7"}}},"commit":{"parents":[{"commit":"49de5c41e561504a9187493f9e38ae6e6ac33412","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49de5c41e561504a9187493f9e38ae6e6ac33412"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:41:52.000000000","tz":-360},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dae5df3afe3f5bad9b05a699b24f2667618b2d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dae5df3afe3f5bad9b05a699b24f2667618b2d9"}]},"branch":"refs/heads/master"},"4af11cd17a6f72ea37e22312e1f94084ef417b8c":{"kind":"REWORK","_number":8,"created":"2025-01-21 16:47:59.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/8"}}},"commit":{"parents":[{"commit":"49de5c41e561504a9187493f9e38ae6e6ac33412","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49de5c41e561504a9187493f9e38ae6e6ac33412"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-01-21 16:47:28.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4af11cd17a6f72ea37e22312e1f94084ef417b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4af11cd17a6f72ea37e22312e1f94084ef417b8c"}]},"branch":"refs/heads/master"},"782e610b5c045f4f880b15c7abdead11e39543e2":{"kind":"REWORK","_number":9,"created":"2025-01-21 18:28:56.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/9"}}},"commit":{"parents":[{"commit":"49de5c41e561504a9187493f9e38ae6e6ac33412","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49de5c41e561504a9187493f9e38ae6e6ac33412"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-01-21 18:28:40.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/782e610b5c045f4f880b15c7abdead11e39543e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/782e610b5c045f4f880b15c7abdead11e39543e2"}]},"branch":"refs/heads/master"},"0399ccea2d9b2701b75b0f79b6b7f51e72fe7a56":{"kind":"REWORK","_number":10,"created":"2025-02-04 17:48:57.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/10"}}},"commit":{"parents":[{"commit":"49de5c41e561504a9187493f9e38ae6e6ac33412","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49de5c41e561504a9187493f9e38ae6e6ac33412"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-02-04 17:48:41.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0399ccea2d9b2701b75b0f79b6b7f51e72fe7a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0399ccea2d9b2701b75b0f79b6b7f51e72fe7a56"}]},"branch":"refs/heads/master"},"b1f3742419a5e4e8cf91cbe47d6f67b743a7df9a":{"kind":"REWORK","_number":11,"created":"2025-02-04 21:45:00.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/11"}}},"commit":{"parents":[{"commit":"49de5c41e561504a9187493f9e38ae6e6ac33412","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49de5c41e561504a9187493f9e38ae6e6ac33412"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-02-04 21:44:24.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1f3742419a5e4e8cf91cbe47d6f67b743a7df9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1f3742419a5e4e8cf91cbe47d6f67b743a7df9a"}]},"branch":"refs/heads/master"},"5b70606cc8d5734af7f88a297921514498a0f982":{"kind":"REWORK","_number":12,"created":"2025-02-08 02:01:05.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/12"}}},"commit":{"parents":[{"commit":"df249434186b9f173edf8f5aa93d1c46643912fa","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df249434186b9f173edf8f5aa93d1c46643912fa"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-02-08 02:00:42.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b70606cc8d5734af7f88a297921514498a0f982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b70606cc8d5734af7f88a297921514498a0f982"}]},"branch":"refs/heads/master"},"995b0990b2ef560b71fe3e13cc358394460f8aa0":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-02-18 19:17:44.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/13"}}},"commit":{"parents":[{"commit":"6fcca0b2c99c5ffb2beace4d213ec14cd91f9204","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fcca0b2c99c5ffb2beace4d213ec14cd91f9204"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-02-18 19:17:18.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/995b0990b2ef560b71fe3e13cc358394460f8aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/995b0990b2ef560b71fe3e13cc358394460f8aa0"}]},"branch":"refs/heads/master"},"0756f43f6eed5b8af3ebada452baed120b37a136":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-04-10 21:43:04.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/932447/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/14"}}},"commit":{"parents":[{"commit":"96379d31f1f150b30b430862be2c8fa19d1fbe26","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96379d31f1f150b30b430862be2c8fa19d1fbe26"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-04-10 21:43:04.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0756f43f6eed5b8af3ebada452baed120b37a136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0756f43f6eed5b8af3ebada452baed120b37a136"}]},"branch":"refs/heads/master","description":"Rebase"},"16d746e7117af6216a6cece48e4cf487e38756a4":{"kind":"REWORK","_number":15,"created":"2025-04-10 23:11:09.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/932447/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/15"}}},"commit":{"parents":[{"commit":"96379d31f1f150b30b430862be2c8fa19d1fbe26","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96379d31f1f150b30b430862be2c8fa19d1fbe26"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-04-10 23:10:58.000000000","tz":-420},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16d746e7117af6216a6cece48e4cf487e38756a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16d746e7117af6216a6cece48e4cf487e38756a4"}]},"branch":"refs/heads/master"},"e8c76305b669f222a96fb26d4251e9ef8ae7f676":{"kind":"REWORK","_number":16,"created":"2025-04-10 23:13:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/932447/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/16"}}},"commit":{"parents":[{"commit":"96379d31f1f150b30b430862be2c8fa19d1fbe26","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96379d31f1f150b30b430862be2c8fa19d1fbe26"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-04-10 23:13:14.000000000","tz":-420},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8c76305b669f222a96fb26d4251e9ef8ae7f676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8c76305b669f222a96fb26d4251e9ef8ae7f676"}]},"branch":"refs/heads/master"},"f557fc86ae8b7532d23c6c00e2acef908f057a81":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2025-04-10 23:18:47.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/932447/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/17"}}},"commit":{"parents":[{"commit":"7653cb710611eff08de0313a677b6d5ebba9c186","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7653cb710611eff08de0313a677b6d5ebba9c186"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-04-10 23:18:47.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f557fc86ae8b7532d23c6c00e2acef908f057a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f557fc86ae8b7532d23c6c00e2acef908f057a81"}]},"branch":"refs/heads/master","description":"Rebase"},"2b2a77a26caaaa34e7b193a868b4b07ab3edc953":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2025-04-11 09:53:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/18"}}},"commit":{"parents":[{"commit":"5445b983ec19c3f498f250a78186b0ad682e223a","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5445b983ec19c3f498f250a78186b0ad682e223a"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-11 09:23:52.000000000","tz":60},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b2a77a26caaaa34e7b193a868b4b07ab3edc953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b2a77a26caaaa34e7b193a868b4b07ab3edc953"}]},"branch":"refs/heads/master"},"4a7f8cf4d955f73b947c0d5c9c5a2b544f833844":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-05-14 18:38:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/932447/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/19"}}},"commit":{"parents":[{"commit":"eb4a1d7014dc94baaa7e23de37f2b7bc9802a61d","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb4a1d7014dc94baaa7e23de37f2b7bc9802a61d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 18:38:18.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a7f8cf4d955f73b947c0d5c9c5a2b544f833844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a7f8cf4d955f73b947c0d5c9c5a2b544f833844"}]},"branch":"refs/heads/master","description":"Rebase"},"884c65f80ab1a0a3f9406b8b91e71397ba634346":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2025-05-20 01:24:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/932447/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/20"}}},"commit":{"parents":[{"commit":"693645d4b5e58f17e63b46b125478097f504e87e","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/693645d4b5e58f17e63b46b125478097f504e87e"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-20 01:24:44.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/884c65f80ab1a0a3f9406b8b91e71397ba634346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/884c65f80ab1a0a3f9406b8b91e71397ba634346"}]},"branch":"refs/heads/master","description":"Rebase"},"094f5b86fd6496f09fe079f14832f89808489224":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2025-05-20 14:27:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/932447/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/21"}}},"commit":{"parents":[{"commit":"55193c82a2f6febf1b079ef1aefc8709dde35ea7","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55193c82a2f6febf1b079ef1aefc8709dde35ea7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-20 14:27:19.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/094f5b86fd6496f09fe079f14832f89808489224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/094f5b86fd6496f09fe079f14832f89808489224"}]},"branch":"refs/heads/master","description":"Rebase"},"f2e4642120b131e25ee29f5edbe26a02442466dc":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2025-05-20 15:56:54.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/932447/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/22"}}},"commit":{"parents":[{"commit":"71fb6f26ca83212b0c7f7a6f4157f4a1e555ff81","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71fb6f26ca83212b0c7f7a6f4157f4a1e555ff81"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-05-20 14:59:06.000000000","tz":-420},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2e4642120b131e25ee29f5edbe26a02442466dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2e4642120b131e25ee29f5edbe26a02442466dc"}]},"branch":"refs/heads/master"},"77ba4df1d9442195e4a844b89e90d4c72dd1e9f5":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2025-05-20 16:48:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/23"}}},"commit":{"parents":[{"commit":"e513dcf93478907e077d39fe5f6d347cac3d6a96","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e513dcf93478907e077d39fe5f6d347cac3d6a96"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-20 16:47:53.000000000","tz":60},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77ba4df1d9442195e4a844b89e90d4c72dd1e9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77ba4df1d9442195e4a844b89e90d4c72dd1e9f5"}]},"branch":"refs/heads/master"},"624fd7a139e9d7c85ab435ddf1efae6bb455ebc9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":24,"created":"2025-07-14 18:58:34.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/24"}}},"commit":{"parents":[{"commit":"a30193118bd82570d0214487a4e319d3bbd1496a","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a30193118bd82570d0214487a4e319d3bbd1496a"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-14 18:57:11.000000000","tz":-420},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/624fd7a139e9d7c85ab435ddf1efae6bb455ebc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/624fd7a139e9d7c85ab435ddf1efae6bb455ebc9"}]},"branch":"refs/heads/master"},"5ae343b4f0726efa3d273ff522e1262eacdf0e60":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2025-08-20 05:28:06.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/25"}}},"commit":{"parents":[{"commit":"0cf48f884d04c8df1b88282a069a8f0e9637fd8f","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cf48f884d04c8df1b88282a069a8f0e9637fd8f"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-20 05:18:31.000000000","tz":-420},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ae343b4f0726efa3d273ff522e1262eacdf0e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ae343b4f0726efa3d273ff522e1262eacdf0e60"}]},"branch":"refs/heads/master"},"8f36683b564abeeb07e4887a203c1654b16ede31":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-09-05 19:46:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/932447/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/26"}}},"commit":{"parents":[{"commit":"2a556b108bc62e7d51b5145517bbd081baf0e727","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a556b108bc62e7d51b5145517bbd081baf0e727"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-05 19:46:26.000000000","tz":-420},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f36683b564abeeb07e4887a203c1654b16ede31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f36683b564abeeb07e4887a203c1654b16ede31"}]},"branch":"refs/heads/master"},"d06072563f71e12b91512756282e2b62f38efaa2":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-11-10 05:18:45.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/27"}}},"commit":{"parents":[{"commit":"7a4c96e3ca9667a0eb555fe9cb3eb95436e99f93","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a4c96e3ca9667a0eb555fe9cb3eb95436e99f93"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-11-10 05:14:10.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d06072563f71e12b91512756282e2b62f38efaa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d06072563f71e12b91512756282e2b62f38efaa2"}]},"branch":"refs/heads/master"},"3309915f025f460ae09a7810f20bd9f37ae10643":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2026-02-10 04:43:10.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/28"}}},"commit":{"parents":[{"commit":"50fc4e28ce577553e993ef3ac01abaa5888d7057","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50fc4e28ce577553e993ef3ac01abaa5888d7057"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-02-10 04:39:12.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3309915f025f460ae09a7810f20bd9f37ae10643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3309915f025f460ae09a7810f20bd9f37ae10643"}]},"branch":"refs/heads/master"},"bab67a58aeac54ab652a2fcdbf03f75d00dbd4cd":{"kind":"REWORK","_number":29,"created":"2026-03-03 05:39:12.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/932447/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/29"}}},"commit":{"parents":[{"commit":"188aa9190160cfde17dc973ac8a81cfc82cb8add","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/188aa9190160cfde17dc973ac8a81cfc82cb8add"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-03 05:38:23.000000000","tz":-480},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bab67a58aeac54ab652a2fcdbf03f75d00dbd4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bab67a58aeac54ab652a2fcdbf03f75d00dbd4cd"}]},"branch":"refs/heads/master"},"842de095b24deb5a207961ae53313187553ca49a":{"kind":"REWORK","_number":30,"created":"2026-03-03 16:02:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/30"}}},"commit":{"parents":[{"commit":"188aa9190160cfde17dc973ac8a81cfc82cb8add","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/188aa9190160cfde17dc973ac8a81cfc82cb8add"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-03 15:49:44.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/842de095b24deb5a207961ae53313187553ca49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/842de095b24deb5a207961ae53313187553ca49a"}]},"branch":"refs/heads/master"},"fdc9c5b7e2ba6a287d9196699332b6b9f76fd351":{"kind":"REWORK","_number":31,"created":"2026-03-18 13:35:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/31"}}},"commit":{"parents":[{"commit":"e33d3cf3b6b255e5685549a634841a792baf1625","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e33d3cf3b6b255e5685549a634841a792baf1625"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-18 13:35:30.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdc9c5b7e2ba6a287d9196699332b6b9f76fd351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdc9c5b7e2ba6a287d9196699332b6b9f76fd351"}]},"branch":"refs/heads/master"},"daaf3c07e2cdf1bf830b7aab77cf3700cf6b6253":{"kind":"REWORK","_number":32,"created":"2026-03-18 15:18:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/32"}}},"commit":{"parents":[{"commit":"c9c5d0e7a8049cb9ee452fb524c108da520483b4","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9c5d0e7a8049cb9ee452fb524c108da520483b4"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-18 15:17:34.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/daaf3c07e2cdf1bf830b7aab77cf3700cf6b6253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/daaf3c07e2cdf1bf830b7aab77cf3700cf6b6253"}]},"branch":"refs/heads/master"},"ef7fa29a74bfe46758573b591be4d4a0a4af522b":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2026-03-24 17:58:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/33"}}},"commit":{"parents":[{"commit":"fc5f461f5537bf422b949ea180e54f648adeb384","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc5f461f5537bf422b949ea180e54f648adeb384"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-24 17:40:01.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef7fa29a74bfe46758573b591be4d4a0a4af522b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef7fa29a74bfe46758573b591be4d4a0a4af522b"}]},"branch":"refs/heads/master"},"9c8ce3a7d5be8e1d940bb9682341560fb2c90c43":{"kind":"REWORK","_number":34,"created":"2026-03-25 16:30:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/34"}}},"commit":{"parents":[{"commit":"9e9081538e263171214c616ba2bdbf05acd1bbc2","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e9081538e263171214c616ba2bdbf05acd1bbc2"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-25 16:24:46.000000000","tz":0},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c8ce3a7d5be8e1d940bb9682341560fb2c90c43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c8ce3a7d5be8e1d940bb9682341560fb2c90c43"}]},"branch":"refs/heads/master"},"7f69047e3382592791e6037a0fdd7980ffa8909b":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2026-03-31 14:54:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/35"}}},"commit":{"parents":[{"commit":"c55440e352873c830aa48265a8ba547605258d6d","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c55440e352873c830aa48265a8ba547605258d6d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-31 14:53:55.000000000","tz":60},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f69047e3382592791e6037a0fdd7980ffa8909b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f69047e3382592791e6037a0fdd7980ffa8909b"}]},"branch":"refs/heads/master"},"7b17aad69ee9b200b8e127c26a7790207de41d74":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2026-04-09 09:21:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/36"}}},"commit":{"parents":[{"commit":"9f952e12489d217a50dac88e1d483243a5fd11e0","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f952e12489d217a50dac88e1d483243a5fd11e0"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-09 09:16:31.000000000","tz":60},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b17aad69ee9b200b8e127c26a7790207de41d74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b17aad69ee9b200b8e127c26a7790207de41d74"}]},"branch":"refs/heads/master"},"c577f964700ad947c0d018f82958c2f27ed38373":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2026-04-13 09:47:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/37"}}},"commit":{"parents":[{"commit":"bbd6fd0a37db95b7acdc735829f3b6632586dc56","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbd6fd0a37db95b7acdc735829f3b6632586dc56"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-13 09:46:54.000000000","tz":60},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c577f964700ad947c0d018f82958c2f27ed38373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c577f964700ad947c0d018f82958c2f27ed38373"}]},"branch":"refs/heads/master"},"6eeda9276892eced269040a9123a4f054d599924":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2026-04-14 15:09:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/38"}}},"commit":{"parents":[{"commit":"7a5136f989f362602c8c25d6668eccc133962fd1","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a5136f989f362602c8c25d6668eccc133962fd1"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-14 15:08:58.000000000","tz":60},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6eeda9276892eced269040a9123a4f054d599924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6eeda9276892eced269040a9123a4f054d599924"}]},"branch":"refs/heads/master"},"8f53e42068b063c20c43de087fdd671374d86b43":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2026-05-06 10:50:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/39"}}},"commit":{"parents":[{"commit":"e2f9298b82f41c3f52d38dcdef159b07150e2831","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2f9298b82f41c3f52d38dcdef159b07150e2831"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-06 10:50:07.000000000","tz":60},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f53e42068b063c20c43de087fdd671374d86b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f53e42068b063c20c43de087fdd671374d86b43"}]},"branch":"refs/heads/master"},"50742f41bf2b0444385e1fb03cb9d0e6dbd76023":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":40,"created":"2026-07-22 23:58:01.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/47/932447/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/40"}}},"commit":{"parents":[{"commit":"af3d9af962f4947a4ea7d8e80be84ce7837e06da","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af3d9af962f4947a4ea7d8e80be84ce7837e06da"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-07-22 23:37:13.000000000","tz":-300},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50742f41bf2b0444385e1fb03cb9d0e6dbd76023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50742f41bf2b0444385e1fb03cb9d0e6dbd76023"}]},"branch":"refs/heads/master"},"a65c31894987ad39b2968658ccd59dc56afe3a54":{"kind":"REWORK","_number":41,"created":"2026-07-23 10:27:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/932447/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/932447/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/932447/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/932447/41"}}},"commit":{"parents":[{"commit":"af3d9af962f4947a4ea7d8e80be84ce7837e06da","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af3d9af962f4947a4ea7d8e80be84ce7837e06da"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-19 00:47:15.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-23 09:50:54.000000000","tz":60},"subject":"Object-server: return 409 if x-delete-at is newer than queue entry","message":"Object-server: return 409 if x-delete-at is newer than queue entry\n\nWe have seen situations where object-expirers issue stale delete\nrequests into the object-servers: e.g. when cluster had config\n\u0027expiring_objects_task_container_per_day\u0027 changed, and user POSTs\n\u0027x-delete-at\u0027 updates into the objects with the existing \u0027x-delete-at\u0027\ndelete task, object-server couldn\u0027t delete those stale delete tasks\nfrom the expirer queue containers; and when expirers handle those\nstale delete tasks, they would got 412 returned from object-server\nand then keep trying on them until \u0027reclaim_age\u0027 is reached.\n\nThis is totally not necessary, as those stale delete tasks could be\ndeleted as long as their creation timestamps were older than the\ntimestamp of the new \u0027x-delete-at\u0027 POST.\n\nCo-Authored-By: Clay Gerrard \u003ccgerrard@nvidia.com\u003e\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I07a82025658eeabf462b8e26b7bc352530ba88b9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a65c31894987ad39b2968658ccd59dc56afe3a54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a65c31894987ad39b2968658ccd59dc56afe3a54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
