)]}'
{"id":"openstack%2Fswift~932625","triplet_id":"openstack%2Fswift~master~I5023aa774d2dd56f0d201ed853ba69027e4f1395","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-10-31 20:45:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5023aa774d2dd56f0d201ed853ba69027e4f1395","subject":"sq: probe test for object_updater_stats","status":"ABANDONED","created":"2024-10-17 15:13:20.000000000","updated":"2024-10-31 20:45:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b0e2a259c7a2ef76fa5ce073d2df0aee7a8972eb","_number":932625,"virtual_id_number":932625,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-17 16:23:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-17 16:23:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6af0b5f8c71af7163e43c46c72411c989eba2790","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-17 15:13:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4704f61553268489d36e67850e3d4918560079ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 16:23:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2259e200ffbf40b5aad396f00ad0cf8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04492b4311bf47ff8f555b7525974e84 : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/ca3f54d4674743f7926a3c572b77317c : SUCCESS in 56m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d71231cc0864d1ca46cf1a0149fe6a1 : SUCCESS in 56m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6276b62964b5466c8839afeff93f5b0c : SUCCESS in 39m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea52ea46192f40ff90ad2b82d65af9a4 : SUCCESS in 59m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/167e235483704ba5942f5299e998fd7e : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c0fc57136e24190a95f8822e09ec2b9 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3cb836ca4b04065a6b9d0f0208cc784 : SUCCESS in 18m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5427118937034a7486abd416fe222492 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dcb75e8229fe46fdaa00a2aa024d86e6 : FAILURE in 11m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5108bc1784c0464091138e4fa63dd0bb : FAILURE in 24m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fa11d5b21ee742d08518ed77489513f6 : FAILURE in 12m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ca0cb47a2bfa421b94bf6974b4c34718 : FAILURE in 21m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3a0dfca479f442caafec4831ed0a5507 : FAILURE in 22m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/46f566152f30466ab6be73ab10546a1e : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df2db0684dba469ea1f4090ab8b6516c : SUCCESS in 16m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8b02bfbbeece4fec823826b50d89993e : SUCCESS in 16m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/56aa8a3bfebd4ee3a047dbf28e768367 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/196041c8a55c4735aebe1d651447b05e : SUCCESS in 16m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a0d97f7fd79445529e9169d8f7ea1590 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/61d38de674dc40ae825d8f7625311966 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/371588c280334888a8ee4aaa373070e1 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6827783ec3a64009a80ef96c1f5ff44e : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1244582580594071995fa91603b46176 : SUCCESS in 51m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e9043b4a36a45da81c0370914386d30 : SUCCESS in 54m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97847e56bdec424f8f9c6d1686b6cd8c : SUCCESS in 49m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/975532ba6dbc42a1844b423c0723f516 : FAILURE in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e56246163c2405998920c8f1719af27 : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/587dfff3c9b74963b3fb1ee66d387f2a : SUCCESS in 23m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fe5496ea2ca4e53976950cff6f10194 : SUCCESS in 38m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9a177f375bb577a57ec0142d8e3b521807ac5a13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 18:01:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4b90ce540f34826b738d524271a52af\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e309c5226e4402aafa38f2f855c1e16 : FAILURE in 2h 31m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b0e2a259c7a2ef76fa5ce073d2df0aee7a8972eb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-31 20:45:52.000000000","message":"Abandoned\n\nsquashed into parent 931979: Add oldest failed async pending manager | https://review.opendev.org/c/openstack/swift/+/931979","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4832a2ac93f886394aa4e96302d4e06953213df","revisions":{"b4832a2ac93f886394aa4e96302d4e06953213df":{"kind":"REWORK","_number":1,"created":"2024-10-17 15:13:20.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/25/932625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/932625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/932625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/932625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/932625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/932625/1"}}},"commit":{"parents":[{"commit":"a908bf7c6a2b55f13dfc5e91fcc8a063e71fb84c","subject":"Add oldest failed async pending manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a908bf7c6a2b55f13dfc5e91fcc8a063e71fb84c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-17 15:13:04.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-17 15:13:04.000000000","tz":-300},"subject":"sq: probe test for object_updater_stats","message":"sq: probe test for object_updater_stats\n\nChange-Id: I5023aa774d2dd56f0d201ed853ba69027e4f1395\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4832a2ac93f886394aa4e96302d4e06953213df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4832a2ac93f886394aa4e96302d4e06953213df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
