)]}'
{"id":"openstack%2Fswift~932633","triplet_id":"openstack%2Fswift~feature%2Fmpu~I32aca026570edb406ba90ec1e91b1d86bb70a582","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-audit-sessions-2","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-29 09:36:10.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-29 12:47:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I32aca026570edb406ba90ec1e91b1d86bb70a582","subject":"mpu mw: add POST request handling","status":"MERGED","created":"2024-10-17 15:50:24.000000000","updated":"2024-10-29 12:50:05.000000000","submitted":"2024-10-29 12:47:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"932633-p-mpu-audit-sessions-2","meta_rev_id":"f53a96a4dc4072acd2299026cdd3f811176a65d9","_number":932633,"virtual_id_number":932633,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-29 12:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-10-29 09:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-10-29 09:36:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-17 16:43:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-18 16:12:22.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-10-22 23:37:43.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"94e522600b716d40325a4970a82dc28f7dc69c5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-17 15:50:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"649a51869c76baa6b06dc3309bc67e562df51fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 16:43:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa736c503d24b769e6d100a48d5ebd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba30ff72476448cb6ad6af8205e5166 : SUCCESS in 9m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7479f77203747fbbe3887b77d9934ff : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3eac1652ad1c4c98ab025717e765e53e : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ffe8710976a47d9a382ecd37743807d : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a4a2ef2d6a1e4a90b9d150749dd0d63e : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f380d09297644c38ee8fcadb5bcbd6e : SUCCESS in 23m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3f7a527d02494fdbb750031a9f72b3d2 : SUCCESS in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2a1b95c7233142adb41c700bfbeb1b19 : SUCCESS in 18m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5b28dfaedd734850bc8623a20ef59ac5 : SUCCESS in 17m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4e89c5a819ae41f88027199e861f6a72 : SUCCESS in 20m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/355dfb6bfbf3496d8a1a63f1e6d352c8 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ee28fbadb3334166b3dc5541de7b9f76 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ccd1b3b151434d20857440939fa1edfd : SUCCESS in 14m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e85e9e33c636493b89753ca73a48dc39 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fa77757f64ed4a038731cb2f5c71083b : SUCCESS in 17m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b692c6b16ef645e2932f2fa1dce67fa0 : SUCCESS in 15m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/95596e5e3e6143f6b89837bf1fd98187 : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e03910a57b74297a5de38ec7afc1203 : SUCCESS in 15m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8c33e278293a4f8b85c7c03ffab688ba : SUCCESS in 5m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c911646161448b0824abf6d93b12711 : SUCCESS in 51m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47293e233435417cbfa66a3d04dfca75 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d274abf6a57e49dda941c1b563452aa6 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cbcff7460df849f8b623b2f914ecf523 : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a643c16d604f45d33cdd66b9775871b4b2604fb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 17:47:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a40e2f4b95946eebff3f472bbc1e03b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29fd7b4348864c5d853d203967244987 : SUCCESS in 1h 37m 00s","accounts_in_message":[],"_revision_number":1},{"id":"10aa66d9e31f09a687a8255ce621a323e6bb59e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-18 14:50:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bab2c0fad5127725809a0ee402ab4537df52f44d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:04:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1c6ddb7a864472495a5bf29033c4df4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ea88e8fa7e4bbeb8584a60946bbe08 : SUCCESS in 6m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ce4a2b822ac4403bd2a30f2453a1250 : SUCCESS in 17m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9e3f3550aa74109875a58eabdf29855 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8de85b78edc94e8cbbb69084803f4981 : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f09e38b00024795938c1e521e7f165e : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6736b5f940494398aa46bdd50265ffba : SUCCESS in 22m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51ebbc1e187943209ac3bd41cd5c050d : SUCCESS in 19m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7892060d580345fba94e9f52fbe6f6ef : SUCCESS in 21m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8e9bd6ca07b8432987db9012ee0f5ef5 : SUCCESS in 21m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/437490db5f3d44e08e08ecb834ccbec9 : SUCCESS in 27m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/01edd2bb76f04df69ea3fce827a9764f : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c4201f71626142df84499b9baca6b236 : SUCCESS in 22m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a2ec778dfacc496797f1ebf0ccdf7c52 : SUCCESS in 15m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1cfd0f3dab81450d8f7f1ed5eaf83684 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b9f85490c5a8435f9d4ba266a29b62fc : SUCCESS in 17m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/81ccb8187143451e98ee08b0b4477392 : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b4676ae63f994ce48c0b6b6a41cdacdb : SUCCESS in 8m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff3519a8ca5c44a2a21554a522ed6a7f : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/de50265ed95e42cbbdcdd79e0472d1da : SUCCESS in 6m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b6d569fb5c944efb729888ed83a46b7 : SUCCESS in 55m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f450dc4099b4e03a881e988037b1f27 : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbf2aa353cb4224820102c7b0c321fa : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/887c7f7828c04a5584c63843ae59f5b0 : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b42b894bedf2725377dac37181456fec7b2ba807","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-18 16:12:22.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d3af9af8306bfe6d7d4f0077ee389803d5f09f50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:25:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/967f461bcb0f466ba0de57682f0346a9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a2f1dbf3c6ed40a58da5df83cb60ff7b : SUCCESS in 1h 33m 45s","accounts_in_message":[],"_revision_number":2},{"id":"2928b179560ce6b81ddd67ff39b61cc9f5b5c312","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-18 17:05:24.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fbd0528813ec4c81e5c7be13f3dfd0e20334bb59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-18 17:18:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"fc628fe6d815508058714572999992c7a5b797ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 18:15:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8678a6db464e41c8b25ea374fd23ca07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a51f57948541dc99ab07a13da9d3ad : SUCCESS in 8m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdcd53ea9299429d846b4d50e5e317ad : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b27e44ad748481ebe2d0c9dfec5f5ba : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a50f85b0762640d0bad11bcbfe6e8485 : SUCCESS in 18m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c77d442f78ca4ddeb41c0bdb89b0e291 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0d783950083d4967af819d2acce830a1 : SUCCESS in 20m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f7897567d0d4cc6a3617fdc60e39c23 : SUCCESS in 18m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b0addcc70b1c4da780202d9efcb7a841 : SUCCESS in 21m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5c102a5288684e42801eee4b2d376eb9 : SUCCESS in 20m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/68f45f958c2941b391db45f048075eb1 : SUCCESS in 34m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1a63c527d2e44e0fb11355f2e0b18ffc : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8c6375355a134349b0887b8f2760bf80 : SUCCESS in 18m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97e071eaefc14af7be2f2ed4ac5af62c : SUCCESS in 17m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6bf6efdcc3314357baa8fb9abe913638 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d81aa3f4d0dd40018aedbe96e530efe9 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dd6e96a9f8ef495d9249dcdb1ddd68b1 : SUCCESS in 15m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30788390825844719963d4bbe951b717 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/032c2d0915a0425aa0e63a747184f470 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/43d60831e64c4f418fa39b412f9f333f : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed16831d59ff4b07ae676882c2349760 : SUCCESS in 51m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/284b3ca007be442790e50be611f9543e : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a781eaafe8340aa908dcd92423c2785 : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd7804eed45f4d168542de2d606262f7 : SUCCESS in 17m 03s","accounts_in_message":[],"_revision_number":3},{"id":"364450ddf0d0f4cf692d506820d3cbed79c6be67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 19:36:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62b0525d7c194015b7b9b68952ebab20\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ab19b710b794191b6446b1fafaf5fd6 : SUCCESS in 2h 13m 20s","accounts_in_message":[],"_revision_number":3},{"id":"c31beb902380355891d9b8a66a5dec0a0d95c7ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-25 18:10:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c9514d2e06bf5739fbc5c7494339bd705001352a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 19:30:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0519006fd6f4d54959d9dc7d0856f80\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/431a99f24f504acbb955fcb2225cc50d : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":4},{"id":"4221013c48589518b30903dcd1635116340126fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 20:12:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36ec94c3557847cea5ad4d0d51f8f3f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e4f94249554d8b82fa6adba77f676d : FAILURE in 6m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe4bd2a1b3724f89b47e9a1fb11d0bec : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40e20bd24d7e44b9a0dab9d5722fefbe : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9001f35f506843dabd7dc7caba4f1682 : SUCCESS in 17m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/402bf9da1d7b44f5a4928b8b48bbcf7b : POST_FAILURE in 33m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3675743038ff4c52a334acdc3669cd9f : SUCCESS in 19m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d4af8a281714f6fbbd4c0b5959aa0ff : SUCCESS in 15m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bf231850e6cb40b4b2d664d4e48c003e : SUCCESS in 20m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3109346f557f4d04a8ec53fecc42591e : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/76ded36e09ba43f894f16fe098e300a8 : SUCCESS in 20m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/218a8cc9a2724c9099bd6dd8dded20a7 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/72d6cd800f48437dbde4da0b760f4bc7 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d74e4453f6f44b39793f3fcfe37ee17 : SUCCESS in 17m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/97b20731aceb40268af22275b9e46fc1 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9046b1dc4b9440308f58ad7588c4b63a : SUCCESS in 20m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38d92621b4f343b48637f6fbd9d99062 : SUCCESS in 8m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8eb86d91403546a28428c88b71de257e : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fdb5aa52391b4360a99607d2c67ad1ba : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf19a5c7d61b4a3d99ef4828468912f7 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/18aba39882e949e48526e6dcc5e69b46 : TIMED_OUT in 2h 00m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c916335ebf9042eba70a16b06cb8ac1e : SUCCESS in 19m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbcbfe7004e0435ab735e5855c4226de : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/20acc2efe4d04388ac80e5105d02a3f3 : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":4},{"id":"d6f05afffd1626df73eaac353cdf44d1289ee276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-28 10:27:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9920cee638e43e0e35827ae9050d98e81de8b515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 11:27:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c325d0b301345f48168e2186a55af39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/693ba65649ff482b8797c49ce3824b98 : SUCCESS in 8m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3207c2ec324844c099f5a1eeffa83708 : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffe8a8cb9e1c42a5a61029da67bf7059 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/047318916e1a48a3831660191d57359c : SUCCESS in 19m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a913f4cd69f4a32a2358538fb8ac58a : POST_FAILURE in 34m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c237ad03fb94d8e840ce29d7d91b1b9 : SUCCESS in 21m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5deb189d1eae4b8881ffd42e0df4bed9 : SUCCESS in 18m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6a538254c32941e285c343c1fd98e6e5 : SUCCESS in 18m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5208ed31ee5b4656a80770c49572f5d9 : SUCCESS in 13m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/891ab49653f241babad43883ae8005c1 : SUCCESS in 20m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0bc2841d3f7642b6a18b2fff651ce96d : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2dabdc2c3b484e3984130482cb988aa5 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2efe89c61ad241a9b7ba52f4a0ef1f3f : SUCCESS in 17m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/967de710af1347e89ee40b3fa3fe9ad6 : SUCCESS in 11m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/514f54739330458a83c8ac8c11a082dd : SUCCESS in 14m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c264c63d53694299976ad202564c30da : SUCCESS in 12m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e551a4bed97e403ca3d5b0f1b9f7b868 : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a4a153bd50548ac90c54d3726a0c300 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/da7e71a8303c41dd8b930f1068909133 : SUCCESS in 5m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4990c8a08844e51ba24e0dccd66c20c : SUCCESS in 58m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c92913259b3477e9ac394b9c287f194 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9377023589c7452c8f8ec400b56bf4a5 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6ba8a758a63441e8bd5a5d6be394b303 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":5},{"id":"ad0e7767d66ab560ae72cee17fd38223b11c6e01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:03:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b68f3d6b0bf4ac89d68ddc7c6db510f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c8884218b2244755a9fc18416a843087 : SUCCESS in 1h 34m 29s","accounts_in_message":[],"_revision_number":5},{"id":"77935ad4ca045c92b8132e01a85e273e319dae82","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-29 09:36:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf4bb7be8e96f5c760bfc11aef66d79687dd2e89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 09:37:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"74fa58f74ad001b459ac6da5b491d53508d91f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 12:47:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96a1f467a3264707832eaa3be18618d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81419b9511784663ab65a8b052fb30cd : SUCCESS in 7m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/91916bf7df274f59a27b4bac698a247e : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3d5bf0dbcc0446f790f3f49dbedcc2a8 : SUCCESS in 19m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5d0434535b64dd981fa7876d14107b7 : SUCCESS in 18m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1dced9a351e0469cad5f63ce66e5bfde : SUCCESS in 21m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c27daf4df79a46049fdb42504b18d9ee : SUCCESS in 20m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8862084c8f50469fbe43cd859feae535 : SUCCESS in 21m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7841b3c8f3d942e6bb6e27267fa0cbec : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/27cd666330904e8a878be2d8d901e7bb : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/803621f540934e6ea659b8e513393d13 : SUCCESS in 17m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/808b582edce64955bbb155d151fa0b9c : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c7c7ac008df46efb9113d6afe42b76b : SUCCESS in 14m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fd9d22e09f5b4f42b306ab8b8b6eee72 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6c074b95a4884c35be75bd219f74de85 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/938b535587c04f149b3ab0304580919f : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1928c326402d45b1a6fa235394bff8ef : SUCCESS in 53m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/783457041baa48b6b1c13c743a100dd3 : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f47cb3a1bd5d457380a08d30156ad9b6 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c318bf16cbe846d2927d96ba1bcdf450 : SUCCESS in 19m 08s","accounts_in_message":[],"_revision_number":5},{"id":"6c173797601c2c07459db7ca9a8330b9c4fcdb8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 12:47:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f53a96a4dc4072acd2299026cdd3f811176a65d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 12:50:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa594e5d8749438b81225d98dc0308d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35cd46ce3f0246fab79e19160e7efb14 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/19f769382d7245ee93dc588a3aed194c : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ff5b399b4a179f3ba6f3c8fd85cf0286272d2eae","revisions":{"54a06ce99cae1bd75d61c34573375517eccf715d":{"kind":"REWORK","_number":1,"created":"2024-10-17 15:50:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/932633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/932633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/932633/1"}}},"commit":{"parents":[{"commit":"aeee0e3540542b56f191644eb6284217e7522de5","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aeee0e3540542b56f191644eb6284217e7522de5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-17 15:42:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-17 15:50:04.000000000","tz":60},"subject":"mpu mw: add POST request handling","message":"mpu mw: add POST request handling\n\nPOST requests to a user-namespace mpu object are redirected by MPU\nmiddleware to the manifest object using the location header returned\nby symlink.\n\nChange-Id: I32aca026570edb406ba90ec1e91b1d86bb70a582\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54a06ce99cae1bd75d61c34573375517eccf715d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54a06ce99cae1bd75d61c34573375517eccf715d"}]},"branch":"refs/heads/feature/mpu"},"8963b5016f2a26e7cb3697f38beeaefa6723ee6a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-18 14:50:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/932633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/932633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/932633/2"}}},"commit":{"parents":[{"commit":"564b57835f81a5726a5ecda92abad1d3a14917ff","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/564b57835f81a5726a5ecda92abad1d3a14917ff"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-17 15:42:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-18 14:49:41.000000000","tz":60},"subject":"mpu mw: add POST request handling","message":"mpu mw: add POST request handling\n\nPOST requests to a user-namespace mpu object are redirected by MPU\nmiddleware to the manifest object using the location header returned\nby symlink.\n\nChange-Id: I32aca026570edb406ba90ec1e91b1d86bb70a582\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8963b5016f2a26e7cb3697f38beeaefa6723ee6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8963b5016f2a26e7cb3697f38beeaefa6723ee6a"}]},"branch":"refs/heads/feature/mpu"},"b3b318d3e57facadca4d2cd0e08e3d189b1c1d9a":{"kind":"REWORK","_number":3,"created":"2024-10-18 17:18:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/932633/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/932633/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/932633/3"}}},"commit":{"parents":[{"commit":"564b57835f81a5726a5ecda92abad1d3a14917ff","subject":"mpu: Cleanup MPUSession class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/564b57835f81a5726a5ecda92abad1d3a14917ff"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-17 15:42:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-18 16:55:11.000000000","tz":60},"subject":"mpu mw: add POST request handling","message":"mpu mw: add POST request handling\n\nPOST requests to a user-namespace mpu object are redirected by MPU\nmiddleware to the manifest object using the location header returned\nby symlink.\n\nChange-Id: I32aca026570edb406ba90ec1e91b1d86bb70a582\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3b318d3e57facadca4d2cd0e08e3d189b1c1d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3b318d3e57facadca4d2cd0e08e3d189b1c1d9a"}]},"branch":"refs/heads/feature/mpu"},"7a0f2626dd3338b2666bc3c709cbf64f2021d66e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-25 18:10:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/932633/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/932633/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/932633/4"}}},"commit":{"parents":[{"commit":"e8b7f1f4abb7707153ada11d6f57d7ebfcb04059","subject":"mpu-auditor: audit sessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8b7f1f4abb7707153ada11d6f57d7ebfcb04059"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-17 15:42:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-25 18:09:54.000000000","tz":60},"subject":"mpu mw: add POST request handling","message":"mpu mw: add POST request handling\n\nPOST requests to a user-namespace mpu object are redirected by MPU\nmiddleware to the manifest object using the location header returned\nby symlink.\n\nChange-Id: I32aca026570edb406ba90ec1e91b1d86bb70a582\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a0f2626dd3338b2666bc3c709cbf64f2021d66e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a0f2626dd3338b2666bc3c709cbf64f2021d66e"}]},"branch":"refs/heads/feature/mpu"},"ff5b399b4a179f3ba6f3c8fd85cf0286272d2eae":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-28 10:27:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/932633/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/932633/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/932633/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/932633/5"}}},"commit":{"parents":[{"commit":"b9552a1741c2b580e4241f2c5cfd4916c336dc21","subject":"Merge \"mpu: probe tests: add a helper to check mpu resources\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9552a1741c2b580e4241f2c5cfd4916c336dc21"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-17 15:42:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-28 10:26:42.000000000","tz":0},"subject":"mpu mw: add POST request handling","message":"mpu mw: add POST request handling\n\nPOST requests to a user-namespace mpu object are redirected by MPU\nmiddleware to the manifest object using the location header returned\nby symlink.\n\nChange-Id: I32aca026570edb406ba90ec1e91b1d86bb70a582\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff5b399b4a179f3ba6f3c8fd85cf0286272d2eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff5b399b4a179f3ba6f3c8fd85cf0286272d2eae"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
