)]}'
{"id":"openstack%2Fswift~932935","triplet_id":"openstack%2Fswift~master~Ibab68b459c9510148b67ec1aecab73db11192afe","project":"openstack/swift","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-30 19:24:17.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-30 23:15:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibab68b459c9510148b67ec1aecab73db11192afe","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","status":"MERGED","created":"2024-10-21 20:11:01.000000000","updated":"2024-11-30 23:20:02.000000000","submitted":"2024-11-30 23:15:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"932935-migrate-to-noble","meta_rev_id":"c8f25256b8a10b4e674691a1651b679f32403441","_number":932935,"virtual_id_number":932935,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-30 23:15:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-13 19:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-13 19:38:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-21 20:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-21 21:37:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-11-13 19:38:56.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"04cf9cb635c500b9a079f1608c8c7617b2e60310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-21 20:11:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d36740dbf9010d191ba0b341636f644b370dc50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 20:14:21.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  while parsing a block mapping\n    in \"openstack/swift/.zuul.yaml@master\", line 746, column 7\n  did not find expected key\n    in \"openstack/swift/.zuul.yaml@master\", line 782, column 8","accounts_in_message":[],"_revision_number":1},{"id":"098bcb9857d76d2cef027b9e5882e47bd38e1ac1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 20:14:25.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  while parsing a block mapping\n    in \"openstack/swift/.zuul.yaml@master\", line 746, column 7\n  did not find expected key\n    in \"openstack/swift/.zuul.yaml@master\", line 782, column 8","accounts_in_message":[],"_revision_number":1},{"id":"03511775dd870790342015f8a9ae719a8d135e87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-21 20:21:15.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a04c7851d1c396e3788a5149da1a7e123298baea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 21:29:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6835bc31805420fbcd0b03490e516e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f6e9163d134803ba224e6af45b2bbc : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/a1e0d5a2313e4904a5606f94a11b8856 : SUCCESS in 43m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a935c5927e944e88acb905c45cb669e6 : SUCCESS in 56m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d541824a3f214a269425093094e812c0 : SUCCESS in 54m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d82629f8f054cdab5a1b252f0d542a7 : SUCCESS in 1h 02m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c3a7d4302584c2398b33e4071bc17ae : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e4bd06e0f7c466bbf7fe7b5b7d68411 : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/726bcfc276054fdfb43d0464b3ddd783 : SUCCESS in 15m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/776fcb9e7fc0459682ad37cc08e68a08 : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f1049e75170420e82f78c5e801b8a06 : SUCCESS in 16m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ecf6db4f02ea4e159d578ed241b6df35 : SUCCESS in 19m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fa425504543c479b85c29f3c230b60c8 : SUCCESS in 17m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d00344bbfa8540b5a9aa074c48f8a4dd : SUCCESS in 13m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/38ed1b821f0a499c893a192f8e8eeb3b : SUCCESS in 30m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8e3fc26bdde0454ebb1a46985b77c54a : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9225bb4d78b947f5afa636c25ef975f2 : SUCCESS in 10m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c117478c4d13446bab5f6de6f2ada504 : SUCCESS in 17m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8af9cc8c663c4f398e367b89efc77531 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/12a99d4151f647b89c866889fa91e850 : SUCCESS in 14m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a3929aac878141d58a193ace9f135bbc : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e5f1e8eaa0848a0a92fbfa761756a16 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/95a88c4a9feb44bda1a4f6654e061931 : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e1f039c857f4c56822d4573bb42104d : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6437875d08494b4798cb254eb536ff4a : SUCCESS in 45m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2028d40fe54f4da98b2823d09e7fa04f : FAILURE in 9m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/63a1b8f88a734d2195b2cc4217807946 : FAILURE in 23m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a732ad8e4d640cd8acedc44b6d48636 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85585086eb99498797d45ddeae4e1773 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77a0c75dbb1d4c3c8ba5c2575e5588b0 : SUCCESS in 15m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c749df3993b4bcfa2239176cfee0b9b : SUCCESS in 32m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ec92225918c47deb00764657390d03b : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d8bcf2554083013989f12deef96cbe4d096a79d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-21 21:37:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b55631f84100f169d7c34f1a0611f10c307741e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 23:10:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f65aae4f549c49b3bb1e068dd10f5375\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ecdea557740e46c6bdfc7cb55577416d : SUCCESS in 2h 46m 15s","accounts_in_message":[],"_revision_number":2},{"id":"9a50a394172b4c613902bfa89ae866b22bf3e76b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-22 03:24:45.000000000","message":"Topic set to migrate-to-noble","accounts_in_message":[],"_revision_number":2},{"id":"f9b2c12f22862b48f0c784ef60f87506c5e2aa37","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-12 03:28:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"baf1d87175a7b8067d1ee8cac86a05d329606a35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 04:39:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4dce251bc554ee5bfa723acd084bb00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6733385ae5fb4bdb86cd6f161896d3a0 : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/08bcd9403f3d4ce1a59edff107069aab : SUCCESS in 57m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82bceb5f2df64f98b476106bdf2adee0 : SUCCESS in 56m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6c58fec25b04c64b0b5525b34c8b91b : SUCCESS in 48m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df9aaf457801408b8312fca21e1c650c : SUCCESS in 1h 02m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a26d75ac3c2c47beb7db25e1d9f5d544 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2657c118c3440018cf27d80c18e3830 : SUCCESS in 11m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58a8c6e689564e2db33b78fe30fb024d : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c13a8e6bedeb4b31bae9e589fb075826 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2f81800939b14fb690d564e135115901 : SUCCESS in 9m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b23dcf75f1314c448644a976a9707cd8 : SUCCESS in 14m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/08e6aa2689ba4b37920a84ce25e97da5 : SUCCESS in 21m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a577254f48aa469180220008b7e41bec : SUCCESS in 17m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c79f28693f90418e8a673f73eddb9a3a : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a4d8ca81c394a39afcf10e5722eff2e : SUCCESS in 15m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f7e9aac84c1b43659d7992632921df64 : SUCCESS in 17m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4dad8e5fdce443dabc05540b20ed2ea1 : SUCCESS in 13m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/30485cfd22b442858ba93aa2e19f951f : SUCCESS in 13m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c2cbed141ec44c7187ca355a83d340ca : SUCCESS in 15m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5880bda2188b422aaed98519020e040c : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/895f9bf3654f430a8460970029247787 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9c29919be4b544649e6cffff02a4b238 : SUCCESS in 8m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/77d200786ff845139a748c0e35d9ab49 : SUCCESS in 55m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/848b57e51f614fa19561d43d2cda22f9 : SUCCESS in 50m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbfefb93619b4763b195228451020b50 : SUCCESS in 54m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ea368e40bb04f7d822b38cd170c069c : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c87a8e8db4b4b2a9dde2d1f06aaece7 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52928e123b064cfcbbcfdf750522e28e : SUCCESS in 24m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67dc28e25ca444a1b6375865b95f4ddb : SUCCESS in 41m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c46f36962b814052a2484c772973c4a6 : SUCCESS in 1h 03m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9bf3486d22ced0754b4feccaf27d39034c7dab34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 05:16:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d89432d4cb8b47ff9bffe328dba0264b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e034c18524945669d493152e327bde3 : SUCCESS in 1h 44m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a72e353ab5c4adf00db7e330c336407a48543e0b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-13 19:38:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7f7c22584320c75d77af4d69dae9c28243298ca","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-30 19:24:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86b70394d05e97ffd39f28b94baee997bfe4b54e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 21:15:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f3021fe5ab9490bb20401ac752fe67f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8729f741af264ad892b214cc47fb5aa7 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/cc23c1d418024cb0b6af488e170c412d : SUCCESS in 45m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/feecb7abefcc40e7913b9a79ed52280b : SUCCESS in 53m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e714b81290c54f848bfe2d6577c26bb2 : SUCCESS in 37m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87411cb853b84238a9ac41b97c9ffa04 : SUCCESS in 1h 39m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a3dd20220ee4d33ab8e33687b5aa641 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54953629ff5341b5811949fe77888ce5 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8550ac93587847c8b4c2a0be98fb1e95 : SUCCESS in 23m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fc8db1fbea374065908a76aa94e298f4 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a4658f99bfa433bacdb2aa1f2abd522 : SUCCESS in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0aa123041db84b548e02ccfbdfcaa899 : SUCCESS in 9m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/90bb5f082deb4ce9a269eb51504e732f : SUCCESS in 17m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4e608fc9250b42b3a0add4a2f7aba21a : SUCCESS in 21m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c598ac6e75ca42e0859a79300b3738c4 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/867f426170dc4d76bc75c68e4e6a2e69 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/02feb9d1715a4670bb907df935e575e3 : SUCCESS in 17m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/57abe3c5da264cca8991cb531d028a12 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4e404e31eb3a4429a5a047cfd11227cd : SUCCESS in 15m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5ec8f560d4514a1f94b240119b6941e9 : SUCCESS in 15m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3a6727cc284c4b2aaa58748417f071de : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fb0e77772e1948759d00f8818b3567ad : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/511518c0397347f896a6d80b7152ca74 : SUCCESS in 4m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9a5d41c3cd641948a94f159fc11b9eb : SUCCESS in 53m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/73f87d8c498f47d48d8162e260706b55 : SUCCESS in 51m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c4f7429edd04bae87d8bd47afc4c839 : SUCCESS in 31m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8da6df1915d4e59816e08022989e538 : SUCCESS in 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f418a5d36a344a39953029cab7d134d1 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6fd1319d92e74e498eab1d26f0f3e218 : SUCCESS in 15m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29dac8b74d8f4aaca9bb744997df4c75 : SUCCESS in 40m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41371f61b56d4a17bd9b7e22bc94a2d0 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":2},{"id":"12f0377ef9c626b9222865b171f53c57e05c48e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 21:16:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"982e550f07a6c6953d61a9192e787175135ac061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 21:48:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34ba56eda1794f59878a9b4533aefee3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/33e51df5af6e4b50b5ffc7ec8ab26c09 : SUCCESS in 2h 08m 08s","accounts_in_message":[],"_revision_number":2},{"id":"11d935a06b406ebd97c769671fbb99cb1a544750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 23:15:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec3acec66cc448d8aa079a398a2cc5e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b19e094958442caa9c54c4397a94bfd : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1fb208ac6446464b8aba8445fa20df0b : SUCCESS in 1h 02m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37bdf3995e0b45ce88de2cd02f9006c0 : SUCCESS in 1h 04m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5eacdefde3c428089ab715703ca6ee6 : SUCCESS in 36m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bab4ec4376a04011b719b4bdde21de03 : SUCCESS in 1h 06m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/98dc32feaca24f77b23426869afd4281 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/815ae823b81f4f18b6c581bf10d59c6a : SUCCESS in 16m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a75b62cbd55416ebac8c0a5fc62640f : SUCCESS in 14m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/61e1a2496d324d0eae54b846c2969526 : SUCCESS in 16m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ef399e63076496ab04af92568a6fd8f : SUCCESS in 17m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e78d1743393e49ae95d05931e7c047f2 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa9ae1abc60f4770a2375d863e3577d5 : SUCCESS in 10m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6184d951b174df0bf7ccdad4d433a37 : SUCCESS in 14m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0eb93819e4f444c097a3dd434687919e : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6f783ade50f549e9b0a6776e52038371 : SUCCESS in 12m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4fd545c440b4487e9f53bb1d55a137a5 : SUCCESS in 9m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/554edd6da967420a8b88497f62d2be27 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/945ab87d845240749f84a3e7c79405a9 : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0fa83cff9dd4d148a24c5eb54b8b350 : SUCCESS in 59m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d19a5db8e8a14687922c87ee90dc0b9b : SUCCESS in 52m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4307bfc8cea84666944c9a2cbaa352ae : SUCCESS in 53m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80267ce1071d41be8a0726a699f7e9c1 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a17fbc8a6bd74136bf62161bfa91724e : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9bf4401e9acd4cc19d14d0d2933a88a3 : SUCCESS in 10m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/721b2c609f654c819f42030238215c7b : SUCCESS in 52m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6ae09569b5e419e8aba716759b28520 : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":2},{"id":"e170bfa1f9e7f8de2049b3f8b2481ce2eb4af732","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 23:15:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c8f25256b8a10b4e674691a1651b679f32403441","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 23:20:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ab2a936d9624cdb920d1a21b41ecd46\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a517c78cda2483dbf3350d9182fdf95 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4bbd584b593d4f5985505a64d462e006 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"92929705482e2cf7b5297d027a71aedb92b291f7","revisions":{"c90b09b3fbca56650cda137da5b0c2326c1f9f03":{"kind":"REWORK","_number":1,"created":"2024-10-21 20:11:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/932935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/932935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/932935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/932935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/932935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/932935/1"}}},"commit":{"parents":[{"commit":"4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82","subject":"fix x-open-expired 404 on HEAD?part-number reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-21 20:02:38.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-21 20:10:59.000000000","tz":-420},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Ibab68b459c9510148b67ec1aecab73db11192afe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c90b09b3fbca56650cda137da5b0c2326c1f9f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c90b09b3fbca56650cda137da5b0c2326c1f9f03"}]},"branch":"refs/heads/master"},"92929705482e2cf7b5297d027a71aedb92b291f7":{"kind":"REWORK","_number":2,"created":"2024-10-21 20:21:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/932935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/932935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/932935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/932935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/932935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/932935/2"}}},"commit":{"parents":[{"commit":"4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82","subject":"fix x-open-expired 404 on HEAD?part-number reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-21 20:02:38.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-10-21 20:21:15.000000000","tz":0},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Ibab68b459c9510148b67ec1aecab73db11192afe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92929705482e2cf7b5297d027a71aedb92b291f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92929705482e2cf7b5297d027a71aedb92b291f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
