)]}'
{"id":"openstack%2Fswift~933061","triplet_id":"openstack%2Fswift~master~Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261","project":"openstack/swift","branch":"master","topic":"refactor-test-slo","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-02-14 17:52:38.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-02-18 21:10:46.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-18 21:10:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261","subject":"refactor test for x-delete-at w/t part_num and x-open-expired","status":"MERGED","created":"2024-10-22 20:53:05.000000000","updated":"2025-02-18 21:13:37.000000000","submitted":"2025-02-18 21:10:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"933061-refactor-test-slo","meta_rev_id":"bee2ca53a2c32a37d7ae88d856b0b7a4fc11cfe1","_number":933061,"virtual_id_number":933061,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-18 21:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-02-18 20:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-02-18 20:01:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-22 20:55:07.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-10-22 20:55:07.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-10-22 22:10:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da40c18a1625f4a2cb4ee96e5b802c205c60cc7e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-22 20:53:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"176acb8235c5339a5808c726e061126efdf559d9","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-22 20:55:07.000000000","message":"Patch Set 1: Code-Review+1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"b1d1d5536e319d16cf62bef01f6bb117ec691715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-22 20:56:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"77795ac7ceb81f35b146d085d9ff3aee7e3afb2c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-22 21:28:11.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7891201afeed72332a342af7c5ec7312ee177e46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 22:10:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3d0a98bb1d84c3a96446c149a9f35a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd4581622c5479591af50f5011d5c0d : SUCCESS in 7m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69dfb40634c44f3ab3e92a6a3b6b42bd : SUCCESS in 1h 07m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17f4b0844c0d407c99d29ecc1e19f909 : FAILURE in 14m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bb42634c811449c99efa31545a7967a : FAILURE in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8ecf5966eda4d689118c9907ca4133c : FAILURE in 19m 11s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6116efab998c4ccdaee12c0db8ba08a4 : FAILURE in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f52788e5dc14188af11700d8b34c883 : FAILURE in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a33fdef514c14f83bf43933bc6fd8257 : FAILURE in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/de99d5107ba144a98f2a3ce3555c2b91 : FAILURE in 14m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0d7781d986a6400d89f2b17c3a573641 : FAILURE in 14m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4a5ba17e8a034b9dabe181745e4162d8 : FAILURE in 15m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6fa3e0e7ccd425697457511a5f74010 : SUCCESS in 46m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0bff182e12b54c6ba09d033ac87bfacc : SUCCESS in 45m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/167b5529f66740898c9322690fdbf9b9 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/614cf62ac57b4345a93169983bd43eb8 : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":2},{"id":"906ab5903890b40aad22c53552d5ff0e10de6f28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 23:38:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afbb7505bc2f4883b5f415cf949814f6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3d25415e8c748d98edafb921177f7a8 : SUCCESS in 2h 40m 40s","accounts_in_message":[],"_revision_number":2},{"id":"d22174e6ac98d2fd05e402065c2cb171c8f37c0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-23 15:40:26.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3d598f00e127668a48ed9e810b38ae97f2f72fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:48:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da0124da3f35430d82e1b5d91155b75c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a051c80bd2248ce9b6a13bd90012a3f : SUCCESS in 10m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5709a6173f2448aab20d6209584379f0 : SUCCESS in 51m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/557b702c582b40e6b5717f601c926c2c : FAILURE in 21m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4eb0eaca15e4cc3a8dfd06dd557ff69 : FAILURE in 24m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ceb31e4cbbe41d8bcbf02c7a3267eda : FAILURE in 12m 01s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/610c13af85554a4097be11d149a88a5c : FAILURE in 21m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/801d5ad013b44282bfaa255e2eabde09 : FAILURE in 22m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8337deb63b484314aa2b23145acbe82c : FAILURE in 12m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7570600ae3754fd1b7074287c7d2a531 : FAILURE in 16m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/407a08108ffe46039aa5c7897720b9f8 : FAILURE in 9m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ccf85a31443d4fe1a92d89e245a66037 : FAILURE in 16m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5223af15e26492b9792053b5e79ade5 : SUCCESS in 32m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/addbcb25b55a492e8a53f23945ba88b9 : SUCCESS in 49m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/697c1ce6e94f4f42a30bff77ffc4122e : SUCCESS in 3m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a40df94798c4e93909fef04187c7bc8 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":3},{"id":"b64f58fa485a4634b55586393c996861ae508375","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:14:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea0539e44a36447b88493e5cc019ae93\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/338f3bfefe9649b8b0d4558bfc3d3a85 : SUCCESS in 2h 31m 09s","accounts_in_message":[],"_revision_number":3},{"id":"d7f32cc9c57ce660d1c17dc93b3a99ef4670d3df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-23 21:59:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"57343600889d65e6c9a70fd3672178aef9a45310","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-23 21:59:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2a4efaa565cd8b70748bf743397b1cf72cbe1b76","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-23 22:00:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4d276be18006210b1034acb840db3c41ea8772c0","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-23 22:00:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e6043526b217ad5bd75dab9f6410f8b06bd85d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 23:22:24.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad8aefbf3020461bbcf57dc1dd5ee24b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc17e4d5f8e435b998d541f4b895d49 : SUCCESS in 8m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1dc0ed5d2b1846309c1510abc0b75dbc : SUCCESS in 1h 01m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df8360e2718c4bc699e4391b0d48d479 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cf450537bdf410a8cc4a500af651c15 : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44879145e75342e3b05f61e61022ba31 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75531621647c460a83f264ada9da197d : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/30dcc7dce92d41188062d274ba282ccb : SUCCESS in 16m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df7a35b75eb44ef583b9426e41673a57 : SUCCESS in 17m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c938fab56b114ed78b252986358248f1 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/76c010b252214f02bc126f6bc4447663 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b5c112f1d6cc4cfc832a291a540c5cf2 : SUCCESS in 15m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f422763013b4bed9c8498d15d7f630f : SUCCESS in 27m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/09876683881d4905a47d6ac4c0a498e0 : SUCCESS in 51m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9e24549883e4870ab8fa576add70859 : FAILURE in 3m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dbdbf6a6be0d42af8acc23e6651d6dde : FAILURE in 1h 10m 42s","accounts_in_message":[],"_revision_number":4},{"id":"75ada50c4708603d9c52cf0dda47488fade4eb94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 00:55:25.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/954a60027a924e86bc12e74f8aabcb53\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad9405f251e84bfca1cbb7af2b2ebe78 : FAILURE in 2h 27m 34s","accounts_in_message":[],"_revision_number":4},{"id":"8247c2c03a895a2372d7ad0aeb8bedd27fe375d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-24 17:54:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c4a6f29b462fd614bd0e0391662b71591569a4c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:04:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a71fb7eb9d584fc6ab6804eb0f4724a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364e7442f30c4171bc8cf38222da7f61 : SUCCESS in 10m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3bd05949cc294d3d8afcb439ac64c9ca : SUCCESS in 1h 02m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aef462b895914bff9d28a2c380ff2e2c : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d6895ca55a2403191cb22925a0065c4 : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d984c6cf6833430ba837ca695a0c931e : SUCCESS in 22m 30s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14cfa03ca69340e8b955201eb8f37743 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/61abce0aa85b4eb6a064d72fc650eb7c : SUCCESS in 16m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37be7d5b90e04352802259d86c2ef9d7 : SUCCESS in 21m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3e6281630c16490bb44a611cdcd56c58 : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d29256e82cb94a9b8227285a33c714ab : SUCCESS in 8m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d904d753652a4dc1a12c48b3c1cf4878 : SUCCESS in 14m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b02472b58814fdf808a3c40dd1dda58 : SUCCESS in 41m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1a21321cc4224b88ae37e09696e53df9 : SUCCESS in 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f6d2c1ec13048e289a5c8969bbdb5a6 : POST_FAILURE in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21071ebe4fce4810a94c261e8265ff65 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":5},{"id":"aa4b75e2bfef290e9d0dbdcb53b2caffa5c0740a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:34:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/147d939b22a74a43b4455cfb864f894c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1adfc8f31e294ab98881697b1e5bcb97 : SUCCESS in 1h 37m 20s","accounts_in_message":[],"_revision_number":5},{"id":"9ae0641fd9731b2c96e7c253c3ba8292c380822e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 19:09:45.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"dc6f41862b603c3347ca6e92d437df78ebeba4b9","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 19:11:30.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6769f2f4e1c220f855330466511e9160d5f9d1c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 20:21:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78e9ed09f8034e67b34598c91497ccfc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b5b0dca3884a19af8fd3fc39583bf0 : SUCCESS in 8m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1bcdd25d78f64608a7eace9347cf50ec : SUCCESS in 1h 04m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a09c685b53c34d08a74b562d714f2bf5 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbaed3c5c6984a35b7fd7ec91fe340f6 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36a370c0fd854d41ab35bed733520990 : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b4593b8dc804d5f8bb0bc1a1bb79393 : SUCCESS in 22m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/526676817ac34976a575d673fc2e9e84 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/abbfe2ea14e2435dbf5b4f9e3a52d5c0 : SUCCESS in 15m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ed13bdea2c7f41c887438c1baca17d8c : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1a6142b15e59499b8dba500949228457 : SUCCESS in 13m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a5658e270d354c2080b5204938935ecc : SUCCESS in 17m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d496608910eb4ce09a23535ade075b33 : SUCCESS in 54m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6bb9ea93e6b446348f2139a5065e55f1 : SUCCESS in 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add40df8ddfe4dcd806eae16513d1cfe : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7301da97838c484b8890b593af278eb6 : SUCCESS in 20m 10s","accounts_in_message":[],"_revision_number":6},{"id":"e1805dad6de57e1b64b2a637329fa9a448ceb0ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 23:18:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b273479be24f4347a1c1f269c91c738b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2eb871233a90442b84a967c8b3725421 : SUCCESS in 1h 37m 07s","accounts_in_message":[],"_revision_number":6},{"id":"33cb15a062421ec6af968542095644d2c797061c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-24 17:29:23.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e4d66e4d6be0f909b7c2bbcf8aaf0666e14cb7f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 19:24:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff6cda13da2a4edd882746ebb01c05e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b95e3029d04337a91790a5a4dd5fd2 : TIMED_OUT in 30m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a3d6da58286485d954652d060950f7f : SUCCESS in 1h 41m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0a6bc3ff081425cba80bbaf6753806a : SUCCESS in 30m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f76c81294ab6462ba83ff4e8b29626ff : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbafe0f9e46a4351b8888bf5fb0beb3d : SUCCESS in 18m 04s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/94c81354571b40158d67d8399d7d2a1f : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/201bc95acd674aa19656cceac0b335e8 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b5e12672ff3f414eaa7cb65a0ab63412 : SUCCESS in 16m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/458628f6caa5418ead60294206f228ab : SUCCESS in 26m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8470bacc0b1849bbad4ea8faad716eda : SUCCESS in 1h 02m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a85f77f279145a2b553d10da568a86f : SUCCESS in 12m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cf07594747f4dd2a2d164082a6a3b3e : SUCCESS in 21m 59s","accounts_in_message":[],"_revision_number":7},{"id":"cbf0680183374eb05efa8521bb97155c4674ecef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 22:20:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74a586a41ad84ba5b362ee3a585c2a10\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/17026514dcb54cdc80c68c5df42d85cd : SUCCESS in 1h 34m 48s","accounts_in_message":[],"_revision_number":7},{"id":"0abc1cd3fb48e3d70ee4b4d897cb95b54634e97e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-25 16:03:37.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"4cf5d723bc3cf3e79326c045b2510c829da6d752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 16:55:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/976aea511aaf4664b23f2839c6bff670\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7bb86562d974eeab3a9649ba2e3133e : SUCCESS in 4m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/acb0e57c4fca4953ba7fe88ed55af164 : SUCCESS in 32m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/961534582e16451f965753fc8b187ac4 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7edadc1c1d440b09c531a5d8c813d0e : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6e68df359d1465493271e2b55c12beb : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c2b7d7342a9a4ccbaa230c21574e9fce : SUCCESS in 12m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e2490157ae7043c39cf173e2cbb492ab : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7bab38d80d9a4056a6c085c0d1687a7b : SUCCESS in 12m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce10fb9e209d4c2d8f2cd8d49c350358 : SUCCESS in 46m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0e2718ede674e039514a734d4159e2b : SUCCESS in 29m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d3afdfc085424a99ebc1c1880d8529 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1fb628c54cd14b0f8b30781dc734f890 : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":8},{"id":"7294ae2ec6abfcf47a5a6a482663493a9a10c756","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 17:37:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9ece27f5b744548892b694db4874522\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe1b3fd153e84c178670f056406e9b21 : SUCCESS in 1h 31m 42s","accounts_in_message":[],"_revision_number":8},{"id":"793c3be55f74edaa8885db86a43eb2d83956368a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-28 20:14:13.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"f5ca050466890f427fe09c036fa3cf94198235fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 17:39:26.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"96045b4686be43b710d78d5ab6d5f8b7506dc40d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 19:00:27.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/943e58834f8d495588fdb9b08f73fc7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7b740563c345f783f0c8f4316182dd : SUCCESS in 8m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9ae0870254f4877ae9ba7c5566ef0d6 : SUCCESS in 1h 06m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2e298d8e34043749d4ccf6366287da1 : SUCCESS in 20m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a8927cb05f54fbb9eabac8bed88b5f6 : SUCCESS in 21m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb44092bac9448e8b93513183295c4cb : SUCCESS in 22m 21s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c31c369a89f44bfaa50cc9fb81930cdc : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6f0aa5c1f7654ba4b691ee1e518f7733 : SUCCESS in 12m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c719988cd2e244178e13cd73c8a3f88e : SUCCESS in 13m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/493c6ef78db94e6bad888c8715b3edc1 : SUCCESS in 33m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ec7fb8082ff4756a9385b053ed69f92 : SUCCESS in 48m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c64a04b2684565ac6b1227829bbb59 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c168bbf1d0174bc8b198a8e202614a3e : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":9},{"id":"10972bdcf5f05e0d7d125cfd63db59577f0e5f9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 19:00:35.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"29b8c6e2090d78859016d73e34e8ffc7cf324470","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 19:01:15.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"bf1be0d4a3840c71304c3ae38e2dbff6b4d45f56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 19:39:38.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4b05393dd2a4c8e43bf0d3112450b455eba0a6ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:37:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ce37736b982403a893f8ac07bcd59c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cf44a16d53b4d4a9197d62553575d07 : SUCCESS in 8m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3987dd83682642fda0b0ed1fb5f5e558 : SUCCESS in 45m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fc0ae8007f84da99dd226403fac8c69 : SUCCESS in 25m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d1da711d6f1425992606099ff91b94d : SUCCESS in 22m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78e922ab5cc349fe8ad2b4a35f0ef7d0 : SUCCESS in 22m 45s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/eb3c1353110945d5adfa7c1e4b3ea2d9 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eb5bfc522e134109a9334b9266d57537 : SUCCESS in 9m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/88ee9b05afa840f088f36308114222e2 : SUCCESS in 14m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a5c49aed4cb40ecad33760eadfed81d : SUCCESS in 26m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0824718f6bb9418f802bdbd338fde79d : SUCCESS in 57m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1718097f4b23494fae20a5ed57f4697a : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/af8d88e934bc4307b6734f522fc09cda : SUCCESS in 20m 50s","accounts_in_message":[],"_revision_number":11},{"id":"d3122c6432ed4fb748ee22e2fc265d8ac8bd4d52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:56:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2fc0a6d374b43068261a58703e7adfa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac9441aeea154e01b261be15a96fd7f6 : SUCCESS in 1h 15m 23s","accounts_in_message":[],"_revision_number":11},{"id":"dc9c36bd4f492f3c7d76f03bd90d4940b45cde98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-14 00:27:38.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"2f09f47f7ab1926528e93c4dc7117fb52ec69606","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 00:27:59.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad521cd2746d474a91cdacae4fc01203\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e8f6de1d8c984ddeb4bbc9d4f2d43d12 : NODE_FAILURE Node(set) request 300-0026334751 failed in 0s","accounts_in_message":[],"_revision_number":12},{"id":"5db0498bea710b77ea59133f2808e3916d32fe0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 01:20:30.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62601da9aabc4769959d308e903cd1c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac57a869dbe548b58225e8800da0d7c9 : SUCCESS in 5m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eae037d941cc49eb9aea7c815ddb0c75 : SUCCESS in 46m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94a7eb5d1e714b888ae8f53ad5c161d9 : SUCCESS in 12m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66fa51b2c48548ea996b3135940ff7cd : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d89e2775842347a9a79a56215d5afe4e : SUCCESS in 13m 21s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c766b0eaf5994d6da497a6d5c1ca3b26 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7028ac83ba5749149a5a25e3d6c405cb : SUCCESS in 9m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c009618ba89b458eb0c5ee0e34aa6537 : SUCCESS in 14m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52cbcdd4f37b4031829e2692be6f54be : SUCCESS in 50m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b3de2f89a58a49aba73345d1b359fb2d : SUCCESS in 38m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10a65ab6f1f041dd9451d41b1151dbc6 : FAILURE in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de20964e4be541058f9713999d083090 : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":12},{"id":"ef7972daeb067d163b89034a92d1c517a95fe876","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-14 17:52:38.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"fde3f97fc9dd2dbe90ae2945653ccce7e1dd6e85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-14 19:53:14.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"9ffdb1db0bbc67f6a9779fee9928e83cde598d4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 19:53:35.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef815d40ef4b4c55bc5d9cc7eadb5a73\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28e943b3a019402eb9c44ee369367288 : NODE_FAILURE Node(set) request 300-0026342359 failed in 0s","accounts_in_message":[],"_revision_number":13},{"id":"887b7ce442af59c37cb00359349157813189b0fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 20:58:05.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/539823a5f25b4a2ba3fa3c5b0b6fe71a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8f711937ec449bb1270e2ad5a07a9a : SUCCESS in 7m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a47ede42b5f3476f9127f3b3e9d8bce8 : SUCCESS in 44m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e97711357656407285e44be28d461cf3 : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c46026dcecaa4ea4943ceadd96ea6d40 : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80269d7babe54b43b0ff1df65e7f8ea5 : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c2cead474dba4a45bc463b069d81a938 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bff2bfeacbc84986affe2f5c5d1705d5 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fdfe129a04954061819ae5ba2be2369f : SUCCESS in 15m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c947ce13aa74c5b8615e1da452f9dd1 : SUCCESS in 27m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2934ecd2dc774b2dbff5380eb3349c7d : SUCCESS in 58m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5528449899d5427c9f52b536ee9afe46 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f18afc844c67408586d41eeeb13124d1 : SUCCESS in 21m 19s","accounts_in_message":[],"_revision_number":13},{"id":"a64b16989f242c2e210fdc29ae10c005ac4dadd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-18 18:50:56.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"5b314aacaf83b41a01303675fe3eca4cacc64a52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 19:52:50.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed2815819aee419c80139d64c0d9c006\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8ccf31b16ac44bb90156d9043eabe28 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":14},{"id":"20a74e5d909af59975c0df406e0ec4cfa87aa295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 19:56:00.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/512cf4dbcd1b453f9cacede062aea9e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d094a3fa38724e23b66d3ecb7717a665 : SUCCESS in 8m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40ac6733a453427392f4ba499578aba7 : SUCCESS in 1h 04m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5154891cb18430392a7869aa52c6fb1 : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24b622f6a4964b7c873376c5a8af9527 : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f452cb4785bf4f2e96dccaaae8a5e9ab : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/74b27c889c234e7689c60f5162b4d68c : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aace605547404b5ba59f2f0f5fda59ec : SUCCESS in 14m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8842731c59304ef9bda89a5372f8659c : SUCCESS in 14m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a29a9fe6d673482bae67dac1cf995349 : SUCCESS in 36m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4331e27760e4412a44d2e2ceb03e699 : SUCCESS in 52m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01e0ea0e85c841b1b3ba6f50457b2bfb : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d261b5a9dcdb42ca897205a2749c2f57 : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":14},{"id":"04e2adede0970b8b74ecbc941ed914b2ab0797c6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-18 20:01:58.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":14},{"id":"231c12211aa0900517d0303a4d77783112a8d9c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 20:02:31.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"53b0987cf2825da861c2231e5b9bcb083a0f9052","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 21:10:46.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/891e64d795ea40dd99a27b2e3a86f48f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0bcf8d8588447c2b73f855ad5f36a96 : SUCCESS in 6m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/942c8ee8046e4d90ae36b9659cc9ec6e : SUCCESS in 46m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01f65cf239b94416ad70086cd4702a14 : SUCCESS in 14m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/182861829cce427f99393c11fbf7b15b : SUCCESS in 15m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/547769791a494be48cb8e7918d4da814 : SUCCESS in 27m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/64f1846830f44bf7ba85ee17cc317c4c : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cb1885f2a53c411d9e1ed08d95ad591e : SUCCESS in 15m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/17f869eecae643fe9b9b19231a39409e : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d6dddbe2c93348f18ab56505fcacbe21 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/753d99b58fbe48a0a519b8a1a709f282 : SUCCESS in 7m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/837fdfb5759f41b4b9f06f455af6bd25 : SUCCESS in 49m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b80e2b3e730e47179d2855d0def7f006 : SUCCESS in 1h 01m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a426d191ee38470493f633b81bb2c4f3 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/254e5604216440cdbd7b253ea8b71a59 : SUCCESS in 19m 59s","accounts_in_message":[],"_revision_number":14},{"id":"7196c4d4369c95f089b528a20bd5f90c7e56d69c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 21:10:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"bee2ca53a2c32a37d7ae88d856b0b7a4fc11cfe1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 21:13:37.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61172ba1b7ef4e8ab65b4d6a37f56dc0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ea5bb000ac464f8c55a66ed6201c01 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7d9e950d45e740129614446a4afb0a77 : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"c3237aba14a87b78f63721fbf92fd07e770cda97","revisions":{"0f274d78989400e86b51d75908483f542151cc76":{"kind":"REWORK","_number":1,"created":"2024-10-22 20:53:05.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/1"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f274d78989400e86b51d75908483f542151cc76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f274d78989400e86b51d75908483f542151cc76"}]},"branch":"refs/heads/master"},"051b0294ad884b10cebafbd0d53db0fd9eb7e10a":{"kind":"REWORK","_number":2,"created":"2024-10-22 20:56:29.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/2"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:56:19.000000000","tz":-420},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/051b0294ad884b10cebafbd0d53db0fd9eb7e10a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/051b0294ad884b10cebafbd0d53db0fd9eb7e10a"}]},"branch":"refs/heads/master"},"3cd8203485966b59f083e7b86698d5e3cf764ada":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-10-23 15:40:26.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/3"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-10-23 15:40:26.000000000","tz":0},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nsince we have strict requirements on client/server time\nsynchronization the test needs to be modified so that the\nfunctional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cd8203485966b59f083e7b86698d5e3cf764ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cd8203485966b59f083e7b86698d5e3cf764ada"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0fec7cd45150edf417050c49e5d872b8a4d8cfbc":{"kind":"REWORK","_number":4,"created":"2024-10-23 21:59:25.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/4"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-23 21:59:06.000000000","tz":-420},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nsince we have strict requirements on client/server time\nsynchronization the test needs to be modified so that the\nfunctional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fec7cd45150edf417050c49e5d872b8a4d8cfbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fec7cd45150edf417050c49e5d872b8a4d8cfbc"}]},"branch":"refs/heads/master"},"9a5b60309f67ba2b0de7eb9f8716b35a9d19ca27":{"kind":"REWORK","_number":5,"created":"2024-10-24 17:54:59.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/5"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-24 17:54:50.000000000","tz":-420},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nsince we have strict requirements on client/server time\nsynchronization the test needs to be modified so that the\nfunctional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a5b60309f67ba2b0de7eb9f8716b35a9d19ca27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a5b60309f67ba2b0de7eb9f8716b35a9d19ca27"}]},"branch":"refs/heads/master"},"c6ad17f2df771ed96fb929a85f277ca8f1d5fde5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-30 19:09:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/6"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-10-30 19:09:45.000000000","tz":0},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nsince we have strict requirements on client/server time\nsynchronization the test needs to be modified so that the\nfunctional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6ad17f2df771ed96fb929a85f277ca8f1d5fde5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6ad17f2df771ed96fb929a85f277ca8f1d5fde5"}]},"branch":"refs/heads/master","description":"Rebase"},"197ad8fc69a42f74b01e4a4b59f32539329896cd":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-01-24 17:29:23.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/7"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-01-24 17:29:23.000000000","tz":0},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nWe shouldn\u0027t allow a test to assume a strict coupling of\nclient/server time and the test needs to be modified so that\nthe functional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/197ad8fc69a42f74b01e4a4b59f32539329896cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/197ad8fc69a42f74b01e4a4b59f32539329896cd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"35619434a1498fd40e29d076f1744e268aa84944":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-01-25 16:03:37.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/8"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-01-25 16:03:37.000000000","tz":0},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nWe shouldn\u0027t allow a test to assume a strict coupling of\nclient/server time and the test needs to be modified so that\nthe functional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35619434a1498fd40e29d076f1744e268aa84944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35619434a1498fd40e29d076f1744e268aa84944"}]},"branch":"refs/heads/master","description":"Rebase"},"1f3eb1e7949bd0c311806936972907ae5868988e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-02-13 17:39:26.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/9"}}},"commit":{"parents":[{"commit":"11557559b36fe566fefba49ad3f87b0b5d8b8ce3","subject":"Merge \"versioning: 411 PUTs with neither content-length nor transfer-encoding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11557559b36fe566fefba49ad3f87b0b5d8b8ce3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-02-13 17:39:26.000000000","tz":0},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nWe shouldn\u0027t allow a test to assume a strict coupling of\nclient/server time and the test needs to be modified so that\nthe functional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f3eb1e7949bd0c311806936972907ae5868988e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f3eb1e7949bd0c311806936972907ae5868988e"}]},"branch":"refs/heads/master","description":"Rebase"},"5fa143a12f8de8967a86a90b01c61127af91d091":{"kind":"REWORK","_number":10,"created":"2025-02-13 19:00:35.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/10"}}},"commit":{"parents":[{"commit":"11557559b36fe566fefba49ad3f87b0b5d8b8ce3","subject":"Merge \"versioning: 411 PUTs with neither content-length nor transfer-encoding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11557559b36fe566fefba49ad3f87b0b5d8b8ce3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-02-13 19:00:23.000000000","tz":-480},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nWe shouldn\u0027t allow a test to assume a strict coupling of\nclient/server time and the test needs to be modified so that\nthe functional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5fa143a12f8de8967a86a90b01c61127af91d091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5fa143a12f8de8967a86a90b01c61127af91d091"}]},"branch":"refs/heads/master"},"e480a0364894802db0a5cff4be2b32b6575efff5":{"kind":"REWORK","_number":11,"created":"2025-02-13 19:39:38.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/11"}}},"commit":{"parents":[{"commit":"11557559b36fe566fefba49ad3f87b0b5d8b8ce3","subject":"Merge \"versioning: 411 PUTs with neither content-length nor transfer-encoding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11557559b36fe566fefba49ad3f87b0b5d8b8ce3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-02-13 19:39:29.000000000","tz":-480},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nWe shouldn\u0027t allow a test to assume a strict coupling of\nclient/server time and the test needs to be modified so that\nthe functional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e480a0364894802db0a5cff4be2b32b6575efff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e480a0364894802db0a5cff4be2b32b6575efff5"}]},"branch":"refs/heads/master"},"480698152a5518cd7f6fcda091525eabe20edf00":{"kind":"REWORK","_number":12,"created":"2025-02-14 00:27:38.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/12"}}},"commit":{"parents":[{"commit":"11557559b36fe566fefba49ad3f87b0b5d8b8ce3","subject":"Merge \"versioning: 411 PUTs with neither content-length nor transfer-encoding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11557559b36fe566fefba49ad3f87b0b5d8b8ce3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-02-14 00:27:29.000000000","tz":-480},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nWe shouldn\u0027t allow a test to assume a strict coupling of\nclient/server time and the test needs to be modified so that\nthe functional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/480698152a5518cd7f6fcda091525eabe20edf00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/480698152a5518cd7f6fcda091525eabe20edf00"}]},"branch":"refs/heads/master"},"e00a0f90fc5b8f25497c9ed96515cb933d7fc188":{"kind":"REWORK","_number":13,"created":"2025-02-14 19:53:14.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/13"}}},"commit":{"parents":[{"commit":"11557559b36fe566fefba49ad3f87b0b5d8b8ce3","subject":"Merge \"versioning: 411 PUTs with neither content-length nor transfer-encoding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11557559b36fe566fefba49ad3f87b0b5d8b8ce3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-02-14 19:53:04.000000000","tz":-480},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nWe shouldn\u0027t allow a test to assume a strict coupling of\nclient/server time and the test needs to be modified so that\nthe functional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e00a0f90fc5b8f25497c9ed96515cb933d7fc188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e00a0f90fc5b8f25497c9ed96515cb933d7fc188"}]},"branch":"refs/heads/master"},"c3237aba14a87b78f63721fbf92fd07e770cda97":{"kind":"REWORK","_number":14,"created":"2025-02-18 18:50:56.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/933061/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/933061/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/933061/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/933061/14"}}},"commit":{"parents":[{"commit":"11557559b36fe566fefba49ad3f87b0b5d8b8ce3","subject":"Merge \"versioning: 411 PUTs with neither content-length nor transfer-encoding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11557559b36fe566fefba49ad3f87b0b5d8b8ce3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-22 20:52:55.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-02-18 18:50:41.000000000","tz":-480},"subject":"refactor test for x-delete-at w/t part_num and x-open-expired","message":"refactor test for x-delete-at w/t part_num and x-open-expired\n\nWe shouldn\u0027t allow a test to assume a strict coupling of\nclient/server time and the test needs to be modified so that\nthe functional test-suite pass for us.\n\nChange-Id: Ia1d8d9085bad5c6df744a4551eef9dcf56e6e261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3237aba14a87b78f63721fbf92fd07e770cda97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3237aba14a87b78f63721fbf92fd07e770cda97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
