)]}'
{"id":"openstack%2Fswift~933156","triplet_id":"openstack%2Fswift~master~I905b6437f3f5011c96faba7f542597680f2bfd39","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-04-20 22:24:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I905b6437f3f5011c96faba7f542597680f2bfd39","subject":"sq? ring-info: Work on builders, too","status":"ABANDONED","created":"2024-10-23 21:20:18.000000000","updated":"2026-04-20 22:24:47.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"821266133b18a1a8f0959c437c8d1ed08cc6c9c0","_number":933156,"virtual_id_number":933156,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-25 04:01:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-23 22:50:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3282653bd16c35d8e1a15619c8ae03084a5d6500","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-23 21:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76ee83ff8b57325c96b0b3e459ae48e01f9d82a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-23 21:36:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d193fd51ba7f99afbf0dbffe9183ce608cc2d6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 22:50:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ec8383d632848eb8382df01a6b71c93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b94ba889e90445faa7b6c1f5e2a2464 : SUCCESS in 7m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/c28467e4d6c443bcad95ce578557f18f : SUCCESS in 1h 06m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/045b348d9dad4f38af97f604bfd09ce4 : SUCCESS in 42m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1def88be461405fa70630bd920cb8c5 : SUCCESS in 36m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e66fe54e13046a89816c640205a5c65 : SUCCESS in 47m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92d71f5ba1b749d7be14aa79c56e0cb6 : SUCCESS in 13m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20179ad3094647e4b8ad16187a79149c : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1119aa9359e46f08160a26d32e866e5 : SUCCESS in 18m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9587b0d638ee4c8790ee5ba0e9fd0085 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8118f4b723b5421b8761efea73ea34cd : SUCCESS in 17m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd09541753074a0fa63bab792189d0bf : SUCCESS in 19m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/08ea696a7241497287b1fef5f7e83fdc : SUCCESS in 20m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/81ce182d01564f05b492799097d02942 : SUCCESS in 19m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eee5a685429348f6946d9914d928d726 : SUCCESS in 21m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e7ebaf477d5c4e90ae1604e4787cac90 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/968ec169f2f14ee6b9cd4c96a848c72e : SUCCESS in 15m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d48ddfbaba94e3c88735935c33ce7ed : SUCCESS in 17m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b58ec3b8d0fc4f07bb6d8132fc2e9ba7 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/80e8f7ef61f6479ab7153b0acf955595 : SUCCESS in 8m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e6531cb9eb14de19d0a89ebc29e61be : FAILURE in 3m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1224fedf6b046318a4e175c07c5e09c : SUCCESS in 5m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4c920e44f06b43ce85eaf9853eeb50aa : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/90e400c11f96424f8e136b39f21ac6d5 : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab19529811884e2286f694e27a3faa42 : FAILURE in 4m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2db502ea5574fd19b994a54979376f5 : SUCCESS in 51m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b7e78e34b38442ba608471a727fdef8 : SUCCESS in 37m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ffead528e0b48a7b3d87944e3dbb26f : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41076b56a8a4a76a790b92a2e73b33f : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/947a878a6e4c4d8d9d2f75c16cc01552 : SUCCESS in 23m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/260dc4c35b804f4b93d17af1052d4329 : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ec9231fa33424663d37040d1b31deea8aa1b3857","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 00:12:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dac6210f20f0430f9ae903c923cc1dc3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4132175f62424d44b65ed9ef6f53bddc : SUCCESS in 2h 07m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c228d2e06f468a1140ae06f998dda38c8df3339f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 19:09:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f145edbee44914f85ad8a07dff293c7b28160797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:15:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a06ffa56f07540f7928b80dcbd5abe11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26a1e4194c1f43c8b4935e98bae3160a : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/c29b7803627145438945f2803ae13ed2 : SUCCESS in 57m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0aa0c06364a34cc28cca9a560675fa05 : SUCCESS in 1h 05m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5bf4a8e32fe0454ca35dcfab926575db : SUCCESS in 43m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37c7ad5f52fb4c57ae836e418dcc8976 : SUCCESS in 48m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08fbcd700d814acf88db1f28fe5db20c : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75b9494c33a9457a93564233b9b75035 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdc15d8781464ae7ae719ec105c902e4 : SUCCESS in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5effa86aa50a4a259d71a1b642b455a1 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/395bf1b1f34e4377b5dab281ec60407e : SUCCESS in 20m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc80c4d46c144f5e8b9fd931fb530c12 : SUCCESS in 20m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/386d61acb3d2431085335d3b449786f5 : FAILURE in 2m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a27d1538ae814c6ba9da709b983ec85b : SUCCESS in 22m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44cd5d5054c74ba398d8fa058eaeca9e : SUCCESS in 26m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c8f6f110acbd4f4e8100255de4dab0cd : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da3dfc4567624347abd403d9b1bef921 : SUCCESS in 11m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89b8f4568e3f473b8106e366dae425a5 : SUCCESS in 18m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9971c3831f5b4dbfb5eebeab0b26b033 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/80f7498cb16649409fcc49c88557b81c : SUCCESS in 11m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e453f43bc689471f913b3ea7c543d9e1 : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a72ee662a1364302a87ada9b318ed937 : SUCCESS in 7m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f9fbb32c75034f8c8956e0dca1918e65 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce48f14d497942f1a033c6183d53566d : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/844a9ddbf8944a9089a0d8fd14d7f47d : SUCCESS in 38m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4524742462f8420e9ef51c52ab57709b : SUCCESS in 33m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b4f4b4dae2a84a54a19245ab10eb553b : SUCCESS in 53m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cc0341d00ef411cbccdbdfed53e725c : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aee68696b5f423fbe97e7c32236e164 : SUCCESS in 3m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7a29450fd6d4f78bef2541f63f49779 : SUCCESS in 20m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0941dd82e2b946499d2ad7085a0cfbc1 : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":2},{"id":"4fd6a3be615dd81808d640eaf1a52de4c7bf63c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:51:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a43ba9a51c417381562cc106dc3824\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/90ceb9f5c40941f894476c1f21004d60 : SUCCESS in 1h 39m 46s","accounts_in_message":[],"_revision_number":2},{"id":"2028a68fea95f8b063ad119e8f599ac1057bd297","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-25 02:35:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f3af899b89ed8dd51f0c14d0dffe4d1236f682e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 04:01:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8ce9cd7940a49a48a8293f72a0ef127\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab74d361018d44e5b1f555e15291d285 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/4a0db039848042c0b7352cca790bb5da : SUCCESS in 47m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/688dcf4cdae94bc4b0899a2d2c79ede5 : SUCCESS in 1h 08m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ffb8dce4c96d475daa4f1ffcf045ad33 : SUCCESS in 43m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9872baef72a4484b3417ff5eee67a47 : SUCCESS in 1h 01m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e031ebeaaada4f79ac93e4f34dd6594c : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3475ec3279045359c3c4ae97924867e : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf769400e03e4bd09e037ffa0be60bf3 : SUCCESS in 16m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/beff66fbad3c48438671ecdca7426105 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed8d33688cf7430e9fde6711d342a836 : SUCCESS in 15m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d21abe5b2044941bd57b5ee6ab0ed76 : SUCCESS in 14m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/be4c9118044846acacd6586fe00388b7 : SUCCESS in 15m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2b88fc1f49c94dc19a64d14d41cfacae : SUCCESS in 16m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f7a14b2e2df549d9aef8b1d4adf74cad : SUCCESS in 23m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/95cc03696ce345bea6f712472d4bdf12 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f703f9ba90df4beab93bb2e435ab7750 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cbbfe79da9fa4d19a4b85d9785304876 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eadeec1b0dbd4182b266f48b2b486886 : SUCCESS in 13m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5f3f2c553fa44c04bff74f8c3c5eeae3 : SUCCESS in 13m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c94d6f6264d64ff5bde72ca99852e2ce : SUCCESS in 13m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a9745323906849b596971c617753a2d9 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5ec42c2fc82846488695987ace9697d6 : SUCCESS in 13m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1233c2d007214907afc37249be7544a3 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9f38a59ec3b4270a52ba283ee128786 : SUCCESS in 37m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a22a485b44149208d528632897b4fd5 : SUCCESS in 38m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5c06aa10bc6438da24d26b2d71dc1f4 : SUCCESS in 52m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6def901d063845e3b3b0c60c2b7ad488 : SUCCESS in 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19b726bf04934661b6866fcdfb117fcf : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3c3bae2813a7411eb904bb6ca5fcd4b4 : SUCCESS in 17m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e98b61878e1d4a4d916057c6075d3812 : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":2},{"id":"88cbcb2ef6e6baf9ca0611ce61bd735b6d8c99a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 04:47:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a32a806c796242f1806b52307d017fe3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c496afc991942318a33bfc3effc7c8d : SUCCESS in 2h 10m 03s","accounts_in_message":[],"_revision_number":2},{"id":"821266133b18a1a8f0959c437c8d1ed08cc6c9c0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-20 22:24:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"867393592f37d23bf4b306216c4f66b98f1c9d70","revisions":{"fe260b21e8fda0bc6ff5cd5f0010457dfaf3d40d":{"kind":"REWORK","_number":1,"created":"2024-10-23 21:20:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/933156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/933156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/933156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/933156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/933156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/933156/1"}}},"commit":{"parents":[{"commit":"fe21cf04a3bd7c9e4bc69c25ffade87edab00b40","subject":"Add swift-ring-info utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe21cf04a3bd7c9e4bc69c25ffade87edab00b40"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 20:39:11.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 20:48:03.000000000","tz":-420},"subject":"sq? ring-info: Work on builders, too","message":"sq? ring-info: Work on builders, too\n\nChange-Id: I905b6437f3f5011c96faba7f542597680f2bfd39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe260b21e8fda0bc6ff5cd5f0010457dfaf3d40d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe260b21e8fda0bc6ff5cd5f0010457dfaf3d40d"}]},"branch":"refs/heads/master"},"867393592f37d23bf4b306216c4f66b98f1c9d70":{"kind":"REWORK","_number":2,"created":"2024-10-23 21:36:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/933156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/933156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/933156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/933156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/933156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/933156/2"}}},"commit":{"parents":[{"commit":"fe21cf04a3bd7c9e4bc69c25ffade87edab00b40","subject":"Add swift-ring-info utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe21cf04a3bd7c9e4bc69c25ffade87edab00b40"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 20:39:11.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 21:36:41.000000000","tz":-420},"subject":"sq? ring-info: Work on builders, too","message":"sq? ring-info: Work on builders, too\n\nChange-Id: I905b6437f3f5011c96faba7f542597680f2bfd39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/867393592f37d23bf4b306216c4f66b98f1c9d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/867393592f37d23bf4b306216c4f66b98f1c9d70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
