)]}'
{"id":"openstack%2Fswift~933157","triplet_id":"openstack%2Fswift~master~I9aed885e766ca8e1b5d573ea8cb81398513763f4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-30 04:56:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I9aed885e766ca8e1b5d573ea8cb81398513763f4","subject":"ring-builder: Add revert subcommand","status":"NEW","created":"2024-10-23 21:20:18.000000000","updated":"2025-03-03 20:47:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"e5cedc31ef733ef085d8585b6870b1e97c5c81d5","_number":933157,"virtual_id_number":933157,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-03 20:47:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-23 22:54:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8b5dfde5dac0647705203e63b5a40cc35a42cec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-23 21:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dc8c5ad3df82bff2a6cff8399b90590a3429523","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-23 21:36:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eb3ec8bf612edd77f78792971925ffa57dcd4bbb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-23 21:37:45.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6c3ceb0369e9c637de571fba4c302cb2f3e3aeca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 22:54:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52fc13dd241848609d787d2ac38b77d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0983ea49dbb4a88bc8e36696f0cfeb8 : SUCCESS in 6m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/45ff89cbfb534891ae329e13000668d4 : SUCCESS in 1h 05m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e66e7aaf4c5b452ea5eccfb3d3c61a51 : SUCCESS in 41m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6dedb6f2feac4a8fbdce71930be6dc44 : SUCCESS in 59m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3066ad98bc1b41ca9b5ce9a497568c63 : FAILURE in 36m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea0b9f6ea0f94066a8273da49370d598 : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fc56ca1418340f180e19306fe1df8f6 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b31dca820b744f6d97cd0a9cf56304c8 : SUCCESS in 12m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee065841eddc4665a723fb9933a68b2a : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/57a918c828ea4ee6b46c972d445c2251 : SUCCESS in 20m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/375a692a674a441b9d6b0fb625b8b221 : SUCCESS in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/26b41be2b1194c93aa97669dc1009628 : SUCCESS in 14m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/10b74d907c9c4c8496875ab6cb4cd67d : SUCCESS in 27m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7a5076733c524a5a9784312657a4e095 : SUCCESS in 24m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/78b7d81f54ca4963a2f34439e263119c : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c2d4da57b8404ea6984c34832bbdd898 : SUCCESS in 16m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4f0130872c64987848b6152b2bb0d26 : SUCCESS in 15m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/479d4a8daf0f4020a7944087168a3d84 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/de1c8931f5404c859700523d05846c5b : SUCCESS in 15m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b81eee0ecb4342af9e3ca318a2a4970c : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89d965b5429b4f1a9ccfec4e0bb27549 : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97bce1d40e34432b9a830cd02e913188 : SUCCESS in 16m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e834f056099e406a8727ba952368ffcb : SUCCESS in 8m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ede4df43889142f2890752099f4e636f : SUCCESS in 50m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b741506f76514efc890a16a0b8587a37 : SUCCESS in 45m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0df5977b4c9f4b348fa21bc6000b430a : SUCCESS in 51m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d40fe277f2d947b681566901892b94c7 : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8596751bada545da9e758c443ce07a63 : SUCCESS in 3m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5e035db4b0d4081a3d506edcd3f828c : SUCCESS in 10m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dd8beb210b343ae87f157817f11d723 : SUCCESS in 43m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fb290d97d933c49b24838f0aff2ab149ffdf19f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 01:28:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbce335fae51435c85872716b65c7a3b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f174b4f9d2294c6e9a28d7ade42598bd : SUCCESS in 3h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"510a88c81e68cb423455ad28d8c5db11f8ea59db","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 19:09:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da039ffec78144069e0a5050e82601c6e75b1dcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:19:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94b2d07cdfc847459fa3e384769ceb08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3da53896bab43aba2817c3b6b6b5f33 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/912137c4ae2e40ca826a385e08c8dd36 : SUCCESS in 51m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cfb580fbb3914724a048b263b9216fc1 : SUCCESS in 38m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a14fdea3a39b489faa784b22fef26e1d : SUCCESS in 48m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11d02036be2d4d0b98f147e828dc6b38 : FAILURE in 37m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93b90520a4c34c4aa4f55115540a9c29 : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/551e2ae00f2545c6adfa540c8ccdf06b : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a99a64b1f964a51b9717a1f499c1928 : SUCCESS in 10m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd296363a2cf48dcb7b5464656e71304 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78d54cd674df49cb9d506aab5b72d857 : SUCCESS in 10m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e11cca5b5f8449d0a38414a6ad0952dd : SUCCESS in 18m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2945e107b3f64088bf491fbeefcf01e8 : FAILURE in 2m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e106a69a710f4ceabed90cc5e8a12e93 : SUCCESS in 22m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/00f0657c7240479d88f4b1e1c2da90ba : SUCCESS in 21m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/97e0432d19f54545b961f53af628c462 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97f83069214042939676fe9925b5996f : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/21e0efa3176845669b67e5da3bd5fd99 : SUCCESS in 13m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6679f09c3a464bf79f4efba5c5256f18 : SUCCESS in 12m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0ca4220ff5bf41eb8a635dfb1db40adc : SUCCESS in 12m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f5a302bb10df45f1be33e9073746519c : SUCCESS in 18m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a97f7537e1604bb0916cbddd974b0a98 : SUCCESS in 4m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da2ff7af6e1d4077a159017ddfdc1865 : SUCCESS in 14m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c5b7adaafe549fa83b8e57a3f5f496c : SUCCESS in 8m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/10c5b3d94e884e5f96c35c7a71b32d07 : FAILURE in 54m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b16f8d62e9347118e8f9cfc4d62b7e0 : SUCCESS in 29m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c20b41fd21924d0aae3aa34a856fbecb : SUCCESS in 39m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dad05b9146f49069b9cb43e52f8c2de : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fd1a206757e4d59ae5e81aed5ab6538 : SUCCESS in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4ec5cafd41e4d9c9e9bf1d2b5adbf07 : SUCCESS in 18m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c99d632d973e4c35bfc737a6920e0cb9 : FAILURE in 1h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c29b53d7f74285b698732c4d2552b6f671f35cd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:49:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d9380cb8f5a456699142e21c5c2c9f4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f81f30efe3e64b2ea40f420d8c28adb3 : SUCCESS in 1h 37m 30s","accounts_in_message":[],"_revision_number":2},{"id":"fd897d33178e8f0a3a66063f0f72237f6bbb48a9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-25 02:37:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d7dfeb1e5ff890af74959bcfe40145e077fb9be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 04:41:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afd8e31f7a7f45e6bdaf31b7d09958b8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a90ef67e4274793bfe856a53448cc80 : SUCCESS in 1h 58m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4649ea843fd695aa836e019bfb2ac2be004614a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 04:56:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/345aa6ba11f24d8caa34f9cc0f9d2006\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473aa009c0c2483297af96b11300e096 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/643710dbd9304ac493f560f28b8f2300 : SUCCESS in 1h 03m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ebbbd5dae8c9473b9c59ac0f46e2e635 : SUCCESS in 1h 01m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c64cadd2e5b0416cb2a1f304a5b8fc8d : SUCCESS in 38m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6cb0f9c402f4e01b66539dbc1c5795f : SUCCESS in 54m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d08566498e72446dabd5fc377caeb975 : SUCCESS in 10m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fc9888e03604ccf989d81105f2d5249 : SUCCESS in 15m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59c61e1c469649adb3b15761c0d0c65d : SUCCESS in 18m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb19a8b6eb6f4fa3aee8798990e10449 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e213c0e4ffd460482ff5efc97dbe1f0 : SUCCESS in 16m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a2947665e85403a82435249dda8ad88 : SUCCESS in 17m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ae9125f72ed24ceea15270f7a5cf98fe : SUCCESS in 22m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f4e32a94effa4f5d99a943989b84af72 : SUCCESS in 24m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f08290cc565f4e2191ea0d029757805a : SUCCESS in 21m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/10db7e3ae93546b6964e8da5d796eee8 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b951b1dcb5dc4d79b6eef80a7e190e79 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/949267a39f8343da920090ffc446525e : SUCCESS in 16m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7f243d3881894fc1a9f514102d6e5fc4 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9d3af76cb8d946c3855cf1ac2d8013cf : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8926546e4da1403ea1ec410491a09e8a : SUCCESS in 14m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/353e52cb51434b97b11817219026db36 : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/20c36b3dfbcf4303b968a15ac1f89a08 : SUCCESS in 13m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/63724818bcff45d3aac1d8aff75dfc47 : SUCCESS in 5m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/71b6396a1f5f435fbb142cbb05ecac77 : TIMED_OUT in 2h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a734940c56a42b79f158776d345d9dd : SUCCESS in 52m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d584ae60ce74b438c6f31b519814b9d : SUCCESS in 56m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb072c4b32494a09ba3484767fa6b899 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/267b3d6d74c44cecb6c07a37e99c1f4b : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00dadddbc47140fa944232868814e4d5 : SUCCESS in 18m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eeb50f668eba4a77a8e6e5eff35cff9c : SUCCESS in 48m 14s","accounts_in_message":[],"_revision_number":2},{"id":"9010c2d1c8503e0d3013319cbfa2792be8066334","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-30 04:56:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b4942af95fc94be407080917b06ce0264a00463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 06:06:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423c722d13094aab84b023bf71b48f6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef4703ba93e437f9c43e31693df8114 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/a43879c35efc47f097b197f0cdd9faf7 : SUCCESS in 46m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e51cd96d5a74fe3ab541269b147ed3e : SUCCESS in 59m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4fc578c6e524b4db9f1283604b138ab : SUCCESS in 35m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbc0b202640540c3b23b21e4e125f8e0 : SUCCESS in 44m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d445460b48174e9a9b5cde0db17793ea : SUCCESS in 13m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee6b8a1f2cb14ff39d4eaa79511bf77e : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d97ee6162e3f4d9e943bda36db096e8e : SUCCESS in 16m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6174131bcb1846a9a1aa12775c8ff678 : POST_FAILURE in 35m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c511bfca44104c8cac887298c90d0603 : SUCCESS in 15m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a5c1f0d59804a828ec746cbccd8f5c3 : SUCCESS in 19m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7ed01dd8b6cc4d298331cd024f0d9c0a : SUCCESS in 16m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6600f90b712c43a69a721bd88ef79078 : SUCCESS in 9m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/72fdd419685e4322a5aacee2333abbea : SUCCESS in 17m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/47d897aca28f4cf4a96468872ed6f681 : SUCCESS in 9m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4ef4350ad5d7443bbaba50f2e8b5bf6d : SUCCESS in 16m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a40ae583d1084993a7586d505a7e364d : SUCCESS in 9m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2b337db955044624881a64c70e68df8d : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e5d5f6d2aa724b7cb908dea86d5fd1bb : SUCCESS in 15m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d01e57cd0ed44177adeed99ba24f2d06 : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/13e94475e07a40099360b053552e1136 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3bbbdc5759e24710b20381541ab6620b : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8eb9617c99624f27b062a0a3cfc3509e : SUCCESS in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b516b651702844c79377ec2573cd015d : SUCCESS in 55m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f3ca9bdf282248d8ba2afd173964ef47 : SUCCESS in 49m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f85bf1a40f8410e8e0bb179443e3451 : SUCCESS in 50m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec7eef321e564efbbc75d589614747f6 : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e0d9a0b6f14fb6ba9ef3abba06e7b4 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b14d01ae57c4d1195c03569fe1bf021 : SUCCESS in 16m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4f9ceb99e6b46d4aa2aa3db3178e8d2 : SUCCESS in 39m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e8516d0477350d0926a2e7b5f5019ebb1bfa859d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 08:39:17.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1141b5c127ef40a980b68db25baeba65\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c81573c27b09442d8a3a89dce9627a89 : TIMED_OUT in 3h 00m 57s","accounts_in_message":[],"_revision_number":2},{"id":"fdccd2d1609c8b421a3b3addb9f634b1091de5fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-03 19:30:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2a3d2ccdae52bb78b9b31094e462dbcc0df48ae1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 20:15:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97f2f0801013420aa4670defde207af8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04d37147edbd40a9a51dac7c103b2a90 : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":3},{"id":"e5cedc31ef733ef085d8585b6870b1e97c5c81d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 20:47:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cdaa1804b4d4130ac77d727bf49b048\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a184d7932ace4b7a8e8a508ae1fe5955 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/34a7cda26d264b02bb7704812f334e85 : SUCCESS in 50m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48db533486de4dee844fb438701deacb : SUCCESS in 58m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f85fdc88fce469992868b6774438ca2 : SUCCESS in 40m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41ac1f11a34349e58995432d5ce24c36 : SUCCESS in 1h 08m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ab0a46c9f6a43d19a01eec827d81d63 : SUCCESS in 22m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31ab6cb67c074f5a97c9c83611f8789d : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23d0008187524b6ea8dfd943c0443e59 : SUCCESS in 22m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c29cd8059fd64d8e9e20e65292f669c8 : FAILURE in 5m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d81948c50324db5bbd690edd2092bd1 : SUCCESS in 15m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/91679a1d63f64dfda87cbe7b5eb09ff7 : SUCCESS in 20m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a8090d8dc1640438c916949d30f5222 : SUCCESS in 30m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/95a01bc4309142098d8f4c75cefab5b2 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/db0e690edb2846daab983981cd034557 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/920cac934143453aa65b281969bff037 : SUCCESS in 14m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/70a552332bee4e559678d69c95fee5e3 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/545ff0ca2cc04fb6bcd72aa13ce81904 : SUCCESS in 14m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2a1ba59d0810471c89abf32d1b643a6a : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1289757ecea743adbb2ca98de1e47ca8 : SUCCESS in 58m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec04f0ce96d5465cb1cedc63567418df : SUCCESS in 50m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae54bc222d74438bb36b66d31b499951 : SUCCESS in 48m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb3e24ceaf524e19b4ff7c191136c6fa : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1721a766ff7340269b30d247184a350c : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29990247793e4eb886c926b19f7df943 : SUCCESS in 25m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/737f22398f6a4828955ed3929cb67054 : SUCCESS in 50m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1224488c9968438b8d1fa5648221bab4 : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3384d7c08521066537903532ce3e459b64f393f4","revisions":{"cbcb5b081806032ffea323c72457566b8a0799c3":{"kind":"REWORK","_number":1,"created":"2024-10-23 21:20:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/933157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/933157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/933157/1"}}},"commit":{"parents":[{"commit":"fe260b21e8fda0bc6ff5cd5f0010457dfaf3d40d","subject":"sq? ring-info: Work on builders, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe260b21e8fda0bc6ff5cd5f0010457dfaf3d40d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 20:40:02.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 21:19:57.000000000","tz":-420},"subject":"ring-builder: Add revert subcommand","message":"ring-builder: Add revert subcommand\n\nSometimes, operators need to roll back a ring change. Fortunately, we\ncreate backups of ring builders upon completing (most) significant\noperations. However, simply copying an old builder into place has\nseveral downsides:\n\n- It also rolls back the version number in the ring; this throws off\n  tooling such as the ring-composer which expects the version to be\n  monotonically increasing.\n- It\u0027s relatively easy to accidentally copy the wrong builder backup.\n- There\u0027s no feedback about how disruptive the roll-back may be. This\n  is particularly significant if rolling back more than one version.\n  This can be catastrophic if there was a part-power increase.\n\nAdd a new revert subcommand to swift-ring-builder to help with this.\nWith no further arguments, it will list the available backups, along\nwith their ISO-formatted timestamp, version number, and device count.\n\nIf a builder to which to revert is specified (either as a full file\npath or simply the name within the backups directory), several checks\nare first run:\n\n- The file name is checked to ensure it seems to be a backup of the\n  appropriate builder.\n- The part-power and next-part-power are checked; if either has\n  changed, refuse to revert.\n- The replica counts are compared; any change is displayed.\n- The device lists are compared; additions, removals, and changes are\n  displayed.\n- The ring assignments are compared. Number of partitions with 1, 2,\n  etc. assignments changed are displayed, as well as total changed\n  assignments.\n- If neither devices nore assignments have changed, refuse to revert.\n\nAfter that, confirm that the user wants to proceed; assuming they do,\nupdate the version to be at least one higher than the current builder\nversion, save a backup, and replace the current version.\n\nRelated-Bug: #1968177\nChange-Id: I9aed885e766ca8e1b5d573ea8cb81398513763f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbcb5b081806032ffea323c72457566b8a0799c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbcb5b081806032ffea323c72457566b8a0799c3"}]},"branch":"refs/heads/master"},"20a8a324d4cb1e07b69f4558ece60805925375a7":{"kind":"REWORK","_number":2,"created":"2024-10-23 21:36:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/933157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/933157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/933157/2"}}},"commit":{"parents":[{"commit":"867393592f37d23bf4b306216c4f66b98f1c9d70","subject":"sq? ring-info: Work on builders, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/867393592f37d23bf4b306216c4f66b98f1c9d70"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 20:40:02.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 21:36:41.000000000","tz":-420},"subject":"ring-builder: Add revert subcommand","message":"ring-builder: Add revert subcommand\n\nSometimes, operators need to roll back a ring change. Fortunately, we\ncreate backups of ring builders upon completing (most) significant\noperations. However, simply copying an old builder into place has\nseveral downsides:\n\n- It also rolls back the version number in the ring; this throws off\n  tooling such as the ring-composer which expects the version to be\n  monotonically increasing.\n- It\u0027s relatively easy to accidentally copy the wrong builder backup.\n- There\u0027s no feedback about how disruptive the roll-back may be. This\n  is particularly significant if rolling back more than one version.\n  This can be catastrophic if there was a part-power increase.\n\nAdd a new revert subcommand to swift-ring-builder to help with this.\nWith no further arguments, it will list the available backups, along\nwith their ISO-formatted timestamp, version number, and device count.\n\nIf a builder to which to revert is specified (either as a full file\npath or simply the name within the backups directory), several checks\nare first run:\n\n- The file name is checked to ensure it seems to be a backup of the\n  appropriate builder.\n- The part-power and next-part-power are checked; if either has\n  changed, refuse to revert.\n- The replica counts are compared; any change is displayed.\n- The device lists are compared; additions, removals, and changes are\n  displayed.\n- The ring assignments are compared. Number of partitions with 1, 2,\n  etc. assignments changed are displayed, as well as total changed\n  assignments.\n- If neither devices nore assignments have changed, refuse to revert.\n\nAfter that, confirm that the user wants to proceed; assuming they do,\nupdate the version to be at least one higher than the current builder\nversion, save a backup, and replace the current version.\n\nRelated-Bug: #1968177\nChange-Id: I9aed885e766ca8e1b5d573ea8cb81398513763f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20a8a324d4cb1e07b69f4558ece60805925375a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20a8a324d4cb1e07b69f4558ece60805925375a7"}]},"branch":"refs/heads/master"},"3384d7c08521066537903532ce3e459b64f393f4":{"kind":"REWORK","_number":3,"created":"2025-03-03 19:30:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/933157/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/933157/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933157/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933157/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933157/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/933157/3"}}},"commit":{"parents":[{"commit":"e4cc228ed09eb30a4e186515f801d9cde0140c4d","subject":"Refactor some file-like iters as utils.InputProxy subclasses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4cc228ed09eb30a4e186515f801d9cde0140c4d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-23 20:40:02.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-03 19:30:14.000000000","tz":-480},"subject":"ring-builder: Add revert subcommand","message":"ring-builder: Add revert subcommand\n\nSometimes, operators need to roll back a ring change. Fortunately, we\ncreate backups of ring builders upon completing (most) significant\noperations. However, simply copying an old builder into place has\nseveral downsides:\n\n- It also rolls back the version number in the ring; this throws off\n  tooling such as the ring-composer which expects the version to be\n  monotonically increasing.\n- It\u0027s relatively easy to accidentally copy the wrong builder backup.\n- There\u0027s no feedback about how disruptive the roll-back may be. This\n  is particularly significant if rolling back more than one version.\n  This can be catastrophic if there was a part-power increase.\n\nAdd a new revert subcommand to swift-ring-builder to help with this.\nWith no further arguments, it will list the available backups, along\nwith their ISO-formatted timestamp, version number, and device count.\n\nIf a builder to which to revert is specified (either as a full file\npath or simply the name within the backups directory), several checks\nare first run:\n\n- The file name is checked to ensure it seems to be a backup of the\n  appropriate builder.\n- The part-power and next-part-power are checked; if either has\n  changed, refuse to revert.\n- The replica counts are compared; any change is displayed.\n- The device lists are compared; additions, removals, and changes are\n  displayed.\n- The ring assignments are compared. Number of partitions with 1, 2,\n  etc. assignments changed are displayed, as well as total changed\n  assignments.\n- If neither devices nore assignments have changed, refuse to revert.\n\nAfter that, confirm that the user wants to proceed; assuming they do,\nupdate the version to be at least one higher than the current builder\nversion, save a backup, and replace the current version.\n\nRelated-Bug: #1968177\nChange-Id: I9aed885e766ca8e1b5d573ea8cb81398513763f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3384d7c08521066537903532ce3e459b64f393f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3384d7c08521066537903532ce3e459b64f393f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
