)]}'
{"id":"openstack%2Fswift~933203","triplet_id":"openstack%2Fswift~master~Ibc5b96421721e8e0a257acd06fdbfa5d82907bea","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-06 17:03:31.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ibc5b96421721e8e0a257acd06fdbfa5d82907bea","subject":"Prevent installation with pyeclib 1.6.2 and 1.6.3","status":"MERGED","created":"2024-10-24 03:39:28.000000000","updated":"2024-11-06 18:50:46.000000000","submitted":"2024-11-06 18:45:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933203","meta_rev_id":"f14518d4bff0a391df8ded3195283ec70f130c83","_number":933203,"virtual_id_number":933203,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-06 18:45:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-06 17:03:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-06 17:03:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 04:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ef18edd1308db5c931e93f7b229aee50bb2f154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 03:39:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afa7773d9f4c453dfbd2e6aefc5929592383d383","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 03:52:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e91a58911953e429c7c9e604ba883a44e6a0a9c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 04:43:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dad44b6312694f3fb3d1269db24a140d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a86d5c4170403a89473791cc77c7a7 : SUCCESS in 7m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e3f37f1d10114a0188d4b7c6c0e763c3 : FAILURE in 2m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f3f7c3d7dcac4d2cbe8b0a65c4507675 : SUCCESS in 41m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0d0fb5c93a24118a6c59480fceea8fe : SUCCESS in 1h 02m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3f8a826d458f43e990a423163ec63889 : SUCCESS in 35m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/381f7061cf2041de9d961eb4affa49b0 : SUCCESS in 1h 03m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a99013d23e14005a799ed91ae9a6a74 : SUCCESS in 16m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f5e786f4be140548cb26893d811c095 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a784ae09e7454b7e82d3681dc2e79de9 : SUCCESS in 16m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7940f7a597d74b669004c4fdd1264f2d : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/04bb320b136b41c6baae20c534aee2ed : SUCCESS in 20m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34fb8985395d4541aaf2dadba4b61384 : SUCCESS in 16m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e7a2afb34ffb492b8a8a6a9cd52ea688 : SUCCESS in 22m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ae6fc8cc3e0e46dba842e4082f58fe64 : SUCCESS in 23m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/040df9e9624d44918192913f553f7b28 : SUCCESS in 27m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fd1e5ab76c814051907d28a63c168553 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/baeefbdc323c4f86b8231f89c16cf8db : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97424c27946040bdb8260d2ce8b07a54 : SUCCESS in 19m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/25868103bdbe4028a0b9049dd10e1757 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/de59184b3fc746648f8f1299c863cef8 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0f1ee108a32e40699d3544f5a97ffb30 : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03f32b0349cc4f9db84104765059416a : SUCCESS in 8m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/43a8cb993c7e4b4d8b0b4147e6920b03 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/67b5a705da854093ba7ad3fd165c42ca : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb6b8fd509534359930e4aba951c54c6 : SUCCESS in 52m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5c991ec98904cbaa04bd77b0fc12534 : SUCCESS in 52m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d75d3f63c12a47e88474c09115557a16 : SUCCESS in 54m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb872e85f8674fd89f644820a9b1a6d0 : SUCCESS in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fb7d754db5485ba92894c3c637f71b : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/659c61828e7e4462832e655f4b951cde : SUCCESS in 19m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cff0fd1cce9417781a635b7627674e9 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"aa77a32cc43fadd3a21c83ccfcdbc6d17ef09019","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 06:42:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be0806b833c425ea86b68fe3bc39aef\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29c634a21a7a4dd6b13147dffb866d83 : TIMED_OUT in 3h 01m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2cac08a5371d721cd92c4794dd875a72b240efde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 18:50:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"faf2402b93e35319220313f03203ec2dabbe6b4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 20:02:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b02f988e1940dc9f1ff9b49686f585\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1520ba04f994aa1804f9b7ed14769e3 : SUCCESS in 9m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a9ddb588b89a43c89aa1c9b8775d6e2f : SUCCESS in 3m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/927641f7d46e46699dea9c76b8c0bec6 : SUCCESS in 1h 09m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c90b0829ca54e99b43140c0e915f951 : SUCCESS in 54m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6a50bd7b26894d8a80e5b07f9612f86b : SUCCESS in 56m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae58902a7bd84b8bb1f7486032ddc4e3 : SUCCESS in 1h 05m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fec12187484f45f4a28e03ac40465ea8 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f901794bc6a4609aca821bf11943612 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd32bd3550cc413bba2fcb10f7a1ea52 : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/02418bd1149542f4b7e701f6ee80e6a0 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ce3849dbc87d490e9225dfe2327a5b8f : SUCCESS in 18m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa24603c3cfc4d58b10761407e63222c : SUCCESS in 18m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cdf37c651f974f40bd2991d793cebf05 : SUCCESS in 20m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6430dca9473e4df2999bd702de21ec05 : SUCCESS in 24m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3b11e6e8353f4ecb9a4a5967e20d3d5b : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d2b7120ce80b4f888e452cbd7d861482 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ffb0fbdd0c7c4ff6bbf52f323bffa235 : SUCCESS in 20m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e301826f5c724394a7ec39ee897b4afa : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/74e34fdecfc443c7bd665e89c1778405 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f0a4fcf8c0464ef5b2aa920b11dd9798 : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0d7d9cd7a47d4e0598d24468c1062f5a : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a89fb219d44746178e64f32b0230b6db : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d77878d1bc0a40798c9d19304711b438 : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/17c895fb51fd453c940c320e1cfaa73b : SUCCESS in 49m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4db4eb2968b74fc7bb878abe9c767be2 : SUCCESS in 42m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1eeec7d33c8e4fec80face80de2410c0 : SUCCESS in 47m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc32c8cfa9f343eca67055a9243861b4 : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8517d47f43c342f59375434a095413c6 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d94881634484c68bf1d061bdcbed206 : SUCCESS in 19m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a0d8a531660421498a25e1f4cf4f662 : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e5259e02f0e649d4e0610dade6c413f227cd59ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 20:48:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b46d7aa9093d4318a149661549814052\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9158fd3e4ed24c08bfed00dc2a20e124 : SUCCESS in 1h 17m 05s","accounts_in_message":[],"_revision_number":2},{"id":"5f39046708ff9607e1b1e7c04852925bead6334f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-06 17:03:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"18068fe5b5163c2428cdcc63d27f53267709f048","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 17:04:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"033ed7e5c6076d7002494db90644365b95696155","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 18:45:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74177d6278bb4a489724d256f04f7883\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c202aa096d3646fbbd5358d6b50b2e92 : SUCCESS in 9m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5dfc7511024646fda7076cb6434e992c : SUCCESS in 2m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/b8852355757a4c0596d185cbddfdf685 : SUCCESS in 40m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/04a90bb35309410bbcc0a93f66cc3076 : SUCCESS in 1h 10m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4851f1cab0b14a3c9024e03a976e3a79 : SUCCESS in 1h 03m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5951cbd7206d4f5998961415d7ed50ae : SUCCESS in 42m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/17da33b362e3415aafc78e0731526f1d : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6a290ea158f844779dd1b612de501af8 : SUCCESS in 17m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f1f419bbee3348f1862946dfa4d15ed4 : SUCCESS in 11m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9b01927f6c8d40c4a60470915ff08f81 : SUCCESS in 16m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2eecad6967be49c2888809d61418ee40 : SUCCESS in 16m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ab06b052a364a1eb3a22f81627b71db : SUCCESS in 21m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3473188d0602414ab858a72ef787f0e1 : SUCCESS in 12m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a5bfac05c56c49af8925395c587b4d0a : SUCCESS in 20m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5995e76631454fc58bbc17e2cc77f62a : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1eaf7947f3614cc0b31a5808be762b26 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/927a783d93794eccbbe8e940a279b9b5 : SUCCESS in 17m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4eb3cca968fe4c918439b18b4589f3c1 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a435d38a408e44ea937ca859b39b4f3c : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1c13031b9c0455ea004daf7065bb7d1 : SUCCESS in 54m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4970db5bb470427ea26f892c310fc09b : SUCCESS in 42m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c08df49dffaa463b81a457e0ea2efe9e : SUCCESS in 43m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e5bf776d9840e7a83330138795f57e : SUCCESS in 22m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68ad75d632264a6d9f927c31c56aab7a : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c18515a9bfaa4060b9c371b6d9eb6f82 : SUCCESS in 18m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/990062b49d8941968b7471436ceaacfe : SUCCESS in 43m 50s","accounts_in_message":[],"_revision_number":2},{"id":"bfd2ad6c82b1a717f0145f1b502918908061aa33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 18:45:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f14518d4bff0a391df8ded3195283ec70f130c83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 18:50:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acf8c0cf7d4743f789616418562b3190\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73a38e291ff48afbd4ed77b1294d0d7 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4e29c6a2d30e4719a5ff25993c117fbc : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e954712cef99b32b0fe815f3fbef6860b51c1343","revisions":{"67a242f1b19589d6268dcb0fe91b95d749e9c69a":{"kind":"REWORK","_number":1,"created":"2024-10-24 03:39:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/933203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/933203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/933203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/933203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/933203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/933203/1"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 03:38:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 03:38:59.000000000","tz":-420},"subject":"Prevent installation with pyeclib 1.6.2","message":"Prevent installation with pyeclib 1.6.2\n\nIt\u0027s a little too prone to crashing python.\n\nChange-Id: Ibc5b96421721e8e0a257acd06fdbfa5d82907bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67a242f1b19589d6268dcb0fe91b95d749e9c69a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67a242f1b19589d6268dcb0fe91b95d749e9c69a"}]},"branch":"refs/heads/master"},"e954712cef99b32b0fe815f3fbef6860b51c1343":{"kind":"REWORK","_number":2,"created":"2024-11-04 18:50:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/933203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/933203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/933203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/933203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/933203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/933203/2"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 03:38:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-04 18:45:59.000000000","tz":-480},"subject":"Prevent installation with pyeclib 1.6.2 and 1.6.3","message":"Prevent installation with pyeclib 1.6.2 and 1.6.3\n\nTheir binary wheels may crash python on older processors.\n\nRelated-Change: https://review.opendev.org/c/openstack/pyeclib/+/933338\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/934074\nChange-Id: Ibc5b96421721e8e0a257acd06fdbfa5d82907bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e954712cef99b32b0fe815f3fbef6860b51c1343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e954712cef99b32b0fe815f3fbef6860b51c1343"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
