)]}'
{"id":"openstack%2Fswift~933363","triplet_id":"openstack%2Fswift~master~I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-10-24 23:39:44.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-24 21:48:55.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5","subject":"CI: use py36 constraints for py38","status":"MERGED","created":"2024-10-24 19:34:55.000000000","updated":"2024-10-24 23:46:15.000000000","submitted":"2024-10-24 23:39:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933363","meta_rev_id":"0e7f59903424855dac2acc76686d6840290079a7","_number":933363,"virtual_id_number":933363,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-24 23:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-24 19:56:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-24 19:56:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 21:00:18.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-10-24 21:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-24 21:45:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"280e80b70f627320ed390b1347096e29ce5b4bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 19:34:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52591c958ac06a3f1b6ead8766aafb2f0feff5d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 19:56:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c94ae5402c71daefaac1f8db9566694ac307b09a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-24 21:00:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e43eea3b9b62e13609f37ad4cb565b8082c3835","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 21:12:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/770428aa88d24ada9bac1b334f36c37c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/592294b266644b30bcd7ba1480568f15 : SUCCESS in 1h 35m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2532e03444f41184c6d7232b4b66b89304fe811b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 21:45:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8728499506cf41a5b01e4cafd64c0d2c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/696a862a457f4ba393229c87ea8ed0a6 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15cd5fdf1044ad488d2b395aa1a6b67 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f52fda87004f49b2a7208a21284e174c : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/165aa786b26c482eacd056ec439616e1 : SUCCESS in 59m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16429176a7074ecc89dbbaeaf368f576 : SUCCESS in 37m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/19fc2b0e476e4f8e801ccbce6daaeb67 : SUCCESS in 40m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d66878b87cc4334a1a7cc7da3622d5d : SUCCESS in 59m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0dcc4e7f069403a97af39514c93f7fc : SUCCESS in 11m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c13188bbae124e3394e48f16393dd5b7 : SUCCESS in 18m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/460bdf7f83014b6bb65285b80557a856 : SUCCESS in 16m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c528555297414211b6264e5ed535cc4c : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea0950c707f84eaabc54227abca95e7d : SUCCESS in 20m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2bd68a8c7fe143599caa1fa5650b995a : SUCCESS in 16m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a478ddf8e8ac49ba8084299529cb1438 : SUCCESS in 19m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d48265e694834c52b7be5df0e93e53bc : SUCCESS in 21m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e6ccb3e2dca34ccba6bae4da2e57eb2b : SUCCESS in 21m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c28fae4074844f7191f1a46ebbafc6e1 : SUCCESS in 10m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f003d93281242008585df980dcb6bce : SUCCESS in 17m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9733145db6e442a9bbd96a9ec1092276 : SUCCESS in 16m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c5c0c3f9066e4631bc51f95969d9ab7d : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/242b9f7655764e1fba4dc2c11c4aa12d : SUCCESS in 18m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d9bbbdd91364e31b37d057421e16e81 : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/71255f6c344d47d981ba6e21c09be2eb : SUCCESS in 4m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d58e0a24fdf648019578a9c875733075 : SUCCESS in 13m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/54fa8c2ac9cb4ba08c2c51fa4b4c2d12 : SUCCESS in 5m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/750561f51b534883a16d1d54ab635687 : TIMED_OUT in 2h 00m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07fc7d463e534bf6b6f4761d0ed3afad : SUCCESS in 41m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40fdcec09f7342ef8866757fc90be8ba : SUCCESS in 50m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8786eebabd9146c793c03767d2ea4a57 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f9d38796ed470ea8cc58a261089ad7 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d91d87bb8eb4a6da27747b446dd1425 : SUCCESS in 19m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92939f23363b41b4a5dc6e1fe982847f : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d8345614f7daff83641e261a4817a894709b8637","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 21:48:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c9012aea521aa197769d76cf36ebf7f602ca79e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 21:55:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"149ef1bb4a13b2c540aa3128d7c26037f8d083c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:36:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/334f7911d1af4e2ea05597101fc98815\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/abde5d3241ae4e2c885a11e5e1287b19 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7cfcd85c59475482073c12b00727c9 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26d2c1c444764025a9038732b9ef9d6f : SUCCESS in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/1da561c5f11041858919e94b4f4ae218 : SUCCESS in 43m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f70e67bb0ca4813946ce6ef2b8c29b2 : SUCCESS in 46m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60f8ae72dfd942c486b9935d1b24509d : SUCCESS in 44m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/533c5946534d42599f86bd2546519d88 : SUCCESS in 46m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2776d67a2b24f279eecc9e0ac2b4dac : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe6722ecec9a4021b478cfe620c7b59e : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f9e379f26ef402fb64c7b799972cb80 : SUCCESS in 17m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d63a0de5fa541bca9f9b9dcf1cae78c : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0fa828aa4d8a4f3bbd8348508235bc16 : SUCCESS in 22m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c655568db4a84135a3b9bca06c1e21f2 : SUCCESS in 17m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8e8b1e1c16984d748ad468d8aa7d4ea4 : SUCCESS in 8m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/31fdbde3b377409b862d07eca738dc7e : SUCCESS in 20m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c437fbd6868d481e87680a16476ea288 : SUCCESS in 12m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d15be3c11a6c440fbcf7a603571d3893 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/15abd60926444ac3a3528ebebc6a6350 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e0f1bb55a8114c9f9dd56e424c62c386 : SUCCESS in 9m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4677fa3d2e3441679056719f6a416b96 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8cb9810465ad4eefb2b016b79b45d941 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5eb995fc27944102858fdfc2ec8fe075 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f3ce8201cfb3463697b202f179198bf3 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/92d9df2a2067402e99b1270e67f22015 : SUCCESS in 13m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b48cbaff5ce64952808596726713111d : SUCCESS in 6m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf13de2d31034b6393966b7e01967b33 : SUCCESS in 40m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/89b0121745ed454ab976344b20e09f7f : SUCCESS in 32m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b4facf977c74f79851957ae5cf56b28 : SUCCESS in 34m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/766710eb3fc244d7b19285bd96762fb9 : SUCCESS in 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd5f5f983cd4c7a9c023e269c77a303 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46d60225d2c44370bdd4149943c1f8ec : SUCCESS in 17m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6f997a03ddd46119aa4b1af3e557b44 : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2d33beb9d5484acf4bcfdad3b52e9f6c77628c17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:37:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"945201e7226915cba42f0902e101b8b98ad8148f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 23:39:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73f9a7b91b5840ee9c3dc3488a01809c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1e2484f17fcb46289446294b1ae4ee54 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5849ee4077045529508e7c316bd1915 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/160b8f68728e4d3c87d46acab5ccd0ab : SUCCESS in 5m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/746c6bca75994c3b8f83bfeae9ab3e5d : SUCCESS in 42m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eab37b1368ff435d9a9932482dc80789 : SUCCESS in 48m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7270001c4f224c07bd33ebc40a3676c7 : SUCCESS in 39m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6f337ce7eb148b9a1e771158ec25aa3 : SUCCESS in 45m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6bb638dbf1014b0b86886200e2856956 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad4b28395f1840a1b7c6ef1357baf767 : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e47c1a90a8d748b68378335554cb59d9 : SUCCESS in 17m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2f39590e21f4429e983098389e0a1abc : SUCCESS in 17m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5906ee7d3dbf4df490eb149aa47f04bc : SUCCESS in 18m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d3c886e6e869473b9822f4f63a18532a : SUCCESS in 35m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dc1fe1916ec5452d84e5bfd6735b60e5 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac81a954b81941eab70efcdeb0f94d49 : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/52ff95b44cc1449b96b9bf9b8598402a : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5d9cfeae370c49db83e6284670fadf61 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3a25e156cfba48e78cbdc731b1e8f640 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/311dc04e46224aeb8ea38f1a0fd8a209 : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e224db3798b5499194728d7e717267a0 : SUCCESS in 5m 29s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/440bf2e9710c42319a15d6f5511aa7d9 : SUCCESS in 4m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b2570b54ff642739a7eb04250d4e054 : SUCCESS in 55m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/521687bf72a24a6fa37e49308ee44372 : SUCCESS in 42m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9268036e5bd43038b7e32b5980ba03d : SUCCESS in 42m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3701ced0da354bf8a46427f3e6a48102 : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d22208d82a49db9f21af1b72bd0e72 : SUCCESS in 2m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d161d8d350b4cabad9b55baf4c1943f : SUCCESS in 18m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d8990aa93b942cdb021aec57696c093 : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c12693548393c4c5c48c1410c12d4488d481920e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 23:39:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"77f4dff82663014ab8cccf061eb5df84a8cc74e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 23:40:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a594652d7b1459fa8afa7a31dc82ef4\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a5789b8a06d8427281ef008e92727b12 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b65aab040a4b1ba1610c3184271751 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83c59519f9af46d58e18428c520452ca : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/da7a2f397cfe45bda72bfb7c3f9ce851 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e7f59903424855dac2acc76686d6840290079a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 23:46:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8922d3daa6754f7bbab0b8ec1fc721b0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/73efa0853bbe40c581add7c263d93c36 : SUCCESS in 1h 55m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fdb93c980651d5672d000b354410d3ca3d1cc43","revisions":{"0fdb93c980651d5672d000b354410d3ca3d1cc43":{"kind":"REWORK","_number":1,"created":"2024-10-24 19:34:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/63/933363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/933363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/933363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/933363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/933363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/933363/1"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 19:33:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 19:33:55.000000000","tz":-420},"subject":"CI: use py36 constraints for py38","message":"CI: use py36 constraints for py38\n\nNow that openstack/constraints has dropped support for py38, we have\nto manage constraints ourselves.\n\nChange-Id: I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdb93c980651d5672d000b354410d3ca3d1cc43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdb93c980651d5672d000b354410d3ca3d1cc43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
