)]}'
{"id":"openstack%2Fswift~933369","triplet_id":"openstack%2Fswift~master~I8c65c7829fbe167e0c79239d577c38e90b93b5cf","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-31 19:42:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I8c65c7829fbe167e0c79239d577c38e90b93b5cf","subject":"CI: Move a bunch of func test jobs from py38 to py312","status":"MERGED","created":"2024-10-24 20:34:25.000000000","updated":"2024-10-31 21:45:08.000000000","submitted":"2024-10-31 20:41:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"933369","meta_rev_id":"cd1be14bac021a5952ffadc0a5369da593aad76d","_number":933369,"virtual_id_number":933369,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-31 20:41:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-31 19:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-31 19:42:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 20:59:28.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-10-24 21:48:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcf5e5df7ce36ed7e0cf7ba9226a950928042ebb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 20:34:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f81ffe08233a380d1cbc14eafae005c4af4a00ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 21:48:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92f374e471164e31b6c7418123b24802\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b4300d73e23849c3b18e6ea9b3439e7b : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c253b000148a4c7f90874ec91463abe6 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28f0fd24202a4db0afecbedf6637cc38 : SUCCESS in 6m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/a801dfd66ecc459381f13fdf9fd35a82 : SUCCESS in 37m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d0526b2d3404dc4b126f24cf8a33eb2 : SUCCESS in 1h 01m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9b738b7f328c4829aad063ecbde83fed : SUCCESS in 45m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e689ac8435564ba59a584697d7eb32a4 : FAILURE in 37m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82c886a8bdd0425ba5c9413c38e39eed : FAILURE in 18m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdf9ff62e7a24cfcaaf85c6a0c5cf207 : FAILURE in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8237c96db0141a4a1f07017041eab55 : FAILURE in 26m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27431f106d4449788af0a6128d96e79b : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad2425bb0d954d84ab7f9b03c16e4e21 : SUCCESS in 20m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a2226fa2c8d42d6ab936bd8cc5d742a : SUCCESS in 18m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c1edb2c14b17417883070125c516e1db : SUCCESS in 18m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/08a04373362d4394b440f779cece8745 : SUCCESS in 22m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2750da7be9c242b4a4a91aa9989df4fc : SUCCESS in 25m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b132c4210ddd4857828d0ea3204f017f : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69e0291205bb4769bbc9624d8c3ec201 : SUCCESS in 18m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1133f9e426b248bbbe5336557ce77cbb : SUCCESS in 20m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cbaf3b5662564ba3816d80e1ace4371d : FAILURE in 15m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ce57bb5b50764171b2d52b5af86e4caf : FAILURE in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/091c6e48b1ad4077ad6779a885cb55dd : FAILURE in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb5696b2e3b143d7a085687c8a51a44e : SUCCESS in 8m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba2a43adf4674d2a8621bd1a98b11697 : SUCCESS in 9m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/546dac6df4734111a05661be3f3d1616 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e907682c03f94b98b8e876ab250f5f6b : SUCCESS in 56m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/426f71ddabca469dbb3f97a932615d92 : SUCCESS in 31m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b91c35446a5045f0b380d3db12391d43 : SUCCESS in 34m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5003b391267436f84ec9d211f456dc2 : SUCCESS in 24m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47cf4cce7a274460959019ee1935f443 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4bb48de724b64a2ea02a72bbbda211df : SUCCESS in 16m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82e176cc66bd4a56b6a2e68f22535af4 : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":1},{"id":"28712580f8625545455eaa215d4ee8ba6b7fcff9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 22:12:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c6cf5e55c03a21b7f6bad6b5bfe4c5b5f14b820","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:13:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81434aaa63f24503a43effb1d09dfbd6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3eb95f2723964c109da7b84be6a2b520 : SUCCESS in 1h 36m 44s","accounts_in_message":[],"_revision_number":1},{"id":"66653474e3d24879efe5ad7f7866771f6c7b389d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-24 22:17:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31dfe54f60be96173ce15f1542db3d743ec751d8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 00:09:32.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da06d5251a7411ab45a6c9eb0f3a820\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eac861e55894b1b94befd301a719ec4 : SUCCESS in 1h 16m 54s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/c6f3f685aee74d708788844345349136 : FAILURE in 1h 54m 28s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe3635a57ad04ac3b608533da086e3a9 : FAILURE in 1h 41m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d615b21abb7b433abbc126d25429191d : SUCCESS in 17m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/87018409cb9943f2b976360365220f47 : SUCCESS in 21m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/10978d65c9dd4c6983b37a683536343e : SUCCESS in 21m 06s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd715e8180a44336b3644a50828b198d : FAILURE in 18m 33s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d766f0f531064fe9804f29d503d6d26b : FAILURE in 19m 26s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b7e714d8c41466eaf9f1d1efb52f5c8 : FAILURE in 19m 52s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/5d1e9c734b314f28b982963b143e9f88 : SUCCESS in 10m 24s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/783efd8250874176962f1b471e263d7e : SUCCESS in 19m 58s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/a189541df20b43bc8cf6ef1d08b6b5c8 : SUCCESS in 19m 24s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/39f4e2b574a549b7a6e42af619de52f6 : SUCCESS in 10m 08s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/15d58ac0009540d1b6cf0b9333a4b00b : SUCCESS in 20m 00s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/33f2ce4e59a94e7bb26a13e415f1fcf0 : SUCCESS in 19m 44s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/afe6ffea785c4b738d5bb168f66e5656 : SUCCESS in 18m 51s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/8c9880ec154f4d278af4e0f4f7ae8cb0 : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"43d7d7b754f8210196865c4d73f8e5d61b3e121b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-30 16:39:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"936dc2f009eb9d597e87155a724ebedffc99f91b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-30 16:40:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f88bd30228577cfc487b3341713f307b040e3578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 17:50:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58b0771e562540f9bdc84a6190be1a1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ac59251a6e450a991777ccf62c6157 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/26d4e3d877cd427289d22f86dd1ed59a : SUCCESS in 54m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e253be6d0d4455383a3330bfdbce56d : SUCCESS in 53m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92a734d2c0f84aebbc9d41470c81eea4 : SUCCESS in 40m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a855d94437644aca25f29bb3df3d9e2 : SUCCESS in 37m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52029833eb054e00a0e3fed207b082b7 : SUCCESS in 12m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5f0718aaedd420a84b952c01b115d1c : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f9e1d6b557547dbaf1057e3497df762 : SUCCESS in 27m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/702fd8bb57664075b01b5c55d44a37c1 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47f4d92236f64c13aaad6bfed28ecfcb : SUCCESS in 23m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44e5cf64e8bf4a979b2223239fba0f78 : SUCCESS in 22m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/90b690029c784d8b8446d1ee0dfbbcab : SUCCESS in 23m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e01bc6c84c9d4bfd822be2213e31df67 : SUCCESS in 23m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dc756e2770544e989001919f5e604d42 : SUCCESS in 22m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c7afcbb872d542668e16c440d731c050 : SUCCESS in 22m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a42861109cc4436791997c89a4d922eb : SUCCESS in 23m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dce13ff604b64a1e8e087ce0724b6c01 : SUCCESS in 23m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b43549a6f13042b2a8bce6e7bd2e5a30 : FAILURE in 3m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/32a3b0bf083b465087d22d6c7171dbb1 : FAILURE in 3m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/30dd90ec99ea4c35909e7674db182103 : FAILURE in 4m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ec3fca7f0a8f4f7b920685f568148b68 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e689623b14e643218cc4634d5b602949 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8fad8623f86f4fba945fc0a62a2f0a15 : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/99df5bd3c36d4946a46848c554e50906 : SUCCESS in 1h 00m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3da3f4bcd68649b6a891c93837c79434 : FAILURE in 19m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5110145fd41b44889602beea80434496 : FAILURE in 20m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/903eaeb15582451a8747afd55b56f3f2 : SUCCESS in 19m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b072dbe70c14b63a129e18b40920b96 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f06d5e5c4634430589beb9c9ff35bde7 : SUCCESS in 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76a1069c34a147b1a4fb62136e2d044d : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":2},{"id":"fecefef7c05c7d969c6746d38eaa2687808d5094","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 19:24:51.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c622a8bafb4a4611970471e67b8282c5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0487114238fd47bbac3a90401155c48b : SUCCESS in 2h 33m 26s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ede216261b548379e9a5e602d054165 : FAILURE in 9m 09s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/98489da759d1469dab6ea1a7f848262d : FAILURE in 10m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b8edb0671c51468daa01c464a6abe8e9 : SUCCESS in 17m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f33ba0016d534b94801ef70aa897d738 : SUCCESS in 11m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8d0e4301b17f41a798b648eb2d215cee : SUCCESS in 18m 04s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/552ce5251ef1434cb3b36388ba1e42ef : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/43ceec0602c343bb8d51caa8531a44bc : SUCCESS in 18m 02s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e57efd44f14c40deafb67e4d5b2e7362 : SUCCESS in 19m 50s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/0d274aea85e544c7bf87e21bf141f303 : SUCCESS in 19m 15s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/9eadcbcf1cf74d4faff4a5eadd8888d2 : SUCCESS in 18m 03s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/9f72415e58c94a83abeb36f498bab83b : SUCCESS in 19m 02s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/09d038bf99ac49949ccbe1ed3d2483f3 : SUCCESS in 18m 54s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/e6fc86aa3f7e4127b41275f26ca7f16a : SUCCESS in 18m 16s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/26de0838058b41ce852e6b4298a1263d : SUCCESS in 18m 56s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/b7bcaa145a374d4e856a272aa4ebda3d : SUCCESS in 17m 52s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/f94229505bc9417c86c923266ea2c5b9 : SUCCESS in 18m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8ffe449a6e58fe947194dc2afab7efcf61433eab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 21:09:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61124c41c4cf4a87974e3236b663a221\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a8d40486f54450bad5a0e3b48728643 : SUCCESS in 1h 52m 22s","accounts_in_message":[],"_revision_number":2},{"id":"23753ddabbc498a1276ec011d03583d0ff6a06ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 05:43:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b9312a045fecc65cd6c06fa1edaf78fff1b2ed2e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 05:46:09.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4f46b8eb5aaff6b6c9b90668e1d262be36fc4aed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 06:10:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"84607b6cf2cd130860e047589dfcfef26c0e789d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 07:13:23.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99a0cece1adb46499b9ce3f79fd1ed6a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f419a2392fe0486dbeb0f58d72351528 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81432458e4f043e08969da0946b66778 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/705af171f1a740d89583fcb280087c3d : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/5b817c1e51964fb09511579b1ac943da : SUCCESS in 56m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4a59dd429d2a45f49d818efa0a7c634c : SUCCESS in 56m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d77a680e78874944a0f5d243e75c1b88 : SUCCESS in 41m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea498623494d4c5abc2f23bd7b95c78b : SUCCESS in 57m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff488aa9910a404c880cbec942d44c7c : FAILURE in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35ff0257c47244eeb3b056e402cc9154 : FAILURE in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5971297cfbd47609003cbce104f831b : FAILURE in 21m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26c9c13e2d8743abb68e9ff01d757779 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5da74d08233443ca3d4a1fb6712f33d : SUCCESS in 20m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a99d2d5454849e6afdbf9a4785a3ae3 : SUCCESS in 19m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e77a5a62f1bc4636bccdf44706ff4a39 : SUCCESS in 20m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/25e00a6183034194b730a03d8d40177e : SUCCESS in 13m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b6cde9ae7b0344d4bdee3e085229f71c : SUCCESS in 28m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c098c38a4f974c3b9c48b93722a70f21 : SUCCESS in 13m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ece83e556a3c4d2c96c647a607c7e85a : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ea0bf49f05af41b8a256586d3b200582 : SUCCESS in 18m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7628c8f6b06441399e3c9bef2e401327 : FAILURE in 12m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8ef82f8c8d3b4c61a14abb0774738932 : FAILURE in 11m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dee404f33dd74f7fb45e37257bec9b1e : FAILURE in 14m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d80b459e24bf4dd39dc16182823b2343 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e1891aea266464d8c4b5d6683580f53 : SUCCESS in 12m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/365f0ee871f84b0d9f1f57cf24d2fea9 : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd54c58e45314b689944b600af30526f : SUCCESS in 54m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bee2b985d3c84df4b6c36ae94c2039aa : SUCCESS in 30m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1d3612bb15434a86a3d512336ca40600 : SUCCESS in 47m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efecc6b1ad7f40b0a2bc70ecd3b08dd9 : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0726e6c6a6a4461fb92203cde1e9a997 : FAILURE in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d872b68f806449a8fd755284d6bcd16 : SUCCESS in 11m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0df9a39c19fe4199840273e8f9cef88a : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":4},{"id":"b8e6144ab9d3427eb094aced9ec07535e650bc43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 08:04:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad753e14dd6d4a02a0853a02b55bee47\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/365562cdbbf24a6f85204e2f6b341bd6 : SUCCESS in 1h 52m 02s","accounts_in_message":[],"_revision_number":4},{"id":"b76fb8b0241b02f143856f34e3dfb2e029463373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 16:00:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"93ed2a22cb13bc7adac8c0c19b3e5819dd2974be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 16:19:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f5f42d627b02bceb9e056a5c71b3588000728778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 16:20:04.000000000","message":"Patch Set 6: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/swift/.zuul.yaml@master\", line 656, column 31","accounts_in_message":[],"_revision_number":6},{"id":"99d909adf464797599bb4f0751398bdf4984543b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 16:20:08.000000000","message":"Patch Set 6:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/swift/.zuul.yaml@master\", line 656, column 31","accounts_in_message":[],"_revision_number":6},{"id":"5dedbeb2cc7f58dedc6903c6a8762de61f49f2f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 16:22:33.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f2317ded863a246c5d01f15309d719c956c2ff4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 17:36:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b44924b09e674ab9a8e2afe09b67b7f1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bd5a150fd4544b32be13c28d71307c3f : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b848cae19eb4f2280a565f36cbe314f : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6981d4500f441e2b33a8f29cb35326f : SUCCESS in 6m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/3c3c63b5aa6143ed93c53dffcbff3330 : SUCCESS in 1h 04m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e3bc2f4eec1c4ebe9362dfe18479ffb1 : SUCCESS in 58m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45641a17eebb48a99401573d6db846c9 : SUCCESS in 50m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52a8ea70bf3545cb9af49df4c61c3269 : SUCCESS in 1h 02m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f2334674ae14dabb189656f01dc3f20 : FAILURE in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0310c1b09419473dbc9053c35ae57862 : FAILURE in 21m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e7d74489edc40f0b7d805862b0bea59 : FAILURE in 19m 00s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/86e159f89daa48f88f3b7c820f3d2a9c : FAILURE in 15m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/979172ebc7db45dc93e33e46db87e7be : FAILURE in 12m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c917299fefcb443b9a3f3c369b98bca2 : FAILURE in 16m 45s","accounts_in_message":[],"_revision_number":7},{"id":"4f2218c7d1f7b1064ca8531dc1c580c950cbfb27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 17:44:27.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8223e8a66085254bcfa62f0105fee2798d6da625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 18:05:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dd5bed0f77945e28192b22877ecb4635d91d1ca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 19:05:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ac35ea8027740c3af2b021eb8f8a9ee\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e602e303526b4a6f9f3e7cedd2023105 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d23c40eb5ef465990bae57117242525 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbab71c4d66a4366aee3f4fe07332676 : SUCCESS in 5m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/746adcb3311248fabe150a4eba729e53 : SUCCESS in 55m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5dc8ef1647d14598acd7d941fd011561 : SUCCESS in 44m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48a60121f6f34148be55b3a814d656af : SUCCESS in 52m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c6e24772ce744129df28838017a4ed2 : SUCCESS in 48m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f962afc80ed447129d910676f3ad8aa5 : SUCCESS in 22m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70d5a93369db4ac9a9f472d82ea81690 : SUCCESS in 13m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d74a1490c7e24527a7f921e49bb1b10c : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69b696561a5a40e88b43309c3f12b672 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8d27fcf549c4914aa3517c1ceb80c65 : SUCCESS in 17m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/844ee7a15a06407b864f1eaa27dd3ffb : SUCCESS in 23m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7fb3ec85cf914d41b14682564ff75a76 : SUCCESS in 14m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c09540f3224944d4a6835af2d2bb8c1b : SUCCESS in 22m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/761d483be9824685a7363801a3a42e25 : SUCCESS in 30m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/48f1195946b849c2ac7a56ea516dd9ea : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9cfe58e1605943d9b5962d2cfe43a793 : SUCCESS in 10m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5610ff86690f4d709f756eee0d667f23 : SUCCESS in 18m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e123126dc6fb43d6a4e7c255b35f39bb : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bd7ab63f00ff4877b20e945d6e3bbd28 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/71174f511d534b1e88510ab22d138b01 : SUCCESS in 18m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d191e32f418d49b1818a9b6ef4332a06 : SUCCESS in 8m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b958ebcfe81c4b40a19b6d8c6360a268 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd0e957926e745829d7306f7297c9fd1 : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfd2ac2384d041098b327ad5e2b67c28 : SUCCESS in 47m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a0d656b8067d44abbfa1d9f4b6fb83d8 : SUCCESS in 43m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0f778ad19a24361a4781ea7d10c2834 : SUCCESS in 53m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65972ff8254c46f78c9a338e6e1cc3cd : SUCCESS in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d341aa831b1494fa9652ed5006303b5 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e3e29f9d11bd43e998fab84fa3f896de : SUCCESS in 22m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1139dd96800645c7bf208fe75ef36daf : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":9},{"id":"3633683b2712e83481a4805e084249afec0ac4aa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 19:42:23.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"21ce606e6eee22c9a277cac68eb2ae3716f58475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 19:43:06.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"c64adfaa71ed44573bf02e32d01d7c08ba4609d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 20:41:05.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36b514ba481c4e19b942df8bf8e7bc4b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bb115688894446109408ab470fe2146d : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cca53cd4dce548d4b2d46fa095d34ae8 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6328cf1c4384d4d9a2030abae40c5c7 : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/e3c65ab7ab5e4ae086e26892cc4d0362 : SUCCESS in 47m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b78105abf22247af83cf92996b737380 : SUCCESS in 45m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab0044dc22bb49ab8e6d28397b41f032 : SUCCESS in 41m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89d6ed98c08046f79a5f382e193d1d57 : SUCCESS in 39m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/086a8305f08f47d288edb7454f15cd73 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/961ff666a9144e54a534e648c3c761ab : SUCCESS in 15m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6440e8f4f3824a72b3ca802b4a624d88 : SUCCESS in 18m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1d02cef4cc084f5988e70380346e79a6 : SUCCESS in 20m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9f96035e7bf842e4b4b7489178062c9f : SUCCESS in 15m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0de4199810b44f30b7c5a77932b46f58 : SUCCESS in 22m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77dc7061b6eb4f3ab66ca01e2f5502fa : SUCCESS in 10m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d7ec23ff11f4d1cb841afd426ce6206 : SUCCESS in 11m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d45fcd9b4479455396d61eeb2ad2dce8 : SUCCESS in 11m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/74628de1c3f6430abcf3b8575447867c : SUCCESS in 9m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/480f23949ac9420e87712804b8a54d42 : SUCCESS in 9m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/afa9ccbc7e344c9b86cdf422725d6163 : SUCCESS in 10m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7cf4feae33384f03a3dec90d17ce82af : SUCCESS in 4m 26s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2c581a1bc5f54c3e8018229b123aa4f5 : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/892cda8003a141cfbad8db3da6b09b62 : SUCCESS in 44m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f3e7118fc4064ef8a980f68d4a69ff5d : SUCCESS in 30m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/823b6b560a8743c583d2f7a7aeb67b56 : SUCCESS in 27m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eda4bc5388e240f0a85123fe43a75f90 : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ecb012b99ed43e793343044fcdb5cac : SUCCESS in 3m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eed8a85015124a10846d16dc7b4dcf97 : SUCCESS in 17m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dd8f04654314e4f992ef95d6a1f562d : SUCCESS in 48m 31s","accounts_in_message":[],"_revision_number":9},{"id":"9b99d355f10af0e7cbd9415ef502042a242543f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 20:41:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"280ec4048a8325466ef526b252d537aeb6bae884","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 20:43:36.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7a34c36b67342218eec3af1df12703b\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/108c6a214ebc4cea84f821ddf5cd996e : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b709967e90944b5da6f9ce57903f22f6 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45d1321c3c2548cf9a639d036fb3bba3 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d3e781b245194a878e8657b2aeecfeb9 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cd1be14bac021a5952ffadc0a5369da593aad76d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 21:45:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce5d8691691042f3aadeed0f3eb79a78\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce38e2f1239e4313846da5a9ff2b6491 : SUCCESS in 2h 27m 11s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"c3f1ed3df178ffa09dec02346f3644ea8970c203","revisions":{"12c8feba28d3fcf0edad3a5c17b048411d440b63":{"kind":"REWORK","_number":1,"created":"2024-10-24 20:34:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/933369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/933369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/933369/1"}}},"commit":{"parents":[{"commit":"0fdb93c980651d5672d000b354410d3ca3d1cc43","subject":"CI: use py36 constraints for py38","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdb93c980651d5672d000b354410d3ca3d1cc43"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:34:01.000000000","tz":-420},"subject":"CI: Move a bunch of func test jobs from py38 to py312","message":"CI: Move a bunch of func test jobs from py38 to py312\n\nChange-Id: I8c65c7829fbe167e0c79239d577c38e90b93b5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12c8feba28d3fcf0edad3a5c17b048411d440b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12c8feba28d3fcf0edad3a5c17b048411d440b63"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdb93c980651d5672d000b354410d3ca3d1cc43","is_merged_in_target_branch":true,"change_id":"I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5","change_number":933363,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"92b608940cd82f82a9c5d558e86f214ccf06cea6":{"kind":"REWORK","_number":2,"created":"2024-10-30 16:39:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/933369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/933369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/933369/2"}}},"commit":{"parents":[{"commit":"c3280e5337f15987e595c3854821c398f5e363b0","subject":"CI: Drag forward more constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3280e5337f15987e595c3854821c398f5e363b0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-30 16:39:01.000000000","tz":-420},"subject":"CI: Move a bunch of func test jobs from py38 to py312","message":"CI: Move a bunch of func test jobs from py38 to py312\n\nChange-Id: I8c65c7829fbe167e0c79239d577c38e90b93b5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92b608940cd82f82a9c5d558e86f214ccf06cea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92b608940cd82f82a9c5d558e86f214ccf06cea6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c3280e5337f15987e595c3854821c398f5e363b0","is_merged_in_target_branch":false,"change_id":"I0e55d8b4c5ed520c0aad8cc952a9fa1683905356","change_number":933680,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a2925468d8c860184b6ae0b1380bc971b815f970":{"kind":"REWORK","_number":3,"created":"2024-10-31 05:43:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/933369/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/933369/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/933369/3"}}},"commit":{"parents":[{"commit":"0fdb93c980651d5672d000b354410d3ca3d1cc43","subject":"CI: use py36 constraints for py38","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdb93c980651d5672d000b354410d3ca3d1cc43"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-31 05:43:25.000000000","tz":-420},"subject":"CI: Move a bunch of func test jobs from py38 to py312","message":"CI: Move a bunch of func test jobs from py38 to py312\n\nChange-Id: I8c65c7829fbe167e0c79239d577c38e90b93b5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2925468d8c860184b6ae0b1380bc971b815f970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2925468d8c860184b6ae0b1380bc971b815f970"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdb93c980651d5672d000b354410d3ca3d1cc43","is_merged_in_target_branch":true,"change_id":"I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5","change_number":933363,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2e0377181a173b632fc2cb5ea7b77de5dc0afd6a":{"kind":"REWORK","_number":4,"created":"2024-10-31 06:10:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/933369/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/933369/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/933369/4"}}},"commit":{"parents":[{"commit":"0fdb93c980651d5672d000b354410d3ca3d1cc43","subject":"CI: use py36 constraints for py38","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdb93c980651d5672d000b354410d3ca3d1cc43"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-31 06:10:15.000000000","tz":-420},"subject":"CI: Move a bunch of func test jobs from py38 to py312","message":"CI: Move a bunch of func test jobs from py38 to py312\n\nChange-Id: I8c65c7829fbe167e0c79239d577c38e90b93b5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e0377181a173b632fc2cb5ea7b77de5dc0afd6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e0377181a173b632fc2cb5ea7b77de5dc0afd6a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdb93c980651d5672d000b354410d3ca3d1cc43","is_merged_in_target_branch":true,"change_id":"I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5","change_number":933363,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a2630df62942a30cc17701641d5c27bd65863ee6":{"kind":"REWORK","_number":5,"created":"2024-10-31 16:00:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/933369/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/933369/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/933369/5"}}},"commit":{"parents":[{"commit":"0fdb93c980651d5672d000b354410d3ca3d1cc43","subject":"CI: use py36 constraints for py38","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdb93c980651d5672d000b354410d3ca3d1cc43"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-31 15:55:18.000000000","tz":-420},"subject":"CI: Move a bunch of func test jobs from py38 to py312","message":"CI: Move a bunch of func test jobs from py38 to py312\n\nChange-Id: I8c65c7829fbe167e0c79239d577c38e90b93b5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2630df62942a30cc17701641d5c27bd65863ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2630df62942a30cc17701641d5c27bd65863ee6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdb93c980651d5672d000b354410d3ca3d1cc43","is_merged_in_target_branch":true,"change_id":"I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5","change_number":933363,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6f665d47d9b1b8ca76c651b7aac8766334f4a140":{"kind":"REWORK","_number":6,"created":"2024-10-31 16:19:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/933369/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/933369/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/933369/6"}}},"commit":{"parents":[{"commit":"0fdb93c980651d5672d000b354410d3ca3d1cc43","subject":"CI: use py36 constraints for py38","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdb93c980651d5672d000b354410d3ca3d1cc43"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-31 16:19:30.000000000","tz":-420},"subject":"CI: Move a bunch of func test jobs from py38 to py312","message":"CI: Move a bunch of func test jobs from py38 to py312\n\nChange-Id: I8c65c7829fbe167e0c79239d577c38e90b93b5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f665d47d9b1b8ca76c651b7aac8766334f4a140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f665d47d9b1b8ca76c651b7aac8766334f4a140"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdb93c980651d5672d000b354410d3ca3d1cc43","is_merged_in_target_branch":true,"change_id":"I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5","change_number":933363,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"27a255d461c69e18e20bb5303ac3c2dae98201a4":{"kind":"REWORK","_number":7,"created":"2024-10-31 16:22:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/933369/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/933369/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/933369/7"}}},"commit":{"parents":[{"commit":"0fdb93c980651d5672d000b354410d3ca3d1cc43","subject":"CI: use py36 constraints for py38","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdb93c980651d5672d000b354410d3ca3d1cc43"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-31 16:22:28.000000000","tz":-420},"subject":"CI: Move a bunch of func test jobs from py38 to py312","message":"CI: Move a bunch of func test jobs from py38 to py312\n\nChange-Id: I8c65c7829fbe167e0c79239d577c38e90b93b5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27a255d461c69e18e20bb5303ac3c2dae98201a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27a255d461c69e18e20bb5303ac3c2dae98201a4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdb93c980651d5672d000b354410d3ca3d1cc43","is_merged_in_target_branch":true,"change_id":"I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5","change_number":933363,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ece954b5f3ff81e9b7ce1cc7a7d51a1e7037ba06":{"kind":"REWORK","_number":8,"created":"2024-10-31 17:44:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/933369/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/933369/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/933369/8"}}},"commit":{"parents":[{"commit":"0fdb93c980651d5672d000b354410d3ca3d1cc43","subject":"CI: use py36 constraints for py38","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdb93c980651d5672d000b354410d3ca3d1cc43"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-31 17:44:20.000000000","tz":-420},"subject":"CI: Move a bunch of func test jobs from py38 to py312","message":"CI: Move a bunch of func test jobs from py38 to py312\n\nChange-Id: I8c65c7829fbe167e0c79239d577c38e90b93b5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ece954b5f3ff81e9b7ce1cc7a7d51a1e7037ba06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ece954b5f3ff81e9b7ce1cc7a7d51a1e7037ba06"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdb93c980651d5672d000b354410d3ca3d1cc43","is_merged_in_target_branch":true,"change_id":"I9046c6ea91c98f15c4c491ce3de4a5aaf682f6f5","change_number":933363,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c3f1ed3df178ffa09dec02346f3644ea8970c203":{"kind":"REWORK","_number":9,"created":"2024-10-31 18:05:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/933369/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/933369/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/933369/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/933369/9"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-24 20:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-31 18:05:15.000000000","tz":-420},"subject":"CI: Move a bunch of func test jobs from py38 to py312","message":"CI: Move a bunch of func test jobs from py38 to py312\n\nChange-Id: I8c65c7829fbe167e0c79239d577c38e90b93b5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3f1ed3df178ffa09dec02346f3644ea8970c203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3f1ed3df178ffa09dec02346f3644ea8970c203"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"288677695dbc24730f4f7f114dc2c6c418057cc4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
