)]}'
{"id":"openstack%2Fswift~933379","triplet_id":"openstack%2Fswift~master~I289d3e9b6fe14159925786732ad748acd0459812","project":"openstack/swift","branch":"master","topic":"comn-db-log","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-10-13 13:31:33.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-10-13 12:23:50.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-10-13 13:31:33.000000000","reason":"Change was submitted"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-10-13 13:31:33.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-10-13 13:31:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I289d3e9b6fe14159925786732ad748acd0459812","subject":"common.db_replicator: log container, db paths consistently","status":"MERGED","created":"2024-10-24 22:53:14.000000000","updated":"2025-10-13 19:42:04.000000000","submitted":"2025-10-13 13:31:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":353,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"933379-comn-db-log","meta_rev_id":"2709dd5aadbc71db1d03dd166e6b2a886ad17690","_number":933379,"virtual_id_number":933379,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-13 13:31:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2025-10-13 12:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2025-10-13 12:23:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 22:54:29.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-10-24 23:42:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-11 16:41:00.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-08-07 22:25:46.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-08-08 20:07:09.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2025-08-11 15:58:26.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-08-25 16:22:47.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-09-30 06:40:40.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"1bfd2efcdf0881e06fec5383af6636a17c753823","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-24 22:53:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1da1a323fb0969625cf0c021136e43603b76805","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-24 22:54:29.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"884065ec03fe24b1160ad4679dd107cfb059aa43","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-24 22:54:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7c1c538843fb335247aa64cffc0d588a38ec248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 23:42:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4af7c03eecdc449ea2d4b92f6ef7a3f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a3c538ca4f54b2d96b5c9e01407dd29 : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/3485e47d79d6425891f1efc953b56e92 : SUCCESS in 41m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb0f5d2818f947138506422e21fdec4a : SUCCESS in 46m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8e6f55a4e09c4d4ba39fa1a0c04a3dc2 : SUCCESS in 40m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/791d70de358f49638747b1edd6ce7352 : SUCCESS in 47m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56d0911ac5e94e72b8730978aaaee724 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5330c3b0c4f4cca9c1b26a04282497b : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b9f66260f704a7bb183f1ec6c438018 : SUCCESS in 18m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a6794722bdd40e3ac7d524ea36fe843 : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/65f4218aa2584717af9170a0d26b7749 : FAILURE in 20m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33c6de2fcf3a4f77a337069dd04c96ed : FAILURE in 15m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8f8c9f975ff34c69b8a851b96cb32fff : FAILURE in 2m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/93012edf695843e7950ec9f12cad5c2b : FAILURE in 16m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a40b3579246647d696306b71da2fe326 : FAILURE in 34m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/536d0fc42aad4b57bf67e0d708b4d801 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe409a62e12e4f5cb071623e29d38861 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e671b4bebb8143e0b97efd6ed806f719 : SUCCESS in 17m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4e9a1d0a17da43adb793e3b0b19a0b36 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/77a0b989383d4055a4b3a7501e73fd48 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/825d7f2cd15f4617bbe70a83b322616b : SUCCESS in 13m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03f0b164408747e4bdd5d503df48459c : SUCCESS in 7m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e754183e05c64b1caa77152e7a31533c : SUCCESS in 13m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6180dee3df0a4436b7ddfcb7198e1dbb : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/538e7ceba26d4539944557e75d46e1fd : SUCCESS in 43m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c070ccdcc97472ba936e5f3909c5366 : SUCCESS in 33m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/520e9a7e60734600a5a1770602bca9cd : SUCCESS in 31m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c0a35823223465792ec6141afbeb104 : FAILURE in 22m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e96941e5dfa4abd806dd68703edd67b : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a9006533f6d84a58a48f92f42cfdd36c : SUCCESS in 19m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb41bfae0dfa459d88ab79538ff41bfd : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"32fb3abd04f1786aafbd73d7a5c0b57e3ca26752","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 00:29:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/304b655b13584d8ea2ed0d149d0a0c09\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9402016bc2a64c0dba43d5275a873454 : FAILURE in 1h 34m 23s","accounts_in_message":[],"_revision_number":1},{"id":"54dc14e736c0b169459e8138cb3be09047efacb1","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-28 16:03:44.000000000","message":"Patch Set 1: Code-Review-1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"d2e69db3e75ed107bd4cd535d16914ec306809e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-21 22:40:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"696cd55f1f8650252558951d5cbecfc3778de3be","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-21 22:40:34.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"f18239310f9fd50d7630e5807d00a35810f08707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:54:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75bfb67d6028408ba2dfa112d4ec3ea8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6895292cc744ef9ab8a3dfa44e48b57 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/cd549636798448a7b0101002c5949e88 : SUCCESS in 39m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e96a0737b844b16ae6afb53eb20f404 : SUCCESS in 45m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa85629b48b546e7a22a4c5e2109da6f : SUCCESS in 52m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0d1eb28b2d34625ba1665a5cc91017e : SUCCESS in 1h 07m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c37003ac24a543dda4caf4aa860834a4 : SUCCESS in 13m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/182a85a69bac4bcd81e7456992b362a5 : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ad398d1dcc7424dae877acb4deffd85 : SUCCESS in 23m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/402ce5d38f154049acabe6bd84d556b6 : FAILURE in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/378cb60980f94475b2e927fbaadb4fee : FAILURE in 13m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4cfd2bf52d9e4047b295c9ff0580e453 : FAILURE in 11m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a38fdb259734f30880f75fee332be1c : FAILURE in 22m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2fd8396cfcb6436396fe87f3e6d6d0af : FAILURE in 29m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/22efc04e76e94a79b196e5d325e56cfd : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/60303958ef004de59eb8677ec085a797 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5729ae7261374f7aafd6d1a332c89573 : SUCCESS in 20m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9a5aa451c27841cf9e6a10d559ab21bb : SUCCESS in 10m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d8ee8161bfc545cc8984e2904525c6ed : SUCCESS in 16m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f9cf939f6f994914996543910b702b3f : SUCCESS in 14m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b92d98a61674eb5a3f848f789391711 : FAILURE in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5ca47892e40c42d7a992deba21e3d15e : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb6f14cd7ec64240aa69698b7a1c3a6f : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e1fb520655242d3aa7e1db8694eb953 : SUCCESS in 49m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/648554eb30ed42a88e29686548b4465a : SUCCESS in 59m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e5a7d80d177421490982f3444754b41 : SUCCESS in 53m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/deb68e0456c942f8b4369d1a6bc1affe : FAILURE in 19m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcf2c9239a9042c3a8d35dac775bc3d6 : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5bb7442cc7cb401c86f31a81d33dd0bd : SUCCESS in 17m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/965ba99820784b9c82cba0e365cfc5fb : SUCCESS in 41m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6f7b847f275f06c5a51b9ffc839a66bce01b5de4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 00:39:01.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0d8ea5d049a4119afc8fb5573303d79\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1403949240114d4881fa2a51a06d6f1b : FAILURE in 1h 56m 34s","accounts_in_message":[],"_revision_number":2},{"id":"428165edfbdd6214dba8a617335e09fbdf94a703","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 19:14:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"48e665c23797664327755775f3397c3e3242de91","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 19:15:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"6e1ed7259bbff6a8a305d73a6c0b3c1ed7fe91a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 20:25:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc149ea544b746b38d286cbd879b3d34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ea252be72342df91b3366c09e616a6 : SUCCESS in 5m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/a6a4594c5ccc451595252ab13696ab4b : SUCCESS in 39m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d61142558174754874d361088068a68 : SUCCESS in 37m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a437656bbbfa436da858bd39bc8b4ae0 : SUCCESS in 43m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb861916929f400d9e7a572baf4786ca : SUCCESS in 41m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1eb2898ffa6c4bdebc34273f4f9ecd76 : SUCCESS in 24m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/279ffff3cdfa40828ba6944d8a7d3364 : SUCCESS in 24m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17ad613c07644b4fbd92858ed9269d6d : SUCCESS in 24m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8579a199cc914fef806833257a94e8ac : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dec08578fcd5415fbca3dfe5557ee5d7 : SUCCESS in 21m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f15e634a3aed418b9839c5c29742f5b7 : SUCCESS in 18m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1640d23ac96a4b59a3d73232c2eb0ba0 : SUCCESS in 23m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bea36bf258334e05be2a2a64400faff1 : SUCCESS in 27m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f99a894531fc4b189278be8f3a254b59 : SUCCESS in 20m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20bbe0541c19416a98fbb3925609db98 : SUCCESS in 17m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/04dfe656ba6947ac88690e02d24d916c : SUCCESS in 20m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/79351b50be094ea0bbf72dfea25449ff : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3c76668c6390468cab4fb16a08d7521c : SUCCESS in 16m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7402af8afc374fa3bef10c39fa242cf2 : SUCCESS in 15m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c93b9c5740954b06b728f10bed367f57 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a1869b7a051e43308b12c774b8888761 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/85e83c9f93124d599541561f7bddd386 : SUCCESS in 6m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa091c2438df48d7b913dfc1731283f0 : SUCCESS in 51m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc515c146fe441ccb7afb7adeebde5e2 : SUCCESS in 27m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd8304e3d07649398f791e30761da37b : SUCCESS in 52m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a08a3c2b5ed4fdfb5c7798ba57f1f62 : SUCCESS in 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4530e731551b4951bbd503d3006f9310 : SUCCESS in 2m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a8d1a8528304a74bf918ddeeb35310f : SUCCESS in 20m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e63cda0b71e84ab7b23617c57d543ab9 : SUCCESS in 50m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff3821075c04430abc1361e47de59f37 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":3},{"id":"1d429465f9e6644627019e22fcd9818470e17b14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 20:35:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8231f449f6464bb58bb9c8c131c74c40\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e977f391c4a48c1acbff53fa53abcb9 : SUCCESS in 1h 18m 47s","accounts_in_message":[],"_revision_number":3},{"id":"ab0313a96e4ac8f1e81227c8cd11b3e421c7b347","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-11 16:41:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(16 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e4cdbebbfd3d28113275394c2b5e45e07094d2fb","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-12 17:11:15.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"7c7ccde3b22fc397f735d2d6107681ed49f64dde","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-17 20:00:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c4993b7ed00b3949ff3357bf6e42c3f02b60d438","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 00:26:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b15c2ce943e14d6ba5ef988e34da2976\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad8a6c22a8514f968fdb448b8fd0821a : SUCCESS in 1h 30m 42s","accounts_in_message":[],"_revision_number":4},{"id":"11445fb504089a47dacf455bcfa67a50c3cdaf9e","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-03 17:17:57.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"1d3541b1bd05a218e92efff018cafd6fe43ceed0","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-03 17:18:45.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f82501e97267794a18486fbc7dd814285c8d0ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-13 17:13:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f9dd9d0f5e6a9f4073e3db16b94fe3e8bb103233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:46:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b57cb2fc5974b67be7c9bc9ad56dd22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46570e707739442b8429620467ae6ab2 : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/3118335a6b1e4aafb217dfbbae54a670 : SUCCESS in 1h 05m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d8885269b954739824351e5612ba76e : SUCCESS in 1h 21m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/163ea3abf4fe4b8fbba51a722b5bd86d : SUCCESS in 1h 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b56e2ad5c6f544188215f2d325f2c6a8 : SUCCESS in 40m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93357977790b43c69287f48dcc0b4a26 : SUCCESS in 22m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1a9181ea6a94819abe93104a443f4b6 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/298f50b297c64e3aa77e04d2026551ad : SUCCESS in 23m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cb802da46fba47b988f664c053cb46fc : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1301506ebeab413c8d19c8b4bdca002f : SUCCESS in 19m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/20d10079b4ba4ecfbbf0eddb9db6999d : SUCCESS in 21m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5197bcac9dd24bb28bb5014f83ed285a : SUCCESS in 23m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f9fa7a9b46674d589b6d7f43be353830 : SUCCESS in 20m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ac736c77a1dd48e3bab4b4d8be8d66a7 : SUCCESS in 17m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/573cac567bd24ff6b329e52ca5bf2e3b : SUCCESS in 21m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75112a46b090452ba9e55a525b85f5a9 : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5edd448b4f1e49dfa98d623e7a2431e5 : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9283ff2de664052b50d8915f13afb23 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fe03a769146400b804cdee2407e6c17 : SUCCESS in 1h 03m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25ad8b0ff1364c25ac3e6e6830af0f05 : SUCCESS in 1h 02m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8852adf362e54602bd3ef9e11f19680f : SUCCESS in 52m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e6d25081a51475ca82becf0eceb00d8 : SUCCESS in 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd123b0f59454fd399ef1c343206fb6b : FAILURE in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0c6aef343e542f3b5b0ac9ca11644b7 : SUCCESS in 18m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f31b32b8cda549df95cdffd456e52a05 : SUCCESS in 54m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6266fcd521994e7f843e5bfd0290aa7c : FAILURE in 38m 45s","accounts_in_message":[],"_revision_number":5},{"id":"198c9836d96a959aef49498ba196045ef996adde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 21:24:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe7a72d6b6d74518a61f9cefee2bc8b5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba1bae972f4e4a418ebc283d0c8525cf : SUCCESS in 2h 30m 42s","accounts_in_message":[],"_revision_number":5},{"id":"0e05a1bd7f164d300239dea23b361b97e78111f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-15 20:34:39.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1dc8b7b6e8fc21a1da61b55ff42c4c4e643d4d2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-15 21:24:49.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"c025bcfe7257d5f262e94b581ba23844091c5944","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-15 21:25:08.000000000","message":"Patch Set 7: Code-Review-1 Workflow-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"75ade03c64be144fe14664c5d47607686514505d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 22:51:01.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c13174e6669e474391edd0d690d996f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1b53df7a354482bf9a471c50d6ba54 : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0ed908e0b4384028a34ee46896f646be : SUCCESS in 1h 10m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8aa3067037ea42209f7dcb6c4f03a81a : SUCCESS in 49m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/06f0b04f64654212add036ece0082dd6 : SUCCESS in 43m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e8d90add12d4786b79a43ecf2039007 : SUCCESS in 1h 04m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a36d823c64f94083a7eef448030bc7ec : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfe3d241518f4c168c02d1040880b94f : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/486eb56de6374bc7aad7d3fe34b44ca3 : SUCCESS in 12m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/620f15085bed4a5fa0f9329f583f444b : FAILURE in 4m 02s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9cc83d506c3b445ab80639d71a707ea7 : FAILURE in 9m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/da2a275b84e74113a287c471b3aad811 : FAILURE in 22m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3433fcfa2c644e67b58ba8038dd2e5fc : FAILURE in 21m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fa299b45fc124d5b9c2585f371d5c899 : SUCCESS in 12m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8eedeec65459454ea68954606cb90ed2 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/49ef3bae93194877a904e3a4959495dd : SUCCESS in 17m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c89ceaf516db492ca9776204111482d0 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10bf69f645744d7c849c3c49e756094f : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd4b021652a94affb0a796cc3413ee22 : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea36519bfdf34fc982946b9f0755d565 : FAILURE in 1h 00m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00e0a084bca94d55b92a8d0109f605d2 : SUCCESS in 34m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72c5225eb3624f0cb04634ca749e11d7 : SUCCESS in 46m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2df6de0629194038915bfb04b1c67ab9 : FAILURE in 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc3e72b9e1b43698080af5bc41e2b26 : FAILURE in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6a5a9aa69d34638a5a381dededa288d : SUCCESS in 17m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9667d65f3c724335ba21efe35a90e1d7 : SUCCESS in 54m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7634df6ba183415183a130b4190ada90 : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":7},{"id":"6c9cd2b25a88d537591197926669a1e8153fde2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 23:15:50.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2456fd57cb5d4d35909c4c37cf85d260\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/92157c399b424afcbe091abbd304b248 : FAILURE in 1h 44m 47s","accounts_in_message":[],"_revision_number":7},{"id":"052a4ffc7aea68af4098bcd19b4aceec4844a052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-01 22:06:02.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"69884ed247980c5a0c590f70ece6098b1194074a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 23:10:57.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9941e15c707a40b087cc0aafe91666cc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/06cef9d93d4e44a6a9a9eb5df26e0e73 : FAILURE in 1h 03m 11s","accounts_in_message":[],"_revision_number":8},{"id":"3de09ad940a6f472653940f7b07f713e95419056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 23:19:20.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dc24256e4c24f199e34d4f73e860eaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba185993c0b4d5bae3cd31725be96b9 : SUCCESS in 5m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/cf548712510c4c3fa0d5725ffa28130e : SUCCESS in 37m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db063eec1269476b8aee417e998ccf6a : SUCCESS in 35m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f843d5ea6be847fe8dd16f6a738828e6 : SUCCESS in 39m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07ea693fce1e43c8b787720f533f23bc : SUCCESS in 1h 06m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf0a736f62ad4d36a1029d0169f11d6e : SUCCESS in 12m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09c95951f96545dab00bf4edb0d4c0a5 : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eed5efde79284d0c9d4767d2434ee132 : SUCCESS in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b7394064c18848ab8e219bc1f7aaeea2 : SUCCESS in 4m 31s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2be1c4295c0b4e43917beee19e8aeaa5 : FAILURE in 15m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5484b925707c4506a7a3dce72cd9688a : FAILURE in 16m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/62e42b17c7a74fc58c72dfe758c28b76 : FAILURE in 17m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6e4eddd82da64908a7e81d2f976c3b42 : SUCCESS in 13m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b16c9bc8676844f9a5169c29c0ad57f0 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5f05ed1d58a0410989531ba8f9c683c8 : SUCCESS in 14m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9f90c5c6e4a4a06a61a2c77322d2ad2 : SUCCESS in 5m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cfcea53ddb7242b098a3b6d552aa7f93 : FAILURE in 13m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b6a82ad82fd42adb84956814b4f85f5 : SUCCESS in 5m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ee48690eacc4f06b7cfc67ac09cc4cc : FAILURE in 58m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fde5a1dca5a74bab81262477c230dc07 : SUCCESS in 28m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c8ddf4d066c4cb9b32a86beb004f257 : SUCCESS in 30m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c18410f0ebb4737af4f1fba20b113b4 : FAILURE in 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/938e8b7824cb4211addeecae53892426 : FAILURE in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/787194e5682b4480b48db7cc197d3531 : SUCCESS in 11m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c382d9a94444d3d98d9ffde007f117a : SUCCESS in 53m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/474e5a0640c648a3ae0d89a4e8936968 : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":8},{"id":"f5415789e0a07980d222c1af62d9aab6bfabc7f2","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-03 14:18:15.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"39e3fb3f671d0d49172576e14c4cef81d7909240","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-07 21:12:45.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"13b3c21ae375d01734d0c3b7f5f7fa1ef25848c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 22:21:33.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2083388c9a4a4c8faab1a52e2675eccb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d05bafc79b4592afeccf9c1715a1dc : SUCCESS in 10m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/9b81fdb9d9414d9ba942b4b62ceac8b6 : SUCCESS in 56m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34b785f4fb9443359a43c39488afd9a2 : SUCCESS in 55m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93f015d1a70a48ca8549e84ec01c9eb2 : SUCCESS in 50m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f496850a9704d4dbf28a75613ec3d76 : SUCCESS in 1h 03m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45505a2b71ba4757ad8a155bd31ada02 : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f2e7d61314548018a44812e203b49ab : SUCCESS in 11m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2f0aedb738a45fda7d5782222958c63 : SUCCESS in 17m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e5afd9ad19ee45539977cfecf55af823 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/783153c69f4f46a6bf294a130e91654b : SUCCESS in 9m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bdff429bf5414aafb1a67def8878ef28 : SUCCESS in 15m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/12d408afe69e40e1b68b97d7bcbb31a7 : SUCCESS in 27m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bbbfe21126d34631ad223f60929ffcce : SUCCESS in 29m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f7e877acaded42259f901232ec128485 : SUCCESS in 7m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/46df9fb2cc73458e86035bb224a068b9 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8cfde42845974a4dba04c0e6ef47e1dc : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a079b185b6c94e71a3f76d64884f2488 : SUCCESS in 4m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b945ae998e4e4fbda8393143ae825836 : SUCCESS in 8m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3bcb9bfc399d4187b4e4876a22da2579 : SUCCESS in 5m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/00075365da6a4cebbdfa33a46ee91f77 : SUCCESS in 59m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3813db6cff1f44a2b15ca9b8b72bf4c2 : SUCCESS in 51m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2db851ddb61244da94485788ba6cc62a : SUCCESS in 21m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7eef71466ffa4e31ac515682bfaf11ce : SUCCESS in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df87296fd564873a7145f489b28ad73 : FAILURE in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d2d59e400704b17a875acaaad820bbf : SUCCESS in 18m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/234bfbfdb3c544bb977caee5ed47bb62 : SUCCESS in 53m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bf2c78410b24bf08b2a7b9e4c77cd44 : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":9},{"id":"40d498f2afb354fc57c2d37810e50aaa720956e9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-07 22:25:46.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"57b95f1f8acc79f7868dbbf9b4eb41c3ca63afb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-07 22:38:50.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"c11a5002eae6caf7eb20590a2e06d14db83ce311","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-07 22:40:52.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"07d00994144b3006f8a905114195eeefc9800f02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 23:50:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/286573ad930b4f6993c6df7f5db0a279\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e8ee1c11a3c4afcb7d3054778bc8d78 : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/1c9ab1424d1748d18ca0f30b3aa5e790 : SUCCESS in 33m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e3e7f2f5a0343f882c61453d6491ebf : SUCCESS in 1h 05m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c4f90a67ad6c440e920c83f6f42f31e0 : SUCCESS in 42m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f98e9226b16449949672271ec8303e58 : SUCCESS in 51m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82aabb4e5be9422f91ebd34872cb2b6d : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0bb68ca1cea4b3abb293c1c10aba7f6 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afa3e6d21ef648ca89d213dc6762d634 : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42afbeee4bdf408b9b175a20a23456a2 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/963e01a61a56482c855b3cecf6a9ceda : SUCCESS in 10m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e6dc4ba5554d43fc8082ce379688fc87 : SUCCESS in 18m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ac6565a6af694ad4b782059d1ca657fa : SUCCESS in 10m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a72706c91a6e42c9923c32dcafa88fb0 : SUCCESS in 26m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5481a65242b34ef69b001013549fa20b : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e548729baf774d42b38fd7c70d78b7e5 : SUCCESS in 19m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6de25809cab64d898979b543796b0210 : SUCCESS in 7m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bfce95c92b4e41b7b2373f94dc6ed0c5 : SUCCESS in 6m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce82379e1978460580de40066293fb3f : SUCCESS in 9m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ad66baeb22b4225ad4ebf3dbbbf9582 : SUCCESS in 8m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a78993786d654b46899e8210203b4c09 : SUCCESS in 53m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/454d40e7898944e9a86b47251946be8c : SUCCESS in 1h 05m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/583bbc028d644c25ac7584e238e439fe : SUCCESS in 48m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bed58aabb1b4d1888497b9ec0bab6dd : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc8e7bcc1d445f79c57a3928e65249f : SUCCESS in 3m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a1803d268754158bed727bf60d18e27 : SUCCESS in 17m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e69876263a441cfb25e6bdca74d238d : SUCCESS in 28m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0afa1f14d35447969360d6b88d36fd5e : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":10},{"id":"22202803e233f425674245683d2fbba832e3e77b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 01:57:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/584a905c865743e3b12419483eee5282\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/24fcf17c451a4ba88dc5e3c83ed6da5f : SUCCESS in 1h 30m 30s","accounts_in_message":[],"_revision_number":10},{"id":"875e1a38455dd65abd2f2109d74e1f4b4241415d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-08 20:07:09.000000000","message":"Patch Set 10:\n\n(13 comments)","accounts_in_message":[],"_revision_number":10},{"id":"5524a4d30e1fcea91c4caf147b9afaa107c40cf8","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-08 21:02:03.000000000","message":"Patch Set 10:\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6d07b91eec2a42cc7665095c8b143fb82d1c59a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-08 21:02:48.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"a973a0cfffd45fcf78e789a2f1f58e860e72c6fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 22:07:34.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e74a1622af4e55ab45180146b66870\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63757f5acf1847c2a20e9a0c1fd05f5a : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/820c2eef4d934010b2c0a045a8bc05ec : SUCCESS in 1h 00m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20e350434d064138beb5ff8284d419a5 : SUCCESS in 59m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8cdf21b3c60e458bac4e262d96660575 : SUCCESS in 47m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcfb288e83fa4c7e84300205b7b07bc2 : SUCCESS in 55m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c0676ae3b6b42a0b677be791aee8e1e : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65f71713f1e3489c92c5eccb2623c6bd : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a73a9533a0b4024a0881fbd20ecde67 : SUCCESS in 20m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/342e7f205fe5444d9fd3b1a01dcdd5f1 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7be2b516527e4f029c6f9943877ba6a3 : SUCCESS in 23m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d0de3740ce564c30a88440cd1e2b1f98 : SUCCESS in 19m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/13702dd4b8924a049633988a3eceb55e : SUCCESS in 23m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/14c70aa264aa493d84eac16968c4c69b : SUCCESS in 24m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bf6566b2de6441bbadf03f0fd87335ef : SUCCESS in 13m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7c18796a6ca045608507993fbbd1f5b1 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f7dbf9aba0c34c9ea13c2ff8e2c45741 : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a77360956b5d448d9842e9f8eb6f0d1f : SUCCESS in 5m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2281e03d986542fc93b1d35a6c7cf28d : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/737f3f4d85c34506a5468dffc0ec177a : SUCCESS in 10m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec9912a4083e45b4b03c19720f85fd8a : SUCCESS in 53m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52769265892e4955be57df2c12a44c8d : SUCCESS in 53m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e24aef31bc4e4bbea52ac0f57b076503 : SUCCESS in 49m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79f7a99c07984e0691b26560c3acedca : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c46c995f38544659d5cf1cb8be86688 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a4b3a4d38e943efab3c8f1d49d63060 : SUCCESS in 18m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7dfab32bdd54e2aa6c447becc4cbec1 : SUCCESS in 43m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e59998c8d4a4c8d8501f9d71f6aef9a : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":11},{"id":"83da5455c53eb8b80496a8e5de9dba59dfa90896","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 00:33:12.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea1d76465fb74b078d8e6cb8ae5b3d0c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/280a73aa3dc34b529286360125899b1b : SUCCESS in 2h 08m 31s","accounts_in_message":[],"_revision_number":11},{"id":"1f9d53f2212c5a88731f9ec10f36255fba3be7ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-10 18:46:46.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"430d37cb5c6ceb67443359b76abbb58768659672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-10 19:51:45.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56b92122882b41b0969403391ef8ae2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d78cad1723469db8764c1d3b34a363 : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/a1b8f6df7439427b861f8227a515a182 : SUCCESS in 29m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/42538e397e79412cb1f6b4a0be0d50b6 : SUCCESS in 37m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9bcf9d2c0ac2485e885955210500ce4a : SUCCESS in 53m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6071cf03cc2f4f519aa2b4176712599c : SUCCESS in 57m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/275673c53ae742938e25758c475fd533 : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f63823a06c2420bafc297a33cf9c9a8 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/373a7eba118241be810a0b56cf51857b : SUCCESS in 19m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b162e50962b4e2f8d28c736d4d86335 : SUCCESS in 3m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a546265074db419f8291885154c7d713 : FAILURE in 17m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/247c2788535d43bebfec30672a560003 : FAILURE in 25m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/55efcedc17c147f99f068a3cea3c9834 : FAILURE in 33m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1329874f84604caa9879894b7cbbea7e : FAILURE in 42m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8182649b6cde45368e18a288b3bff801 : SUCCESS in 7m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a33b5ba834b449e3b43d5c89bcd26343 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cc33d364568d481a81fd62028fabfd14 : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/df2764d9e6a14c2c994298539fbc71ff : SUCCESS in 4m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc155a8dba184eb29da380afbcc816d2 : SUCCESS in 14m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2907c285b4814d79b683aa20f6796f7c : SUCCESS in 7m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a0c052c62bc4fc7b9c6910e9523eabf : SUCCESS in 53m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2927190237746ccb4bed5b360e25d9a : SUCCESS in 48m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8fb9e2df9c70408c9243a89e94417f05 : SUCCESS in 56m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ce3487170d64e86832df05e412a782c : FAILURE in 25m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c2b0d9738f4bac84b68229a87d6e43 : FAILURE in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a22bb371b314b6993571616b515adab : SUCCESS in 23m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/419bb4dd764d4163a598d2433c6fb42b : SUCCESS in 27m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e1ee44710024d4d9eb8dbbe0e1baf4c : SUCCESS in 41m 55s","accounts_in_message":[],"_revision_number":12},{"id":"a106cd515585d047e7d0026766f36216476ac5e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-10 20:10:09.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe648bb5ea84601b1fc5c1cc6059cfb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/504a48ffad944e77b31fe403f98f8279 : FAILURE in 1h 21m 45s","accounts_in_message":[],"_revision_number":12},{"id":"53a65637ece9d6a83b24aabf6d955dd08c02e43a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-11 09:22:58.000000000","message":"Patch Set 12: Workflow-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":12},{"id":"14b25c6c29b3546bd16543854b922f950c2c4411","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-11 09:28:27.000000000","message":"Patch Set 12: -Workflow Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ee99bf437df4fe26947fe314236a9cc1d9935536","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-11 15:58:26.000000000","message":"Patch Set 12: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":12},{"id":"5a5835d9ec63412e0296f3eda3e67a216e9eb933","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-11 17:15:24.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"5f9ffb17955562401d6ce9fb50afb5bc90d40ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-13 20:31:50.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"a98db65fbf3bf9a8395253a96c00e099a914b4e4","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-13 20:33:46.000000000","message":"Patch Set 13: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":13},{"id":"1eef177f2c72e5415aa4ae4f8c57a234d41456c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-13 21:22:42.000000000","message":"Patch Set 14: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":14},{"id":"64e8eb771ee3a517bd62cf0ccea313d66f7bdd56","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-13 21:23:06.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"fa5fbfe5f44b631dd1ff6f315fa63a61cc83267a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:29:18.000000000","message":"Patch Set 14: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dff879c5d2a4273bd5247ff6d3156f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6f29910c6846bdb2b0d7ed88aeb522 : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/42145078c8464d669f4171e29d49561b : SUCCESS in 55m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02a13d2b9ed44f34b42050a91f7785d2 : SUCCESS in 51m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3d5ede9ef8041fab629f44e8375e9a1 : SUCCESS in 43m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bd8e23769364924bb0b5f95c401cc26 : POST_FAILURE in 10m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d2666e1a4ed42a3b9c4ad262565b627 : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/209dead894404e949a3dfd8d7108ad93 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03fa85afd15249658e7a43c1279c7163 : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38fedb0191b743e989dfc7f004da1082 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e014be8ef68a471086d5384b911b82e2 : SUCCESS in 8m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/95f83fcb15c14b2fbc77fefc9e45c909 : SUCCESS in 8m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/857aba6a50aa4935a43eb29467ae5d54 : SUCCESS in 19m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/504d185f64b94bbd91cde25c79a9439c : SUCCESS in 23m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/95dc1ee2fefb438e953eef06cee4c55f : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b27783585de448bca26b5bab5bc4fb0f : SUCCESS in 14m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2972163369c54e27a9c9509641f3ddaa : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ae1e0935fb69420aa2392ccb814549b2 : SUCCESS in 6m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/77455684f35b4f35aafe43c738711f91 : SUCCESS in 17m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0feadc1d8f64fc6b1f13845442c3341 : SUCCESS in 10m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/93da1a78863d44bdb0f530a2a88137e7 : SUCCESS in 33m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/934dc4c591a045afb29d2247d7ca0ebc : SUCCESS in 21m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e66d921624d4423d9ffcf71923208bcb : SUCCESS in 53m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6630a3f69eeb4611b014323ad1ec3785 : SUCCESS in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff79040aafd440b3b5bbf3ad4323fbc9 : FAILURE in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f0dd2834468f443f93c209163e01a348 : SUCCESS in 16m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dad838a031c47d7b9564d3fc41e90c0 : SUCCESS in 48m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e530be25f4649dc9632dff0b67462e2 : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":14},{"id":"0ce4ab6df3b1d041e067a32a094b0008f844aeeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-14 01:14:32.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"95b4d2a89c5cf63e32f602d4addc3321c6156ee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 02:15:15.000000000","message":"Patch Set 15: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5faeeb773a3c4026a50414250134255d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80afc2ba99844240beaa901bf125f667 : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/79736fa77bdd4b67a01209d386064679 : SUCCESS in 47m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/873eacf439cd436ea1dbc6cc8d8dac0a : SUCCESS in 52m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/598aed65906147f4bf283d66d95ba121 : SUCCESS in 41m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf6b8678a6364df8b21d308504cf1dff : SUCCESS in 32m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f04c34f9fc645c18648bb57dedad4c7 : SUCCESS in 12m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/402c163d1cba491a84383a0be0dd7600 : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ceb5b704bb04bac9c28a60c1c34f7b1 : SUCCESS in 10m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c3cbf8b72814661a14fad1cee5d1093 : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b86274403b03431ab7d4c88108392a32 : SUCCESS in 11m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2b719bf20d5c4aaa985e321e5c0baffd : SUCCESS in 10m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4317dcf2ad124187bd9a587712943041 : SUCCESS in 34m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/65480b89750e4b469c9322290db4bfdc : SUCCESS in 30m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/026c8312d6774555aea204565a456282 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a966dde3e7094d6b878ac5ddeff9c67a : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/121e3997863a47c8896d3dde4f31cbab : SUCCESS in 17m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/39534ea988bd45ce964aa8648cac3086 : SUCCESS in 8m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f9a48ef376b64f59931c236dc4390971 : SUCCESS in 13m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dcb85860cf6449e38e5722ae5c559983 : SUCCESS in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f3c33d5062c42a092f58043678e7639 : SUCCESS in 44m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a05bbde571b4a3e8020311ae02e1477 : SUCCESS in 57m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4150f89fe23b4886b263ed7498ed17f1 : SUCCESS in 55m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3b3a1a5ae9a44bb948e33cf7d613261 : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcfcba01eafa46adb76466cf451c3ffb : FAILURE in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/250fae1dc90348db8f685a56452b98f0 : SUCCESS in 10m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4251e8cbb2914c759db126ce6805fec5 : SUCCESS in 30m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bda02bace8b5450aa0ffb28a24dddd9c : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":15},{"id":"7a6d0ddede1239b201049a58055303f1c527ddcc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-14 08:41:31.000000000","message":"Patch Set 15: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"cb780ea3b8ccb2ed23032a7133f3d4350a7c7938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-14 18:35:47.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"b5154d571933a7db955d18a26aa6c07ff5193456","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 20:35:45.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99d62090a93343a7b4607189aff4541c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f7e15b43fd2046b4922d87cfd8063f52 : FAILURE in 1h 31m 50s","accounts_in_message":[],"_revision_number":16},{"id":"832b27cc2e066ff1956c053d211b4f2fbda24667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 04:29:35.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/920c6409fa7b41da959ed9dafae1b4d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bfe037f2c2410789a168cf6a09ddba : SUCCESS in 5m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/b22409c11292469a8b72ce9575d604d0 : SUCCESS in 28m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b67843b7b454479491f3b8caacc79923 : SUCCESS in 37m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdee94a28a4f4566b275f2ecc3f25c65 : SUCCESS in 32m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16a85405cc40404ab3fc02b14c7cda22 : SUCCESS in 32m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0de84c2a42342b7955c0de9ff7ba788 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95bb1e91df5b40f5bacb8afab583a885 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7209b954919144f7bda594bfc251646b : SUCCESS in 10m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/488967ddad414bfd86e0dbcfe93a2345 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e30543fa98d44749a68e78b7f76666c1 : SUCCESS in 9m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aa8cf41deb4447db96cba429f577f9ea : SUCCESS in 8m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/af3363ebc8af4f8489bfd9d799efa88e : SUCCESS in 11m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fa0a0a4a867b40959d97954e791cddf1 : SUCCESS in 15m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9cc8bc934c2143378afac14628340724 : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6f7a78848779478e9f51134e48ab7e1f : SUCCESS in 7m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/890194eb9f46477babd384795feea940 : SUCCESS in 8m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7942d9c8af7486e8dbb864ef1574c20 : SUCCESS in 3m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c26c5f7ae11c451197989add02b215ed : SUCCESS in 12m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9c6492c1722f4ce3b7465a1c3e804af9 : SUCCESS in 10m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/005610f06583453caf27f55fe631b3d7 : SUCCESS in 56m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d430070b82a34ad09b8d2d80b9c7bad3 : SUCCESS in 31m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35fce190dfb04e5483b8eec6afae8e81 : SUCCESS in 22m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80ec3cf4a4e04ac48c813620546fd5b9 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bdde1befc564dd2a92e87addf72c1a0 : FAILURE in 2m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d6cb9fc574e4f428a3098838a31c7e2 : SUCCESS in 11m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/722aca78a01143d8ad435626509b821d : SUCCESS in 34m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae95741984ee4d7da6a1634330484527 : SUCCESS in 30m 58s","accounts_in_message":[],"_revision_number":16},{"id":"f977a13ea5aa2f36a3ed551691e4bcaee8612247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-19 19:29:12.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"0d8ab002e6d84ffca0c35cc2f01799ffdeb933b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-19 19:44:19.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"96b25c8f0e6d24348f98f3b8b67fb222a85dacfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 20:59:20.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79807fb0fa454f30afba74d18c15c90a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2139b93efe5946fa9a799b5737b27bf2 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/26986afb6312484ea8199a9b10042b33 : SUCCESS in 1h 06m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74e06c98b7764279a7c5cad74d97d975 : SUCCESS in 1h 03m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3917e4c1769443a38b4a8919800acb2b : SUCCESS in 37m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/255a9b2f549548a59cf09228566712c0 : SUCCESS in 1h 03m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc67ddf2dd30483f890ca0493cad40e6 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7213d52b446f473cae31bc0e96811fea : SUCCESS in 27m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/123c80193dec40669324a8d5618e856f : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b2f2f50eeb345998320f548696c6c57 : FAILURE in 3m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c58d4a5bdd74b7f99a02e34b96c21e9 : RETRY_LIMIT in 9s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5e7b91565da74e5b9a210f0812963fb4 : SUCCESS in 23m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9d92c48d787e44e99eef6014f7b307bc : SUCCESS in 28m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f6e6f75e4e9346dba1e80e9d7f0e2321 : SUCCESS in 32m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ca00040baaa84703bf07c62c546421a4 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1f67d78761b4481186c8ab178afd962d : SUCCESS in 13m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5bcc7f4c462e40b3be8adacea1f53454 : SUCCESS in 15m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7344698a36054fdb92f23769dc575ca0 : SUCCESS in 5m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef7086bbea2a4bcc952c92a2a523efb7 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/30eeb524c2154b8b838a8758442c26bd : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d0aaba652144479ae4687d49a6a0725 : SUCCESS in 1h 01m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2590b6a47b7e40e2adfda6854fb6ae7a : SUCCESS in 49m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a18e90a813c74f43ad6333e6bf967ef1 : SUCCESS in 55m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0fad54ef3e945e8adbcf75d382cb9f0 : RETRY_LIMIT in 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e212bae92d4af8b34a4b72d4b73132 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e4e47b402c141668c1881cc7a38ca77 : SUCCESS in 26m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f43bebd48af4a74866569b72b2118e5 : SUCCESS in 1h 00m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f32441651414781aef639ea6948bb0b : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":18},{"id":"0bc05911656c5fa26168330b79a8878a15ff38bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 21:21:08.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0286b55a91814bbb908a1dc10719ec93\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e39641a503440c68b236afc709d7881 : SUCCESS in 1h 33m 44s","accounts_in_message":[],"_revision_number":18},{"id":"512447750c70d90602b79cdbf33c76dcbf8035d8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-20 13:40:09.000000000","message":"Patch Set 18: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":18},{"id":"be6854f11d11ae3a5f85ef73d190e595cc3c6421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 01:08:40.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"01b45c5494fa8905e6a1eb517e55132e4a18c792","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 01:09:16.000000000","message":"Patch Set 19: Code-Review+1\n\n(33 comments)","accounts_in_message":[],"_revision_number":19},{"id":"cb828b81c1920aef4f5f2fc8c419ffea5f8ed5e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 01:13:32.000000000","message":"Patch Set 20: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":20},{"id":"aaba727c5ca79ba9f4990f48cd9574fd56a40e27","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 01:13:49.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"196aab4add5c44912733c7bb5b5f42670cf16450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 01:17:16.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"fc53fb5db7884c671f6771dba9d5483bb3ea9c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 02:34:28.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f33cf367892f4a31b058499d269f228f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bbfa1c29b8b4640a14f57e864cf29f4 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/4b84f8a36cdb4d2e86787c0c8aaa484c : SUCCESS in 1h 01m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3ee72784b1246478351f34cd1359bcb : SUCCESS in 57m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a6a0b7c17084032a3eec538c5fd0a16 : SUCCESS in 1h 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a53d20c3e8a2493b8d4a5f185693964d : SUCCESS in 1h 00m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d1f7511a50f4588b6283257e0ed505c : SUCCESS in 10m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/461c09351ff84ecc84ae284a552d2607 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d2a86ab81134ca1ac5f81198869a7a6 : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef344d3db8714487a1d014e4358d536a : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5e4a5aa98c7b43279fc5487a323ac699 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fa63d139445a4ebf8b22289e69217a7d : SUCCESS in 22m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e51ddd468fa9403f9e01d3293d627f64 : SUCCESS in 28m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c7b11acf4bd1476d8a2dd6897e02590a : SUCCESS in 19m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fadb4cb984b74688b0213fac2a27fbe4 : SUCCESS in 12m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d52f8c8d68fc4dedacc579d8827ce565 : SUCCESS in 8m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f3faef8ec9cc423d902aceb1d10a1b03 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce16ab005247481f9aead5f70d1d489a : SUCCESS in 5m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f5392da94e034b139894046b29c02b5c : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8381db0c6a9a47c39843c526835a149d : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9ea384659f54bf8a0f9a98148f9df53 : FAILURE in 42m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea12ff4481f7425e90c59e843c69b1d4 : SUCCESS in 48m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30cd5d86f79e4fb4a4087e6fa276efea : SUCCESS in 34m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f4c254b619b4e8e9610663f0f856d3f : SUCCESS in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293b25a242d741049b4c50bdb175e17e : SUCCESS in 2m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ebf8ee7cb0348629ec82b0ae8773b55 : SUCCESS in 24m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b332f7ae12348b08a9ca73a2ceb2ba0 : SUCCESS in 48m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82bcfbefc1ae4646b29c0dc2eefcc933 : FAILURE in 1h 04m 31s","accounts_in_message":[],"_revision_number":21},{"id":"d694b3068b0d557dd389b5c0dc89f029f902dfd7","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 02:40:55.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"f254312c4c010f9ceb61186b259df3d1fdc9cf8b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 02:41:02.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"f5f5c2395e598b2d5cfd4ec58dc809e1932cd063","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 03:26:35.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20456d5c608b4f41af6b40cdd28b09c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3cc665f0b27b494bae4a2f1ef677bb9b : SUCCESS in 2h 07m 24s","accounts_in_message":[],"_revision_number":21},{"id":"3c877de723dd76ad3e588cfed4f47d213babe995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 03:59:00.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe3c0c0e983430d88c711bedb12c4d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a808110d5f649ffb5d3d0141f46df58 : SUCCESS in 6m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/a677c112f6de4624be0dccf87c4ab19c : SUCCESS in 44m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc935188688b4e9eb1c12ff272e29bf5 : SUCCESS in 1h 01m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/159b3814d0544547be50c62bceb4dd45 : SUCCESS in 43m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a02ef03e823433ab4bf58810a880b20 : SUCCESS in 53m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccf37998cd434bee8650fa5002c3281f : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/745470f2d0b4408a86db4b5686c577ca : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29c4fc4276e64d3faddc003793d7d2a9 : SUCCESS in 23m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d54166227714f1daf78b8c1c1ff3cb4 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c99511eef1f4dcfa923aaab98a1c71e : SUCCESS in 28m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc2d6b2aaf544628b4d38f7760218a31 : SUCCESS in 19m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/98bb3f15d83e43df87239d9cc040c8e8 : SUCCESS in 26m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9ef8babc0981466fab67acfd05ffa128 : SUCCESS in 18m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4b967838fd944fe19b580c0f0d85cb04 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9859812ba3b040b7b4e9c132cd711b52 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/69e84361f0424f53a74582bfc2d6c99b : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f208904451534c76b2e83e4933b5cbd3 : SUCCESS in 7m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2bec9d6660f4189b804abaa9af605fb : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d6e63b9650dc45b2a2d4836de797a71e : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d7a6bcc1d344e97a93ad7ba16d1bbbc : SUCCESS in 58m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c0e0c3692cf4478b4a33fe60afebb04 : SUCCESS in 40m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a864d75600eb4d3593bbc6ab7ab23fce : SUCCESS in 55m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4c62f51286d42b39f2edc71e579382b : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2577caa4bd4aeaa12c12cfc196d1d3 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0436cace2de44576b6e4c526d2a7e7b4 : SUCCESS in 11m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c70ca2a71bb6431bba3525451c237fac : SUCCESS in 52m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dbc58e622c74c148dfce9d811adaa21 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":21},{"id":"14e47fa60709eafa0767c22348848bbca5561b22","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-21 09:19:42.000000000","message":"Patch Set 21:\n\n(7 comments)","accounts_in_message":[],"_revision_number":21},{"id":"ac52845b7fad8dcd40ab863809b5cdc5eb4bf4ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 23:12:32.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"f970959920e0cebf9f0936b5240b56e7c8659db8","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 23:12:53.000000000","message":"Patch Set 22:\n\n(8 comments)","accounts_in_message":[],"_revision_number":22},{"id":"0f142fdca67257d1337e8f82fbb059b13e13d546","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 23:13:11.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eef0cdc1d7754b9fadffd7852d98132e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d44591067b348dfa8570410d1f62378 : NODE_FAILURE Node(set) request c8564c89a6344815857543631acd89ad failed in 0s","accounts_in_message":[],"_revision_number":22},{"id":"c4d8b6710d030480fc6fc6a33a5db026bc0c9e53","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 23:32:59.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"579f719532357221a846b59681d4101b7f5a39b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 00:22:58.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d42ba14da4d440deb97aeb155bf8fdda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224ad1b59415423c95dd917d84eb21f9 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/3d938db34dbf456691263d99d129b508 : SUCCESS in 1h 05m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb08c3acaed84beaa24a2a0991a895f3 : SUCCESS in 38m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/07db8ed149c94b7dbcb94b75a9899ffa : SUCCESS in 53m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d07239f9e87c41fd833421a5c355c32a : SUCCESS in 32m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e268c5ae5d54dfab795d22e1a010fc3 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f906ccd153142329fe6e3c41feb4e0a : SUCCESS in 11m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db9c4b8ef76844d684bfcd73c5963911 : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c31fd42433024ec9bee0de66d1f7a19c : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1274f5cbbad64d428970865a0008f407 : SUCCESS in 27m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/28a0a4c4ebc0496e936fe22f8bc39b5a : SUCCESS in 21m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b24a1cbb57a74a7789766dc903e475b1 : SUCCESS in 10m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0fc608c634cc4d63b082a2f837009a2a : SUCCESS in 19m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8db0340405b34114a5e1f0ea7d162fbf : SUCCESS in 9m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fa65d695f4a449809487bfcd7e353134 : SUCCESS in 9m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/796c6b39f04645128e10b37b458ea4a5 : SUCCESS in 8m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c17608f192534f99bc512edf97073787 : SUCCESS in 4m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d3c4bda35c60422ab361f6ba5aff7f59 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/572d761f70ba43b1be6df3eacb22784c : SUCCESS in 7m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b9d9e74283d4d6789a7306ce9703c69 : SUCCESS in 46m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15e2cd43355840bf96c2787f612fa47e : SUCCESS in 21m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b3b3be9f3c4445abaf93c463fc767da : SUCCESS in 21m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/898b14f1aa714950a653bee82cf0280a : SUCCESS in 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/570502ecbbea461f9a3e277c925c1a85 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c747205fc464107a1d8888a1942045c : SUCCESS in 18m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04eecae0c47545888b13413347087154 : SUCCESS in 36m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c36a40d98e9a40fb95c20f1a2f4084f5 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":22},{"id":"0e76e70628bb1573567b9facd7123e7d39f982e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 01:02:24.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797a129c57804ae2a697fd2b6bf32f7d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6efe33b0f54c43c5be1d52e5d72c9b78 : SUCCESS in 1h 27m 26s","accounts_in_message":[],"_revision_number":22},{"id":"23fe7c793eaaff27ab9346ac8366684bc051aad2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-25 16:22:47.000000000","message":"Patch Set 22: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":22},{"id":"805f976a86ea467ad88c21f296b734692c19d1f9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-25 21:57:38.000000000","message":"Patch Set 22: Code-Review-1 Workflow-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":22},{"id":"8e50f4ef2b61b5e102e71be8347c7f6f4d5fc043","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-26 14:40:32.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"0c1e51d034d9c1401e9e1b67fbb07d9dfb925ee4","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-26 16:44:10.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"5e35f75848e087150d1927be02a397e941bab5e0","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-26 16:46:22.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"32a168ac736b2b8e97118d94c5b8a4ce704ca6c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-05 17:32:45.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":23},{"id":"89cea0e46f70bbec7fa85ee3f4c97f338e12cfc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-05 17:37:45.000000000","message":"Patch Set 24: Commit message was updated.","accounts_in_message":[],"_revision_number":24},{"id":"f5207fa686518e0dcff155462e2e4bea61d53ab4","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-05 17:43:05.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"80fb5ffb2bc8e01f7af386ee018c42044a286c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-05 17:43:55.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"eb445a7440011a4a74354334ff74f25437cafc80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-05 17:44:37.000000000","message":"Patch Set 26: Commit message was updated.","accounts_in_message":[],"_revision_number":26},{"id":"2662e5bdf68826fff4013a7e20d52c4e0e970ad5","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-05 17:46:36.000000000","message":"Patch Set 26:\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"3d486303fc5601e8f927f2a00591005aec00e052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 18:55:35.000000000","message":"Patch Set 26: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4b9e6e5c77c478094d8f58f345860ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2da4fb6be541738b9e893173bc85d7 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/fddb1fe60976468fa4a86bd25e717e9f : FAILURE in 26m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ea29207628241eab0f81d8464ca6404 : SUCCESS in 42m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ceb267dae2441bdbc79888a6c73da66 : SUCCESS in 32m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8901e6ff1104874b77d39adee88b854 : SUCCESS in 57m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65ff542a7bb64bc687e8b0462f03260c : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b8f3fb364e14cd59b3542bc4818ae13 : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c8d8c2941524ea8838c4cfc2f28cbc4 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1bc0266acd15488982f73291b383711b : FAILURE in 4m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b5215faad356487b9c0f65dc1624065c : SUCCESS in 24m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aedc8c59d97e434c9f851108e97f6ada : SUCCESS in 23m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/460de2d4d23741b8883d60f7e7a46093 : SUCCESS in 24m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7da3756b534c4a13a4fee0adda7ea083 : SUCCESS in 20m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1814029162e84da5b60f0bfaffc2d922 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/95e428f466e948be931f20a9d125d640 : SUCCESS in 12m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9bce44b7918a40eea5e74655186003b2 : SUCCESS in 15m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a83f6aa9367c4d8883561f0d2fb05291 : SUCCESS in 7m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9291ce17d6f049509e164eb3833ad3e0 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/235e11c2e5454b4abc68f15af74aeb52 : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c0158fe27934d3287147d718d86e1aa : SUCCESS in 1h 01m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c718c919bfe24947a20bf59433338104 : SUCCESS in 48m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c96318e2bba3463793e566f0a1c530cf : SUCCESS in 57m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f19d4bdfc5b944259fec099f13c4dc06 : SUCCESS in 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b14ceb87fb4812ae67b17d06bd98c4 : FAILURE in 3m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/580c1fe0f62f4bb7a8c5a513a0cc0a91 : SUCCESS in 19m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/792d27f6e64d42dba57ec17526699d6f : SUCCESS in 35m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4c77c949dac42a9b48f2ba5f5f8f21a : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":26},{"id":"32d0580d6ddbf5086465b90f39469dc5e66439a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 19:02:49.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22f008184e0f40a6bbdffa0d9b85a69a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d52b7cce1004dbca169c4db6a423ccb : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":26},{"id":"8bf269b0f4e4afbb3773a633ff05457b2fb61fd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-08 16:06:46.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"877af2975c5bf47bbe1c4ec83cfed42c9e3530ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-08 16:26:51.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":28},{"id":"b5687be583529d846ed4a1cdffe94ee39dd539f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 17:48:07.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e6d9b6dcb66459195f9eee2d002d84f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ddc0ba8f26409194c9e174d40f7549 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/8c1e0bbc9c4441a6992f85a7ba23d39e : SUCCESS in 1h 13m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/26be93c7d74d427786bdbeb2081160dd : SUCCESS in 1h 15m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09971b0448824456a7bee9bd5638dcb4 : SUCCESS in 59m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/683d4e1af2e2419fb1198007297db705 : SUCCESS in 1h 04m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95ab39b5c6c14e37a6bab9e75d6b7980 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b73e57f14c747778a59ed468d21c4a5 : SUCCESS in 12m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3083907e05e84bdba58daa7a7c17a339 : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4543925e51b4af4b45b076b9f49248f : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1b1f6ec85eae4cfebfec068a3dd50c2a : SUCCESS in 13m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59e733d65f084f33ad83ee934527ee96 : SUCCESS in 17m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3209a7666e0f4d8d996997d67666a7fd : SUCCESS in 32m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/325e1ab701b942b0a28917c726608a25 : SUCCESS in 23m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fc02d930128d46bc9cae6a4e5625e73c : SUCCESS in 9m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/387ea071fd204545a421c8d9173a20c6 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fbad3b0d10814efba685d9f63b513bac : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4f9adc42b8d6474792cfe851e4ad3ee7 : SUCCESS in 7m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/448026b2cf6d4e9bbae55b45358bacf3 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d5e48f6d668141338fe718c0dc1cba83 : SUCCESS in 4m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fa750536efb4dc5ac4bdc7a7da8ad6e : SUCCESS in 52m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f49c7a055a4d450fa0ce305b66e59e9d : SUCCESS in 55m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d30974f11eb141e4834a42634781b68d : SUCCESS in 45m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcae2ba7bd184fd0abf6cf937a7e484d : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88564d06123f47d6aa0a626d403820a8 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2d857ff1c8c346f6a86d140158d86524 : SUCCESS in 8m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f17f6e115c9d4db9b75dd4e1f8b5bac0 : SUCCESS in 48m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f35dfc0fe9c24a8a9256940911f0bbcc : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":28},{"id":"42dca18b205839568c056e7f1377c8329e4e9e26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 20:39:43.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5f8c626d62a4a7e8291794db389747e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56e5fed828db4083b6603c9bd9130a33 : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":28},{"id":"4da2f44a23b7f7108ee544913dbc9e3c058de63c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-09 16:23:06.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"7e9289f4f10bf14345467a3883479050d4dc05b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 17:30:55.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bf90e02847e4a63b58069894c4a1f44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8684a074ad4dbfa6edc3d7860e8248 : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/d91eb513daf44b40830498d1299741b5 : SUCCESS in 58m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79c697375b2d493792f0afa463f58dc3 : SUCCESS in 55m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2640676b19a14a17bc66e7f5b75108c8 : SUCCESS in 45m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b60f284bb5af4f5496b3e6ad8e7f3d1f : SUCCESS in 34m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75f354ff317e4e898114fad100f8858c : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/805cfb76b1234250bbdaeb2eb21eeac9 : SUCCESS in 24m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0e348a72af544fabf4914a4c3e467b3 : SUCCESS in 20m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee386597086e49dab3ea141a878a2b2d : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6e20b2ba193c4f1ab5872ba351b54da4 : SUCCESS in 22m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4b3b0c2b8d064735bee5d92234c432e2 : SUCCESS in 17m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9d97ac655a2c4a76ae74ab256c9084e9 : SUCCESS in 26m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/50710b1828ee4230b0e44d6fc916b740 : SUCCESS in 14m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ca4213c406b44995b95541816c0c3b53 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/297d8e4eaf05465f8789462631c76f01 : SUCCESS in 10m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9378ceaa3681431d8be21ce9e0fcda71 : SUCCESS in 16m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6837c3f3b2e4da59a2f2c2343f58ee3 : SUCCESS in 4m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/871c88bb918d4aaa9842bb48336b5c84 : SUCCESS in 16m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d759a901eb9e43cfb5cfc3353d4e4963 : SUCCESS in 5m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/60bbce5a60764452b4cd4d0ab7a05b0a : SUCCESS in 1h 00m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05b9ae1d3b2d4c4ca46cf6a10966ab02 : SUCCESS in 43m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f1ceb5d4c58489f94566e7039511d11 : SUCCESS in 43m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3ada1ef655845ea8fcc8329f7ab7790 : SUCCESS in 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75974e11077c49e6bdea34fea32e193c : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ed1f43d9a6e74b4fa12c02ca06c8ce22 : SUCCESS in 26m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c3fa19e4d8c425b81f43a72515e21e9 : SUCCESS in 54m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1ce0dc305684313af317027d4f8e1bf : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":29},{"id":"b73153e1a9a3d408d40ed9d4ff61c6ce04844dfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 19:08:42.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/900fcb23469641feb5d900811443fdcf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/353b08a76328489388ff50c2ca4476b4 : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":29},{"id":"ade9a9610f8c6e1de9c631d049fb825d8941e828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-23 15:51:09.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"dc0ab8befe808d41b2247f9f51952584e3e7e073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-23 16:03:44.000000000","message":"Patch Set 31: Commit message was updated.","accounts_in_message":[],"_revision_number":31},{"id":"c77a77e866079640eec303d07caf77c596f74b00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-23 16:08:18.000000000","message":"Patch Set 32: Patch Set 31 was rebased","accounts_in_message":[],"_revision_number":32},{"id":"d1b8e9c9d169f38e3cea19a3e4354cf3c7a246f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 17:16:24.000000000","message":"Patch Set 32: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f66a6db0736d426fa1ea8b4b2cb76406\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d7ac1e264c40c7a3adbbebc33c34de : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/12ef1393e3f643889df5d39645dffe24 : SUCCESS in 1h 02m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69f0ef50742846418a7a9667ac5c3382 : SUCCESS in 1h 02m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9fd140fea46348d192a49c0c8bb43ee4 : SUCCESS in 41m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d62e2b591df04c09887208944c909581 : SUCCESS in 38m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/851c667985224745a784fab77a246e92 : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb1a0e96fabb481aaa6fa7c6daa189eb : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30bb5061b0cb4c6ba1ddc5e761b6c69e : SUCCESS in 19m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3af9a200c104dd7a4c4c5cb049e375b : SUCCESS in 4m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/78ff2dd345e8457bbabba0a8094547f1 : SUCCESS in 25m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4be45a02872f477d9d5538a69bf0777c : SUCCESS in 21m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ba2bd49c1ced491da571d15434f93bb9 : SUCCESS in 29m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c71d7a90ccec44ef8b1d8b7e8ca7dc80 : SUCCESS in 11m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9745bef918374200a28aa8a6732fdc6d : SUCCESS in 11m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bb0243448bbd432b9c5897259713dd94 : SUCCESS in 9m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4cc52287081b474584626e9356261abb : SUCCESS in 16m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03594cfa1069459783e483e8c429ad04 : SUCCESS in 6m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d07378b12144656bbb84f14e97cd5cc : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1bc10b7a52a745739722cc9bab754af9 : SUCCESS in 6m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ff7df956cfa4d61abd36063ad5fcb5f : SUCCESS in 57m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cfb6a3564edb4793bbb4e9619a69d048 : SUCCESS in 51m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f289a9b9b903452287e5452a64f48253 : SUCCESS in 52m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/025a4f71ed2748419306ff665f74fb8f : SUCCESS in 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86e10f8ab25f47e895f18893f7828a3f : FAILURE in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7598e120f89c4be8959545a95ddba1e3 : SUCCESS in 9m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a70604c871b84e6195d6469116cea4cf : SUCCESS in 37m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6345b178e35546259629f6b656d3d660 : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":32},{"id":"23d902b76e5981d40d89bf7487dece6c31f4bf60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 17:55:10.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8fd022fa1a147c08b979e5f4ce483c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e5c391e363848e29d42a185b3c8c401 : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":32},{"id":"32be5f869f1ced4108a7d735b4d61d6859eec9d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-24 21:27:20.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"8f24b7a223c7bf537516a87cdda1c48a5562a513","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 22:27:10.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aeca0a3089c4054ab1626175d520102\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27eb898948a54cd4b15d537f5e80b90a : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":33},{"id":"0266d6363c905006644c3c96da515da682848712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 22:39:51.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9379068644a485ba627e9fc4b9db917\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed3759cd8dda432b869faf3b4e6baea5 : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/f4cd8c2b31584116aa706154f14eb698 : SUCCESS in 55m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0cd45df5b6947b4b9be4b83e5768d41 : SUCCESS in 1h 03m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e25bfa8289914fe1a5cd84e358bc2a27 : SUCCESS in 54m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9472f6b156144e26803299e4d2176579 : SUCCESS in 1h 05m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58c8e1b2e643468aa5d536f1adb1d57b : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3894c8f10849451c982427983742f37f : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09aa3024039843e0b5a30357e0221bc2 : SUCCESS in 21m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0020f1e0c5948deb402546ba1f3750f : FAILURE in 4m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f4368f2372d847838adc8456830c2c88 : SUCCESS in 19m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e59459264bb145ebbace56c5cecb9900 : SUCCESS in 10m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cf1f2df614274ecdb1cb513a578372d3 : SUCCESS in 20m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6057cd7de2b345568d1ec9cd29b034a6 : SUCCESS in 11m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d09137db387b41b7ab47812f74362a03 : SUCCESS in 9m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2481ab5d5c5c4e4ca10346d53df7c578 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2d501e70977249ebbb65860dbee0b63c : SUCCESS in 17m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/98e1c8d794d44ccc92d064a14fe493e0 : SUCCESS in 3m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/abda006fff454a9b9dde45f0a9b67a1e : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/37335b35c0ab49768160dfc0ddd841b7 : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/57ac283c944e4069bac6fee4e35f28c5 : SUCCESS in 58m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c73a5a7db204fc2b927b26d1a3c3de7 : SUCCESS in 58m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/176257582295482ea952a750551e7383 : SUCCESS in 55m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29e22dad892a489aaa43f173789d4c73 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16548e9f2aa7498f954e50effbf18319 : SUCCESS in 3m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/28c974e26e2f43c2be4a2c98eb5a41d3 : SUCCESS in 24m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7cefe02e502454c8f87ba97d88584fa : SUCCESS in 53m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d781aded620b476e988b8eb29d71fdee : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":33},{"id":"794b4e94dfca6fab2fd6d1e7de5034ca5133bdf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-25 17:21:11.000000000","message":"Patch Set 34: Patch Set 33 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"52ca9c375a1e4bab812dff27396c9be37b08b0b4","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-25 17:49:52.000000000","message":"Patch Set 34:\n\n(4 comments)","accounts_in_message":[],"_revision_number":34},{"id":"3ba9696d1c175855287bf0256f14b72e7ac90fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 18:23:03.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c2237a279b845e0ab90b82164b49bde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383d98d0bb964f13983482dcfbf72a96 : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/ce49d35fcf1242319e882c2b8bd01bdb : SUCCESS in 53m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f4e607890c9a454685b63791ec61aed4 : SUCCESS in 36m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c090988d94ef43cba4ffdf4fe7e21755 : SUCCESS in 52m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9f6f0ddea5c4171be92190f92df1949 : SUCCESS in 47m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a19a278f67c490e928c6efbe11bb1d1 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66a02c0f5f074af4b46df6b7c52056a3 : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/773297d94cc34fd68a69ff0394465a27 : SUCCESS in 18m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9073662bde3439990b78eb97e0c71bd : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/712ba9d9b3bf4a9498c2a025d2a0a2ab : SUCCESS in 19m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e3545c0feb2846fcb12397321d6687fc : SUCCESS in 9m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7809ea2a34dc4161bddf107a5949b22e : SUCCESS in 27m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a9b343b88730481daec5ced8f319ce71 : SUCCESS in 30m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/79eae631bbf541158436e0cdd1aebc48 : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b095eab28af74b5fb5e41f4f22a7e1f2 : SUCCESS in 13m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8d72210fefe84a4c8a8ea8f4e48335cc : SUCCESS in 14m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aef9e959c6c74430bfa1636a9ef62fbd : SUCCESS in 8m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/797faeac494844699c326c4d8b46e81b : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3cf875791f4048bab183a1c1f4201e51 : SUCCESS in 8m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/690a57141f104e5db7294d3f526f96d6 : SUCCESS in 56m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f7e669ff44f4a92a1b2e3ff28c650bf : SUCCESS in 50m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/005da561348f43a1b5f98a24bb676eac : SUCCESS in 57m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/927bc86560f7453291284255249a14ce : SUCCESS in 23m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c7341940f1d4367888a7bd80a83f168 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a295fc8f40143c1a1d60c3290b39b74 : SUCCESS in 29m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6ce62688f4449d1b8f6940e47bd09e8 : SUCCESS in 54m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1537a4d12e3640d0a388d24a7e5111a1 : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":34},{"id":"53a1636b774844100da36efe93c6752a78368feb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-25 18:46:30.000000000","message":"Patch Set 35: Patch Set 34 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"a32fbb1da659c50a50edb0753580fa994828027e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 19:54:48.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6ed89a003934b749ec20af4459748d4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fda962e50b6046e5889c19dd903f56fe : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":35},{"id":"5ace59dbc1af994d183d5649e06184ce510ebaf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 19:55:05.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c902f32ffa4142079ea5d7cea8bf7656\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14b4bf7672e410e99a7b97e313283bd : SUCCESS in 6m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/a4ae9f8488e64431aabdfe89163d5c82 : SUCCESS in 1h 03m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6362df0a0ec74f38879f0004a5661d8c : SUCCESS in 48m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/90df543554964424b90a0464839793f2 : SUCCESS in 48m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6661ccb2fd5c4b5da7577452478d7331 : SUCCESS in 1h 01m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb17f96513ba491ebd1f29c5ef700fc0 : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a94f555fd5f640d3bf22b0acedb23617 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8bff903a82542ef86ee324761e1d528 : SUCCESS in 20m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9573979321145c288475a7e12f496d2 : FAILURE in 4m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bbcc05270a1e4952bcd309478fcd0424 : SUCCESS in 20m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/107224a21a954b2fa4b60901de013b9a : SUCCESS in 10m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a22aca937bd748fbb324ae1fdc7a8e69 : SUCCESS in 29m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9fd2e4ae60ed4cadb0c638bfd189f70d : SUCCESS in 28m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/095107ab4cc3484794bf90f49c67faa3 : SUCCESS in 19m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0b50d68dfb8146a8a5ce21baa2634a00 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1dad4be6b9bc4b82bd95ece7102c31b7 : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/35830e5fdf374a788b88527bb3fdec2e : POST_FAILURE in 6m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/870feed870ae4e6380332691bf973dbb : SUCCESS in 15m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b3662ef561ac496b8191026a47d591c6 : SUCCESS in 9m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0fafc55b8b242ddaf2bb2e4cc92a3e7 : SUCCESS in 58m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55a22a5128f248c284dd37da93c646d3 : SUCCESS in 50m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a111ad1d52c74d75b8a9eae294b99a68 : SUCCESS in 53m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ed684b92e5a442182fa4293a32ae5eb : SUCCESS in 22m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/976ae7bd35f745938dd2db05f2801d60 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74adf1be595e443ba08a83408fb46615 : SUCCESS in 19m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1fe473eefc44c69998c61580864df4e : SUCCESS in 48m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f301d5ac513486e961276f385ab2d53 : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":35},{"id":"4627d5e2d929b3d11a7df8574ddace4541f3eba8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-26 18:02:47.000000000","message":"Patch Set 35: Code-Review-1\n\n(23 comments)","accounts_in_message":[],"_revision_number":35},{"id":"fe3265d76cabf2cfddf64e539ee4f740a6d62b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-28 22:20:35.000000000","message":"Patch Set 36: Patch Set 35 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"7148bd1a96b93a70678170eb0718241043c7d1fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-28 22:57:37.000000000","message":"Patch Set 37: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":37},{"id":"fb3d8bb9d0d017f00570b7280f46feea74dd5cb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 23:53:14.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a55a080c82d7425db7586facc37c7b33\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d765324e1d9a45ba86e93b55b23a6377 : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":37},{"id":"4c180c31dd1b035d0c43b3279f6edbea3cd9c080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 00:07:12.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f032397ab2a4c7c9ac1332c065cdfad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36d105f7b304ead869712407599c965 : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/ad48704af3d84e94a662c8a4c6492e74 : SUCCESS in 1h 01m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/595152c956f241ad877dcb2abf247e8f : SUCCESS in 1h 04m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4109baa6ff2e43fd919b89cae4936dc8 : SUCCESS in 1h 00m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9d26b11c349436090efd361784f0d3a : SUCCESS in 57m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7007ddfcc5d41d3b1a2b10c1597c7ae : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aab6a47d4e2d4d208c79ee4ed4f9c6ce : SUCCESS in 15m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96b341a008944486b8016aec42e96d94 : SUCCESS in 20m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/482328647e1244dd818ef8541aabb1c0 : SUCCESS in 4m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1348f5705d00436db92cc74b4e6e2d2a : SUCCESS in 24m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/293e382cc3ed451996e581e04b9419b9 : SUCCESS in 8m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ed132abce7f841c4a3c4a37f3c897ae6 : SUCCESS in 28m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b74d2c153501479fa4a3fd117e5b8f6d : SUCCESS in 30m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4118cffdcbc941208e98602d9bc1f37d : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d218ab2425ef46b3a116a09f471058bd : SUCCESS in 7m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6ade5ef063574a04aa501c681502e23a : SUCCESS in 12m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e73a6239c23949ee871bbddf88f5f19d : SUCCESS in 6m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fcf42fc94b124692acfa6e74585ad0a1 : SUCCESS in 15m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a70d9eaf5c6444aa9b8b7e6f1a2e2f3 : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/28b29c1b096c479ea5eabf3c20192e85 : SUCCESS in 55m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31fc2a56dd9f485cadedaede604d204f : SUCCESS in 33m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e4e2eee2ee540cdb3b5341e50a9df8b : SUCCESS in 52m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6741400976b440569294d88abb5a2c20 : SUCCESS in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d0f91c3f3394421907d1e6b31373fee : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23e2721694c34426999f8938fbb8b57f : SUCCESS in 14m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4237de1b759142e79d97a12fb7cff123 : SUCCESS in 36m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f5fd13b630245a2b077991715888d53 : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":37},{"id":"69538febbf5e32cff6f528440ff0fd543b9a85f1","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-29 01:21:41.000000000","message":"Patch Set 37:\n\n(10 comments)","accounts_in_message":[],"_revision_number":37},{"id":"35b06d7d2a6b84cf209684fff77c7ebb3abec62c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-29 16:01:56.000000000","message":"Patch Set 37:\n\n(3 comments)","accounts_in_message":[],"_revision_number":37},{"id":"5d33b59db4d781fa7db0060473433c9232d91421","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-29 23:35:22.000000000","message":"Patch Set 37:\n\n(16 comments)","accounts_in_message":[],"_revision_number":37},{"id":"33f050f715f9726aa0adf07eeb6e830746a05bfc","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-29 23:37:30.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"85b513ac355f632f21ca680621a68a255fc67ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-29 23:57:03.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"3e9d8ae6a4e46f3fd3f34704ecef8cd7afd00df1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-30 00:00:20.000000000","message":"Patch Set 39: Commit message was updated.","accounts_in_message":[],"_revision_number":39},{"id":"c6751d737f849e21af853f7d69ffca5695044fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 01:03:09.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c100748b2c4c62a6a89865e095d680\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1466b7ad58a24860bc2f71393a4d20bd : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/4eb0274cc2474718b8ce04c71c437642 : SUCCESS in 41m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4fd1e8b79485439087a3de1a4b3ba9e2 : SUCCESS in 46m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/31f974ff9bd94f08b9a6f30e57d0aa28 : SUCCESS in 36m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dbbbd5c6c3f40f0a82f46f7202dc316 : SUCCESS in 45m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fc9a77356614a7bb1c234b3f32beebb : SUCCESS in 15m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43c197a5f07443bb86872b5a3b4de18c : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d1fb0e8854245e899ac51c3e733c5de : SUCCESS in 19m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee336cb9a7f7403fba10225f3a861fd7 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2657f66459294113b48f936567ee3a8e : SUCCESS in 20m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9fa828b501d4100bbdb531d992ed330 : SUCCESS in 15m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/68e5055247a047daa8961c3796a9d07a : SUCCESS in 19m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/010aaf6c77ac4a96833aed6f908918e1 : SUCCESS in 29m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b27bc5e5da7e427da3b276f29dacd3c3 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b9e7f8309dcf4e2eb2193a3ac1dcbfe3 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/68c98d519d4c4cdba14db553ed455934 : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/42da85cbc1c444e5817ace45c31cb3da : SUCCESS in 6m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bcc1e6875e044f64b5ff7a723febe2c8 : SUCCESS in 14m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5bdc95440c2b460892366981f6275d8c : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f67bf4c3cbe94d3daec4cf4295aeb827 : SUCCESS in 52m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae6626413eff46299745fb76301eff50 : SUCCESS in 49m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd9129eb65c24b5fa615d43a77c42580 : SUCCESS in 53m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a7ce185cf240899ccc6f61e7b85c56 : SUCCESS in 20m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba1b8487400b4f7d9d64974f7565c998 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ca80a7e8140427283d4bcec51818d11 : SUCCESS in 15m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a94ad62cfcd42cebbc86bb68039a84c : SUCCESS in 50m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2fe6b77903443d296d89f3704b090bb : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":39},{"id":"c20c4f16c744c48633b8d8b396cede16c75fbc88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 01:15:53.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08094f385af646f09e242f40375d4c45\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/47af26703d4b44e78a53c4417558e14e : SUCCESS in 1h 13m 14s","accounts_in_message":[],"_revision_number":39},{"id":"72dc0946f74d70b1f7af9d0e9c13a5f809329ac5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-09-30 06:40:40.000000000","message":"Patch Set 39: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":39},{"id":"4f685e12b58fa2671cf20d39b2fe01cb1b2272ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-02 19:58:44.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"b0ebb2d7820a99e94c90b3c1143870109f08b64b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-02 19:59:13.000000000","message":"Patch Set 40:\n\n(8 comments)","accounts_in_message":[],"_revision_number":40},{"id":"fc7f1935b17c58c9653b5e1620b8087487cb81b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-02 20:00:26.000000000","message":"Patch Set 41: Commit message was updated.","accounts_in_message":[],"_revision_number":41},{"id":"1b8dfd167b01c5148c1ded37fb2d62590e7cd581","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-02 20:01:09.000000000","message":"Patch Set 41:\n\n(7 comments)","accounts_in_message":[],"_revision_number":41},{"id":"0c04a76732059843849a149c74eea370aa79a6c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-02 20:04:53.000000000","message":"Patch Set 42: Patch Set 41 was rebased","accounts_in_message":[],"_revision_number":42},{"id":"f4019ad42bb846faa299b2ba36fb0c82fa02e6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 21:09:01.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb21117f62854ff5956c45bd6e971114\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec499a3e967746a4b637546c61f067ee : SUCCESS in 4m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/307bdf1365944a17ba0c71e5e611a7c6 : SUCCESS in 35m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a5404136d6b4352afb84b08c0d8a727 : SUCCESS in 37m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/98f182800918460484fb71a6e6dcd531 : SUCCESS in 38m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7c10a90afb94645807ea9f1422ad9f6 : SUCCESS in 59m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc16f16b1c064033b9d844e3e75542b2 : SUCCESS in 10m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fd4da065dc642d0a8666fc5ff10e91e : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0185978913674b75ae77636e5bd2cbe2 : SUCCESS in 10m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cc95c1aba7414c00bd0423df52c643c3 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/15609a05b71942b2b6cf2254d8e3dd25 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/43fc05f4c23b4d4b9d2d45231cda2dfc : SUCCESS in 8m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/51304cc55a5f45b88494dd4de2b8c1ad : SUCCESS in 10m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/24f769f571cf4e70931efbaa876f9a0a : SUCCESS in 11m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9a455c81cf5345759a8ea9961a1c7974 : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/88ca5f9e06bc4a23bccd55cb60d1a0bc : SUCCESS in 9m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/36acc7fca8a4483c968542df3ffc43df : SUCCESS in 8m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3d7619d3cff44b2943a11201cb958c5 : SUCCESS in 3m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2076d6ce0df04ed9bcfe28ffbddf8d35 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a52c0a60b31446e3a8ab883b4cf396e3 : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcef5f39d15f49f68b270114d11d0e17 : SUCCESS in 38m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a05fd612608e48a4aaa581db339355dc : SUCCESS in 31m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db08f6104a1246a5b8a78c1d22522b32 : SUCCESS in 58m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/042717da719c498495e302b8ba63d931 : SUCCESS in 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7600f3083b7c455987d26e50d22a52f7 : SUCCESS in 3m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/42c91b0339a543a8a892b287a43cef22 : SUCCESS in 16m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82cd34cd1f154b2e87ec861819a52a3a : SUCCESS in 49m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e41c2c55b034e1c929d052e3097c990 : SUCCESS in 28m 51s","accounts_in_message":[],"_revision_number":42},{"id":"08eb3dc93fc7539a53914f51e058b45d7ebbe404","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 00:24:05.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c4830639ed04ecab8cd90d46d562e1b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aade36217d0542a0b9e28feac8f7af31 : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":42},{"id":"51befc999641047129f8a7691c4bafe1c4d24b3f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-10-03 18:21:35.000000000","message":"Patch Set 42: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":42},{"id":"c6fef9dc091ced427d65aab64fe3b6a37f2bf7ec","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-06 01:10:52.000000000","message":"Patch Set 42:\n\n(4 comments)","accounts_in_message":[],"_revision_number":42},{"id":"cdeb5c250296a37f7d94159629c877e9af42ef56","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-06 01:10:57.000000000","message":"Patch Set 42: Code-Review+2","accounts_in_message":[],"_revision_number":42},{"id":"41bf4fa2f3d6abf24e11e6a4576003a1ac2abf7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-08 15:42:51.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"9b0d69985dc5c3c7d690ef07053d39653318d961","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-08 15:43:05.000000000","message":"Patch Set 43: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":43},{"id":"1cfa9fbd342ad7be87f1b9ced86f9eb92f211ea8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-08 15:43:59.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":44},{"id":"c3bc803de8fad110bf4ca08052a067e81b6166c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 16:57:42.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c463fa71960e40c9b0963a1424cdfa0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03684d4eae04f0da7605ee54adcce6c : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/3dad18b8b81047b5a1b03d77ee18cb7a : SUCCESS in 36m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bbeb4fb976c54188b05af5b0e3339114 : SUCCESS in 1h 09m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e3792d95e654ffdb98f25a16ac0ca21 : SUCCESS in 55m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/868b1f3527dc4ad79c2c31122e474b9c : SUCCESS in 46m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5016a88bf144cf1a529f6804fc64b0f : SUCCESS in 10m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af5f8a88212346599bda8111be7e874a : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ea74b6f3985420bbabcb16f62f33494 : SUCCESS in 21m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a90e7993b3b949ebb5428c7b64d63f91 : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5d4f98d3ceea4b93ab10e31879061587 : SUCCESS in 22m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b97645a91cc4e85ade8bce8af1787ad : SUCCESS in 14m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fe70af2a40b34646b002f4f877afc49b : SUCCESS in 29m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/877026c4985d4762a857e486e7a79927 : SUCCESS in 19m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/87712c20836f4fa98e2f7fcdd186fbdd : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e0845a3db9aa4ebe999f292e8f168a1b : SUCCESS in 17m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/232c9f86e4ce49e2aa5a2ecbdb8ecfdf : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6cdb8140d825438099737e02b8c8ca8b : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d586e3974f34cb1ba5be1a753437f5b : SUCCESS in 12m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0ec4bedf3dbf4b2b869e6cf59967f262 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a100d5aaa054cb998e7296b27c3f9cc : SUCCESS in 55m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/81f33254a8724885aa4b9ea6e571a961 : SUCCESS in 50m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78b4e203be7047eeb744a98250d4d9a9 : SUCCESS in 56m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76c16ff197834e818f242f12e9b89eb9 : SUCCESS in 30m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5dea2c506c14e59981c51bc87a8718d : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/75f6b7e3a7cf4b14be5a1a92518bc3bd : SUCCESS in 25m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e5bfda5364d474980836c98f6b48490 : SUCCESS in 51m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca939896271f45de915dc2bc8347476b : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":44},{"id":"cc945d211a47641e8c62642dda83edfd24893d9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 17:07:57.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d7225075b345e3884e1b548ef64454\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb082ac44fe640c0b76611b329519ed2 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":44},{"id":"44d16c35e3628fb4c843b1586e43f74eb435db61","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-09 14:53:29.000000000","message":"Patch Set 44: Code-Review-1\n\n(17 comments)","accounts_in_message":[],"_revision_number":44},{"id":"314478c42cd67a38eedb54026a40b3ef66b2e2ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-11 22:14:04.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"187adedb6cfd0da586b6778c6dc008077a240e81","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-11 22:14:28.000000000","message":"Patch Set 45:\n\n(17 comments)","accounts_in_message":[],"_revision_number":45},{"id":"d7980e47c045b63ad6b933e7c5c4007b8dd3811b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 23:00:54.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d497fe32c48041d5a392b6d81336288c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a2c137b48d6a471aabd42846722f40f1 : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":45},{"id":"fd4f70e5db59e2382ecb261f014673c72e91d69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 23:19:15.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2848ed300a4f4df799c6f06defd56e1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04dda41eca348eb9700e13f6c61404f : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/98d46eda1ecd493397b3f00afd4a7586 : SUCCESS in 41m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/58c37d90003f41ff8a09883240fbabb6 : SUCCESS in 44m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c119b6f843c465792655957df89ff0a : SUCCESS in 55m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2e181654bcf407289079f3e88b57d58 : SUCCESS in 59m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dacdbf0654e942c497e579a3ad9b665d : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46968a22bddc4487b5327ce41d62e412 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b72dfbbde014da59f356ab0be5da624 : SUCCESS in 20m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4fd9e035cb7e4199892b9ef1df3eb146 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a3f5b97dbcf1492a8cec567c5e5c2cba : SUCCESS in 16m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/30d4db452710410ca11099f142cc1bc6 : SUCCESS in 14m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/674075b29b764b9fa8ff5947cf94f9d7 : SUCCESS in 22m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/965271b949fa4aa8bd9edb2e43c8cc17 : SUCCESS in 32m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/89f85cb83b334da6aee22048972f8310 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dd844b755a274cb5bde4926476e2b034 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/72d5c85023534b0a8a57d3a4593aec6e : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e4008f026524932a6b674ce08a127d4 : SUCCESS in 6m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c9196c075dae4e1594d5406222d337d6 : SUCCESS in 11m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/13d2e2598c1f4c5eb483c2e5dab4e43c : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4885e9cd6e9d4a508d1770a651dccdc1 : SUCCESS in 33m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76540c5004094033852a54b7e69a833f : SUCCESS in 48m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/24f132ff680144fb84fa8948401f86e4 : SUCCESS in 59m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0841f2dae6b4422b849ee7cfd4ec871a : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533dac6af5014e2daef525f3d0d303ae : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac8a9aaa02e249c6a03f7d65f1705209 : SUCCESS in 17m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b323ce34b71476b990fdc2aff96e944 : SUCCESS in 39m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c01c2e6d842d4c58bff837e833870045 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":45},{"id":"4632feca85c2ad3e542881be2ef9c24a2f2a2424","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-13 12:23:50.000000000","message":"Patch Set 45: Code-Review+2 Workflow+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":45},{"id":"a5f754523075643bddb71c4e7b8f70ebe7fcd3a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 12:24:17.000000000","message":"Patch Set 45: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":45},{"id":"7f34dcadccd122d5244bdc5d5507661237d4e51a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-13 12:28:02.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"3b124d17caf539201ded9208532f8712dbd9bd15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 13:31:32.000000000","message":"Patch Set 45: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c93b35232a6447ba005f896325bfdb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59c346dce55e444b9951903313101288 : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/3ba6031540bf4369842f31461fcded02 : SUCCESS in 1h 00m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/862942993aa34edcbbae9704a261751c : SUCCESS in 1h 02m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c8a7e4d5db25440aa255a10a19d8c9b1 : SUCCESS in 54m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e704ba61a2fa4113af854b94bfa2fcbd : SUCCESS in 45m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d02eda99ffd548f4a0312f6be49b5daa : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/160a76cf3c714f7894d8c5e7109eca34 : SUCCESS in 20m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb19d002a18c42eb91c096e20a55526a : SUCCESS in 21m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cbfe481bd8044f4e9e9af79769a33985 : SUCCESS in 10m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a726d10a3d9441e78feb081b90f12f52 : SUCCESS in 23m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b59576bc6ed14218a30e1d0440b60f18 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1c4988b3adc4477bbb8ae35f79551269 : SUCCESS in 8m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b54935d159e64bd3b5e8e8566fb7598f : SUCCESS in 13m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a569df4c9f445bca240fdcadff8b93a : SUCCESS in 6m 39s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/47692a4246df4037936be66f3a37925f : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/45bc76adb303494ba470bc57fcfd40ec : SUCCESS in 57m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb087c13db8b44709dc4d5681bf67b97 : SUCCESS in 40m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/474c915a9b0346539aff801d95dc4e2d : SUCCESS in 35m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e3de865c7ed49f380f2ce9f96c64b06 : SUCCESS in 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac338597ea8645c9b411a24aa0c1bf97 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e1af6c2c1ee4c5f9c851bec9233fd6e : SUCCESS in 17m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46b14c228a874b1a96f0d7ed07bcedab : SUCCESS in 53m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/970da92b87e94a9a863b36c5957236d6 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":45},{"id":"268b4c099df001e9d1c3d86a123a864abb2b2ed8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 13:31:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":45},{"id":"882ea278bf1f3ff547c21101583e0887bec3e75c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 13:32:29.000000000","message":"Patch Set 45:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5b3203088b745a8bd0223990aaed6b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96f5029a9124b55816997a6af196b08 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/803f79ab5c3a4a8792f76027bb3eef93 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"2709dd5aadbc71db1d03dd166e6b2a886ad17690","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-13 19:42:04.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45}],"current_revision_number":45,"current_revision":"41bf72a5cc5dbda1192732930c78a1fd91282f98","revisions":{"a5c06217fdd2b0d507b87b4a0935ee6067f50bdd":{"kind":"REWORK","_number":1,"created":"2024-10-24 22:53:14.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/1"}}},"commit":{"parents":[{"commit":"7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f","subject":"functest: do not use account2 unless needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a6dc095b46183f8c3ca64c4ed21c1cb8224e37f"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-24 22:51:51.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5c06217fdd2b0d507b87b4a0935ee6067f50bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5c06217fdd2b0d507b87b4a0935ee6067f50bdd"}]},"branch":"refs/heads/master"},"509383e60c16c87204d4b2de02cfdb36fad32858":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-21 22:40:26.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/2"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-21 22:40:26.000000000","tz":0},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/509383e60c16c87204d4b2de02cfdb36fad32858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/509383e60c16c87204d4b2de02cfdb36fad32858"}]},"branch":"refs/heads/master","description":"Rebase"},"e93c0bcc364991ab4758a5cf162f82953f6baf7a":{"kind":"REWORK","_number":3,"created":"2024-12-03 19:14:38.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/3"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-03 19:14:30.000000000","tz":-480},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e93c0bcc364991ab4758a5cf162f82953f6baf7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e93c0bcc364991ab4758a5cf162f82953f6baf7a"}]},"branch":"refs/heads/master"},"17c8af933b5fed6ffba57370eca74ce5216dfec8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-12-17 20:00:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/4"}}},"commit":{"parents":[{"commit":"4faa3523c9c0522c4f28a3028398949be4934483","subject":"Merge \"CI: Configure bandit better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4faa3523c9c0522c4f28a3028398949be4934483"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-12-17 20:00:51.000000000","tz":0},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17c8af933b5fed6ffba57370eca74ce5216dfec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17c8af933b5fed6ffba57370eca74ce5216dfec8"}]},"branch":"refs/heads/master","description":"Rebase"},"19ef39cef01dc61ee5d22f5d7d3deba82666f326":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-01-13 17:13:13.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/5"}}},"commit":{"parents":[{"commit":"94d3a5dee8122e14be0b00694d65c448ce47ea31","subject":"Merge \"obj: Add option to tune down etag validation in object-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94d3a5dee8122e14be0b00694d65c448ce47ea31"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-01-13 17:13:13.000000000","tz":0},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19ef39cef01dc61ee5d22f5d7d3deba82666f326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19ef39cef01dc61ee5d22f5d7d3deba82666f326"}]},"branch":"refs/heads/master","description":"Rebase"},"ceb9dec4cf5dec6c74101a5c2ba40e8d28de3a83":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-01-15 20:34:39.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/6"}}},"commit":{"parents":[{"commit":"9b4d008390e3dd5878730eb07456db97a55c70c5","subject":"Merge \"Leave updater per-device stats in recon for debugging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b4d008390e3dd5878730eb07456db97a55c70c5"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-01-15 20:34:39.000000000","tz":0},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceb9dec4cf5dec6c74101a5c2ba40e8d28de3a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceb9dec4cf5dec6c74101a5c2ba40e8d28de3a83"}]},"branch":"refs/heads/master","description":"Rebase"},"45e53f83e68cb30217e791283a6a35a5412cdddd":{"kind":"REWORK","_number":7,"created":"2025-01-15 21:24:49.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/7"}}},"commit":{"parents":[{"commit":"9b4d008390e3dd5878730eb07456db97a55c70c5","subject":"Merge \"Leave updater per-device stats in recon for debugging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b4d008390e3dd5878730eb07456db97a55c70c5"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-01-15 21:24:42.000000000","tz":-480},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45e53f83e68cb30217e791283a6a35a5412cdddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45e53f83e68cb30217e791283a6a35a5412cdddd"}]},"branch":"refs/heads/master"},"06c148544bde2c6f77a6c88003ce6526760b7d4f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-02-01 22:06:02.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/8"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-02-01 22:06:02.000000000","tz":0},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06c148544bde2c6f77a6c88003ce6526760b7d4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06c148544bde2c6f77a6c88003ce6526760b7d4f"}]},"branch":"refs/heads/master","description":"Rebase"},"1a2186ac105c44a8d31f06f1ed96b976912c3b72":{"kind":"REWORK","_number":9,"created":"2025-08-07 21:12:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/9"}}},"commit":{"parents":[{"commit":"c8d00108a61852143569f5f96c601675374ff095","subject":"Merge \"Fix another way mw may encouter invalid swift paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d00108a61852143569f5f96c601675374ff095"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-07 21:12:34.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a2186ac105c44a8d31f06f1ed96b976912c3b72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a2186ac105c44a8d31f06f1ed96b976912c3b72"}]},"branch":"refs/heads/master"},"f33c4bb32009261c2e3205a4d8ba495e19da13a9":{"kind":"REWORK","_number":10,"created":"2025-08-07 22:38:50.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/10"}}},"commit":{"parents":[{"commit":"c8d00108a61852143569f5f96c601675374ff095","subject":"Merge \"Fix another way mw may encouter invalid swift paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d00108a61852143569f5f96c601675374ff095"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-07 22:38:41.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f33c4bb32009261c2e3205a4d8ba495e19da13a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f33c4bb32009261c2e3205a4d8ba495e19da13a9"}]},"branch":"refs/heads/master"},"d523b58c5a74559fceae1579449b3edb4b581e22":{"kind":"REWORK","_number":11,"created":"2025-08-08 21:02:48.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/11"}}},"commit":{"parents":[{"commit":"c8d00108a61852143569f5f96c601675374ff095","subject":"Merge \"Fix another way mw may encouter invalid swift paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d00108a61852143569f5f96c601675374ff095"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-08 21:02:31.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs include the DB path\nand the container path. But there\u0027s places where the sharder\ninherits replicator code and the logs have not been treated in\nthis way. It would be great to do the same consistently in\nreplicator logging\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d523b58c5a74559fceae1579449b3edb4b581e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d523b58c5a74559fceae1579449b3edb4b581e22"}]},"branch":"refs/heads/master"},"c5b89dd4f6090a829d99b27a8f7d99fe06250845":{"kind":"REWORK","_number":12,"created":"2025-08-10 18:46:46.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/12"}}},"commit":{"parents":[{"commit":"c8d00108a61852143569f5f96c601675374ff095","subject":"Merge \"Fix another way mw may encouter invalid swift paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d00108a61852143569f5f96c601675374ff095"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-10 18:46:38.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nWe previously ensured that all sharder logs included the DB file\npath and the container path. We moved logging changes from container\nsharder to the parent db-replicator logging. It would be great to do\nthis consistently across db-replicator logging\n\n- this change addresses moving container sharding logging to\nits parent db-replicator logging and modifies the dependent\nunit-tests that have assertions. It also improves db-replicator\nlogging by including DB file path and container paths\n\n- this change doesn\u0027t address implementing the aformentioned\nlogging modifications to container replicator\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5b89dd4f6090a829d99b27a8f7d99fe06250845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5b89dd4f6090a829d99b27a8f7d99fe06250845"}]},"branch":"refs/heads/master"},"1643bf0ec86fa9e36f92bb75b36957e5e5e82a97":{"kind":"REWORK","_number":13,"created":"2025-08-13 20:31:50.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/13"}}},"commit":{"parents":[{"commit":"c8d00108a61852143569f5f96c601675374ff095","subject":"Merge \"Fix another way mw may encouter invalid swift paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d00108a61852143569f5f96c601675374ff095"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-13 20:31:20.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments.\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nWhy not (deferred work)\n-----------------------\nThis change intentionally limits scope to the common.db_replicator\nmodule to “bank the win.” A follow-up will audit sharding-related\nmessages in the container.replicator module (e.g _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1643bf0ec86fa9e36f92bb75b36957e5e5e82a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1643bf0ec86fa9e36f92bb75b36957e5e5e82a97"}]},"branch":"refs/heads/master"},"ae31d87a9f8f5ce1c7fcb95f71b9ccafaf982a00":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2025-08-13 21:22:42.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/14"}}},"commit":{"parents":[{"commit":"c8d00108a61852143569f5f96c601675374ff095","subject":"Merge \"Fix another way mw may encouter invalid swift paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d00108a61852143569f5f96c601675374ff095"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-08-13 21:22:42.000000000","tz":0},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments.\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nWhy not (deferred work)\n-----------------------\nThis change intentionally limits scope to the common.db_replicator\nmodule to “bank the win.” A follow-up will audit sharding-related\nmessages in the container.replicator module (e.g _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae31d87a9f8f5ce1c7fcb95f71b9ccafaf982a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae31d87a9f8f5ce1c7fcb95f71b9ccafaf982a00"}]},"branch":"refs/heads/master","description":"Edit commit message"},"048fd2fb8b7475e143246f6cfe0d51c78994382f":{"kind":"REWORK","_number":15,"created":"2025-08-14 01:14:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/15"}}},"commit":{"parents":[{"commit":"c8d00108a61852143569f5f96c601675374ff095","subject":"Merge \"Fix another way mw may encouter invalid swift paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d00108a61852143569f5f96c601675374ff095"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-14 01:14:15.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments.\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nWhy not (deferred work)\n-----------------------\nThis change intentionally limits scope to the common.db_replicator\nmodule to “bank the win.” A follow-up will audit sharding-related\nmessages in the container.replicator module (e.g _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/048fd2fb8b7475e143246f6cfe0d51c78994382f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/048fd2fb8b7475e143246f6cfe0d51c78994382f"}]},"branch":"refs/heads/master"},"d6bc078d388efc41ef53bbd078b238d18283a023":{"kind":"REWORK","_number":16,"created":"2025-08-14 18:35:47.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/16"}}},"commit":{"parents":[{"commit":"c8d00108a61852143569f5f96c601675374ff095","subject":"Merge \"Fix another way mw may encouter invalid swift paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d00108a61852143569f5f96c601675374ff095"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-14 18:35:37.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments.\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nWhy not (deferred work)\n-----------------------\nThis change intentionally limits scope to the common.db_replicator\nmodule to “bank the win.” A follow-up will audit sharding-related\nmessages in the container.replicator module (e.g _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6bc078d388efc41ef53bbd078b238d18283a023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6bc078d388efc41ef53bbd078b238d18283a023"}]},"branch":"refs/heads/master"},"b44a073326b0f2ae27b3c759d5a79e1a97a30119":{"kind":"REWORK","_number":17,"created":"2025-08-19 19:29:12.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/17"}}},"commit":{"parents":[{"commit":"c8d00108a61852143569f5f96c601675374ff095","subject":"Merge \"Fix another way mw may encouter invalid swift paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d00108a61852143569f5f96c601675374ff095"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-19 19:29:01.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments.\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nWhy not (deferred work)\n-----------------------\nThis change intentionally limits scope to the common.db_replicator\nmodule to “bank the win.” A follow-up will audit sharding-related\nmessages in the container.replicator module (e.g _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b44a073326b0f2ae27b3c759d5a79e1a97a30119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b44a073326b0f2ae27b3c759d5a79e1a97a30119"}]},"branch":"refs/heads/master"},"f009241071350438ebc9f7b854771d22024f0b48":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2025-08-19 19:44:19.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/18"}}},"commit":{"parents":[{"commit":"28a56e3f1b3527dabe9658c77d78eee400c4e9c2","subject":"Merge \"zuul: run py3.13 unit tests in the gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28a56e3f1b3527dabe9658c77d78eee400c4e9c2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-08-19 19:44:19.000000000","tz":0},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments.\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nWhy not (deferred work)\n-----------------------\nThis change intentionally limits scope to the common.db_replicator\nmodule to “bank the win.” A follow-up will audit sharding-related\nmessages in the container.replicator module (e.g _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f009241071350438ebc9f7b854771d22024f0b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f009241071350438ebc9f7b854771d22024f0b48"}]},"branch":"refs/heads/master","description":"Rebase"},"a7a88496ff98178da156a981fc89762ea5aa18da":{"kind":"REWORK","_number":19,"created":"2025-08-21 01:08:40.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/19"}}},"commit":{"parents":[{"commit":"28a56e3f1b3527dabe9658c77d78eee400c4e9c2","subject":"Merge \"zuul: run py3.13 unit tests in the gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28a56e3f1b3527dabe9658c77d78eee400c4e9c2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-21 01:08:28.000000000","tz":-420},"subject":"db-replicator: include container name in logs, not just db path","message":"db-replicator: include container name in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments.\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nWhy not (deferred work)\n-----------------------\nThis change intentionally limits scope to the common.db_replicator\nmodule to “bank the win.” A follow-up will audit sharding-related\nmessages in the container.replicator module (e.g _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7a88496ff98178da156a981fc89762ea5aa18da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7a88496ff98178da156a981fc89762ea5aa18da"}]},"branch":"refs/heads/master"},"7a2d0bde12a9cf5d32e400f1ef52a5b1fadd6c0a":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2025-08-21 01:13:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/20"}}},"commit":{"parents":[{"commit":"28a56e3f1b3527dabe9658c77d78eee400c4e9c2","subject":"Merge \"zuul: run py3.13 unit tests in the gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28a56e3f1b3527dabe9658c77d78eee400c4e9c2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-08-21 01:13:32.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a2d0bde12a9cf5d32e400f1ef52a5b1fadd6c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a2d0bde12a9cf5d32e400f1ef52a5b1fadd6c0a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"941e41427c390d407c85fe5f033dcc42aa367527":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2025-08-21 01:17:16.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/21"}}},"commit":{"parents":[{"commit":"dfab57be74cfe20523dccda98dceb2a8fc507991","subject":"Merge \"Revert \"Remove unused MANIFEST.in\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfab57be74cfe20523dccda98dceb2a8fc507991"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-08-21 01:17:16.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/941e41427c390d407c85fe5f033dcc42aa367527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/941e41427c390d407c85fe5f033dcc42aa367527"}]},"branch":"refs/heads/master","description":"Rebase"},"27db680ae44290d8715f127777def65a76012fd2":{"kind":"REWORK","_number":22,"created":"2025-08-21 23:12:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/22"}}},"commit":{"parents":[{"commit":"81df05a9c4702204a8347ab2ba43e7ff75c328aa","subject":"checksums: Ignore broken pyeclib installs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81df05a9c4702204a8347ab2ba43e7ff75c328aa"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-21 23:12:18.000000000","tz":-420},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27db680ae44290d8715f127777def65a76012fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27db680ae44290d8715f127777def65a76012fd2"}]},"branch":"refs/heads/master"},"e93961be8777904f34fe010a41a800a5bc1679a3":{"kind":"REWORK","_number":23,"created":"2025-09-05 17:32:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/23"}}},"commit":{"parents":[{"commit":"81df05a9c4702204a8347ab2ba43e7ff75c328aa","subject":"checksums: Ignore broken pyeclib installs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81df05a9c4702204a8347ab2ba43e7ff75c328aa"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-05 17:32:37.000000000","tz":-420},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e93961be8777904f34fe010a41a800a5bc1679a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e93961be8777904f34fe010a41a800a5bc1679a3"}]},"branch":"refs/heads/master"},"c1b6977493977848beb6d2956d7b1585a468cfa1":{"kind":"NO_CODE_CHANGE","_number":24,"created":"2025-09-05 17:37:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/24"}}},"commit":{"parents":[{"commit":"81df05a9c4702204a8347ab2ba43e7ff75c328aa","subject":"checksums: Ignore broken pyeclib installs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81df05a9c4702204a8347ab2ba43e7ff75c328aa"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-05 17:37:45.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log\nformatting so messages include both the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\n- No functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1b6977493977848beb6d2956d7b1585a468cfa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1b6977493977848beb6d2956d7b1585a468cfa1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f264727ca79916667399ed218a9c8d6ee60eec33":{"kind":"REWORK","_number":25,"created":"2025-09-05 17:43:55.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/25"}}},"commit":{"parents":[{"commit":"81df05a9c4702204a8347ab2ba43e7ff75c328aa","subject":"checksums: Ignore broken pyeclib installs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81df05a9c4702204a8347ab2ba43e7ff75c328aa"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-05 17:43:21.000000000","tz":-420},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f264727ca79916667399ed218a9c8d6ee60eec33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f264727ca79916667399ed218a9c8d6ee60eec33"}]},"branch":"refs/heads/master"},"f1630fded3238585a96276ee7fa957cb9c238369":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2025-09-05 17:44:37.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/26"}}},"commit":{"parents":[{"commit":"81df05a9c4702204a8347ab2ba43e7ff75c328aa","subject":"checksums: Ignore broken pyeclib installs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81df05a9c4702204a8347ab2ba43e7ff75c328aa"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-05 17:44:37.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log\nformatting so messages include both the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\n- No functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1630fded3238585a96276ee7fa957cb9c238369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1630fded3238585a96276ee7fa957cb9c238369"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3444e1ed3f1f46e228ab4be2c3d92aba2e2b3003":{"kind":"REWORK","_number":27,"created":"2025-09-08 16:06:46.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/27"}}},"commit":{"parents":[{"commit":"81df05a9c4702204a8347ab2ba43e7ff75c328aa","subject":"checksums: Ignore broken pyeclib installs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81df05a9c4702204a8347ab2ba43e7ff75c328aa"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-08 16:06:29.000000000","tz":-420},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log\nformatting so messages include both the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\n- No functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3444e1ed3f1f46e228ab4be2c3d92aba2e2b3003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3444e1ed3f1f46e228ab4be2c3d92aba2e2b3003"}]},"branch":"refs/heads/master"},"bcab3fbdff55c9b25a887a2cf2b3582fff0c6bcf":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2025-09-08 16:26:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/28"}}},"commit":{"parents":[{"commit":"82cb5a5d78b91c6af258cd8f06d30f69be90fa18","subject":"AUTHORS/CHANGELOG for 2.36.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82cb5a5d78b91c6af258cd8f06d30f69be90fa18"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-08 16:26:51.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log\nformatting so messages include both the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\n- No functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcab3fbdff55c9b25a887a2cf2b3582fff0c6bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcab3fbdff55c9b25a887a2cf2b3582fff0c6bcf"}]},"branch":"refs/heads/master","description":"Rebase"},"4be903c5763f71af495fd53e858e6b7b1c99557b":{"kind":"REWORK","_number":29,"created":"2025-09-09 16:23:06.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/29"}}},"commit":{"parents":[{"commit":"64327336f657728a80ca3339f22690b01f837dfa","subject":"account-broker: add path property, introduce _populate_instance_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64327336f657728a80ca3339f22690b01f837dfa"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-09 16:22:50.000000000","tz":-420},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4be903c5763f71af495fd53e858e6b7b1c99557b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4be903c5763f71af495fd53e858e6b7b1c99557b"}]},"branch":"refs/heads/master"},"cc4358f47bce7a9d5d772e438b3ae7977e442c2c":{"kind":"REWORK","_number":30,"created":"2025-09-23 15:51:09.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/30"}}},"commit":{"parents":[{"commit":"64327336f657728a80ca3339f22690b01f837dfa","subject":"account-broker: add path property, introduce _populate_instance_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64327336f657728a80ca3339f22690b01f837dfa"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-23 15:50:10.000000000","tz":-420},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator log formatting so messages include\nboth the container path (account/container) and the DB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc4358f47bce7a9d5d772e438b3ae7977e442c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc4358f47bce7a9d5d772e438b3ae7977e442c2c"}]},"branch":"refs/heads/master"},"303e17f18236551f7832deb64b66e17478e3ced5":{"kind":"NO_CODE_CHANGE","_number":31,"created":"2025-09-23 16:03:44.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/31"}}},"commit":{"parents":[{"commit":"64327336f657728a80ca3339f22690b01f837dfa","subject":"account-broker: add path property, introduce _populate_instance_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64327336f657728a80ca3339f22690b01f837dfa"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-23 16:03:44.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log formatting so messages include both the container path (account/container) and the\nDB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/303e17f18236551f7832deb64b66e17478e3ced5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/303e17f18236551f7832deb64b66e17478e3ced5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"dd5c6c087d0a43d62666e52f0c43f61222454e80":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2025-09-23 16:08:18.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/32"}}},"commit":{"parents":[{"commit":"f38ab2d466a5d252d06739606c421173e0c37962","subject":"account-broker: add path property, introduce _populate_instance_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f38ab2d466a5d252d06739606c421173e0c37962"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-23 16:08:18.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log formatting so messages include both the container path (account/container) and the\nDB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd5c6c087d0a43d62666e52f0c43f61222454e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd5c6c087d0a43d62666e52f0c43f61222454e80"}]},"branch":"refs/heads/master","description":"Rebase"},"55b4e257e9279cd86c1275f691f370287d585619":{"kind":"REWORK","_number":33,"created":"2025-09-24 21:27:20.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/33"}}},"commit":{"parents":[{"commit":"f38ab2d466a5d252d06739606c421173e0c37962","subject":"account-broker: add path property, introduce _populate_instance_cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f38ab2d466a5d252d06739606c421173e0c37962"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-24 21:25:38.000000000","tz":-420},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log formatting so messages include both the container path (account/container) and the\nDB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55b4e257e9279cd86c1275f691f370287d585619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55b4e257e9279cd86c1275f691f370287d585619"}]},"branch":"refs/heads/master"},"3693a9237fe8c1d8c08f143924f306dd2dd8fc4e":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2025-09-25 17:21:11.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/34"}}},"commit":{"parents":[{"commit":"cf1a13ebe593dd0c9922411d570b71a5b2008311","subject":"brokers: add resilient path property with lazy cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf1a13ebe593dd0c9922411d570b71a5b2008311"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-25 17:21:11.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log formatting so messages include both the container path (account/container) and the\nDB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3693a9237fe8c1d8c08f143924f306dd2dd8fc4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3693a9237fe8c1d8c08f143924f306dd2dd8fc4e"}]},"branch":"refs/heads/master","description":"Rebase"},"bf78c5b58deb65b33cb831ff73e1c56beeb6b013":{"kind":"NO_CHANGE","_number":35,"created":"2025-09-25 18:46:30.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/35"}}},"commit":{"parents":[{"commit":"07d230a8e8dbf760d35f023806cdda54eb0a399d","subject":"account-broker: add resilient path property with lazy cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07d230a8e8dbf760d35f023806cdda54eb0a399d"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-25 18:46:30.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log formatting so messages include both the container path (account/container) and the\nDB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf78c5b58deb65b33cb831ff73e1c56beeb6b013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf78c5b58deb65b33cb831ff73e1c56beeb6b013"}]},"branch":"refs/heads/master","description":"Rebase"},"e07ce376c74bc5bd85d50ad1359bbb72f9e82d19":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2025-09-28 22:20:35.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/36"}}},"commit":{"parents":[{"commit":"388656ab1b429eb7ec317b3f14fef58dcd4a81bd","subject":"account-broker: add resilient path property with lazy cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/388656ab1b429eb7ec317b3f14fef58dcd4a81bd"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-28 22:20:35.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n- Update common.db_replicator, container.replicator log formatting so messages include both the container path (account/container) and the\nDB file path.\n\n- In _replicate_object(), predeclare broker \u003d None and handle\nfailures to instantiate the broker without raising NameError.\nWhen the broker is absent, log using object_file exactly once\n(avoid duplicating the DB path). Keep the try-scope unchanged,\nper review.\n\n- Update unit tests to match the new log shapes and FakeBroker\nbehavior. tighten assertions for \"db:\" and \"path:\" segments;\nquarantine tests pass the same db_path into _replicate_object()\nand assert the quarantined dirname accordingly.\n\nFurther work\n------------\nFurther work is needed to audit sharding-related messages in the\ncontainer.replicator module (e.g., _choose_replication_mode\nwarnings) and bring them in line with the container.sharder\nmodule’s style.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e07ce376c74bc5bd85d50ad1359bbb72f9e82d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e07ce376c74bc5bd85d50ad1359bbb72f9e82d19"}]},"branch":"refs/heads/master","description":"Rebase"},"7e83f0d29f3eac36b5704af0bc6a758c3986062c":{"kind":"NO_CODE_CHANGE","_number":37,"created":"2025-09-28 22:57:37.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/37"}}},"commit":{"parents":[{"commit":"388656ab1b429eb7ec317b3f14fef58dcd4a81bd","subject":"account-broker: add resilient path property with lazy cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/388656ab1b429eb7ec317b3f14fef58dcd4a81bd"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-28 22:57:37.000000000","tz":0},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n\nExpand db logging consistency from container.sharder into continer.replicator and common.db_replicator\n\n* extract logging consistency helper from container.sharder and move it into common.db_replicator\n* update all log message call sites in common.db_replicator\n* update all log message call sites in container.replicator\n* update test.common.test_db_replicator fakes to better match real brokers db_file and path properties\n* update assertions for existing log messages\n* add assertions/tests for all updated log messages\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e83f0d29f3eac36b5704af0bc6a758c3986062c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e83f0d29f3eac36b5704af0bc6a758c3986062c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"61decd5d42929e6be4894b35e5e24b286ba6d709":{"kind":"REWORK","_number":38,"created":"2025-09-29 23:57:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/38"}}},"commit":{"parents":[{"commit":"388656ab1b429eb7ec317b3f14fef58dcd4a81bd","subject":"account-broker: add resilient path property with lazy cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/388656ab1b429eb7ec317b3f14fef58dcd4a81bd"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-29 23:55:56.000000000","tz":-420},"subject":"common.db_replicator: include container path in logs, not just db path","message":"common.db_replicator: include container path in logs, not just db path\n\nContext\n-------\nA prior change made log messages within the container.sharder\nmodule consistent, but did not touch the common.db_replicator\nor container.replicator modules.\n\nScope of this change\n--------------------\n\nExpand db logging consistency from container.sharder into continer.replicator and common.db_replicator\n\n* extract logging consistency helper from container.sharder and move it into common.db_replicator\n* update all log message call sites in common.db_replicator\n* update all log message call sites in container.replicator\n* update test.common.test_db_replicator fakes to better match real brokers db_file and path properties\n* update assertions for existing log messages\n* add assertions/tests for all updated log messages\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61decd5d42929e6be4894b35e5e24b286ba6d709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61decd5d42929e6be4894b35e5e24b286ba6d709"}]},"branch":"refs/heads/master"},"777cc77329b3e38f3632b7d4e6172875c77973b1":{"kind":"NO_CODE_CHANGE","_number":39,"created":"2025-09-30 00:00:20.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/39"}}},"commit":{"parents":[{"commit":"388656ab1b429eb7ec317b3f14fef58dcd4a81bd","subject":"account-broker: add resilient path property with lazy cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/388656ab1b429eb7ec317b3f14fef58dcd4a81bd"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-30 00:00:20.000000000","tz":0},"subject":"common.db_replicator: log container, db paths consistently","message":"common.db_replicator: log container, db paths consistently\n\nExtract helpers (from container.sharder) that formats log\ncontext from either a broker (preferring broker.path/db_file)\nor a plain db_file string. Use it in common.db_replicator and container.replicator so messages are uniform and robust.\nUpdate tests to cover both cases.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/777cc77329b3e38f3632b7d4e6172875c77973b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/777cc77329b3e38f3632b7d4e6172875c77973b1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a99d4312dc4dc1e3e1fa9e51851da0753b5ed9ac":{"kind":"REWORK","_number":40,"created":"2025-10-02 19:58:44.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/40"}}},"commit":{"parents":[{"commit":"388656ab1b429eb7ec317b3f14fef58dcd4a81bd","subject":"account-broker: add resilient path property with lazy cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/388656ab1b429eb7ec317b3f14fef58dcd4a81bd"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-02 19:57:43.000000000","tz":-420},"subject":"common.db_replicator: log container, db paths consistently","message":"common.db_replicator: log container, db paths consistently\n\nExtract helpers (from container.sharder) that formats log\ncontext from either a broker (preferring broker.path/db_file)\nor a plain db_file string. Use it in common.db_replicator and container.replicator so messages are uniform and robust.\nUpdate tests to cover both cases.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a99d4312dc4dc1e3e1fa9e51851da0753b5ed9ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a99d4312dc4dc1e3e1fa9e51851da0753b5ed9ac"}]},"branch":"refs/heads/master"},"ba859125f40731c2f4066542334e6b3d847faaba":{"kind":"NO_CODE_CHANGE","_number":41,"created":"2025-10-02 20:00:26.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/41"}}},"commit":{"parents":[{"commit":"388656ab1b429eb7ec317b3f14fef58dcd4a81bd","subject":"account-broker: add resilient path property with lazy cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/388656ab1b429eb7ec317b3f14fef58dcd4a81bd"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-10-02 20:00:26.000000000","tz":0},"subject":"common.db_replicator: log container, db paths consistently","message":"common.db_replicator: log container, db paths consistently\n\nExtract helpers (from container.sharder) that formats log\ncontext from either a broker (preferring broker.path/db_file)\nor a plain db_file string. Use it in common.db_replicator and\ncontainer.replicator so messages are uniform and robust.\nUpdate tests to cover both cases.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba859125f40731c2f4066542334e6b3d847faaba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba859125f40731c2f4066542334e6b3d847faaba"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cbe3c88083dfe0d96e7ae33bacbdc6aca811f8f8":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2025-10-02 20:04:53.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/42"}}},"commit":{"parents":[{"commit":"ac5c783d652e66f2f19cdb1d8ba29b5dc3c51710","subject":"Merge \"Assert metadata of SLO PUT from container sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac5c783d652e66f2f19cdb1d8ba29b5dc3c51710"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-10-02 20:04:53.000000000","tz":0},"subject":"common.db_replicator: log container, db paths consistently","message":"common.db_replicator: log container, db paths consistently\n\nExtract helpers (from container.sharder) that formats log\ncontext from either a broker (preferring broker.path/db_file)\nor a plain db_file string. Use it in common.db_replicator and\ncontainer.replicator so messages are uniform and robust.\nUpdate tests to cover both cases.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbe3c88083dfe0d96e7ae33bacbdc6aca811f8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbe3c88083dfe0d96e7ae33bacbdc6aca811f8f8"}]},"branch":"refs/heads/master","description":"Rebase"},"b823b912f3d9678e8d300e6359952f54c2ff6472":{"kind":"REWORK","_number":43,"created":"2025-10-08 15:42:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/43"}}},"commit":{"parents":[{"commit":"97e00e208fc6a1076bef41ba28a17cda8974aeb0","subject":"trivial: reorder AccountBroker.path methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97e00e208fc6a1076bef41ba28a17cda8974aeb0"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-08 15:42:09.000000000","tz":-420},"subject":"common.db_replicator: log container, db paths consistently","message":"common.db_replicator: log container, db paths consistently\n\nExtract helpers (from container.sharder) that formats log\ncontext from either a broker (preferring broker.path/db_file)\nor a plain db_file string. Use it in common.db_replicator and\ncontainer.replicator so messages are uniform and robust.\nUpdate tests to cover both cases.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b823b912f3d9678e8d300e6359952f54c2ff6472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b823b912f3d9678e8d300e6359952f54c2ff6472"}]},"branch":"refs/heads/master"},"f8607fbc8b0dd7898fc224dfc6d2dd231957ae0c":{"kind":"REWORK","_number":44,"created":"2025-10-08 15:43:59.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/44"}}},"commit":{"parents":[{"commit":"97e00e208fc6a1076bef41ba28a17cda8974aeb0","subject":"trivial: reorder AccountBroker.path methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97e00e208fc6a1076bef41ba28a17cda8974aeb0"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-08 15:43:53.000000000","tz":-420},"subject":"common.db_replicator: log container, db paths consistently","message":"common.db_replicator: log container, db paths consistently\n\nExtract helpers (from container.sharder) that formats log\ncontext from either a broker (preferring broker.path/db_file)\nor a plain db_file string. Use it in common.db_replicator and\ncontainer.replicator so messages are uniform and robust.\nUpdate tests to cover both cases.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8607fbc8b0dd7898fc224dfc6d2dd231957ae0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8607fbc8b0dd7898fc224dfc6d2dd231957ae0c"}]},"branch":"refs/heads/master"},"41bf72a5cc5dbda1192732930c78a1fd91282f98":{"kind":"REWORK","_number":45,"created":"2025-10-11 22:14:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/933379/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/933379/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/933379/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/933379/45"}}},"commit":{"parents":[{"commit":"97e00e208fc6a1076bef41ba28a17cda8974aeb0","subject":"trivial: reorder AccountBroker.path methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97e00e208fc6a1076bef41ba28a17cda8974aeb0"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-20 23:18:59.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-11 22:13:46.000000000","tz":-420},"subject":"common.db_replicator: log container, db paths consistently","message":"common.db_replicator: log container, db paths consistently\n\nExtract helpers (from container.sharder) that formats log\ncontext from either a broker (preferring broker.path/db_file)\nor a plain db_file string. Use it in common.db_replicator and\ncontainer.replicator so messages are uniform and robust.\nUpdate tests to cover both cases.\n\nNo functional changes to replication behavior; this is\nlogging/robustness and test updates only.\n\nChange-Id: I289d3e9b6fe14159925786732ad748acd0459812\nRelated-Change: I7d2fe064175f002055054a72f348b87dc396772b\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41bf72a5cc5dbda1192732930c78a1fd91282f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41bf72a5cc5dbda1192732930c78a1fd91282f98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
