)]}'
{"id":"openstack%2Fswift~933468","triplet_id":"openstack%2Fswift~master~Id7e5739057b83bd697962d46128a25e786599d0b","project":"openstack/swift","branch":"master","topic":"chunk-transfer","attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-02-14 01:53:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-25 21:49:55.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Id7e5739057b83bd697962d46128a25e786599d0b","subject":"sq: modify is_sha256 check and change tests","status":"NEW","created":"2024-10-25 22:46:38.000000000","updated":"2025-07-18 08:07:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":19,"unresolved_comment_count":10,"has_review_started":true,"meta_rev_id":"2fb18b248eb71e232c7ae417d04dbbcc0254a7d9","_number":933468,"virtual_id_number":933468,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-14 00:54:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2025-07-18 08:07:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 22:46:55.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-10-25 22:50:09.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2024-10-25 22:50:09.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-10-26 00:01:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-18 08:07:20.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"5604e9559068ff59d9f224a4a3889b6889c6b9a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 22:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"928c89b884a1046099580da11768deb13ac3d411","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 22:46:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"09397a7e984d447f09aa01239e788e76cf356819","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 22:46:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a0e208b082937adcd7c2ab8cd91c3f93305d86a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 22:47:01.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"9dec7301bb83848eab3ebef60e881ddb6803141c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 22:49:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f195aa47cf5171a57f804a5c99e0d6b528a89cae","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 22:49:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"08d574d959978a12493bc6f4be83bc3c6df928b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 22:52:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5048e1bcedc9d7fce7929642b4ad1732a6bfef20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 00:01:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3b3779814834a0e8143f29d6afcf008\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0298e38183ab4d3eb2a072ea1ac4920f : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/f53a2c10866b495b9945c82d294678cf : SUCCESS in 56m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de73e8c834af4ba79fc01be653023f29 : SUCCESS in 1h 02m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/47cd0f50dad045bdb550e4318422f1d3 : SUCCESS in 40m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/456f529efc1649eeb6730ce485ba8986 : SUCCESS in 42m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a39012166e5e4abeba1bb68a33423208 : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca3457ffb6ac41929cd13224eb4ab563 : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d803894728e9425b89cc07af475fa628 : SUCCESS in 15m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1cab8c17e92c4b3e97ff49055062fe1e : FAILURE in 3m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/530ac34300a2477fa0964a6724952a4e : SUCCESS in 13m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4245c22f922247f88bfe0276af6d84f3 : SUCCESS in 16m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ab61a67c3f134c6f8837a676bd53ed62 : SUCCESS in 18m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d9a2709e992444bab7c882a084a002c6 : SUCCESS in 22m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ba1f322901c1431698728919291f27c5 : SUCCESS in 23m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cebcd8ddbe4144778a169c32794e6bca : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/61c3871c396347b294ab3845a33fd1b7 : SUCCESS in 19m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c174ea7481604ee786d8ff34650cf7cb : SUCCESS in 9m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6c5f284cb35c47e28671a08b72455278 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c59825ca47e64854b3a4b66e84f0eeea : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e053e83d2d794bfe8782d99ac4fcebb5 : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c65e15f93de9438aa3058d45e80bbf80 : SUCCESS in 7m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e2ca03860594f3b9dea443313c413a1 : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7ec88faa2a664c798bcdf424a5991f71 : FAILURE in 10m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9f2fbb5c7e84ae0adf5e102931537b6 : SUCCESS in 54m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8d1635d13e64ddc9250f15af2a1409c : SUCCESS in 26m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3015c025a4e643b58dfb427c09abdccb : SUCCESS in 54m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bca2871629f34f5ba4394398055b5208 : SUCCESS in 17m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d808727f7b7f4d988cf63d142e2991a9 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/95dc8b27983e489a8247a6d3adae55c5 : SUCCESS in 16m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a8ba2ba3def413496853342e71cb6c7 : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":3},{"id":"1dc0bcaac6218d83c933049d416aafa2ffde3191","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 00:20:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/024205ed6c7d4468a5e68765af4d796f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/106154ba81174a2db266d1cb77c206b9 : SUCCESS in 1h 25m 35s","accounts_in_message":[],"_revision_number":3},{"id":"092126b667b16d0c196694b454624ab5f9b88997","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-28 15:37:46.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"fd9c7d0b42a74183fa8d3bb64cf2c84da4bda771","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-29 15:50:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d636f68854251e17b737511248feaae6fcf69d87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-29 22:43:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b0b7cf81120538c9b7266b3da3ade541b866fb4b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-29 22:45:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1fd323406cf0e9d53c42c2ad3306c63a1ed4c242","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-29 22:58:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b7ddfe34308f0096ce7d35b4d678b7d2ba131cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 23:48:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4577ff5462a24417ba39a7d64c4f37ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec80efedf8e49a4bf286991a0822538 : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/cafb77783d524f1bbb77d54ca754c17e : SUCCESS in 37m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/10e8ae3f924841a382ec80fdaafdac87 : SUCCESS in 35m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da576e673e1a4d248e9ca278766d7acc : SUCCESS in 33m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3460f072c48440caa15ace8d9a8f8301 : SUCCESS in 59m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42638e243c9f4396842db115cead49e9 : SUCCESS in 11m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7529fbba24de48c5994d1a78dc7452fa : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec7e4281dbb34901a7dce5ccab875e7b : SUCCESS in 19m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/00edd29b41a044228ac08a659e4d30df : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/62116c66f78944cfab8c1b582750c586 : SUCCESS in 17m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2427a170e8d44d0e8134b29bb56b0936 : SUCCESS in 17m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9adee4385e2b4ad49923083874b418a3 : SUCCESS in 17m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cc1984f530a5437ead86e02ea6f929f0 : SUCCESS in 16m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c0ac863884ca49bf87acc1561a2e2bf2 : SUCCESS in 18m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a5c773f5196b4fb5821775d919495e29 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb4e1274b02040dbbcab2e676c7948f4 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e2d4c9c25195401c811df205f64bc4d2 : SUCCESS in 17m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8541dc1719b4466193de05cc49716b2a : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/df4afeea113744c4ae90dbacdde4a957 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/14fc4f6e83f64d06ac0f2e890d05e092 : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e304a83c58bf41a99f751ae0868b7f62 : SUCCESS in 4m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea63fdbb161e40be87405e18f8a390ba : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d95abed99dfc43a4aec50435f8b96bc5 : SUCCESS in 6m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2298b3de05da48ccaf3259d0cd977534 : SUCCESS in 47m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ad9ac92d0a640a9993dc5122886c36b : SUCCESS in 29m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97dc328d3a96473890f316d2a0bc6164 : SUCCESS in 51m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fef1078fd90e409fab3a7015ba98422e : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745eefa03aba46509b86a4bf07bc3220 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/81dd1e2c0fbe4320a9b8dae018fd50bb : SUCCESS in 14m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5959c92e71f2452d8236b69791d98e39 : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d43eaa5aff5768b11a41683c0d4cec0a5ec72926","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-29 23:53:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7b7c4b3f2c63b20034845a52e95bcb09dbcf90f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 04:14:02.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b8c4bc8d301478eada9904afffcab7a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4891693117e6487397c96231bc439801 : TIMED_OUT in 3h 01m 35s","accounts_in_message":[],"_revision_number":4},{"id":"f76b8ef59c6cca1be05bf2361b890d8fd8c05c6c","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 15:19:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b3523108c797d23e0889aafb836d49b093ab9b13","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 16:10:03.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"564ceab988460bedbcabaae5824fc96afe76300f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 16:24:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22e2e2fb999b4737b11431da6689f38a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d41e5af0b643c2b665e7a6d1063232 : SUCCESS in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/d07f11c3f4454c429f9ac8612e8648b3 : FAILURE in 46m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/70f2e480052a4afa93dbf69053f2f38f : SUCCESS in 49m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3d5bd9e23955428aab03a2e4ece6c408 : SUCCESS in 49m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c2d20f46f494ca29bdbc81bafe9f658 : SUCCESS in 58m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f99ae7c0a4e0437991caf0a4d38dc8a8 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccec99c47d484afa8c7c876a197650a2 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87c6c05068f6476bbd9b550243ef04b7 : SUCCESS in 22m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a6d9187eb5cf43008e72a0e599a82828 : POST_FAILURE in 35m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/473faa1e2d80465887d30bb36b0e1a29 : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1aaf69eebd64cf785f650c9dea9ca5e : SUCCESS in 19m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6318b0d300c743eca5b3b44497789d58 : SUCCESS in 18m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/328ff598fd554d56a5b4e38d282f9cc4 : SUCCESS in 21m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0cf11b8aa5cc4112a98dbe1f9e0f09ea : SUCCESS in 33m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e6738f3e241482ca2183ea9e9ce50e4 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d1e181498334b0f830ed111d19afb0c : SUCCESS in 19m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc44c7a874214c55ba1fdc3d26b760d7 : SUCCESS in 12m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f2652907dfee4722b5de2aae93ce1bc8 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/041b8bac4a424f569cdd5ca551566dfd : SUCCESS in 15m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6d832037b7a0455fab8e2ffce7ccbd86 : SUCCESS in 12m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75d7d013c63f409e89ceee18c114fac9 : SUCCESS in 9m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4528cefad34a4c2abb6e322ea1f1f55c : SUCCESS in 14m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e49f9bd2ad647cea8aef701f1d0fbcf : SUCCESS in 10m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/da0f7f4332e845a8a38cafb0de4d7989 : SUCCESS in 47m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e14b375b24d48cc99a6d19ae7fde183 : SUCCESS in 51m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a3268af5cac4ff18c9dab4f37a1ef91 : SUCCESS in 32m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a32f1f3a3b35441b9c5866b759c18345 : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c46e3cddb8748ec95d4ecf19dccba69 : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ab4e01a735348b6a158979b677e37d2 : SUCCESS in 20m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/227e8c11efeb4a64b84c660ce2669087 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e115c4ee9b6d98238523df84a9c46b2bb68c2271","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 17:30:28.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"226db3d7922c4392c7391832bd229dc7bf7373d6","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 17:31:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"427e12ad1393700bf97d87ffecf8d88ee6b0cc53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 19:12:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe7737c74c07474980657b2807583eaf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8dc97ef9857943f29adc6be9f853e18a : SUCCESS in 1h 44m 16s","accounts_in_message":[],"_revision_number":4},{"id":"600b474f6ec9a33d96c7687dcc3ee7556a1a3d09","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 20:17:22.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"236a9e2c4385fb371c8c55408792090cfd51d6db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 18:05:12.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ae07beca54a9f3327f987a52d79439860288f9fe","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 18:05:29.000000000","message":"Patch Set 5: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"bcb790b53fc9c7cfe97362f96f2f45c58ab24619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 19:14:43.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51f30d47a2c24ecb863dbe3d6b5d4bf3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c68220ab3034ef4848e1016f0d37408 : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/221ac60ae66e4e78a12d16f1b14a718f : SUCCESS in 1h 08m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/45d1015381ca414daeef40b1bc79d17e : SUCCESS in 52m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b9c19e30cb4475dbe64bc649a013f26 : SUCCESS in 47m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d7d8a2bece740b295e116d337c684e7 : FAILURE in 35m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b85bbae68804637a938dc5dc3185d2e : SUCCESS in 12m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d8b3c664eb424962a822a2abe181730c : SUCCESS in 12m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f141204d927b48d58905009ef48ea6ad : SUCCESS in 23m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67f4f893649648debbf26f50b46862f4 : SUCCESS in 4m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/effbed52ffdb424ca498aad1b9243403 : SUCCESS in 21m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d8e33be8987d45a1899050505bb5d4d7 : SUCCESS in 9m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/664af8be341d415fa96c8d8573c8c567 : SUCCESS in 21m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c14ff082aeca4120ac12f0c5e10d4f43 : SUCCESS in 22m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d00c7fd54a548b7bb53c1eb7e337eaf : SUCCESS in 9m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b2122efac65b49a28b8c82ab5e5ef781 : SUCCESS in 20m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/67b07c96f2a94570b13d63401ab1557e : SUCCESS in 10m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/df4e716b33ae4bdd834e68070b9188f2 : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/290659445a604b138b00a33d268098e0 : SUCCESS in 9m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c0d90e6f049b4590ba5b7f42233cc34c : SUCCESS in 18m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/65e340eef68d4ed59b60a54005aab179 : SUCCESS in 4m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1b67f9cf600461980e2c85315af058e : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8ec66c6f2e08468a99a9bb0c8ed4fcf3 : FAILURE in 6m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffe0fa503e904ec8b74ae8efc6a5b22b : SUCCESS in 56m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1570b529870a401f959212ec39b302a3 : SUCCESS in 27m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fbe79a667c43428ab59db1b728008a56 : SUCCESS in 48m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/751848ebb7a542389551e62303ae681e : SUCCESS in 26m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e989b5b8dc423b83de3cfdac8221f8 : FAILURE in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bec4cd4ec5e94c59b046e562f0e667be : SUCCESS in 18m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2911dabe18794d058891cd3a9d91d1e2 : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":5},{"id":"ce11ff1be12e4682e7dac63f48b38e6b716f8ac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 19:29:52.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"b69cb913d4c0239dfd3cc89427edaead60a19142","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 19:30:01.000000000","message":"Patch Set 6: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"a3639787a600255614ecb605bbdaf2eb65118d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 20:30:51.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a89b62f5e19e4c478d8798fdb71cd941\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963aedcdae144b7695bd1f9a5ed984eb : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/d3ae7f2e11e64100bf49909aa251b8bc : SUCCESS in 55m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/870808bc817949a3b51e14deb9139bfa : SUCCESS in 55m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/edf7d19df53d4bae9005743ea955f6c9 : SUCCESS in 45m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2dd86025da6047f68fce96b31402a69d : SUCCESS in 55m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a502abb6027b4a25aade16d0d7d56b7b : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/232d9e81e56c44f2bfc4138be999252c : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c31a81aa13fa45d58a9ed136c68def65 : SUCCESS in 17m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94ba01999af94f71a1a9bd2a80fcd968 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/956a52c06ff5427e938ca934183389a7 : SUCCESS in 13m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a2b3ee338954e539573559f4e8e6fb4 : SUCCESS in 9m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dba50ae558ae4742841d2e00cd288c7b : SUCCESS in 20m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/856bc637cf7b406298070b9498d87d1f : SUCCESS in 21m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4546b2cd4ba446cb8516cce3a9a073b9 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/adebb0561c9d4de89f06fb327eebacfa : SUCCESS in 13m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b6469355d08748a68ec0737dba6e38c6 : SUCCESS in 19m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6a73aacefa17443bb1bbf7a42a17e5a3 : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2012b5bab88e4a51a058ea6d5e2b1ae9 : SUCCESS in 13m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/37d97370411e410c93e513372d63f294 : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b0d0ee82a164b888a46f391278ace1b : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/27c0e36acb0d45339abe06cf26438c9a : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/de247effb57646d99f5042f92b97c7b4 : FAILURE in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cb939901a1a48f49260ffc3f296fe8a : SUCCESS in 48m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5424f5a3f4046b594d5a41957e028d1 : SUCCESS in 48m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/510d8b9fe0a940c9badc9a645522dde0 : SUCCESS in 45m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d9d2541d10c4521abd19b0070b3c692 : SUCCESS in 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30fcb68b9ce14bd0b21081439235bd80 : FAILURE in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0d5297530164267aaf426d5d7187e56 : SUCCESS in 10m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7034c6157b654c8ca7987357244a106e : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":6},{"id":"302518ab09138fdc0875852d02fa6c2551692746","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 22:29:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ec4877f971045b298c942858e59498e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a45a5cac13a4dedb64bd3f925910582 : SUCCESS in 1h 40m 37s","accounts_in_message":[],"_revision_number":6},{"id":"fa3f604369db4a73ff7e7a786ab3619b68b8f24c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-25 21:49:55.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nThe following files contain Git conflicts:\n* test/s3api/test_input_errors.py\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"baa15b17f5652660f76b1054d2f3b87483e0613c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-25 22:02:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"227c6d4aa32cad2a8c1460d893d8e0cc025db591","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-25 22:05:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0128baf9f255d87b2c9e467565f398e303f8c638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 23:13:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8132126223848c2841ee395b621a1d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce1f47eabc9249109f4161fca5eb6260 : SUCCESS in 5m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/42f5d4ec8177441ba7068ee684792721 : SUCCESS in 39m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/46e358916c2c460c89e2a2e14410fc0b : SUCCESS in 1h 00m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f70158df56014a5cbbb4e308ffc8b77e : SUCCESS in 49m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83366b88300f4839b0e83a466ef4f0ea : SUCCESS in 1h 01m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57e92b19efee4638ba88d015b011b81d : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e7feab886ec4484b7ccc3167285faf8 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f00c555b379e4584ba22d11730a3fb04 : SUCCESS in 12m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c330feb43464ee7b86cf545f71234d3 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14e13dd3bef8453fa3829e81a13f604e : SUCCESS in 20m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/887c19e9f61749f3a7d2c3a1aa6f854c : SUCCESS in 18m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c405e03b6aa4d838b13cb51c186031a : SUCCESS in 17m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fcb2fe3874b84fa1a31e6dad74388b0c : SUCCESS in 21m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21c20bc7fc1e4cf8807ea0587783b923 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/940a90f3e7ff45869731cee95f8d008a : SUCCESS in 16m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/30a3b6ec71d345f2856db5a199c357ec : SUCCESS in 14m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/32fa19acd839422780d013fdf32f4a5b : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0c23bd680c1541d6be6d4383881b9e02 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f82cf895fd144004b7ee19c595065ca3 : SUCCESS in 15m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a5549bdf4757482690e5992bc9bf98e4 : SUCCESS in 4m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/96841888f15f400e8168d39305ba2d44 : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad46115082194f5084acd0be12fa2387 : FAILURE in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8bb7ebed6a5450f8f307a04d354e15d : SUCCESS in 53m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0788ff0af42347ef8152123ef3a40fd9 : SUCCESS in 29m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/81480cc5697b42608db429e0623738b8 : SUCCESS in 52m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c59fed0b248f4e1f80663fbcccf78fdc : SUCCESS in 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc33a96325424fa4882e828b4cd6e0e1 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e5b48d08ab6c446fa3a33225b6670618 : SUCCESS in 19m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb637aeb32df4c258a414238a94c05d2 : SUCCESS in 33m 39s","accounts_in_message":[],"_revision_number":9},{"id":"917b0c98385e0a8a974ef772dce844a1845bde4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 00:05:59.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9a434bfdd214ec1a7eab8c79ca6e88a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f164762d772d4aff8da16a97acde589a : SUCCESS in 1h 51m 23s","accounts_in_message":[],"_revision_number":9},{"id":"abed900dffe64fb80fa501f5153f15eb0d144d37","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 19:12:28.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/s3api/s3request.py\n* test/unit/common/middleware/s3api/test_s3api.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"55c246df2f5c46443b6c8871b7abc3219d628434","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 19:19:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"24787c7e33d2fd4b6e9191c9bf05e04f2bf7e1ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 19:29:58.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b57290ea86036083a3a1e5765f977655979e1861","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:48:34.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489850de3a6b4bf689452deb9e9b8d0c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/36402b4bfbb54d3aa3c1ffda1524cb9d : FAILURE in 1h 15m 46s","accounts_in_message":[],"_revision_number":12},{"id":"9035c85d09935a0ca5c25c3c137dbc946fe79697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:52:04.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/782346c089a147e49d7eeb1d1da264dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05b623db2b54b5fad4596ad7eed549d : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/05eec84a417b4275a2fca417a31f445f : SUCCESS in 1h 09m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a26c78c64f64139901e09f8cad9f0ba : SUCCESS in 43m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/376c0754b2884454aa10bb3e033c1164 : SUCCESS in 1h 06m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6406b78e33149d9b17cd17e84479aa7 : SUCCESS in 48m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91fe7ff5f35846c1963f5d0281aab1f3 : FAILURE in 19m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94c5f0086c9b4e3092151175045097a7 : FAILURE in 12m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd0ac19bf8a64844ab431696ca199118 : FAILURE in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b25da6e0d734a038b20b23be64f462f : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da4fc4c98359435f8a23b95782247e00 : FAILURE in 11m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f04f5d5f69824f928faa4258b9dd52aa : FAILURE in 20m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d0db4f8f68be4c16ba0ca1f7c4492270 : FAILURE in 34m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0e3d92a3b2b64b9a87bbeb1c00fdbef6 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44775ac5720f43e19951abb14cd96d01 : SUCCESS in 15m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f1c246c6ab1044aba61710806f298fa4 : SUCCESS in 9m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9df6f2eebe3f47019696f04448eb5c1c : FAILURE in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ff9615b83e44294a21da6b111027293 : FAILURE in 17m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d3498e0960c1481488ed748db7f88a45 : FAILURE in 9m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8f95d3cb7294ee8a77cf62ed0551572 : SUCCESS in 55m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/729f89a8d5b446678f32d299cca8bb60 : SUCCESS in 47m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c4ba2250b2549048c5aa5aab721c360 : SUCCESS in 37m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47957bc1b44c4e1ba2e2dd45f4258023 : FAILURE in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2686ae31fd674c12a821821dec013305 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a57e5308a614fc88457d53ccb4a17b0 : SUCCESS in 22m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4490c660f4624b43bfb7a9e9134c9136 : SUCCESS in 1h 00m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eeb0d3c4f54542e891d8da21321fde94 : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":12},{"id":"26dbfffd2208291069821501424e6fe43ee4fc71","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 23:58:26.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"f96a3bae0b66c51a83256e5b6c94a1fa482bdd39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 23:58:48.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/773129046d394273baefe37ec565e34a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ed36c84db07042c09d6c925e9df37366 : NODE_FAILURE Node(set) request 300-0026334662 failed in 0s","accounts_in_message":[],"_revision_number":13},{"id":"a01d5ac16221e4f4637e51106e67fc5ad8bd8fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 00:54:10.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebecd32d59744b6e9f6f6a8092a6d0d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038a49643f794590bd78136a3c218a21 : SUCCESS in 6m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/1d6764b91533449db1473f9721f2f22c : SUCCESS in 39m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2a653a0dc64345a9896c2471a3dc4865 : SUCCESS in 44m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/417fe634277645038f9f130b4b089491 : SUCCESS in 41m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c9f03b26f68459f80f360008e42d1f0 : SUCCESS in 46m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/010a46b117ae4a1dae11a6a4d3fd26b9 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/799d0641cfaf4cb58f286b298d015c88 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/682fa227e5274504a6ffd160ba6da56c : SUCCESS in 13m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/71252346eadb4f0da74a2f5e81f0d994 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/20513a6953c648cb808d471e89c5d4fb : FAILURE in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3d16edb382b4fb88804bcb5356978e1 : FAILURE in 10m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f05f66e7703745bdbb68dc38c9d5c037 : FAILURE in 18m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/106fd6d3e09141e8932727bb8875f043 : SUCCESS in 9m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/05437bb85b1e4b4cb35e38679dc5b2f0 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4796bf83daf74e57bb1df23f357a09ea : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/71f2d2b4b7f8400eb602663d0a093e3c : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf44d524356b42e280a3babacb12550d : FAILURE in 15m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/29408293ece74686932b9eee5e0fa60e : FAILURE in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/088b11b67635486997cf0ac5415e7d47 : SUCCESS in 54m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab34b7ed8b1d424f8a2eb1250051bc3d : SUCCESS in 49m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb926cae93bb4386915b1f9b174a8f79 : SUCCESS in 49m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c399dbc770a743978d16611e8573edbc : FAILURE in 22m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9905649f2f149d5bc5425ea1f6e8add : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fea551dd8a6b4f10ad5a422227baed4b : SUCCESS in 19m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c52c64d7bb6e469c9bb766d6992717a7 : SUCCESS in 37m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/995b47d04025430ca53aaeb833493643 : SUCCESS in 39m 42s","accounts_in_message":[],"_revision_number":13},{"id":"cfb64771f367732f07b4a8152742dd7325bb8fcb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 01:53:04.000000000","message":"Patch Set 13:\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":13},{"id":"2fb18b248eb71e232c7ae417d04dbbcc0254a7d9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-18 08:07:20.000000000","message":"Patch Set 13: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"66ebf11b691958a85b1e0ea764ebe71d03705159","revisions":{"564fedaa6e102135b91b0e20812d3c48b5e9df2c":{"kind":"REWORK","_number":1,"created":"2024-10-25 22:46:38.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/1"}}},"commit":{"parents":[{"commit":"6e06604eeb0a21adbb24e4cbe9035e2758120b9c","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e06604eeb0a21adbb24e4cbe9035e2758120b9c"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/564fedaa6e102135b91b0e20812d3c48b5e9df2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/564fedaa6e102135b91b0e20812d3c48b5e9df2c"}]},"branch":"refs/heads/master"},"14a74dfad2a68be3fdd256dc15c38506a4dc291c":{"kind":"REWORK","_number":2,"created":"2024-10-25 22:49:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/2"}}},"commit":{"parents":[{"commit":"6e06604eeb0a21adbb24e4cbe9035e2758120b9c","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e06604eeb0a21adbb24e4cbe9035e2758120b9c"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:49:22.000000000","tz":-420},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14a74dfad2a68be3fdd256dc15c38506a4dc291c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14a74dfad2a68be3fdd256dc15c38506a4dc291c"}]},"branch":"refs/heads/master"},"2ae4e4b51d07da27016a800ada94f82cf6b0da60":{"kind":"REWORK","_number":3,"created":"2024-10-25 22:52:42.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/3"}}},"commit":{"parents":[{"commit":"6e06604eeb0a21adbb24e4cbe9035e2758120b9c","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e06604eeb0a21adbb24e4cbe9035e2758120b9c"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:52:31.000000000","tz":-420},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ae4e4b51d07da27016a800ada94f82cf6b0da60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ae4e4b51d07da27016a800ada94f82cf6b0da60"}]},"branch":"refs/heads/master"},"e93f4137d3815e9c055b88c4b2d33a9ad8ac8f42":{"kind":"REWORK","_number":4,"created":"2024-10-29 22:43:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/4"}}},"commit":{"parents":[{"commit":"6e06604eeb0a21adbb24e4cbe9035e2758120b9c","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e06604eeb0a21adbb24e4cbe9035e2758120b9c"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-29 22:38:00.000000000","tz":-420},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e93f4137d3815e9c055b88c4b2d33a9ad8ac8f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e93f4137d3815e9c055b88c4b2d33a9ad8ac8f42"}]},"branch":"refs/heads/master"},"cf3b46e9b9e646a40379f595bc1d498b51c6d62b":{"kind":"REWORK","_number":5,"created":"2024-11-04 18:05:12.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/5"}}},"commit":{"parents":[{"commit":"6e06604eeb0a21adbb24e4cbe9035e2758120b9c","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e06604eeb0a21adbb24e4cbe9035e2758120b9c"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-04 18:04:55.000000000","tz":-480},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf3b46e9b9e646a40379f595bc1d498b51c6d62b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf3b46e9b9e646a40379f595bc1d498b51c6d62b"}]},"branch":"refs/heads/master"},"05657163ee1ff993458bad67a96ce074f7677e0d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-11-04 19:29:52.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/6"}}},"commit":{"parents":[{"commit":"b9746fd0ee6018aa1467e746c2c0d72315866794","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9746fd0ee6018aa1467e746c2c0d72315866794"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-04 19:29:52.000000000","tz":0},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05657163ee1ff993458bad67a96ce074f7677e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05657163ee1ff993458bad67a96ce074f7677e0d"}]},"branch":"refs/heads/master","description":"Rebase"},"bfd661461e0cffae044e46af99d4fbf76d61d1e0":{"kind":"REWORK","_number":7,"created":"2024-11-25 21:49:55.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/7"}}},"commit":{"parents":[{"commit":"cb97e987a523ab7fb6ff09566f940f4f2e3f6be3","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb97e987a523ab7fb6ff09566f940f4f2e3f6be3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-25 21:49:55.000000000","tz":0},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfd661461e0cffae044e46af99d4fbf76d61d1e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfd661461e0cffae044e46af99d4fbf76d61d1e0"}]},"branch":"refs/heads/master","description":"Rebase"},"cf0688ed8652b26a34b8848d9c0d42adde4b022a":{"kind":"REWORK","_number":8,"created":"2024-11-25 22:02:50.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/8"}}},"commit":{"parents":[{"commit":"cb97e987a523ab7fb6ff09566f940f4f2e3f6be3","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb97e987a523ab7fb6ff09566f940f4f2e3f6be3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-25 22:02:24.000000000","tz":-480},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf0688ed8652b26a34b8848d9c0d42adde4b022a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf0688ed8652b26a34b8848d9c0d42adde4b022a"}]},"branch":"refs/heads/master"},"0d1e074127aeec3ffd3d8685da4527ff347f657e":{"kind":"REWORK","_number":9,"created":"2024-11-25 22:05:22.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/9"}}},"commit":{"parents":[{"commit":"cb97e987a523ab7fb6ff09566f940f4f2e3f6be3","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb97e987a523ab7fb6ff09566f940f4f2e3f6be3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-25 22:05:01.000000000","tz":-480},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d1e074127aeec3ffd3d8685da4527ff347f657e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d1e074127aeec3ffd3d8685da4527ff347f657e"}]},"branch":"refs/heads/master"},"c3f574183fd13ffb6884fccb9ff91b5d25eabe95":{"kind":"REWORK","_number":10,"created":"2025-02-13 19:12:28.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/10"}}},"commit":{"parents":[{"commit":"1b38348be4bb37c9e335d794b7e52eaff1e9edfb","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b38348be4bb37c9e335d794b7e52eaff1e9edfb"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-02-13 19:12:28.000000000","tz":0},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3f574183fd13ffb6884fccb9ff91b5d25eabe95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3f574183fd13ffb6884fccb9ff91b5d25eabe95"}]},"branch":"refs/heads/master","description":"Rebase"},"ff3d93af67f216d0a798fa1a3201a08e3e57a791":{"kind":"REWORK","_number":11,"created":"2025-02-13 19:19:19.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/11"}}},"commit":{"parents":[{"commit":"1b38348be4bb37c9e335d794b7e52eaff1e9edfb","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b38348be4bb37c9e335d794b7e52eaff1e9edfb"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-02-13 19:19:03.000000000","tz":-480},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff3d93af67f216d0a798fa1a3201a08e3e57a791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff3d93af67f216d0a798fa1a3201a08e3e57a791"}]},"branch":"refs/heads/master"},"ca35953571b11f4509a904c289b5adbecb0eb351":{"kind":"REWORK","_number":12,"created":"2025-02-13 19:29:58.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/12"}}},"commit":{"parents":[{"commit":"1b38348be4bb37c9e335d794b7e52eaff1e9edfb","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b38348be4bb37c9e335d794b7e52eaff1e9edfb"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-02-13 19:29:04.000000000","tz":-480},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca35953571b11f4509a904c289b5adbecb0eb351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca35953571b11f4509a904c289b5adbecb0eb351"}]},"branch":"refs/heads/master"},"66ebf11b691958a85b1e0ea764ebe71d03705159":{"kind":"REWORK","_number":13,"created":"2025-02-13 23:58:26.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/68/933468/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/933468/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/933468/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/933468/13"}}},"commit":{"parents":[{"commit":"1b38348be4bb37c9e335d794b7e52eaff1e9edfb","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b38348be4bb37c9e335d794b7e52eaff1e9edfb"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-25 22:44:50.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-02-13 23:58:03.000000000","tz":-480},"subject":"sq: modify is_sha256 check and change tests","message":"sq: modify is_sha256 check and change tests\n\nChange-Id: Id7e5739057b83bd697962d46128a25e786599d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66ebf11b691958a85b1e0ea764ebe71d03705159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66ebf11b691958a85b1e0ea764ebe71d03705159"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
