)]}'
{"id":"openstack%2Fswift~933557","triplet_id":"openstack%2Fswift~master~Ifd33f6ae98a4a96605a4d3dcc028ea42f23e3940","project":"openstack/swift","branch":"master","topic":"rebase-parallel","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-28 19:26:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifd33f6ae98a4a96605a4d3dcc028ea42f23e3940","subject":"sq? fixups for test_expirer.py","status":"ABANDONED","created":"2024-10-28 11:37:07.000000000","updated":"2024-10-28 19:26:25.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e68a20300900adcb0b286b4a744c91587224cf44","_number":933557,"virtual_id_number":933557,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-28 12:36:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-10-28 19:11:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-28 12:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-28 19:11:02.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"0a991a7cff26dd1fb9ea7f9f8cf55bb5d5f3dd40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-28 11:37:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"025ed685a30855be3e5bf82e83a174eb31e5ad48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:36:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd51cff7cb2f49ce8cb5fae7438bdf71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd211f828cb44cca3b72a7fa862c1c3 : SUCCESS in 7m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0405b12705b54dddb56202b9b1e1961f : SUCCESS in 48m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d97eba6ee5c34c379971cdc903ef0372 : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07af40788d3a4a68a0c5b206cee0cddd : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/881c8f6a72c84d708840ee0ba00511e8 : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e51d14fd3fb24a5ca268ad96e562da7e : POST_FAILURE in 35m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc75ece38c9a45f9b3eae4000826b2db : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/416cbae5a84f49d982b84a7a623c29d3 : SUCCESS in 18m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/af91ede998f14a28885f3c6deaac62fb : SUCCESS in 16m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fb15932b1ea744d9bba2398bd3c8e07d : SUCCESS in 17m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5ffdccf45b114c68988ff7ea9a63ff57 : SUCCESS in 31m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f998c88cb8f842519cec63c92065ba6d : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a48cb790e6f44d37b2e2afc523a51f01 : SUCCESS in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf4e452b989d4fb5a801d83f72037d07 : SUCCESS in 18m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fbf37cec1417496e87676bbfbcd12220 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cb11c97e60a74017b4bd91a42d97b92c : SUCCESS in 15m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d81c6b16891649ad8af1aba16ad3a865 : SUCCESS in 15m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/80cde1f6739e4d77a08814b03033d01b : SUCCESS in 53m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cef4073f4790405d8f72c02335bd9d06 : SUCCESS in 52m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/876f3f901dba4ea0ac0e94ada3dcd703 : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7fd081f02d4c9280a3e3858401f7b4 : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/467e5ee0400f47e7850e23e1f66e2cd3 : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1f36ab58337cb01c054bd093401969f351e04d91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:56:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c9d65d629094e349ac5591afed17982\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2b77525459304830b82cce9ab72e7ea4 : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6ab4debf98d5dc5129a489a0f4285a6b62a72250","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-28 19:11:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e68a20300900adcb0b286b4a744c91587224cf44","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-28 19:26:25.000000000","message":"Abandoned\n\nsquashed 933372: test-expirer: call tracking \u0026 exception stubs | https://review.opendev.org/c/openstack/swift/+/933372","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39fce6d7d39045e776fb0bd752818ac3a693b445","revisions":{"39fce6d7d39045e776fb0bd752818ac3a693b445":{"kind":"REWORK","_number":1,"created":"2024-10-28 11:37:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/933557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/933557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/933557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/933557/1"}}},"commit":{"parents":[{"commit":"19b2a9340ce27959ad3ae3c24d5a23d199f442d0","subject":"test-expirer: call tracking \u0026 exception stubs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19b2a9340ce27959ad3ae3c24d5a23d199f442d0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-28 11:36:02.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-28 11:36:02.000000000","tz":0},"subject":"sq? fixups for test_expirer.py","message":"sq? fixups for test_expirer.py\n\n- make FakeInternalClient respect acceptable_statuses\n- fix some comments\n\nChange-Id: Ifd33f6ae98a4a96605a4d3dcc028ea42f23e3940\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39fce6d7d39045e776fb0bd752818ac3a693b445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39fce6d7d39045e776fb0bd752818ac3a693b445"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
