)]}'
{"id":"openstack%2Fswift~933605","triplet_id":"openstack%2Fswift~feature%2Fmpu~I1e560a2d39a647ef8dbf3f85e2563894f2f94fea","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-part-copy","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-05 01:51:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e560a2d39a647ef8dbf3f85e2563894f2f94fea","subject":"mpu: modify copy middleware to support uploadPartCopy","status":"MERGED","created":"2024-10-28 18:23:49.000000000","updated":"2024-11-05 01:52:42.000000000","submitted":"2024-11-05 01:51:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933605-p-mpu-part-copy","meta_rev_id":"8f2546fe76abcbc2264d4f2359dc1e3fccf2572b","_number":933605,"virtual_id_number":933605,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-05 01:51:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-04 23:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-04 23:33:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-28 19:13:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-29 15:54:29.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-11-04 23:33:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"ad620333ae1b00da753b18d6ae6c9423779e122f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-28 18:23:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b8d2cb4536c5e9ce91be7a14a41a14b58f453a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 19:13:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cd971ca48934a7dbf9d0207358ed05b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d1ce309d28439ab5addd444a0be616 : SUCCESS in 6m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b681ed69bfc746b39e8f198be54866b0 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7163aa49ab5e462eb7d65f00079472ed : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c09251e81c2446e83d9c900f79bf649 : SUCCESS in 19m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bef98c469e68454bb7218e0d5d46f007 : POST_FAILURE in 33m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0fb828aaced4da58008aa54a5fea2cc : SUCCESS in 17m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb9b0998f21047f183d63816e9cad8d7 : SUCCESS in 16m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/04e20af83115471d9b42f302b9dd934e : SUCCESS in 9m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/66f364450bf3413089f1549d3c1ca34e : SUCCESS in 16m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/73fa42e0538e4c2e985cc35a1b46e592 : SUCCESS in 14m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f055c8b15555478a95c58ac2cd2f6d17 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d21ec1ad120f4241b49e8efd4c884283 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/417a085a3649478490fa0bb3891790b8 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/75a4bc797f5e4e95889c50631337a812 : SUCCESS in 8m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fa65a674ac5447ca9f5cd50a7ccc7653 : SUCCESS in 13m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e4817495f71e466994e79020d5e5e714 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b06e321c399441f88cfd6fb7003bc538 : SUCCESS in 5m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/654096398bbd48b4ac179add5c5207b8 : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/18e5035a0a67441aa60b22e6901d17f8 : SUCCESS in 5m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a96ea12b841c4deb95826356f43339b4 : SUCCESS in 47m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/451d9f2cf20743f5b3e5b3d5d8b51378 : SUCCESS in 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eaaa80db4594d2a96ea374c2a7fa0db : SUCCESS in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb019fcf41d14dadaceb3527aa612d70 : SUCCESS in 18m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7c613daaf6f3b609c95fed5e855c7efb90e730b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 19:20:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b787f6929c20485fb7210d6e5e5f9443\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/873bbb9f998945f0a04feb37ec54af33 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":1},{"id":"07c75ceff26e88da7432da702e013d0ca4aa8e31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-29 18:29:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8910a076ec94dc5297f814971c8fd3ccde8428fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 19:14:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/117c67cebfbf4dd8bce0d02b88ee503d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7341f1b1f14ec1b857736c4ee755e8 : SUCCESS in 5m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0d4126834d3433190286e7857051343 : SUCCESS in 10m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4cac189af4d452ebf7a35ca93825234 : SUCCESS in 10m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71d5323a8c404f54a002da5c7c7e278e : SUCCESS in 18m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/48f4f447ea1c418bbcacd02a3f0ca77a : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7a833d60e57427ab78e40eac7691155 : SUCCESS in 16m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/60bf4ace75ce43f7b4f64ef3e8dcb586 : SUCCESS in 9m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6589e9592d4548519bcd73cdeed83384 : SUCCESS in 9m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5b4e76504e3645528356509295a29b39 : SUCCESS in 10m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/630227fb2b804443b7e60ca8ec1aa3d9 : SUCCESS in 14m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/794422dcd6ff453982119115fff005af : SUCCESS in 10m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0e7e1bc58f4436e9157e6016bcb226c : SUCCESS in 11m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/36e666a26c5d4af09f11e90adb9cc120 : SUCCESS in 17m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/93539cb8f10141a08019f896dc34c38b : SUCCESS in 8m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e60b0bf75f494cc19f88b115daee4024 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6d106f2acd5848f991918b2c770dc3ee : SUCCESS in 8m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7fd42239ef94c9891386bea1ab5d4d3 : SUCCESS in 4m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb3fb4eb5b6740b7a7f53357307ad740 : SUCCESS in 9m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/75c4b3da4a094b3980a6acebf70d6120 : SUCCESS in 4m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c20f94ecc59141de9bcd0f711fb06fce : SUCCESS in 37m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a235556af0514c6eae93e728e0e5f2d6 : SUCCESS in 24m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84bbe9ac3fdd41d5a053a18465a3acd7 : SUCCESS in 9m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3c3bb3e06a784ab385b637303ff89de7 : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b1f6fed467865503a96be7e5c536e53f201ca22a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 21:41:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74d91edf24894a14bf05909ba3a6ed22\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3034408d0be34c988968bd92650d6098 : TIMED_OUT in 3h 00m 54s","accounts_in_message":[],"_revision_number":2},{"id":"21ed7c40a790b20c0f48073249f006fb321f1f3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-30 11:46:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a8455ef970e46eb5eee495ebe674a1d0ac1acc03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-30 12:45:32.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"3b1b9d59bf2f0b15544bf9da938b473aa4dd9a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 14:36:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbd9a95bb9474c0f9335089ceb153917\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5106f07debb64b56998ed5841e01c499 : SUCCESS in 5m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/327a0fbe64964a3a92536e84280368c4 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2728bd6a68d147268b0ed6c9685f53a4 : SUCCESS in 10m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8dc0393fbe4b4ab38f2de2f67ef2fdae : SUCCESS in 19m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6fb9c906e7f041b983143a870a8aa823 : POST_FAILURE in 34m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/93cdde16d7004636ba51567f7c7d6c46 : FAILURE in 19m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd99600af0644d27b6efe3f9eeb5976d : FAILURE in 8m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e1a719c0550d4c5889bd89bbd4514d03 : FAILURE in 19m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fbf970386db940989e4d2df1c7662f85 : FAILURE in 21m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fa9888b8cea44578bbb9e325a6bd83f2 : FAILURE in 29m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d4092a40dbf4739a18f962044059b05 : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8792a3e04dac4007967eec5018933a0d : SUCCESS in 22m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e4edf3b879a42b5a6a76df94190c767 : SUCCESS in 16m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/55f2b86082c34b8db8b1c2f9912e11af : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/912fabac6e284acea70116259b9b9ff6 : SUCCESS in 13m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/56e92e2b576d4ceeb5c59a90e8991127 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9faf441f82c648d8832f9d59a3cb523b : SUCCESS in 8m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03108c188b6f4f3897427ec98d8045a1 : SUCCESS in 13m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/06a36a74a0ef405fae642e578ffc809e : SUCCESS in 6m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d99e1ce3ff944b1958051fae89239c1 : SUCCESS in 35m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c6556c7dbda432596811cf36f3a7452 : FAILURE in 16m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513fbf82d5a84755bd7652fd41e0357a : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0625ffeeb68a44c6ba0367e80d8a84ed : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":4},{"id":"98ffaf7d1990f87031f2b9958a3c71126e2fd239","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 17:24:35.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b88318b6dfbd40aebb76ab6662fdfaff\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a5534d0d3d44f0dbd63b907d7e336a6 : TIMED_OUT in 3h 01m 06s","accounts_in_message":[],"_revision_number":4},{"id":"1bfd378aa585df5fbb8a23b08f3c64c0ef9b06e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-30 18:16:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5585b5d5ce52ca424d49e083dbaa79e81684aefd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 19:52:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d2d1801d634cecbf5a27c18c043705\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6753908ecc0447bb45c020c9ea9f060 : SUCCESS in 8m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54d67eebb6074a95bd511ab08a89820f : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b0e2a3d72eb4a7098f4199449386a09 : SUCCESS in 17m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e473bdd58a749d8af47d7dc8bdb38c4 : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8683b429e84c49ada97bbf347cfc7a65 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7be4362992084abe9d58ae28c24bf790 : SUCCESS in 18m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d0b513a75dc14e9eb7b0f35a97bb3c84 : SUCCESS in 18m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/afa45ad290ec451a9c0db53993dfb62b : SUCCESS in 16m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9efe8469eb2543cfbb77f2f50f134f0c : SUCCESS in 18m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6c0ccc01d60e4bf28f9e889f0a59a282 : SUCCESS in 22m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de9e5e8607c643d2b3b720c371b5d3b5 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e13e3592a9749d8860cd971e01e04d5 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c3c3ebc525ac47cca90f97066420c7ce : SUCCESS in 16m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/733299bafe984a34966d2b7f74d4fd48 : SUCCESS in 13m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/efb38706e42a4f9c881efd8b08e3610f : SUCCESS in 14m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aab4df1b562646c2bf98f582a13828a2 : SUCCESS in 19m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f19431caf62b4a0c910d7ca2163e9510 : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7d3ae6a513334948a36361a0413e011e : FAILURE in 15m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/14ccbf795b2a49f3a38e336fa80e7007 : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d1d4da1deac4e94914658f26d1ba076 : SUCCESS in 57m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff7189044c7d4028b37402bb2679ed10 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07263287887e490fad924b4de778c03f : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/84e13e289b4b4a0c87cb5998ec6ff440 : SUCCESS in 26m 20s","accounts_in_message":[],"_revision_number":5},{"id":"164df118c478c3a058e032569cc8c7799d294a99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 22:45:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba55d3f7a5c4487db1726355554e2d10\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4dc8603ef2484064b5367d1bb30dc5f0 : SUCCESS in 1h 55m 41s","accounts_in_message":[],"_revision_number":5},{"id":"bcc3a2b906b9a0aff68af69704e76def8f06934a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 22:02:19.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f3ebb8f05da243973edbe68fc025b3cef523ced6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 22:53:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc5d8e9aa1114bac80b5ad60c5ab3da3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda9ac01ee294f61873ca0c136824fe5 : SUCCESS in 8m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e022a24968347afb2191d1524ce93ab : SUCCESS in 23m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e74361fcd7d464ba44b5b3e48f9195c : SUCCESS in 22m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8cb6a1a5e8c347eba7dce21c298c62a0 : SUCCESS in 24m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c01215c20d54a6fa98a2798efdce031 : SUCCESS in 4m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ee2dd39f68f4acaa883aaea15fa9d78 : SUCCESS in 26m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/63c6286f58714102a9b85c918883ec06 : SUCCESS in 22m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e41f65b0fbcb4b4ea4a4d55aa3021cf3 : SUCCESS in 11m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b02997395b134dbbad3cf24d6a98c893 : SUCCESS in 31m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1a853d77d5f442f9b6ed768f9928dd17 : SUCCESS in 20m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8773df1f990643b48d45b0373259fd47 : SUCCESS in 24m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af227026262f416890e840efc875293a : SUCCESS in 18m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e74be0790f44450985c64a434232c0e7 : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8d9c2b0054084d83b71fc983fbcc32a0 : SUCCESS in 18m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7713f78096524c7ca65082c3e964cbda : SUCCESS in 22m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/88ff8882a0fd46e6999343c439a3f199 : SUCCESS in 4m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0dd6df74b8cd4e7bbf1cd131f9ae805f : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c7b5976f04694277b74d312c777a75ca : SUCCESS in 6m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/53a73f9c80a640ff85dd71a2f978e9e7 : SUCCESS in 41m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9504b6b186b74c28b721cad28d8f62ad : SUCCESS in 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b706692ca91c4eb8bd9c23f416e065f0 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6666f885abb0444caa7a896690aaa717 : SUCCESS in 14m 56s","accounts_in_message":[],"_revision_number":6},{"id":"e87393f5b8fe7c1d55a9ef1cc5c54740cba76109","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 23:33:41.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c3c1cfa39ae75dc8279d87f7912ead96d79dfb20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 23:34:03.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"67e953c8ec779231e748274ac8220c69e805993c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 23:35:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4325a7a177647ba8063f1bf20766b79\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37035c2d8f844fc59f8be728c28d0a54 : SUCCESS in 1h 30m 58s","accounts_in_message":[],"_revision_number":6},{"id":"079209a38da68f611045aae1038539a63cc06a74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:51:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7533250bfd34b9092e8c27b6aec4412\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bed59fc079cf4550b68967aae0c21456 : SUCCESS in 5m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/70e28f46b88642c98a9b05d39c96b238 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d6fbced111604e7d873eab87122d4bca : SUCCESS in 20m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5483f48378e4da0962b43523dc412e3 : SUCCESS in 12m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/369888f628ab4033bb66dbd5d6373c66 : SUCCESS in 21m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4ac6e4f091c44f2e892e6fbc45ce602c : SUCCESS in 12m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cda1a3cf2d354b2db4a878138e29c56c : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dd8a82988991461ebff1b99bb6baf2e2 : SUCCESS in 21m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5f6d257dee984e43a074aae8753171b4 : SUCCESS in 10m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/99dce90cc39b465691ba679c13b39c95 : SUCCESS in 12m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/201021ce4c33471a934a57528222034a : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/69c54e39da744767950f0c6108c49dc0 : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/32c17884828d489999203eee38a53366 : SUCCESS in 4m 00s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6046cda641f1420584a43cb897350b9c : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed4edaaa652d41d191577025177c18c8 : SUCCESS in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c295eb2ea2b4937828613466514d6ee : SUCCESS in 21m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81a7083cfeee4e3699397da4e0a065ac : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b87d97fbf8dd46ebbbe2a9c237657cb2 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":6},{"id":"7ca5d3b29985f791dbee7d48c0b3e8808361582c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:51:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8f2546fe76abcbc2264d4f2359dc1e3fccf2572b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:52:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0d8a50b0ea14d57be3dd918a9e90a87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2cf062bbed4423dbe738949849127ef : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/659940a150a04f5d88f8d5e7c85af7ed : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c9fc7e71509e8ad92a37ab34360d4ae05a86c3ad","revisions":{"d24d3844a90863afbcc0f58c6d0323b164f05650":{"kind":"REWORK","_number":1,"created":"2024-10-28 18:23:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/933605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/933605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/933605/1"}}},"commit":{"parents":[{"commit":"b9552a1741c2b580e4241f2c5cfd4916c336dc21","subject":"Merge \"mpu: probe tests: add a helper to check mpu resources\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9552a1741c2b580e4241f2c5cfd4916c336dc21"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-28 18:20:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-28 18:20:00.000000000","tz":0},"subject":"mpu: modify copy middleware to support uploadPartCopy","message":"mpu: modify copy middleware to support uploadPartCopy\n\nAn uploadPartCopy operation will have part-number and upload-id\nparameters. These must be removed from a copy source GET request in\nthe copy middleware which precedes mpu middleware. They are only\nrelevant to the copy target PUT request and would cause the mpu to\ntreat the copy source GET request as a listParts operation.\n\nChange-Id: I1e560a2d39a647ef8dbf3f85e2563894f2f94fea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d24d3844a90863afbcc0f58c6d0323b164f05650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d24d3844a90863afbcc0f58c6d0323b164f05650"}]},"branch":"refs/heads/feature/mpu"},"3abe6cc2c6937856dd08cfb084fc6f13f205671f":{"kind":"REWORK","_number":2,"created":"2024-10-29 18:29:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/933605/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/933605/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/933605/2"}}},"commit":{"parents":[{"commit":"b9552a1741c2b580e4241f2c5cfd4916c336dc21","subject":"Merge \"mpu: probe tests: add a helper to check mpu resources\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9552a1741c2b580e4241f2c5cfd4916c336dc21"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-28 18:20:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-29 18:28:30.000000000","tz":0},"subject":"WIP: mpu: modify copy middleware to support uploadPartCopy","message":"WIP: mpu: modify copy middleware to support uploadPartCopy\n\nTODO: explore more the impact on SLO copying\n\nAn uploadPartCopy operation will have part-number and upload-id\nparameters. These must be removed from a copy source GET request in\nthe copy middleware which precedes mpu middleware. They are only\nrelevant to the copy target PUT request and would cause the mpu to\ntreat the copy source GET request as a listParts operation.\n\nChange-Id: I1e560a2d39a647ef8dbf3f85e2563894f2f94fea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3abe6cc2c6937856dd08cfb084fc6f13f205671f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3abe6cc2c6937856dd08cfb084fc6f13f205671f"}]},"branch":"refs/heads/feature/mpu"},"152b2f1cc1ab85ecfebb98a1bd1caac35dc65b68":{"kind":"REWORK","_number":3,"created":"2024-10-30 11:46:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/933605/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/933605/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/933605/3"}}},"commit":{"parents":[{"commit":"6a12bc77c51a846ec6a8fd85bc37a3c5193b4ec1","subject":"slo: add more functional tests for copying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a12bc77c51a846ec6a8fd85bc37a3c5193b4ec1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:07:32.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:45:21.000000000","tz":0},"subject":"mpu: modify copy middleware to support uploadPartCopy","message":"mpu: modify copy middleware to support uploadPartCopy\n\nTODO: fix failing unit test; FakeSwift needs to tolerate parameter\nre-ordering\n\nAn uploadPartCopy operation will have part-number and upload-id\nparameters. These must be removed from a copy source GET request in\nthe copy middleware which precedes mpu middleware. They are only\nrelevant to the copy target PUT request and would cause the mpu to\ntreat the copy source GET request as a listParts operation.\n\nChange-Id: I1e560a2d39a647ef8dbf3f85e2563894f2f94fea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/152b2f1cc1ab85ecfebb98a1bd1caac35dc65b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/152b2f1cc1ab85ecfebb98a1bd1caac35dc65b68"}]},"branch":"refs/heads/feature/mpu"},"9555c8313d3d29e910d7fb824ae3586831e44090":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-30 12:45:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/933605/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/933605/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/933605/4"}}},"commit":{"parents":[{"commit":"f339d98f2e226e70e6afc5e41b3fb8bcbfb9550d","subject":"slo: add more functional tests for copying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f339d98f2e226e70e6afc5e41b3fb8bcbfb9550d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:07:32.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 12:45:00.000000000","tz":0},"subject":"mpu: modify copy middleware to support uploadPartCopy","message":"mpu: modify copy middleware to support uploadPartCopy\n\nTODO: fix failing unit test; FakeSwift needs to tolerate parameter\nre-ordering\n\nAn uploadPartCopy operation will have part-number and upload-id\nparameters. These must be removed from a copy source GET request in\nthe copy middleware which precedes mpu middleware. They are only\nrelevant to the copy target PUT request and would cause the mpu to\ntreat the copy source GET request as a listParts operation.\n\nChange-Id: I1e560a2d39a647ef8dbf3f85e2563894f2f94fea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9555c8313d3d29e910d7fb824ae3586831e44090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9555c8313d3d29e910d7fb824ae3586831e44090"}]},"branch":"refs/heads/feature/mpu"},"4b7f77c24a5191fd98bc45f14eec2768587795e8":{"kind":"REWORK","_number":5,"created":"2024-10-30 18:16:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/933605/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/933605/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/933605/5"}}},"commit":{"parents":[{"commit":"f339d98f2e226e70e6afc5e41b3fb8bcbfb9550d","subject":"slo: add more functional tests for copying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f339d98f2e226e70e6afc5e41b3fb8bcbfb9550d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:07:32.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 18:15:46.000000000","tz":0},"subject":"mpu: modify copy middleware to support uploadPartCopy","message":"mpu: modify copy middleware to support uploadPartCopy\n\nAn uploadPartCopy operation will have part-number and upload-id\nparameters. These must be removed from a copy source GET request in\nthe copy middleware which precedes mpu middleware. They are only\nrelevant to the copy target PUT request and would cause the mpu to\ntreat the copy source GET request as a listParts operation.\n\nChange-Id: I1e560a2d39a647ef8dbf3f85e2563894f2f94fea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b7f77c24a5191fd98bc45f14eec2768587795e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b7f77c24a5191fd98bc45f14eec2768587795e8"}]},"branch":"refs/heads/feature/mpu"},"c9fc7e71509e8ad92a37ab34360d4ae05a86c3ad":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-11-04 22:02:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/933605/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/933605/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/933605/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/933605/6"}}},"commit":{"parents":[{"commit":"73dfe3bacf815fd059ce47edccca50a559c10fe4","subject":"Merge branch \u0027master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73dfe3bacf815fd059ce47edccca50a559c10fe4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:07:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-04 22:01:11.000000000","tz":-480},"subject":"mpu: modify copy middleware to support uploadPartCopy","message":"mpu: modify copy middleware to support uploadPartCopy\n\nAn uploadPartCopy operation will have part-number and upload-id\nparameters. These must be removed from a copy source GET request in\nthe copy middleware which precedes mpu middleware. They are only\nrelevant to the copy target PUT request and would cause the mpu to\ntreat the copy source GET request as a listParts operation.\n\nChange-Id: I1e560a2d39a647ef8dbf3f85e2563894f2f94fea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9fc7e71509e8ad92a37ab34360d4ae05a86c3ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9fc7e71509e8ad92a37ab34360d4ae05a86c3ad"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
