)]}'
{"id":"openstack%2Fswift~933680","triplet_id":"openstack%2Fswift~master~I0e55d8b4c5ed520c0aad8cc952a9fa1683905356","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-31 19:45:25.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I0e55d8b4c5ed520c0aad8cc952a9fa1683905356","subject":"CI: Drag forward more constraints","status":"MERGED","created":"2024-10-29 21:06:36.000000000","updated":"2024-11-01 10:49:41.000000000","submitted":"2024-10-31 22:49:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933680","meta_rev_id":"2cc559c70304af0f3e90372aa0bad7c8ce6b6934","_number":933680,"virtual_id_number":933680,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-31 22:49:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-31 21:21:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-31 21:21:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-29 22:14:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-30 21:20:13.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"ec7387e047d717e9058b2cd237785e72168fbd9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-29 21:06:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c2cc2dcd0d3a0909428d935dee28c07fb66a619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 22:14:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea426d88ad68479da6ae41edf7d3c0b4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/088fd299ef4e4034b40fea84f9ea73f7 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/796690bdf95b443683347039c1ef0bf3 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14d59fc23bca433196dc068ab30ae337 : SUCCESS in 6m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/50ff7244a881469aaf237e778d8ad552 : SUCCESS in 56m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec76e24f0ed846ebbccfec6e98c6a801 : SUCCESS in 49m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1c9bc71572a4cd4acc1f624939bb62f : SUCCESS in 41m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87acd3ead24441f2aea1bfe3ea7fd0ba : SUCCESS in 58m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf71d8ef10ac444aa3ae9ceee98f5fee : FAILURE in 5m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45d01fcc48494846a149c318b7928e50 : FAILURE in 5m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93bb404d4721430580d5c58b70e30707 : FAILURE in 5m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eec7753f2cb849489c9128245a933111 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/841df0c39bc1431e87862936a262a15c : SUCCESS in 11m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ab75bfd4f5342c897e1d36a3dbddb58 : FAILURE in 2m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a589fa8a56ab4effa91dd575e2af688e : FAILURE in 2m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a99697ed9c6846ce99daa75cb8292f9b : SUCCESS in 24m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5a89950bae044a69b177427cb043385e : SUCCESS in 13m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e8d7b6c6635a4e1582415cf99bfd3c1c : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e3c63433b9b5439e96d118b75563dc0d : SUCCESS in 10m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f009c26ccdf84f7c8c59ebcb45c68463 : SUCCESS in 19m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1424b5b89da1439e9bdc2c4414bf102b : FAILURE in 2m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/25c984ec2db34c5a9568729756136c91 : FAILURE in 2m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aa36407ad01d474291f6e37d097c6af6 : FAILURE in 2m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a91b4ce09e2b47328be356d33049ce5c : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e9f06607a324f778850e81c16458b80 : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb318b52926442f59c4b4cd40d852e9e : FAILURE in 3m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/22444561f641450ca151e37dab738097 : SUCCESS in 49m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/da21a0958e8549a0b3e2979ab48f138a : FAILURE in 20m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d460f5e70f84793acf362294a84da13 : FAILURE in 9m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af63f944c6a747b490f0da5fa057b2dd : SUCCESS in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c43df66012a34741b3b79590a7b65e7a : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae69285059ab46ec8bbbc87e223bc94b : SUCCESS in 11m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06d742360cff4ee1989d3ce54c58faeb : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":1},{"id":"066229c935128f2b3c5e28d83df3d9f464e43c27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 23:06:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bb408847bba4d899cfcc2dd7c7434f8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b7134b5c7134937ace29b957f104c88 : RETRY_LIMIT in 19m 30s","accounts_in_message":[],"_revision_number":1},{"id":"207fe1a43872aca9de6492bcf938cea2d4ed3f91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-30 04:39:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2577371843bd8ce2e6b55510528d4ef8ee645fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-30 04:50:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c2e7b96c0eda4670dca10a797a2152a97d3b8456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 05:58:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f9706ea22074d9eb7cc50b9069a2585\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fecb4fa01d684764abcc17f9cbd39780 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97a54791ce3c47a39fc6eca7180ec947 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72a5d575a14648948697d71329c62b4b : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/8a151b609ded4d05bd44f86545f5ee93 : SUCCESS in 59m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d5adb1e883046dca44624cf80501714 : SUCCESS in 54m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6477d329897c487f9daa020d00170454 : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e6ee4cac6e54c4181fe626c6fdf8f5a : SUCCESS in 1h 03m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b9f008000b44183a36f1588eba3028e : FAILURE in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2953e596319849c78a912db8ef121dc4 : FAILURE in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbf4ea186ea346f1bb4121a627b12bd1 : FAILURE in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21fd2fa5c180428c91f1dcfed4173086 : POST_FAILURE in 37m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b19190af7c3e4697adfd116f514ec5e5 : SUCCESS in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0e39255554a7438b92ce1d3c50e8d13d : SUCCESS in 14m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e2cf2e1d12448519199991484f39acb : SUCCESS in 20m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c79dfd06e7f4453da7d10a763028a728 : SUCCESS in 23m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1bd009c3ac43446f8b618760ee5f3e42 : SUCCESS in 30m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/10e6560d43cb4c96aac2129d3e04fbd4 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5c58d237d5e644498dcd1594c767d6c7 : SUCCESS in 13m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40109ce31bb44207be832f331929db4c : SUCCESS in 17m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/539631b075b24088b4afe0870e314c2a : FAILURE in 13m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/70a3a2b9476f4638b8dabee797920e05 : FAILURE in 15m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/67b8dabce5e149eeb1954b873dcf8f1b : FAILURE in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96ab97f1af044b61bcf2cb05d3cd93f6 : SUCCESS in 5m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6293263843b248b6a7b135d06ce353c8 : SUCCESS in 9m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/882a2f8c871c40189fad6f421dd477eb : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c9e1bdc4ba8433fb21c858716d5998e : SUCCESS in 45m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a0f67299702423babd62d8598a60e62 : SUCCESS in 38m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee3f0112f0fb4b7abf11f772962b6bef : SUCCESS in 30m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1296da083f04b858354cd941e6efa77 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9dcadbe74244eacb6354c0f9b71efe7 : SUCCESS in 3m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c122c318e1b47a3a09bdf95c5b2be3b : SUCCESS in 10m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc9e09d990ec44c6885fee6dcb2459a2 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":3},{"id":"7ed86a30caba0162312d2409ed184b10ec386332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 07:57:38.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ba7e0dbe5bc4a82af3167b471f7e4f7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c3724704ba642749d47725b26654b7d : TIMED_OUT in 3h 01m 03s","accounts_in_message":[],"_revision_number":3},{"id":"bfb6f19852d0df336cf2b0a0ea7ddd0ad95875ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-30 16:13:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"149e71c66b670e60360aa02c9317d39d9bf28736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 17:46:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5ddb7f91f5145728ea256340808ff30\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/51196b364a624ee9afc9927f219ee503 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89479ad160f44b959262af147868d0f8 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acce93c695084cbaaf60e7380e6d1959 : SUCCESS in 5m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/6a7f0ebb3244496bbc8dbfb4e280d3f1 : SUCCESS in 55m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a2e807ad220414f892fa953734d04b6 : SUCCESS in 1h 00m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c00726ad56c483e9264f87db131f459 : SUCCESS in 52m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f0614f86f544327b4f844c93fd9b4cb : SUCCESS in 1h 03m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4954722e1c94c2dbb86329b0fa3e260 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5497447315ab478984c62bc68b6d06eb : FAILURE in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b0ecdc165ee49adafd498632598a1ba : SUCCESS in 18m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1b5a8a7f33714604b63a8f9c87844f80 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b58421d898124b9fa972bfac7afa1080 : SUCCESS in 16m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af4966f33a5a462bb0bd5a8429f16195 : SUCCESS in 17m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/91b0a83e97cb45daa7e12a8c5a12956c : SUCCESS in 10m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/684e7fef5b7646cabf64b139072755ae : SUCCESS in 27m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f93a5ff43a5b4c6484b32aa52ffdfc06 : SUCCESS in 27m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/845010f3f6b849b9be11fd4408f780d9 : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b874e528d8b4fd1aaac42b20fb20c44 : SUCCESS in 21m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6dd055c799084eecad5402bb4da66cc7 : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb5a6f5cc9a04e9f966d32ed38423e74 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b7d41822f96a43a786baa3c3df41a466 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d8b948b386034c8792d0aa851e250cc4 : SUCCESS in 10m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/71a2b6ce55f3423d82af52f1e16a9b97 : SUCCESS in 4m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2f2c6fa4e92419491abb0e8ba789e84 : SUCCESS in 14m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5226eef2375a4f15bc26a59586eb8296 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/33506e9e6fd14097a4022611465c0af9 : SUCCESS in 52m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58b83afcb7bb4236b392d5b231c2f2ba : FAILURE in 10m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f20f0ac09caa462f97870f8aa0bdc00d : FAILURE in 26m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bd3059dc5514ec19e02e913a65cb3c4 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae91db0520c46a199d1bd6b9f5a237f : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f284cf0b522645a48908d6b9b1094afa : SUCCESS in 16m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae6c48beb90a4174b0933b16d5539fe3 : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":4},{"id":"6354e19a01af12ec78f4919a68f65f22bbac72a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 21:21:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b7a2705b27a4e25ab895ce37af784a3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/908ec12c468f4e129aee17b885da6baf : SUCCESS in 2h 04m 46s","accounts_in_message":[],"_revision_number":4},{"id":"bff7097d7eaefc0913ff6ce09f31b71856fa24da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 17:49:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0ec6322f417a2811b38d862d81b47883f4b7ae89","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 17:56:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c95f8864c9470b050c7b9f6bcda90f49abec03d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 18:05:33.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"71b57729221416b5ee452b4ba0aa326f9407df94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 19:16:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca0a66497c484e3e83096029bc62dae1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7b27a3b67bde41ed8253c2bb4d5981e2 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8312a2835545df8653c50362060803 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f132a6b83718475e8c4d63a7cd529178 : SUCCESS in 5m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/70cd3dde27c847c98b1097a5415022b3 : SUCCESS in 50m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c640ac4644146099391d5c61c9ba1df : SUCCESS in 48m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e29145d72ecb4e2a9557aa1d0dca720b : SUCCESS in 46m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/162984b689c84d6c932e30b6da398fa8 : SUCCESS in 1h 02m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe3cae39376c408e9efc879fe9f0d785 : SUCCESS in 23m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac1edaf2d46d494b96e27b3c324dfdd9 : SUCCESS in 21m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc6d1006b1a54db4ab69f482170e0741 : SUCCESS in 21m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f374ff7e3e994923bebd2c0cf8aaf68a : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b393af26be24673af7654d60ebc5e1e : SUCCESS in 17m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34da69c2f5684196810ac3d829116824 : SUCCESS in 9m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1bd81b3e5f8a47ceb57092f7fe6bca6c : SUCCESS in 21m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/738800fbbce54b2cae31d6ae4b91e04a : SUCCESS in 17m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/296878bf01d04738961a89ae5bf8c923 : SUCCESS in 34m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/687873b2d74047aab700b15321b2cbb4 : SUCCESS in 10m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cfef2c7632c24265bf74543f39c03e6b : SUCCESS in 16m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/879a1eda20c54806b22d931c2ceffe01 : SUCCESS in 18m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a56db854aeb348b28bdf21378daab6d5 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a2e36a90f9ab411686bebdcb241e1d45 : SUCCESS in 9m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bf37f51ccc044289b07ca5d81a063af4 : SUCCESS in 18m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b4373a1ae44490d9d42672302a707a7 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49384822af9f49b3a5aceecf9327b6be : SUCCESS in 15m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d4c1041a6d44490892054a6b58bbe2de : SUCCESS in 4m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e499f73b65b341c4ad979e6ded673a96 : FAILURE in 50m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/84e1a67c6e904437a585ced3b8517ca1 : SUCCESS in 48m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c4312ac71fd46f19c993d7e38419255 : SUCCESS in 36m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50cf416afc66403f9104e6f156642b34 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900e79c0db304003b8d4c39ac4d7011c : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/346f09bf6eda4523a56b6c8a7f00a68b : SUCCESS in 15m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6cf2f81c24c451780dbac4b920da4d5 : SUCCESS in 38m 47s","accounts_in_message":[],"_revision_number":6},{"id":"1240b42a73ef0075f062a31fd04cb07c51efed96","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 19:45:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"626b6ebf100c20cdabf16a5d8555195b8d4b4707","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 19:45:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d0e2a52055c79ac9c8a0db7ddb94cd0eac943d21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 21:03:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bfcb51d51a84b01958755afe1dfdac8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0a2edfa1366549949f517e457ea48a0b : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29c1e63995c4cf798d519b2e946ef3f : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8f28d5212d347b58e4b00251df3a379 : SUCCESS in 5m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/6c6ce6e4ceae4a3c81df2f16fa050ecb : SUCCESS in 39m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dee7aec0b1fa47c1a388e0a01247186c : SUCCESS in 38m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/64a47302d38248769c62318c8157a183 : SUCCESS in 40m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c38835678d9c496f92f4a0e0b95d9df5 : SUCCESS in 54m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/260bad627be14b7fba3d3d3d6180a7b8 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94deea5c9069496c8e463b3ac83c57e3 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3d9b606ca404c3fb8478d55d1a11cc5 : SUCCESS in 18m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee5acbeb1b184a40b2bd61415d8531d7 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/341c79ca312e4882a6507625a9424842 : SUCCESS in 19m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a868eb18121465d98258441fd56fadd : SUCCESS in 10m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/220c0c3c1aee4779a0f5c8102cc21385 : SUCCESS in 25m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/56c14691637149f398247e63e041218d : SUCCESS in 10m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eddee63fc2344f11868ffd4a07c7bbce : SUCCESS in 21m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0f6d51c35d264be1b791bddb8b7fbdbd : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/440ba4979c124892affbccf2d7064f92 : SUCCESS in 13m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/de80a9434c1547d787fc29d907eb2a6b : SUCCESS in 16m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/af24d8f569204d088ec93cab9947d134 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2f969a7310f14cfea40bc5fd08e7d7e1 : SUCCESS in 8m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b86fad60e7024432b3c45917c8b6ba16 : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d7e40277f6b4cb78f2a05bcb43c4fe5 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0bdbf2371c148178d07d9750e5aa3b1 : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d135872a62b74e7ab874d134aff2475f : SUCCESS in 7m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/27af7bc0aec84cbb9510ae9d64e350a1 : SUCCESS in 42m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9efd6176146e40989f9f3017fe0d47d7 : SUCCESS in 29m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14a0429ba62c47eda94116087bd77a4b : SUCCESS in 29m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca680a5b06f34ff28defd6ac0cec7417 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07aec5636ed7464aa3948ca39b7849ad : SUCCESS in 3m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/058f330fdbb44e0b9978494de5ae210f : SUCCESS in 11m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90520c24cef342d289244d0f2a205e39 : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":6},{"id":"61d2967bdebda4c8c7bcd45d6db974ffffe5915e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-31 21:21:37.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"437e7603907127966405f89fb17e81e381c81d76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 21:22:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0c0d96266c8cb5740bda5d880f3f2a31a4e6bf95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 22:49:39.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d0e864f9dcc4543a46303293ebb1a75\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/73997745de9b4cfea80022c5c9b681ac : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa6613241d04065ad39eeb00f645eca : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb266c44f1f442819e81fa20082d9c0b : SUCCESS in 6m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/44aace06de554a03a772a3d79c5e0614 : SUCCESS in 53m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca9231b304e043ec9b0fe277951e6bd1 : SUCCESS in 55m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e94accf2ac5479e8884c37e2663ddb3 : SUCCESS in 57m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f819dfe5ae7944c79ce55a2b85d5c36e : SUCCESS in 1h 00m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d42b8efd23eb403c95be154207c52690 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a5d3dc54a3046178e7cd4ba0142a5bd : SUCCESS in 15m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/775c35d805f0430ca685a35d9f3b210b : SUCCESS in 15m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/73922ad183d240229889f0b3d7814713 : SUCCESS in 11m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d2e43af1d74547cc86c64a1f317e52e7 : SUCCESS in 16m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c0361e33246d495da23e03f75d241d93 : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0ae3962548f649de85e090de326345a8 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/52d029e435dd467ab11b177466b1e43d : SUCCESS in 10m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3f1e868db7db4c61b98919c850f068b1 : SUCCESS in 15m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6c1a05a8793744e882f5fbd55e655c9d : SUCCESS in 9m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f0e3c5558a274be7953aefc3252bdfc3 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d5a73cea31c04b359a0e73abdba58e57 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3570e4cadc3e475bb0e95e3cc443050e : SUCCESS in 6m 19s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ebe088c17f24553962e2953a0f82b4c : SUCCESS in 6m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b005be656cc04a8d94404113d7d92aeb : SUCCESS in 47m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eea9a30d987a4653bc6dfb35d980340e : SUCCESS in 33m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a45388f4771412e865ba64dc776584c : SUCCESS in 50m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3d4990b5d5344859ccff907e4413599 : SUCCESS in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7b34ed6f4644f9b6f42c16b1451678 : SUCCESS in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fcaf65b8c40445b3baf6d664ee8071af : SUCCESS in 11m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/decc63b207c84b8c9a5fe989ef9b5e1d : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":6},{"id":"7ffd8a4e22d80d56541ea5213a12f38192ea66a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 22:49:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"bfe265b2ca36f281f02f07ba8c343617f0eda981","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 22:50:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c749408b247144a187f13398cf7a3d21\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e32970a1da4a49c78cddd85ba19ef90b : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd488b671e44bb0a573c78bd6b20b31 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c76c13460d94ca3a7fcb59bdcc18a50 : SUCCESS in 35s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5f36bee16a904508aabe386753c71d8b : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f3bf600fdd2d7fb75accb7d5599070e95c67ab99","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 03:56:17.000000000","message":"Patch Set 6:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18338a77f77f4b4691a33701f37902d1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bf0f032ddd14d0da8f30b26b3ba88f7 : TIMED_OUT in 3h 01m 14s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/f58857433ea949dfbcf96dbf43685756 : TIMED_OUT in 3h 01m 25s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/37d1e8979bf640ceb81fa8fbd5c834c6 : TIMED_OUT in 3h 01m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/66f2a2b4b6584237b0dbb0f4a983d5d1 : SUCCESS in 20m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5eeac12051b545df95d2f42b5a46c848 : SUCCESS in 23m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0600d7e76ef041a7842b29694938d161 : SUCCESS in 16m 09s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/269209a375374cc18c128801b26b39a7 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b190af560c6840a9a0808bfcd4b27c50 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f14609c86ff147aaab8bf5a489c3079c : SUCCESS in 18m 13s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/0d97725a803e4f9eb423a612713980d2 : SUCCESS in 16m 25s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/e88d88cb043e424ea57847ab21880706 : SUCCESS in 25m 46s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/9135132933fe445d99541259cef1b7cd : SUCCESS in 16m 46s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/8c5c3f0f880a42f3988972ad70194645 : SUCCESS in 16m 49s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/5871f2cb994d4322b3b8bdd591a45a5d : SUCCESS in 11m 08s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/33eab0fa32874e78bbfc5b218a69ab52 : SUCCESS in 16m 45s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/ceaa4b4007db4b8d9f4d188ab5e77af5 : SUCCESS in 23m 37s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/ee30ae455b8e4681a932aaab72ad09b7 : SUCCESS in 22m 37s","accounts_in_message":[],"_revision_number":6},{"id":"2cc559c70304af0f3e90372aa0bad7c8ce6b6934","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 10:49:41.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af23800d0c9f4cb09a78c6cf6ed645ed\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dd121db81caa4e17a04ef57852ea59c9 : TIMED_OUT in 3h 01m 02s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a8ad97f838141b0d6636b27c46663db99b68dbe0","revisions":{"aca608fbc08bb02b250962609c2253d932683069":{"kind":"REWORK","_number":1,"created":"2024-10-29 21:06:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/933680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/933680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/933680/1"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-29 21:01:25.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-29 21:01:54.000000000","tz":0},"subject":"CI: Drag forward more constraints","message":"CI: Drag forward more constraints\n\nPreviously, we\u0027d install the lowest (mostly common) denominator of deps\nfor our func-py3 jobs. This left us not testing up-to-date deps for\nmany python versions, though, and could even prevent us from testing on\nnew python versions entirely. For example, py312 cannot install\ngreenlet 2.0.2, the last version to support py36; this caused gate\nfailures when OpenStack as a whole tried to run Swift\u0027s func tests on\nUbuntu 24.04.\n\nChange-Id: I0e55d8b4c5ed520c0aad8cc952a9fa1683905356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aca608fbc08bb02b250962609c2253d932683069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aca608fbc08bb02b250962609c2253d932683069"}]},"branch":"refs/heads/master"},"9dd7b092cd4bfa0c872a78c0272e2d56a1ad9168":{"kind":"REWORK","_number":2,"created":"2024-10-30 04:39:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/933680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/933680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/933680/2"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-29 21:01:25.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-30 04:36:58.000000000","tz":0},"subject":"CI: Drag forward more constraints","message":"CI: Drag forward more constraints\n\nPreviously, we\u0027d install the lowest (mostly common) denominator of deps\nfor our func-py3 jobs. This left us not testing up-to-date deps for\nmany python versions, though, and could even prevent us from testing on\nnew python versions entirely. For example, py312 cannot install\ngreenlet 2.0.2, the last version to support py36; this caused gate\nfailures when OpenStack as a whole tried to run Swift\u0027s func tests on\nUbuntu 24.04.\n\nChange-Id: I0e55d8b4c5ed520c0aad8cc952a9fa1683905356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dd7b092cd4bfa0c872a78c0272e2d56a1ad9168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dd7b092cd4bfa0c872a78c0272e2d56a1ad9168"}]},"branch":"refs/heads/master"},"d996e24407cfe76727550e1ee336854184014e9f":{"kind":"REWORK","_number":3,"created":"2024-10-30 04:50:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/933680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/933680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/933680/3"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-29 21:01:25.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-30 04:50:28.000000000","tz":0},"subject":"CI: Drag forward more constraints","message":"CI: Drag forward more constraints\n\nPreviously, we\u0027d install the lowest (mostly common) denominator of deps\nfor our func-py3 jobs. This left us not testing up-to-date deps for\nmany python versions, though, and could even prevent us from testing on\nnew python versions entirely. For example, py312 cannot install\ngreenlet 2.0.2, the last version to support py36; this caused gate\nfailures when OpenStack as a whole tried to run Swift\u0027s func tests on\nUbuntu 24.04.\n\nChange-Id: I0e55d8b4c5ed520c0aad8cc952a9fa1683905356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d996e24407cfe76727550e1ee336854184014e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d996e24407cfe76727550e1ee336854184014e9f"}]},"branch":"refs/heads/master"},"c3280e5337f15987e595c3854821c398f5e363b0":{"kind":"REWORK","_number":4,"created":"2024-10-30 16:13:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/933680/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/933680/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/933680/4"}}},"commit":{"parents":[{"commit":"288677695dbc24730f4f7f114dc2c6c418057cc4","subject":"Merge \"Rename obj to path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288677695dbc24730f4f7f114dc2c6c418057cc4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-29 21:01:25.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-30 16:13:00.000000000","tz":-420},"subject":"CI: Drag forward more constraints","message":"CI: Drag forward more constraints\n\nPreviously, we\u0027d install the lowest (mostly common) denominator of deps\nfor our func-py3 jobs. This left us not testing up-to-date deps for\nmany python versions, though, and could even prevent us from testing on\nnew python versions entirely. For example, py312 cannot install\ngreenlet 2.0.2, the last version to support py36; this caused gate\nfailures when OpenStack as a whole tried to run Swift\u0027s func tests on\nUbuntu 24.04.\n\nChange-Id: I0e55d8b4c5ed520c0aad8cc952a9fa1683905356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3280e5337f15987e595c3854821c398f5e363b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3280e5337f15987e595c3854821c398f5e363b0"}]},"branch":"refs/heads/master"},"b79ee152a95ae89bdc8d3d115739db2ea3dd0f67":{"kind":"REWORK","_number":5,"created":"2024-10-31 17:49:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/933680/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/933680/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/933680/5"}}},"commit":{"parents":[{"commit":"ece954b5f3ff81e9b7ce1cc7a7d51a1e7037ba06","subject":"CI: Move a bunch of func test jobs from py38 to py312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ece954b5f3ff81e9b7ce1cc7a7d51a1e7037ba06"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-29 21:01:25.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-31 17:47:46.000000000","tz":-420},"subject":"CI: Drag forward more constraints","message":"CI: Drag forward more constraints\n\nPreviously, we\u0027d install the lowest (mostly common) denominator of deps\nfor our func-py3 jobs. This left us not testing up-to-date deps for\nmany python versions, though, and could even prevent us from testing on\nnew python versions entirely. For example, py312 cannot install\ngreenlet 2.0.2, the last version to support py36; this caused gate\nfailures when OpenStack as a whole tried to run Swift\u0027s func tests on\nUbuntu 24.04.\n\nChange-Id: I0e55d8b4c5ed520c0aad8cc952a9fa1683905356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b79ee152a95ae89bdc8d3d115739db2ea3dd0f67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b79ee152a95ae89bdc8d3d115739db2ea3dd0f67"}]},"branch":"refs/heads/master"},"a8ad97f838141b0d6636b27c46663db99b68dbe0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-31 18:05:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/933680/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/933680/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/933680/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/933680/6"}}},"commit":{"parents":[{"commit":"c3f1ed3df178ffa09dec02346f3644ea8970c203","subject":"CI: Move a bunch of func test jobs from py38 to py312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3f1ed3df178ffa09dec02346f3644ea8970c203"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-29 21:01:25.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-31 18:05:15.000000000","tz":-420},"subject":"CI: Drag forward more constraints","message":"CI: Drag forward more constraints\n\nPreviously, we\u0027d install the lowest (mostly common) denominator of deps\nfor our func-py3 jobs. This left us not testing up-to-date deps for\nmany python versions, though, and could even prevent us from testing on\nnew python versions entirely. For example, py312 cannot install\ngreenlet 2.0.2, the last version to support py36; this caused gate\nfailures when OpenStack as a whole tried to run Swift\u0027s func tests on\nUbuntu 24.04.\n\nChange-Id: I0e55d8b4c5ed520c0aad8cc952a9fa1683905356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8ad97f838141b0d6636b27c46663db99b68dbe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8ad97f838141b0d6636b27c46663db99b68dbe0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
