)]}'
{"id":"openstack%2Fswift~933729","triplet_id":"openstack%2Fswift~feature%2Fmpu~I712ac34b09ef3b4219fea6340faa2f71d374fe4c","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-copy","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-05 01:54:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I712ac34b09ef3b4219fea6340faa2f71d374fe4c","subject":"mpu: fix copying from an mpu","status":"MERGED","created":"2024-10-30 11:46:28.000000000","updated":"2024-11-05 11:50:49.000000000","submitted":"2024-11-05 01:54:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"933729-p-mpu-copy","meta_rev_id":"9d389b6e68ee41866bfd03e07b7cf7660db4b928","_number":933729,"virtual_id_number":933729,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-05 01:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-04 23:39:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-04 23:39:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-30 12:18:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-04 23:39:35.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"9275c49275c67d1aad0d8d0c7be8cc3ffcd4b716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-30 11:46:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfa05bbb95637fb00640f9e3a9d5ae5457a1feaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 12:18:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c5b39001e0c482bb96e38146ab4c4b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccef6f9eb00e4961beafe6a4e677a183 : SUCCESS in 7m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19c65423bdb042d0bfabdfe8c4ad72f4 : FAILURE in 17m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9575184dcd5e4f679d26e2b2a7dd2c33 : FAILURE in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2e156c2553f4bea9ef5d2668e0e66fe : FAILURE in 18m 04s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d15c5c964a544959aa37a3df40cb4b2 : FAILURE in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f06012e294646f48b49d08f94dd5963 : FAILURE in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/beae07b22dbc4e5c9b6fc92634789582 : FAILURE in 18m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9837945ac6eb4a1094ead46ef36dd965 : FAILURE in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/93e0b4f852dd4db59c662d47cdade925 : FAILURE in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3a1e91b870f743bbb453cb6d754a50f7 : FAILURE in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02b2d3415dc40168b7e51338a04be18 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36c7c538a1ae4160810cedf291f35cbf : SUCCESS in 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9051eef89e7ecd37649da790e2d95b0d562e8171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-30 12:45:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5cd15db35acf60304d8d8dbef1dfe40d0c6310e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 13:09:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1433534aa5342739364eb70fdfded4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d118a1f1d0848a9ad6f22a8ba523128 : SUCCESS in 8m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/486992f7bde14d0d8eeb8991c514306a : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9472abcf2bed455da64ce82df861d205 : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8cdefe42ba8f4a7db4e35d6e7ed8b165 : SUCCESS in 16m 59s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3bf29f2aa49345a08aed8cc67309890c : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/98789708b6b04192b38204f9c2c63346 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e98fa057423e49f58a9d6b5a09897dca : SUCCESS in 16m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/52a4edbdce1c489fb3551e326abf9043 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b028e0cc20e34f6392088e66fba01d62 : SUCCESS in 14m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e2a310b6d9b241a0b2d366d283f6ce7c : SUCCESS in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d66dfc8a7d24bdcb10e440772cbc4b5 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5ba9aecd1954f4c8fdc78ac6896002d : SUCCESS in 18m 04s","accounts_in_message":[],"_revision_number":2},{"id":"02e4f5b5fb7aa2e25b09d41508d573d89cf43edd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 17:07:45.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5b1533d77a46ed876bb72db6b8f50e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6bc59329d2b3421ea724b14af2a49641 : FAILURE in 2h 09m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0749298f409c0003db261c7f77db63a640d24f91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-30 17:30:35.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3a1bcdc9bf8adf7f2432772ecfbf8d9f4937d433","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-30 17:30:47.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"3454b0b803728a0db254bab366688496d6591569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 18:24:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65392f624e7b4957b674171a59dddc7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd1a7fd7ffd4682ad89b5934168c74a : SUCCESS in 8m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6dfc163dfdc04617bb2c6c0c98db3cb5 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/368be6a0d01045e6994fa652fb0a5d8c : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38137fdeddba4357b9f7ab14df2c506f : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c981539afa041579abeffd75a6f946f : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/03341106716f470c83c8f750ee87d644 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/52eae857e55a44298aab68e7af003d82 : SUCCESS in 18m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/73f2a549b4294359bc3f2fbf2c9bbd6e : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4c2c0cfba13240dfb7998e43f5f5fa52 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/491f557ec4464be3b2899a14e504971d : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa2bc05a8e44a63b162becb982a33c2 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/591b00a50a4c4ce59550167da8a4f90b : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":3},{"id":"4562f2c081008d25baa684a6d0801571ceb5e1df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-30 18:30:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9df095df0bd6ed300c6f196e0ec9deed9ed0636a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 19:43:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb2a8d9613a4a32a55714d5e9316a52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68b235c670a46fda5bc92dd83db1130 : SUCCESS in 8m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbbabd1489d94c9da8b824628c2eb983 : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf6809c30c364e1e9bb3f31b32ae394a : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adb3149534cf4523bcb987278edbb5f0 : SUCCESS in 17m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba93ca3774d14f35bb53fbda181e20de : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6251bb6fabce46c19346182932a1b058 : SUCCESS in 10m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86d25b8c8b454c4786382fa2b2377f63 : SUCCESS in 9m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b9f0b32b18e84b829b41c6842a0cc506 : SUCCESS in 10m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0d01c50697d7430f8d85fd8b41448457 : SUCCESS in 11m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e191e4377a0b4c92881d6f2266cc007c : SUCCESS in 14m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b072aca7f5d242f191645fa7dac6b3a0 : SUCCESS in 11m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0074a30f41394d21a2d8892fe0b722bd : SUCCESS in 19m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ea796e90c42425aac2954376e874dfb : SUCCESS in 21m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e5c1c50d78c84301b50d8a39292a61a7 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1d314f05bdd64523818c938d58270f7b : SUCCESS in 14m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fb04fca51b9d43e192f9500c1286f915 : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab327f3cd08047ce8285cb6c1786b355 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/23b7011f3cdc44028a90fb7d3925057e : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/58e35c322ae042948e7eebc1aa6819ea : SUCCESS in 6m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48489a45e30e45d3b549abe536e33afe : SUCCESS in 1h 00m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd708e78237949bca501ab3afe8f2c42 : SUCCESS in 20m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32250f3965ec4d9ab61ea80de674e9c2 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d4fb507b0164bfcadb11ecae38c57b9 : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":4},{"id":"136f380ee30132836b5a180c4682939f037d608d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 23:57:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b89b4802534cbea5982ec0282d93f6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80651b8f1f994bc9aa5808a9b0823a62 : SUCCESS in 2h 32m 00s","accounts_in_message":[],"_revision_number":4},{"id":"e3f6e469362cf369d902a5114605b8e72ebf1cf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-31 15:38:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0e56907a69e43090ca153243787c9bd5dde3e7f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 16:24:10.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44d6ad3f61314b88830cd98a25d0c4d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9e525748ee541a1afd26a21535cfc56 : SUCCESS in 4m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a8cac3ed2f14f8ab5954ca0e4620ec8 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8194f792132b4226a05706ca69501892 : SUCCESS in 16m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c10b42b3786445b08bef9dcb595969a2 : SUCCESS in 15m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ecf42d1d80e547a59f69b0cba060eb3e : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b861739116e24daa8dadfaf0c8928f2a : SUCCESS in 21m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19b1cce979df46ae9b1f0c960093fd8c : SUCCESS in 18m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b3fc6cd606254744b1e015c1dcc6d6a1 : SUCCESS in 15m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dc7ef790ecdd481e936a3efe71e1798d : SUCCESS in 27m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b85d2e58c3014bc09f1cca01920b758f : SUCCESS in 22m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/01a327cedda248f4bbf237788a66e283 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff3b7b35f6734ea3ab3a36e909d05d2a : SUCCESS in 15m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a8462f8c82045f49a720060f4127476 : SUCCESS in 16m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb2cf8f3c5a0431daf4d58be81da2e56 : SUCCESS in 7m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f266808a2c3546ce8dda485cfc605f8a : SUCCESS in 14m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fda65b83c7084fb7a3dbad30b2fca6e7 : SUCCESS in 19m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/48d203faaef3448b9080372c6c53d840 : SUCCESS in 9m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e37e1cc9635b46c093ef6f7cd8ac5f24 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a6eaf8ecd3124ba98d757ecef559bf84 : SUCCESS in 4m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/13af38ed5de141bc8debb4e839617ac0 : SUCCESS in 36m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a911e4072ff40f5aed82399f525a1d7 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce13b71a7f8455c8fc493555467ba53 : FAILURE in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5425f72502ff4b60b60e81dd0bb7b639 : SUCCESS in 25m 01s","accounts_in_message":[],"_revision_number":5},{"id":"46cd3326acd473eb5b99a066356e293b3909e965","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 10:04:19.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa633f40a3994cf586c72ee6cad710ad\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/78f05308798a4ea2a84d6dafea7fb31c : TIMED_OUT in 3h 01m 03s","accounts_in_message":[],"_revision_number":5},{"id":"61d95b7f9fa196410c85242a781c1a75f9b24b08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-01 18:25:00.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"edd25636810e7d89c0e6c158ffc51de00e08cdfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 19:13:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f18c4c974d2e401d82f8caa2fd24b4d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf2cad6014140c58e8ef8e8f4ea7c2f : SUCCESS in 6m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af298efded534e26b24c76f1f4c8f268 : SUCCESS in 11m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6ccb0c1e2eb4eff97e12467dd5db7ee : SUCCESS in 18m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4620e668aa645dc8eadbbc207d551ee : SUCCESS in 15m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c442ef8b32764aa3afb2d00bee04d05c : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d9b819c738241609bd26f9c4a8ab31f : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/46ff2dcc65c64259b2051f3c6501d410 : SUCCESS in 9m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a6316b88a882491ab072406193b3cfab : SUCCESS in 9m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/45515049cc164dec977c267b478bb2d3 : SUCCESS in 14m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2b5eaafe31a24ffc94d8191013eeeada : SUCCESS in 32m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7354c05895184f8fac688d5cf6c325c8 : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b4bd37063184030b41a9785b8f24259 : SUCCESS in 12m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1899187e23e48e2ac98c58adde733da : SUCCESS in 18m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3637a34a638c4e13b46a5e92ee05aefb : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9aaf48cd154b4624876171765defa058 : SUCCESS in 15m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4a74df8bb4414cafb839d69c1e145313 : SUCCESS in 9m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a46478e1d270404683130d3eae5ff152 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/43ade825e21c4f7697c64da41a58ee25 : SUCCESS in 10m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ba0c2a7bcbd74307a698492cba141b26 : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/abac68189d9147bf9e32109ccac97b00 : SUCCESS in 47m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e68bd69ab8994164b5b2dfa39aa912bc : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d3b085368054cc597c629effa53a721 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/537806de370044008eb16b1c1e1f86e5 : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":6},{"id":"a01de6c1a312ecf96b162dee184a123036c31cca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 20:30:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c1dbfe99bec40d2a668f7959c708e90\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb67bc8302b9485cbd65c86244190a2c : SUCCESS in 2h 02m 26s","accounts_in_message":[],"_revision_number":6},{"id":"6bf59bdd22a5ee421931c4f598f3021fda5cc1c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 22:02:19.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"50e690b2c402e866daf00b9a8dc678ecb32d9a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 22:38:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf7b334bda8140b5bba0a3d048bf9767\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de08cce3b44145cabe324e6667adfbf4 : SUCCESS in 9m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/153f0a6ce7c84ca5b12dc26ce7ff34d5 : SUCCESS in 12m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b1e5bdeedf4472d9bfb3e73bc7e105c : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c195fbcd3574d469c1466d1b45f7d1e : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd0ac191e9554aaaa9fa2125bb4e9897 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d48c1235e61147bbac5609bc03570b2a : SUCCESS in 10m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ec3db74670e47598956a9eb6eb6a38b : SUCCESS in 19m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7f76629d865e4bbfb17e52103167fba5 : SUCCESS in 23m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cb7159ac127140df96a4e36ef6872441 : SUCCESS in 30m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23ccfb3207984aeebb33e1ff54ef09ff : SUCCESS in 24m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ebe80b937ba4300a9a651176fc13f5a : SUCCESS in 18m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ee1583e86fe4073853b4608b3908f6d : SUCCESS in 18m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1131cd84b0b34417a61f94b3e537419b : SUCCESS in 9m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/06baa84805b5450fb4d9edab351c9e22 : SUCCESS in 19m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8b25257970414d0794e4f7e08e3e40ff : SUCCESS in 18m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d195994ca35c40f7b0ae5bdf1bb98cc1 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a1602b1d77c74c4cac2dff657fcf11a8 : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a1dceb89d4d4acd8a3feb8dfb1010b9 : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e61828b7a69440a3a1aad0a38a1ee9a1 : SUCCESS in 34m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ccf34ee1ac8436e8a6dffe265c8ad27 : SUCCESS in 23m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a373e1e48c471bad382ee3a6a5afd6 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb677cabec1d42a38e229777b5706b37 : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":7},{"id":"4150685fdcf1e5f766da7ce75fb1fa89ed4af950","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 23:23:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4786199cd15495cbb85c232248bef7b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/799e6b01e5ed400a9075ff60495f180c : SUCCESS in 1h 19m 13s","accounts_in_message":[],"_revision_number":7},{"id":"a9b47e8e56733251d2cc24ed43de4b14547dd8a8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 23:39:35.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ab2c7331f21351a5f31c483c7a4279b52a36f46a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 23:40:01.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1234e79fdfebb2727897cdbb9c54a832d34c0fda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:54:35.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77b114d887bb4352a852543768507512\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64dd3d03758a427289dd756f01ebd96e : SUCCESS in 7m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2a71f40d57794dd4ba1312c1ead4eebc : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5520c8dd19394f17b50ee203a64b063c : SUCCESS in 13m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/093bfb7cd5584c69b52c87726f39c88f : SUCCESS in 18m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d65d58c2a1dc41a3a98d3ac30b5c459b : SUCCESS in 14m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3780049331bc4d64b98366eaf3ba9ff6 : SUCCESS in 22m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9a1fe9e36e14768b41d99250043d48a : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/abb277d81b5f4fc380db7a0227c39696 : SUCCESS in 10m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bc869bead7d746bca4b1ef21b65c1256 : SUCCESS in 18m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/463ba32c293c4616a897b9e73b18901a : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1138f654d0ca4e809db83b8396dcd6a7 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/45fd042fe76040a5bd286101cded27f6 : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de574bb6f7d748a8ab8d972af892a6b6 : SUCCESS in 5m 35s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/558a3c2c4e424d4f93b6eb98893d6c57 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a4f2a7727804d2abf3bdc0cde9d0ee2 : SUCCESS in 43m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cde9f40e45f6447cbda448c1efa28f8b : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cbaeb965e514ae79b40fa0b61bff51c : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ceff88f3a5384f64b4b474f625be5d8e : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":7},{"id":"6b7beb093ff9143295c660a0fd3402d3add0f08b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:54:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"97e6e8feee6e9d51c72577d494b6c435bac9622d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:55:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23c41c93764a4d25a3951e88e5595d1e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/983474ceba744bb68bbf8dbf48ecab2a : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b71dc66f93f54f3684e566cdb73be0a4 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9d389b6e68ee41866bfd03e07b7cf7660db4b928","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-05 11:50:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b65674b1c97b75f64a6b38503d64abaa95674795","revisions":{"728067fcf6cdc06cd72d1e70c741278186650cf3":{"kind":"REWORK","_number":1,"created":"2024-10-30 11:46:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/933729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/933729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/933729/1"}}},"commit":{"parents":[{"commit":"152b2f1cc1ab85ecfebb98a1bd1caac35dc65b68","subject":"mpu: modify copy middleware to support uploadPartCopy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/152b2f1cc1ab85ecfebb98a1bd1caac35dc65b68"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:17:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:46:09.000000000","tz":0},"subject":"mpu: func test for copying from an mpu (fails/skipped)","message":"mpu: func test for copying from an mpu (fails/skipped)\n\nChange-Id: I712ac34b09ef3b4219fea6340faa2f71d374fe4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/728067fcf6cdc06cd72d1e70c741278186650cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/728067fcf6cdc06cd72d1e70c741278186650cf3"}]},"branch":"refs/heads/feature/mpu"},"ad19c474f38a50db4ff5bdbffc094df74401ccaa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-30 12:45:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/933729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/933729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/933729/2"}}},"commit":{"parents":[{"commit":"9555c8313d3d29e910d7fb824ae3586831e44090","subject":"mpu: modify copy middleware to support uploadPartCopy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9555c8313d3d29e910d7fb824ae3586831e44090"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:17:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 12:45:00.000000000","tz":0},"subject":"mpu: func test for copying from an mpu (fails/skipped)","message":"mpu: func test for copying from an mpu (fails/skipped)\n\nChange-Id: I712ac34b09ef3b4219fea6340faa2f71d374fe4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad19c474f38a50db4ff5bdbffc094df74401ccaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad19c474f38a50db4ff5bdbffc094df74401ccaa"}]},"branch":"refs/heads/feature/mpu"},"e5675b77cf1ed8788b8fa47348cf04ba7adcd624":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-30 17:30:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/933729/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/933729/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/933729/3"}}},"commit":{"parents":[{"commit":"55a40a816108ef35ea2115d9fc5b1799a600254b","subject":"mpu-auditor: Add config class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55a40a816108ef35ea2115d9fc5b1799a600254b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:17:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 17:30:25.000000000","tz":0},"subject":"mpu: func test for copying from an mpu (fails/skipped)","message":"mpu: func test for copying from an mpu (fails/skipped)\n\nChange-Id: I712ac34b09ef3b4219fea6340faa2f71d374fe4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5675b77cf1ed8788b8fa47348cf04ba7adcd624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5675b77cf1ed8788b8fa47348cf04ba7adcd624"}]},"branch":"refs/heads/feature/mpu"},"b41839768be686117d75b90044e627396d82db9d":{"kind":"REWORK","_number":4,"created":"2024-10-30 18:30:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/933729/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/933729/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/933729/4"}}},"commit":{"parents":[{"commit":"55a40a816108ef35ea2115d9fc5b1799a600254b","subject":"mpu-auditor: Add config class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55a40a816108ef35ea2115d9fc5b1799a600254b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:17:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 18:28:24.000000000","tz":0},"subject":"mpu: fix copying from an mpu","message":"mpu: fix copying from an mpu\n\nWhen copying from an mpu the GET response etag should not be copied to\nthe PUT request (same as for copying from an SLO).\n\nChange-Id: I712ac34b09ef3b4219fea6340faa2f71d374fe4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b41839768be686117d75b90044e627396d82db9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b41839768be686117d75b90044e627396d82db9d"}]},"branch":"refs/heads/feature/mpu"},"3fc032536c2c583f592dead9fb3a10d2b94aa77e":{"kind":"REWORK","_number":5,"created":"2024-10-31 15:38:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/933729/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/933729/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/933729/5"}}},"commit":{"parents":[{"commit":"55a40a816108ef35ea2115d9fc5b1799a600254b","subject":"mpu-auditor: Add config class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55a40a816108ef35ea2115d9fc5b1799a600254b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:17:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-31 13:28:01.000000000","tz":0},"subject":"mpu: fix copying from an mpu","message":"mpu: fix copying from an mpu\n\nWhen copying from an mpu the GET response etag should not be copied to\nthe PUT request (same as for copying from an SLO).\n\nChange-Id: I712ac34b09ef3b4219fea6340faa2f71d374fe4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fc032536c2c583f592dead9fb3a10d2b94aa77e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fc032536c2c583f592dead9fb3a10d2b94aa77e"}]},"branch":"refs/heads/feature/mpu"},"4e5ed823d259bd0ca223ef8e89db5585177b3aa8":{"kind":"REWORK","_number":6,"created":"2024-11-01 18:25:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/933729/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/933729/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/933729/6"}}},"commit":{"parents":[{"commit":"4b7f77c24a5191fd98bc45f14eec2768587795e8","subject":"mpu: modify copy middleware to support uploadPartCopy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b7f77c24a5191fd98bc45f14eec2768587795e8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:17:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 12:20:10.000000000","tz":0},"subject":"mpu: fix copying from an mpu","message":"mpu: fix copying from an mpu\n\nWhen copying from an mpu the GET response etag should not be copied to\nthe PUT request (same as for copying from an SLO).\n\nChange-Id: I712ac34b09ef3b4219fea6340faa2f71d374fe4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e5ed823d259bd0ca223ef8e89db5585177b3aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e5ed823d259bd0ca223ef8e89db5585177b3aa8"}]},"branch":"refs/heads/feature/mpu"},"b65674b1c97b75f64a6b38503d64abaa95674795":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-11-04 22:02:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/29/933729/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/933729/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/933729/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/933729/7"}}},"commit":{"parents":[{"commit":"c9fc7e71509e8ad92a37ab34360d4ae05a86c3ad","subject":"mpu: modify copy middleware to support uploadPartCopy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9fc7e71509e8ad92a37ab34360d4ae05a86c3ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:17:08.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-04 22:01:11.000000000","tz":-480},"subject":"mpu: fix copying from an mpu","message":"mpu: fix copying from an mpu\n\nWhen copying from an mpu the GET response etag should not be copied to\nthe PUT request (same as for copying from an SLO).\n\nChange-Id: I712ac34b09ef3b4219fea6340faa2f71d374fe4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b65674b1c97b75f64a6b38503d64abaa95674795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b65674b1c97b75f64a6b38503d64abaa95674795"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
