)]}'
{"id":"openstack%2Fswift~933931","triplet_id":"openstack%2Fswift~feature%2Fmpu~I9d8939788bb843b072c3565df298c72acb2596d2","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-copy","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-05 01:59:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d8939788bb843b072c3565df298c72acb2596d2","subject":"mpu: trivial renaming for clarity","status":"MERGED","created":"2024-11-01 18:25:00.000000000","updated":"2024-11-05 01:59:56.000000000","submitted":"2024-11-05 01:59:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933931-p-mpu-copy","meta_rev_id":"d05ac52282a4ff340648866c9d1597865c2ebb60","_number":933931,"virtual_id_number":933931,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-05 01:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-04 23:45:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-04 23:45:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 19:32:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-04 23:45:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"bfe58f30bba203fcc708d47094fb0b52446aa8e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-01 18:25:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95255bd954be046d61435131916ec798a4f3932c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 19:32:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a02443d0a9e4b5abf148a1f22ac47b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10550a0c2d164a998aca3f301433408e : SUCCESS in 8m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f945b38c405444709d80e0c115ea757e : SUCCESS in 14m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1d1b7028411462d8e423e5a44adce95 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cf2c13318be470198b9126feb1b22bf : SUCCESS in 15m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7993fa1af6ec47cdb1f69126707ee4ff : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/09b5505ebf124050b257224b2af0c1ae : SUCCESS in 10m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d129895f1c0549e6a54ffb39dc3fc942 : SUCCESS in 12m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ba8a8698646040d99d11be18b1fd00de : SUCCESS in 19m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6e5f4566b0a241968566ae6d87852b21 : SUCCESS in 11m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/904eccd2e4ec45029def37ff1688d3b3 : SUCCESS in 14m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8d9855787eea441ea7f4ff159eae0975 : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d770230631fe4362830dcf58719a7114 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/91883845384b4b06803bf176243c9000 : SUCCESS in 11m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c14d609a21e442cfafc2cb63f7332308 : SUCCESS in 8m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cd13212ef415437283f4b557dc4dd839 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e7e61556b5604ac3b14644fdf23065d8 : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a5cb8175c534dd7b1603bbba693739e : SUCCESS in 5m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e53d0185865541bca1301ed0c8744726 : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fc81878ddd9b4ce4bf39dd9e3fc73ac1 : SUCCESS in 6m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e156281505c744d28e01d62864998cbb : SUCCESS in 36m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bf528095bee46bc8d64de5e85bd6854 : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40806699ef54b4bb8abb65fffe35429 : SUCCESS in 3m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3526d18d7044469583bc96fd24f8077e : SUCCESS in 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1b981535ed04d0d3a627cb22b88288a3d75d11ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 20:41:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6597fd60334243549661689adf363335\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/463b1592f13046acb2ce62602e569ce8 : SUCCESS in 2h 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a79704f7720a437acfe2f7b7eef7bf083c1f7c0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 22:02:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"846b7e4272d2cce09276de83f6d0ee4b0e74102e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 22:52:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d421bd344cd480da387b236c467a239\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79e5a9fce7d415ba2c03d62f346921e : SUCCESS in 7m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1fddf85b4fa346bcad64f3dfda2ddedd : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7db18fbdb3b44a88b1dd44cab53ac6e : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d690507fb7d419491f409b643737315 : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a23a2dc59b5742648e4ffa72483abec7 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/88cb40e8bf6540b690ef2797d713480b : SUCCESS in 16m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24fe708f12ff4e519b346e153cfd3bd4 : SUCCESS in 12m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f0240dbfec2143da9572fe56260e8327 : SUCCESS in 15m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f7fc90f10ddd4900aebca8f739466e9f : SUCCESS in 18m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9452f3c032cc48e7b57c94a7b0a3139d : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce82aac3877e4e5ea50a974540f876b5 : SUCCESS in 12m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/64f5b7cf2735484385406294c3ad72e2 : SUCCESS in 13m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f8d2e9205f704b999bace835f9f25ed6 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/49c2cd02100540b0af48d655aed4ea06 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/365762b722e348bf9cba8901cbb3caae : SUCCESS in 13m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0eb0d50b38a946e5946b005342927e34 : SUCCESS in 7m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/934ae6db8d8b43ae8d9501eed7adf922 : SUCCESS in 12m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/df3f63b5aa2b490ea2c210fc570235e3 : SUCCESS in 9m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddc4a9639427443ca1bd2fdf14807738 : SUCCESS in 41m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d6c4e3dffea4ba6b029cceb2fcea106 : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21fc6f25dd464ece81b5599d776fdbef : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13fc28fe2e6a47cfb8845815a7598643 : SUCCESS in 24m 35s","accounts_in_message":[],"_revision_number":2},{"id":"4441ff41607ff67c40b7f921102f21064f99f7ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 23:38:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d0ff83b45df4f16bfd1192e1c472ff2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee7c3fb9c1314e56ac27aea7448a2e73 : SUCCESS in 1h 33m 47s","accounts_in_message":[],"_revision_number":2},{"id":"0a4b380f5846869d0cc51f725c3992b48892ed4e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 23:45:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ebf3ae8d6d8f0649ece44e2caeb293e31d7499fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 23:45:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f89c9ab1ad30f521b401becab72b8d28955d6736","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:59:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe34d564a61f4531bbd16af753947b82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136be2d8c99c40b0a25e129ad41a35a9 : SUCCESS in 10m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/048bd09a6e6d49ee97d9684b08814166 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/132fec7cdc324f398ddc8f638b93d8d8 : SUCCESS in 12m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7289ff0aac88451cb469cd4bc344d673 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3c1cc87599664f678eef65806a0e1455 : SUCCESS in 11m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0b282702a4fc4431945faa2585bafb6a : SUCCESS in 20m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ece47e93215f4d0ca19eeba0f5110ba7 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/671e59f073eb49f59332ee1421adc1ee : SUCCESS in 17m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c9cf07b125c244b19b91cbcdb6ab7cd1 : SUCCESS in 13m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/63d20634d3614c1ca57e8f0fd2bcac7e : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/69a9d96000d04d3c9cb233f90bd30233 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b624444bb93d45769606dfb71720c518 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e0b50d8b0d44f6fb69f245eea3f53de : SUCCESS in 5m 50s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38ecfbfb65224fd9b2f8f89173dedff2 : SUCCESS in 5m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/83775ac2ad8146ba8a7877e779792f96 : SUCCESS in 48m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02b8e2d511584da387b74ab9f0aa7efa : SUCCESS in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c96554163ef4c74b6d7ca7483cfba2a : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f95e6820a224bc98c929afa198d7dfd : SUCCESS in 19m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2ba91193e52fef7e1b21f874d983483e5e417976","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:59:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d05ac52282a4ff340648866c9d1597865c2ebb60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:59:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a6878ff381847088bde0257a82ff58d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4a7663142547b8873d550d25d69ea3 : SUCCESS in 37s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/806179b105a74588b2902baff11494b0 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fa360b98f7e18f8168d5286ee87f01851ca30067","revisions":{"73c806bdda119423412f06c803e658a50a564053":{"kind":"REWORK","_number":1,"created":"2024-11-01 18:25:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/933931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/933931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/933931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/933931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/933931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/933931/1"}}},"commit":{"parents":[{"commit":"4e5ed823d259bd0ca223ef8e89db5585177b3aa8","subject":"mpu: fix copying from an mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e5ed823d259bd0ca223ef8e89db5585177b3aa8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 17:00:42.000000000","tz":0},"subject":"mpu: trivial renaming for clarity","message":"mpu: trivial renaming for clarity\n\nChange-Id: I9d8939788bb843b072c3565df298c72acb2596d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73c806bdda119423412f06c803e658a50a564053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73c806bdda119423412f06c803e658a50a564053"}]},"branch":"refs/heads/feature/mpu"},"fa360b98f7e18f8168d5286ee87f01851ca30067":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-04 22:02:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/933931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/933931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/933931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/933931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/933931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/933931/2"}}},"commit":{"parents":[{"commit":"b65674b1c97b75f64a6b38503d64abaa95674795","subject":"mpu: fix copying from an mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b65674b1c97b75f64a6b38503d64abaa95674795"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:52:18.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-04 22:01:11.000000000","tz":-480},"subject":"mpu: trivial renaming for clarity","message":"mpu: trivial renaming for clarity\n\nChange-Id: I9d8939788bb843b072c3565df298c72acb2596d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa360b98f7e18f8168d5286ee87f01851ca30067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa360b98f7e18f8168d5286ee87f01851ca30067"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
