)]}'
{"id":"openstack%2Fswift~933933","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ided25598694badc6e9ddf2887adcf5263ec263ea","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-copy","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-06 10:29:57.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-05 17:42:01.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ided25598694badc6e9ddf2887adcf5263ec263ea","subject":"mpu: fix param passing to sub requests","status":"MERGED","created":"2024-11-01 18:25:00.000000000","updated":"2024-11-06 12:32:29.000000000","submitted":"2024-11-06 12:31:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933933-p-mpu-copy","meta_rev_id":"751e21cf27de997fade1ce3e6f42fa3de86872fe","_number":933933,"virtual_id_number":933933,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-06 12:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-06 10:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-05 17:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-06 10:29:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-05 17:42:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 19:34:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-05 17:42:01.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"fe18b02c257f4ba5f59653b91c15558adf4a05d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-01 18:25:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6f9a242f05d4bba0b5be75c5f4deea168e116d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 19:34:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fbace5277934548be259d030815fda3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f679e55dc3404e57880b5f75b6d2f681 : SUCCESS in 6m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3581d15615c1410d917241156ac69cb1 : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1a42fefe00d47c4be0551fe23d61666 : SUCCESS in 18m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c1a51cdf5f748a98f1020b6c34db7ee : SUCCESS in 16m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2cc981c483d547e0b3b447e7e3ccc0ec : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6186d596bdcf4a288204890c98be3d7c : SUCCESS in 15m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0826ca4520342c79c86fd499f260b24 : SUCCESS in 10m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5a2cfa1d3a624f318ee2fc60e8c3a75f : SUCCESS in 18m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dc593aa250114a429ce85f11fa510f22 : SUCCESS in 18m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/443261dfed994396b1961cd3658fc9ad : SUCCESS in 31m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/426bc71eacbe45389a1ce401c3820216 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f1f5bf2a2d4442cad68de8d8a2a6d72 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e2a1ac4dfeb4e0f8d1f65e0f1768c46 : SUCCESS in 9m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/efda253a274e4fca9bcf1a935aabc1a9 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/57efd4c2b27a46868db5fe9997ef3943 : SUCCESS in 9m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/647c58b2930e4a84a017d079468359b3 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/294dd0e201f941e5a155b64d468f7216 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c5a76978c48d4448949834883b6f5b3c : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/934399d2b9164ec4a4b484b0a7398586 : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf4ee46ef6b6424d8d7920d8c540c990 : SUCCESS in 50m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3162a490920412dbe89447658279acc : SUCCESS in 17m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1304e74a050b4336adedf279cfd2136d : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/84a422a575994c888f06be5d12ccc2b0 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4a237e06f1b33c0913011716c4804c16b8d60b66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 20:55:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4ac7fad92c49698558e92de8919df9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de9785e0dcda46dda6bbc802e659fd6e : SUCCESS in 2h 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cdde7f0ddab9c407ab13990b53cd55d964086cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 22:02:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"782e41b775fe6b4fcfe431934c1eb3594e59441b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 22:56:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a34df1ca6a044368be24fb03b8b6c742\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de55c650277b4b678645f32e256330e0 : SUCCESS in 7m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14c1d088495c43cea7c5de9bbd05323b : SUCCESS in 23m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49cf12ce1ba44cf88bf22a75ec0a99e7 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2119b28761fa4d78adcc2d0d808abbb7 : SUCCESS in 14m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f6ccb9319274cc4aa02c62544d17525 : POST_FAILURE in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cd08e70e62ac4d9eab16e97fcca897a4 : SUCCESS in 23m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cde0d9ed79be4e468eda0f22bee31620 : SUCCESS in 18m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce78e0fa9171437eba5118f66690e98d : SUCCESS in 12m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/51c5ae71e8c54193be3685ccdb9c167f : SUCCESS in 21m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/041bfb788a704df380fc109ac93b02a9 : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/38f65152a7054eeb967f42081f58aabf : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3d4ad424fd4d4edf93fee3b35005417e : SUCCESS in 24m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3ca6c8423f2e4cd09f7d2edec45bdcd1 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/50eb332a27e244c4aaa7a9bb7516ee04 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d08215c7fbc54cf1bfab39532acc827f : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f92c5a4e4b604400a0fb0acce2acf3aa : SUCCESS in 5m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb2ac0074ed3411ca32d714529085021 : SUCCESS in 13m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9078142cbe634b2ab7f9ba7f97212945 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6de81e63c174aaab393788da184c38b : SUCCESS in 49m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a99098fd6e49440b97aa9b41dfc0f007 : POST_FAILURE in 21m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7841ca2da72b4c8886b8a9811fd2feab : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/64fe4bccfa034329b8d1528b7bcf04e3 : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":2},{"id":"44d7e4bd12aaf28aa5cb55d57da831449ffbba03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 23:34:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f3e3f48cead498c9406c8f1adf14d8a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/adb3aa5fd49c4687ac7e3e0253086ffc : SUCCESS in 1h 29m 17s","accounts_in_message":[],"_revision_number":2},{"id":"273d683d26dfac1b38ce2d36b9186117b1aefc18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-05 16:45:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fe89bf901764d287e4a4082c7c5d28e01523787a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-05 17:42:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7be2142573750435af551eec2dea22dd3bd8798c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 17:55:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3251e63604e84154be57b4eaba63e937\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741d7efd27484df08297f88a3564aa77 : SUCCESS in 9m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9241403ad956412aba4551094a8390ff : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57b7b4f8ca654e7b800c6cdca4cb170e : SUCCESS in 13m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38cd4df312484ecf8c98ca5230a6098d : SUCCESS in 16m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84b3ba6969ad40acb2ccc213076a3089 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c7b59a40b9c47b8a21c92aaa942976e : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a429dd925de4e30931d6e2575804108 : SUCCESS in 19m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a59d73830874219b91463d0d5c13141 : SUCCESS in 23m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/106af7fedbf642ccb128b4ef30b984d9 : SUCCESS in 13m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3128bddce6fe4498a6373f5f9a16cd94 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3fefdf66bfd94db2b983d5f45a434c5f : SUCCESS in 19m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4642c5d0cc34442da6d0abd4b336dade : SUCCESS in 15m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dd1877080e2344a5977bf3ec2bb1fd45 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/765a52c62e59462b8d63f05f7d4975ff : SUCCESS in 12m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8dd624f16fb44787bf7f99895c946868 : SUCCESS in 17m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e35daf3a45c04442975e74cebddbe742 : SUCCESS in 4m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb474d919ea84c658c9bd0964c880bfc : SUCCESS in 16m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b4c8a6c8ebc412583f6d6c1d3190d49 : SUCCESS in 6m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1582d4713637405e987ad25e5abcedcb : SUCCESS in 57m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca74e940f0c540aba8ab02fd308dc2bd : SUCCESS in 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9680136e854cad927ab6845a575f51 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8f6bc74389fd4bf7872b3c393b9d03cc : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":3},{"id":"ce330c8b559898eac643eb14a8f6ffca51c0e24a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 18:36:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80ebe07c8cd490b9585f37c6cb2548a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c6f1a9fefe64bebb7964aa524b14d4d : SUCCESS in 1h 49m 04s","accounts_in_message":[],"_revision_number":3},{"id":"74d84910d119bfd71e99fcd103064b1a2a08a063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 20:43:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8291c35bbe2a214bd69f958bb3c0b81b7a2e42d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 22:49:57.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/933932 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"1c556c5d0ebf38d7b33aff8583ba1f0d7c93b1f3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-06 10:29:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b943c3e1186a86215e88fc1124738ba174708e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 11:26:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0296ec68a22642229d335c3f302ddd76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc4ebd4f35042a0a69879ae4c71eafd : SUCCESS in 8m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a74ec6fa5fa47dd9ac912520fbb852a : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66cc77381a084c7698ffd623501d5b56 : SUCCESS in 12m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ab8a1aef2f14e26820c17353be4e3c2 : SUCCESS in 23m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01313df9d68d47b3a6ce38273f02155d : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3aaedcbe2eab4480ac66bd7f3eabaaa3 : SUCCESS in 14m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ed44fc37d61543c8b76848cfba6ad5b8 : SUCCESS in 15m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/600d8573c9bf4898af4fd140b76be4d1 : SUCCESS in 18m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/53c125c6820341b7b504d0f363cdb562 : SUCCESS in 25m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b2b038218fa4010b11c0aa58f5c4538 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/daa8c85c344146f588ccad4a6dba0ebc : SUCCESS in 19m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfc62f2cd5fc406f832502da9fdd0410 : SUCCESS in 18m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1d46e378e1ff455abae3393bcfa248a6 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a4c7dedf31794c4f99f615d74602a6d3 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6c8974067ce3407d8a682a944efe5169 : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbd5f6ba13734366b6fe9be00e24a20b : SUCCESS in 5m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/38a5e21ecbba46cc9d34a165ec54b6bb : SUCCESS in 10m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e12d234a550a45f9a83eb0a44176925e : SUCCESS in 8m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/80a3b18431cb421fb0314067e6e0109c : SUCCESS in 55m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/627160ebb64949b28c90459bfce829fa : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe25a0a96d384f76a4a2460ff40e78f4 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/660ca37010474483996e015422cd3f29 : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f5946d1bdb7a5381bc626fe8c0c03880ee1573e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 11:27:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b4448551a555516ad7d0e1c0f60aede4a2ddb55b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 12:31:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9acf89bd52c4a048b72189aaace6245\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b760c7f765e342b39927c04921a03c9d : SUCCESS in 8m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a3062bfe5c7246799659eb8ba2f00f7f : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dae1da7d45a449569a9a317b10d4ee1d : SUCCESS in 21m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a66d3aa2b26b4021854ab52e2aec21eb : SUCCESS in 19m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6354fc688c64f418de191cab347584b : SUCCESS in 26m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8471c156a88d4eeca1b5aeef06900b7b : SUCCESS in 21m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9556ee881744eaead3ffaa39553b05e : SUCCESS in 21m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4f6df3156cea454d8b6fb8d34ef7fec9 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a26fa8dff134087a4a034f190204d1d : SUCCESS in 18m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d42db1b418c24a3d848a86ec753fcc9c : SUCCESS in 9m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3635dccf61b44ea39a186db8ced684b6 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d595d9613ed74bfdb9d6b546f826da22 : SUCCESS in 18m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c1df9e412c44bee80cb73ed74361e95 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7032f900b337416fa38e2b1112ad4e7d : SUCCESS in 7m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffed542e1bb9424b986ed5cb2eaba94a : SUCCESS in 53m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61cd093e44ac4327b59d675fb3c23182 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be88ed01bd248e5ade0f8c7a4ddbe55 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3bb772d9d4834d85a51685ca9aac531b : SUCCESS in 25m 59s","accounts_in_message":[],"_revision_number":3},{"id":"484dee0da085604eeb18aefe58f0d2351a3f8f28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 12:31:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"751e21cf27de997fade1ce3e6f42fa3de86872fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 12:32:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90d34bc9be1444a4b2a009cf2fe72fbb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ee478d1fc247ceb7f2e9c24a1b232e : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c6d6a3d708cd4db99fdb72cf4f4dca67 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"50dfff7db1251aff0b549449b4e41633bacff4b7","revisions":{"72f6f5a2319bc4a36054275ec6bd9caf7ff24e75":{"kind":"REWORK","_number":1,"created":"2024-11-01 18:25:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/933933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/933933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/933933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/933933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/933933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/933933/1"}}},"commit":{"parents":[{"commit":"41295e0b89abdd78c56955a90e166f7be14204df","subject":"mpu: add x-backend-etag-is-at header to GET/HEAD requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41295e0b89abdd78c56955a90e166f7be14204df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:54:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 18:08:38.000000000","tz":0},"subject":"mpu: fix param passing to sub requests","message":"mpu: fix param passing to sub requests\n\nPass relevant params from client requests to sub-requests\ne.g. \u0027marker\u0027, \u0027end-marker\u0027.\n\nChange-Id: Ided25598694badc6e9ddf2887adcf5263ec263ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72f6f5a2319bc4a36054275ec6bd9caf7ff24e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72f6f5a2319bc4a36054275ec6bd9caf7ff24e75"}]},"branch":"refs/heads/feature/mpu"},"5b795540740f2dc46db940a2f9bcfa9fa16644d1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-04 22:02:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/933933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/933933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/933933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/933933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/933933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/933933/2"}}},"commit":{"parents":[{"commit":"d7e659aff117c03165b3d55615e7fc8ae38785ce","subject":"mpu: add x-backend-etag-is-at header to GET/HEAD requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7e659aff117c03165b3d55615e7fc8ae38785ce"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:54:16.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-04 22:01:11.000000000","tz":-480},"subject":"mpu: fix param passing to sub requests","message":"mpu: fix param passing to sub requests\n\nPass relevant params from client requests to sub-requests\ne.g. \u0027marker\u0027, \u0027end-marker\u0027.\n\nChange-Id: Ided25598694badc6e9ddf2887adcf5263ec263ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b795540740f2dc46db940a2f9bcfa9fa16644d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b795540740f2dc46db940a2f9bcfa9fa16644d1"}]},"branch":"refs/heads/feature/mpu"},"50dfff7db1251aff0b549449b4e41633bacff4b7":{"kind":"REWORK","_number":3,"created":"2024-11-05 16:45:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/933933/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/933933/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/933933/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/933933/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/933933/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/933933/3"}}},"commit":{"parents":[{"commit":"afe6157773487f8c1b6d4544334c5c0c4b1eda85","subject":"mpu: add x-backend-etag-is-at header to GET/HEAD requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afe6157773487f8c1b6d4544334c5c0c4b1eda85"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:54:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-05 15:00:06.000000000","tz":0},"subject":"mpu: fix param passing to sub requests","message":"mpu: fix param passing to sub requests\n\nPass relevant params from client requests to sub-requests\ne.g. \u0027marker\u0027, \u0027end-marker\u0027.\n\nAdd support for part-number-marker in listParts.\n\nChange-Id: Ided25598694badc6e9ddf2887adcf5263ec263ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50dfff7db1251aff0b549449b4e41633bacff4b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50dfff7db1251aff0b549449b4e41633bacff4b7"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
