)]}'
{"id":"openstack%2Fswift~933935","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ib00d55b3b787da981649c5cf0c3cc9bb199857c3","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-cross-compat-tests","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-27 15:40:16.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-05 17:37:51.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"Ib00d55b3b787da981649c5cf0c3cc9bb199857c3","subject":"mpu: migrate s3api multi_upload to use native mpu middleware","status":"MERGED","created":"2024-11-01 18:25:00.000000000","updated":"2024-11-28 04:21:23.000000000","submitted":"2024-11-27 18:15:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933935-p-mpu-cross-compat-tests","meta_rev_id":"a10108c77e5af6ba10e33296249d5ece46b194c2","_number":933935,"virtual_id_number":933935,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-27 18:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-27 12:35:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-27 12:35:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 19:32:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-21 23:23:00.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"085d2b819bd4e5469649c862ea10e679b4d2f1c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-01 18:25:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e3d37adb30b77358421be54b216207cb774407b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-01 18:25:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e80333349cd6a37e088b14c0bfd338552a535847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 19:32:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/960fb852de794dcda894d947787d5d19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3cee3400b445148c3ad6681e714563 : SUCCESS in 9m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5132a248ecdd43eb9ebd4fb2e93b8362 : FAILURE in 13m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9894580cbc5f44aeb34a3310e22c1df8 : FAILURE in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9e6cb504c0c49ed93ec7fdb8359e073 : FAILURE in 18m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c37ea873bd54523aa56781a0442b499 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/548992ac57374109a7154ed6d103f561 : FAILURE in 22m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0da30af98ca54de4a14c9fd3c4950d52 : FAILURE in 17m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f57510e29aa4401094ec53eb8365b59e : FAILURE in 22m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/66a66eae3b84427784647790de5bcd2e : FAILURE in 26m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f90bbdf1065b4ecd9a13fdd3cf9b3972 : FAILURE in 23m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8783cb6409014aafb0077c126fab0046 : FAILURE in 11m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/453f44ee129842e8ba8dfecec559e935 : FAILURE in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9231cac695d441178b8bbda99c2f0e15 : FAILURE in 18m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/10be617bc5a145bfb6d6e951e6c99c83 : FAILURE in 12m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3c57f3cffadb4a808efc949b693c5c43 : FAILURE in 9m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9a12dd26302f4a7b9320c5a74735613d : FAILURE in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc8c5f2d08734aa5990e8bfd277c2101 : SUCCESS in 5m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0b161dd08f304fd8921648a7e66ed410 : FAILURE in 15m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b4ecd6767a04bfd80ad4741beebe08b : FAILURE in 5m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/54bcf51911fe430dae04c39bb43432a1 : FAILURE in 52m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b164916a8324cbca8b199f698a81c02 : FAILURE in 21m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a5dcc6b5afe41f6aec6a0dd71463720 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb561359057f41a2b060930f75a9cec6 : SUCCESS in 25m 13s","accounts_in_message":[],"_revision_number":1},{"id":"141e9cd1e3d4e5e9e830f8730beb6208bfed3608","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 21:28:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9352d6edf8c453682a5c39ec476f6c3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91d1519d73564bed977549fca57e6eb4 : TIMED_OUT in 3h 01m 12s","accounts_in_message":[],"_revision_number":1},{"id":"823a64a3b7f6532d4f7d27f123ee613f0b5822ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 22:02:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"afe5cefa5054a1beb9f26a2c96c13e9cebfe1532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 22:57:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e2ecbbaa5e942f5ae53da5322ce7336\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8660acaee9e84d43a4e07c8a75de833e : SUCCESS in 8m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ff4500601f442718b1c632c5b427d9a : FAILURE in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/959a2161c3624d48b1db9e71d9f6ae12 : FAILURE in 15m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb1fa570f5fc4e698e96a70753eef201 : FAILURE in 16m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb81151c11894d3587ee0cda7cbfa06a : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06552853848c475d81dc78efd52c6179 : FAILURE in 16m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/78d74336b9a84505ab703a1bf43d9449 : FAILURE in 13m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfcac48c049548f6b89e6efd38f05dd9 : FAILURE in 19m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c8d844d84f764c81b5730b8f246a4cad : FAILURE in 24m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64a61bb4d5d84d02aae1f4618b449ced : FAILURE in 19m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c52109f47a10455fabbb6ccd7f648926 : FAILURE in 14m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fdf1d6c98d1840849ca7cdb91606e51b : FAILURE in 18m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1de0c2250915483cb9650631f59e8f1e : FAILURE in 9m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/14e2c11c4f7d40e6a47f9277895c8480 : FAILURE in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b4147d22cc2942e19e2c0db06d0c12b7 : FAILURE in 15m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c614906e09b4ce893d9c6d1bce74afc : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2996fff969f64fc69672f9ef6f9baba4 : FAILURE in 13m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a5a41b9685114a828946223aeefd3b0f : FAILURE in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/681797ffa4654baaa3eec61b76d16846 : FAILURE in 35m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dce260c0e6f4e668829e5999863ecb7 : FAILURE in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/371634fc82c046ba86f28e3ad075aedc : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bc175d6bf3254db394ee5ee8f355f84c : SUCCESS in 17m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b95c20d0b74c22a7268eb7952fbcc88634a5b927","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 23:26:12.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2163d3f918534892852b5be10c60f834\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a238a2cce8c44fd4942c52a27beae2b3 : FAILURE in 1h 21m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e34d332a5c71507d94c431bbe5f4cf81157540d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-05 16:45:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cd04f5409cb783ea168fa7d467fd2e465a3bf52d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 17:37:51.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80e4a9ab9b8c40bdab365d3cd1924b36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29b8a2110ec44a9afdb434060233ea1 : SUCCESS in 8m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5e6fe368026496191bc7798c685a6dc : FAILURE in 18m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8937b37642114db892a868da9df064ea : FAILURE in 22m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c120f7763ed54ef2b7e6d71a0ecd6b2a : FAILURE in 17m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a6767ea33bd4ddc84afc7307ca2af65 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d104a17cef264ffa8fda072b62c5319d : FAILURE in 22m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee3ef7ef88f54952b088caee1b1083b5 : FAILURE in 15m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eb3909919cca4e9d847f1a2667554000 : FAILURE in 25m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/93969ead23264fb480f10614f8d69526 : FAILURE in 29m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3ce3628f471646978910dbe83bdea0a0 : FAILURE in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b4ff36ef37424450a87ff8665e7605a4 : FAILURE in 16m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/86886803123d4181bd86a84cfdea5033 : FAILURE in 9m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a40e9ea6679d4d5bafa2d785b026fac8 : FAILURE in 13m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/787ddbb6a442454fb0d7bd38375a0e79 : FAILURE in 12m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c8d2e35586a74ec9bb4cd9a9deb53434 : FAILURE in 14m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ae328ae464d4cb888e5806d8f8ef119 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0bca18aa0b924303af4e5d94fcd9c231 : FAILURE in 12m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/838a993039ad401cad3ea30513533ef2 : FAILURE in 6m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5e3c5b101e948ee8c6f9c20c16e4875 : FAILURE in 47m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/428643e8c5e343a1acfebfeff5ea3471 : FAILURE in 19m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/561bd86a5240497abc0663c3a5ffc326 : FAILURE in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50f1c4aeeaaa4de99ad4a483300d0c47 : SUCCESS in 16m 25s","accounts_in_message":[],"_revision_number":3},{"id":"8e2e4c915c5bad3d87b66e4edd96a4f2329f37b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 18:28:58.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da31ef700e574ccda6814354b693c8b5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/754bc87da9b64b2d9d1e0941fc8148c0 : FAILURE in 1h 41m 18s","accounts_in_message":[],"_revision_number":3},{"id":"f6bfe723cb41c195390c2050a971c18df7715b54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-13 10:37:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b2e4cfd9700a12ff4ab35ceb02dc410a46a5d54f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 11:41:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc3bf4f724764305893376789f040144\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335f04f31c154a6e801f2fabe44c16aa : SUCCESS in 8m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa01f15189a64972960f4b81d8fad59e : FAILURE in 19m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49fe005061d24ecca0aa7f0df1dd0221 : FAILURE in 18m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3139fea23804246bec3a94599049489 : FAILURE in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a6a52a1ba7d8465fb5ead456c1034278 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/251ef1aeff99413da28bd37d4e8e1b05 : FAILURE in 18m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/12816bc492724c2c83fc8892849ad631 : FAILURE in 19m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/989df97bcc254eb39de92723e86c1dc0 : FAILURE in 10m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/927a596345e44a1ebfc47ed9e99ddf48 : FAILURE in 23m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/08c004b7244e459a9c12148bea7a461b : FAILURE in 15m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d7f4f0a03bba4d838703c4fdd29b5451 : FAILURE in 16m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/22c9ff8a415c433ea1700c25a52266e8 : FAILURE in 10m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c1b9e56a2ade46d1a46978b595e5b39a : FAILURE in 15m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/29335ad3e68a426d8de09d9d323d77f6 : FAILURE in 12m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aa04da6d0adc4b54a3401a91b2cdaa9e : FAILURE in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/727a02bb4d03453cba1ab7acee9f439e : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a6612db2933347dd9b19f84d46747c9c : FAILURE in 10m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3608a50b607a413884e230d60be87e7c : FAILURE in 6m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9d2f754159b411598db72d85aef54af : FAILURE in 59m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d16656a2f074cf99b068d870a98abb5 : FAILURE in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/055031cc6bb64b75ae3d8ac2d13e65a9 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/34d56740dfe348859ef8248b3b287d04 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":4},{"id":"fd120877899d7f7d5d32dec698a60fb8a00f5e4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 12:17:25.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dfd59dc6056440fbafb41488c5bd57e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4bd716dbd7f84873afdd911809a8162b : FAILURE in 1h 36m 31s","accounts_in_message":[],"_revision_number":4},{"id":"3ade5b8c1e6012cafaef1ca35ac3a673cbdf652e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-15 17:16:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9b78720a4324805df72ff3010d2de7b284bb2b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:09:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c12081d6e6cb4294ab449f9e63b70b47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cda7c67d5294354a6fad9e077a3e700 : SUCCESS in 6m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b26b350a56fd48dca02e933b9713fd17 : FAILURE in 20m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb4a05736f804ebaaf6dc9671288c23e : FAILURE in 22m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/220fad0a974f42ab8dbb9aa609310dbc : FAILURE in 21m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/360ed37ee8f8469fa2efc93a07a74b3c : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5a0c1c9d95a45ba8d70edeed9e0329e : FAILURE in 16m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2fc267a1ae2b4f2ba0e49a42558aa6af : FAILURE in 19m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4177b0f986d44a42bacc5f48681df1e6 : FAILURE in 21m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7d7eba92ac7e495daddf547d1b287cbc : FAILURE in 27m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b2262724f974a23af0f540216ac17f5 : FAILURE in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/243cdcb628cd41c686ccfda48038bb63 : FAILURE in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/19a7de5bb83e4640a8aaee5eb6ea5f37 : FAILURE in 15m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b9692e469ece46a19fc185e490eac60e : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3ac450570f2a4033bb58ff173afb5acb : SUCCESS in 14m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ac739a15b55445fe82fe5d398feb8457 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d994cb2d6f9489689ad39e592e42503 : SUCCESS in 8m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c51f93b5cdf4884be86094fd6c72443 : FAILURE in 13m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a5ca04b86929431ca93b1e749cbc3fbd : FAILURE in 8m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e59b08c174ce43c6babf6bdcbf2e3246 : SUCCESS in 46m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49bd39ec3d854800a82703e2b2df8392 : FAILURE in 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81f969b8815f4fec888d444fe839a4d9 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eab5b0884fd842a9adfdc6e1968f10c7 : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":5},{"id":"6d90d2fc55da59fe46c79bb6df29337d61a7f61f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:53:46.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9c0641ce69f48caa9629c5a119e7394\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1d4e63408a3847f3ac4bd78c25ae98c2 : FAILURE in 1h 34m 51s","accounts_in_message":[],"_revision_number":5},{"id":"259f597f829ee6baa05053603e05faeb4558c333","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-19 13:12:37.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"d4adecf78df040a6667a3cce2181730aad6cf159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:20:33.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b39c5ebe9cd44849963600c79c19463f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011281a3bd99468a966f0e61a6661cdc : SUCCESS in 12m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aff8750d2b6549fb87ad0123485dc91e : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec0b508180a24646b1f4aea7067ad164 : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bb6075440404b9aac6516ac76665cd2 : SUCCESS in 22m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d41ec2491a84f06a74f27729e01a1d9 : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ffe297319a624db6af1b977ee284080a : SUCCESS in 20m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/214e91e1293541c494d34a8cf3a6282a : SUCCESS in 24m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3f2565be211e4a3a9a1f6ce6000affbc : SUCCESS in 22m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ae94ad7f7ede4798936ae93354297157 : SUCCESS in 37m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d090e26c86441fbbc472a571af4fd27 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fa33b8c778e44a582321e5e0035e202 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1764f15669f34fdeb7fec38c009e9c8d : SUCCESS in 27m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/25109847a98e46608b22bb118ce4bd45 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3c499e98327d4274a9a380280fe0fa92 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cc8ced5830614c02bf9a0449460343a8 : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ae7c36cb6d7f4f3d8d86e522f3283e6a : SUCCESS in 7m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55b91e76d535437ba61eb1e11c8c83a9 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/863892e41afe4a55a8cd1171b6fbf00f : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/591b9d8e36b849c1a815c28f8365f5fe : SUCCESS in 48m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15f3a21948e64a5d95bccdc02e1a9209 : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/243e424ebc6b4039817bf733dba1156d : FAILURE in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7b7b5ac5bde4e7293afce8d9c9d8d83 : SUCCESS in 17m 24s","accounts_in_message":[],"_revision_number":6},{"id":"28d00b734ee3f6f644b8cc1cdd672b835da2418b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 16:18:43.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2047c4ed6344dd68d24e53aa837f87c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c1816fa824ef45999290a772e2fa7302 : TIMED_OUT in 3h 02m 10s","accounts_in_message":[],"_revision_number":6},{"id":"d7e4b04d6e8283d08731701187f6fe056eab1cf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 12:32:50.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f9991b60a7a8cc9436189212876ed0c6f71fb5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 13:36:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b94585eb8414eb8a609cd7932f88e42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce447a61f804c1a8bc17175f7717f9e : SUCCESS in 8m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf2872a0faae43ba9a86b3d9b91d7ae3 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aef344378fc14e5ba23bdd2dc5d297c5 : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30e9141c03d746a79a5233a9fad1ca61 : SUCCESS in 22m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a8cd446509084d24b73759bee3855717 : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/00144b06de6b438685fb62ce51a7b2d4 : SUCCESS in 20m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d2b428244cb4564abfcb89dc43c38b3 : SUCCESS in 19m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d7ff6e5bd014494ca77541d7dc7bc4f3 : SUCCESS in 21m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/81e7322ef0c94aba91bf93b6723c1c3d : SUCCESS in 30m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/86ba26b4735348fda63c02c41ed86f89 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08cb89d5bb844835beadbe798b15f4c3 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/148dc1b4e6de484abe2cc6295183bc37 : SUCCESS in 16m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a3bd81a85c88420b94a2ead71e0ee937 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/387122b8bb3447eab0aaf43291da6511 : SUCCESS in 14m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8783c20be275473fa0b8a78e015dbe24 : SUCCESS in 14m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/339383a2a90447c6b69d05cf7df23778 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eeade54f8ecd4f6297e19bc2868401e5 : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ccace7da88840df9b176128dfdbcce4 : SUCCESS in 7m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3839a1c6612b450b9983387678ec5e9f : SUCCESS in 52m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/668d2d861c4d41869cbb9b4218c9698f : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07d39189a90a42a6b1dbcbc934ee78eb : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d66df32b37514aab8fedd2cecab75dc2 : SUCCESS in 19m 29s","accounts_in_message":[],"_revision_number":7},{"id":"d2dcda1123891d636da98fd312d2e49c3d688bf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 14:01:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b67a24875b6e47f1b1aedca4783cb021\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f25201d5031d4f039e9ce9efb5d7bbcd : SUCCESS in 1h 26m 05s","accounts_in_message":[],"_revision_number":7},{"id":"e70985a3f252662da1719613c178502a71638f21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 15:43:02.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"f55ce76e5db40ea1dca149b20b5d90db840084c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:34:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81fac5d1096f413ba57691ed8acb6b3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/177d5c814298472b92d78deb600288f6 : SUCCESS in 7m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8136d5d7ac749a9a9dca1707e35a6de : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b15c6d6c80144c99934f6212bc0e581 : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d65696d1e3040869d5924f870639fc0 : SUCCESS in 17m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2ca7d3beeabf46498608f88076907422 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14caa9c296e6450f88fe0dbda18b899c : SUCCESS in 15m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f271dd162994811b27092fdc7a572ed : SUCCESS in 18m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fbc8ebdfea1e4e2d95296d7466238924 : SUCCESS in 19m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/68544efe75bc4c638846a6bc6b7341ad : SUCCESS in 21m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c50e3c4cf72248ed86a3ea8c78ec422e : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d726a60c5164a52bfe0c5d8c2076419 : SUCCESS in 9m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/703c008f787241d8aef12c28ddd4cac7 : SUCCESS in 15m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f455d6362f7f4beda4ad47af9605e298 : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bcad133724f94b44abcbabca975f5497 : SUCCESS in 12m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c50719406ce64a268999363bf1fa3ef7 : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d6cd8c39429b4796aa5a9d67b48bbec1 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e15036446e34301b42763eaf7233e48 : SUCCESS in 13m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/85249cb68cb64b04b4ffe019cfdb8e70 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbc11d8f103b40ba9155590af307e3dc : SUCCESS in 46m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0607b616c134205897e6e059d0f72ed : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d15acd8956b44d6b9b73e7f8fb21914f : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4084a9ed970a44a5a489efe161fb39ab : SUCCESS in 17m 46s","accounts_in_message":[],"_revision_number":8},{"id":"683885a43b51303bd91906f5836b8b56ce7a667b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:50:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2a740b516db4728860aa298e3e180e8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2174197a168d4f889c539c81cde413a9 : SUCCESS in 1h 57m 28s","accounts_in_message":[],"_revision_number":8},{"id":"35b0af91bd187412b918183a6e797017a90392cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 09:50:21.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"0e52d53ac0e87e5181937a2df77b1eec496b1222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 12:06:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9cdc9e65acc4949afb5c56021c86b30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7d04ccaa7564edaba908c5309f603a0 : SUCCESS in 9m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c84f4d8d369349f2be240e97131c37bf : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2a1c3fac7164e3493346857d66c3c99 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d965c2856974c1297e599232beb1d89 : SUCCESS in 28m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c12ba43546243b2b97a6c14e4e7278c : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78371186bd6a4bf1a6092a934cd6518b : SUCCESS in 19m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac928b0b1f19441698074424af54fb21 : SUCCESS in 16m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f58d6a4748c4caaae7c7f0b0f2f01eb : SUCCESS in 15m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1bd5b83f9d0e431e80c4231c04a00e12 : SUCCESS in 23m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/713702d68433411eba7da571581e4ebf : FAILURE in 20m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ebf7c0a6199340b9887d3bd9ca8a6a47 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae1594ba2675403aa732df4afa512477 : SUCCESS in 16m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9f050b4d018f4118a01f9607c040c997 : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c1572f1744764070a641b6a4bbdd5098 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ed856ec579f8425da1534891dbbaabca : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2eac95bb03684f9e9f659382d502618d : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aa58344dc10d4f5cbfd9d58e77a99143 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/82e3602c623b435d8ad7aae9848926e4 : SUCCESS in 7m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/07d3c6817373460cbe2c54acf36d6fcc : SUCCESS in 52m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c4855eb6b2b4ee48d940c85aae0192d : SUCCESS in 22m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dbde266aabc4313bb4e726574c9c6c2 : SUCCESS in 3m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06db5367f15e4b51a0ead0394e069987 : SUCCESS in 20m 35s","accounts_in_message":[],"_revision_number":9},{"id":"3d54c3dbdf7d3608c6d76dae75b9ebc6d7a1fe1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 14:13:33.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"9e7f76a74e34680a87776c2bbbf94d0d1b06a2da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 15:31:09.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88e52f5f1b7640a495070275cd663217\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d01099b9256246bf8d298a25c3c9e57f : SUCCESS in 7m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73ff3e9eba774f6e9835d97b09b76356 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b4c3fff30434bcfab3470f129102141 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9068a14728ba4c16ab38c814803a59c6 : FAILURE in 18m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1680b73463714c41af3cfddfd230eeb8 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8420f4859da64279a5d5be57fa12e619 : SUCCESS in 18m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d71d1caa53044385b05c6bc1c2712d39 : SUCCESS in 17m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6d65f31fd0014195a6f52fb3f2d7dc30 : SUCCESS in 15m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e935fc33b7994257b9aa29adc08ac7c9 : SUCCESS in 28m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a3e75cc674345398c6e23422227af01 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0fb971ea12334ba397531147b5265fb0 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5eb0c199b702476da5d1933dfd769a46 : FAILURE in 18m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/639ce947eb7942cba39ea34797fc41e1 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c34686a6104f4ba5ba8619ffaa836e39 : SUCCESS in 16m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bf88de98728542478c1959fb21011551 : SUCCESS in 17m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d7d0489b45b47eb96111e94a1f30699 : SUCCESS in 7m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c49245c1f7c74f908b02977195d5d03d : SUCCESS in 15m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4913518ba6094fc09401a6df90dca457 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1712f315198041b69975b40ec1e22119 : SUCCESS in 59m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82f6d1c9a6324b61847fe36fe1faa36c : SUCCESS in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8405bb9004df41f89731b2082d14b0e0 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/96e1386e1c2a44b68f697d673bb076f1 : SUCCESS in 24m 33s","accounts_in_message":[],"_revision_number":10},{"id":"b6344b43477068f35ea1b7a53d8d12b4384a9059","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 19:52:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b84c401e8843fab847ba259acdf43f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44de68c9230e4377bcba0db6b11035f8 : SUCCESS in 2h 13m 16s","accounts_in_message":[],"_revision_number":10},{"id":"3688d5e9dbda34fad62a31b148b748711a7dcdff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 15:44:40.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"24d0b94d3582e79fbb9866d0139e75bca6426cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 17:31:35.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6d32f46ff3646afba66356080eda911\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/321c58c7162348a8815f8ae62fcc9002 : SUCCESS in 8m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc12a0356a5d49ee813bd92f69a30045 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88eeb259e01b468dbdc6192059a3f6e4 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff5c9f1ccd9c4f8898b624f5d56a8551 : SUCCESS in 21m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b52095d6b49f4b3a87ae8e90ac856ce1 : FAILURE in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d742d2e10ba4ade9eda744d721c681a : SUCCESS in 10m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ccd09ddaa9e3440d94774ab8450ed797 : SUCCESS in 18m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d408b2bf7354950b95f0b21e0e35fd8 : SUCCESS in 19m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d226a27643c04c55ab5be86c7d949082 : SUCCESS in 35m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8592c6edcdc9496e807b1f9db3b1a0d9 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8e94b649acd4915b9b26c6c7dd13571 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ceae6ada80741098b43006f09bbdfc5 : SUCCESS in 19m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/66be8be45efd4669a6a1327f2c7fc790 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a3de667dfc334b2e87ee1ea9a79e5923 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3c6c8c21ab80437aa2d364b84c3df2e0 : SUCCESS in 15m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05ef1c882ba0424d8c243866fb3acac3 : SUCCESS in 8m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b795638a2b41421b89b779de98900b77 : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f177c0a629cd4e538d52d584874e5c70 : SUCCESS in 7m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cd131408ac1448ba642c96f723fb696 : SUCCESS in 1h 02m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/feb4d6b0d95248f4a26add25b55c110b : SUCCESS in 18m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52962fff66194869a29b8c8beb4f6ea6 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d46f113e643c43ad919253ba50e33057 : SUCCESS in 25m 06s","accounts_in_message":[],"_revision_number":11},{"id":"5fe5c719b01dd9364cf0fcb8eff1dd494a1c9b3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 21:39:09.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe5aa0983ca42ef81b8d2aa92172e85\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e122dcb9ad774a9e98dbc75fe36272a5 : SUCCESS in 1h 41m 56s","accounts_in_message":[],"_revision_number":11},{"id":"24f16e98867a990ab08432bfe3de94602f2936cb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 12:35:43.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"c70077aa2a269acdd036c6b5c76cef6998b27e18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:36:14.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"2dac126fa536ec6aa4f4af7b79e2495739626239","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 14:53:16.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/386d18f4abb446ee9dea808fc591266a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc45a624e57547e6a60089d372907db3 : SUCCESS in 8m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a4c198ab6fd74e68a224b64934718d86 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad4dbca9d8684922b7763bf1e44a5b32 : SUCCESS in 17m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bddb1bc3c25143b084ddcc572a92ef4a : SUCCESS in 16m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/de37958a698149bfbb34839e9ac0dafa : SUCCESS in 10m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/68d0e5c89dfc461797f04b852e7171d8 : FAILURE in 28m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4761e593fea443e78355cd03d62a81d5 : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/79cf0f3632624cf49c33412e86bf6bb2 : SUCCESS in 19m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4918a599ad574164a5e5d8f0025f4dc6 : SUCCESS in 17m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cf55485c9dbb47968e2d74193249fd85 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/26e52671e9eb44429fdbc04ddbd3a1be : SUCCESS in 17m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fddaa5817d6244b4bcc34100bc884898 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab74de72135542e48ef27ba9b39cb92c : SUCCESS in 6m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/90172b7c53fb4823ad27b6565bf723f7 : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b00cf6ede20f432790edf1937d90b0a1 : SUCCESS in 50m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d2a24b8b17a4aaf96b733e5a21a0fc3 : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d94470334949f58281477bcbd6df6a : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92553b6a05f74510840f74f13ab894fa : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":11},{"id":"54131cc3fa0767e81d57ae329014e822ac4e60dd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 15:40:16.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"febb7b1a517f789d17b20623201abe9fd92fd404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 16:26:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/767b277acced4122b071ac0e3df3f372\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c84763b713464eb7518dee5a5a139e : SUCCESS in 8m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c03048317b0942589cb2150bff5bdb89 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2692796ba61b447d9c6b20f5368815c5 : SUCCESS in 17m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c413b22a22d34ac498bd9ccdacfb6e56 : SUCCESS in 21m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/365471ffd9cc4b548c3d0a1fbc6732f7 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f670e1a208342f18a8ab43df4e77a91 : SUCCESS in 20m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9e9986290a6c44d88c4ee3a101a8e0c8 : SUCCESS in 17m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9d99f5df87064bd0aafd876145571895 : SUCCESS in 17m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/12b2bf4bb6ef417cbb06e3845330b3a4 : SUCCESS in 30m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dcacff3e60fc48bbaff1b86ade4c5097 : SUCCESS in 19m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc96ffd47f1a448fa10788943c0836d2 : SUCCESS in 10m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf84c2a49fb24bde92ecad8aaa3fc025 : SUCCESS in 17m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/985db8e6d7764fea96bf2d9ddd23a2d5 : SUCCESS in 9m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8782a1fee4aa4f62b40ae73c9b5e63c9 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2c5642a81fcc482a8ee796889c98b652 : SUCCESS in 17m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ddccb3c66c4440c5bd5609fd97a97203 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a11a1fb438c48db9ebb1f751fe58ddc : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7dad8122ce0d4ad3b14dc852ca971802 : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/78b73fab98b24612ae221aa11f72f943 : SUCCESS in 36m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44374e31f4ac4215be325884a9a78ede : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23cd35cc1c0e40f79d73b0b35fd74f9e : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c80c9f75a00e4a56a486034a8e6912d0 : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":11},{"id":"8ed5476012b26940a22ad7d5f6c391a3acdcac93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 16:26:43.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"d88ae13bb40987a7d2d0a938e533435ca32acead","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 18:15:59.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59ac8e8ea34c4f57850bc85409539558\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e28f5ddd05410aa3da7d0de855d4a1 : SUCCESS in 9m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8cb02df3eff14cbb90bcb4a2db4ec9f8 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8039135105784bc8891b5eca075d4cd9 : SUCCESS in 22m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/feb4a26f3643447d9cf25fe2eb43c6e5 : SUCCESS in 15m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cc22f66da6ce4246ba4bc625994135b9 : SUCCESS in 24m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f4b647e5018d48759ca36a86bf059c59 : SUCCESS in 22m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/48d67487bfbb4c4e9374487c90a064f1 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22d5338b6b6e46d3aea92f01a8fbc62c : SUCCESS in 15m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d2b4f5f58e149f98c3a12107bacb9e1 : SUCCESS in 15m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dacae94fec40469aa76a208c64e15055 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4a7d81e0f87748b1bdff9040b64ea01c : SUCCESS in 16m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/513fdb33e70b422dac3af754a6208c1d : SUCCESS in 17m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/454ac2cd6f9d4ae19aca5d226898732c : SUCCESS in 8m 32s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/86f972f4aba845cf916589c1426f37db : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5539c4373b24b81b4138c988af35ac0 : SUCCESS in 51m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40244a05df244138b137412e0b122ca5 : SUCCESS in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ed209be87c4299accb2634ce3fbdd9 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba210b67793e4c738fb3e2c75df31cd0 : SUCCESS in 22m 01s","accounts_in_message":[],"_revision_number":11},{"id":"55e742946a9579e170f48f6c0386f4ab590984c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 18:15:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"dd9c472a686f0bcfdde529faac334657f85b2bd0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 18:18:02.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae9ac7a66084b71988658354e228ad3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0d4f387c864490abba48b5ad3927cb : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1a41e93bc532495e945a4696ad015c83 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a10108c77e5af6ba10e33296249d5ece46b194c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 04:21:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb3ce1f7a9c4875a9abeae1f32d3bef\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e80165ebbd8498696e2a1150e43c546 : SUCCESS in 2h 51m 20s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"819b63b76a8553668466fd699f5e234f14597ea0","revisions":{"3413429b08b6e49c1bab391711a0f7c1750a0e3d":{"kind":"REWORK","_number":1,"created":"2024-11-01 18:25:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/1"}}},"commit":{"parents":[{"commit":"250e76f95a2f392cc2e6ac57eb34e4ecd9f40e7f","subject":"mpu: change some response codes to keep s3api happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/250e76f95a2f392cc2e6ac57eb34e4ecd9f40e7f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 18:24:23.000000000","tz":0},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO: only one functional test has been verified!\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3413429b08b6e49c1bab391711a0f7c1750a0e3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3413429b08b6e49c1bab391711a0f7c1750a0e3d"}]},"branch":"refs/heads/feature/mpu"},"f9777abf0fd12c3393af1a7c1628b4a9e8851c23":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-04 22:02:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/933935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/2"}}},"commit":{"parents":[{"commit":"b2ad03933714dc879cb4d6c3a3207c020cb33669","subject":"mpu: change some response codes to keep s3api happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2ad03933714dc879cb4d6c3a3207c020cb33669"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-04 22:01:11.000000000","tz":-480},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO: only one functional test has been verified!\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9777abf0fd12c3393af1a7c1628b4a9e8851c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9777abf0fd12c3393af1a7c1628b4a9e8851c23"}]},"branch":"refs/heads/feature/mpu"},"ecb0f2b6d94333818d80ef158650fb32543cb361":{"kind":"REWORK","_number":3,"created":"2024-11-05 16:45:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/3"}}},"commit":{"parents":[{"commit":"37d6b264a00cd3f57aa014caeb9699c617f6eb2a","subject":"mpu: tolerate complete and abort mpu operations being retried","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37d6b264a00cd3f57aa014caeb9699c617f6eb2a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-05 16:45:15.000000000","tz":0},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO: only one functional test has been verified!\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecb0f2b6d94333818d80ef158650fb32543cb361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecb0f2b6d94333818d80ef158650fb32543cb361"}]},"branch":"refs/heads/feature/mpu"},"44600fb808bd26092bbcf27d6d15a39bb7fde4bc":{"kind":"REWORK","_number":4,"created":"2024-11-13 10:37:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/4"}}},"commit":{"parents":[{"commit":"fdcdc24a5e8c0400003e283bab92f2d5721f697b","subject":"mpu: forward headers with uploadPart subrequest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdcdc24a5e8c0400003e283bab92f2d5721f697b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-13 10:10:03.000000000","tz":0},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO:  unit tests need work\n\ntest/functional/s3api/test_multi_upload.py passes with one TODO.\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44600fb808bd26092bbcf27d6d15a39bb7fde4bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44600fb808bd26092bbcf27d6d15a39bb7fde4bc"}]},"branch":"refs/heads/feature/mpu"},"b1a6c40d79ba59e693a19134e2647b60029cf3cd":{"kind":"REWORK","_number":5,"created":"2024-11-15 17:16:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/5"}}},"commit":{"parents":[{"commit":"c67104c4f4f76ad09e70e80247a729fddbccec52","subject":"mpu: detect content-type during createUpload when not given","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c67104c4f4f76ad09e70e80247a729fddbccec52"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-15 16:57:12.000000000","tz":0},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO:\n\n- test_multi_upload unit tests all skip.\n- test/functional/s3api/test_multi_upload.py passes with one TODO.\n- Some other func/compat tests skip.\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1a6c40d79ba59e693a19134e2647b60029cf3cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1a6c40d79ba59e693a19134e2647b60029cf3cd"}]},"branch":"refs/heads/feature/mpu"},"97e35e34c8ce51fff1f4f53d458778841ca2c407":{"kind":"REWORK","_number":6,"created":"2024-11-19 13:12:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/6"}}},"commit":{"parents":[{"commit":"abec3d3113934f750d8c712a1524a3896692abeb","subject":"mpu: copy sysmeta from create request to manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abec3d3113934f750d8c712a1524a3896692abeb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 13:05:01.000000000","tz":0},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO:\n\n- test_multi_upload unit tests all skip.\n- test/functional/s3api/test_multi_upload.py passes with one TODO.\n- Some other func/compat tests skip.\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97e35e34c8ce51fff1f4f53d458778841ca2c407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97e35e34c8ce51fff1f4f53d458778841ca2c407"}]},"branch":"refs/heads/feature/mpu"},"308dcd33939ece0b62afddbc681fd32ec4ef3853":{"kind":"REWORK","_number":7,"created":"2024-11-20 12:32:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/7"}}},"commit":{"parents":[{"commit":"20f17c75b6311846f8cc434ae16066a6f2a41e48","subject":"mpu: copy sysmeta from create request to manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20f17c75b6311846f8cc434ae16066a6f2a41e48"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-20 12:32:19.000000000","tz":0},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO:\n\n- test_multi_upload unit tests all skip.\n- test/functional/s3api/test_multi_upload.py passes with one TODO.\n- Some other func/compat tests skip.\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/308dcd33939ece0b62afddbc681fd32ec4ef3853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/308dcd33939ece0b62afddbc681fd32ec4ef3853"}]},"branch":"refs/heads/feature/mpu"},"59b4cf27274e9b7f87c8071fa8791e7d0c13200a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-11-20 15:43:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/8"}}},"commit":{"parents":[{"commit":"763afa0e2e8a40dcf699004b92cb348b1ee6bc37","subject":"mpu: copy sysmeta from create request to manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/763afa0e2e8a40dcf699004b92cb348b1ee6bc37"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-20 15:42:32.000000000","tz":0},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO:\n\n- test_multi_upload unit tests all skip.\n- test/functional/s3api/test_multi_upload.py passes with one TODO.\n- Some other func/compat tests skip.\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59b4cf27274e9b7f87c8071fa8791e7d0c13200a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59b4cf27274e9b7f87c8071fa8791e7d0c13200a"}]},"branch":"refs/heads/feature/mpu"},"74bff4396845c9313a9cabe4e5561c1f7fe89645":{"kind":"REWORK","_number":9,"created":"2024-11-21 09:50:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/9"}}},"commit":{"parents":[{"commit":"a835b5a0b83cdfc9c95352814dc244e4151ef559","subject":"mpu: copy sysmeta from create request to manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a835b5a0b83cdfc9c95352814dc244e4151ef559"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 09:47:52.000000000","tz":0},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO:\n\n- test_multi_upload unit tests all skip.\n- test/functional/s3api/test_multi_upload.py passes with one TODO.\n- Some other func/compat tests skip.\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74bff4396845c9313a9cabe4e5561c1f7fe89645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74bff4396845c9313a9cabe4e5561c1f7fe89645"}]},"branch":"refs/heads/feature/mpu"},"43844ecf1fd722f1ff9ba5fef524988dc182f42f":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-11-21 14:13:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/10"}}},"commit":{"parents":[{"commit":"f35234193bafd24e414093589666f45ec31922ee","subject":"mpu: copy sysmeta from create request to manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f35234193bafd24e414093589666f45ec31922ee"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 13:33:13.000000000","tz":0},"subject":"WIP mpu: migrate s3api multi_upload to use native mpu middleware","message":"WIP mpu: migrate s3api multi_upload to use native mpu middleware\n\nTODO:\n\n- test_multi_upload unit tests all skip.\n- test/functional/s3api/test_multi_upload.py passes with one TODO.\n- Some other func/compat tests skip.\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43844ecf1fd722f1ff9ba5fef524988dc182f42f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43844ecf1fd722f1ff9ba5fef524988dc182f42f"}]},"branch":"refs/heads/feature/mpu"},"819b63b76a8553668466fd699f5e234f14597ea0":{"kind":"REWORK","_number":11,"created":"2024-11-22 15:44:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/933935/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/933935/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/933935/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/933935/11"}}},"commit":{"parents":[{"commit":"1b7f427d6e3158cb118e73836b65d1a7f2c5dfe8","subject":"mpu: return 404 early for list parts on aborted mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7f427d6e3158cb118e73836b65d1a7f2c5dfe8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 16:49:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-22 15:42:54.000000000","tz":0},"subject":"mpu: migrate s3api multi_upload to use native mpu middleware","message":"mpu: migrate s3api multi_upload to use native mpu middleware\n\nPartial migration of s3api to using native MPU. Functional and\ncompatibility tests pass.\n\nRemaining work:\n- test_multi_upload unit tests all skip.\n- test/s3api/test_mpu.py skips tests re versioning\n- still some references to segments container in s3api modules\n\nChange-Id: Ib00d55b3b787da981649c5cf0c3cc9bb199857c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/819b63b76a8553668466fd699f5e234f14597ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/819b63b76a8553668466fd699f5e234f14597ea0"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
