)]}'
{"id":"openstack%2Fswift~934031","triplet_id":"openstack%2Fswift~master~Ic40c4b104b11de9ab09485e82543521b37ea1daa","project":"openstack/swift","branch":"master","topic":"p-slo-copy-tests","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-04 18:41:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic40c4b104b11de9ab09485e82543521b37ea1daa","subject":"slo: add more functional tests for copying","status":"MERGED","created":"2024-11-04 11:21:22.000000000","updated":"2024-11-04 18:42:57.000000000","submitted":"2024-11-04 18:41:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934031-p-slo-copy-tests","meta_rev_id":"d2de72c8e17135d9fa33a1b2295a0a4ad339bc1c","_number":934031,"virtual_id_number":934031,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-04 18:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-04 17:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-04 17:45:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-04 12:24:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-04 17:45:56.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"9589f1d3af868949da371f1a474c4769605d3c16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-04 11:21:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a9a45aa79922815488780f78518c285e819925c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 12:24:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82e9cc2159ec4728bcbc4d1a6997d346\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2260684b7ec4eb69789aa295805721a : SUCCESS in 8m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a9d53deb204b4daba2b58bde611e9397 : SUCCESS in 44m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6c78f113eb049de971fe8f13c73deab : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d47ac10a3dd481db25fb8b5693744ff : SUCCESS in 21m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a6b57e8369548919c479b442f2ee303 : SUCCESS in 11m 49s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3b5839d70ec24e0a9211260eeedddc18 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/896bb2999987442cad0c97a1b79111bf : SUCCESS in 16m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3480039065ff4cee9f50adbd2787b67a : SUCCESS in 9m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/440ad53da4664d708733de2de31d464b : SUCCESS in 13m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c1a4fa4c132f44d196404bc3173c7f08 : SUCCESS in 13m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0f2dbedc445345f6824ea8f64cbcd123 : SUCCESS in 9m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0c9a842e4d34dd488c0bc81bccf9d56 : SUCCESS in 39m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35475e6d07e541c7907d98b339b1db6c : SUCCESS in 45m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c62d6634ed4cc6a11ca45e4ba3168f : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/edd09e0acb864ce0b6d86123a98700d9 : SUCCESS in 17m 45s","accounts_in_message":[],"_revision_number":1},{"id":"22d6b367fc7f8bc90c70ceba2ae4ab8ca2a1ea29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 12:56:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a729d19d8e4b0c937cb27ec2a5d106\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b9780cc6c6e0464198566b023d248bc4 : SUCCESS in 1h 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7e290b1b6905132bf4de22a355b9aeb227c60769","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-04 17:45:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b38831d56134d813bb4e39950f990ecfe5d3c64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 17:46:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba783053e8e04885d8e9e19316fbcc376a8b0a5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 18:41:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccf47bac0ffe4f2da63e5f1b0adca6f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd72632e0574acb9a7a78343244bd22 : SUCCESS in 8m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea296ae000424de685bda09a256d7f8a : SUCCESS in 38m 23s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/86f4259a78504717b221ee210b0ed571 : SUCCESS in 21m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59eaad23af4e4b019747e95686d806f4 : SUCCESS in 13m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c86df63205e49db94c49ee0e760c0e8 : SUCCESS in 24m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3eb965150b7442cdb9442431f3b81a2a : SUCCESS in 20m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/81f9aa3354f64b1695df0600b2703cfd : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/437d278bf2964fddadf82efb33987433 : SUCCESS in 12m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f40e343171c46dfb31533e0badad6a0 : SUCCESS in 18m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b414d5a7ad9e44e39ec34980642eaa1e : SUCCESS in 8m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c949a744c11245c29c9153113028ba1f : SUCCESS in 15m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1690588bb9184461bb1b46756abac218 : SUCCESS in 9m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ad11303d1354072962f26424f421029 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/24837697bdcd42879f5ed475996c6a60 : SUCCESS in 8m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/407ea37188114131b93462e34a1292c7 : SUCCESS in 52m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dad9fdeea3f34a089d97f887f8e014f0 : SUCCESS in 53m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62237fa51e7b47848a2287e4b5863d87 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf5830fadffc400197c928266a693f16 : SUCCESS in 22m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ab670af0e22a1764a585a9164590b909b365bd44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 18:41:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d2de72c8e17135d9fa33a1b2295a0a4ad339bc1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 18:42:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32eadbe02b334e1eb545745be7cfb214\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c18ca9fff743408aa98a62ec3cc32748 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/88feaf20d0ab4d46bacb07057e56cc41 : FAILURE in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"639251d4c58d0820f5917a2f7ffeff0c7b420959","revisions":{"639251d4c58d0820f5917a2f7ffeff0c7b420959":{"kind":"REWORK","_number":1,"created":"2024-11-04 11:21:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/934031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/934031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/934031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/934031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/934031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/934031/1"}}},"commit":{"parents":[{"commit":"baeef8517835ed52b2c7d09f65b8a594b9c31c2f","subject":"Merge \"CI: Bring check/gate in line with PTI for 2025.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baeef8517835ed52b2c7d09f65b8a594b9c31c2f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-30 11:05:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-04 11:21:07.000000000","tz":0},"subject":"slo: add more functional tests for copying","message":"slo: add more functional tests for copying\n\nAdd functional tests for copying from an SLO using PUT requests with\nan x-copy-from header.\n\nAdd functional tests for copying from an SLO using part-number parameter.\n\nChange-Id: Ic40c4b104b11de9ab09485e82543521b37ea1daa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639251d4c58d0820f5917a2f7ffeff0c7b420959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639251d4c58d0820f5917a2f7ffeff0c7b420959"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
