)]}'
{"id":"openstack%2Fswift~934117","triplet_id":"openstack%2Fswift~master~I3a9716d3f5f936ae7b3f5b1c73e2bb961dd610c0","project":"openstack/swift","branch":"master","topic":"rebase-parallel2","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-05 22:50:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3a9716d3f5f936ae7b3f5b1c73e2bb961dd610c0","subject":"sq? fixups for test_expirer_rebalancer.py","status":"ABANDONED","created":"2024-11-05 11:36:21.000000000","updated":"2024-11-05 22:50:23.000000000","total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"af128b6dadd4eb29a268ef11c990e9775f1cf423","_number":934117,"virtual_id_number":934117,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-05 12:22:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-11-05 15:40:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-05 12:22:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-05 15:40:05.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"25a4365c9d0f5fe6b44447b45c2e6b0bd0665739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-05 11:36:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c00b70db2c1d8e082ce34fd384eb027eefcda929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 12:22:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99fd372d5ac040778bdcb50fde7cce8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be5978a185548a1aae6e38df235c21d : SUCCESS in 9m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8260853c62b244deaf5db1bbc19a0bd7 : SUCCESS in 36m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/affdd356e48a4860afbeb8301e2a2103 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/119e3af1bbcf44439095d26580e97486 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b47f7723fdd4e05952a0f40b51973cb : SUCCESS in 21m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d32ca0bbf7c4d7aaf88a3211b564b86 : SUCCESS in 4m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ff77c6f07d6a4d3a95affe3e62a3eeb4 : FAILURE in 5m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/139224e7c8b84e1dadd0c68f65dfb234 : FAILURE in 20m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb5755789bbd46a1999a8ee602aaceda : FAILURE in 22m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b980acdc5cb4453b827e501aac68f524 : FAILURE in 27m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21f26c4f1abe4fbca4cbf6fc3e66881e : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a7b5b3b7d14347bab48142dde7e9b948 : SUCCESS in 20m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/96329cd5d4f342d686aa4de097506f10 : SUCCESS in 20m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e793190715734671ab5fb4b8f608f930 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/74010e34e63e46d691d95466bf9d6f58 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/59b7717248704548a91ad36b0c98dde8 : SUCCESS in 17m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13f5210c2dd94d9599dd7e886209daa0 : SUCCESS in 45m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d822a777ec8e48df9d18478b541d82fe : SUCCESS in 39m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5500a578108444eba541dca44d79c5ab : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fabec967ca14b858800ef3c3fa4dfcf : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/348950dd72484e778e1e41ca94ab105b : SUCCESS in 21m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7dd74a13c1d4b570b92883d04f48f51197442d73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 13:09:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de7f22ca2c8f41f5b7613ab1bb87a331\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e91ee9a3982436fb213fa6f27639dbb : FAILURE in 1h 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6d73eb548c0d3f677a3118187e07c2f6c368f452","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-05 15:40:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"af128b6dadd4eb29a268ef11c990e9775f1cf423","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-05 22:50:23.000000000","message":"Abandoned\n\nsquashed into 933373: Configurable expiring_objects_task_container_per_day | https://review.opendev.org/c/openstack/swift/+/933373","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bf663394e8b63f1e0490e94f132af79481a6444","revisions":{"8bf663394e8b63f1e0490e94f132af79481a6444":{"kind":"REWORK","_number":1,"created":"2024-11-05 11:36:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/934117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/934117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/934117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/934117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/934117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/934117/1"}}},"commit":{"parents":[{"commit":"36bf7141ead73626e8281e3f4ff989f875cb5565","subject":"Configurable expiring_objects_task_container_per_day","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36bf7141ead73626e8281e3f4ff989f875cb5565"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-05 11:35:53.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-05 11:35:53.000000000","tz":0},"subject":"sq? fixups for test_expirer_rebalancer.py","message":"sq? fixups for test_expirer_rebalancer.py\n\nChange-Id: I3a9716d3f5f936ae7b3f5b1c73e2bb961dd610c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bf663394e8b63f1e0490e94f132af79481a6444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bf663394e8b63f1e0490e94f132af79481a6444"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
