)]}'
{"id":"openstack%2Fswift~934153","triplet_id":"openstack%2Fswift~master~I9874a140027cf143df1ac670e2b8a2d09b398c1e","project":"openstack/swift","branch":"master","topic":"improve-s3-err","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-02-04 17:07:31.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}}},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-02-04 17:07:25.000000000","reason":"Change was marked work in progress"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-04 17:07:25.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I9874a140027cf143df1ac670e2b8a2d09b398c1e","subject":"s3api: fix errors for x-amz-copy-source usage","status":"NEW","created":"2024-11-05 17:07:53.000000000","updated":"2025-03-06 23:23:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":36,"unresolved_comment_count":13,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"5533664917b0db5e465f822f585c93e1380e6e60","_number":934153,"virtual_id_number":934153,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-06 23:20:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-03-06 22:09:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"reviewer_updates":[{"updated":"2024-11-05 17:08:09.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-11-05 17:08:09.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-11-05 18:27:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-17 18:53:16.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"0b50d9e528ec4382f9079b3e5ceb4bca21068193","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-05 17:07:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"223f27561b136f2e8ffd18ff73ab22eaf8679171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 18:27:17.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c9dfc152d6e472abfd897370606c2cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d33aee944740bb9c113c5fe72a02cc : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/b6bc912b836e41fab0e716933576f90f : SUCCESS in 1h 04m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9036753a82164256b77f2d1873de46dd : SUCCESS in 1h 06m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0bda9bcb8bf948cc8caf5a7385fbab02 : SUCCESS in 57m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3aa4bb8c39cb4e748a04507781d6a25e : SUCCESS in 48m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b7efbf80202472da6a8234c489914d5 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a46d82bc8c4d42c0967d0f40c90452b9 : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4ebc50f10cc41369e5888537bd89655 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a469730b840445d9d044eb2f763f0c2 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/adf85e79341842019f564fadb9bd5020 : FAILURE in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/10096d58437e47bbacaafd5e2b4ef580 : FAILURE in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c2f368fb2b524f6893a78765da78240d : FAILURE in 19m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/308ed0355fbe47b08c1af6b97a0ca75f : FAILURE in 41m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f2787a4f1e7e498bbecd5fbc6562ba93 : SUCCESS in 10m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ede06100a204cad9438a28a6cef1253 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1de8e2fe5d1e44df8149caef78cb3243 : SUCCESS in 18m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6c3e423cfecf4f868c9e610e43479e8b : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6a1f4c84238c4a1da1979ea36a3bfeba : SUCCESS in 18m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/227606a8497a4aee956c59731312929e : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/19a0c8eddc2b406d966ff55eeda535ab : SUCCESS in 4m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e54d5fb52514421ae6637c81816c486 : SUCCESS in 14m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a2fc470eee31411a923908fd09050e6f : SUCCESS in 6m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ebaf2986a784ab6936f30c9cd2e9ae6 : SUCCESS in 54m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e98afeac35d54390a5457397a93887e3 : SUCCESS in 52m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/614f6e34170041e4b00f86070a2ea4fe : SUCCESS in 47m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04e153e18c0642429b4ee3355d06c4a0 : FAILURE in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d682fa1df62f44ecaf5eef69d5fd55c8 : FAILURE in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc4674fa966c4ae39eb4af61cf81f3ea : SUCCESS in 20m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ea5a804b34a4bf59cd44a55e658da6b : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":1},{"id":"637f37d938a9ecd68b781facfb2944225e710da5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 18:44:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceeb46e3179840598af88d90173d0449\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37dc4bb58c21435bb851e5ab8a6b691d : FAILURE in 1h 34m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ce671f0d11ca07fc5d56d95f0431eae6b774f30c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-05 19:30:44.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a27b627ec3b713d23fed1cc77e743b7d932ec3ea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-21 20:50:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7553c9fc7fca952e1932f4970d65a26b8940789d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-21 21:44:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"36b96bf063235e24553674a8b229d486f27891a3","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-21 21:44:24.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2561ac63716f245e00dba2154bbe4494f168b0d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-21 21:48:42.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"36ad091a4f67138fe4646483627a65f9573d6155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 22:46:10.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c6def4edece4a21a80d7ebd3b2a2a6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d481e08f144ca69e68c22f15634d31 : SUCCESS in 7m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/2254a912e2c14cf0bd5e1a56f6a701d9 : SUCCESS in 50m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f3a8f15da1a45eba57d4a9f03975f07 : SUCCESS in 42m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62ac1a0cd01d43308cd6493392d9fc6a : SUCCESS in 33m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea6f8ffdeaef4b13bca5cf4eb08acc52 : SUCCESS in 48m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eac836980d96444097d301346fabdd35 : FAILURE in 15m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99f857d6feae4bfc904f1426b714ae67 : FAILURE in 16m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80bb38e618fc46a68a338c6d82c0d1f9 : FAILURE in 21m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae2e4dbb9456425383d0da5543ccdb37 : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/13911782791d4c92b7872d74d45d627a : FAILURE in 14m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d1a35023f7b249668391d86e291c45bb : SUCCESS in 18m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55b216a01a1646aa880303d0a6d3ee93 : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ab8a758f2c7468f964fc966b626ba3b : SUCCESS in 30m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5f067eac5dc84a5cb2503bec9a9d6c0d : FAILURE in 18m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/150e4bb95f154b9ebcc1429379b70d32 : FAILURE in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e33d131cbd904ff59e0b8c6a2cc415a4 : FAILURE in 20m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/838d6d82047547beb9c164571f709f86 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fd8bca9a565a4b6780af37b8e6db23d1 : SUCCESS in 9m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1b3328f83c5545e2a82ee9a73a27e30a : SUCCESS in 8m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/990d3c8cc1e04ffabd9e26cb3caf2f8d : FAILURE in 3m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8465c5d484ee412994a24c450027d5d9 : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/12e48279db2a4916915a3c555ed7c81c : SUCCESS in 8m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/00ab9c6c5a9f415bba4608d0dc7d57af : SUCCESS in 33m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d65360c517ed4c3d8d517ca1f3725be3 : FAILURE in 29m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b61559999f474b17aa5be370e83b093e : FAILURE in 21m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c59b19fc7bb4f4ea46f6d6fe0325139 : FAILURE in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79c025310604f41962cc3d0d7333957 : FAILURE in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91444bac0d3d4d04bab8d8a92e084a8c : SUCCESS in 17m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9d24553849845018b5d924d35b1d995 : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":4},{"id":"d632bf6de0749e62201e5c8636ab86a6b3e4bcce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:19:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d186ffd348e0493f9a656944b4c85eb7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42691742c3bd4152aac1051585e1ca3a : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":4},{"id":"74028ae5ccd89b962ab69680706c841a9a441159","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-22 17:08:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"8fc7ab6bd9b6a6b4df0267855514db5b816b7fd4","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-22 17:14:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"75130f9688905d6a0f7e6af8f6a5900601be2d7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 19:36:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"69668e7a661005b9350aab8c0e3c052b2a909735","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 20:44:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a2cda11ef84320926a24ed52e7de54\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fb4faef4c4e4e2a9ff63cb01f8c2e6d : SUCCESS in 1h 05m 59s","accounts_in_message":[],"_revision_number":5},{"id":"174768a1f3c522d5a664ad3482217ddef108d62c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 20:55:35.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1942779982e4d898fc24fb04e3cd733\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b154e20b034843a10e54de912e73da : SUCCESS in 4m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/9d14b72a779446df9f3580be45b019da : SUCCESS in 59m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47902d84ff934b179e4c4e39bfaf7dae : SUCCESS in 1h 04m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1639bc3ece9f49049876d8062e5bdf9c : SUCCESS in 58m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8851b5c1e0c74878a00a1d073fbb1318 : SUCCESS in 1h 03m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca6532d4fe5b4137968270adaec1a445 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5219069ea8c94a5ea31681cf592650bd : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf50d10e02834793919388c3809bab03 : SUCCESS in 23m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/71ed7b2b356742ad968f9c1899fec342 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/77c2af4b7ca048fbbf7fe880c5d79286 : FAILURE in 20m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/591867291a5c4b0da67a7d5c5af57334 : SUCCESS in 17m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/75463d3a2a0b4744bbfb868f76463490 : SUCCESS in 18m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/df359f78144b44e38974f632e00b8ddc : SUCCESS in 14m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19d61b6b18a2412489062cc25bad75ca : FAILURE in 21m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/02b6d94bd60b4c25a9b40f2f9b49c43f : FAILURE in 12m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7876344d852a49d5a73359635e94c008 : FAILURE in 19m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/75b7dc54549b4ddc8b72470c299e5a2d : SUCCESS in 9m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3a202944f776473880a26c7dd0c8f469 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/852267341ba74c4f8412dd356781198d : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/49cb7289ed684c5aa29e6b730c4b24ee : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/503dd237d0654cd69799f5806d1029f8 : SUCCESS in 10m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ab91d908cda64f64ab5373a196db23e9 : SUCCESS in 6m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9a394ad1d284606844819e6ce260e40 : SUCCESS in 36m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f4fd03de8c34c458104dc730aaf0c08 : SUCCESS in 26m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74684d7cc55445d6adad6156d03d6871 : SUCCESS in 23m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dc90ee1779044d8b3d6ecf7ed706c8f : FAILURE in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31051415321b499da35385e91b5528b7 : FAILURE in 2m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/61f12e103fac44fa84e96bdb15d66465 : SUCCESS in 16m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42c38bfa757045868d99efd078433c74 : SUCCESS in 36m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/798379804bc6417d86e2991d00c2e498 : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":5},{"id":"bfc374570b6b609b1d941c054ea1d20a4eef8721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 21:56:34.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0abc8525a90ca01086c92f3b3ff66cc85babdb96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 22:58:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0274d6a3fd104cab995ada28a947da62\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/189de1b42eff42b3a04489392e39d7ba : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":6},{"id":"a148cf8a870ea3cf0b7bbe5339bd9b9223836879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 23:07:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40cd33df7dcc4da6bc3c802a8880ad6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35500ba888cf4010b5ed7487ee8c39be : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/a8ac531958264d6397da936094f6b1a0 : SUCCESS in 42m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7876107204f4e18b655f65d70656527 : SUCCESS in 38m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6759df1eaea944498ae2e0027f359805 : SUCCESS in 40m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7f606885a1e462f8152f803e9884437 : SUCCESS in 1h 04m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3ff64b3dc5c4ad6a2e26ab865d88853 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab5056bd37a94a0abc258611ea3a6f51 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97103d83f1ae4f01a478e201a2d98ec0 : SUCCESS in 20m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e8ba1bd695464d089c3e295b24c2417c : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5764990b4764a3c9529c13adcc027dd : FAILURE in 17m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6459013a48814de38a989bb188374735 : SUCCESS in 18m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/636268196f4340a5902e3db23088a7e8 : SUCCESS in 22m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7f6f2d2c18bc42229ea61b9c74078b00 : SUCCESS in 29m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9b09a60243a42a6b8a27a11109ed549 : FAILURE in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/534d28690e29477a95325375b777894a : FAILURE in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7f6bbc12ad994bd8a6bd52e6c035fb4b : FAILURE in 18m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5d8c29000ac74fab9923bfbff046e6c1 : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8fde43a69cc649eea8b3141565affc2d : SUCCESS in 15m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0179efcd862b4feda48c8747f2e0e9c6 : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6c160e3b4224911bf9cfc99c4bdb4ce : SUCCESS in 5m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/41d60d215eed4652a2820ae1c8fec48b : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2b08db577f3b4e479888d4ef5200a87b : SUCCESS in 4m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/984eddbe85a24a7e8d909f0cc1282499 : SUCCESS in 51m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa8018359a6e49678473e693ce280989 : SUCCESS in 33m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db28c45ff28d45bf86155841d2e02ccd : SUCCESS in 24m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8db0ea5a6b3452e943a35e680b33af2 : FAILURE in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894e18e292e841719489ff44bfae2883 : SUCCESS in 2m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e02217ef7724e2184f81107a7c3601b : SUCCESS in 10m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f7fd31184f042729c92808d701bf144 : SUCCESS in 34m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/460e79c06e7c4d8f93fcde02942b8898 : SUCCESS in 47m 07s","accounts_in_message":[],"_revision_number":6},{"id":"1938cfce1d40e9ec7862eedd322ac93f88a4be54","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-04 01:38:01.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"581556ce191cf397e98d7b61ecf39b975dd35460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-05 16:36:22.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"75c5d079935dec999d57f8611e7c8d82335a9ffb","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-05 16:39:00.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e8c4e3e7c0d0cd749573e0e8127d46969b235d23","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-05 16:39:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bb5c97b1b979eb072e2ebaa0c5bbcae3323eca4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 17:55:49.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad5963845ff24119904cbe1faef2ca3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79adda416f5b42749cb3d56a04f76630 : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/e9e043a0b073478d80be350cfe6b26c8 : SUCCESS in 41m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf0c7523e14d4ef3bfc3f73b9c856285 : SUCCESS in 44m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cf1dbf6bd1064e81a7b0223c6e4e3de3 : SUCCESS in 47m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65289fe26419460d83efaa01aee36ebb : SUCCESS in 1h 04m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d92256da606424580a4f913a5d61989 : RETRY_LIMIT in 4m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4f5e8e4d4934ed6a9a3a783c8104a5c : RETRY_LIMIT in 5m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/caaa53de5def466b8a414d49fc99d75b : RETRY_LIMIT in 5m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6666261490ff4a4ba72f28c88773f6f8 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f79704076c5d4c0cbbffdd8b55743f37 : SUCCESS in 20m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff19ec8102cb46c49c73e1f00ec24f42 : SUCCESS in 19m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a1967be0ad6f4b7a9ed7451eef3eb7f5 : SUCCESS in 20m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8e78d8502f9f4264be20b7ad39f35cc8 : SUCCESS in 24m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c502e54694b457c96b598a0bf03d61e : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bea33c1263a741aaacd5b2ed218d893e : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e28b36960a9145cb9dfbf3504e8a9916 : SUCCESS in 16m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bbb150cbfd1b423a8b27b2e1f34794af : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/790c82b3bbb443b7a27ab70bc0454f4f : SUCCESS in 15m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2c386bfd6a1041099409313edf169416 : SUCCESS in 17m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d87f0abcc07e470c9abdf0f69279e5f7 : RETRY_LIMIT in 3m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd6ea663319a454f997e3c495b76c6f1 : RETRY_LIMIT in 2m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/96cf618b8b544b4a8a20782d7323d29f : RETRY_LIMIT in 2m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/719fd0f6d82c40b08cf2d58bd7b3c72b : RETRY_LIMIT in 2m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5900010d5b87467db4a34f4a1feadb54 : SUCCESS in 49m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/777cb4ea45f745c2ac87ef707c4cc89c : SUCCESS in 38m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfb5b9fc8f674d6fb49bf3753fdb6bed : SUCCESS in 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9782cf87f72443b92785edcb12bda29 : FAILURE in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/af9463affc84481489c8ca0cb884939d : RETRY_LIMIT in 3m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3001efa71d2c4904bcf7add4073b4f63 : SUCCESS in 47m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/097d80787b1a4ce79a89717e860e12b7 : SUCCESS in 47m 57s","accounts_in_message":[],"_revision_number":7},{"id":"6aeb76563623197c76a6b36f533f271b6806b954","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 19:43:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81dcf1a1ac5942c79914478f2a611670\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7505094208f3420fbb0bd7dbc80124bb : SUCCESS in 2h 26m 40s","accounts_in_message":[],"_revision_number":7},{"id":"374ec18412181f749dab449ccd36c81f77248ee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-05 22:43:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"9630fa81bee179ce8efc35392cc51f4e4f61ff17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 23:52:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1461bd791717403cbe2c54e6f9f2be28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f64dcb4618f444eb2ee52b3c0c53f05 : SUCCESS in 4m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/fe40feb33cf64c809d99ded55ca435d5 : SUCCESS in 35m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4a36b722390c4106a6b2b9775fb19c05 : SUCCESS in 1h 02m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09010a0c829142fda54d7d3a1865118d : SUCCESS in 33m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9884507a28f640dd8f744e16d431f236 : SUCCESS in 49m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d4245f114314e7db57ef2e4736e06a5 : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c762b462372415399dcc91afc108455 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b13d8869717e4da1b4a31ce0f8c43380 : SUCCESS in 17m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2046195558d3487cb57cd9149e8fa108 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/616320a9939944c3846e080447522860 : SUCCESS in 20m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27032592c2514ec18511d168674245fc : SUCCESS in 18m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/721e2042ff6e464ab3ec19d2fca061cb : SUCCESS in 22m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/51073f03f32f4b03af2ccaf242380ede : SUCCESS in 26m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/662c29c98581483dbb54332cccc9a5f5 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bedc845197574dbe815dd4b3346718ea : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d259d0f6e4ec412fb8f164054dd3c8ab : SUCCESS in 18m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/707c27d8950a40189703eebddf2255cd : SUCCESS in 8m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/24d948bb71d34c769f3a53534823dd2f : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d558e4a520794f2b9498ca0755f85bf7 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e28dc0eb97941229a8180cb6e7a2b90 : SUCCESS in 4m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f2ebc8dad1a4f59ac912c0afaaa709c : SUCCESS in 15m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d35977c3432a416ebca2f11e65fc18ab : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fa706c2d4fd471fa533c20cce734627 : SUCCESS in 53m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50dc1e1c56114d20b04c00eda8815fbb : SUCCESS in 24m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78ac0b276f19427c88381b08b8d5473c : SUCCESS in 50m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb44dd7a10a7421cb15822210530fbda : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c70ad556a914dc3a6e7ba00c16463a8 : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f858243106f442ec9c2a420226cf399a : SUCCESS in 17m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07fa99c65a144d04a578cd032ee9ae13 : SUCCESS in 31m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/514d1ed1c45247c580bfcf87c336ff24 : SUCCESS in 34m 29s","accounts_in_message":[],"_revision_number":8},{"id":"1fab407e7240ba046fcf0561e1e0c1dfbe5a7b09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 00:14:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35929fbe021043759c061d324b46dfd1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4086942188ef4908907b69764f74cece : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":8},{"id":"c07d78c8083b439367ce4e91560912faca3e0f5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-12 17:11:25.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"93bb3ba6a152680f9221ccddc21ec6da980a389d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 18:29:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e693543954b4867a8b82a90489667b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/506c92d0e8f145d2b70b161010fe62bb : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/658c1024e4944c69aacb18b654a7162c : SUCCESS in 57m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eafe2f2c0973428ea3d90cb0cc56f15d : SUCCESS in 54m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39aea9612e134fc8a95d744cb5ac4ac6 : SUCCESS in 59m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53d72571333f44cebaa9a4342476c6f6 : SUCCESS in 1h 12m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07fdf65d05e84086ad2d494952cc2fc0 : SUCCESS in 16m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53ad113e84ca455884713e9a28e83eec : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4da9a1a138949fa91b40c7eb66708c1 : SUCCESS in 12m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/206d24fa76404378ae5ef872b2355dd6 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3dc1f06b0894cc7a0f745ae04a25366 : SUCCESS in 9m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/76c88e4ffa254c91927fdb5ac9e475b0 : SUCCESS in 15m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bb371546465248ce9111152300e2114d : SUCCESS in 21m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c8edbb866c5e4b72a134917a71b9dbd7 : SUCCESS in 21m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/443221e1678940dda71b71df24b65cfa : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ecc8b051d044362957a4d99ef95a21b : SUCCESS in 21m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/817b70e613aa4e41ab9e095a63a07e43 : SUCCESS in 9m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/61143286b99f4b7b90c2cf8d3ad0b978 : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c8aca0a08c9e4ecea7cb7ff41577e933 : SUCCESS in 16m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/baebfeb64a5a47acacd6942ab41e5326 : SUCCESS in 19m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f72f6c82d6d64cf38fd5da28e070e68b : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c15b1c93bb264122bff785b0c7cae0bd : SUCCESS in 16m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/67b78164b66446edb52198ebffc84e56 : SUCCESS in 4m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/07507c4167a547de88a6aa427a7130b7 : SUCCESS in 57m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22aa6ee55249482d928d9adce534a165 : SUCCESS in 41m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54f4c7892e71475c84124022b4c6fcbc : SUCCESS in 25m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bef90d70ad6a438a9e4e8c3b3ad539da : SUCCESS in 22m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df67378bdfb34849b5b95dd6fb209129 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/622fa441d4bb492b8dcc227e0265dce1 : SUCCESS in 18m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7af75564b26b49cc900a274d646cfef8 : SUCCESS in 44m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8b964e4b7c44c0ca9339d85e2087766 : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":9},{"id":"c54bce8fd16d5ca4942018593be50343d4e43b33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 19:55:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/118029408bb14e22aa9e740fd2c1dfe6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef250e496428473480b58379ca27db55 : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":9},{"id":"d0b49d51bc424af18fa48699cab65dd6e99c8f3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-17 17:11:17.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"860390cdd9c3291d335ee8ade51593d905f65a2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 18:32:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76964631c7614e4b9bf3306c87f57bd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d54ca1a33c4580bde934f339d99a1f : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/43cca7f6f84d4411b8289fa70b54b983 : SUCCESS in 42m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a9632d79d904656a7a0377d537fe13a : SUCCESS in 47m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f6fcd432ac834075a323ba48ddcaa9c5 : SUCCESS in 40m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d70c71173c554be38d6e96b31b216420 : SUCCESS in 1h 05m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da837ef0853d46b5a7a912c85d92abf6 : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e82d17b81b1f48edb3a72057df8cb78e : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/669684d3676e401186b65495203499a2 : SUCCESS in 21m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b4b3cb50f994238ac950927e7393222 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/439c8a5c6ed544d0b70223910033886e : SUCCESS in 23m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1b9f5abf48bc483288a1dc2b3a942268 : SUCCESS in 10m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/df0b0041df914de79af466f423cc41f1 : SUCCESS in 20m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5b441dd6f5f04597a33dbf22225c443e : SUCCESS in 19m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c2679fdae9204fbfa65be587b5796fa0 : SUCCESS in 11m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73572661167e4a54aa9ddd47ea589daf : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/996c09a4e1274b63aa01bce1db372231 : SUCCESS in 10m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/56f4e3382fc947d8b6ce21162e79e312 : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c5fe8f2bbefa4e88beb1a03c85390c1c : SUCCESS in 16m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/02aac18fe1a4401e95eba27888e0885b : SUCCESS in 13m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d515c81339e04f1ab89d1deaf3a62060 : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/38494f9b47354b70ae25debab18c47b4 : SUCCESS in 15m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0026352b92a94a27a20b599591dc3edb : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b94664eddb6d46639fcafb47cfc86c32 : SUCCESS in 47m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77f563d88e314259bd83a8a27655f045 : SUCCESS in 36m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9af1c2cfc02347368a434648a64fdefa : SUCCESS in 35m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7cd47a184bb45da94e6192148836cd7 : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbb09b74313416cb510cdf492ca8e84 : SUCCESS in 3m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/556dc9bf1d1d4f9b9c6ba2cbc9bdf158 : SUCCESS in 16m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7cbd23d53c3459f80d0da30525e5dd4 : SUCCESS in 45m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43bf6a7eb7ca49179fc51db00ad7038c : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":10},{"id":"fd2967ef4b525bad62db7e50bfde9a5fc5681f41","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-17 18:53:16.000000000","message":"Patch Set 9: Code-Review-1\n\nCopied votes on follow-up patch sets have been updated:\n* Code-Review-1 has been copied to patch set 10 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\").\n\n(12 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0c7dc6134f886bf28b18284cf9856e4952838d3f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-17 20:48:28.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e76d2fb71b757c7765d4e79c95e9b7f69aca5439","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 22:34:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49f68b3e09ce40339db504b5898fbcc0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f92afce079a3462aa608a77a120d6e71 : SUCCESS in 1h 57m 21s","accounts_in_message":[],"_revision_number":10},{"id":"92c2af29da2ea73edfd04b61b2c3154f624d14e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-13 17:11:02.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"91e0e9d2f93dcb7cc036d5e286aea724abfa7d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:39:22.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c6268d03b8d4b48b8ff46bfdef1c701\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39a2025f506496082fe0ccec724e424 : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/c5f0d018ab2346e4bbe6c0a22ee92d93 : SUCCESS in 1h 03m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7ad72d8ceb2a4dc3a772437e78374817 : SUCCESS in 59m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e0b66054e064ec286b11dc89d4d48d6 : SUCCESS in 51m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b191feffa1d44043996a714500253f61 : SUCCESS in 1h 03m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/746f864891a7495d833ef6f133b47750 : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3365b32d925b4c62971868eeeefd6626 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/825697e8686c4a0b9b997946ddf57d91 : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e38f219d14d41a6889f67de5a412790 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24c4446acff7444d8a2fd8c674529dae : SUCCESS in 15m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/271624f856564a3b9f6b5ea9c1f65921 : SUCCESS in 20m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1c3c1a8c2f9f403eaaac0328e310768b : SUCCESS in 33m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ae0f97746ad24ca3a12aee650a684d78 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9241834754dd4df6b209936eb90c1dc2 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1e8cb457c0a4467693d1c8277801d80b : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/204aa69e1dd04a81bb6dcd6ccc3122c0 : SUCCESS in 8m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3151a4e4a4d3487a820e881842f3538f : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5af9ff82ef624fa9976c661dfea5948e : SUCCESS in 5m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d180f147031043509be552484917c7e9 : SUCCESS in 57m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0982578d80354a0d98e0e92edb26c77b : SUCCESS in 48m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/69b5bbd78edf42edb884b10466dec936 : SUCCESS in 53m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7804ce6c5d646d79c7f5c97e9a99ff8 : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9523ebea77497d98a464aecff854cf : FAILURE in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/64d3173b14e547d4b46b23cac41ebe5e : SUCCESS in 20m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd5b0084576b4f63af0d0076229a5633 : SUCCESS in 38m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9297be3d3cd84facb0ce075e7c4504a3 : SUCCESS in 48m 15s","accounts_in_message":[],"_revision_number":11},{"id":"c49a430c4da3f259e9f67826bc96a641347ecf67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 21:30:49.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9263358308464a1e8f370f69f8a216d3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff2576b8ecc146cd8d1e6912b148561b : TIMED_OUT in 3h 01m 01s","accounts_in_message":[],"_revision_number":11},{"id":"33c6921159d3bead49753af9d31d1fdce62fdbf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-03 15:51:52.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"12cc61114216734a457b1910cdb15b7108fb595f","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-03 15:56:07.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d6139aeb9bdf7e0eba169eb79c456a1d3b99b34a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-03 16:39:56.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":13},{"id":"e56df08d5ab7c44fd433639c6a9b55bc2be5bfc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 17:41:20.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50f2763914554c8da78e997446fda836\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fc0facecf44ff780d5059f4be7a639 : SUCCESS in 7m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/edd2e9419d7a4129b9d8c93b7c410879 : SUCCESS in 46m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/439c0bd6edcc41fdb866d97d516cb602 : SUCCESS in 46m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/023f1c1430e0411f8ad2b8222e7dc1e9 : SUCCESS in 39m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96b2ce372c7647ab982b83da19b8d130 : SUCCESS in 38m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2311c6d13a7499c8847c1ee4ceda960 : FAILURE in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59f76e81a82c43678723f2b6096b209f : FAILURE in 18m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88921724775b47f5a3b998a138a91d2a : FAILURE in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/587c47e1adcd46fba6ded487b7fda1b8 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/538942e7e613408e9c9baea575628eb1 : FAILURE in 10m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b6696c772b324addbe469cf5d4c1d27c : FAILURE in 23m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c85c7b5d100b4c79ab46448c29e803e5 : FAILURE in 30m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d639235d1b284ce3a5e6421bc32aafba : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a3a019685a9b4f528c247d4bac9e6130 : SUCCESS in 13m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/db49b39544ee4c719ae8310e2916b8ed : SUCCESS in 9m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/66291e059bf34fb681bc7c82ca6773bd : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a829148c5ea04d5095ae43afd3a99bad : FAILURE in 14m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9bee8ce8b8c04f14bf90c0afacb7d216 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67aa75a02ad54d1e84c75c5968cbee35 : FAILURE in 37m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0eafafd07f74c6ebfc48243c58d27f8 : SUCCESS in 53m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/178ec578e91e4a3c87163c0e9dbcfbeb : SUCCESS in 55m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab24801721494c7981ad517d9a4ab580 : FAILURE in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe278086f38945e588f7a5233a6f7d0a : FAILURE in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01057b8432ad4b6093c3e8a12b9ec732 : SUCCESS in 18m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5a698e09eab4b7fab199c318f8bfeb2 : SUCCESS in 37m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfa271b65df74a188be67f5a6d6ab3d7 : SUCCESS in 45m 04s","accounts_in_message":[],"_revision_number":13},{"id":"d8b296594f023153636d14704eae2a4ab1e903d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 19:16:16.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21b0aadfee0742e889b02e607b7b03ea\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5402ac447cf642e9a3951c3b823502ff : FAILURE in 1h 18m 42s","accounts_in_message":[],"_revision_number":13},{"id":"8fd00a81a222a6aafe9d2e6c3631c9e8e73c349b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-04 17:07:25.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":13},{"id":"af6e5d72674620e9cea337f856ecb5cb68f55652","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-04 17:07:31.000000000","message":"Patch Set 13: Code-Review-1","accounts_in_message":[],"_revision_number":13},{"id":"f436462dd5b5e2dbe95d4ef65211d157d4d5b5d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-03-06 22:09:57.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"65d7ac5be4f608c133b7e69f387f1a08ea6672ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 23:20:44.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d75b7157fffa4c05a90681993b6f6fcc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9fc8988e0f94c7ba4239427c0f63b8b : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/2e9a72fd3fb048198cb6587281fb226d : SUCCESS in 52m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ceaee56965c843feaa294826f95e5ef5 : SUCCESS in 58m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a20e01bd81a4aa29bcd82987de4b17f : SUCCESS in 50m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4785ac84d46e4577bf83dfd9ec4c29c1 : SUCCESS in 1h 04m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba3b8c429cdb4c259c3c42b385d3f095 : FAILURE in 21m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59d412b03f5f4837b5e2b742828785cf : FAILURE in 20m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebe144eedf79412cab40952e5b66a9ee : FAILURE in 22m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/970e150c7ed6419e87f4bf30b9eb881e : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3674bba0ce634e758fca88ae2cd6d1c1 : FAILURE in 21m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1485e1841fd0432b9088d45631d38277 : FAILURE in 23m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e4531568219146e5a377ade3ace427d4 : FAILURE in 29m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7034383699f9475b9cd0b508e07e310e : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e506253799f8450da5be06c4688b9809 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/01aa6069ae284b1fa1961e3aab9b63d7 : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ba63881169343d5998ff0ac51e413d8 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c73a7ac78723472785ae908b20d95ba9 : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/20028ef7cca349c8b35afff4dfb39087 : SUCCESS in 6m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/20471baad87a46c3af422b187377ef12 : FAILURE in 53m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9bfd3edfd724c4ba375cf0f444f78d6 : SUCCESS in 49m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7444624bdea4e60a4ecd37f043ff5a5 : SUCCESS in 35m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9662c2c05d9409c847d62e07a988955 : FAILURE in 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b237302ee87642bdaedb212785526d72 : FAILURE in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8e2e8d0d6e0c4f0a87ac617515891a1f : SUCCESS in 20m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36f6ec1cd399482da7ceb771095fd877 : SUCCESS in 42m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d415b371a4b74f55b7446f0528deeb0f : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":14},{"id":"5533664917b0db5e465f822f585c93e1380e6e60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 23:23:24.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4892d24b9d8a47dba9a98d0ee79d087e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/528ff276c3d044ba90f7cae6c2edc6a2 : FAILURE in 1h 11m 19s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"fafdd4c5ba6ff1e106da56f94e6a35dfa35cdabb","revisions":{"310422bbafa7a645dc5628623887415d5ac4ae99":{"kind":"REWORK","_number":1,"created":"2024-11-05 17:07:53.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/1"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 17:05:15.000000000","tz":-480},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware\nwhen we send incorrect/invalid values for the\naforementioned header. More coverage in cross-compat\ntests for the same\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/310422bbafa7a645dc5628623887415d5ac4ae99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/310422bbafa7a645dc5628623887415d5ac4ae99"}]},"branch":"refs/heads/master"},"933f1d48779075c575af0da0295335774ad32c53":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-21 20:50:09.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/2"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-21 20:50:09.000000000","tz":0},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware\nwhen we send incorrect/invalid values for the\naforementioned header. More coverage in cross-compat\ntests for the same\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/933f1d48779075c575af0da0295335774ad32c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/933f1d48779075c575af0da0295335774ad32c53"}]},"branch":"refs/heads/master","description":"Rebase"},"fb5daeea3925148dce7f7a4edbb092b9e569f0b2":{"kind":"REWORK","_number":3,"created":"2024-11-21 21:44:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/3"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-21 21:43:58.000000000","tz":-480},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware\nwhen we send incorrect/invalid values for the\naforementioned header. More coverage in cross-compat\ntests for the same\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb5daeea3925148dce7f7a4edbb092b9e569f0b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb5daeea3925148dce7f7a4edbb092b9e569f0b2"}]},"branch":"refs/heads/master"},"4552769cbb070a0448a2180ffae17da9aaae062d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-11-21 21:48:42.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/4"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-21 21:48:42.000000000","tz":0},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4552769cbb070a0448a2180ffae17da9aaae062d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4552769cbb070a0448a2180ffae17da9aaae062d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"10ebeff1fbacdfad4b8bea0e1589aa75a3ba7a7c":{"kind":"REWORK","_number":5,"created":"2024-12-03 19:36:18.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/5"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-03 19:20:47.000000000","tz":-480},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10ebeff1fbacdfad4b8bea0e1589aa75a3ba7a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10ebeff1fbacdfad4b8bea0e1589aa75a3ba7a7c"}]},"branch":"refs/heads/master"},"52421981fd629f987a7039e41fce351803ee812a":{"kind":"REWORK","_number":6,"created":"2024-12-03 21:56:34.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/6"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-03 21:56:17.000000000","tz":-480},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52421981fd629f987a7039e41fce351803ee812a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52421981fd629f987a7039e41fce351803ee812a"}]},"branch":"refs/heads/master"},"b54548203a6461bacf9497f0a4e09c30af0612a3":{"kind":"REWORK","_number":7,"created":"2024-12-05 16:36:22.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/7"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-05 16:36:18.000000000","tz":-480},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b54548203a6461bacf9497f0a4e09c30af0612a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b54548203a6461bacf9497f0a4e09c30af0612a3"}]},"branch":"refs/heads/master"},"a711f7edc98922fc959cfe331a724b4faace44a5":{"kind":"REWORK","_number":8,"created":"2024-12-05 22:43:40.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/8"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-05 22:43:33.000000000","tz":-480},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a711f7edc98922fc959cfe331a724b4faace44a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a711f7edc98922fc959cfe331a724b4faace44a5"}]},"branch":"refs/heads/master"},"7433672f2631c1fa08c0f611fb84650fd27cdf01":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-12-12 17:11:25.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/9"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-12-12 17:11:25.000000000","tz":0},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7433672f2631c1fa08c0f611fb84650fd27cdf01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7433672f2631c1fa08c0f611fb84650fd27cdf01"}]},"branch":"refs/heads/master","description":"Rebase"},"b12c8849af7f82305bff1108bf63cc9574cfa1ae":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-12-17 17:11:17.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/10"}}},"commit":{"parents":[{"commit":"4faa3523c9c0522c4f28a3028398949be4934483","subject":"Merge \"CI: Configure bandit better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4faa3523c9c0522c4f28a3028398949be4934483"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-12-17 17:11:17.000000000","tz":0},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b12c8849af7f82305bff1108bf63cc9574cfa1ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b12c8849af7f82305bff1108bf63cc9574cfa1ae"}]},"branch":"refs/heads/master","description":"Rebase"},"81ffd0e8981e9ab7849c617fa46545f36d94bd9b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-01-13 17:11:02.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/11"}}},"commit":{"parents":[{"commit":"94d3a5dee8122e14be0b00694d65c448ce47ea31","subject":"Merge \"obj: Add option to tune down etag validation in object-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94d3a5dee8122e14be0b00694d65c448ce47ea31"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-01-13 17:11:02.000000000","tz":0},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81ffd0e8981e9ab7849c617fa46545f36d94bd9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81ffd0e8981e9ab7849c617fa46545f36d94bd9b"}]},"branch":"refs/heads/master","description":"Rebase"},"5812709aff38b7f46c905a40ce0792f46558bfcf":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-02-03 15:51:52.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/12"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-02-03 15:51:52.000000000","tz":0},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5812709aff38b7f46c905a40ce0792f46558bfcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5812709aff38b7f46c905a40ce0792f46558bfcf"}]},"branch":"refs/heads/master","description":"Rebase"},"1ac4c8ab630fff3cbfbff38765ec3b64d2da5515":{"kind":"REWORK","_number":13,"created":"2025-02-03 16:39:56.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/13"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-02-03 16:39:47.000000000","tz":-480},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ac4c8ab630fff3cbfbff38765ec3b64d2da5515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ac4c8ab630fff3cbfbff38765ec3b64d2da5515"}]},"branch":"refs/heads/master"},"fafdd4c5ba6ff1e106da56f94e6a35dfa35cdabb":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-03-06 22:09:57.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/934153/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/934153/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/934153/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/934153/14"}}},"commit":{"parents":[{"commit":"5554f432f5d285c82cd82e40607067cd7a214542","subject":"Merge \"tests: Allow configuring normalized_urls \u003d false\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5554f432f5d285c82cd82e40607067cd7a214542"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 14:01:31.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-03-06 22:09:57.000000000","tz":0},"subject":"s3api: fix errors for x-amz-copy-source usage","message":"s3api: fix errors for x-amz-copy-source usage\n\nWe want the correct error from our s3api middleware when we send incorrect/invalid values for the aforementioned header.\n\nDrive-By Change: Add and fix other s3api error responses\n\nChange-Id: I9874a140027cf143df1ac670e2b8a2d09b398c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fafdd4c5ba6ff1e106da56f94e6a35dfa35cdabb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fafdd4c5ba6ff1e106da56f94e6a35dfa35cdabb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
